)]}'
{"id":"openstack%2Ftooz~911682","triplet_id":"openstack%2Ftooz~master~I2a3e0d08b6742ee7b473bc382bfc021523719251","project":"openstack/tooz","branch":"master","topic":"bug/2056429","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-03-07 17:08:09.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"I2a3e0d08b6742ee7b473bc382bfc021523719251","subject":"Fix sentinel tests not running","status":"MERGED","created":"2024-03-07 10:01:42.000000000","updated":"2024-03-07 19:57:30.000000000","submitted":"2024-03-07 19:56:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"911682-bug/2056429","meta_rev_id":"0052bbc7cd51151cfdecc299f13fac0ea0e0a138","_number":911682,"virtual_id_number":911682,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-03-07 19:56:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":2,"date":"2024-03-07 13:52:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-03-07 14:28:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"all":[{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-03-07 14:31:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-03-07 11:17:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-03-07 13:52:04.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2024-03-07 14:28:40.000000000","updated_by":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"reviewer":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"state":"REVIEWER"}],"messages":[{"id":"c0f1eeb9ab47f3db2e61b5b309dd4603c0bd25a4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-03-07 10:01:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"81c5b02fba3d4a15d58e9d7812c13f9a943f84fd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-03-07 10:02:17.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"907af9613e6a25a83966abf03a47feecc01bd675","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-07 11:17:05.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1fe37658d9d14fd4963ce31d9dee4ed2\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/322c23258029485e80b99cc21e1b71fb : SUCCESS in 1h 12m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6957cded0fd14439b647d3cdd9b11de2 : SUCCESS in 4m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b690f076ebeb4d19a61a0d4d9c6f1372 : SUCCESS in 3m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2eda49591022490c957f4a58495dee0b : SUCCESS in 4m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/59920475569a41219e8d8ba25ad4ec82 : SUCCESS in 4m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bcc0131fec49474f82d3f04d0b49f27a : SUCCESS in 4m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c7905d87a73040de95777907c0c98730 : SUCCESS in 3m 20s\n- tooz-tox-py38-etcd3gw https://zuul.opendev.org/t/openstack/build/befef0525ead465dbe2c0c1e6ebbeea6 : SUCCESS in 6m 18s\n- tooz-tox-py310-etcd3gw https://zuul.opendev.org/t/openstack/build/401a1be196264de99dfa900ee96d4a2f : SUCCESS in 6m 11s\n- tooz-tox-py38-zookeeper https://zuul.opendev.org/t/openstack/build/faad831f7d444c12bd0fa724d0953ebb : FAILURE in 4m 21s\n- tooz-tox-py310-zookeeper https://zuul.opendev.org/t/openstack/build/bc566c3d2f1e467291615dc37f0270a9 : SUCCESS in 4m 22s\n- tooz-tox-py38-redis https://zuul.opendev.org/t/openstack/build/69db9d67a7964ce4ad1c4e0440340fd2 : SUCCESS in 3m 43s\n- tooz-tox-py310-redis https://zuul.opendev.org/t/openstack/build/c43b2dd107684e889c783676a0751c6c : SUCCESS in 5m 57s\n- tooz-tox-py38-sentinel https://zuul.opendev.org/t/openstack/build/9e96c564d086407eb6a67fd87e7b0f5d : SUCCESS in 5m 23s\n- tooz-tox-py310-sentinel https://zuul.opendev.org/t/openstack/build/3e737779155541fa865bd1dda7856c4e : SUCCESS in 5m 55s\n- tooz-tox-py38-memcached https://zuul.opendev.org/t/openstack/build/1e7890e820de43a68962be27ce728cdd : SUCCESS in 3m 59s\n- tooz-tox-py310-memcached https://zuul.opendev.org/t/openstack/build/96affe0647be457b9d27f41658f8d256 : SUCCESS in 4m 09s\n- tooz-tox-py38-postgresql https://zuul.opendev.org/t/openstack/build/140cfc7183214135945c268de53da1f1 : SUCCESS in 3m 46s\n- tooz-tox-py310-postgresql https://zuul.opendev.org/t/openstack/build/aabe11eb0b6e41a98f7c5ee9a9560c6d : SUCCESS in 4m 12s\n- tooz-tox-py38-mysql https://zuul.opendev.org/t/openstack/build/0d4e6c38eb2449448a1513904bc03939 : SUCCESS in 4m 22s\n- tooz-tox-py310-mysql https://zuul.opendev.org/t/openstack/build/539df6cab57d4f7194399a81939c553f : SUCCESS in 5m 59s\n- tooz-tox-py38-consul https://zuul.opendev.org/t/openstack/build/134a1a66a2224eb4a8adf5544c80548b : SUCCESS in 4m 25s\n- tooz-tox-py310-consul https://zuul.opendev.org/t/openstack/build/7f229281c99b49daa8a3a9b764682051 : SUCCESS in 3m 51s","accounts_in_message":[],"_revision_number":2},{"id":"61bcd3a2b36b31e843ab6507bd019ed3027df35b","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-03-07 12:31:57.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b6614a70eb01bc76500305e613048dbb8ba8d6ea","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2024-03-07 13:52:04.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"82d927f295e2bded974531500ecc108087b76361","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-07 14:01:35.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/21ff74304e184abf8358128463b8803a\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9863b392d53b47399ff4fa7e3c1df26b : SUCCESS in 1h 28m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/539c6b16c6a04feea43e0543cb954ec5 : SUCCESS in 4m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/584593ba5eca4a93abad3e59297c21a8 : SUCCESS in 5m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3cc1bfce91ec4bb7abe4debfbe185a32 : SUCCESS in 4m 18s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e83c97d2a1b84bbbab7f1fb85e67dde0 : SUCCESS in 4m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d9b19369f0246eeb9f5092ca48bdaca : SUCCESS in 4m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/aa80a0ac8fae414199b29d690d3e7b1a : SUCCESS in 5m 21s\n- tooz-tox-py38-etcd3gw https://zuul.opendev.org/t/openstack/build/a92c353fd4cb4aeeaef249c3669443ff : SUCCESS in 4m 34s\n- tooz-tox-py310-etcd3gw https://zuul.opendev.org/t/openstack/build/64a2b2cfebd14c7c911fbdea460c4a94 : SUCCESS in 4m 34s\n- tooz-tox-py38-zookeeper https://zuul.opendev.org/t/openstack/build/334ba163ab6e47e9b5d6fce62a80c3e4 : SUCCESS in 4m 06s\n- tooz-tox-py310-zookeeper https://zuul.opendev.org/t/openstack/build/af888e030a58421bb246921e0e2b6d55 : SUCCESS in 4m 14s\n- tooz-tox-py38-redis https://zuul.opendev.org/t/openstack/build/b01bf991c8b44f35b0e3c65318257858 : SUCCESS in 4m 12s\n- tooz-tox-py310-redis https://zuul.opendev.org/t/openstack/build/ad49a0f690bc40b18d7d3a7c7039bdc9 : SUCCESS in 3m 47s\n- tooz-tox-py38-sentinel https://zuul.opendev.org/t/openstack/build/8399d980aa1d476c88fbbcc033cc9856 : SUCCESS in 4m 07s\n- tooz-tox-py310-sentinel https://zuul.opendev.org/t/openstack/build/c6958062d25247ceb536c88ad7ad03f5 : SUCCESS in 4m 10s\n- tooz-tox-py38-memcached https://zuul.opendev.org/t/openstack/build/c138bf9010a9473f93bda7e6d07c3b68 : SUCCESS in 3m 45s\n- tooz-tox-py310-memcached https://zuul.opendev.org/t/openstack/build/da932cc08ebc4a0694b5f71dd9d005eb : SUCCESS in 4m 08s\n- tooz-tox-py38-postgresql https://zuul.opendev.org/t/openstack/build/dc9b3a28cddd41e49f3a46b54a17d660 : SUCCESS in 4m 06s\n- tooz-tox-py310-postgresql https://zuul.opendev.org/t/openstack/build/900ed8102ae64baab267711def9c7eac : SUCCESS in 3m 33s\n- tooz-tox-py38-mysql https://zuul.opendev.org/t/openstack/build/2e427cff933e4fc5922297b03de67996 : SUCCESS in 4m 08s\n- tooz-tox-py310-mysql https://zuul.opendev.org/t/openstack/build/46db652d64334b3093a9f174603e720c : SUCCESS in 4m 25s\n- tooz-tox-py38-consul https://zuul.opendev.org/t/openstack/build/f81a2911ab2c40d289de6325993dccd2 : SUCCESS in 4m 01s\n- tooz-tox-py310-consul https://zuul.opendev.org/t/openstack/build/44daea6014cd4da899ca6265a3db1f89 : SUCCESS in 3m 37s","accounts_in_message":[],"_revision_number":2},{"id":"8d4e07534379352d6dc12737196fb76473a895a8","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2024-03-07 14:28:40.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"16a111f6e7674310134a7c3adb907c0b3f43170b","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2024-03-07 14:31:01.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"ee1170622a1e6310eb60f87ccb7b92948eb6fe11","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-07 14:33:34.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"0262b08e4668f79a8e5c40d266eb37c7a5f052c4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-07 16:32:07.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6118d8b1b11a41f39842fb8616da410d\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6fd7b5ebf6364f31b07d5b63875e5780 : CANCELED in 12m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0e888a2329754f12986b8ac83150fbca : SUCCESS in 4m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a082fc2db5b843df923db3fae1acadb1 : SUCCESS in 6m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8713f8dec05a4b5493cef83b2a2f9600 : SUCCESS in 6m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/89a1621d9c7f4946a669717a737b9319 : SUCCESS in 3m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/45522da2f54b4545b34c3e9de942dc95 : SUCCESS in 3m 52s\n- tooz-tox-py38-etcd3gw https://zuul.opendev.org/t/openstack/build/faa89679fe12468d96f0b2732559f010 : CANCELED\n- tooz-tox-py310-etcd3gw https://zuul.opendev.org/t/openstack/build/92c48cbc786d47cba553371faa99a39d : CANCELED\n- tooz-tox-py38-zookeeper https://zuul.opendev.org/t/openstack/build/adeaef2a534c4e72a84fa8877fa75034 : SUCCESS in 3m 41s\n- tooz-tox-py310-zookeeper https://zuul.opendev.org/t/openstack/build/704cc2a97700474caa9787fd1a17ffc8 : SUCCESS in 4m 56s\n- tooz-tox-py38-redis https://zuul.opendev.org/t/openstack/build/db57d24f42b64ce98cb1899adae155a1 : SUCCESS in 4m 24s\n- tooz-tox-py310-redis https://zuul.opendev.org/t/openstack/build/baef7a47aaa845dc9af72c646eeb4ccf : SUCCESS in 3m 43s\n- tooz-tox-py38-sentinel https://zuul.opendev.org/t/openstack/build/00ecefc930c4461fbbeb0ee29da186e7 : SUCCESS in 4m 19s\n- tooz-tox-py310-sentinel https://zuul.opendev.org/t/openstack/build/7eed51f5dc64484d8a2c163c09b32732 : SUCCESS in 4m 54s\n- tooz-tox-py38-memcached https://zuul.opendev.org/t/openstack/build/3aa37c8ed6644bbe960293bdcf28e683 : CANCELED\n- tooz-tox-py310-memcached https://zuul.opendev.org/t/openstack/build/bddbf28d15f344aab3616802f1055dff : CANCELED\n- tooz-tox-py38-postgresql https://zuul.opendev.org/t/openstack/build/25722e0b1f6f459fa94ceac98c7f3bd7 : CANCELED\n- tooz-tox-py310-postgresql https://zuul.opendev.org/t/openstack/build/55ced3c6847a40e5a8625d42c0bf0f20 : CANCELED\n- tooz-tox-py38-mysql https://zuul.opendev.org/t/openstack/build/357ec86e2939412b8cf8692029465ebc : CANCELED\n- tooz-tox-py310-mysql https://zuul.opendev.org/t/openstack/build/2fbeae0410a34b4c9ce5bb4409c2c454 : CANCELED\n- tooz-tox-py38-consul https://zuul.opendev.org/t/openstack/build/26d68bfc67124a8ebae28fe5e856fb7e : CANCELED\n- tooz-tox-py310-consul https://zuul.opendev.org/t/openstack/build/401aa3fe538e4f3fad4f4a96dd304478 : CANCELED","accounts_in_message":[],"_revision_number":2},{"id":"c7d644ec75c86e43c527fce5f6ce7eca6b6522ea","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-03-07 17:08:09.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"08a623e55cb6d7ce9b1d612d68669177ad33edc2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-07 18:38:28.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a8512af0c8ed4aab9db2ee7b8af01703\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e43d0769062c4af89d7b7e1a7ea7b943 : SUCCESS in 1h 24m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5cf0a21a3bce4985b492498fcc9b7e1b : SUCCESS in 4m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d196a8bc56f1479894887cdf7b5806d8 : SUCCESS in 3m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/81852c30331e435f8b7fd7abcd0368e3 : SUCCESS in 4m 06s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a1a712a52893462ebe5c6a31fb291a47 : SUCCESS in 4m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/321bb99128a249018ce0659250812e4c : SUCCESS in 4m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/38d890eef9bf42ec80c00031b55ab5f7 : SUCCESS in 4m 11s\n- tooz-tox-py38-etcd3gw https://zuul.opendev.org/t/openstack/build/e5c9079f01d949bb953197347164665b : SUCCESS in 6m 15s\n- tooz-tox-py310-etcd3gw https://zuul.opendev.org/t/openstack/build/584b3697701e48858d7ec804d7854271 : SUCCESS in 4m 20s\n- tooz-tox-py38-zookeeper https://zuul.opendev.org/t/openstack/build/5a97c234a62d45409320507eb42f07ec : SUCCESS in 4m 20s\n- tooz-tox-py310-zookeeper https://zuul.opendev.org/t/openstack/build/cd8a4568545e49588a0a949072ae70f9 : SUCCESS in 4m 26s\n- tooz-tox-py38-redis https://zuul.opendev.org/t/openstack/build/e82a6761ec5445fdb2e32db5e4208b1f : SUCCESS in 4m 59s\n- tooz-tox-py310-redis https://zuul.opendev.org/t/openstack/build/7642275070d84f46b063aa665ae2c0e0 : SUCCESS in 6m 08s\n- tooz-tox-py38-sentinel https://zuul.opendev.org/t/openstack/build/7116b570ec944eaa8bff489398c0fd9b : SUCCESS in 4m 56s\n- tooz-tox-py310-sentinel https://zuul.opendev.org/t/openstack/build/eeff3ffcc9f044f79183fc909f2073f0 : SUCCESS in 5m 51s\n- tooz-tox-py38-memcached https://zuul.opendev.org/t/openstack/build/22f700fb7da44792bbbfd9bd17a4ab41 : SUCCESS in 4m 51s\n- tooz-tox-py310-memcached https://zuul.opendev.org/t/openstack/build/bb32a5d12bfd4c559cc29c4a52268f1a : SUCCESS in 5m 51s\n- tooz-tox-py38-postgresql https://zuul.opendev.org/t/openstack/build/87e8e071b073470e8f2862f94b18d2fe : SUCCESS in 5m 02s\n- tooz-tox-py310-postgresql https://zuul.opendev.org/t/openstack/build/4b46262ad4174f73b90028befab0ab63 : SUCCESS in 3m 55s\n- tooz-tox-py38-mysql https://zuul.opendev.org/t/openstack/build/63ee17c91cff43a78c28f0c4e73aa62c : SUCCESS in 4m 24s\n- tooz-tox-py310-mysql https://zuul.opendev.org/t/openstack/build/b065a21c565246f2bdd45a4eb22da98c : SUCCESS in 4m 29s\n- tooz-tox-py38-consul https://zuul.opendev.org/t/openstack/build/952410e3c5614e44a323f83f8692c24f : SUCCESS in 4m 26s\n- tooz-tox-py310-consul https://zuul.opendev.org/t/openstack/build/4a36745df62f454b9525a2396b9ecaaf : SUCCESS in 4m 46s","accounts_in_message":[],"_revision_number":2},{"id":"6090cc400ea820ba8e0bc3041a77ad7476ff6359","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-07 18:38:56.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"c22c0c77a4dbdc92672b63bbde4aa847a970da64","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-07 19:56:34.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6c8757e4a078477792322a6aff78952a\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bf62497d2b75408cada35f6b33d83c34 : SUCCESS in 1h 16m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/94ad567f43d240ed9888635b55175ecf : SUCCESS in 4m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/54e223532f0742bfb4181dc8b0dd7aff : SUCCESS in 5m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/23519e2e558c44ca8c73765a694abe0f : SUCCESS in 5m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7eb74ddf82894de1ad9c51ac8383fb50 : SUCCESS in 5m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/abffe03c83c240e1ae9fc49930c94768 : SUCCESS in 4m 04s\n- tooz-tox-py38-etcd3gw https://zuul.opendev.org/t/openstack/build/80564beaef18462da107a0ebc3c64d3f : SUCCESS in 4m 27s\n- tooz-tox-py310-etcd3gw https://zuul.opendev.org/t/openstack/build/11bbd097baed4a7cb8fd6b668e38de30 : SUCCESS in 4m 22s\n- tooz-tox-py38-zookeeper https://zuul.opendev.org/t/openstack/build/0084269138f2448f81be3556f2aa8206 : SUCCESS in 4m 35s\n- tooz-tox-py310-zookeeper https://zuul.opendev.org/t/openstack/build/dd528ab8554f496fa30b96a037caaf84 : SUCCESS in 3m 56s\n- tooz-tox-py38-redis https://zuul.opendev.org/t/openstack/build/90ce8afa4d2744cd8219d2c5dabfcde9 : SUCCESS in 4m 35s\n- tooz-tox-py310-redis https://zuul.opendev.org/t/openstack/build/5f4701e7d6ba470a87e7faa5e5676497 : SUCCESS in 4m 06s\n- tooz-tox-py38-sentinel https://zuul.opendev.org/t/openstack/build/67d8421268bb4e4fa4c599773c7d58e4 : SUCCESS in 4m 18s\n- tooz-tox-py310-sentinel https://zuul.opendev.org/t/openstack/build/40ef1d87001248c1a2431b5354a2146d : SUCCESS in 4m 43s\n- tooz-tox-py38-memcached https://zuul.opendev.org/t/openstack/build/9596a1db46a64b8782adec4aaae32810 : SUCCESS in 3m 46s\n- tooz-tox-py310-memcached https://zuul.opendev.org/t/openstack/build/af166313bb3246569cd5b5720d91351e : SUCCESS in 4m 07s\n- tooz-tox-py38-postgresql https://zuul.opendev.org/t/openstack/build/1f780f8f0c84449a826cee3a186573db : SUCCESS in 4m 30s\n- tooz-tox-py310-postgresql https://zuul.opendev.org/t/openstack/build/da6f1a1309394c489bc66ef5fa15c616 : SUCCESS in 4m 14s\n- tooz-tox-py38-mysql https://zuul.opendev.org/t/openstack/build/77903a6a6988492e8e5f3eb7fc0eca05 : SUCCESS in 4m 35s\n- tooz-tox-py310-mysql https://zuul.opendev.org/t/openstack/build/e55570b230a74a70ad5eab9aa02073f9 : SUCCESS in 4m 06s\n- tooz-tox-py38-consul https://zuul.opendev.org/t/openstack/build/6aaaa87a6422442aae33dd652be8177e : SUCCESS in 4m 16s\n- tooz-tox-py310-consul https://zuul.opendev.org/t/openstack/build/f82bf73d9ac0498e8363a03b77ede3b9 : SUCCESS in 4m 11s","accounts_in_message":[],"_revision_number":2},{"id":"eb449376eda277c8358d8352e1a743419b535b4a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-07 19:56:35.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"0052bbc7cd51151cfdecc299f13fac0ea0e0a138","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-07 19:57:30.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e1f6586c2a894ce1a20b6edb4dbabec8\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/30c773f09d06421ca019f98ee25846a4 : SUCCESS in 38s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/07e6f43deba54014ae580d33b48efacc : SUCCESS in 35s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"29f9043b7bd494134aacf202394b0aba64740378","revisions":{"8b114aed456c1e6a49d53daf6a4866369c6a0be3":{"kind":"REWORK","_number":1,"created":"2024-03-07 10:01:42.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/82/911682/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/82/911682/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/82/911682/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/82/911682/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/82/911682/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/82/911682/1"}}},"commit":{"parents":[{"commit":"1d0ece27a4f8b0d3aed13b603fd834eba3567183","subject":"Merge \"redis: Fix wrong type used to parse socket_keepalive\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/1d0ece27a4f8b0d3aed13b603fd834eba3567183"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-03-07 10:01:08.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-03-07 10:01:08.000000000","tz":540},"subject":"Fix sentinel tests not running","message":"Fix sentinel tests not running\n\nWe have to tweak the environment set by pifpaf to use redis sentinel\nas backend.\n\nChange-Id: I2a3e0d08b6742ee7b473bc382bfc021523719251\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/8b114aed456c1e6a49d53daf6a4866369c6a0be3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/8b114aed456c1e6a49d53daf6a4866369c6a0be3"}]},"branch":"refs/heads/master"},"29f9043b7bd494134aacf202394b0aba64740378":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-03-07 10:02:17.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/82/911682/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/82/911682/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/82/911682/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/82/911682/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/82/911682/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/82/911682/2"}}},"commit":{"parents":[{"commit":"a20ca740cf5ad29c74026937d75314a8823d3361","subject":"Fix broken redis sentinel support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/a20ca740cf5ad29c74026937d75314a8823d3361"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-03-07 10:01:08.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-03-07 10:02:17.000000000","tz":0},"subject":"Fix sentinel tests not running","message":"Fix sentinel tests not running\n\nWe have to tweak the environment set by pifpaf to use redis sentinel\nas backend.\n\nChange-Id: I2a3e0d08b6742ee7b473bc382bfc021523719251\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/29f9043b7bd494134aacf202394b0aba64740378"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/29f9043b7bd494134aacf202394b0aba64740378"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
