)]}'
{"id":"openstack%2Ftooz~914635","triplet_id":"openstack%2Ftooz~master~I2319edaf17589e22567c352d3d79cb925febbd8d","project":"openstack/tooz","branch":"master","topic":"remove-distutils","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2024-04-01 23:17:46.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I2319edaf17589e22567c352d3d79cb925febbd8d","subject":"pre-commit: Bump version, add doc8","status":"MERGED","created":"2024-03-28 10:26:40.000000000","updated":"2024-04-01 23:18:42.000000000","submitted":"2024-04-01 23:17:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"914635-remove-distutils","meta_rev_id":"f15dda41a03242e6b415c2043cb33b5bb8aae588","_number":914635,"virtual_id_number":914635,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-04-01 23:17:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2024-03-29 00:31:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2024-04-01 21:50:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":1,"date":"2024-04-01 21:50:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-03-28 12:06:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-03-28 16:00:18.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"CC"},{"updated":"2024-03-29 00:31:47.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2024-04-01 21:50:22.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"9fe6fe006b34e64cb477c1a4f152d9ce270d384f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-03-28 10:26:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ac8acc2f0c414b1b0408f667099620648baf0e28","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-28 12:06:47.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e191f8c6148b414aba0bb24018d4b704\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8b3d05f9b63844a7aeaeee57961068a1 : SUCCESS in 1h 34m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/66af9a8e8a764b219538af6439f255f9 : SUCCESS in 4m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0cea0e11574e4f688f743d4ffeea2553 : SUCCESS in 4m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ce374d3bb4fe443c932d5eab3eb92934 : SUCCESS in 4m 38s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/141ce3739d464f67a32cc4145c07d078 : FAILURE in 4m 36s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/55aa3b39b205499ea0f6aca9e4a853c2 : SUCCESS in 6m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f5a8b7b022a04a459c17481528c2db20 : SUCCESS in 4m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5cfdbe4fa2814a6fb013809d1501b1a9 : SUCCESS in 3m 50s\n- tooz-tox-py38-etcd3gw https://zuul.opendev.org/t/openstack/build/88fed7a5e692438c8cdfa2d14d6962f4 : SUCCESS in 4m 31s\n- tooz-tox-py310-etcd3gw https://zuul.opendev.org/t/openstack/build/616b11c3d323465da9157fd200b84e92 : SUCCESS in 4m 53s\n- tooz-tox-py38-zookeeper https://zuul.opendev.org/t/openstack/build/a025f6a5e8624d32ac177d318c72baf1 : FAILURE in 5m 25s\n- tooz-tox-py310-zookeeper https://zuul.opendev.org/t/openstack/build/482bad9714504aafb7c7b48079dca37a : SUCCESS in 4m 30s\n- tooz-tox-py38-redis https://zuul.opendev.org/t/openstack/build/a8307f387d6f459a8a546b8d447cd828 : SUCCESS in 4m 12s\n- tooz-tox-py310-redis https://zuul.opendev.org/t/openstack/build/8e975ceb359e49d08db1c63c9abf7759 : SUCCESS in 4m 06s\n- tooz-tox-py38-sentinel https://zuul.opendev.org/t/openstack/build/a73ef3830cd44ea1b9cc2741b41aa6cf : SUCCESS in 4m 25s\n- tooz-tox-py310-sentinel https://zuul.opendev.org/t/openstack/build/b82b99e60fa34e6584c571e1edd7a6a6 : SUCCESS in 3m 41s\n- tooz-tox-py38-memcached https://zuul.opendev.org/t/openstack/build/e6373ebbb04a4b68857cccf2ac50f316 : SUCCESS in 4m 04s\n- tooz-tox-py310-memcached https://zuul.opendev.org/t/openstack/build/11244ea934b24e4cb07fba20cb4a20df : SUCCESS in 4m 40s\n- tooz-tox-py38-postgresql https://zuul.opendev.org/t/openstack/build/ff40cca96470418a9124115f3644f788 : SUCCESS in 4m 08s\n- tooz-tox-py310-postgresql https://zuul.opendev.org/t/openstack/build/6b7ad6f986af41f394f3c3cc763f1458 : SUCCESS in 3m 33s\n- tooz-tox-py38-mysql https://zuul.opendev.org/t/openstack/build/bdabf414efbb481187024f3d85063f62 : SUCCESS in 4m 27s\n- tooz-tox-py310-mysql https://zuul.opendev.org/t/openstack/build/5fcede1f3ada44089cd2041bd54a3553 : SUCCESS in 4m 10s\n- tooz-tox-py38-consul https://zuul.opendev.org/t/openstack/build/4a7fe6a908e941f6b7c22295bec79ebe : SUCCESS in 4m 41s\n- tooz-tox-py310-consul https://zuul.opendev.org/t/openstack/build/a6df569b902d477fa66cebe5eee4796e : SUCCESS in 3m 56s","accounts_in_message":[],"_revision_number":1},{"id":"835cc39dba5af8a451dfb85e4e0549f7dad9621c","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-03-28 16:00:18.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"501dbc028ef5153ceab9a1fc48af1184cb5b1803","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-28 17:22:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cc82416c8fcf446f9dd5a54e758238d2\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ff33a987f0a04d8b9735dc554196cef6 : SUCCESS in 1h 15m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e995dd44d2254930b01cef59f0c00fdf : SUCCESS in 4m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/995ef770b90f44309fb6c3efe7d43b6e : SUCCESS in 3m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a14e2f9f70c64e019ea698850dca352c : SUCCESS in 4m 54s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b8031c641db64a47a97d71c6a038746d : SUCCESS in 5m 04s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c4dd64be73e7455a9ac084096189717a : SUCCESS in 7m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5748944e916d4d9181897f683957b42b : SUCCESS in 3m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/56f720f0c7c545cc91b59b43f6a72a82 : SUCCESS in 3m 54s\n- tooz-tox-py38-etcd3gw https://zuul.opendev.org/t/openstack/build/4701739132ec42868b14ab016be5a604 : SUCCESS in 4m 32s\n- tooz-tox-py310-etcd3gw https://zuul.opendev.org/t/openstack/build/2121a59776984e399e7d405478f3f308 : SUCCESS in 4m 44s\n- tooz-tox-py38-zookeeper https://zuul.opendev.org/t/openstack/build/4f9aef543dde4350baae8f40e122ea4b : SUCCESS in 4m 33s\n- tooz-tox-py310-zookeeper https://zuul.opendev.org/t/openstack/build/93f76863334b4184b2d5a9c795647c10 : SUCCESS in 4m 23s\n- tooz-tox-py38-redis https://zuul.opendev.org/t/openstack/build/8a5fad9e18464afc94162061a0369c28 : SUCCESS in 4m 27s\n- tooz-tox-py310-redis https://zuul.opendev.org/t/openstack/build/ef74b5108cd64352b06d8a33eb52bab9 : SUCCESS in 4m 09s\n- tooz-tox-py38-sentinel https://zuul.opendev.org/t/openstack/build/7ad8afe70f3746949108cc260befecda : SUCCESS in 4m 20s\n- tooz-tox-py310-sentinel https://zuul.opendev.org/t/openstack/build/bf2d2f3c35544d3195a0ef84801f572c : SUCCESS in 4m 13s\n- tooz-tox-py38-memcached https://zuul.opendev.org/t/openstack/build/74d8b48e4fcf4f7a9912886aadf0dacb : SUCCESS in 4m 28s\n- tooz-tox-py310-memcached https://zuul.opendev.org/t/openstack/build/cb577fc334124592a3fca3d6792e116d : SUCCESS in 4m 35s\n- tooz-tox-py38-postgresql https://zuul.opendev.org/t/openstack/build/500e2991066847a791e0bdf600ed07a1 : SUCCESS in 4m 50s\n- tooz-tox-py310-postgresql https://zuul.opendev.org/t/openstack/build/d94282e6ac2e4178aee263a6ba5e070b : SUCCESS in 4m 27s\n- tooz-tox-py38-mysql https://zuul.opendev.org/t/openstack/build/6a9792edcfcf48d6925627a8f52d90a3 : SUCCESS in 4m 14s\n- tooz-tox-py310-mysql https://zuul.opendev.org/t/openstack/build/38d544021fe84482bb7eafed27e0373b : SUCCESS in 4m 30s\n- tooz-tox-py38-consul https://zuul.opendev.org/t/openstack/build/fe8fe0303d6d4fac90aaa9f89541cdea : SUCCESS in 4m 08s\n- tooz-tox-py310-consul https://zuul.opendev.org/t/openstack/build/fc94b432f56d4249b2ea0f4bec492e30 : SUCCESS in 4m 23s","accounts_in_message":[],"_revision_number":1},{"id":"5720a733ed10a40b1db97eeee0941a8513e8925c","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-03-29 00:31:47.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e52218c9f9186e210b75ee32dbe9221a2472d1a7","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2024-04-01 21:50:22.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c4a52c4f68a249bcd6610354addefe07d72956d7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-01 21:50:58.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"cc465c510efb453fe534f037eaefa2aac79cb7e3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-01 23:17:46.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/67bc2b2c339a44299544de502abc87d4\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/428ee35d1b15449b9d13daca12a716dd : SUCCESS in 1h 17m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/04afe3eff9d049839785273b31835148 : SUCCESS in 3m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/203b37c3049341cbb488723f2361e710 : SUCCESS in 4m 02s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2ca0a940850c43aca925ff975988bcad : SUCCESS in 3m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/26e76dda3691497c8232825699c855ab : SUCCESS in 3m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b052e9e6a80c45dbacdf92a685fffde3 : SUCCESS in 4m 12s\n- tooz-tox-py38-etcd3gw https://zuul.opendev.org/t/openstack/build/2bb99bc73c4c43ce8e26ab2b5985eb94 : SUCCESS in 4m 33s\n- tooz-tox-py310-etcd3gw https://zuul.opendev.org/t/openstack/build/7ea13fe9331042aa83aefc86d870238d : SUCCESS in 5m 44s\n- tooz-tox-py38-zookeeper https://zuul.opendev.org/t/openstack/build/aa182240c58346aabec2c1e5850f996d : SUCCESS in 4m 50s\n- tooz-tox-py310-zookeeper https://zuul.opendev.org/t/openstack/build/01eb8d8fa69245f591d8f51c7d049f55 : SUCCESS in 3m 48s\n- tooz-tox-py38-redis https://zuul.opendev.org/t/openstack/build/b3ba54562f3e43839573b14c4e068335 : SUCCESS in 7m 41s\n- tooz-tox-py310-redis https://zuul.opendev.org/t/openstack/build/ab0109f57cbe40f2a16359976bc079c7 : SUCCESS in 5m 04s\n- tooz-tox-py38-sentinel https://zuul.opendev.org/t/openstack/build/0160891fd83a4a7f94b112382c375068 : SUCCESS in 4m 05s\n- tooz-tox-py310-sentinel https://zuul.opendev.org/t/openstack/build/f335da73cd4e4805a49e43c082291dd0 : SUCCESS in 4m 07s\n- tooz-tox-py38-memcached https://zuul.opendev.org/t/openstack/build/8f5317c70baf4d69a8cf8733b86eb1cc : SUCCESS in 3m 48s\n- tooz-tox-py310-memcached https://zuul.opendev.org/t/openstack/build/5f37d2e2fb1c487ba3654476fe79a170 : SUCCESS in 4m 14s\n- tooz-tox-py38-postgresql https://zuul.opendev.org/t/openstack/build/9c2515eed6bf4e8a898bd89bc29ca17b : SUCCESS in 3m 58s\n- tooz-tox-py310-postgresql https://zuul.opendev.org/t/openstack/build/fecf7e3ec5954e44a47c58bd18b5f126 : SUCCESS in 5m 07s\n- tooz-tox-py38-mysql https://zuul.opendev.org/t/openstack/build/f294fdee7fd34da29e4b3181ee398b1a : SUCCESS in 3m 44s\n- tooz-tox-py310-mysql https://zuul.opendev.org/t/openstack/build/2179aa26833c451891e27161712f9c08 : SUCCESS in 4m 01s\n- tooz-tox-py38-consul https://zuul.opendev.org/t/openstack/build/8052a57b897d4c7cb9bf6ba711e67dbc : SUCCESS in 4m 04s\n- tooz-tox-py310-consul https://zuul.opendev.org/t/openstack/build/2774a69bb23c4135a18e03f42753b3e0 : SUCCESS in 4m 06s","accounts_in_message":[],"_revision_number":1},{"id":"bf37d8e903e77525d5a85b22e5f5f3e029515546","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-01 23:17:46.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"f15dda41a03242e6b415c2043cb33b5bb8aae588","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-01 23:18:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4963ec517e894ec0bd95ccf21c0ef362\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/18dd2c54a0294785913ae1ee98c48c60 : SUCCESS in 37s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6203a699f7b345f0b3a1fa2bb4320a1f : SUCCESS in 36s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0e2010d93055d94834cc5faf2733941e49fd6a1f","revisions":{"0e2010d93055d94834cc5faf2733941e49fd6a1f":{"kind":"REWORK","_number":1,"created":"2024-03-28 10:26:40.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/35/914635/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/35/914635/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/35/914635/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/35/914635/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/35/914635/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/35/914635/1"}}},"commit":{"parents":[{"commit":"aeb1f4c19f8c38b06fc0ed14e5ad3ad492d39662","subject":"Remove unnecessary dependencies","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/aeb1f4c19f8c38b06fc0ed14e5ad3ad492d39662"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-03-28 10:21:40.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-03-28 10:21:40.000000000","tz":0},"subject":"pre-commit: Bump version, add doc8","message":"pre-commit: Bump version, add doc8\n\nChange-Id: I2319edaf17589e22567c352d3d79cb925febbd8d\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/0e2010d93055d94834cc5faf2733941e49fd6a1f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/0e2010d93055d94834cc5faf2733941e49fd6a1f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
