)]}'
{"id":"openstack%2Ftooz~922595","triplet_id":"openstack%2Ftooz~master~Iab4a5b050f9960b1e125aa9dd657cc690f06d32f","project":"openstack/tooz","branch":"master","topic":"remove-TestSkipped","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2024-06-27 19:28:10.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iab4a5b050f9960b1e125aa9dd657cc690f06d32f","subject":"Replace use of testtools.testcase.TestSkipped","status":"MERGED","created":"2024-06-24 09:52:33.000000000","updated":"2024-06-27 19:29:06.000000000","submitted":"2024-06-27 19:28:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"922595-remove-TestSkipped","meta_rev_id":"e63c3cf8319a2da7d2fc5b4f6ca2227a3894142c","_number":922595,"virtual_id_number":922595,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-06-27 19:28:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2024-06-24 10:01:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2024-06-26 07:51:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":1,"date":"2024-06-26 07:51:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-06-24 10:01:54.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2024-06-24 10:40:37.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2024-06-24 12:03:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-06-26 07:51:03.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"3cb2a5edcd3de0ac7c32cfbcafdca811c00d1b09","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-06-24 09:52:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"47460502c550a6d5027e96ab23b88e0c4e914f4c","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-06-24 10:01:54.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"fd55fe7958eb6af572eb518529d4db6c338d81ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-24 12:03:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/49f88454fd2a4969aaf3464089d922a6\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c847b4881420453186f3784142234b54 : SUCCESS in 1h 51m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d05a6d51fb184b53b2a331ad3103d353 : SUCCESS in 5m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/04aa19e66f4c4677908bfe2b66a72deb : SUCCESS in 3m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e56cd0188a7a48508f3853d9f76124df : SUCCESS in 4m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c31e46d7a84646078c0c44a57af5a268 : SUCCESS in 4m 50s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e39665d1a6614580820bfb064977cafa : FAILURE in 8m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eadba1a5d7924c66a9b25ece208e076d : SUCCESS in 4m 06s\n- tooz-tox-py38-etcd3gw https://zuul.opendev.org/t/openstack/build/9a2bc178ac6f49fc9b47f6f65b9726a0 : SUCCESS in 4m 31s\n- tooz-tox-py310-etcd3gw https://zuul.opendev.org/t/openstack/build/3ec7caab543545529d9e35f7fb09d4be : SUCCESS in 4m 19s\n- tooz-tox-py38-zookeeper https://zuul.opendev.org/t/openstack/build/c6dfd5df1dd546e0bd9466f62a994d53 : SUCCESS in 3m 49s\n- tooz-tox-py310-zookeeper https://zuul.opendev.org/t/openstack/build/51079738df3a46828cd88cd96b533a28 : SUCCESS in 4m 26s\n- tooz-tox-py38-redis https://zuul.opendev.org/t/openstack/build/3a1a4db44aa54b358d02beb9ede5561f : SUCCESS in 3m 59s\n- tooz-tox-py310-redis https://zuul.opendev.org/t/openstack/build/a05a3c97b91640fab60c36e6485b7a08 : SUCCESS in 3m 36s\n- tooz-tox-py38-sentinel https://zuul.opendev.org/t/openstack/build/7bc01e86db474dc39c165cea37fea15c : SUCCESS in 4m 09s\n- tooz-tox-py310-sentinel https://zuul.opendev.org/t/openstack/build/1410fb70ddb1431587994c88d37dc841 : SUCCESS in 5m 45s\n- tooz-tox-py38-memcached https://zuul.opendev.org/t/openstack/build/993826d4e2d74976820541ce55d1df21 : SUCCESS in 4m 21s\n- tooz-tox-py310-memcached https://zuul.opendev.org/t/openstack/build/c906778694c3431aa5b9e016c836496d : SUCCESS in 5m 33s\n- tooz-tox-py38-postgresql https://zuul.opendev.org/t/openstack/build/b8ddb8af1ba04210b9686eacb67134fd : SUCCESS in 5m 46s\n- tooz-tox-py310-postgresql https://zuul.opendev.org/t/openstack/build/bf4383e07aef46bc9c9bfabe4cbb6fe6 : SUCCESS in 5m 28s\n- tooz-tox-py38-mysql https://zuul.opendev.org/t/openstack/build/8d51d5b48316407889cb5cdbee08c143 : SUCCESS in 3m 54s\n- tooz-tox-py310-mysql https://zuul.opendev.org/t/openstack/build/57a38fafc3064ceea520e2d833795765 : SUCCESS in 3m 59s\n- tooz-tox-py38-consul https://zuul.opendev.org/t/openstack/build/4fea42284b394a97ba8a1176229c3474 : SUCCESS in 6m 02s\n- tooz-tox-py310-consul https://zuul.opendev.org/t/openstack/build/c01e06244afb4121bdd240469961bc3a : SUCCESS in 5m 43s","accounts_in_message":[],"_revision_number":1},{"id":"84df082e3c4de77bb6a56143d8802b263821517b","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2024-06-26 07:51:03.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"186185f20bc70c9513c8ef44791d9f99e3ceca5c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-26 07:51:34.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c78691c009e0792fa3aec7869e92a2477bbf0266","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-26 09:28:32.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/396c222c677a4cf0bef8f5961dd5d036\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/709e9499d05747239841f58db4cb9c9f : SUCCESS in 1h 34m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cd67ed39e0a0475b8bea28492b6856bf : SUCCESS in 4m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/13761b8409754bbf806a047ffc6644a6 : SUCCESS in 4m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f03e0328b5fc4b7ca89b16c93d9bd2a1 : SUCCESS in 5m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e69ae629a1c417a9a88b909a3ee76b3 : SUCCESS in 4m 01s\n- tooz-tox-py38-etcd3gw https://zuul.opendev.org/t/openstack/build/38828261796a4f2686b8a10dbf2723a7 : SUCCESS in 4m 26s\n- tooz-tox-py310-etcd3gw https://zuul.opendev.org/t/openstack/build/e3630ba9086040798a9af47385d82fbc : SUCCESS in 4m 27s\n- tooz-tox-py38-zookeeper https://zuul.opendev.org/t/openstack/build/14966a8e5fbe4ac68b10c507bcd3d4ae : SUCCESS in 3m 49s\n- tooz-tox-py310-zookeeper https://zuul.opendev.org/t/openstack/build/ca305deadaf54969ba79e39a47bda91c : FAILURE in 4m 52s\n- tooz-tox-py38-redis https://zuul.opendev.org/t/openstack/build/af6a76425b044f2f971341e8eb344b3d : SUCCESS in 3m 37s\n- tooz-tox-py310-redis https://zuul.opendev.org/t/openstack/build/4ff44baffb8d46b898496f24d94e6fdf : SUCCESS in 3m 37s\n- tooz-tox-py38-sentinel https://zuul.opendev.org/t/openstack/build/b2f09a92ccf74595a101f58c14f310ee : SUCCESS in 3m 34s\n- tooz-tox-py310-sentinel https://zuul.opendev.org/t/openstack/build/e03e736433d9449f92f59d9a532869c7 : SUCCESS in 4m 38s\n- tooz-tox-py38-memcached https://zuul.opendev.org/t/openstack/build/d15d09c38eb34155919c6782e3fdd8e6 : SUCCESS in 4m 29s\n- tooz-tox-py310-memcached https://zuul.opendev.org/t/openstack/build/c5f798435408460ebea57c08c0e6028e : SUCCESS in 3m 48s\n- tooz-tox-py38-postgresql https://zuul.opendev.org/t/openstack/build/8189737f9e9b49cda1aee0148dfe53bd : SUCCESS in 3m 52s\n- tooz-tox-py310-postgresql https://zuul.opendev.org/t/openstack/build/5b15d1f38dd046899405ca52da51d890 : SUCCESS in 5m 24s\n- tooz-tox-py38-mysql https://zuul.opendev.org/t/openstack/build/cb8b42b2d2634e13b78bc900c5de6d48 : SUCCESS in 5m 27s\n- tooz-tox-py310-mysql https://zuul.opendev.org/t/openstack/build/828b2a80c8d844269b4679d68e42866d : SUCCESS in 5m 05s\n- tooz-tox-py38-consul https://zuul.opendev.org/t/openstack/build/1b8d39407ca748e58cab1d23bb7664fd : SUCCESS in 3m 37s\n- tooz-tox-py310-consul https://zuul.opendev.org/t/openstack/build/304b9d2eeff74485b6988cb8dc9a34b0 : SUCCESS in 5m 37s","accounts_in_message":[],"_revision_number":1},{"id":"a1ffc75a5262581d784512df166e61154ef1173b","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-06-26 16:29:04.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b6a48815bc8479ac18580cc6a8c52573603d0546","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-26 18:39:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/100de975d33d45828885d28eb1128518\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4747c9bdc9264c5e9d6b397dfa9d217f : SUCCESS in 1h 51m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ec2409d8f4fa4d0e9b54b68ef7470000 : SUCCESS in 7m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e54e3b897fa449aa932e8030152e0087 : SUCCESS in 4m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c900c9c0ac2742aaa0cd17032d90270a : SUCCESS in 4m 20s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/467a5d1f781748228388d5dea5b0adcf : SUCCESS in 9m 06s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b2d921c418b744ffbbeb5597cb893bfe : FAILURE in 7m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5547037827a04279b52f6e793acef81a : SUCCESS in 5m 10s\n- tooz-tox-py38-etcd3gw https://zuul.opendev.org/t/openstack/build/6ef895c7b31b4ba686d7e693532a65d8 : SUCCESS in 4m 25s\n- tooz-tox-py310-etcd3gw https://zuul.opendev.org/t/openstack/build/ac965a0f5a9d4954a6e1a087e5b6263c : SUCCESS in 4m 35s\n- tooz-tox-py38-zookeeper https://zuul.opendev.org/t/openstack/build/ccef3c1314c44abd93526447f3f6af00 : SUCCESS in 4m 33s\n- tooz-tox-py310-zookeeper https://zuul.opendev.org/t/openstack/build/60d08b13103f4650a2c5d6cd635a1234 : SUCCESS in 4m 19s\n- tooz-tox-py38-redis https://zuul.opendev.org/t/openstack/build/a68b4274d6864f98b4e2fd3797c25a3c : SUCCESS in 7m 39s\n- tooz-tox-py310-redis https://zuul.opendev.org/t/openstack/build/11daf7b6aa324768a64dfacd0ff96187 : SUCCESS in 4m 19s\n- tooz-tox-py38-sentinel https://zuul.opendev.org/t/openstack/build/b97449776c3241c8a19e82fe751a2600 : SUCCESS in 4m 08s\n- tooz-tox-py310-sentinel https://zuul.opendev.org/t/openstack/build/a99629621ad54eaf8223e09b5c3fc478 : SUCCESS in 4m 11s\n- tooz-tox-py38-memcached https://zuul.opendev.org/t/openstack/build/8b7d363eb94e4e9d8ff8bce827cfd4b4 : SUCCESS in 7m 15s\n- tooz-tox-py310-memcached https://zuul.opendev.org/t/openstack/build/d88242824a614d9896c8f7a931eaed32 : SUCCESS in 4m 16s\n- tooz-tox-py38-postgresql https://zuul.opendev.org/t/openstack/build/6d9927b35c11429f80cb4bd941b07b47 : SUCCESS in 4m 06s\n- tooz-tox-py310-postgresql https://zuul.opendev.org/t/openstack/build/3294f2f306fc45c39e3bee04a4521231 : SUCCESS in 4m 14s\n- tooz-tox-py38-mysql https://zuul.opendev.org/t/openstack/build/49267bc9ca314ac8b9132f3023f4cd1f : SUCCESS in 4m 23s\n- tooz-tox-py310-mysql https://zuul.opendev.org/t/openstack/build/2faa7e3475c24159b0cd283bc2513975 : SUCCESS in 7m 15s\n- tooz-tox-py38-consul https://zuul.opendev.org/t/openstack/build/940b335740e440099b9335d4fcd2c0d9 : SUCCESS in 4m 20s\n- tooz-tox-py310-consul https://zuul.opendev.org/t/openstack/build/f4d7223f01c6475796d1d4c3f87cefde : SUCCESS in 4m 23s","accounts_in_message":[],"_revision_number":1},{"id":"c5051425cb00121aad89c016f3dd632fb4163d8c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-26 18:40:08.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e0a6b969a09a642fc166d78ff268dd6c7d9adf06","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-26 20:13:10.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0d57dd968f684111a7d89c4b5cc883c2\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/57b84a58f1654f8f8d7dd773d177d5e0 : SUCCESS in 1h 31m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ac80157390d043ada3697fc91ca3cd82 : SUCCESS in 5m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/84d216ca0ba745e59cf034fcb1084ac8 : SUCCESS in 3m 52s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b428d3c73d1141dfaac7bf6dd37337bb : SUCCESS in 9m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a579fa305a28459e99364a7d12577c82 : SUCCESS in 4m 41s\n- tooz-tox-py38-etcd3gw https://zuul.opendev.org/t/openstack/build/d03fcdc7c2504abbb0e841a4c266fd2c : SUCCESS in 4m 31s\n- tooz-tox-py310-etcd3gw https://zuul.opendev.org/t/openstack/build/e9794abce3d74f45aa6698356679743a : SUCCESS in 4m 08s\n- tooz-tox-py38-zookeeper https://zuul.opendev.org/t/openstack/build/eb0ea167779440aa9e624cee5bf0222c : FAILURE in 5m 35s\n- tooz-tox-py310-zookeeper https://zuul.opendev.org/t/openstack/build/db4d2fd3c6c94da1b47d0900e4f59c41 : SUCCESS in 6m 02s\n- tooz-tox-py38-redis https://zuul.opendev.org/t/openstack/build/23f93e38f3fc4cd19cd1335d09fc1e33 : SUCCESS in 3m 56s\n- tooz-tox-py310-redis https://zuul.opendev.org/t/openstack/build/641a4734b1da4e0fa6633d86bf5dfbc8 : SUCCESS in 4m 04s\n- tooz-tox-py38-sentinel https://zuul.opendev.org/t/openstack/build/93d3a05360024d85bb3fdacd80f3806b : SUCCESS in 4m 16s\n- tooz-tox-py310-sentinel https://zuul.opendev.org/t/openstack/build/810b585741be49b9bf0d29800deb01b4 : SUCCESS in 3m 53s\n- tooz-tox-py38-memcached https://zuul.opendev.org/t/openstack/build/540b8eb545144137a536a2c61a3cbf2e : SUCCESS in 4m 39s\n- tooz-tox-py310-memcached https://zuul.opendev.org/t/openstack/build/86174a16e10c47539796fd2d4c9dfab3 : SUCCESS in 5m 28s\n- tooz-tox-py38-postgresql https://zuul.opendev.org/t/openstack/build/17c06186f885466eaac00ce0a2ee1799 : SUCCESS in 3m 42s\n- tooz-tox-py310-postgresql https://zuul.opendev.org/t/openstack/build/2194aceeecf64bbca58565553a65c355 : SUCCESS in 3m 51s\n- tooz-tox-py38-mysql https://zuul.opendev.org/t/openstack/build/f365e994e1c04153b8d0688f4cffdcca : SUCCESS in 4m 32s\n- tooz-tox-py310-mysql https://zuul.opendev.org/t/openstack/build/fba3a84afb574f02a61a10551ce513b4 : SUCCESS in 3m 59s\n- tooz-tox-py38-consul https://zuul.opendev.org/t/openstack/build/d5e431179dbc4faeb06709260e56374e : SUCCESS in 4m 03s\n- tooz-tox-py310-consul https://zuul.opendev.org/t/openstack/build/d8fc96a827bd4659b6a5029a3ceafcb8 : SUCCESS in 3m 44s","accounts_in_message":[],"_revision_number":1},{"id":"f78eedd6e4b10b4207038628b6e69b91ab71519d","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-06-27 15:49:18.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"06c30081f4e1adead1ea8f8a539841108a778f67","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-27 17:38:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/795f280a86014fd1998345f94a409ecb\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5df205aef5c242ee9a9dd6c0f5cbca82 : SUCCESS in 1h 42m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ac84de17375d4d96abab13c8717aeab9 : SUCCESS in 6m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dbfc753a097145b9a05e3df90b36035a : SUCCESS in 5m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bcd12af5e49d4b8b95a940d68dce9502 : SUCCESS in 4m 20s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9259fa2959494e5bbd46cd2179e32b27 : SUCCESS in 5m 04s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ce7df2e81263496dbf61402198b30dc6 : FAILURE in 10m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1dd70a47a5594de9a5a553f4d1a6e8a8 : SUCCESS in 4m 04s\n- tooz-tox-py38-etcd3gw https://zuul.opendev.org/t/openstack/build/fc1d976580c34fe49a400f9c2308b64c : SUCCESS in 4m 34s\n- tooz-tox-py310-etcd3gw https://zuul.opendev.org/t/openstack/build/b1e6133099624b418f55a89d72c6d334 : SUCCESS in 4m 13s\n- tooz-tox-py38-zookeeper https://zuul.opendev.org/t/openstack/build/2075ca3385b04e93b98c3b4f251d125e : SUCCESS in 4m 26s\n- tooz-tox-py310-zookeeper https://zuul.opendev.org/t/openstack/build/c40ca212ed6440a9b9c76a117e647626 : SUCCESS in 4m 33s\n- tooz-tox-py38-redis https://zuul.opendev.org/t/openstack/build/234ff5fbb98848fa8832112d1ca39c84 : SUCCESS in 4m 30s\n- tooz-tox-py310-redis https://zuul.opendev.org/t/openstack/build/7ff4ddd837c349989f3e4b99677f7c5e : SUCCESS in 3m 41s\n- tooz-tox-py38-sentinel https://zuul.opendev.org/t/openstack/build/241c175824f34137a424c5aabde5de34 : SUCCESS in 4m 11s\n- tooz-tox-py310-sentinel https://zuul.opendev.org/t/openstack/build/a377bc5830354ad98a285c2cb492e51f : SUCCESS in 4m 30s\n- tooz-tox-py38-memcached https://zuul.opendev.org/t/openstack/build/cdcc2bdf83b644638b4cb9a4d4fc79d9 : SUCCESS in 4m 23s\n- tooz-tox-py310-memcached https://zuul.opendev.org/t/openstack/build/f8c08581c08b4366b34ba89ba093fd4d : SUCCESS in 3m 56s\n- tooz-tox-py38-postgresql https://zuul.opendev.org/t/openstack/build/9f3e9bdcd45646a1a6ac151c5bffa006 : SUCCESS in 3m 40s\n- tooz-tox-py310-postgresql https://zuul.opendev.org/t/openstack/build/b1b0f646fefd41ff848df49a6c1fab35 : SUCCESS in 4m 50s\n- tooz-tox-py38-mysql https://zuul.opendev.org/t/openstack/build/c570b2b1b2cf4a0bb2fb19dc48e7d3ce : SUCCESS in 6m 13s\n- tooz-tox-py310-mysql https://zuul.opendev.org/t/openstack/build/5a777bd2576641a2be1cbe4d91b1b049 : SUCCESS in 3m 45s\n- tooz-tox-py38-consul https://zuul.opendev.org/t/openstack/build/3a1810adc6f348ab8fecca8c969c1357 : SUCCESS in 3m 37s\n- tooz-tox-py310-consul https://zuul.opendev.org/t/openstack/build/65dee5dcea004e16a9c23ea9c98f0005 : SUCCESS in 5m 07s","accounts_in_message":[],"_revision_number":1},{"id":"3f60ebc76be481270fcf2c055c20d99fbe115f30","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-27 17:39:21.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2a67dccf2377575f99908fbef9e0e0a54f5836ad","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-27 19:28:10.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1aacd2205bfd4ec9ab8b3d64b8388fa3\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d412c257989b4bc4916a7aa7904ee95b : SUCCESS in 1h 43m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/afd1371537e04dbdb3de59a6935ad63d : SUCCESS in 3m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/046d81d223544f8dbcae1ea6c229b750 : SUCCESS in 4m 14s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/16fded021a1f4d4ea6cf7e54b8f0ec6d : SUCCESS in 4m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57e9d5ac79a64d9f85c45a38940e3f4d : SUCCESS in 3m 49s\n- tooz-tox-py38-etcd3gw https://zuul.opendev.org/t/openstack/build/3987ebe09210446c800bd65b4673c9ee : SUCCESS in 4m 20s\n- tooz-tox-py310-etcd3gw https://zuul.opendev.org/t/openstack/build/93c3efbe403c4ff8b98e49abfeacbc0a : SUCCESS in 4m 37s\n- tooz-tox-py38-zookeeper https://zuul.opendev.org/t/openstack/build/441d2bc4087c42ddb4987ff9f430da99 : SUCCESS in 5m 28s\n- tooz-tox-py310-zookeeper https://zuul.opendev.org/t/openstack/build/9bd33013b2a24e27a11710007913c55d : SUCCESS in 4m 09s\n- tooz-tox-py38-redis https://zuul.opendev.org/t/openstack/build/9ac1b2c5c51e466d8eed205345aa5e88 : SUCCESS in 5m 20s\n- tooz-tox-py310-redis https://zuul.opendev.org/t/openstack/build/c8d0e89761ce4b23b6dd708def15bcb3 : SUCCESS in 5m 52s\n- tooz-tox-py38-sentinel https://zuul.opendev.org/t/openstack/build/4d3498ede4bf4ff289e5e2e76e133c27 : SUCCESS in 4m 02s\n- tooz-tox-py310-sentinel https://zuul.opendev.org/t/openstack/build/5e58a1fa9977433690da6f2294a7f1f5 : SUCCESS in 5m 55s\n- tooz-tox-py38-memcached https://zuul.opendev.org/t/openstack/build/e47ba01f8ba147faa02738bbff6b6a52 : SUCCESS in 5m 08s\n- tooz-tox-py310-memcached https://zuul.opendev.org/t/openstack/build/80b1814cd73641fd8dd6663adb8358bc : SUCCESS in 5m 37s\n- tooz-tox-py38-postgresql https://zuul.opendev.org/t/openstack/build/5e65edbd38e143668d416f5a501f9660 : SUCCESS in 6m 00s\n- tooz-tox-py310-postgresql https://zuul.opendev.org/t/openstack/build/565cba2d4eff4ba28df16af99719b02c : SUCCESS in 4m 02s\n- tooz-tox-py38-mysql https://zuul.opendev.org/t/openstack/build/280e61640062450786cc550501bf7733 : SUCCESS in 3m 54s\n- tooz-tox-py310-mysql https://zuul.opendev.org/t/openstack/build/deeeffee69694018b911b68111d1fb86 : SUCCESS in 3m 50s\n- tooz-tox-py38-consul https://zuul.opendev.org/t/openstack/build/2419e42cdec8448689050787c4bbb6df : SUCCESS in 4m 15s\n- tooz-tox-py310-consul https://zuul.opendev.org/t/openstack/build/1bbd26680fa64682abcbad796d8e8866 : SUCCESS in 4m 03s","accounts_in_message":[],"_revision_number":1},{"id":"8d460fe6e0ebf1a5fcf46e98e8b2f2911a2d4fa7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-27 19:28:10.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"e63c3cf8319a2da7d2fc5b4f6ca2227a3894142c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-27 19:29:06.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c1e6692a3edc49a183d22aa305780b91\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/529a7f1fee12493b8279261354eb6e13 : SUCCESS in 40s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"af0a60b256024bdb9817778493c5e30e1b119f42","revisions":{"af0a60b256024bdb9817778493c5e30e1b119f42":{"kind":"REWORK","_number":1,"created":"2024-06-24 09:52:33.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/95/922595/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/95/922595/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/95/922595/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/95/922595/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/95/922595/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/95/922595/1"}}},"commit":{"parents":[{"commit":"c65282fcb0737dd3250e09a8271e16f0d0a10387","subject":"Merge \"Loose lower bound of packaging library version\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/c65282fcb0737dd3250e09a8271e16f0d0a10387"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-06-24 09:52:16.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-06-24 09:52:16.000000000","tz":60},"subject":"Replace use of testtools.testcase.TestSkipped","message":"Replace use of testtools.testcase.TestSkipped\n\nThis has been deprecated for removal [1]. Use the stdlib variant\ninstead.\n\n[1] https://github.com/testing-cabal/testtools/commit/59b890db3c\n\nChange-Id: Iab4a5b050f9960b1e125aa9dd657cc690f06d32f\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/af0a60b256024bdb9817778493c5e30e1b119f42"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/af0a60b256024bdb9817778493c5e30e1b119f42"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
