)]}'
{"id":"openstack%2Ftooz~930983","triplet_id":"openstack%2Ftooz~master~I90d65f3e5b1be6b32e8c44a018c85c01ca47d3b2","project":"openstack/tooz","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-10-05 03:54:11.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"I90d65f3e5b1be6b32e8c44a018c85c01ca47d3b2","subject":"Use oslo.utils implementation to parse sentinel address","status":"MERGED","created":"2024-10-01 07:59:10.000000000","updated":"2024-10-05 05:04:33.000000000","submitted":"2024-10-05 05:03:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"930983","meta_rev_id":"a93583708488084b4e5e8c6bb384624e6ad18d33","_number":930983,"virtual_id_number":930983,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-10-05 05:03:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"recommended":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"all":[{"value":1,"date":"2024-10-02 17:43:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2024-10-04 11:40:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":1,"date":"2024-10-05 03:54:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-01 10:07:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-10-02 17:43:20.000000000","updated_by":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"reviewer":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"state":"REVIEWER"},{"updated":"2024-10-04 11:40:56.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"bf5963017091f968e663c04dc200ed960d2e271c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-01 07:59:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"135d085793c1397f2df646e42ac05a7b12ba6b92","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-01 08:00:20.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"40c2c98a62938e11e30a1af534b7ad48cdb7a129","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-01 08:22:30.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"895bb6670d877f38acb8f2f0f13aef48828e6466","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-01 08:25:05.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"4966f04b7f52fec73de9d102ae80f4c2340b2ab1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-01 10:07:58.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ffc00cbea98748c3aca0efa6119022cd\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7fcbcf01fca24e4faae1448e88bd97c2 : SUCCESS in 1h 36m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8330fa3627a24eb9bee466c6a5c54d88 : SUCCESS in 3m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e50484b302134bd0b956f05f8fc0eff6 : SUCCESS in 4m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0e868b1c6cd3419ca13f56c2180cf61a : SUCCESS in 5m 19s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/cbf6b4b8c60a4d48905502b77b9d83be : FAILURE in 10m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff3e748cad224db2b5bbc0ea32d61368 : SUCCESS in 4m 40s\n- tooz-tox-py38-etcd3gw https://zuul.opendev.org/t/openstack/build/6a460e6c27b34ff4919713dec5d75c05 : SUCCESS in 4m 47s\n- tooz-tox-py310-etcd3gw https://zuul.opendev.org/t/openstack/build/83af4cbfab3149c29d6eef5124b5ab90 : SUCCESS in 4m 47s\n- tooz-tox-py38-zookeeper https://zuul.opendev.org/t/openstack/build/76c068be8f904f75bbc3549046259b98 : SUCCESS in 4m 19s\n- tooz-tox-py310-zookeeper https://zuul.opendev.org/t/openstack/build/3baa6a5bd29741228fbf22597d06efe4 : SUCCESS in 4m 07s\n- tooz-tox-py38-redis https://zuul.opendev.org/t/openstack/build/74949c73af05454eb2242483a328dac0 : SUCCESS in 4m 27s\n- tooz-tox-py310-redis https://zuul.opendev.org/t/openstack/build/9a9586a7febe4acb9be8fe9ce609c3d8 : SUCCESS in 3m 02s\n- tooz-tox-py38-sentinel https://zuul.opendev.org/t/openstack/build/d1395e02c0244607bd0949033e9fa7b8 : SUCCESS in 3m 49s\n- tooz-tox-py310-sentinel https://zuul.opendev.org/t/openstack/build/70e2b4e920414bef8a5e11b6b1081ffe : SUCCESS in 4m 43s\n- tooz-tox-py38-memcached https://zuul.opendev.org/t/openstack/build/07c04a650763439fa1f0226990865ea0 : SUCCESS in 4m 08s\n- tooz-tox-py310-memcached https://zuul.opendev.org/t/openstack/build/b704d77aeefa4dc8a74733275b768dbd : SUCCESS in 3m 04s\n- tooz-tox-py38-postgresql https://zuul.opendev.org/t/openstack/build/4ba5a94d3a4d4f9bb70af9bcc72ac6d3 : SUCCESS in 4m 12s\n- tooz-tox-py310-postgresql https://zuul.opendev.org/t/openstack/build/e380962875594c0c8718ed6c4a7650c6 : SUCCESS in 4m 01s\n- tooz-tox-py38-mysql https://zuul.opendev.org/t/openstack/build/ccec6453bf524c3d904f19085be91940 : SUCCESS in 3m 56s\n- tooz-tox-py310-mysql https://zuul.opendev.org/t/openstack/build/37b4db42052b434aa54a428706b12df4 : SUCCESS in 4m 13s\n- tooz-tox-py38-consul https://zuul.opendev.org/t/openstack/build/5c05c01523e34f899be03271513abcc3 : SUCCESS in 3m 04s\n- tooz-tox-py310-consul https://zuul.opendev.org/t/openstack/build/56a36623aa484d73b81a32398c392c06 : SUCCESS in 4m 31s","accounts_in_message":[],"_revision_number":4},{"id":"5c2d22bf2f0f04aa90ace17ab0648168415fa400","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-01 18:14:43.000000000","message":"Patch Set 5: Patch Set 4 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"850184488fe6b65b5b8eaa1d1f31d57db2059944","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-01 19:20:25.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4c2a0f5f671d4268bc3c885575469c2c\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2e19594addbc47bd82c1dbd2d7be7cf0 : SUCCESS in 1h 04m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9067b8b8af024695aca8e17fed35db2e : SUCCESS in 5m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f28788baa8b948079b82b83a933671cb : SUCCESS in 5m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e284d69406b2427abe1513571ebf1d6b : SUCCESS in 4m 38s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/342ac069ef0e4d5394f021f1a3b0a1df : SUCCESS in 6m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c07112e87fa4765847bfe56c2eedb3a : SUCCESS in 3m 55s\n- tooz-tox-py38-etcd3gw https://zuul.opendev.org/t/openstack/build/03d9f0256f394b399727ace218ea576a : SUCCESS in 5m 11s\n- tooz-tox-py310-etcd3gw https://zuul.opendev.org/t/openstack/build/a08a9bde52884fadb2fe8b51326875dc : SUCCESS in 4m 43s\n- tooz-tox-py38-zookeeper https://zuul.opendev.org/t/openstack/build/3bbdae98c86143de90807af902e11596 : SUCCESS in 4m 25s\n- tooz-tox-py310-zookeeper https://zuul.opendev.org/t/openstack/build/8ae96189eede4381bcb39e24e812f24b : SUCCESS in 4m 20s\n- tooz-tox-py38-redis https://zuul.opendev.org/t/openstack/build/275e7c5353e740aca0470ebe70ccedc9 : SUCCESS in 4m 37s\n- tooz-tox-py310-redis https://zuul.opendev.org/t/openstack/build/d8ed99c8035948dd93f6fdb80669e2c2 : SUCCESS in 3m 24s\n- tooz-tox-py38-sentinel https://zuul.opendev.org/t/openstack/build/6c68fff377384bf2a067b484a35e216d : SUCCESS in 4m 04s\n- tooz-tox-py310-sentinel https://zuul.opendev.org/t/openstack/build/45dd6388e9754988828c9355feaee328 : SUCCESS in 3m 58s\n- tooz-tox-py38-memcached https://zuul.opendev.org/t/openstack/build/3c9f3358dad8422c810f60c669767a79 : SUCCESS in 4m 19s\n- tooz-tox-py310-memcached https://zuul.opendev.org/t/openstack/build/5929f966f7cc4f468781b7b0b51b5373 : SUCCESS in 5m 26s\n- tooz-tox-py38-postgresql https://zuul.opendev.org/t/openstack/build/5a07c2099cf946b9a11ed642517020fc : SUCCESS in 4m 08s\n- tooz-tox-py310-postgresql https://zuul.opendev.org/t/openstack/build/3a5cad09a3084453ad4fba02f36211f6 : SUCCESS in 4m 11s\n- tooz-tox-py38-mysql https://zuul.opendev.org/t/openstack/build/e7d4d0063af2473c9c80f7ecc1311873 : SUCCESS in 4m 36s\n- tooz-tox-py310-mysql https://zuul.opendev.org/t/openstack/build/42be89d0241a4a2d9dd0cda988bb1431 : SUCCESS in 4m 23s\n- tooz-tox-py38-consul https://zuul.opendev.org/t/openstack/build/76dc4c94b9e94bcfafd50ff7df279fda : SUCCESS in 4m 09s\n- tooz-tox-py310-consul https://zuul.opendev.org/t/openstack/build/c781b141e2d24c3e80ce01c1ebfad081 : SUCCESS in 5m 30s","accounts_in_message":[],"_revision_number":5},{"id":"2edc1a6ad1a235d0a0479fe2ab1670806024660f","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2024-10-02 17:43:20.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"1cb975ea23c32f4a65428e6697bbb4c9a3dd0b15","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2024-10-04 11:40:56.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"86aa335786c1d32d194a4dacd7dc51cfcdd68083","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-05 03:54:11.000000000","message":"Patch Set 5: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"376afdf5a76b91dfa11565627fa0729248bc648c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-05 03:54:42.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"1b1390d6193e868982b310a7f0d67669c38a41b3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-05 05:03:29.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fb122e1693094bd5b748aa47f9703b70\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4becedb5ebf446fa9549511587cee8a6 : SUCCESS in 1h 08m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d5a2ead3cf4a4816adb50583ccd1cc2c : SUCCESS in 3m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f77044203b16499094b7652051569be4 : SUCCESS in 4m 11s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e2ae147312204c4383969f179f3a1bbc : SUCCESS in 4m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/171fb8e02f75462ca322b4e365d61bbd : SUCCESS in 4m 10s\n- tooz-tox-py38-etcd3gw https://zuul.opendev.org/t/openstack/build/3b90cf8c019a4b68bbff7a7218d1de69 : SUCCESS in 4m 20s\n- tooz-tox-py310-etcd3gw https://zuul.opendev.org/t/openstack/build/322b085271b34ae5bdc58f7b04dd8743 : SUCCESS in 4m 48s\n- tooz-tox-py38-zookeeper https://zuul.opendev.org/t/openstack/build/af0338ea6d014d34823b5b84d243af3c : SUCCESS in 4m 15s\n- tooz-tox-py310-zookeeper https://zuul.opendev.org/t/openstack/build/a052adc3bc56468ca8423d8a7a5f93a6 : SUCCESS in 3m 59s\n- tooz-tox-py38-redis https://zuul.opendev.org/t/openstack/build/fc576c94e797461292c6130a40253adc : SUCCESS in 6m 53s\n- tooz-tox-py310-redis https://zuul.opendev.org/t/openstack/build/8914c86ca8324a73874f16c5e1df034f : SUCCESS in 3m 48s\n- tooz-tox-py38-sentinel https://zuul.opendev.org/t/openstack/build/ea9487334a60485b82b606b163c53fd5 : SUCCESS in 4m 19s\n- tooz-tox-py310-sentinel https://zuul.opendev.org/t/openstack/build/34ef3081d26047f9ac5e270e4901ca5b : SUCCESS in 4m 20s\n- tooz-tox-py38-memcached https://zuul.opendev.org/t/openstack/build/62898513774249fbb147d79e650084b6 : SUCCESS in 5m 18s\n- tooz-tox-py310-memcached https://zuul.opendev.org/t/openstack/build/7741a756984843e7ac598a2efab5e53c : SUCCESS in 4m 04s\n- tooz-tox-py38-postgresql https://zuul.opendev.org/t/openstack/build/a2f8d7d9bbf1434f8146fc1da43d54c4 : SUCCESS in 3m 51s\n- tooz-tox-py310-postgresql https://zuul.opendev.org/t/openstack/build/e08d7b095eb04ebe9fd9aba03bd3498c : SUCCESS in 3m 44s\n- tooz-tox-py38-mysql https://zuul.opendev.org/t/openstack/build/9b30a66936014c29aec39dafbc0ed85e : SUCCESS in 5m 32s\n- tooz-tox-py310-mysql https://zuul.opendev.org/t/openstack/build/db5af3eafa0341a48bde4601c7f9a1a2 : SUCCESS in 4m 01s\n- tooz-tox-py38-consul https://zuul.opendev.org/t/openstack/build/77b409bc676d479ca93a1327bde4c86f : SUCCESS in 4m 13s\n- tooz-tox-py310-consul https://zuul.opendev.org/t/openstack/build/2763e90c197d46b99ee7a72fb8b48441 : SUCCESS in 5m 34s","accounts_in_message":[],"_revision_number":5},{"id":"a9e9647c3b166c95c18bd52f10d14fa5f5a2163c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-05 05:03:29.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"a93583708488084b4e5e8c6bb384624e6ad18d33","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-05 05:04:33.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c31cc72b9bfa456fb7c4ad86c4ca4b6d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fbca789a716c47da8841addfe032027d : SUCCESS in 45s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"cf6d4ecac1c6d2050260e5181354b4a602f9924e","revisions":{"bb408cbd41477a6e448324a418cc23091f5a1069":{"kind":"REWORK","_number":1,"created":"2024-10-01 07:59:10.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/83/930983/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/83/930983/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/83/930983/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/83/930983/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/83/930983/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/83/930983/1"}}},"commit":{"parents":[{"commit":"b12f59e2ec50e1e5e5bf15ffbdaf4815c4db3174","subject":"zookeeper: fix argparse related to ssl","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/b12f59e2ec50e1e5e5bf15ffbdaf4815c4db3174"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-01 07:58:47.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-01 07:58:47.000000000","tz":540},"subject":"Use oslo.utils implementation to parse sentinel address","message":"Use oslo.utils implementation to parse sentinel address\n\n... instead of maintaining own logic.\n\nChange-Id: I90d65f3e5b1be6b32e8c44a018c85c01ca47d3b2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/bb408cbd41477a6e448324a418cc23091f5a1069"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/bb408cbd41477a6e448324a418cc23091f5a1069"}]},"branch":"refs/heads/master"},"544a4685b244a76a0773a44bd63d8ee455048170":{"kind":"REWORK","_number":2,"created":"2024-10-01 08:00:20.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/83/930983/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/83/930983/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/83/930983/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/83/930983/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/83/930983/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/83/930983/2"}}},"commit":{"parents":[{"commit":"b12f59e2ec50e1e5e5bf15ffbdaf4815c4db3174","subject":"zookeeper: fix argparse related to ssl","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/b12f59e2ec50e1e5e5bf15ffbdaf4815c4db3174"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-01 07:58:47.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-01 08:00:12.000000000","tz":540},"subject":"Use oslo.utils implementation to parse sentinel address","message":"Use oslo.utils implementation to parse sentinel address\n\n... instead of maintaining own logic.\n\nChange-Id: I90d65f3e5b1be6b32e8c44a018c85c01ca47d3b2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/544a4685b244a76a0773a44bd63d8ee455048170"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/544a4685b244a76a0773a44bd63d8ee455048170"}]},"branch":"refs/heads/master"},"dab62bd395dd366a1fa299f7be59c224928a2bc1":{"kind":"REWORK","_number":3,"created":"2024-10-01 08:22:30.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/83/930983/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/83/930983/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/83/930983/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/83/930983/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/83/930983/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/83/930983/3"}}},"commit":{"parents":[{"commit":"b12f59e2ec50e1e5e5bf15ffbdaf4815c4db3174","subject":"zookeeper: fix argparse related to ssl","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/b12f59e2ec50e1e5e5bf15ffbdaf4815c4db3174"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-01 07:58:47.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-01 08:22:23.000000000","tz":540},"subject":"Use oslo.utils implementation to parse sentinel address","message":"Use oslo.utils implementation to parse sentinel address\n\n... instead of maintaining own logic.\n\nChange-Id: I90d65f3e5b1be6b32e8c44a018c85c01ca47d3b2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/dab62bd395dd366a1fa299f7be59c224928a2bc1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/dab62bd395dd366a1fa299f7be59c224928a2bc1"}]},"branch":"refs/heads/master"},"0ff1151f9e63a814ddc1804e82a21a5a41e77a24":{"kind":"REWORK","_number":4,"created":"2024-10-01 08:25:05.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/83/930983/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/83/930983/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/83/930983/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/83/930983/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/83/930983/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/83/930983/4"}}},"commit":{"parents":[{"commit":"b12f59e2ec50e1e5e5bf15ffbdaf4815c4db3174","subject":"zookeeper: fix argparse related to ssl","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/b12f59e2ec50e1e5e5bf15ffbdaf4815c4db3174"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-01 07:58:47.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-01 08:24:57.000000000","tz":540},"subject":"Use oslo.utils implementation to parse sentinel address","message":"Use oslo.utils implementation to parse sentinel address\n\n... instead of maintaining own logic.\n\nChange-Id: I90d65f3e5b1be6b32e8c44a018c85c01ca47d3b2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/0ff1151f9e63a814ddc1804e82a21a5a41e77a24"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/0ff1151f9e63a814ddc1804e82a21a5a41e77a24"}]},"branch":"refs/heads/master"},"cf6d4ecac1c6d2050260e5181354b4a602f9924e":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2024-10-01 18:14:43.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/83/930983/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/83/930983/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/83/930983/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/83/930983/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/83/930983/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/83/930983/5"}}},"commit":{"parents":[{"commit":"2e666107fdaf3ea7de0688eee0432fc3981803be","subject":"Fix wrong usage of \u0027wraps\u0027","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/2e666107fdaf3ea7de0688eee0432fc3981803be"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-01 07:58:47.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-01 18:14:43.000000000","tz":0},"subject":"Use oslo.utils implementation to parse sentinel address","message":"Use oslo.utils implementation to parse sentinel address\n\n... instead of maintaining own logic.\n\nChange-Id: I90d65f3e5b1be6b32e8c44a018c85c01ca47d3b2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/cf6d4ecac1c6d2050260e5181354b4a602f9924e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/cf6d4ecac1c6d2050260e5181354b4a602f9924e"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
