)]}'
{"id":"openstack%2Ftooz~930987","triplet_id":"openstack%2Ftooz~master~I315de78098d4a7cbe77fffd59ca848caf7d91eb6","project":"openstack/tooz","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-10-05 10:12:42.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"I315de78098d4a7cbe77fffd59ca848caf7d91eb6","subject":"Use oslo.utils implementation to compare versions","status":"MERGED","created":"2024-10-01 08:49:34.000000000","updated":"2024-10-05 13:23:35.000000000","submitted":"2024-10-05 13:22:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"930987","meta_rev_id":"d17b36fee231392b668c00196de53807918c6da9","_number":930987,"virtual_id_number":930987,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-10-05 13:22:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"recommended":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"all":[{"value":1,"date":"2024-10-02 17:42:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2024-10-04 11:40:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":1,"date":"2024-10-05 03:54:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-01 10:18:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-10-02 17:42:29.000000000","updated_by":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"reviewer":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"state":"REVIEWER"},{"updated":"2024-10-04 11:40:28.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"c989a1552d69751359b3b102499210f7a970dbe7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-01 08:49:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"44012d844fb16264442ec18ef291f346147c9512","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-01 10:18:23.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/abf7bdc8200d40f281206e95be66e613\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/e2d6acc3e8234f3ba824d9f3056a4de5 : SUCCESS in 2m 32s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d211cff712f0452aafab21d45bad589c : SUCCESS in 1h 25m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f4dd281ca9a648cd83ddaba548f5ac74 : SUCCESS in 5m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4103ec8f70204089b2ee1bb5c887e686 : SUCCESS in 4m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/493096b358e54cf1ad44459d888e5f15 : SUCCESS in 4m 28s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a1a00e740f324056a7a87bf49435e688 : FAILURE in 6m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc7183a7ee12490ab8d8af9b93138144 : SUCCESS in 2m 56s\n- tooz-tox-py38-etcd3gw https://zuul.opendev.org/t/openstack/build/6163a074f3b7437babf7fc65a7688d42 : SUCCESS in 4m 19s\n- tooz-tox-py310-etcd3gw https://zuul.opendev.org/t/openstack/build/016c0891e25e409b9a2d355ba30b8d1b : SUCCESS in 4m 16s\n- tooz-tox-py38-zookeeper https://zuul.opendev.org/t/openstack/build/84657ed3317a4a19b337e1f5dcaf3381 : FAILURE in 3m 44s\n- tooz-tox-py310-zookeeper https://zuul.opendev.org/t/openstack/build/792813832d9e4d68964d12e24752eb9e : SUCCESS in 3m 32s\n- tooz-tox-py38-redis https://zuul.opendev.org/t/openstack/build/01523c709434485b8dbcec742325c1e4 : SUCCESS in 4m 08s\n- tooz-tox-py310-redis https://zuul.opendev.org/t/openstack/build/51f9aad62cc346c9b2144e5d381ecadb : SUCCESS in 3m 51s\n- tooz-tox-py38-sentinel https://zuul.opendev.org/t/openstack/build/eff7a6e9e01944b3a4fcd7987d25d8db : SUCCESS in 3m 52s\n- tooz-tox-py310-sentinel https://zuul.opendev.org/t/openstack/build/8227b3c7e4f64813959d453f3a093781 : SUCCESS in 4m 10s\n- tooz-tox-py38-memcached https://zuul.opendev.org/t/openstack/build/efa0be1a500749449a1b9b4d03c44af5 : SUCCESS in 4m 12s\n- tooz-tox-py310-memcached https://zuul.opendev.org/t/openstack/build/0d245b54278c4fbeb94d4ed6afdd41f4 : SUCCESS in 4m 08s\n- tooz-tox-py38-postgresql https://zuul.opendev.org/t/openstack/build/7390911f2914495bb7a7c175d14ed659 : SUCCESS in 4m 11s\n- tooz-tox-py310-postgresql https://zuul.opendev.org/t/openstack/build/454ed6ee0d6f4dfcb43d210783919ec8 : SUCCESS in 4m 12s\n- tooz-tox-py38-mysql https://zuul.opendev.org/t/openstack/build/3dbd797396b94906957d980ec5558c4f : SUCCESS in 4m 04s\n- tooz-tox-py310-mysql https://zuul.opendev.org/t/openstack/build/498d840a693940479962133811e514e9 : SUCCESS in 4m 21s\n- tooz-tox-py38-consul https://zuul.opendev.org/t/openstack/build/06c33b09dbf7430a87b6ae65ea08f672 : SUCCESS in 4m 18s\n- tooz-tox-py310-consul https://zuul.opendev.org/t/openstack/build/2eae8bcbb82b4181803aa8f0ec728170 : SUCCESS in 4m 12s","accounts_in_message":[],"_revision_number":1},{"id":"f003e46e01dd884ac9757ecfc313d7d85c7c8391","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-01 18:14:50.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"badb11331cef7b44793dc17919a02833579e43e5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-01 19:50:31.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/78d43a5ac2154e2aba60183942202a2a\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/7875f8da5d544b91a0dbd1800a6e7c15 : SUCCESS in 2m 31s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/86e58240e4cb4851847ed95e56776627 : SUCCESS in 1h 34m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b266304f6afd4a7a90746917969e16d7 : SUCCESS in 5m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e8e18c35a7af4dbd848df505280c3433 : SUCCESS in 3m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e870808c96364438b73a676c447900af : SUCCESS in 4m 36s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a80419278c5f4969b859264956eab0bc : SUCCESS in 34m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/643d565ba7b849f4babc89533d26897e : SUCCESS in 4m 20s\n- tooz-tox-py38-etcd3gw https://zuul.opendev.org/t/openstack/build/5dc958d24756414aa29441e8b09af429 : SUCCESS in 4m 59s\n- tooz-tox-py310-etcd3gw https://zuul.opendev.org/t/openstack/build/d5e7b5d164564175a3ba4a25b0e7557a : SUCCESS in 4m 41s\n- tooz-tox-py38-zookeeper https://zuul.opendev.org/t/openstack/build/c10c83c963fe4d6f8599e90b191d9e4a : SUCCESS in 4m 21s\n- tooz-tox-py310-zookeeper https://zuul.opendev.org/t/openstack/build/040ba9f3a46a466fbff7981ea0a82443 : SUCCESS in 4m 30s\n- tooz-tox-py38-redis https://zuul.opendev.org/t/openstack/build/e680eacc67c34324a129fdd7ef682ac9 : SUCCESS in 4m 47s\n- tooz-tox-py310-redis https://zuul.opendev.org/t/openstack/build/527f11813ab94a3d9b39380d0933ce16 : SUCCESS in 4m 12s\n- tooz-tox-py38-sentinel https://zuul.opendev.org/t/openstack/build/3fb8138c94134599b8920344466ce1a5 : SUCCESS in 4m 15s\n- tooz-tox-py310-sentinel https://zuul.opendev.org/t/openstack/build/8768b01261754354b6cdab74686008c3 : SUCCESS in 5m 45s\n- tooz-tox-py38-memcached https://zuul.opendev.org/t/openstack/build/f982af5fd2c94851a660d2f26f771ab0 : SUCCESS in 4m 46s\n- tooz-tox-py310-memcached https://zuul.opendev.org/t/openstack/build/88b9a3ab8d9a408b9f92bc6af3070216 : SUCCESS in 3m 18s\n- tooz-tox-py38-postgresql https://zuul.opendev.org/t/openstack/build/ad70af1fc3114f53bb5e732c60a12692 : SUCCESS in 4m 16s\n- tooz-tox-py310-postgresql https://zuul.opendev.org/t/openstack/build/c47b4e976d594c0dbdaac06206872097 : SUCCESS in 4m 04s\n- tooz-tox-py38-mysql https://zuul.opendev.org/t/openstack/build/595917d632624737b46b58846fafa3f7 : SUCCESS in 4m 31s\n- tooz-tox-py310-mysql https://zuul.opendev.org/t/openstack/build/ed233e5af2d44cd48495962cb3ebd2a4 : SUCCESS in 4m 27s\n- tooz-tox-py38-consul https://zuul.opendev.org/t/openstack/build/1364582b62bb486fa7481fe600895bf7 : SUCCESS in 5m 25s\n- tooz-tox-py310-consul https://zuul.opendev.org/t/openstack/build/e0080664ef8e4b28b0ea5c46521fe675 : SUCCESS in 4m 22s","accounts_in_message":[],"_revision_number":2},{"id":"ab93d0c8dba1acba45d38396deb00d49483c5d8b","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2024-10-02 17:42:29.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"c33500d0f619af1d6de4b81b95ef851e0201a9c7","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2024-10-04 11:40:28.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"f65a32fa612505d1f85bb25fab173f5049a83e54","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-05 03:54:30.000000000","message":"Patch Set 2: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3f92dd8f90c9b82e15f7eef13ff93b276b5bb788","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-05 03:55:01.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"0ba3615f01627eb17d3f57787adbb5310cc53257","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-05 05:03:30.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ba06af6abddf466392d90f53d0682361\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/b5e6d029382a4dc5bd861c4e7cca5ffa : SUCCESS in 1m 40s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/73625c0c830a4722bfc549d81a4f0ab1 : SUCCESS in 1h 01m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3fb2c0a8e21d476ebe1b2b075e03b05d : SUCCESS in 4m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/433213bdec944e4c81fdea58963a333f : SUCCESS in 3m 13s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/cccb435bab3e48bea751d7dd20e43c06 : SUCCESS in 4m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/72a92da930ea47c197c7941584eb2ae4 : SUCCESS in 3m 35s\n- tooz-tox-py38-etcd3gw https://zuul.opendev.org/t/openstack/build/bfbb051dc972497d90fe045ecee5de47 : SUCCESS in 4m 26s\n- tooz-tox-py310-etcd3gw https://zuul.opendev.org/t/openstack/build/e343a1ac22e4456e8c821fa89c2340b7 : SUCCESS in 4m 54s\n- tooz-tox-py38-zookeeper https://zuul.opendev.org/t/openstack/build/e013ddd4b0b74607bb53e0929a7d4dea : SUCCESS in 4m 01s\n- tooz-tox-py310-zookeeper https://zuul.opendev.org/t/openstack/build/5e208ba341184d71b9199fac0e4b7549 : FAILURE in 3m 51s\n- tooz-tox-py38-redis https://zuul.opendev.org/t/openstack/build/d6a74c25cfe14bd29d571dab47941874 : SUCCESS in 5m 10s\n- tooz-tox-py310-redis https://zuul.opendev.org/t/openstack/build/676c40bcecdf4e75ac6d53ae96668b90 : SUCCESS in 3m 19s\n- tooz-tox-py38-sentinel https://zuul.opendev.org/t/openstack/build/49ba121dc6994b2482299cb189cb18f3 : SUCCESS in 4m 10s\n- tooz-tox-py310-sentinel https://zuul.opendev.org/t/openstack/build/8a72bafca5264099b75be49f726260eb : SUCCESS in 3m 54s\n- tooz-tox-py38-memcached https://zuul.opendev.org/t/openstack/build/228d1e3a8d004837b36a5675d5a53ecf : SUCCESS in 4m 03s\n- tooz-tox-py310-memcached https://zuul.opendev.org/t/openstack/build/2c47bb00a8ca4e74bc586f68fe488af5 : SUCCESS in 4m 04s\n- tooz-tox-py38-postgresql https://zuul.opendev.org/t/openstack/build/903956de41b24600a441329da369b9ef : SUCCESS in 4m 21s\n- tooz-tox-py310-postgresql https://zuul.opendev.org/t/openstack/build/fa3c62eadfb947cbb2b84082362d6336 : SUCCESS in 5m 05s\n- tooz-tox-py38-mysql https://zuul.opendev.org/t/openstack/build/064377f5767f49b39ae9345555f4eb98 : SUCCESS in 4m 25s\n- tooz-tox-py310-mysql https://zuul.opendev.org/t/openstack/build/2b8d3632fb784c9d88407e08a4e28606 : SUCCESS in 3m 11s\n- tooz-tox-py38-consul https://zuul.opendev.org/t/openstack/build/d94bb0056e92478db0cacdc2a2cf488f : SUCCESS in 4m 17s\n- tooz-tox-py310-consul https://zuul.opendev.org/t/openstack/build/c6714352e80248ca8d7ff94e5f7adf3e : SUCCESS in 3m 55s","accounts_in_message":[],"_revision_number":2},{"id":"524cb0a581d9a21e9573a0573d0e899433f2ccbe","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-05 07:28:37.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9b38d732a9dbdf41cbf74c4398c5668d04e822a4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-05 09:06:48.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0d08cc42286b4e8290cfb0539576ffea\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/b85a0aef66bf485b8b1e53398e07183c : SUCCESS in 1m 46s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/795ae48197ea47fd8eb44204871c28c8 : SUCCESS in 1h 37m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b8a1ead363644f778a571c3941e55766 : SUCCESS in 3m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/41ea88a7d74f45b09383480142629a42 : SUCCESS in 3m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/81b95fc19aab407ab7e896c073556bb9 : SUCCESS in 3m 54s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a4ea13bbd8bc4a6da6a23396aefbe6bf : SUCCESS in 8m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/729db65b5d95489fbc0c42cba9c35ed9 : SUCCESS in 3m 23s\n- tooz-tox-py38-etcd3gw https://zuul.opendev.org/t/openstack/build/23c91c36e7ec44a583f681291b5df0ad : SUCCESS in 4m 36s\n- tooz-tox-py310-etcd3gw https://zuul.opendev.org/t/openstack/build/35b871a7a3cc4e218a43adad3b0f7378 : SUCCESS in 4m 39s\n- tooz-tox-py38-zookeeper https://zuul.opendev.org/t/openstack/build/adc225a5cc9a41598508d89ad4517574 : SUCCESS in 3m 58s\n- tooz-tox-py310-zookeeper https://zuul.opendev.org/t/openstack/build/71212241783d41cf8020c7059f418d1a : SUCCESS in 4m 05s\n- tooz-tox-py38-redis https://zuul.opendev.org/t/openstack/build/3ed9f41896674df0b8d6d0a467cd3ffc : SUCCESS in 3m 44s\n- tooz-tox-py310-redis https://zuul.opendev.org/t/openstack/build/e9ce7eb8a3124f6f99e42804deaf7488 : SUCCESS in 3m 49s\n- tooz-tox-py38-sentinel https://zuul.opendev.org/t/openstack/build/5b4d473cc52c4c2f8cb0d616ab7b0a44 : SUCCESS in 3m 45s\n- tooz-tox-py310-sentinel https://zuul.opendev.org/t/openstack/build/700609e08b01484a8ad6e331ce0f299f : SUCCESS in 3m 53s\n- tooz-tox-py38-memcached https://zuul.opendev.org/t/openstack/build/333a665ed6784a60b3082c11664f2bb1 : SUCCESS in 3m 58s\n- tooz-tox-py310-memcached https://zuul.opendev.org/t/openstack/build/6b68c109213e4e8fb1e39d4b5aa080e5 : SUCCESS in 3m 50s\n- tooz-tox-py38-postgresql https://zuul.opendev.org/t/openstack/build/6fa2f30eccb348c7ad776ce3c66c8476 : SUCCESS in 3m 27s\n- tooz-tox-py310-postgresql https://zuul.opendev.org/t/openstack/build/dfd9e785f42f40dd819568f49ea66c21 : SUCCESS in 3m 48s\n- tooz-tox-py38-mysql https://zuul.opendev.org/t/openstack/build/fe09a68195c54d83bec343c4d463c594 : SUCCESS in 3m 47s\n- tooz-tox-py310-mysql https://zuul.opendev.org/t/openstack/build/4583040c01e04e5c9c0fdc3f554b069a : SUCCESS in 4m 54s\n- tooz-tox-py38-consul https://zuul.opendev.org/t/openstack/build/c034c35fa131450bad50bc922fd6a9f3 : SUCCESS in 3m 52s\n- tooz-tox-py310-consul https://zuul.opendev.org/t/openstack/build/e48e9c2df25c408f962d605d04bb627d : SUCCESS in 3m 48s","accounts_in_message":[],"_revision_number":2},{"id":"03e8b388c7f66dacc54286f282c808ddc72c86cb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-05 09:07:18.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"5ec7a90ee9f5a80c65959cb971276073c8d1246b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-05 10:10:44.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/17d4cfea101f42bda865751ce4556eec\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/391314ff85a641ab8cbba07846b4e26e : SUCCESS in 2m 10s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9a25e234db4a41d18b890bd068795971 : SUCCESS in 1h 02m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b88a4e72f8584e21ba5915b06ffca2e0 : SUCCESS in 2m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d3a8a8fe63764ce89aceaae9e4d010b0 : SUCCESS in 4m 38s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f8f64d81b7b84b5c85116a018a5eec70 : SUCCESS in 4m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d6f2ec4c8cf643b1bd2a49a242d820d4 : SUCCESS in 4m 05s\n- tooz-tox-py38-etcd3gw https://zuul.opendev.org/t/openstack/build/d831cbce08bc4c5cb5a9adb2dba38118 : SUCCESS in 4m 08s\n- tooz-tox-py310-etcd3gw https://zuul.opendev.org/t/openstack/build/e4cffaa5fff546979c1d1751e14b6b96 : SUCCESS in 4m 19s\n- tooz-tox-py38-zookeeper https://zuul.opendev.org/t/openstack/build/daeeaca3b3bc48a1a993b92e53533ae8 : SUCCESS in 3m 50s\n- tooz-tox-py310-zookeeper https://zuul.opendev.org/t/openstack/build/6eb04b37e5c54f6bb4e6ba2f6366633a : FAILURE in 3m 57s\n- tooz-tox-py38-redis https://zuul.opendev.org/t/openstack/build/b0550f4e8e2f4abd824db891c8f3f0d4 : SUCCESS in 5m 00s\n- tooz-tox-py310-redis https://zuul.opendev.org/t/openstack/build/d8b89733c57944a1bc61e5964951dd8b : SUCCESS in 4m 18s\n- tooz-tox-py38-sentinel https://zuul.opendev.org/t/openstack/build/1d5b38cbfc204169b39966f6c47bb1df : SUCCESS in 4m 59s\n- tooz-tox-py310-sentinel https://zuul.opendev.org/t/openstack/build/7eac21da075d4c6aa54a454564ef00a8 : SUCCESS in 4m 18s\n- tooz-tox-py38-memcached https://zuul.opendev.org/t/openstack/build/608925235a6e4226b9e376477b416ad1 : SUCCESS in 4m 51s\n- tooz-tox-py310-memcached https://zuul.opendev.org/t/openstack/build/0aa6008b73ee41b2993981b41d3154e7 : SUCCESS in 3m 19s\n- tooz-tox-py38-postgresql https://zuul.opendev.org/t/openstack/build/b3b6d1415b81426a8011ed95f770fa07 : SUCCESS in 3m 44s\n- tooz-tox-py310-postgresql https://zuul.opendev.org/t/openstack/build/7d3abc3377cc43b99ded09e5cccfe04e : SUCCESS in 3m 14s\n- tooz-tox-py38-mysql https://zuul.opendev.org/t/openstack/build/ffd45fb163bd43509f318cd5813717cb : SUCCESS in 3m 59s\n- tooz-tox-py310-mysql https://zuul.opendev.org/t/openstack/build/ff533ae6c8c548e78e13ae44bdcac96e : SUCCESS in 3m 42s\n- tooz-tox-py38-consul https://zuul.opendev.org/t/openstack/build/7939ac4dd08247c9830e49dc31f2a8fc : SUCCESS in 3m 44s\n- tooz-tox-py310-consul https://zuul.opendev.org/t/openstack/build/0310914be0324d88b010c3b812ea77c4 : SUCCESS in 4m 15s","accounts_in_message":[],"_revision_number":2},{"id":"8c486bcc64dee31fe0b0c1fea8e0b484d2feb5b2","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-05 10:12:42.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"60a181255fcae5513858c53ede0926050b577bb6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-05 11:42:54.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5a0a722d2a7a4157872b935c75c3d844\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/ce347eb9345f497c81d15de3474e9ba6 : SUCCESS in 1m 57s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/99185fad6f0147e0b507027ef6d5e756 : SUCCESS in 1h 28m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2607548a6d86407db5aa96cab769f13a : SUCCESS in 3m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/60cb0900f17d4e818310bb075e59cd92 : SUCCESS in 3m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/45089f8499634b15aaff98053e01c780 : SUCCESS in 4m 01s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/697949ff89574b05bb43cedc6541537f : SUCCESS in 5m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/76771105f2f145b0b205e22c5d82ff0b : SUCCESS in 3m 40s\n- tooz-tox-py38-etcd3gw https://zuul.opendev.org/t/openstack/build/86dcb2c5aaab4313aa11697bd61f0b2c : SUCCESS in 4m 31s\n- tooz-tox-py310-etcd3gw https://zuul.opendev.org/t/openstack/build/0fd4292c6067475ea9f92b0aa9d190e9 : SUCCESS in 4m 31s\n- tooz-tox-py38-zookeeper https://zuul.opendev.org/t/openstack/build/cefa462f6c41414289f7c5efe8fb06b0 : SUCCESS in 3m 56s\n- tooz-tox-py310-zookeeper https://zuul.opendev.org/t/openstack/build/db5b0a404f39480280058736afb5a684 : SUCCESS in 3m 52s\n- tooz-tox-py38-redis https://zuul.opendev.org/t/openstack/build/e330c66fa68f4731b7d71daee9341812 : SUCCESS in 4m 17s\n- tooz-tox-py310-redis https://zuul.opendev.org/t/openstack/build/8882a99eef2e48a1831c6cb835ad3abd : SUCCESS in 5m 08s\n- tooz-tox-py38-sentinel https://zuul.opendev.org/t/openstack/build/35b94db3e03442f58539cb5dff4fea07 : SUCCESS in 3m 17s\n- tooz-tox-py310-sentinel https://zuul.opendev.org/t/openstack/build/93026d3d8e34400f853fba362d987460 : SUCCESS in 3m 51s\n- tooz-tox-py38-memcached https://zuul.opendev.org/t/openstack/build/e6f7ef12000048e5b6af3cbc532ba4f7 : SUCCESS in 3m 50s\n- tooz-tox-py310-memcached https://zuul.opendev.org/t/openstack/build/1223490c939e460a80dee825dadf6376 : SUCCESS in 5m 56s\n- tooz-tox-py38-postgresql https://zuul.opendev.org/t/openstack/build/fbd34404cd9345af867fcee319028cc2 : SUCCESS in 3m 42s\n- tooz-tox-py310-postgresql https://zuul.opendev.org/t/openstack/build/995c1551c8fe46848af34f398b9ccb56 : SUCCESS in 2m 57s\n- tooz-tox-py38-mysql https://zuul.opendev.org/t/openstack/build/f9a541c7a71a4043b3933f007444dc8c : SUCCESS in 5m 09s\n- tooz-tox-py310-mysql https://zuul.opendev.org/t/openstack/build/2a258661c734469a8c203d860fcd554c : SUCCESS in 3m 51s\n- tooz-tox-py38-consul https://zuul.opendev.org/t/openstack/build/7836b1a323c04ba6b5192ed581597f54 : SUCCESS in 4m 12s\n- tooz-tox-py310-consul https://zuul.opendev.org/t/openstack/build/1f7fe17b59394fa3893eaf8cae9c1b83 : SUCCESS in 3m 10s","accounts_in_message":[],"_revision_number":2},{"id":"53f99643d83d865c5e4bcb46dacb50e6d83cbf61","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-05 11:43:33.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"2eb1b9892b7c4ee2c6c70193aa0ad607ae3e7bf9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-05 13:22:40.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c22aa423c74e4c01abc2cdc9bc3eedd4\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/86c39402b0284782a472d5f268cff220 : SUCCESS in 3m 21s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/595557106d24429da0e9761611750dcd : SUCCESS in 1h 39m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b2e76ef81da8485fb4f3e8c919d83d7b : SUCCESS in 5m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/707cabc0d0c44dad86e2b7e01701da28 : SUCCESS in 3m 52s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ad1c1d85fa994e55b978ab3ed38fcabb : SUCCESS in 8m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/39cd6b8397f247d78cd7a4d930a4e1f4 : SUCCESS in 3m 38s\n- tooz-tox-py39-etcd3gw https://zuul.opendev.org/t/openstack/build/20196ce87dd043b99303b880bcaf90bd : SUCCESS in 7m 57s\n- tooz-tox-py312-etcd3gw https://zuul.opendev.org/t/openstack/build/3be9c8990f674e1eafaba0ece3165334 : SUCCESS in 5m 33s\n- tooz-tox-py39-zookeeper https://zuul.opendev.org/t/openstack/build/7852f52dd0b14f429556fa951643fdf6 : SUCCESS in 4m 25s\n- tooz-tox-py312-zookeeper https://zuul.opendev.org/t/openstack/build/7903ea29eefe48f7addac004093bb591 : SUCCESS in 3m 42s\n- tooz-tox-py39-redis https://zuul.opendev.org/t/openstack/build/1de226e1fd3d4157ac08d58db506eef3 : SUCCESS in 4m 08s\n- tooz-tox-py312-redis https://zuul.opendev.org/t/openstack/build/3caaabafa1a84b6789dd54c684caa856 : SUCCESS in 3m 36s\n- tooz-tox-py39-sentinel https://zuul.opendev.org/t/openstack/build/b1f344c905d64189ae062528ebcfd9df : SUCCESS in 4m 21s\n- tooz-tox-py312-sentinel https://zuul.opendev.org/t/openstack/build/8022f22340e44cecacec5d1cb4f8b443 : SUCCESS in 5m 07s\n- tooz-tox-py39-memcached https://zuul.opendev.org/t/openstack/build/4d83e195a4314e45a72a6f3c0dcb7c15 : SUCCESS in 4m 12s\n- tooz-tox-py312-memcached https://zuul.opendev.org/t/openstack/build/29df19d608d34edfa8518b86d2cdeade : SUCCESS in 4m 42s\n- tooz-tox-py39-postgresql https://zuul.opendev.org/t/openstack/build/a8855db8302744cca552855f5122e1dd : SUCCESS in 4m 47s\n- tooz-tox-py312-postgresql https://zuul.opendev.org/t/openstack/build/fb6a4e0dcfe942fb9ada52c70b29ee7b : SUCCESS in 5m 44s\n- tooz-tox-py39-mysql https://zuul.opendev.org/t/openstack/build/d887b2d23fc8450ab3774380854b5a23 : SUCCESS in 3m 54s\n- tooz-tox-py312-mysql https://zuul.opendev.org/t/openstack/build/353f5257b4354ecdb3e7ed63adb5f3de : SUCCESS in 5m 09s\n- tooz-tox-py39-consul https://zuul.opendev.org/t/openstack/build/2d9fc60c9838448485e4f7be29317408 : SUCCESS in 5m 22s\n- tooz-tox-py312-consul https://zuul.opendev.org/t/openstack/build/6b93266885b3469ba13ad56152d6d834 : SUCCESS in 5m 01s","accounts_in_message":[],"_revision_number":2},{"id":"954ad84134823d4b9abcd625ef4b99b03817c0a4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-05 13:22:40.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"d17b36fee231392b668c00196de53807918c6da9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-05 13:23:35.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bce2af3aa5324144a70aab1e76dc219b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/16d1d5428ed74cbc947e7c8ba6994a27 : SUCCESS in 36s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"46366e0d25392e6e8f20b172dd48c818f97abbd2","revisions":{"951b3dc4f8707cc74c982c602182bd7a0fbbeae4":{"kind":"REWORK","_number":1,"created":"2024-10-01 08:49:34.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/87/930987/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/87/930987/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/87/930987/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/87/930987/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/87/930987/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/87/930987/1"}}},"commit":{"parents":[{"commit":"b12f59e2ec50e1e5e5bf15ffbdaf4815c4db3174","subject":"zookeeper: fix argparse related to ssl","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/b12f59e2ec50e1e5e5bf15ffbdaf4815c4db3174"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-01 08:48:05.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-01 08:49:00.000000000","tz":540},"subject":"Use oslo.utils implementation to compare versions","message":"Use oslo.utils implementation to compare versions\n\nThe oslo.utils library provides the utility method to compare versions.\nUse it instead of directly using packaging library, because oslo.utils\nis already required by core tooz logic.\n\nChange-Id: I315de78098d4a7cbe77fffd59ca848caf7d91eb6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/951b3dc4f8707cc74c982c602182bd7a0fbbeae4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/951b3dc4f8707cc74c982c602182bd7a0fbbeae4"}]},"branch":"refs/heads/master"},"46366e0d25392e6e8f20b172dd48c818f97abbd2":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-10-01 18:14:50.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/87/930987/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/87/930987/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/87/930987/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/87/930987/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/87/930987/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/87/930987/2"}}},"commit":{"parents":[{"commit":"2e666107fdaf3ea7de0688eee0432fc3981803be","subject":"Fix wrong usage of \u0027wraps\u0027","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/2e666107fdaf3ea7de0688eee0432fc3981803be"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-01 08:48:05.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-01 18:14:50.000000000","tz":0},"subject":"Use oslo.utils implementation to compare versions","message":"Use oslo.utils implementation to compare versions\n\nThe oslo.utils library provides the utility method to compare versions.\nUse it instead of directly using packaging library, because oslo.utils\nis already required by core tooz logic.\n\nChange-Id: I315de78098d4a7cbe77fffd59ca848caf7d91eb6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/46366e0d25392e6e8f20b172dd48c818f97abbd2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/46366e0d25392e6e8f20b172dd48c818f97abbd2"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
