)]}'
{"id":"openstack%2Ftooz~930990","triplet_id":"openstack%2Ftooz~master~Ic73b9876b6e3d8a3ac836974b24b7cb1bc7795f1","project":"openstack/tooz","branch":"master","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2024-10-03 02:56:19.000000000","reason":"Change was submitted"},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-10-03 01:04:25.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},"16137":{"account":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"last_update":"2024-10-02 11:03:16.000000000","reason":"\u003cGERRIT_ACCOUNT_16137\u003e replied on the change","reason_account":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}},"hashtags":[],"change_id":"Ic73b9876b6e3d8a3ac836974b24b7cb1bc7795f1","subject":"Fix wrong usage of \u0027wraps\u0027","status":"MERGED","created":"2024-10-01 10:15:13.000000000","updated":"2024-10-03 02:57:54.000000000","submitted":"2024-10-03 02:56:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"930990","meta_rev_id":"3ee78fda6bd0c161a8652679b639c7a8fd32ad1c","_number":930990,"virtual_id_number":930990,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-10-03 02:56:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"recommended":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"all":[{"value":1,"date":"2024-10-02 17:41:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2024-10-03 01:04:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2024-10-02 11:03:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":1,"date":"2024-10-03 01:04:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-01 10:56:04.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2024-10-01 11:39:58.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2024-10-01 14:58:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-10-02 17:41:22.000000000","updated_by":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"reviewer":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"state":"REVIEWER"}],"messages":[{"id":"b0474876235d154831dca470f0d8bc351cddbdcd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-01 10:15:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ac4aadd7c86ccf8d2fdf5aa24f0b62f80357e367","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-01 10:18:24.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"93d0205680a7b200bd891f3c4aedce7acd991f1c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-01 10:25:02.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"4ec996af486e1d12e9bed67ab8a206044c6c5869","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-10-01 10:56:04.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"5a069acace4ffc214df310e58220860df8ea3960","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2024-10-01 11:39:58.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"099a283a357075449b6c543b7d759b0db48e7def","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-01 11:53:43.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Workflow+1\n","accounts_in_message":[],"_revision_number":4},{"id":"80e3b1e0333c1405ec50b6d801417e6f36aa246b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-01 12:45:01.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"323f535aa039837212c26ae6f9981fe929da0544","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-01 13:07:39.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"faef92a9c5c8f817007c639ab6087f36d00a9212","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-01 14:58:53.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/117cffd33ab34a17a1e581d7eb3921dc\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5c34adc41e78438ca2897cee12fe06a4 : SUCCESS in 1h 45m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0de9da15f57e43b89eb41f8d5e409423 : SUCCESS in 4m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2eedfc5e72c14a9c961d76f487e44ea8 : SUCCESS in 3m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e8cdef3fc26f4f5eb6e9919e39e7b56b : SUCCESS in 4m 04s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7bbcb90d12ef495db5e48159e1e96231 : SUCCESS in 4m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a73a7e1482b4379b196a2256b9ff77c : SUCCESS in 2m 47s\n- tooz-tox-py38-etcd3gw https://zuul.opendev.org/t/openstack/build/2fb61c05d811485196971001b0551295 : SUCCESS in 5m 10s\n- tooz-tox-py310-etcd3gw https://zuul.opendev.org/t/openstack/build/c17ef587e62f46dd86b82949ca9d1bba : SUCCESS in 4m 55s\n- tooz-tox-py38-zookeeper https://zuul.opendev.org/t/openstack/build/c9ace766857c4eec9843a801e6946e95 : SUCCESS in 4m 54s\n- tooz-tox-py310-zookeeper https://zuul.opendev.org/t/openstack/build/0d5e41d00bb44e06b3f46bd118a5ab77 : SUCCESS in 4m 27s\n- tooz-tox-py38-redis https://zuul.opendev.org/t/openstack/build/0f3e6412a7134306bcdda2dc8e5e2d6d : SUCCESS in 4m 05s\n- tooz-tox-py310-redis https://zuul.opendev.org/t/openstack/build/b1dce4d682ba42c2a316e184fc0d1bb4 : SUCCESS in 4m 10s\n- tooz-tox-py38-sentinel https://zuul.opendev.org/t/openstack/build/55b7ad879a5b4f7a98c737944f78f547 : SUCCESS in 4m 03s\n- tooz-tox-py310-sentinel https://zuul.opendev.org/t/openstack/build/b3e2aa49e2f14c239449102959336c72 : SUCCESS in 4m 16s\n- tooz-tox-py38-memcached https://zuul.opendev.org/t/openstack/build/77c62d1ef5484c3facf02733003aaf8b : SUCCESS in 3m 54s\n- tooz-tox-py310-memcached https://zuul.opendev.org/t/openstack/build/ba3225d818974ae9a1f269590f27f835 : SUCCESS in 4m 11s\n- tooz-tox-py38-postgresql https://zuul.opendev.org/t/openstack/build/37b86d9adb424c8082862957c955b24f : SUCCESS in 3m 48s\n- tooz-tox-py310-postgresql https://zuul.opendev.org/t/openstack/build/d2f33879f1f440e4ad1af57a9d90a597 : SUCCESS in 3m 53s\n- tooz-tox-py38-mysql https://zuul.opendev.org/t/openstack/build/4613228035b246b0bc51aad5bd3ed45c : SUCCESS in 4m 53s\n- tooz-tox-py310-mysql https://zuul.opendev.org/t/openstack/build/2a17de759dda4ef68eece624477ec1bb : SUCCESS in 4m 06s\n- tooz-tox-py38-consul https://zuul.opendev.org/t/openstack/build/a544871073714df9ba4175b4a8ee1344 : SUCCESS in 3m 58s\n- tooz-tox-py310-consul https://zuul.opendev.org/t/openstack/build/ab5b89a5482e4635a42b3546664aada3 : SUCCESS in 4m 06s","accounts_in_message":[],"_revision_number":6},{"id":"6bf320398fd8268dd0effe6f28df85b6d391a95f","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2024-10-02 11:03:16.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"3401cff8edd29796c1aed6be1118576e678ee978","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2024-10-02 17:41:22.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"f219e80391927e10f242112f9cd2a7105964d308","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-03 01:04:25.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"5c5571429ad29faa014a5071243d391f606f8ae0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-03 01:04:55.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"fc742d10b7b26a7ac5e2fa41cd7f007ea18a3dab","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-03 02:56:19.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cb610cf838a64a128229a0e0d248ff26\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ab5f9c3d4c064a8bbf4371c52e1c6ca5 : SUCCESS in 1h 50m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2dcbd5c0bbd342b79cd78039fe0aa903 : SUCCESS in 4m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7f250d06c89a4f4fa5927053adfab044 : SUCCESS in 5m 44s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b3cb786fcb334637aa006db56e28bafb : SUCCESS in 8m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f2e65bfb45a945b6ad1d594a0a61885e : SUCCESS in 4m 34s\n- tooz-tox-py38-etcd3gw https://zuul.opendev.org/t/openstack/build/6531be6ffe1e4040961af1f5fc74e736 : SUCCESS in 5m 33s\n- tooz-tox-py310-etcd3gw https://zuul.opendev.org/t/openstack/build/b386105e93204440b8a6bd266c53746c : SUCCESS in 5m 37s\n- tooz-tox-py38-zookeeper https://zuul.opendev.org/t/openstack/build/4e61f77573c24c7c805aa2fabba2cb40 : SUCCESS in 6m 43s\n- tooz-tox-py310-zookeeper https://zuul.opendev.org/t/openstack/build/0fcd4c839190420f9e5362c092b3ca94 : SUCCESS in 5m 14s\n- tooz-tox-py38-redis https://zuul.opendev.org/t/openstack/build/f84cc426cad74e9198680ff54aae13a8 : SUCCESS in 6m 09s\n- tooz-tox-py310-redis https://zuul.opendev.org/t/openstack/build/96669b1d6b8a46e5ab0f7491266b508a : SUCCESS in 4m 23s\n- tooz-tox-py38-sentinel https://zuul.opendev.org/t/openstack/build/0e2f2e11de03472ba086e6d4c60157fa : SUCCESS in 6m 10s\n- tooz-tox-py310-sentinel https://zuul.opendev.org/t/openstack/build/b0ad7694bda1434fbfc00a4c37381abe : SUCCESS in 4m 21s\n- tooz-tox-py38-memcached https://zuul.opendev.org/t/openstack/build/66b9042f4c3b4b529ea172270ddbebc3 : SUCCESS in 4m 46s\n- tooz-tox-py310-memcached https://zuul.opendev.org/t/openstack/build/d1cb5cc9382e4abeb0917275cb2bd9d5 : SUCCESS in 4m 04s\n- tooz-tox-py38-postgresql https://zuul.opendev.org/t/openstack/build/eab3e2b920f54eabbe6be0656130e7ee : SUCCESS in 3m 50s\n- tooz-tox-py310-postgresql https://zuul.opendev.org/t/openstack/build/ac4725f35bd0405eac47d759b84a5022 : SUCCESS in 4m 31s\n- tooz-tox-py38-mysql https://zuul.opendev.org/t/openstack/build/a3b0194bd56a4710b832afde76261b5c : SUCCESS in 5m 37s\n- tooz-tox-py310-mysql https://zuul.opendev.org/t/openstack/build/f89fcedf89314062a72881a2b24b56ee : SUCCESS in 4m 47s\n- tooz-tox-py38-consul https://zuul.opendev.org/t/openstack/build/aa10cc713451464f8227913f125c8899 : SUCCESS in 3m 59s\n- tooz-tox-py310-consul https://zuul.opendev.org/t/openstack/build/c54817cf65e54ba18f2c9c25f91cbad7 : SUCCESS in 4m 35s","accounts_in_message":[],"_revision_number":6},{"id":"dbfd0176e7b34fc0c82162963dcdd534e6be0f1a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-03 02:56:19.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"3ee78fda6bd0c161a8652679b639c7a8fd32ad1c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-03 02:57:54.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/60453fdf82c44c5489c91f88cf705737\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5fa79d58996b4c0fa60b7e5e90a3bf58 : SUCCESS in 59s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"2e666107fdaf3ea7de0688eee0432fc3981803be","revisions":{"00fb1c60a243d0b0660aeea9a42eead5c37f7013":{"kind":"REWORK","_number":1,"created":"2024-10-01 10:15:13.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/90/930990/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/90/930990/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/90/930990/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/90/930990/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/90/930990/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/90/930990/1"}}},"commit":{"parents":[{"commit":"b12f59e2ec50e1e5e5bf15ffbdaf4815c4db3174","subject":"zookeeper: fix argparse related to ssl","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/b12f59e2ec50e1e5e5bf15ffbdaf4815c4db3174"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-01 10:14:13.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-01 10:14:13.000000000","tz":540},"subject":"Fix unit tests in Python 3.12","message":"Fix unit tests in Python 3.12\n\nWe\u0027ve seen a few test failures caused by wraps\u003dTrue . Drop the argument\nfor now to unblock CI.\n\nChange-Id: Ic73b9876b6e3d8a3ac836974b24b7cb1bc7795f1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/00fb1c60a243d0b0660aeea9a42eead5c37f7013"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/00fb1c60a243d0b0660aeea9a42eead5c37f7013"}]},"branch":"refs/heads/master"},"d3bb1f3f8447f4d7838d6a3edca53e8bbca42be8":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2024-10-01 10:18:24.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/90/930990/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/90/930990/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/90/930990/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/90/930990/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/90/930990/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/90/930990/2"}}},"commit":{"parents":[{"commit":"b12f59e2ec50e1e5e5bf15ffbdaf4815c4db3174","subject":"zookeeper: fix argparse related to ssl","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/b12f59e2ec50e1e5e5bf15ffbdaf4815c4db3174"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-01 10:14:13.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-01 10:17:44.000000000","tz":540},"subject":"Fix wrong usage of \u0027wraps\u0027","message":"Fix wrong usage of \u0027wraps\u0027\n\nwraps should point what is being patched. Drop the incorrect usage\nwhich breaks these tests in Python 3.12.\n\nChange-Id: Ic73b9876b6e3d8a3ac836974b24b7cb1bc7795f1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/d3bb1f3f8447f4d7838d6a3edca53e8bbca42be8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/d3bb1f3f8447f4d7838d6a3edca53e8bbca42be8"}]},"branch":"refs/heads/master"},"42380b7f2be2bc367a2ea0bb5567c1bae8170a1b":{"kind":"REWORK","_number":3,"created":"2024-10-01 10:25:02.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/90/930990/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/90/930990/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/90/930990/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/90/930990/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/90/930990/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/90/930990/3"}}},"commit":{"parents":[{"commit":"b12f59e2ec50e1e5e5bf15ffbdaf4815c4db3174","subject":"zookeeper: fix argparse related to ssl","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/b12f59e2ec50e1e5e5bf15ffbdaf4815c4db3174"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-01 10:14:13.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-01 10:24:32.000000000","tz":540},"subject":"Fix wrong usage of \u0027wraps\u0027","message":"Fix wrong usage of \u0027wraps\u0027\n\nwraps should point what is being patched, instead of True/False.\n\nChange-Id: Ic73b9876b6e3d8a3ac836974b24b7cb1bc7795f1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/42380b7f2be2bc367a2ea0bb5567c1bae8170a1b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/42380b7f2be2bc367a2ea0bb5567c1bae8170a1b"}]},"branch":"refs/heads/master"},"58489d8c7901212fd29b6c5342a8c0d1eba91db0":{"kind":"REWORK","_number":4,"created":"2024-10-01 11:53:43.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/90/930990/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/90/930990/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/90/930990/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/90/930990/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/90/930990/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/90/930990/4"}}},"commit":{"parents":[{"commit":"b12f59e2ec50e1e5e5bf15ffbdaf4815c4db3174","subject":"zookeeper: fix argparse related to ssl","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/b12f59e2ec50e1e5e5bf15ffbdaf4815c4db3174"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-01 10:14:13.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-01 11:52:13.000000000","tz":540},"subject":"Fix wrong usage of \u0027wraps\u0027","message":"Fix wrong usage of \u0027wraps\u0027\n\nwraps should point what is being patched, instead of True/False.\n\nChange-Id: Ic73b9876b6e3d8a3ac836974b24b7cb1bc7795f1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/58489d8c7901212fd29b6c5342a8c0d1eba91db0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/58489d8c7901212fd29b6c5342a8c0d1eba91db0"}]},"branch":"refs/heads/master"},"bca377f09ea98fdc7129cd594bb0dbf8ebe6f1bd":{"kind":"REWORK","_number":5,"created":"2024-10-01 12:45:01.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/90/930990/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/90/930990/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/90/930990/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/90/930990/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/90/930990/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/90/930990/5"}}},"commit":{"parents":[{"commit":"b12f59e2ec50e1e5e5bf15ffbdaf4815c4db3174","subject":"zookeeper: fix argparse related to ssl","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/b12f59e2ec50e1e5e5bf15ffbdaf4815c4db3174"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-01 10:14:13.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-01 12:44:32.000000000","tz":540},"subject":"Fix wrong usage of \u0027wraps\u0027","message":"Fix wrong usage of \u0027wraps\u0027\n\nwraps should point what is being patched, instead of True/False.\n\nAlso fix the incomplete mocking in MySQL driver tests.\n\nChange-Id: Ic73b9876b6e3d8a3ac836974b24b7cb1bc7795f1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/bca377f09ea98fdc7129cd594bb0dbf8ebe6f1bd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/bca377f09ea98fdc7129cd594bb0dbf8ebe6f1bd"}]},"branch":"refs/heads/master"},"2e666107fdaf3ea7de0688eee0432fc3981803be":{"kind":"REWORK","_number":6,"created":"2024-10-01 13:07:39.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/90/930990/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/90/930990/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/90/930990/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/90/930990/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/90/930990/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/90/930990/6"}}},"commit":{"parents":[{"commit":"b12f59e2ec50e1e5e5bf15ffbdaf4815c4db3174","subject":"zookeeper: fix argparse related to ssl","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/b12f59e2ec50e1e5e5bf15ffbdaf4815c4db3174"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-01 10:14:13.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-01 13:07:29.000000000","tz":540},"subject":"Fix wrong usage of \u0027wraps\u0027","message":"Fix wrong usage of \u0027wraps\u0027\n\nwraps should point what is being patched, instead of True/False.\n\nAlso fix the incomplete mocking in MySQL driver tests.\n\nChange-Id: Ic73b9876b6e3d8a3ac836974b24b7cb1bc7795f1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/2e666107fdaf3ea7de0688eee0432fc3981803be"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/2e666107fdaf3ea7de0688eee0432fc3981803be"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
