)]}'
{"id":"openstack%2Ftooz~931053","triplet_id":"openstack%2Ftooz~master~I876aaf57561f4072955cee5db2aa4d66c19885cb","project":"openstack/tooz","branch":"master","hashtags":[],"change_id":"I876aaf57561f4072955cee5db2aa4d66c19885cb","subject":"Move upper functional jobs to Python 3.12","status":"ABANDONED","created":"2024-10-01 18:12:20.000000000","updated":"2024-10-02 01:14:02.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"c8b1ddea7b219675354b476143684a8134920e82","_number":931053,"virtual_id_number":931053,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2024-10-01 19:49:13.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-01 19:49:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8fc442c01ce206f590479fd31c37098f8700e23e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-01 18:12:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"60661148ce2a67c301a8b6dc9162960b9f324ef9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-01 18:13:33.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"a2f9db74cd781c0decdd2909a0f4657f18fb249a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-01 19:49:13.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5f6202d80eb042e998993fe5be027af4\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/f3c8d29908c1491cb8800b52b00ea85c : SUCCESS in 2m 25s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/71501d0e1bd945599a510504cbb9cd16 : SUCCESS in 1h 34m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e6de3a427d814ad09d1957c4d0f162ff : SUCCESS in 4m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b1f71c003b440f2a4db5ee8c26a57b4 : SUCCESS in 5m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ed64941e67a94566807210ef66a383ad : SUCCESS in 4m 20s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/84f31363deda4125b4066657dcd9558f : SUCCESS in 34m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ddae8f3589741058c99719464be64d2 : SUCCESS in 3m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5363bfddbef74dd38944ff7dfed17053 : SUCCESS in 3m 18s\n- tooz-tox-py39-etcd3gw https://zuul.opendev.org/t/openstack/build/1fcaeab63bec4f498b325a007b708bce : SUCCESS in 4m 51s\n- tooz-tox-py312-etcd3gw https://zuul.opendev.org/t/openstack/build/fbc33dad08364d1a836bd9f63cefc7bb : SUCCESS in 6m 09s\n- tooz-tox-py39-zookeeper https://zuul.opendev.org/t/openstack/build/29aede49bfba40bc92e4fde57f977d60 : SUCCESS in 4m 42s\n- tooz-tox-py312-zookeeper https://zuul.opendev.org/t/openstack/build/81255e2845604f459baa422b23652a21 : SUCCESS in 4m 01s\n- tooz-tox-py39-redis https://zuul.opendev.org/t/openstack/build/622653738e8f4e118121deb1c115a0d9 : SUCCESS in 4m 56s\n- tooz-tox-py312-redis https://zuul.opendev.org/t/openstack/build/ed01b82c6f164076a07641c214e78451 : SUCCESS in 5m 39s\n- tooz-tox-py39-sentinel https://zuul.opendev.org/t/openstack/build/1a4628880a7447f5bc1d333e84fe80bc : SUCCESS in 4m 40s\n- tooz-tox-py312-sentinel https://zuul.opendev.org/t/openstack/build/c7885c8ea44340eba5de1036e9f3f060 : SUCCESS in 7m 05s\n- tooz-tox-py39-memcached https://zuul.opendev.org/t/openstack/build/c8cb908f960743dc8b348fea7ced217f : SUCCESS in 4m 41s\n- tooz-tox-py312-memcached https://zuul.opendev.org/t/openstack/build/ab6c754dfcc14b489a2bc7d8e6846118 : SUCCESS in 10m 32s\n- tooz-tox-py39-postgresql https://zuul.opendev.org/t/openstack/build/6f775286ea8642eb9f47f5f4efdc33d1 : SUCCESS in 3m 38s\n- tooz-tox-py312-postgresql https://zuul.opendev.org/t/openstack/build/ea5726b62b1c44a6ac917575dd6972b5 : SUCCESS in 6m 12s\n- tooz-tox-py39-mysql https://zuul.opendev.org/t/openstack/build/fa174300844745d397616d9b89b61726 : SUCCESS in 4m 30s\n- tooz-tox-py312-mysql https://zuul.opendev.org/t/openstack/build/05a148befe404a67a5b2f38ee9e2fb91 : SUCCESS in 6m 56s\n- tooz-tox-py39-consul https://zuul.opendev.org/t/openstack/build/d9f00775282e4ed08a8fed0be2e8253e : SUCCESS in 3m 39s\n- tooz-tox-py312-consul https://zuul.opendev.org/t/openstack/build/6c91e9c1af2a406ba917fe74d1f22e7b : SUCCESS in 5m 20s","accounts_in_message":[],"_revision_number":2},{"id":"c8b1ddea7b219675354b476143684a8134920e82","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-02 01:14:02.000000000","message":"Abandoned\n\nThis is squashed into the parent patch.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"d516c2e15676728a7b8875377a872f1b925fcc17","revisions":{"defa99910fdb70f7bc8a93e0f19ad2232a404cf2":{"kind":"REWORK","_number":1,"created":"2024-10-01 18:12:20.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/53/931053/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/53/931053/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/53/931053/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/53/931053/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/53/931053/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/53/931053/1"}}},"commit":{"parents":[{"commit":"bd5e026e8bb98d9ada67b1ca21048ab5369a1ac9","subject":"Bump python versions used in tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/bd5e026e8bb98d9ada67b1ca21048ab5369a1ac9"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-01 18:11:51.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-01 18:12:11.000000000","tz":540},"subject":"Move upper functional jobs to Python 3.12","message":"Move upper functional jobs to Python 3.12\n\nChange-Id: I876aaf57561f4072955cee5db2aa4d66c19885cb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/defa99910fdb70f7bc8a93e0f19ad2232a404cf2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/defa99910fdb70f7bc8a93e0f19ad2232a404cf2"}]},"branch":"refs/heads/master"},"d516c2e15676728a7b8875377a872f1b925fcc17":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-10-01 18:13:33.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/53/931053/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/53/931053/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/53/931053/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/53/931053/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/53/931053/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/53/931053/2"}}},"commit":{"parents":[{"commit":"480553a4de4f19b749db7042656caadb94c51eb0","subject":"Bump python versions used in tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/480553a4de4f19b749db7042656caadb94c51eb0"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-01 18:11:51.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-01 18:13:33.000000000","tz":0},"subject":"Move upper functional jobs to Python 3.12","message":"Move upper functional jobs to Python 3.12\n\nChange-Id: I876aaf57561f4072955cee5db2aa4d66c19885cb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/d516c2e15676728a7b8875377a872f1b925fcc17"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/d516c2e15676728a7b8875377a872f1b925fcc17"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
