)]}'
{"id":"openstack%2Ftooz~932299","triplet_id":"openstack%2Ftooz~master~I2b35b69bb3064de688f2cbd049d54537bc39167d","project":"openstack/tooz","branch":"master","topic":"fix-tox-minversion","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-10-13 15:14:00.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"I2b35b69bb3064de688f2cbd049d54537bc39167d","subject":"Fix outdated tox minversion","status":"MERGED","created":"2024-10-13 12:40:25.000000000","updated":"2024-10-13 17:52:28.000000000","submitted":"2024-10-13 17:51:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"932299-fix-tox-minversion","meta_rev_id":"0718c055f0ac106c0b0200e0fe6a24617d4f9abd","_number":932299,"virtual_id_number":932299,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-10-13 17:51:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2024-10-13 12:46:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2024-10-13 12:46:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-13 14:24:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2ebfd087eb38ddfa6074b063629a0982625ec626","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-13 12:40:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"86ce2228fc2f786ae975f96b714347e1faffdf31","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-13 12:45:29.000000000","message":"Topic set to fix-tox-minversion","accounts_in_message":[],"_revision_number":1},{"id":"33130521a0bba895b04bdba062f5383b7d5f49a6","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-13 12:46:01.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2e72f76af2627b2d57f15bdb79a5f7d1a59a118b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-13 14:24:10.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/80e63da8491940e8a69891502d4e48c1\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cf78be56bbec440b9912c3f2e3b72e3f : SUCCESS in 1h 36m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/40bff5e0dd5a4306b183f66bdec48e12 : SUCCESS in 3m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f5e282eb1174ee9ad38c5d77368d579 : SUCCESS in 5m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/103ac5fc98054462b8726660ae010965 : SUCCESS in 3m 56s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c54159a07c4b4c8999dd74ab354c8948 : SUCCESS in 7m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a69d007af50743a2bafab22ccd00bac4 : SUCCESS in 2m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/58bb766a42934491a7efc507b3b08f7d : SUCCESS in 2m 55s\n- tooz-tox-py39-etcd3gw https://zuul.opendev.org/t/openstack/build/e546eec865e64729983fa0a54d123672 : SUCCESS in 4m 21s\n- tooz-tox-py312-etcd3gw https://zuul.opendev.org/t/openstack/build/91809fc848514e7dbd4ec574178efd2c : SUCCESS in 7m 19s\n- tooz-tox-py39-zookeeper https://zuul.opendev.org/t/openstack/build/8be2c8b2cd6f42ae83ce22fba772a6a7 : SUCCESS in 4m 12s\n- tooz-tox-py312-zookeeper https://zuul.opendev.org/t/openstack/build/62883aa800f84d85926c470c8157ea63 : FAILURE in 4m 53s\n- tooz-tox-py39-redis https://zuul.opendev.org/t/openstack/build/f4f19b26c0504e6bb7d1789fd38c1e11 : SUCCESS in 3m 17s\n- tooz-tox-py312-redis https://zuul.opendev.org/t/openstack/build/ebe1bb0a294f452eb16bc1fd9ad8995f : SUCCESS in 4m 45s\n- tooz-tox-py39-sentinel https://zuul.opendev.org/t/openstack/build/0336cc59d930465e8499d395fa082d21 : SUCCESS in 4m 00s\n- tooz-tox-py312-sentinel https://zuul.opendev.org/t/openstack/build/1c72b61189df49738c31c78298532ab8 : SUCCESS in 5m 43s\n- tooz-tox-py39-memcached https://zuul.opendev.org/t/openstack/build/6a8665e18721436b8a96c6bc76ed778a : SUCCESS in 4m 16s\n- tooz-tox-py312-memcached https://zuul.opendev.org/t/openstack/build/45dcc5cc8a2a4c09a789b0f61d9a4e21 : SUCCESS in 5m 31s\n- tooz-tox-py39-postgresql https://zuul.opendev.org/t/openstack/build/f0947d809baf42f296a7a228c6f91b4a : SUCCESS in 4m 23s\n- tooz-tox-py312-postgresql https://zuul.opendev.org/t/openstack/build/466cecb1ed4a4d419381c1b055c324ec : SUCCESS in 5m 08s\n- tooz-tox-py39-mysql https://zuul.opendev.org/t/openstack/build/a95f95a9e5834a8986730ae40e640663 : SUCCESS in 3m 54s\n- tooz-tox-py312-mysql https://zuul.opendev.org/t/openstack/build/95a5955030b140ffa73cb0d3b7e82c88 : SUCCESS in 5m 55s\n- tooz-tox-py39-consul https://zuul.opendev.org/t/openstack/build/ef7f403632464848bdca631ed1755092 : SUCCESS in 7m 45s\n- tooz-tox-py312-consul https://zuul.opendev.org/t/openstack/build/756610755d6444cd84552fd22f34445d : SUCCESS in 5m 51s","accounts_in_message":[],"_revision_number":1},{"id":"c302db68ef18dc72ccec73978a26532ce488fdb2","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-13 15:14:00.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"84cd5dc5625fd327e66fc3f7a5b32ffc0723c8ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-13 16:45:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0d74184ec8de4ef280dafa2f9f321721\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/15bbba74c40d4548b2910ae63aac3d11 : SUCCESS in 1h 29m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ed3aefe2b06b4d97a20b943e2d0a26a9 : SUCCESS in 5m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ec920f880df472c8b4b32343f4a6d89 : SUCCESS in 2m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0fdecfb79ab94f9d8b243a784e7e6191 : SUCCESS in 4m 07s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d91a7806776441f8958042d1732ee1e2 : SUCCESS in 3m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e5c11e5de1714167af684d60d20e53e9 : SUCCESS in 2m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/790f0e18ca494cfda5080fe6ab25cbcd : SUCCESS in 3m 11s\n- tooz-tox-py39-etcd3gw https://zuul.opendev.org/t/openstack/build/b56aa3370e9f4cf49a401393aebd8c67 : SUCCESS in 4m 24s\n- tooz-tox-py312-etcd3gw https://zuul.opendev.org/t/openstack/build/dc5f4d948c19427d8a0f31da82302cc4 : SUCCESS in 5m 07s\n- tooz-tox-py39-zookeeper https://zuul.opendev.org/t/openstack/build/0f6f016d4dd641fe88391fa4d90cf93c : SUCCESS in 4m 00s\n- tooz-tox-py312-zookeeper https://zuul.opendev.org/t/openstack/build/5bd3424a76a4460a8f74e5e09287ebe3 : SUCCESS in 5m 04s\n- tooz-tox-py39-redis https://zuul.opendev.org/t/openstack/build/69ba25288f804195b8f224ca9eb4913d : SUCCESS in 4m 11s\n- tooz-tox-py312-redis https://zuul.opendev.org/t/openstack/build/188f87fc94bb4d2391f464a0d5649148 : SUCCESS in 4m 36s\n- tooz-tox-py39-sentinel https://zuul.opendev.org/t/openstack/build/58b325f1574240ee9515f7ff8dd827e3 : SUCCESS in 4m 14s\n- tooz-tox-py312-sentinel https://zuul.opendev.org/t/openstack/build/b4890b1b378942c3bb03b434dabf918b : SUCCESS in 4m 33s\n- tooz-tox-py39-memcached https://zuul.opendev.org/t/openstack/build/fc8d2a8b972944fb9505a1f04e4dbd3c : SUCCESS in 3m 08s\n- tooz-tox-py312-memcached https://zuul.opendev.org/t/openstack/build/ffce5e1506f944bbae2a6419f97f030f : SUCCESS in 5m 28s\n- tooz-tox-py39-postgresql https://zuul.opendev.org/t/openstack/build/c16ddcd01dcf45f4af189ae54b273256 : SUCCESS in 4m 15s\n- tooz-tox-py312-postgresql https://zuul.opendev.org/t/openstack/build/7a870397938b48a589178029311900b9 : SUCCESS in 5m 04s\n- tooz-tox-py39-mysql https://zuul.opendev.org/t/openstack/build/e3e94ac73c5d493598ad251f22c997a2 : SUCCESS in 3m 57s\n- tooz-tox-py312-mysql https://zuul.opendev.org/t/openstack/build/d61c821281ac4ac9a5606bc0fd008544 : SUCCESS in 5m 22s\n- tooz-tox-py39-consul https://zuul.opendev.org/t/openstack/build/81f961dab850475b8324e506fa8d94dd : SUCCESS in 4m 23s\n- tooz-tox-py312-consul https://zuul.opendev.org/t/openstack/build/ca2dbd13d2574f5698c78e02123ca646 : SUCCESS in 7m 14s","accounts_in_message":[],"_revision_number":1},{"id":"96c5db7e139cf50b6ecad84e63f3325898ca7893","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-13 16:45:45.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"85df7c12063e6089fc7cf8ffa13a0b6a104c98e7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-13 17:51:31.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ccaed559ae954be1b04c65e22cb87f13\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d486cbc97fd745c08ff017a49364f303 : SUCCESS in 1h 05m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9a2671b5dd004614a4a99c9a704fa177 : SUCCESS in 3m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fd826b4d259f4130bb2ca01e269f089e : SUCCESS in 4m 23s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7e9e15fbf5fd46f994289a1976a2bef0 : SUCCESS in 5m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d3c87ec2b7f543748db197ccd09bff74 : SUCCESS in 3m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8e5d6726a48742a08a27b0cecdbea4d9 : SUCCESS in 2m 20s\n- tooz-tox-py39-etcd3gw https://zuul.opendev.org/t/openstack/build/af78863ff13344ab91875cd4c1a97319 : SUCCESS in 4m 41s\n- tooz-tox-py312-etcd3gw https://zuul.opendev.org/t/openstack/build/848884e156024db8b42e2e29853c1506 : SUCCESS in 5m 16s\n- tooz-tox-py39-zookeeper https://zuul.opendev.org/t/openstack/build/941fe4a006f440b0b0d420488a9a8208 : SUCCESS in 4m 26s\n- tooz-tox-py312-zookeeper https://zuul.opendev.org/t/openstack/build/6a1189f3d08a407abf2d5ee650e6b2b1 : SUCCESS in 3m 36s\n- tooz-tox-py39-redis https://zuul.opendev.org/t/openstack/build/f241b644bba14dfe81c67f09ce4000c6 : SUCCESS in 4m 11s\n- tooz-tox-py312-redis https://zuul.opendev.org/t/openstack/build/c09d3279bd50435aba401207b9d7b5e0 : SUCCESS in 5m 00s\n- tooz-tox-py39-sentinel https://zuul.opendev.org/t/openstack/build/c557995bbf71468e873bb5d800f51385 : SUCCESS in 3m 20s\n- tooz-tox-py312-sentinel https://zuul.opendev.org/t/openstack/build/2a46f12ae9584b5ba3962e3a4b062ebf : SUCCESS in 4m 58s\n- tooz-tox-py39-memcached https://zuul.opendev.org/t/openstack/build/fc5e0b6348b440f7b44f9ccd647afebd : SUCCESS in 4m 03s\n- tooz-tox-py312-memcached https://zuul.opendev.org/t/openstack/build/199efe0e8b9a4a52a234f0108fb84d41 : SUCCESS in 3m 29s\n- tooz-tox-py39-postgresql https://zuul.opendev.org/t/openstack/build/c165fd6adaab45dcbae4d0ae32d8fdc5 : SUCCESS in 4m 20s\n- tooz-tox-py312-postgresql https://zuul.opendev.org/t/openstack/build/d035d47dbd8b41969c5ce66d5dbf106f : SUCCESS in 5m 25s\n- tooz-tox-py39-mysql https://zuul.opendev.org/t/openstack/build/a35ec78edfb143659767e876a71372e7 : SUCCESS in 4m 15s\n- tooz-tox-py312-mysql https://zuul.opendev.org/t/openstack/build/0efdf4f5558d4a328ba5c270d84469d4 : SUCCESS in 5m 04s\n- tooz-tox-py39-consul https://zuul.opendev.org/t/openstack/build/303dd1ee92434760ab15038d4b3ad276 : SUCCESS in 4m 20s\n- tooz-tox-py312-consul https://zuul.opendev.org/t/openstack/build/ccb7dca479c74feb9bb835772b816dfe : SUCCESS in 4m 48s","accounts_in_message":[],"_revision_number":1},{"id":"2fbb537d196829fe8bf301184be5c9825ad1771f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-13 17:51:32.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"0718c055f0ac106c0b0200e0fe6a24617d4f9abd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-13 17:52:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ebda006deaca4434b519c3850e8b7446\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/74e2986de4ba43d784936ca3379ba2c6 : SUCCESS in 39s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7a2481f559f84a63b59259b6e73c4fc1 : SUCCESS in 38s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7404fbf8d92a673e1f2ac09f2c439d2827f49853","revisions":{"7404fbf8d92a673e1f2ac09f2c439d2827f49853":{"kind":"REWORK","_number":1,"created":"2024-10-13 12:40:25.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/99/932299/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/99/932299/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/99/932299/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/99/932299/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/99/932299/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/99/932299/1"}}},"commit":{"parents":[{"commit":"53b085b20e1c14ed4992d6afd5ee7bb30f905eb3","subject":"Merge \"mysql: Use oslo_utils implementation to parse boolean\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/53b085b20e1c14ed4992d6afd5ee7bb30f905eb3"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-13 12:40:16.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-13 12:40:16.000000000","tz":540},"subject":"Fix outdated tox minversion","message":"Fix outdated tox minversion\n\nThe allowlist_externals option is supported since 3.18.0 .\n\nChange-Id: I2b35b69bb3064de688f2cbd049d54537bc39167d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/7404fbf8d92a673e1f2ac09f2c439d2827f49853"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/7404fbf8d92a673e1f2ac09f2c439d2827f49853"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
