)]}'
{"id":"openstack%2Ftooz~947711","triplet_id":"openstack%2Ftooz~master~If7575b314fcc69597f3eae4c59ad202000001b30","project":"openstack/tooz","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-05-08 01:18:10.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"If7575b314fcc69597f3eae4c59ad202000001b30","subject":"Drop environment for protobuf library","status":"MERGED","created":"2025-04-20 15:01:43.000000000","updated":"2025-05-08 03:20:44.000000000","submitted":"2025-05-08 03:19:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"947711","meta_rev_id":"c2af714551f5cf6532af51ce503e96f99513a275","_number":947711,"virtual_id_number":947711,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-05-08 03:19:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2025-05-08 01:18:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2025-04-22 07:17:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2025-05-08 01:18:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-04-20 16:15:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-04-22 07:17:46.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"04e6f31948e1ae0e424366ee517b3e8463151409","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-04-20 15:01:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0f18aed30c33a671ff3ad83178c6a59bf7626b4e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-20 16:15:06.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/069f39f3edf244d9b9b999387262b302\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ede3494a09314067b2ffc73a3f5975f9 : SUCCESS in 1h 12m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6b3d6ba3730c4c51b9f80e64c9128099 : FAILURE in 4m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dad41b2e1c4e46219ce79195a0c7b9f3 : SUCCESS in 4m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/de331a1714de4dec8e2c1bad4f39dabd : FAILURE in 3m 42s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c246c1a03f074b288db0a54cda3b945d : SUCCESS in 2m 51s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d41d1f87aded404c861f980ea2ece315 : SUCCESS in 6m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b79a4aa0c1b94eb89a0318f4a5d097c8 : SUCCESS in 4m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/16233b539530402b8999a1c4769fc2b0 : SUCCESS in 3m 04s\n- tooz-tox-py39-etcd3gw https://zuul.opendev.org/t/openstack/build/25c52e8adb0b4afa923c2f9d684c2d37 : SUCCESS in 3m 16s\n- tooz-tox-py312-etcd3gw https://zuul.opendev.org/t/openstack/build/537cdb79c0c14524bfbff60aafe64669 : SUCCESS in 4m 48s\n- tooz-tox-py39-zookeeper https://zuul.opendev.org/t/openstack/build/3760713ddd9545ab8c98f353de2c0394 : SUCCESS in 3m 23s\n- tooz-tox-py312-zookeeper https://zuul.opendev.org/t/openstack/build/95681a02ea1e40ec8ec0eb4daab33bdf : FAILURE in 4m 38s\n- tooz-tox-py39-redis https://zuul.opendev.org/t/openstack/build/747bf3001da74d1984d0c92eaa3901a7 : SUCCESS in 2m 28s\n- tooz-tox-py312-redis https://zuul.opendev.org/t/openstack/build/7bfad00054614679a873d19e78c49a2a : SUCCESS in 4m 55s\n- tooz-tox-py39-sentinel https://zuul.opendev.org/t/openstack/build/a6778300aad4474d80471e2baec99d81 : SUCCESS in 7m 29s\n- tooz-tox-py312-sentinel https://zuul.opendev.org/t/openstack/build/29679fbdfaf44baa9ef1d19746bd6a82 : SUCCESS in 2m 42s\n- tooz-tox-py39-memcached https://zuul.opendev.org/t/openstack/build/e9924d3057804d9aa4839809892f8944 : SUCCESS in 3m 44s\n- tooz-tox-py312-memcached https://zuul.opendev.org/t/openstack/build/0d86a7ed99fa4ada950542e87f42d03c : SUCCESS in 4m 34s\n- tooz-tox-py39-postgresql https://zuul.opendev.org/t/openstack/build/fe5109ac5530413caa935fd6855398a1 : SUCCESS in 2m 51s\n- tooz-tox-py312-postgresql https://zuul.opendev.org/t/openstack/build/97b5b88684534defbe80978e88e3581a : SUCCESS in 2m 52s\n- tooz-tox-py39-mysql https://zuul.opendev.org/t/openstack/build/db004c552157491493b0513845213899 : SUCCESS in 3m 55s\n- tooz-tox-py312-mysql https://zuul.opendev.org/t/openstack/build/1bca8ad46b0e4ff2b86f99889d45efa9 : SUCCESS in 3m 26s\n- tooz-tox-py39-consul https://zuul.opendev.org/t/openstack/build/e68d20677bb548478714945204295eda : SUCCESS in 2m 25s\n- tooz-tox-py312-consul https://zuul.opendev.org/t/openstack/build/0c7bb3a5d6a94601ad123a4170c5aa9e : SUCCESS in 4m 55s","accounts_in_message":[],"_revision_number":1},{"id":"44da02f278c87513e4a189b30ccf6be758ec8d83","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-04-21 05:19:49.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"ce89ded97a6c3d5f806718430767f3e38060d880","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-21 07:10:07.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ef7c647162b24d8e91dce4f765463ccd\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cdaee2463a7749feab526dd7ee3f8f85 : SUCCESS in 1h 46m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/adbe8408106242c398d8f7a8f41d1151 : SUCCESS in 5m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b5223e04e10d419a86a1060ef02527df : SUCCESS in 4m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f1a098b5363145799af79e9fb4028014 : SUCCESS in 4m 26s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3de9d02a43664a5889c1ff9f5cce03b1 : SUCCESS in 5m 10s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4b0bfaa51bca4923a5ab63e0ea3f6e77 : SUCCESS in 7m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca28f8cd00ff48dbaa81a80b43e79097 : SUCCESS in 5m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/52371df389dd46b9a3b0d1c18c646518 : SUCCESS in 3m 41s\n- tooz-tox-py39-etcd3gw https://zuul.opendev.org/t/openstack/build/6470a4194e254b06b408db8bd55c4f58 : SUCCESS in 3m 59s\n- tooz-tox-py312-etcd3gw https://zuul.opendev.org/t/openstack/build/c40d8bcc6c374290befdf1950f99e555 : SUCCESS in 5m 53s\n- tooz-tox-py39-zookeeper https://zuul.opendev.org/t/openstack/build/ddb8c6435ebe43d197ad52848ad501f0 : SUCCESS in 4m 30s\n- tooz-tox-py312-zookeeper https://zuul.opendev.org/t/openstack/build/8dd25355d85c49f3b6ed39d51f769e2f : SUCCESS in 5m 37s\n- tooz-tox-py39-redis https://zuul.opendev.org/t/openstack/build/fb50370bc18b4ab0a0f2bdb8103109d5 : SUCCESS in 7m 08s\n- tooz-tox-py312-redis https://zuul.opendev.org/t/openstack/build/4cfe828dc8764af8953872da3a38f33e : SUCCESS in 4m 34s\n- tooz-tox-py39-sentinel https://zuul.opendev.org/t/openstack/build/b831c45346b64038a824f8f8206bada2 : SUCCESS in 5m 31s\n- tooz-tox-py312-sentinel https://zuul.opendev.org/t/openstack/build/2a32a5c663ed435f92312867404374af : SUCCESS in 4m 43s\n- tooz-tox-py39-memcached https://zuul.opendev.org/t/openstack/build/1caaa01da4224cd988dbe9f999c90737 : SUCCESS in 6m 10s\n- tooz-tox-py312-memcached https://zuul.opendev.org/t/openstack/build/7a8e04c4506f49bb8585bd9e451c1bfb : SUCCESS in 8m 46s\n- tooz-tox-py39-postgresql https://zuul.opendev.org/t/openstack/build/91cd1b81927549f68d3ee7503018dbbc : SUCCESS in 2m 33s\n- tooz-tox-py312-postgresql https://zuul.opendev.org/t/openstack/build/0663f122dfa44f7eadbdde092641a9c2 : SUCCESS in 5m 57s\n- tooz-tox-py39-mysql https://zuul.opendev.org/t/openstack/build/0b2e38f282c945feade7da3676b31312 : SUCCESS in 4m 07s\n- tooz-tox-py312-mysql https://zuul.opendev.org/t/openstack/build/d4e56a22c3374d36bab90e3ba0582148 : SUCCESS in 4m 26s\n- tooz-tox-py39-consul https://zuul.opendev.org/t/openstack/build/ca12e67045e4456082a3db86304cdc3c : SUCCESS in 5m 11s\n- tooz-tox-py312-consul https://zuul.opendev.org/t/openstack/build/820680080b4d475e94d7f51d0679b5dd : SUCCESS in 4m 48s","accounts_in_message":[],"_revision_number":2},{"id":"01e0b9763f4b0a7856b1fcfee665a2828bb503ca","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2025-04-22 07:17:46.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"496ac69de96f81661c5677b3e24ec03d198fd807","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-05-08 01:18:10.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"da3ccc6d61026d7ca34d9699777464efda8b6bad","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-08 01:18:30.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"108580952ba5eadaf7f05a862d62d5359727297d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-08 03:19:58.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/54c1941e99c14abdb4138b03dc10a6f0\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7c59c74a3a6845bcbee20cac375b5790 : SUCCESS in 1h 47m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/475f7482313745c08163f6435161856a : SUCCESS in 4m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/07f5fc20103d4d5b980294d059355634 : SUCCESS in 3m 50s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/cc67ee7af1df427aa18f0af69fa34af1 : SUCCESS in 4m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a663123e17764bacb80cc3d3ebcf6de1 : SUCCESS in 5m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a5bff54b378544dea219c513ae12cca6 : SUCCESS in 3m 24s\n- tooz-tox-py39-etcd3gw https://zuul.opendev.org/t/openstack/build/3b3b0c0a63bd432e85527a418e3567dd : SUCCESS in 6m 15s\n- tooz-tox-py312-etcd3gw https://zuul.opendev.org/t/openstack/build/5b2b26ba80d246819b5c913c8394b870 : SUCCESS in 5m 29s\n- tooz-tox-py39-zookeeper https://zuul.opendev.org/t/openstack/build/91c8d58236c94c1fbc76ead9bb8f0d59 : SUCCESS in 5m 06s\n- tooz-tox-py312-zookeeper https://zuul.opendev.org/t/openstack/build/06af5e85cd8b431e9c531bf68b501719 : SUCCESS in 4m 45s\n- tooz-tox-py39-redis https://zuul.opendev.org/t/openstack/build/9a9bd502f5ef430681006e5aa348ce8c : SUCCESS in 5m 50s\n- tooz-tox-py312-redis https://zuul.opendev.org/t/openstack/build/c51b0c509e0149d8a55cb44117dc8d42 : SUCCESS in 4m 41s\n- tooz-tox-py39-sentinel https://zuul.opendev.org/t/openstack/build/293fc9b44f9c4900b5e29e88a264ebf9 : SUCCESS in 5m 43s\n- tooz-tox-py312-sentinel https://zuul.opendev.org/t/openstack/build/d4df8f9b314a43f880e88e128187efd6 : SUCCESS in 4m 54s\n- tooz-tox-py39-memcached https://zuul.opendev.org/t/openstack/build/723aa539cf2e478c8436196c6708f256 : SUCCESS in 2m 20s\n- tooz-tox-py312-memcached https://zuul.opendev.org/t/openstack/build/610332903e8e4c5b9eebde7226a2bca4 : SUCCESS in 4m 17s\n- tooz-tox-py39-postgresql https://zuul.opendev.org/t/openstack/build/0e95c2317fad444b9c6fb1d075b00976 : SUCCESS in 4m 43s\n- tooz-tox-py312-postgresql https://zuul.opendev.org/t/openstack/build/766d3cfefbda4d1ea2cd13f26607756a : SUCCESS in 4m 32s\n- tooz-tox-py39-mysql https://zuul.opendev.org/t/openstack/build/d25271f0eb4f4470964127c750179dc3 : SUCCESS in 4m 16s\n- tooz-tox-py312-mysql https://zuul.opendev.org/t/openstack/build/12a853da1c5a4fe5b51df55ee1f5c7b0 : SUCCESS in 5m 13s\n- tooz-tox-py39-consul https://zuul.opendev.org/t/openstack/build/260e4f40948447cf9b62b21acd99727c : SUCCESS in 4m 25s\n- tooz-tox-py312-consul https://zuul.opendev.org/t/openstack/build/46098a89934c4a9eb483d7141ef0ef4d : SUCCESS in 4m 40s","accounts_in_message":[],"_revision_number":2},{"id":"ccef3f16e220112f76b29831991222c249da14bf","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-08 03:19:58.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"c2af714551f5cf6532af51ce503e96f99513a275","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-08 03:20:44.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/db63a923a08542ee8612264bdc4200b0\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc8917dc9ffd4d22a557cfcfdcd22ecf : SUCCESS in 33s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e376acd7cc214a0486eae263cdd230ed : SUCCESS in 35s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"44549e2412503928e97619127bcd753ed99b60bc","revisions":{"aa82ee010d8bfc9d02d816906821e99170e98611":{"kind":"REWORK","_number":1,"created":"2025-04-20 15:01:43.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/11/947711/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/11/947711/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/11/947711/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/11/947711/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/11/947711/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/11/947711/1"}}},"commit":{"parents":[{"commit":"ea73947c5a6d2861b4faa5750e17adf1973ec7dc","subject":"tox: Remove basepython","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/ea73947c5a6d2861b4faa5750e17adf1973ec7dc"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-04-20 15:00:52.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-04-20 15:00:52.000000000","tz":540},"subject":"Drop environment for protobuf library","message":"Drop environment for protobuf library\n\n... because the library was removed from the dependency when etcd3\ndriver was removed[1].\n\n[1] 6bc02cda5b452bbf4821621eafc031bd676f8a2f\n\nChange-Id: If7575b314fcc69597f3eae4c59ad202000001b30\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/aa82ee010d8bfc9d02d816906821e99170e98611"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/aa82ee010d8bfc9d02d816906821e99170e98611"}]},"branch":"refs/heads/master"},"44549e2412503928e97619127bcd753ed99b60bc":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-04-21 05:19:49.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/11/947711/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/11/947711/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/11/947711/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/11/947711/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/11/947711/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/11/947711/2"}}},"commit":{"parents":[{"commit":"0aa2dd596e6ee3463bd970172d2131fead89f850","subject":"Merge \"Remove remaining uses of oslo.util md5 wrapper\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/0aa2dd596e6ee3463bd970172d2131fead89f850"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-04-20 15:00:52.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-04-21 05:19:49.000000000","tz":0},"subject":"Drop environment for protobuf library","message":"Drop environment for protobuf library\n\n... because the library was removed from the dependency when etcd3\ndriver was removed[1].\n\n[1] 6bc02cda5b452bbf4821621eafc031bd676f8a2f\n\nChange-Id: If7575b314fcc69597f3eae4c59ad202000001b30\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/44549e2412503928e97619127bcd753ed99b60bc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/44549e2412503928e97619127bcd753ed99b60bc"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
