)]}'
{"id":"openstack%2Ftooz~955434","triplet_id":"openstack%2Ftooz~master~I851e17c739f8000885404cbf2bfaa7d241f0b247","project":"openstack/tooz","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-08-13 07:15:24.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"I851e17c739f8000885404cbf2bfaa7d241f0b247","subject":"Skip installation to speed up pep8 check","status":"MERGED","created":"2025-07-20 13:10:08.000000000","updated":"2025-08-13 09:11:09.000000000","submitted":"2025-08-13 09:09:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"955434","meta_rev_id":"4d4d53840217e3a5022b6109badb246b147a600a","_number":955434,"virtual_id_number":955434,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-08-13 09:09:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2025-07-24 22:05:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2025-08-13 07:15:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-07-20 15:37:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-07-24 22:05:41.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"8a10f5cd695a31b550a60e71b0354c7cc0d15ac9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-07-20 13:10:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"739892c0d0668141334899b18ad3f86dfeb29f3a","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-07-20 14:41:06.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1},{"id":"81cf304674a29c9471a08a7c67ae30f1eb891f8c","tag":"autogenerated:gerrit:restore","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-07-20 14:44:49.000000000","message":"Restored","accounts_in_message":[],"_revision_number":1},{"id":"a127058954031cf28453af5bd9e6113e18ea1e34","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-07-20 14:46:47.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c3f7507017546158c77254c205a0c169ee176a01","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-20 15:37:06.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/92adc320d78e46c29f79aca8b7fb8606\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9e958a8929434a33aadc513b1f4fd425 : SUCCESS in 49m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/359108196ab04809bde824b80532a533 : SUCCESS in 4m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2df2cc5feaf54d9ca190271af9f49b8b : SUCCESS in 2m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bb0b4378d8c942229c3694b9c2923c86 : SUCCESS in 4m 18s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c4053b21cc0244f8bcfc46f339929fb1 : SUCCESS in 2m 47s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b4fcd8f718df4b70bde162111468d149 : SUCCESS in 3m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ddaabf36b0474819a0b848937b9b2915 : SUCCESS in 4m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3f429e54f69f423f9960458533cbda74 : SUCCESS in 3m 37s\n- tooz-tox-py39-etcd3gw https://zuul.opendev.org/t/openstack/build/171050ef65de435f829558cafa8f1d51 : SUCCESS in 3m 00s\n- tooz-tox-py312-etcd3gw https://zuul.opendev.org/t/openstack/build/f928c055d0e3449ea865ee53ea6a04c2 : SUCCESS in 4m 36s\n- tooz-tox-py39-zookeeper https://zuul.opendev.org/t/openstack/build/44c2b27e698c48ca98dcc31422c10a60 : SUCCESS in 4m 23s\n- tooz-tox-py312-zookeeper https://zuul.opendev.org/t/openstack/build/fcba7d47548a4b11a260fbfa18075bca : SUCCESS in 4m 14s\n- tooz-tox-py39-redis https://zuul.opendev.org/t/openstack/build/6b70aa693c1d4b6a89e337b12c571b61 : SUCCESS in 2m 59s\n- tooz-tox-py312-redis https://zuul.opendev.org/t/openstack/build/e05a76bd4c5f40d1b4c69c10d3b4ea2b : SUCCESS in 3m 15s\n- tooz-tox-py39-sentinel https://zuul.opendev.org/t/openstack/build/5e32ad1eb6834df0b0facb3efdec12e5 : SUCCESS in 4m 23s\n- tooz-tox-py312-sentinel https://zuul.opendev.org/t/openstack/build/6c64a16435164f42b159d7b8ab5542f9 : SUCCESS in 4m 13s\n- tooz-tox-py39-memcached https://zuul.opendev.org/t/openstack/build/db4db4f7afdf4164ae116c467bb9a607 : SUCCESS in 3m 01s\n- tooz-tox-py312-memcached https://zuul.opendev.org/t/openstack/build/f1f36f89d9d74a7181f931491105da62 : SUCCESS in 3m 25s\n- tooz-tox-py39-postgresql https://zuul.opendev.org/t/openstack/build/fb4ddd09b68449f383346ccc5a87039e : SUCCESS in 3m 10s\n- tooz-tox-py312-postgresql https://zuul.opendev.org/t/openstack/build/c5d3de888d1c4693800930360da4399b : SUCCESS in 4m 17s\n- tooz-tox-py39-mysql https://zuul.opendev.org/t/openstack/build/4ab7ef5e02474af8bf1129eb9d75e5fd : SUCCESS in 2m 31s\n- tooz-tox-py312-mysql https://zuul.opendev.org/t/openstack/build/068a2186462443f9b5dd6889feab9bc9 : SUCCESS in 2m 41s\n- tooz-tox-py39-consul https://zuul.opendev.org/t/openstack/build/c070f05c0e6d442398d3b6cbf4ee216b : SUCCESS in 4m 29s\n- tooz-tox-py312-consul https://zuul.opendev.org/t/openstack/build/32e2df1f19e44c3db586f4583abdcfeb : SUCCESS in 4m 46s","accounts_in_message":[],"_revision_number":2},{"id":"f345b116269eb29c926b07d8c30b38f2431c11c7","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2025-07-24 22:05:41.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"37645de236d829ccb3780556d29721258a79c591","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-08-13 07:15:24.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"de639aae2e628130462c0fd184f687e88845508c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-13 07:15:50.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"1fdbf91e785cbd0399b7c0e8481609090042a03d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-13 09:09:30.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eb7d6e15e22e4480ae270c2780353efd\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/046a97b42ec3434e878936076b57ebbc : SUCCESS in 1h 40m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8835a316713446f08cb1e0d31f56095c : SUCCESS in 3m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/35a06b37015d4b6385f9b22f30ddf2c3 : SUCCESS in 3m 55s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/103e6a9163f74ff88dbf25d63ff879c4 : SUCCESS in 5m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/08c9402b81934b18ae90c29a363a5d3d : SUCCESS in 3m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7482da5a09524a4bad4d69fe011390c9 : SUCCESS in 2m 30s\n- tooz-tox-py39-etcd3gw https://zuul.opendev.org/t/openstack/build/7b1b0bd9a8c14b83b29b3ed3f9b3a273 : SUCCESS in 6m 53s\n- tooz-tox-py312-etcd3gw https://zuul.opendev.org/t/openstack/build/88a0edd9129d4281b5cb78297ebb72b1 : SUCCESS in 5m 06s\n- tooz-tox-py39-zookeeper https://zuul.opendev.org/t/openstack/build/39587f14cac84bd7a87fa22e8e0c4642 : SUCCESS in 3m 25s\n- tooz-tox-py312-zookeeper https://zuul.opendev.org/t/openstack/build/4b7cd98cdc674a8594a5a13980af841e : SUCCESS in 3m 46s\n- tooz-tox-py39-redis https://zuul.opendev.org/t/openstack/build/fc39478c41fe4305b2a39415aefc7858 : SUCCESS in 5m 01s\n- tooz-tox-py312-redis https://zuul.opendev.org/t/openstack/build/eca3a82a799c43919a9f6d29603a9c9d : SUCCESS in 3m 46s\n- tooz-tox-py39-sentinel https://zuul.opendev.org/t/openstack/build/9923ffbdbb3444c2a6efb14ab8bf3c89 : SUCCESS in 3m 37s\n- tooz-tox-py312-sentinel https://zuul.opendev.org/t/openstack/build/eba50feaa77b40b8a6ee60df61661e4f : SUCCESS in 7m 11s\n- tooz-tox-py39-memcached https://zuul.opendev.org/t/openstack/build/886b9ab0073f48c883d2cbb588d0e34c : SUCCESS in 4m 41s\n- tooz-tox-py312-memcached https://zuul.opendev.org/t/openstack/build/1208d9ce2bd64ac587c03029cdbcd8aa : SUCCESS in 6m 34s\n- tooz-tox-py39-postgresql https://zuul.opendev.org/t/openstack/build/d68946eaf92a46f7ba3772089cfafefb : SUCCESS in 3m 03s\n- tooz-tox-py312-postgresql https://zuul.opendev.org/t/openstack/build/924ccfbf9c9b435fa83438625cc18e04 : SUCCESS in 4m 36s\n- tooz-tox-py39-mysql https://zuul.opendev.org/t/openstack/build/c413cb7826a7439d8e6b3ea692962af4 : SUCCESS in 7m 00s\n- tooz-tox-py312-mysql https://zuul.opendev.org/t/openstack/build/5323f4c111ff40e3b5374fd0a514026b : SUCCESS in 7m 54s\n- tooz-tox-py39-consul https://zuul.opendev.org/t/openstack/build/03666eafa9824596a9d84dd0de007228 : SUCCESS in 7m 32s\n- tooz-tox-py312-consul https://zuul.opendev.org/t/openstack/build/10299505223f44c9ad66170d57bd0e83 : SUCCESS in 8m 02s","accounts_in_message":[],"_revision_number":2},{"id":"b7ba764c324054d0e02c9ad6329891ee43df1669","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-13 09:09:30.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"4d4d53840217e3a5022b6109badb246b147a600a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-13 09:11:09.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3f6fb7be58824c6ebb562c02e4844090\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b8d8b4e4fad4453957fa32ab2cd7a5e : SUCCESS in 37s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1eb2a941d7ac4b51827cb9e83b8c67ee : SUCCESS in 38s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ca30baef770893985f909483cb916f1d6dc5d47e","revisions":{"1a99ccb8dac1ee043b5338535ea54258e2d93de3":{"kind":"REWORK","_number":1,"created":"2025-07-20 13:10:08.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/34/955434/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/34/955434/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/34/955434/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/34/955434/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/34/955434/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/34/955434/1"}}},"commit":{"parents":[{"commit":"8df645f382e348f83049d54fb72ed08ae0496cfc","subject":"Merge \"Add TLS support for Memcached driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/8df645f382e348f83049d54fb72ed08ae0496cfc"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-07-20 13:09:29.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-07-20 13:09:47.000000000","tz":540},"subject":"Skip installation to speed up pep8 check","message":"Skip installation to speed up pep8 check\n\nChange-Id: I851e17c739f8000885404cbf2bfaa7d241f0b247\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/1a99ccb8dac1ee043b5338535ea54258e2d93de3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/1a99ccb8dac1ee043b5338535ea54258e2d93de3"}]},"branch":"refs/heads/master"},"ca30baef770893985f909483cb916f1d6dc5d47e":{"kind":"REWORK","_number":2,"created":"2025-07-20 14:46:47.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/34/955434/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/34/955434/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/34/955434/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/34/955434/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/34/955434/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/34/955434/2"}}},"commit":{"parents":[{"commit":"8df645f382e348f83049d54fb72ed08ae0496cfc","subject":"Merge \"Add TLS support for Memcached driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/8df645f382e348f83049d54fb72ed08ae0496cfc"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-07-20 13:09:29.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-07-20 14:46:01.000000000","tz":540},"subject":"Skip installation to speed up pep8 check","message":"Skip installation to speed up pep8 check\n\n... because most of the checks do not require tooz installed.\n\nNote that changelog is removed from doc, to be replaced by release\nnotes published at https://docs.openstack.org/releasenotes/tooz/ .\n\nChange-Id: I851e17c739f8000885404cbf2bfaa7d241f0b247\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/ca30baef770893985f909483cb916f1d6dc5d47e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/ca30baef770893985f909483cb916f1d6dc5d47e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
