)]}'
{"id":"openstack%2Ftooz~961059","triplet_id":"openstack%2Ftooz~master~Icb9ce2966e20eb189a39e51fe43b282f3190a0a6","project":"openstack/tooz","branch":"master","topic":"py313","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2025-12-05 16:09:29.000000000","reason":"\u003cGERRIT_ACCOUNT_15334\u003e replied on the change","reason_account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-12-05 18:26:15.000000000","reason":"Change was submitted"},"16137":{"account":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"last_update":"2025-12-05 18:26:15.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Icb9ce2966e20eb189a39e51fe43b282f3190a0a6","subject":"Move upper functional job from 3.12 to 3.13","status":"MERGED","created":"2025-09-14 04:01:39.000000000","updated":"2025-12-05 18:28:00.000000000","submitted":"2025-12-05 18:26:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"961059-py313","meta_rev_id":"e52da610da79460c8e225596e7ec5a65a7e13d05","_number":961059,"virtual_id_number":961059,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-12-05 18:26:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2025-12-05 16:09:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2025-12-05 16:09:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-09-14 05:44:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-09-15 06:00:59.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2025-09-15 10:36:09.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"}],"messages":[{"id":"b1cc425a31b6798e48193302d5502fcabc55d952","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-09-14 04:01:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e268a8e39237aed1e60bd368b494af0f2b61b5c2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-14 05:44:54.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4685ec1af69d4366a895827b19f65d1f\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f8cd199537a9499d90adad9b13c894ad : SUCCESS in 1h 22m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/29efc138ee02416cbde7326194f62fd4 : SUCCESS in 6m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/82d2fc3874974bf293c36d0ca706f598 : SUCCESS in 4m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9c178ee982354e04b5673489f0cd26fe : SUCCESS in 6m 15s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/84c29f1bd8d740b6aef63b2c3b4d7f6a : SUCCESS in 5m 47s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/37821cb0ac164972abec552795098f83 : SUCCESS in 7m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/18bb810e1ecd49848d1a8de2feb0d5f3 : SUCCESS in 5m 16s\n- tooz-tox-py310-etcd3gw https://zuul.opendev.org/t/openstack/build/c02540a6d4eb4b02b7ba221fff2b0f62 : FAILURE in 4m 46s\n- tooz-tox-py313-etcd3gw https://zuul.opendev.org/t/openstack/build/cd18d7ea3e7f4045bf2a5cdd18eedac0 : SUCCESS in 7m 07s\n- tooz-tox-py310-zookeeper https://zuul.opendev.org/t/openstack/build/25cb7576536047728f4c34a2826ee3ff : SUCCESS in 4m 16s\n- tooz-tox-py313-zookeeper https://zuul.opendev.org/t/openstack/build/2add9a51095e4e2f8a9851d564b3093c : SUCCESS in 7m 00s\n- tooz-tox-py310-redis https://zuul.opendev.org/t/openstack/build/99679ce59cde497abc2b075af0591290 : SUCCESS in 4m 26s\n- tooz-tox-py313-redis https://zuul.opendev.org/t/openstack/build/fbbdd0a458bd42a0ba77b4e6870517bf : SUCCESS in 7m 52s\n- tooz-tox-py310-sentinel https://zuul.opendev.org/t/openstack/build/ea69480d33cf4d2a969484a6cd765eee : SUCCESS in 6m 27s\n- tooz-tox-py313-sentinel https://zuul.opendev.org/t/openstack/build/3b5aec6c35e04e189d4b67779f74f728 : SUCCESS in 10m 03s\n- tooz-tox-py310-memcached https://zuul.opendev.org/t/openstack/build/7be31e1afc064130ba7e95179d169065 : SUCCESS in 4m 38s\n- tooz-tox-py313-memcached https://zuul.opendev.org/t/openstack/build/06761384cf5d4e7f90085bc777cfe0a0 : SUCCESS in 7m 42s\n- tooz-tox-py310-postgresql https://zuul.opendev.org/t/openstack/build/20a94f59811d4995a5b50566a9981f90 : SUCCESS in 4m 32s\n- tooz-tox-py313-postgresql https://zuul.opendev.org/t/openstack/build/ad2edda8934c47039aff2720822936b9 : SUCCESS in 6m 39s\n- tooz-tox-py310-mysql https://zuul.opendev.org/t/openstack/build/9bacf351c3d7425fbb13a41baf51577f : SUCCESS in 4m 35s\n- tooz-tox-py313-mysql https://zuul.opendev.org/t/openstack/build/0d5f78c8729d45f8a39d74a7b967203c : SUCCESS in 7m 55s\n- tooz-tox-py310-consul https://zuul.opendev.org/t/openstack/build/ab80ff07a1604a0d804439f46a30fb36 : SUCCESS in 4m 20s\n- tooz-tox-py313-consul https://zuul.opendev.org/t/openstack/build/26f241f261a84885ba60591feeea4c4b : SUCCESS in 8m 00s","accounts_in_message":[],"_revision_number":1},{"id":"a126b069feb08780d799755a67cc0be62af77f4d","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-09-14 05:50:40.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8f0048eb8250b5fc927979a4973b5ae893737e45","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-14 07:32:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/06f6984c62854ad9b546a51892e438cf\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/15b8590060634a619c28407bc8a6e6b7 : SUCCESS in 1h 36m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cdc72f11cb8149c28925d50afa75780a : SUCCESS in 6m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7e9c3d19bfde4ef9bc102f0023282d03 : SUCCESS in 4m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3ad210a7f772472bb5efd8c2c9f2e700 : SUCCESS in 6m 33s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/03980f2117a6487f9dfc09b631c1ef94 : SUCCESS in 5m 51s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0097471b2d0941a6aaec60339f6bf337 : SUCCESS in 8m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/253c9e2508fb4f5599368be7e2d74308 : SUCCESS in 7m 02s\n- tooz-tox-py310-etcd3gw https://zuul.opendev.org/t/openstack/build/8c8f118d09624b4eb154cda45bd16b6c : SUCCESS in 6m 41s\n- tooz-tox-py313-etcd3gw https://zuul.opendev.org/t/openstack/build/27fbf6fdc19c42d9a3752229faee1ba2 : SUCCESS in 8m 00s\n- tooz-tox-py310-zookeeper https://zuul.opendev.org/t/openstack/build/149128f9ea944314a23e140b71be3ec9 : SUCCESS in 4m 41s\n- tooz-tox-py313-zookeeper https://zuul.opendev.org/t/openstack/build/a815be42e00b4e0eb521af1c9f8e2c53 : SUCCESS in 7m 53s\n- tooz-tox-py310-redis https://zuul.opendev.org/t/openstack/build/090dcf65e30b4f60b3752fe19177a2f0 : SUCCESS in 4m 34s\n- tooz-tox-py313-redis https://zuul.opendev.org/t/openstack/build/4726fa47cfa14dcf8e9beab103cc0cf6 : SUCCESS in 9m 45s\n- tooz-tox-py310-sentinel https://zuul.opendev.org/t/openstack/build/248a4e1fd4f84cfbad9a4a5cd5d3b174 : SUCCESS in 4m 41s\n- tooz-tox-py313-sentinel https://zuul.opendev.org/t/openstack/build/ac750591623c4a248292c4278eaaf925 : SUCCESS in 8m 02s\n- tooz-tox-py310-memcached https://zuul.opendev.org/t/openstack/build/3627bbce11b843c7b0b24ce704c8da94 : SUCCESS in 5m 42s\n- tooz-tox-py313-memcached https://zuul.opendev.org/t/openstack/build/5eba48728a084a7e8fec4908833c0fa5 : SUCCESS in 8m 02s\n- tooz-tox-py310-postgresql https://zuul.opendev.org/t/openstack/build/9861b413d1a34287a2cf4209c465009b : SUCCESS in 4m 23s\n- tooz-tox-py313-postgresql https://zuul.opendev.org/t/openstack/build/623e48ea57004a64a78f78160d0ed7aa : SUCCESS in 7m 42s\n- tooz-tox-py310-mysql https://zuul.opendev.org/t/openstack/build/616bd8e6acd54b70886562bc9f6066a6 : SUCCESS in 4m 28s\n- tooz-tox-py313-mysql https://zuul.opendev.org/t/openstack/build/d297b8f882744381a609d4108466de45 : SUCCESS in 7m 53s\n- tooz-tox-py310-consul https://zuul.opendev.org/t/openstack/build/22968397365a4bbfb72dc19efb4cc7a2 : SUCCESS in 4m 37s\n- tooz-tox-py313-consul https://zuul.opendev.org/t/openstack/build/89b4d32d60384e8db77d9ee9900dc543 : SUCCESS in 6m 47s","accounts_in_message":[],"_revision_number":1},{"id":"2f1352af4ad2978b97046dd6b518485aea0f9c5f","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2025-09-15 06:00:59.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"7c9e247c3b23ed6f4a5518093806aaa622d7321e","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-09-15 10:36:09.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e9f3708ae08c04f074899a850acb6db6fbf26438","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-15 10:36:11.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/321f07122147448d9d2a8e41aa2c509a\n\n\nWarning:\n  Change 960639,1 in project openstack/openstack-zuul-jobs does not share a change queue with 961059 in project openstack/tooz","accounts_in_message":[],"_revision_number":1},{"id":"7ac5eae99dde7866e9b2c1a5fcc58b1c79e7894b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-12-05 15:56:47.000000000","message":"Patch Set 2: Commit message was updated.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-2\n* Workflow+1\n","accounts_in_message":[],"_revision_number":2},{"id":"3a81b634e092bc14ec44d0bdd916180665a3c47e","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-12-05 16:09:29.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"7efd6709ab1481ced0cb2d602907f8eff3e99a26","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-05 16:53:27.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/81ea4af84e2140bfbb50541863d42e46\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/eea2a419cb02430d9fd62fdaf4fb003d : SUCCESS in 55m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/25c2e3a4eec64ab5957c3cf3064c05f6 : SUCCESS in 4m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/300efa3f09674b6a9432de3e6b2e8a03 : SUCCESS in 4m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e3ec2ddf734c441397a590132978d064 : SUCCESS in 4m 59s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5cc2ac3678cc48599e1479b7cf03d913 : SUCCESS in 8m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b165f3792cb450380808c3839179df1 : SUCCESS in 3m 23s\n- tooz-tox-py310-etcd3gw https://zuul.opendev.org/t/openstack/build/5f448b830bac462683813fa44e9e42e0 : SUCCESS in 6m 37s\n- tooz-tox-py313-etcd3gw https://zuul.opendev.org/t/openstack/build/cdbeba54f44940c0a9fc6a62beeaaabb : SUCCESS in 5m 47s\n- tooz-tox-py310-zookeeper https://zuul.opendev.org/t/openstack/build/6e61bba64c87451199d70948325cbc11 : SUCCESS in 6m 12s\n- tooz-tox-py313-zookeeper https://zuul.opendev.org/t/openstack/build/624df04fafd24200b255ea78fc9b70d8 : SUCCESS in 8m 20s\n- tooz-tox-py310-redis https://zuul.opendev.org/t/openstack/build/8448c3c2849a4022b67c53b0289d5de8 : SUCCESS in 4m 19s\n- tooz-tox-py313-redis https://zuul.opendev.org/t/openstack/build/e7597641b4db4173b2be72282a58777e : SUCCESS in 5m 00s\n- tooz-tox-py310-sentinel https://zuul.opendev.org/t/openstack/build/6f704427ad5d400896b62da1fe7f92be : SUCCESS in 4m 31s\n- tooz-tox-py313-sentinel https://zuul.opendev.org/t/openstack/build/e5d56754aeed498b948e5fedeb71696e : SUCCESS in 5m 26s\n- tooz-tox-py310-memcached https://zuul.opendev.org/t/openstack/build/5686f91a67af4102b83aba24328ef6b4 : SUCCESS in 4m 42s\n- tooz-tox-py313-memcached https://zuul.opendev.org/t/openstack/build/6141ddf6932442fa8150dc996d56dc62 : SUCCESS in 5m 23s\n- tooz-tox-py310-postgresql https://zuul.opendev.org/t/openstack/build/ba6d6adb29f14eb59f33204eb68ccd83 : SUCCESS in 5m 26s\n- tooz-tox-py313-postgresql https://zuul.opendev.org/t/openstack/build/2327808da74b4ea2aa2d73a8e248199d : SUCCESS in 10m 32s\n- tooz-tox-py310-mysql https://zuul.opendev.org/t/openstack/build/3ff2e18274be450dada645d5f34808f7 : SUCCESS in 4m 27s\n- tooz-tox-py313-mysql https://zuul.opendev.org/t/openstack/build/d0e48b0b16ec4765a308dd2a8168bd8a : SUCCESS in 10m 22s\n- tooz-tox-py310-consul https://zuul.opendev.org/t/openstack/build/77133138192c41b592013850e5280673 : SUCCESS in 6m 24s\n- tooz-tox-py313-consul https://zuul.opendev.org/t/openstack/build/98dfe6e048244b398363231d3437c606 : SUCCESS in 11m 48s","accounts_in_message":[],"_revision_number":2},{"id":"eda6437fded524200e58fd10501d03b4dbd17411","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-05 16:54:07.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"d8fd01f080f3c8b2ffeb625eda2e1791f23c582a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-05 18:26:14.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4fb12902b7454d01a51fbea2dd3d35ba\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8422e0202d324aadae25232a9eadd7e3 : SUCCESS in 1h 29m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1da99cb5e4154bf0a85e2bbb279358a0 : SUCCESS in 3m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a8048ffeba914ed2b0cbe48888881a14 : SUCCESS in 3m 59s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/be4126cc103046859798ad98d30a5615 : SUCCESS in 8m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6376f2c5e0e042b5a9bc3093cc2402e3 : SUCCESS in 5m 48s\n- tooz-tox-py310-etcd3gw https://zuul.opendev.org/t/openstack/build/7072d4b71107493996f07d59bf4f63b8 : SUCCESS in 4m 36s\n- tooz-tox-py313-etcd3gw https://zuul.opendev.org/t/openstack/build/c4a463b6f5d64f789c176ebae733a7d8 : SUCCESS in 9m 09s\n- tooz-tox-py310-zookeeper https://zuul.opendev.org/t/openstack/build/10d70ea3c5504f388c5c2dea140dbbbb : SUCCESS in 4m 39s\n- tooz-tox-py313-zookeeper https://zuul.opendev.org/t/openstack/build/2702eada0f4049c382ea19a6dd253204 : SUCCESS in 4m 52s\n- tooz-tox-py310-redis https://zuul.opendev.org/t/openstack/build/7b4611109d604aa49e86b9d0044b1155 : SUCCESS in 4m 54s\n- tooz-tox-py313-redis https://zuul.opendev.org/t/openstack/build/f029c88b2ddd446aab8a667b15b84b77 : SUCCESS in 5m 15s\n- tooz-tox-py310-sentinel https://zuul.opendev.org/t/openstack/build/7d131da4a8154ea1b8de9f896b828ad5 : SUCCESS in 4m 15s\n- tooz-tox-py313-sentinel https://zuul.opendev.org/t/openstack/build/934872969e2c48e09be4277e46ba740d : SUCCESS in 8m 11s\n- tooz-tox-py310-memcached https://zuul.opendev.org/t/openstack/build/22fd6575710d40bb90ed0883405722a4 : SUCCESS in 5m 07s\n- tooz-tox-py313-memcached https://zuul.opendev.org/t/openstack/build/b59bd0220edf427a8f2a36a201caacc8 : SUCCESS in 7m 42s\n- tooz-tox-py310-postgresql https://zuul.opendev.org/t/openstack/build/1c72c3f327d34fb9b24cd779d56190e4 : SUCCESS in 4m 50s\n- tooz-tox-py313-postgresql https://zuul.opendev.org/t/openstack/build/a7d5a7b21d2b48bc9e2ca03b10f363fa : SUCCESS in 5m 38s\n- tooz-tox-py310-mysql https://zuul.opendev.org/t/openstack/build/4098363f805c4e5d907714448e0abd9e : SUCCESS in 8m 16s\n- tooz-tox-py313-mysql https://zuul.opendev.org/t/openstack/build/ac9ebfd15bf942a29451e21d27eb1d80 : SUCCESS in 8m 05s\n- tooz-tox-py310-consul https://zuul.opendev.org/t/openstack/build/cfe9eb685b5f46a9bf07788b9d237e8e : SUCCESS in 4m 17s\n- tooz-tox-py313-consul https://zuul.opendev.org/t/openstack/build/9b0a0128b3cc4f7297e94a5ed0fa60ac : SUCCESS in 9m 54s","accounts_in_message":[],"_revision_number":2},{"id":"4592133ac0321d98f760086b3ca659a86f64acf7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-05 18:26:15.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"e52da610da79460c8e225596e7ec5a65a7e13d05","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-05 18:28:00.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/870beb067bbb422d8a4820440370bdfa\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3582a7f9ee5f4a1fb1cdcdf0dd92a1c7 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"377e354ce0ff04d0e63e8f667f82d9973e5a9084","revisions":{"74000f37e046c109d3d103bc1d4200890292d2b2":{"kind":"REWORK","_number":1,"created":"2025-09-14 04:01:39.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/59/961059/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/59/961059/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/59/961059/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/59/961059/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/59/961059/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/59/961059/1"}}},"commit":{"parents":[{"commit":"93c9aa91e3f50d305c00912972b32ad9cbe5872e","subject":"pre-commit: Bump dependencies","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/93c9aa91e3f50d305c00912972b32ad9cbe5872e"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-09-14 04:00:33.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-09-14 04:01:25.000000000","tz":540},"subject":"Move upper functional job from 3.12 to 3.13","message":"Move upper functional job from 3.12 to 3.13\n\n... because 3.13 will be the latest python version supported in\n2026.1 release.\n\nDepends-on: https://review.opendev.org/960639\nChange-Id: Icb9ce2966e20eb189a39e51fe43b282f3190a0a6\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/74000f37e046c109d3d103bc1d4200890292d2b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/74000f37e046c109d3d103bc1d4200890292d2b2"}]},"branch":"refs/heads/master"},"377e354ce0ff04d0e63e8f667f82d9973e5a9084":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2025-12-05 15:56:47.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/59/961059/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/59/961059/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/59/961059/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/59/961059/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/59/961059/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/59/961059/2"}}},"commit":{"parents":[{"commit":"93c9aa91e3f50d305c00912972b32ad9cbe5872e","subject":"pre-commit: Bump dependencies","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/93c9aa91e3f50d305c00912972b32ad9cbe5872e"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-09-14 04:00:33.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-12-05 15:56:47.000000000","tz":0},"subject":"Move upper functional job from 3.12 to 3.13","message":"Move upper functional job from 3.12 to 3.13\n\n... because 3.13 will be the latest python version supported in\n2026.1 release.\n\nChange-Id: Icb9ce2966e20eb189a39e51fe43b282f3190a0a6\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/377e354ce0ff04d0e63e8f667f82d9973e5a9084"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/377e354ce0ff04d0e63e8f667f82d9973e5a9084"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
