)]}'
{"id":"openstack%2Ftooz~963089","triplet_id":"openstack%2Ftooz~master~Iae501ca85ea947c86cb10c2631ea265ef97741ed","project":"openstack/tooz","branch":"master","topic":"py310-plus","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-10-07 00:50:47.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"Iae501ca85ea947c86cb10c2631ea265ef97741ed","subject":"Bump pyupgrade target to 3.10+","status":"MERGED","created":"2025-10-05 10:45:54.000000000","updated":"2025-10-07 04:17:25.000000000","submitted":"2025-10-07 04:16:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"963089-py310-plus","meta_rev_id":"228379bd56d04886796cd9ae952a8456ee53e926","_number":963089,"virtual_id_number":963089,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-10-07 04:16:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2025-10-05 12:43:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2025-10-06 09:15:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2025-10-06 09:15:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-10-05 12:13:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-10-05 12:43:43.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2025-10-06 09:15:07.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"}],"messages":[{"id":"2fb2b7cf265f1d1a66f87ae6cac573ee16b32c74","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-10-05 10:45:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a8b127f3e6f544c714f5eaa8b546e1ef2bb196e1","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-10-05 10:51:36.000000000","message":"Topic set to py310-plus","accounts_in_message":[],"_revision_number":1},{"id":"9ecc83887c7678e10537efd24b19d6a2048e844a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-05 12:13:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9fad09416491421ab1886fc9450bc556\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9f932a77cf334312b1a2d8355a331bbb : SUCCESS in 1h 26m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e1c3359fc6ec48728e2eed84e494c557 : SUCCESS in 6m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a9c349126ae144218f82473513cec182 : SUCCESS in 3m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e3cc6dda5f574cacabb1fa2901b35cf9 : SUCCESS in 7m 01s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0d3dcc81e8234101bc9dcbfe7cb09898 : SUCCESS in 5m 24s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e1dce6805e81407386e9994bfd0ede27 : SUCCESS in 7m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/94969bd977564cbc96c68df3bb0fc351 : SUCCESS in 8m 58s\n- tooz-tox-py310-etcd3gw https://zuul.opendev.org/t/openstack/build/92cf832b96e7454ba5a118c913b36eef : SUCCESS in 4m 42s\n- tooz-tox-py312-etcd3gw https://zuul.opendev.org/t/openstack/build/bdea1329b8644e09841abf1193e977b7 : SUCCESS in 5m 30s\n- tooz-tox-py310-zookeeper https://zuul.opendev.org/t/openstack/build/c7a3bb6ee4624446b069aca93b2df3a6 : SUCCESS in 3m 34s\n- tooz-tox-py312-zookeeper https://zuul.opendev.org/t/openstack/build/ccba9b7a6c8f4fbbb4f4ca3f334b2025 : SUCCESS in 9m 15s\n- tooz-tox-py310-redis https://zuul.opendev.org/t/openstack/build/1a5f9a3f40084285b2dbd43405f62f9b : SUCCESS in 4m 38s\n- tooz-tox-py312-redis https://zuul.opendev.org/t/openstack/build/4c0462943d104abfa28c77722fcb46c9 : SUCCESS in 9m 18s\n- tooz-tox-py310-sentinel https://zuul.opendev.org/t/openstack/build/77becb87e1094e8b83e74fd81e3dc5bc : SUCCESS in 4m 10s\n- tooz-tox-py312-sentinel https://zuul.opendev.org/t/openstack/build/75ef1cc8932044919f1572e9051756c0 : SUCCESS in 5m 02s\n- tooz-tox-py310-memcached https://zuul.opendev.org/t/openstack/build/681f4fa900664229a0f83d2945c12aaa : SUCCESS in 4m 19s\n- tooz-tox-py312-memcached https://zuul.opendev.org/t/openstack/build/ea3cf533b04b485ea786b4f703845fcc : SUCCESS in 9m 28s\n- tooz-tox-py310-postgresql https://zuul.opendev.org/t/openstack/build/fd0aa6d5bdf44407883ae93a4c900299 : SUCCESS in 3m 20s\n- tooz-tox-py312-postgresql https://zuul.opendev.org/t/openstack/build/cf221de51bca49ddafb1ec3a8774a401 : SUCCESS in 4m 49s\n- tooz-tox-py310-mysql https://zuul.opendev.org/t/openstack/build/653d7161996c4ceeb6722dbda46fddbb : SUCCESS in 4m 08s\n- tooz-tox-py312-mysql https://zuul.opendev.org/t/openstack/build/2c0d5d64405a4ef5b3fe81d6f65f07bf : SUCCESS in 3m 22s\n- tooz-tox-py310-consul https://zuul.opendev.org/t/openstack/build/220c01975a4e46e1be74a432a9ecfa88 : SUCCESS in 4m 36s\n- tooz-tox-py312-consul https://zuul.opendev.org/t/openstack/build/424a3233459b411bb19aabab25053c6e : SUCCESS in 5m 29s","accounts_in_message":[],"_revision_number":1},{"id":"959ddc6297c9dbbbb4634622ef083f1c4594b65b","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2025-10-05 12:43:43.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"6e9eb876de1ce0454b1ace03b832c7d2c3e7f094","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2025-10-06 09:15:07.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b8a0503c1003ac3593b814e331b1364b528cbca3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-06 09:15:27.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"144aaf1c9ff3590290a2d598cab6af2e8b1acfdd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-06 11:11:01.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9a4b6dcf1090474a8e8f7815da97cdba\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8e0506373b4b43d4b5a7bae194109861 : SUCCESS in 1h 43m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/191d6e218a514674888a295694e13ca1 : SUCCESS in 2m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/32d26ce8f010497ba52571bd9f9b2988 : SUCCESS in 4m 35s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/fc8bef2668fc4569a5bf6d5dbcb72a2b : SUCCESS in 5m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e64ddd8b51764a24b06c7eff89cbf08b : SUCCESS in 5m 04s\n- tooz-tox-py310-etcd3gw https://zuul.opendev.org/t/openstack/build/70440c894cf24ec1a57a4a310b6058ef : SUCCESS in 5m 05s\n- tooz-tox-py312-etcd3gw https://zuul.opendev.org/t/openstack/build/57678b871071431c905f54bd3cc03331 : SUCCESS in 5m 55s\n- tooz-tox-py310-zookeeper https://zuul.opendev.org/t/openstack/build/4c590867b9b946f8b3cfe7e35efc0ca8 : FAILURE in 2m 59s\n- tooz-tox-py312-zookeeper https://zuul.opendev.org/t/openstack/build/b30c1465b23f475590dd737cce250464 : SUCCESS in 5m 33s\n- tooz-tox-py310-redis https://zuul.opendev.org/t/openstack/build/803c7ea8f31c4e55ae700a2219441b89 : SUCCESS in 3m 41s\n- tooz-tox-py312-redis https://zuul.opendev.org/t/openstack/build/2a643a67f57644b4820828f2c131b94c : SUCCESS in 4m 08s\n- tooz-tox-py310-sentinel https://zuul.opendev.org/t/openstack/build/f849f2a2f92f48a0bd02c432d9562682 : SUCCESS in 4m 15s\n- tooz-tox-py312-sentinel https://zuul.opendev.org/t/openstack/build/964e9fb8ff984be3a06864943ae53030 : SUCCESS in 6m 18s\n- tooz-tox-py310-memcached https://zuul.opendev.org/t/openstack/build/6baed3736fcb4342b2d3f8b3e094cc81 : SUCCESS in 4m 21s\n- tooz-tox-py312-memcached https://zuul.opendev.org/t/openstack/build/a5bbffc4af924e62a8d01e5e116e3dd6 : SUCCESS in 5m 11s\n- tooz-tox-py310-postgresql https://zuul.opendev.org/t/openstack/build/219f1bcabc964fca9305adcac3a2a22d : SUCCESS in 2m 49s\n- tooz-tox-py312-postgresql https://zuul.opendev.org/t/openstack/build/1f61100300bf4deba2ad448c662294ea : SUCCESS in 4m 06s\n- tooz-tox-py310-mysql https://zuul.opendev.org/t/openstack/build/32cc5f1c9c034d87affe8ac615734e16 : SUCCESS in 4m 51s\n- tooz-tox-py312-mysql https://zuul.opendev.org/t/openstack/build/be9987fd10184aec9252fe1f907a363a : SUCCESS in 5m 02s\n- tooz-tox-py310-consul https://zuul.opendev.org/t/openstack/build/4c1d69ae09a64c0e91a912b60e0b2998 : SUCCESS in 3m 43s\n- tooz-tox-py312-consul https://zuul.opendev.org/t/openstack/build/4447397735b14c888799ccf78a13ccb8 : SUCCESS in 5m 29s","accounts_in_message":[],"_revision_number":1},{"id":"490164c959f7b1ad3978bd561e5254cbea86993b","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-10-07 00:50:47.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"69dad90a1bfa3b547cb9597a975580d15b188029","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-07 02:41:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4a1ecbf7ddf745c5a76db6f260926d70\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4d924572982a4f579fb66e21173f2eda : SUCCESS in 1h 45m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/944f34247c624556a9e5925eed7e3db4 : SUCCESS in 7m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/02644ac871cc4f8aba70813c5175dd91 : SUCCESS in 4m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/86ff68f1e967404d8874eeed03849126 : SUCCESS in 4m 29s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/cd2e66b9eb1d4d6b880fa147d67eae6f : SUCCESS in 5m 41s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a60cfb1a66db4c03bb301a41b38298f1 : SUCCESS in 8m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe1a6a711a764b2b8d502a61fccaa5e3 : SUCCESS in 5m 01s\n- tooz-tox-py310-etcd3gw https://zuul.opendev.org/t/openstack/build/f7cf624e95434bb3b18d639f27d269f9 : SUCCESS in 4m 57s\n- tooz-tox-py312-etcd3gw https://zuul.opendev.org/t/openstack/build/714ccc2fdc924e198f9bbf7401f8faa8 : SUCCESS in 5m 56s\n- tooz-tox-py310-zookeeper https://zuul.opendev.org/t/openstack/build/77ef44c3f5a74a65ba57123246d0a39d : SUCCESS in 4m 26s\n- tooz-tox-py312-zookeeper https://zuul.opendev.org/t/openstack/build/20ab2b007e4d48ea8785f79d150a89ef : SUCCESS in 5m 51s\n- tooz-tox-py310-redis https://zuul.opendev.org/t/openstack/build/cbd703b31f6041db8b46e050c0f76b81 : SUCCESS in 4m 32s\n- tooz-tox-py312-redis https://zuul.opendev.org/t/openstack/build/fced981fe11444c980757c0cbbd1cdf1 : SUCCESS in 3m 37s\n- tooz-tox-py310-sentinel https://zuul.opendev.org/t/openstack/build/a5adc1a5ed4e4fcdb960afb3b8a24326 : SUCCESS in 4m 50s\n- tooz-tox-py312-sentinel https://zuul.opendev.org/t/openstack/build/7800b825e15f4b47b05a70131020d670 : SUCCESS in 5m 35s\n- tooz-tox-py310-memcached https://zuul.opendev.org/t/openstack/build/f0ed0d1a83124601b77215ae28aa966a : SUCCESS in 4m 21s\n- tooz-tox-py312-memcached https://zuul.opendev.org/t/openstack/build/8ed1305f92564f7fb7710d1b7746b786 : SUCCESS in 5m 27s\n- tooz-tox-py310-postgresql https://zuul.opendev.org/t/openstack/build/a92258ed541a4d0796fb1392c4b73026 : SUCCESS in 4m 05s\n- tooz-tox-py312-postgresql https://zuul.opendev.org/t/openstack/build/db752a8e1c3a45b788fa7abdf4488554 : SUCCESS in 5m 30s\n- tooz-tox-py310-mysql https://zuul.opendev.org/t/openstack/build/dba4eeff06564a0a9867588cdec50f9a : SUCCESS in 4m 24s\n- tooz-tox-py312-mysql https://zuul.opendev.org/t/openstack/build/257ddb279ca14667b7f1b6afe5cf91b1 : SUCCESS in 3m 25s\n- tooz-tox-py310-consul https://zuul.opendev.org/t/openstack/build/726be894171c4b9f8d006f8a9f6cef35 : SUCCESS in 4m 38s\n- tooz-tox-py312-consul https://zuul.opendev.org/t/openstack/build/4506d41632d34fbc9e19e0ce741746ce : SUCCESS in 5m 13s","accounts_in_message":[],"_revision_number":1},{"id":"7d568ef14bc98b126b8fc83dcf25232bdfd6ff99","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-07 02:41:27.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f1fd715658e8dad7bf849e32737bef84417e0f49","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-07 04:16:38.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dcf41f6814874de58e3f4bba43d53d99\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e4969cca6f574f2a865402b475228979 : SUCCESS in 1h 35m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a701e94b130a4986aabb5f7c35362d3e : SUCCESS in 5m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/dc4a0371ee254759a3ca9bfe339f4e11 : SUCCESS in 4m 33s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e7325f00052247ddbda46c0d10ac7b5d : SUCCESS in 5m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5db4ab90810a4e44b9818ef9ad2784be : SUCCESS in 6m 42s\n- tooz-tox-py310-etcd3gw https://zuul.opendev.org/t/openstack/build/44f1f1ce67e348d9924ea4dff9859ebc : SUCCESS in 5m 14s\n- tooz-tox-py312-etcd3gw https://zuul.opendev.org/t/openstack/build/ea83128a49ec4021b8dd3ec6a5e02a9e : SUCCESS in 7m 50s\n- tooz-tox-py310-zookeeper https://zuul.opendev.org/t/openstack/build/41924cc69cb9433c82b6b93d6761d7d4 : SUCCESS in 4m 15s\n- tooz-tox-py312-zookeeper https://zuul.opendev.org/t/openstack/build/c46df0e6be6e4519a2ce1d1d3877827f : SUCCESS in 5m 30s\n- tooz-tox-py310-redis https://zuul.opendev.org/t/openstack/build/08db2107f50046109cc1728164227ac9 : SUCCESS in 4m 31s\n- tooz-tox-py312-redis https://zuul.opendev.org/t/openstack/build/c69c16ddc3f1447ea84a85e1c9501226 : SUCCESS in 5m 56s\n- tooz-tox-py310-sentinel https://zuul.opendev.org/t/openstack/build/3e2c7172045a41c7b6ed208818f621bb : SUCCESS in 6m 01s\n- tooz-tox-py312-sentinel https://zuul.opendev.org/t/openstack/build/4878fd2658a0440c874d2c35e14142af : SUCCESS in 7m 06s\n- tooz-tox-py310-memcached https://zuul.opendev.org/t/openstack/build/798c3bf4df67454f93e81283dc026257 : SUCCESS in 6m 16s\n- tooz-tox-py312-memcached https://zuul.opendev.org/t/openstack/build/b3f29e80114341378b7d20922fc30888 : SUCCESS in 7m 18s\n- tooz-tox-py310-postgresql https://zuul.opendev.org/t/openstack/build/6a1addf583ed41d19b6d13e30f1c23b2 : SUCCESS in 5m 59s\n- tooz-tox-py312-postgresql https://zuul.opendev.org/t/openstack/build/98382ee3b5554136afbd534b297a7d20 : SUCCESS in 6m 44s\n- tooz-tox-py310-mysql https://zuul.opendev.org/t/openstack/build/9679f3b8ba38458981df5cdbcf90bcd3 : SUCCESS in 4m 25s\n- tooz-tox-py312-mysql https://zuul.opendev.org/t/openstack/build/1045056d9dcb4aa19714662f2699a3d3 : SUCCESS in 7m 17s\n- tooz-tox-py310-consul https://zuul.opendev.org/t/openstack/build/40486ef7d61a4d7cbcbbcde9e743e3c9 : SUCCESS in 6m 04s\n- tooz-tox-py312-consul https://zuul.opendev.org/t/openstack/build/998b56d56d40416588f517000824465c : SUCCESS in 7m 31s","accounts_in_message":[],"_revision_number":1},{"id":"ec276d71f0d1e81ee0454da666ed74f61c77c459","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-07 04:16:38.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"228379bd56d04886796cd9ae952a8456ee53e926","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-07 04:17:25.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e841a9aa56044c3bbd40780b99038546\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/556e209e066e41a2b19b2fead01dbad7 : SUCCESS in 37s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7f3c0b99b724059035a3f89cb6c7bdc4780132ac","revisions":{"7f3c0b99b724059035a3f89cb6c7bdc4780132ac":{"kind":"REWORK","_number":1,"created":"2025-10-05 10:45:54.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/89/963089/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/89/963089/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/89/963089/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/89/963089/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/89/963089/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/89/963089/1"}}},"commit":{"parents":[{"commit":"93c9aa91e3f50d305c00912972b32ad9cbe5872e","subject":"pre-commit: Bump dependencies","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/93c9aa91e3f50d305c00912972b32ad9cbe5872e"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-10-05 10:44:50.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-10-05 10:45:41.000000000","tz":540},"subject":"Bump pyupgrade target to 3.10+","message":"Bump pyupgrade target to 3.10+\n\n... according to the versions currently supported.\n\nChange-Id: Iae501ca85ea947c86cb10c2631ea265ef97741ed\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/7f3c0b99b724059035a3f89cb6c7bdc4780132ac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/7f3c0b99b724059035a3f89cb6c7bdc4780132ac"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
