)]}'
{"id":"openstack%2Ftooz~973984","triplet_id":"openstack%2Ftooz~master~I93c4f50a5e1898eb5e4297b566ab2873171e4cbe","project":"openstack/tooz","branch":"master","topic":"typing","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2026-01-27 18:15:59.000000000","reason":"\u003cGERRIT_ACCOUNT_15334\u003e replied on the change","reason_account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-01-29 17:23:35.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I93c4f50a5e1898eb5e4297b566ab2873171e4cbe","subject":"Add ruff","status":"MERGED","created":"2026-01-20 18:51:32.000000000","updated":"2026-01-29 17:24:34.000000000","submitted":"2026-01-29 17:23:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"973984-typing","meta_rev_id":"1e4380bb925d0fc316439c555181d6444a334d9f","_number":973984,"virtual_id_number":973984,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-29 17:23:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2026-01-29 16:13:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2026-01-29 16:13:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-20 20:26:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-01-27 13:26:09.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"CC"}],"messages":[{"id":"d59eeef2ad388fdea1327c04d8e6cade07a35804","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-01-20 18:51:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7b26c9ed340c9ecc62208834fc20066d8d6b691e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-20 20:26:25.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/124194bb5658404badbab0ad3a8704fd\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/1fe2d6f4809a489ea95a50434e67d5d9 : SUCCESS in 2m 59s\n- requirements-check https://zuul.opendev.org/t/openstack/build/c7b148548e654f939bb713c775315025 : SUCCESS in 2m 34s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/55a676c655b24a158ece16cd0e3c5149 : SUCCESS in 1h 32m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/587d2ffe2ad34942837c553b6cbaf515 : SUCCESS in 6m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9abaa20188ad48cfb833b06aa29eafac : SUCCESS in 4m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/55a362f11b404ee99d4932e4050164fd : SUCCESS in 4m 55s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2cc9aa7a5730468eafbeaca389a497d1 : SUCCESS in 6m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/49cccd6d5b1d4cdc87803d0a9ee7ebd8 : SUCCESS in 6m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9aeaa3b21fdd4c88876d512c24e1ce50 : SUCCESS in 3m 04s\n- tooz-tox-py310-etcd3gw https://zuul.opendev.org/t/openstack/build/b08540d534884156a717d7cd03b8f46d : SUCCESS in 5m 22s\n- tooz-tox-py313-etcd3gw https://zuul.opendev.org/t/openstack/build/119da8bb1811420ea5014c6bef2af174 : SUCCESS in 7m 37s\n- tooz-tox-py310-zookeeper https://zuul.opendev.org/t/openstack/build/306bf43e6984462199c8e7cb3470a64c : SUCCESS in 4m 47s\n- tooz-tox-py313-zookeeper https://zuul.opendev.org/t/openstack/build/2b01dca1056e4321aa30f699cd81d5e3 : SUCCESS in 5m 45s\n- tooz-tox-py310-redis https://zuul.opendev.org/t/openstack/build/adfdafc156404776ab9bf2d5bf47e024 : SUCCESS in 6m 20s\n- tooz-tox-py313-redis https://zuul.opendev.org/t/openstack/build/f46f89a2ddcf4657b6afd92e086b4175 : SUCCESS in 5m 23s\n- tooz-tox-py310-sentinel https://zuul.opendev.org/t/openstack/build/dbfd3bed86ac44269f8aa228dd5669da : SUCCESS in 4m 48s\n- tooz-tox-py313-sentinel https://zuul.opendev.org/t/openstack/build/7c9783055f4e43f2821fac0affaadf67 : FAILURE in 4m 16s\n- tooz-tox-py310-memcached https://zuul.opendev.org/t/openstack/build/99fa71f743f3419fafa4eb4168a51b62 : SUCCESS in 4m 19s\n- tooz-tox-py313-memcached https://zuul.opendev.org/t/openstack/build/7ecaf3f6f93c46e89da05956fa1ffbb6 : SUCCESS in 7m 23s\n- tooz-tox-py310-postgresql https://zuul.opendev.org/t/openstack/build/06b56cbf55a440158afa230a1d9aa8c6 : SUCCESS in 6m 05s\n- tooz-tox-py313-postgresql https://zuul.opendev.org/t/openstack/build/79fd334042ed4599a2fa6714c9d26d10 : SUCCESS in 12m 29s\n- tooz-tox-py310-mysql https://zuul.opendev.org/t/openstack/build/5c2f1e2ca785439cb7b72cae8e8d9495 : SUCCESS in 4m 18s\n- tooz-tox-py313-mysql https://zuul.opendev.org/t/openstack/build/f60a04c8cca048269ff568724f1bbb44 : FAILURE in 6m 42s\n- tooz-tox-py310-consul https://zuul.opendev.org/t/openstack/build/542f7dcfc9cc44a49313bac09a145426 : SUCCESS in 4m 41s\n- tooz-tox-py313-consul https://zuul.opendev.org/t/openstack/build/821c9d5ce5b643a9b1fe6dd2e9c2273c : SUCCESS in 5m 14s","accounts_in_message":[],"_revision_number":1},{"id":"8e5bbbcab2f5007a8bc4d896bf3def760cc71733","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-01-27 13:26:09.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"6203944f1f4a47732f50b3cb051de6013c430af5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-01-27 18:15:56.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"4069907dd4fc6aa9a0e4b9350eb167ab309d6115","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-01-27 18:15:59.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"ded211290290dd02ec8f8cfaee6f9da6049b6d59","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-27 19:59:21.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/acc0c622033e43ad9b9a72caf9df6c61\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/5022f8c2e8a5415ab697b1ff5fb87334 : SUCCESS in 2m 27s\n- requirements-check https://zuul.opendev.org/t/openstack/build/518c72be992f4934a1a4c1059b9049d8 : SUCCESS in 3m 23s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/dd5a971a72e6435695d9f05f2a5b0abd : SUCCESS in 1h 38m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/83d134d7c4e346a5895845cfd63fd9aa : SUCCESS in 8m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/05938fce1705433189fcd4c12d3993be : SUCCESS in 2m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/49079fe03f494593b84cb4bc43dd834f : SUCCESS in 4m 35s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0bd5d4b15ee04766b86a722c1525c929 : SUCCESS in 6m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c4faf0def5b74099b445a5c85e1a4bde : SUCCESS in 4m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b3fe32b110ee4a4f97cc5312ba142d01 : SUCCESS in 3m 24s\n- tooz-tox-py310-etcd3gw https://zuul.opendev.org/t/openstack/build/3fc2ca31c9fa427583e9dba9aff9d167 : SUCCESS in 3m 27s\n- tooz-tox-py313-etcd3gw https://zuul.opendev.org/t/openstack/build/f01520941e844f3d89cae075cdfb9dd9 : SUCCESS in 3m 48s\n- tooz-tox-py310-zookeeper https://zuul.opendev.org/t/openstack/build/d53350c2380446438fb7b3b774183b67 : SUCCESS in 4m 16s\n- tooz-tox-py313-zookeeper https://zuul.opendev.org/t/openstack/build/ee9b1675f07d4df6b28cf792b47e0a79 : SUCCESS in 5m 18s\n- tooz-tox-py310-redis https://zuul.opendev.org/t/openstack/build/0dab0892d33c4a9a94d4a50e4323e2fd : SUCCESS in 4m 35s\n- tooz-tox-py313-redis https://zuul.opendev.org/t/openstack/build/384c3fdd3d534381a8de25841c5ce2fa : SUCCESS in 8m 40s\n- tooz-tox-py310-sentinel https://zuul.opendev.org/t/openstack/build/03eedb1feb7c46bf9d33699cf98d0ac9 : SUCCESS in 4m 43s\n- tooz-tox-py313-sentinel https://zuul.opendev.org/t/openstack/build/a983ad2bfc814c85bb2a002f0ec84760 : SUCCESS in 8m 52s\n- tooz-tox-py310-memcached https://zuul.opendev.org/t/openstack/build/ed3ebdfbbc544c9ba9a40e42fe9c7bd8 : SUCCESS in 6m 05s\n- tooz-tox-py313-memcached https://zuul.opendev.org/t/openstack/build/07548d4cff1e486aac1d9284478a103c : SUCCESS in 9m 41s\n- tooz-tox-py310-postgresql https://zuul.opendev.org/t/openstack/build/e9584c17ec0f4b1fa568e4558bddf56c : SUCCESS in 4m 11s\n- tooz-tox-py313-postgresql https://zuul.opendev.org/t/openstack/build/9d9b693e2bba45b2a46e2a875623d10b : SUCCESS in 4m 41s\n- tooz-tox-py310-mysql https://zuul.opendev.org/t/openstack/build/fe983de4a57a42d98cf25cbae609053c : SUCCESS in 5m 41s\n- tooz-tox-py313-mysql https://zuul.opendev.org/t/openstack/build/ea514b127a5a4db2a43f1f3d604514ee : SUCCESS in 9m 19s\n- tooz-tox-py310-consul https://zuul.opendev.org/t/openstack/build/8454f853fa8a430b8d5d9e3ca33ae8f1 : SUCCESS in 5m 54s\n- tooz-tox-py313-consul https://zuul.opendev.org/t/openstack/build/7aa8be5125504c33ac08e6f74471237d : SUCCESS in 5m 07s","accounts_in_message":[],"_revision_number":2},{"id":"62e3e3e06b4e598a4156019a6ac75f84bfe2ecbd","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-01-29 16:13:18.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"2a3092856b9c0a81cc39457f378389097da1528b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-29 16:13:40.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"75b51cba91ebc9b9e72defe7dff98ca18552eb4f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-29 17:23:35.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a1bb8500fdce4258ac0284cad7ee8c52\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/872d9e789dd441d199949f13130cf82f : SUCCESS in 1m 29s\n- requirements-check https://zuul.opendev.org/t/openstack/build/b737676483524099aaee896f85965e46 : SUCCESS in 3m 07s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/15107d0c47914568bfda34387773db61 : SUCCESS in 1h 08m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/40652ad23b9749eab0b6fb17df111166 : SUCCESS in 4m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e4e89b91c26c48d7b3316318131beb58 : SUCCESS in 4m 25s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/61264c8b4c49421db9a0730b2da5ae1c : SUCCESS in 3m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bbcb7129685940798be9da420c3bd835 : SUCCESS in 5m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/acdc6e2f6c524c8293deb90e7ff61483 : SUCCESS in 3m 45s\n- tooz-tox-py310-etcd3gw https://zuul.opendev.org/t/openstack/build/fe6980096a3e49dcac867f616e010e1b : SUCCESS in 4m 25s\n- tooz-tox-py313-etcd3gw https://zuul.opendev.org/t/openstack/build/ba54d97e83ee4a4db5ea39fd8c1260c0 : SUCCESS in 7m 49s\n- tooz-tox-py310-zookeeper https://zuul.opendev.org/t/openstack/build/413221ea0d474acf847b404b245f03e0 : SUCCESS in 4m 15s\n- tooz-tox-py313-zookeeper https://zuul.opendev.org/t/openstack/build/d0151568e0244dbe85708924b71e419a : SUCCESS in 2m 51s\n- tooz-tox-py310-redis https://zuul.opendev.org/t/openstack/build/f36105e155cd4b5cb4f52abff1189a1c : SUCCESS in 4m 33s\n- tooz-tox-py313-redis https://zuul.opendev.org/t/openstack/build/b248fdce62fe4648bd3f7c5abb2b4fed : SUCCESS in 7m 40s\n- tooz-tox-py310-sentinel https://zuul.opendev.org/t/openstack/build/7df8f081e571413391bd84fed3594da1 : SUCCESS in 4m 39s\n- tooz-tox-py313-sentinel https://zuul.opendev.org/t/openstack/build/1ebdf9c1612c463ba6eea2b91a5009e4 : SUCCESS in 4m 54s\n- tooz-tox-py310-memcached https://zuul.opendev.org/t/openstack/build/c475d398ea7a4e1f9e5057fbeb79c51d : SUCCESS in 3m 09s\n- tooz-tox-py313-memcached https://zuul.opendev.org/t/openstack/build/dfc88f80b9314813b76968c0e0c36738 : SUCCESS in 3m 24s\n- tooz-tox-py310-postgresql https://zuul.opendev.org/t/openstack/build/4a2cfed016794b05a5a47ce9b9717ce4 : SUCCESS in 4m 21s\n- tooz-tox-py313-postgresql https://zuul.opendev.org/t/openstack/build/bd046d93f1bc45ddbfb932589b76c7da : SUCCESS in 4m 29s\n- tooz-tox-py310-mysql https://zuul.opendev.org/t/openstack/build/ef3c11ef39214e9eafe385d3c5fe4b71 : SUCCESS in 2m 54s\n- tooz-tox-py313-mysql https://zuul.opendev.org/t/openstack/build/86bc3acf01624df59a450345cdaac390 : SUCCESS in 7m 24s\n- tooz-tox-py310-consul https://zuul.opendev.org/t/openstack/build/b93a0fad6bad4d189099e881b28f927c : SUCCESS in 4m 13s\n- tooz-tox-py313-consul https://zuul.opendev.org/t/openstack/build/a33e3b56a0ac4e618b0e355455fa2ba9 : SUCCESS in 5m 22s","accounts_in_message":[],"_revision_number":2},{"id":"57cc225c9daa285c0a79bf5e07113a0ef7debcbc","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-29 17:23:35.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"1e4380bb925d0fc316439c555181d6444a334d9f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-29 17:24:34.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6c21942b976242649c591dfc57fe7b54\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/817dd73ee7de41de8f5aaded78f3a4b9 : SUCCESS in 48s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/23e7c157c14d4675927bdaf9ba4ff200 : SUCCESS in 40s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"aa92b7f09923e91e8f0602600cd2ee48b3537775","revisions":{"85c964a9f7c2e07ed17784742826620fbea0d83e":{"kind":"REWORK","_number":1,"created":"2026-01-20 18:51:32.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/84/973984/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/84/973984/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/84/973984/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/84/973984/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/84/973984/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/84/973984/1"}}},"commit":{"parents":[{"commit":"685b888b1b293ea70248a41f9fd94d20b8bfb840","subject":"Merge \"Move upper functional job from 3.12 to 3.13\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/685b888b1b293ea70248a41f9fd94d20b8bfb840"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-01-20 18:49:52.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-01-20 18:51:07.000000000","tz":0},"subject":"Add ruff","message":"Add ruff\n\nChange-Id: I93c4f50a5e1898eb5e4297b566ab2873171e4cbe\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/85c964a9f7c2e07ed17784742826620fbea0d83e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/85c964a9f7c2e07ed17784742826620fbea0d83e"}]},"branch":"refs/heads/master"},"aa92b7f09923e91e8f0602600cd2ee48b3537775":{"kind":"REWORK","_number":2,"created":"2026-01-27 18:15:56.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/84/973984/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/84/973984/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/84/973984/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/84/973984/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/84/973984/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/84/973984/2"}}},"commit":{"parents":[{"commit":"3de5906ce237edd74e4429400497815f9a456469","subject":"Pin tooz-tox-py313-mysql to Ubuntu Noble","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/3de5906ce237edd74e4429400497815f9a456469"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-01-20 18:49:52.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-01-27 18:15:48.000000000","tz":0},"subject":"Add ruff","message":"Add ruff\n\nChange-Id: I93c4f50a5e1898eb5e4297b566ab2873171e4cbe\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/aa92b7f09923e91e8f0602600cd2ee48b3537775"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/aa92b7f09923e91e8f0602600cd2ee48b3537775"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
