)]}'
{"id":"openstack%2Ftooz~976150","triplet_id":"openstack%2Ftooz~master~I05607a8116eaf5dc18e5ea132eda6236966242ec","project":"openstack/tooz","branch":"master","topic":"remove-zake","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-02-13 10:41:37.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I05607a8116eaf5dc18e5ea132eda6236966242ec","subject":"Remove zake driver","status":"MERGED","created":"2026-02-09 17:04:58.000000000","updated":"2026-02-13 10:43:54.000000000","submitted":"2026-02-13 10:41:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"976150-remove-zake","meta_rev_id":"cd5a51770418fc12517c2ac7aa8e1fb8472b535d","_number":976150,"virtual_id_number":976150,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-02-13 10:41:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2026-02-11 15:01:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2026-02-13 08:59:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"date":"2026-02-12 16:57:50.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":1,"date":"2026-02-13 08:59:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-09 17:44:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-02-11 15:01:06.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2026-02-13 08:59:46.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"9c98058449fb024d1cebcf81d6d95fd9e1911028","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-02-09 17:04:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"28dccea61503a73ce2ee39856a5d80c9aca75a9c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-09 17:44:46.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4686a4e2e40f4a0aa915141952018866\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/f3dcb762f7fc4764a9b1ad07077a6f28 : SUCCESS in 2m 22s\n- requirements-check https://zuul.opendev.org/t/openstack/build/da1723f3785d40178a5a41d9b808f33c : SUCCESS in 4m 14s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8d5aab4cf7db42948b3f0da2b3caf469 : FAILURE in 31m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a5d5f59e7126473eba4a10108e828260 : FAILURE in 5m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2f557ce940b2469489ceddd56471aabd : SUCCESS in 3m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/962a1cbeb3d34d21815f7eb12e5ce76f : FAILURE in 4m 17s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/388f2b33f93946b4b75ee3f57f7114ab : FAILURE in 6m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ab1bbfcb63148a7b7e522249430c982 : SUCCESS in 4m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bcebc78ba1f84dfe9e8ccc7100971275 : SUCCESS in 3m 37s\n- tooz-tox-py310-etcd3gw https://zuul.opendev.org/t/openstack/build/9e49de0374004fdba8c02b08dc6690f3 : FAILURE in 3m 49s\n- tooz-tox-py313-etcd3gw https://zuul.opendev.org/t/openstack/build/5373d614506d40f08293fa44fb21382a : FAILURE in 3m 35s\n- tooz-tox-py310-zookeeper https://zuul.opendev.org/t/openstack/build/7de5890734b641dfad643bdc3241f77f : FAILURE in 4m 22s\n- tooz-tox-py313-zookeeper https://zuul.opendev.org/t/openstack/build/b9ea8035cac54ab991d183dfc2b91c7c : FAILURE in 2m 56s\n- tooz-tox-py310-redis https://zuul.opendev.org/t/openstack/build/e7892488f5e84d5e8cde9fd25be2105f : FAILURE in 4m 41s\n- tooz-tox-py313-redis https://zuul.opendev.org/t/openstack/build/493d90c89664485f98b86fd3e1d80e43 : FAILURE in 3m 18s\n- tooz-tox-py310-sentinel https://zuul.opendev.org/t/openstack/build/ebe87c4caa2e4861a2c868648649378e : FAILURE in 4m 40s\n- tooz-tox-py313-sentinel https://zuul.opendev.org/t/openstack/build/03febb498c504a69ad22d83eafbac507 : FAILURE in 4m 55s\n- tooz-tox-py310-memcached https://zuul.opendev.org/t/openstack/build/179b589824b8448680e7d45a18ae8cf0 : FAILURE in 3m 31s\n- tooz-tox-py313-memcached https://zuul.opendev.org/t/openstack/build/ef4b79bb0ff047e29c5122fb730b0468 : FAILURE in 6m 14s\n- tooz-tox-py310-postgresql https://zuul.opendev.org/t/openstack/build/3b33573e7f4145da8133a0f22b42f0d3 : FAILURE in 3m 57s\n- tooz-tox-py313-postgresql https://zuul.opendev.org/t/openstack/build/a939016b9aa0448093ea81d45d8bb037 : FAILURE in 5m 04s\n- tooz-tox-py310-mysql https://zuul.opendev.org/t/openstack/build/5b34c7683b904319bcdf6e3f37e00285 : FAILURE in 4m 48s\n- tooz-tox-py313-mysql https://zuul.opendev.org/t/openstack/build/5a761c916fb845d3842f0ad0f2e277ee : FAILURE in 7m 06s\n- tooz-tox-py310-consul https://zuul.opendev.org/t/openstack/build/0cdd71efa0dd439795ca829d51f22acc : FAILURE in 4m 10s\n- tooz-tox-py313-consul https://zuul.opendev.org/t/openstack/build/4d3802610009443fbe88dfcd96711a19 : FAILURE in 4m 44s","accounts_in_message":[],"_revision_number":1},{"id":"5212225cdd4b2d8aeea2287c3c6d3daef8fd94b2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-02-11 05:22:22.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"2ba41497fbf0fa8540040ab0a404a426c654e02b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-02-11 05:24:24.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"e302a6f8e6a6ee1f20b4d0f894bfb199f82e621d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-11 05:53:04.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1bbf5ec15ddc41b98bc5020ffa26f677\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/98f10f864f6d4625828d3e0673cd811b : SUCCESS in 2m 43s\n- requirements-check https://zuul.opendev.org/t/openstack/build/6a29039b7a994a5bb96a24fed7b0d282 : SUCCESS in 2m 28s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/66dafdf76c2a425585c3c3437a4da5c7 : FAILURE in 25m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ffae062e98954976954f877e94655965 : SUCCESS in 5m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6298114fdc9a47fca6cd268564d61cc8 : SUCCESS in 4m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/aef24e109a454a2bb5c22af7f18a8be6 : SUCCESS in 4m 24s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ac5f4e6a08004397a37e8230ec42fcb7 : SUCCESS in 12m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/deb1bad3e4a24c63bc375159a98cbeab : SUCCESS in 4m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/021c1bcb442948ac8a5a6835394f41dc : SUCCESS in 3m 20s\n- tooz-tox-py310-etcd3gw https://zuul.opendev.org/t/openstack/build/a39b7906b9514d84ba03a46d8aa7e813 : SUCCESS in 4m 50s\n- tooz-tox-py313-etcd3gw https://zuul.opendev.org/t/openstack/build/d031391fe7e041a4b345292d897709f1 : SUCCESS in 5m 23s\n- tooz-tox-py310-zookeeper https://zuul.opendev.org/t/openstack/build/db3c21598c324777a127f19907406d3a : SUCCESS in 2m 43s\n- tooz-tox-py313-zookeeper https://zuul.opendev.org/t/openstack/build/a0e221ecb5cd4308a5b3ee31e4de60ff : SUCCESS in 6m 53s\n- tooz-tox-py310-redis https://zuul.opendev.org/t/openstack/build/8b9cff7bc3ec457196b07381041c5ede : SUCCESS in 4m 30s\n- tooz-tox-py313-redis https://zuul.opendev.org/t/openstack/build/033beda0f44e46f18b82b826963733e4 : SUCCESS in 10m 46s\n- tooz-tox-py310-sentinel https://zuul.opendev.org/t/openstack/build/7c26be0e9efa43878916953e8435703a : SUCCESS in 4m 06s\n- tooz-tox-py313-sentinel https://zuul.opendev.org/t/openstack/build/1b72b9dd6f804f7d98a72dec41761609 : SUCCESS in 5m 36s\n- tooz-tox-py310-memcached https://zuul.opendev.org/t/openstack/build/1a65f0b759554edb98bb2646b9514390 : SUCCESS in 4m 18s\n- tooz-tox-py313-memcached https://zuul.opendev.org/t/openstack/build/26714b5f359e4beeba2c4fc244df7fd5 : SUCCESS in 6m 10s\n- tooz-tox-py310-postgresql https://zuul.opendev.org/t/openstack/build/4d0bb61987d54269b0bea04550463a7f : SUCCESS in 5m 40s\n- tooz-tox-py313-postgresql https://zuul.opendev.org/t/openstack/build/541dae6546ed4ccc9cd074a0f5ebfa58 : SUCCESS in 4m 23s\n- tooz-tox-py310-mysql https://zuul.opendev.org/t/openstack/build/3853c465876d4e3d9b2324f93e0f1654 : SUCCESS in 4m 56s\n- tooz-tox-py313-mysql https://zuul.opendev.org/t/openstack/build/815438d1cf5f412f93b7c88831397bf8 : SUCCESS in 4m 21s\n- tooz-tox-py310-consul https://zuul.opendev.org/t/openstack/build/a04e45065a06451fab4d9d5bb14e0f17 : SUCCESS in 6m 25s\n- tooz-tox-py313-consul https://zuul.opendev.org/t/openstack/build/e6117cad0be74dcfb0190f1c2c01570a : SUCCESS in 4m 39s","accounts_in_message":[],"_revision_number":3},{"id":"7097d0c237df61e2d491649a1f57071708521eb5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-02-11 11:29:32.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"2e99c47fa07e12db6b9ea5ca5f25d834aa70130f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-02-11 11:31:04.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"b65ee52d1a3c6897481709208974cfba1b9e948a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-11 11:49:56.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/de25bbe5b8f44443a40c7b646ed18046\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/56e09197eb414d939fb3927ed6de1242 : SUCCESS in 5m 00s\n- requirements-check https://zuul.opendev.org/t/openstack/build/cb1fd0d974c14bf2bf115b7c6ed24ad5 : SUCCESS in 3m 45s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0d0c4eb133f644439de38c4e90a4552e : FAILURE in 17m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ad49cd4a8a3149b28b2612ace2ad7675 : SUCCESS in 3m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1ef2b6587203460683c278665445e6f2 : SUCCESS in 4m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/efadbde9cffe4716a022f48090eefc22 : SUCCESS in 4m 42s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f18cb2be426d4ef4badd994e4d82f790 : SUCCESS in 5m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ca880569e574ddc8827b5332ee29154 : SUCCESS in 2m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/40f4cacbf95942e0ac20a6bb7c30dbb1 : SUCCESS in 4m 15s\n- tooz-tox-examples https://zuul.opendev.org/t/openstack/build/61d4f34833294a92b5fef96d6058df0b : SUCCESS in 4m 36s\n- tooz-tox-py310-etcd3gw https://zuul.opendev.org/t/openstack/build/ca11aa8f02dc430a976e839480666519 : SUCCESS in 5m 21s\n- tooz-tox-py313-etcd3gw https://zuul.opendev.org/t/openstack/build/8786d6c810864997878084c67bcace39 : SUCCESS in 7m 12s\n- tooz-tox-py310-zookeeper https://zuul.opendev.org/t/openstack/build/d1a11cce99c34b3793ee54988ba59d73 : SUCCESS in 4m 45s\n- tooz-tox-py313-zookeeper https://zuul.opendev.org/t/openstack/build/dfde53a5045240d09dea0b0a6868346b : SUCCESS in 6m 06s\n- tooz-tox-py310-redis https://zuul.opendev.org/t/openstack/build/eb379fbe4b4d44408a12265cddefb0d6 : SUCCESS in 4m 36s\n- tooz-tox-py313-redis https://zuul.opendev.org/t/openstack/build/957e6fbe34b74d01bec021a4c00dd055 : SUCCESS in 4m 58s\n- tooz-tox-py310-sentinel https://zuul.opendev.org/t/openstack/build/e71826c0860243c0a5af66fff32884e5 : SUCCESS in 5m 01s\n- tooz-tox-py313-sentinel https://zuul.opendev.org/t/openstack/build/889f064ec1e643c0b72d3567cbb6df0c : SUCCESS in 4m 49s\n- tooz-tox-py310-memcached https://zuul.opendev.org/t/openstack/build/1a52e2cd65e041fe9b2bafc8240f3d39 : SUCCESS in 4m 06s\n- tooz-tox-py313-memcached https://zuul.opendev.org/t/openstack/build/4a251ab17b7a410384bddbf6b2bcb699 : SUCCESS in 7m 29s\n- tooz-tox-py310-postgresql https://zuul.opendev.org/t/openstack/build/f02b232a9cd14c9ea1051eaaa534f4af : SUCCESS in 3m 45s\n- tooz-tox-py313-postgresql https://zuul.opendev.org/t/openstack/build/d9c8151bab874aaab3945ee356868898 : SUCCESS in 5m 31s\n- tooz-tox-py310-mysql https://zuul.opendev.org/t/openstack/build/86debf0b1632464d8d0c6099bb11743b : SUCCESS in 4m 13s\n- tooz-tox-py313-mysql https://zuul.opendev.org/t/openstack/build/92fb8c200f0840babd5158aca6d4db3e : SUCCESS in 12m 22s\n- tooz-tox-py310-consul https://zuul.opendev.org/t/openstack/build/20dc35df2ef4496586751c5229ef777c : SUCCESS in 7m 47s\n- tooz-tox-py313-consul https://zuul.opendev.org/t/openstack/build/33c5bf988fa2410eae1ab07924c231bc : SUCCESS in 6m 34s","accounts_in_message":[],"_revision_number":5},{"id":"fb7605c63ba467379d5282017cf311ca6be491d5","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-02-11 11:57:11.000000000","message":"Topic set to remove-zake","accounts_in_message":[],"_revision_number":5},{"id":"f7aab266b94e329fa25150cdfddc24f407601d40","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-02-11 12:44:59.000000000","message":"Patch Set 5: Workflow-1","accounts_in_message":[],"_revision_number":5},{"id":"ed90373ca0bf9b31689821861b4d059d9eb25dbc","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-02-11 15:01:06.000000000","message":"Patch Set 5: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"4065d78060fae4fcc1c19e8273c9676d754f4d99","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-02-11 15:02:47.000000000","message":"Patch Set 5: -Workflow\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"a721190a6c1cf6c926d3b12c0a8453d981a58412","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-02-12 16:57:50.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"8967772beb8cc85ed39255d3cc7186f06bed51a6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-12 18:22:06.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8a7fd182dc3c4ac0b36ca6eb25c58a79\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/78fd8b4f637b4140b42ba621c4248054 : SUCCESS in 1m 34s\n- requirements-check https://zuul.opendev.org/t/openstack/build/cd15cce016c640caaaf2f0e038390b48 : SUCCESS in 3m 35s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/47bb134eecc348648dfefefdf710740b : SUCCESS in 1h 22m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6de8fa2485624f3e9ab4d1b411c164d5 : SUCCESS in 7m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/915a9a160891426984a1d8235b060b6d : SUCCESS in 4m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e0d7162c5ac442ae9f706d6bbf3fbe4d : SUCCESS in 3m 15s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f120087f91864a04bb295b92a59f6008 : SUCCESS in 5m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ffbccca633e473083b568c8f2a824b6 : SUCCESS in 4m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/15b76c32b4944b08b72c2552671eb2af : SUCCESS in 5m 40s\n- tooz-tox-examples https://zuul.opendev.org/t/openstack/build/1c8fb1271a1e45e58e7963c7e009acdb : SUCCESS in 5m 05s\n- tooz-tox-py310-etcd3gw https://zuul.opendev.org/t/openstack/build/f6d1ea27d2db4cb49886bcaa72065d8b : SUCCESS in 6m 29s\n- tooz-tox-py313-etcd3gw https://zuul.opendev.org/t/openstack/build/c21c794bd4544d1f96e39d473367cb29 : SUCCESS in 5m 19s\n- tooz-tox-py310-zookeeper https://zuul.opendev.org/t/openstack/build/c2de24c9ff6f45a6bc0aedbdaa79581b : SUCCESS in 4m 18s\n- tooz-tox-py313-zookeeper https://zuul.opendev.org/t/openstack/build/21b4b87ee8b644309cb05d5bbfa08d4c : SUCCESS in 6m 47s\n- tooz-tox-py310-redis https://zuul.opendev.org/t/openstack/build/f5840c9124194285aad76d0be6797ab1 : SUCCESS in 4m 16s\n- tooz-tox-py313-redis https://zuul.opendev.org/t/openstack/build/199d8cc1f51e48b699fd14f69d4549cb : SUCCESS in 3m 26s\n- tooz-tox-py310-sentinel https://zuul.opendev.org/t/openstack/build/b9cbf1a210e24d5b9c9696f9a28d0dc4 : SUCCESS in 3m 17s\n- tooz-tox-py313-sentinel https://zuul.opendev.org/t/openstack/build/f335550f1e134e21af316b2ff60566f0 : SUCCESS in 5m 05s\n- tooz-tox-py310-memcached https://zuul.opendev.org/t/openstack/build/dd102d05fec04260806ed72e53c610a0 : SUCCESS in 4m 34s\n- tooz-tox-py313-memcached https://zuul.opendev.org/t/openstack/build/53cd92322ffe42fbbb77d8164c9ea909 : SUCCESS in 5m 11s\n- tooz-tox-py310-postgresql https://zuul.opendev.org/t/openstack/build/77bacdbadf1c43c38343af2915762784 : SUCCESS in 4m 13s\n- tooz-tox-py313-postgresql https://zuul.opendev.org/t/openstack/build/2758bc727c3948428c93c6c869d0122b : SUCCESS in 2m 58s\n- tooz-tox-py310-mysql https://zuul.opendev.org/t/openstack/build/2240932f5b4b43029e15608c9cc92aad : SUCCESS in 4m 10s\n- tooz-tox-py313-mysql https://zuul.opendev.org/t/openstack/build/698afbe9f2ff4907b3dd64f33d45fb88 : SUCCESS in 7m 23s\n- tooz-tox-py310-consul https://zuul.opendev.org/t/openstack/build/a577fc1504a44ab7b1b3f85997fae042 : SUCCESS in 5m 26s\n- tooz-tox-py313-consul https://zuul.opendev.org/t/openstack/build/9c6ec919b25147c79f10931a7cb2f089 : SUCCESS in 3m 20s","accounts_in_message":[],"_revision_number":5},{"id":"8373301721cb3ad9157d9d88394829227db1cce3","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2026-02-13 08:59:46.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"508fc03bde3115cf1c96cc2df84ba21c579f2825","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-13 09:00:23.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"17de64a5a5f5d68cb6493385ade5540a224e933b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-13 10:41:37.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b6062a967e114f54927b4d16985db773\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/a32a16675fc6457aba66778f20f55cb4 : SUCCESS in 1m 32s\n- requirements-check https://zuul.opendev.org/t/openstack/build/d57a0d2a4161422fb01c4fc3c058fbaf : SUCCESS in 2m 01s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/39260172f0a64970931f64088bf84024 : SUCCESS in 1h 40m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c33e3dfca674688bc42fd81178023ee : SUCCESS in 4m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/99c0eeeb85aa41f5a9bee17c43eb5c5b : SUCCESS in 4m 40s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/97f73400d627486884e2e080378aa496 : SUCCESS in 5m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da5cf13f6f0e4a54b6e3676d87eadffb : SUCCESS in 2m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/12264df338fe4232a51fe4f36026e88b : SUCCESS in 4m 49s\n- tooz-tox-examples https://zuul.opendev.org/t/openstack/build/944b07fb95c34b1c8c81d7df88c811df : SUCCESS in 5m 03s\n- tooz-tox-py310-etcd3gw https://zuul.opendev.org/t/openstack/build/b531d0f2c391453e8a96d82547931915 : SUCCESS in 3m 26s\n- tooz-tox-py313-etcd3gw https://zuul.opendev.org/t/openstack/build/97c7edb29cd94b5abbd54307f2bf9e04 : SUCCESS in 5m 17s\n- tooz-tox-py310-zookeeper https://zuul.opendev.org/t/openstack/build/f219dedc245443b789ebbb67f78e6126 : SUCCESS in 4m 33s\n- tooz-tox-py313-zookeeper https://zuul.opendev.org/t/openstack/build/54b0610493834fb98b69da76e3ff589d : SUCCESS in 5m 13s\n- tooz-tox-py310-redis https://zuul.opendev.org/t/openstack/build/5006682d7f374146acd28b8b6cecb079 : SUCCESS in 4m 07s\n- tooz-tox-py313-redis https://zuul.opendev.org/t/openstack/build/f5758acd1ad248a199d3bad0412ad963 : SUCCESS in 3m 17s\n- tooz-tox-py310-sentinel https://zuul.opendev.org/t/openstack/build/bc210b701c5e4ea79ce08d044e4b36b4 : SUCCESS in 3m 33s\n- tooz-tox-py313-sentinel https://zuul.opendev.org/t/openstack/build/0bf21fbdbaae4213add8608cb1ce0449 : SUCCESS in 5m 00s\n- tooz-tox-py310-memcached https://zuul.opendev.org/t/openstack/build/adf134427d004cb8b5ecebe6cba2dd96 : SUCCESS in 4m 43s\n- tooz-tox-py313-memcached https://zuul.opendev.org/t/openstack/build/33724bfe43d94487adbaa9dce9a11b31 : SUCCESS in 5m 44s\n- tooz-tox-py310-postgresql https://zuul.opendev.org/t/openstack/build/b71a3f31aa5449c4b79f1dddcbf24d08 : SUCCESS in 2m 41s\n- tooz-tox-py313-postgresql https://zuul.opendev.org/t/openstack/build/59f5c5303bd6461999a15a6ebb99318e : SUCCESS in 9m 27s\n- tooz-tox-py310-mysql https://zuul.opendev.org/t/openstack/build/aa124267d71141e1bc81ee09a2e028e8 : SUCCESS in 5m 54s\n- tooz-tox-py313-mysql https://zuul.opendev.org/t/openstack/build/1b36c1a3955d4673ac16b214da6d3444 : SUCCESS in 4m 13s\n- tooz-tox-py310-consul https://zuul.opendev.org/t/openstack/build/220403e0d0834b48bb052a28d556e72c : SUCCESS in 4m 41s\n- tooz-tox-py313-consul https://zuul.opendev.org/t/openstack/build/8566873171214f7fb2c2b43b8f8769fd : SUCCESS in 5m 50s","accounts_in_message":[],"_revision_number":5},{"id":"0eb0a7664463306d5f2522c715c0a527277fed32","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-13 10:41:37.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"cd5a51770418fc12517c2ac7aa8e1fb8472b535d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-13 10:43:54.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b3f315dfe0494225840b4b0c3f98bb68\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d152b38650214a20b7fdfda68a5a6753 : SUCCESS in 51s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6a43f72497724d4ba0bb33dd6e766c76 : SUCCESS in 46s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"d1b91ef8eb0b74f4ee0d4544cf491f5ccf562229","revisions":{"667e07025bd9fb54e4a191e7dcef2d427d8b8252":{"kind":"REWORK","_number":1,"created":"2026-02-09 17:04:58.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/50/976150/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/50/976150/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/50/976150/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/50/976150/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/50/976150/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/50/976150/1"}}},"commit":{"parents":[{"commit":"aa92b7f09923e91e8f0602600cd2ee48b3537775","subject":"Add ruff","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/aa92b7f09923e91e8f0602600cd2ee48b3537775"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-02-09 17:03:49.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-02-09 17:04:45.000000000","tz":540},"subject":"Remove zake driver","message":"Remove zake driver\n\nIt was deprecated due to the zake library abandoned. Now the library\ncan\u0027t be installed due to remaining usage of pkg_resources.\n\nChange-Id: I05607a8116eaf5dc18e5ea132eda6236966242ec\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/667e07025bd9fb54e4a191e7dcef2d427d8b8252"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/667e07025bd9fb54e4a191e7dcef2d427d8b8252"}]},"branch":"refs/heads/master"},"e31171e86013f14927202f7ecf10ca158cca0329":{"kind":"REWORK","_number":2,"created":"2026-02-11 05:22:22.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/50/976150/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/50/976150/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/50/976150/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/50/976150/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/50/976150/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/50/976150/2"}}},"commit":{"parents":[{"commit":"aa92b7f09923e91e8f0602600cd2ee48b3537775","subject":"Add ruff","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/aa92b7f09923e91e8f0602600cd2ee48b3537775"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-02-09 17:03:49.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-02-11 05:21:24.000000000","tz":540},"subject":"Remove zake driver","message":"Remove zake driver\n\nIt was deprecated due to the zake library abandoned. Now the library\ncan\u0027t be installed due to remaining usage of pkg_resources.\n\nTesting of example implementations is disabled now. To restore this\ntest coverage we need to set up the backend like redis or zookeper\nwhich supports the full features.\n\nChange-Id: I05607a8116eaf5dc18e5ea132eda6236966242ec\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/e31171e86013f14927202f7ecf10ca158cca0329"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/e31171e86013f14927202f7ecf10ca158cca0329"}]},"branch":"refs/heads/master"},"0a29681e05528baea02cd72f20a99af25ef7cc07":{"kind":"REWORK","_number":3,"created":"2026-02-11 05:24:24.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/50/976150/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/50/976150/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/50/976150/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/50/976150/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/50/976150/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/50/976150/3"}}},"commit":{"parents":[{"commit":"aa92b7f09923e91e8f0602600cd2ee48b3537775","subject":"Add ruff","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/aa92b7f09923e91e8f0602600cd2ee48b3537775"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-02-09 17:03:49.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-02-11 05:23:58.000000000","tz":540},"subject":"Remove zake driver","message":"Remove zake driver\n\nIt was deprecated due to the zake library abandoned. Now the library\ncan\u0027t be installed due to remaining usage of pkg_resources.\n\nTesting of example implementations is disabled now. To restore this\ntest coverage we need to set up the backend like redis or zookeper\nwhich supports the full features.\n\nChange-Id: I05607a8116eaf5dc18e5ea132eda6236966242ec\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/0a29681e05528baea02cd72f20a99af25ef7cc07"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/0a29681e05528baea02cd72f20a99af25ef7cc07"}]},"branch":"refs/heads/master"},"c964da49f519e6dad105c7e0e8f7fb112f3d7bf2":{"kind":"REWORK","_number":4,"created":"2026-02-11 11:29:32.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/50/976150/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/50/976150/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/50/976150/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/50/976150/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/50/976150/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/50/976150/4"}}},"commit":{"parents":[{"commit":"aa92b7f09923e91e8f0602600cd2ee48b3537775","subject":"Add ruff","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/aa92b7f09923e91e8f0602600cd2ee48b3537775"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-02-11 11:26:58.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-02-11 11:29:14.000000000","tz":540},"subject":"Remove zake driver","message":"Remove zake driver\n\nIt was deprecated due to the zake library abandoned. Now the library\ncan\u0027t be installed due to remaining usage of pkg_resources.\n\nChange-Id: I05607a8116eaf5dc18e5ea132eda6236966242ec\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/c964da49f519e6dad105c7e0e8f7fb112f3d7bf2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/c964da49f519e6dad105c7e0e8f7fb112f3d7bf2"}]},"branch":"refs/heads/master"},"d1b91ef8eb0b74f4ee0d4544cf491f5ccf562229":{"kind":"REWORK","_number":5,"created":"2026-02-11 11:31:04.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/50/976150/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/50/976150/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/50/976150/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/50/976150/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/50/976150/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/50/976150/5"}}},"commit":{"parents":[{"commit":"aa92b7f09923e91e8f0602600cd2ee48b3537775","subject":"Add ruff","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/aa92b7f09923e91e8f0602600cd2ee48b3537775"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-02-11 11:26:58.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-02-11 11:30:56.000000000","tz":540},"subject":"Remove zake driver","message":"Remove zake driver\n\nIt was deprecated due to the zake library abandoned. Now the library\ncan\u0027t be installed due to remaining usage of pkg_resources.\n\nChange-Id: I05607a8116eaf5dc18e5ea132eda6236966242ec\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/d1b91ef8eb0b74f4ee0d4544cf491f5ccf562229"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/d1b91ef8eb0b74f4ee0d4544cf491f5ccf562229"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
