)]}'
{"id":"openstack%2Ftooz~976838","triplet_id":"openstack%2Ftooz~master~Ib9d67fe69635814327695dbd328d9a8427960151","project":"openstack/tooz","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-02-18 14:09:55.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},"16137":{"account":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"last_update":"2026-02-18 16:00:50.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib9d67fe69635814327695dbd328d9a8427960151","subject":"Deprecate etcd driver","status":"MERGED","created":"2026-02-14 14:31:54.000000000","updated":"2026-02-18 16:01:50.000000000","submitted":"2026-02-18 16:00:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"976838","meta_rev_id":"c587fbc7c82078cbbf2afea19f303a4c7f273e87","_number":976838,"virtual_id_number":976838,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-02-18 16:00:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2026-02-16 11:42:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2026-02-18 14:09:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":1,"date":"2026-02-18 14:09:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-14 16:18:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-02-16 11:42:59.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2026-02-16 11:44:28.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"c3489223a785cebe9ab7f4cb1f613a00611d90e6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-02-14 14:31:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d6cbcb4f3ad586ead3127051fffdccdaad501c15","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-14 16:18:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9ade53b5b81942ee8f97e00bc78e2269\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/600c9c2500884d65805d8e54e2a66bf7 : SUCCESS in 1h 41m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fe63c7ccf53b4544a9927880ff7eb3ff : SUCCESS in 5m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/33c9af3ef1cc42d4a326b6c876b20051 : SUCCESS in 3m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1fb78cb2d727427eb524e69bcda0f917 : SUCCESS in 2m 58s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/55584e6aa9774a42a964d252024b0d32 : SUCCESS in 4m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e41360603a14a73a580e7139c501752 : SUCCESS in 4m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/47d4b70c1f7c425caac025a573d8a7b0 : SUCCESS in 3m 07s\n- tooz-tox-examples https://zuul.opendev.org/t/openstack/build/6df37260b3be41f6be69949e41ff9b58 : SUCCESS in 4m 31s\n- tooz-tox-py310-etcd3gw https://zuul.opendev.org/t/openstack/build/c0d7076ea6e4422aa392234b3006f500 : SUCCESS in 3m 22s\n- tooz-tox-py313-etcd3gw https://zuul.opendev.org/t/openstack/build/9b64bf47bc4e408886ecd1d3bd204ca7 : SUCCESS in 6m 37s\n- tooz-tox-py310-zookeeper https://zuul.opendev.org/t/openstack/build/ce72598a022e4292a7ab0e0319cc18e1 : SUCCESS in 4m 28s\n- tooz-tox-py313-zookeeper https://zuul.opendev.org/t/openstack/build/303d0f22fa0d45b9866d8b1b513118c0 : SUCCESS in 4m 42s\n- tooz-tox-py310-redis https://zuul.opendev.org/t/openstack/build/2dde19ad69e2490c8362a7260b10a71f : SUCCESS in 2m 56s\n- tooz-tox-py313-redis https://zuul.opendev.org/t/openstack/build/3aac114d14614c45b19b76707b848e1f : SUCCESS in 5m 25s\n- tooz-tox-py310-sentinel https://zuul.opendev.org/t/openstack/build/4eaa2d88ffaa44a392cb89d7326f79d2 : SUCCESS in 4m 10s\n- tooz-tox-py313-sentinel https://zuul.opendev.org/t/openstack/build/2b0744a057d34703a2d972fce31dfd11 : SUCCESS in 4m 58s\n- tooz-tox-py310-memcached https://zuul.opendev.org/t/openstack/build/26e7a9825e1446eab3cc1d49211091bd : SUCCESS in 4m 19s\n- tooz-tox-py313-memcached https://zuul.opendev.org/t/openstack/build/18235aea9550429589ee9a4fa9a256d1 : SUCCESS in 4m 43s\n- tooz-tox-py310-postgresql https://zuul.opendev.org/t/openstack/build/886c820fd0374ca2989506abff746fdd : SUCCESS in 3m 49s\n- tooz-tox-py313-postgresql https://zuul.opendev.org/t/openstack/build/a80d361c0b0b4a65a25f4dc7ff91441d : SUCCESS in 5m 42s\n- tooz-tox-py310-mysql https://zuul.opendev.org/t/openstack/build/8cfccb0249f04185ac1d5a96cd5323f9 : SUCCESS in 3m 48s\n- tooz-tox-py313-mysql https://zuul.opendev.org/t/openstack/build/412f882cf627435194add810e74adf6f : SUCCESS in 7m 02s\n- tooz-tox-py310-consul https://zuul.opendev.org/t/openstack/build/52f537368529441f9d5f932cda645c58 : SUCCESS in 4m 26s\n- tooz-tox-py313-consul https://zuul.opendev.org/t/openstack/build/d9fa40d133aa411187c389590819758b : SUCCESS in 3m 34s","accounts_in_message":[],"_revision_number":1},{"id":"57e5e833b035987388684d4e864a042db2163ddf","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-02-16 11:42:59.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ef8cd72ecc3b2e91bdedd7b0a6b7a20dd8e8aa6f","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-02-18 14:09:55.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c98ccd691a19ba7029f979e23bedf42d101f4bfe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-18 14:10:15.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"40eb4d8de1664b79552fca4daf442b1fd5503e50","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-18 16:00:50.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a269d06d8728487482772b7a6f2bbff5\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ebd75bd6f7e84f868aa263e4ef8deb92 : SUCCESS in 1h 45m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c369f22359c64ea48b3b051ded67f0c5 : SUCCESS in 4m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/721a61d69780465b8c9f450478742f25 : SUCCESS in 4m 50s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8c67adebc5d54ca08f9700ee5b40de34 : SUCCESS in 5m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a6e47e00dd2e4d14bb78d577fa17b454 : SUCCESS in 7m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2332115b8153424f8a764f0eb49c7787 : SUCCESS in 4m 30s\n- tooz-tox-examples https://zuul.opendev.org/t/openstack/build/395f9acc37c240168267b4af3fcda590 : SUCCESS in 3m 29s\n- tooz-tox-py310-etcd3gw https://zuul.opendev.org/t/openstack/build/a31fc148565b4a6aa405ce7b1ae0845c : SUCCESS in 5m 00s\n- tooz-tox-py313-etcd3gw https://zuul.opendev.org/t/openstack/build/53d32b39009a4e4a9a125d0a8760b86e : SUCCESS in 5m 28s\n- tooz-tox-py310-zookeeper https://zuul.opendev.org/t/openstack/build/ff85050e8e3348eb9320ce6220baa46d : SUCCESS in 4m 47s\n- tooz-tox-py313-zookeeper https://zuul.opendev.org/t/openstack/build/08a48b9571aa4e5791d2333a18b34933 : SUCCESS in 5m 45s\n- tooz-tox-py310-redis https://zuul.opendev.org/t/openstack/build/2b6f2997765b40d481022f08576cf2e1 : SUCCESS in 4m 38s\n- tooz-tox-py313-redis https://zuul.opendev.org/t/openstack/build/64118647e5b64e3fbbc7b541d14ec786 : SUCCESS in 5m 52s\n- tooz-tox-py310-sentinel https://zuul.opendev.org/t/openstack/build/5fc0f069c61d4655884b24e7b8c13a5b : SUCCESS in 5m 12s\n- tooz-tox-py313-sentinel https://zuul.opendev.org/t/openstack/build/2e6af1572622449a81395edf4fd5568f : SUCCESS in 4m 38s\n- tooz-tox-py310-memcached https://zuul.opendev.org/t/openstack/build/968d3be3a0824135b589f272b0ca2e2f : SUCCESS in 4m 55s\n- tooz-tox-py313-memcached https://zuul.opendev.org/t/openstack/build/c3647930b810417f9d0e3d05225a2558 : SUCCESS in 5m 19s\n- tooz-tox-py310-postgresql https://zuul.opendev.org/t/openstack/build/a90494f2b99844778a71a71ef85c4176 : SUCCESS in 6m 03s\n- tooz-tox-py313-postgresql https://zuul.opendev.org/t/openstack/build/84cde4f504ce49a0a41b9594d209b301 : SUCCESS in 4m 59s\n- tooz-tox-py310-mysql https://zuul.opendev.org/t/openstack/build/8b4bb98ce85a4c0bbc575c7c23b61e2b : SUCCESS in 6m 26s\n- tooz-tox-py313-mysql https://zuul.opendev.org/t/openstack/build/21facfc4aba249cab0412164bee203f2 : SUCCESS in 4m 36s\n- tooz-tox-py310-consul https://zuul.opendev.org/t/openstack/build/46287f7960a144ed816c1b44894d0cf1 : SUCCESS in 4m 19s\n- tooz-tox-py313-consul https://zuul.opendev.org/t/openstack/build/d4b7da85bd2144e1bc73f98f7f7b5ab2 : SUCCESS in 8m 02s","accounts_in_message":[],"_revision_number":1},{"id":"306fd9d74956ccbffbc1762fc56805be14d1b389","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-18 16:00:50.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"c587fbc7c82078cbbf2afea19f303a4c7f273e87","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-18 16:01:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/056355ba2692401eb150d3254eec7674\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc444b89006643788a3616d8abac6234 : SUCCESS in 44s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/03267145203244b98a2d75e5d3768a0b : SUCCESS in 40s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8a6270de6ee3aefe1aa0f4fb6c98ac5b363570c5","revisions":{"8a6270de6ee3aefe1aa0f4fb6c98ac5b363570c5":{"kind":"REWORK","_number":1,"created":"2026-02-14 14:31:54.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/38/976838/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/38/976838/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/38/976838/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/38/976838/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/38/976838/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/38/976838/1"}}},"commit":{"parents":[{"commit":"d1b91ef8eb0b74f4ee0d4544cf491f5ccf562229","subject":"Remove zake driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/d1b91ef8eb0b74f4ee0d4544cf491f5ccf562229"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-02-14 14:26:08.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-02-14 14:31:44.000000000","tz":540},"subject":"Deprecate etcd driver","message":"Deprecate etcd driver\n\nThe driver supports only v2 API, which was removed in etcd 3.6 .\n\nChange-Id: Ib9d67fe69635814327695dbd328d9a8427960151\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/8a6270de6ee3aefe1aa0f4fb6c98ac5b363570c5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/8a6270de6ee3aefe1aa0f4fb6c98ac5b363570c5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
