)]}'
{"id":"openstack%2Ftooz~979667","triplet_id":"openstack%2Ftooz~master~I4f250efae66b431cfe15453507fe4b507078074e","project":"openstack/tooz","branch":"master","topic":"linter-sync","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2026-03-09 15:23:52.000000000","reason":"\u003cGERRIT_ACCOUNT_15334\u003e replied on the change","reason_account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}}},"hashtags":[],"change_id":"I4f250efae66b431cfe15453507fe4b507078074e","subject":"ruff: Configure hacking as external linter","status":"MERGED","created":"2026-03-09 14:20:22.000000000","updated":"2026-03-09 19:34:23.000000000","submitted":"2026-03-09 19:33:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"979667-linter-sync","meta_rev_id":"2e7cb014fee6f10f60ea10697dd59c8e55e503f6","_number":979667,"virtual_id_number":979667,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-03-09 19:33:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2026-03-09 18:28:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2026-03-09 18:28:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-09 15:15:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a905e1d35d871580ec7d2deab74b854c651e6201","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-03-09 14:20:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"62fe8d224c7697f2be9c4b73a0940f55a17acfc1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-09 15:15:44.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e24f0b604ea246e8b092ede752910cfa\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/7056fe5ad87447199a0b676477cfbb22 : SUCCESS in 2m 52s\n- requirements-check https://zuul.opendev.org/t/openstack/build/209f7fd24ba34d228fedc80187f1ba84 : SUCCESS in 2m 25s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5a505bdd503b4f4eac581e50fc82c5da : FAILURE in 53m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/18896fd4a7664448adf3ae5cb6c7f9b7 : SUCCESS in 5m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e9fd30d217f14bfd9707aa382d392df4 : SUCCESS in 3m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a8ae324529684fec945bd8792cb0872b : SUCCESS in 4m 33s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9507fe5e86a04d578b486ca180f77643 : SUCCESS in 5m 21s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/36c30c0739184344b77ec83eeaf20768 : FAILURE in 5m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/59da48a5395746739c562f0d44ef1d19 : SUCCESS in 5m 06s\n- tooz-tox-examples https://zuul.opendev.org/t/openstack/build/fe21728b46e2408f88d1e7f1a47520b0 : SUCCESS in 4m 39s\n- tooz-tox-py310-etcd3gw https://zuul.opendev.org/t/openstack/build/96d298fe87564e9eb68178f29b43e6d9 : SUCCESS in 3m 45s\n- tooz-tox-py313-etcd3gw https://zuul.opendev.org/t/openstack/build/177c27e36eb64dcc9b3276ee815c1b3f : SUCCESS in 4m 12s\n- tooz-tox-py310-zookeeper https://zuul.opendev.org/t/openstack/build/5e60c9da8dd64c0092099ac5b1ce21fb : SUCCESS in 3m 26s\n- tooz-tox-py313-zookeeper https://zuul.opendev.org/t/openstack/build/2911c95402bd40c6b2864f6a5d111d13 : SUCCESS in 5m 15s\n- tooz-tox-py310-redis https://zuul.opendev.org/t/openstack/build/eababf468667499f9cdd76a87efb6753 : SUCCESS in 5m 52s\n- tooz-tox-py313-redis https://zuul.opendev.org/t/openstack/build/70cdf39ef15a40eea11b9f4a0167104f : SUCCESS in 7m 59s\n- tooz-tox-py310-sentinel https://zuul.opendev.org/t/openstack/build/2d87ee31383543b1b941bd7c6cc01cab : SUCCESS in 4m 52s\n- tooz-tox-py313-sentinel https://zuul.opendev.org/t/openstack/build/e5f70097b2694ac0a82fd540d45290d0 : SUCCESS in 6m 44s\n- tooz-tox-py310-memcached https://zuul.opendev.org/t/openstack/build/ff6107e5c22c4da0b440f3873084ae6a : SUCCESS in 4m 09s\n- tooz-tox-py313-memcached https://zuul.opendev.org/t/openstack/build/9be82eb722744e57bc2375a4206861a9 : SUCCESS in 6m 30s\n- tooz-tox-py310-postgresql https://zuul.opendev.org/t/openstack/build/70a91b54f37740eeacfed97cde3af2a5 : SUCCESS in 4m 13s\n- tooz-tox-py313-postgresql https://zuul.opendev.org/t/openstack/build/22da74de75e8477c9bcf4fd73767d3b2 : SUCCESS in 11m 23s\n- tooz-tox-py310-mysql https://zuul.opendev.org/t/openstack/build/0bf1b267bfa647679eb213a30f41e3d6 : SUCCESS in 4m 25s\n- tooz-tox-py313-mysql https://zuul.opendev.org/t/openstack/build/de16de0284164c2d903fe155444a52f0 : SUCCESS in 9m 29s\n- tooz-tox-py310-consul https://zuul.opendev.org/t/openstack/build/bc915246429b4bcc99c7126a732a6cf0 : SUCCESS in 4m 36s\n- tooz-tox-py313-consul https://zuul.opendev.org/t/openstack/build/dc70a6ab33c1489f9860cd1f6c53dac4 : SUCCESS in 5m 54s","accounts_in_message":[],"_revision_number":1},{"id":"9e2afdd683b2a13d95c31193b29ba6ac053abb2c","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-03-09 15:23:52.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6b29cc2b0a4e091e5805cf37bfbeea2ac3266592","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-09 16:20:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/65b1f62834a9493184129f4630899492\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/fa1331928ac84c618504b9fe2360a0f8 : SUCCESS in 2m 01s\n- requirements-check https://zuul.opendev.org/t/openstack/build/14a6db1e9244492c97edabc451c4e123 : SUCCESS in 3m 28s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/063a73ddac3e46d29f79a14feba1dbc7 : SUCCESS in 54m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ab40f6ff874146d7aef140e9bd7ae575 : SUCCESS in 7m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f6f00a6eed1a4a10b468561121c16b49 : SUCCESS in 4m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6925f8f4be0f4ba4bb65a6e119e0a41b : SUCCESS in 4m 35s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a44992eeec3c486db02b8ee1c9e31f8d : SUCCESS in 5m 23s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/03ea2c18575947de9a753b0e49dc24c4 : FAILURE in 7m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/85f259ab21ea4bfe893a0d79b10986a8 : SUCCESS in 5m 05s\n- tooz-tox-examples https://zuul.opendev.org/t/openstack/build/348c9becd57347ef82974bc35c975b91 : SUCCESS in 5m 37s\n- tooz-tox-py310-etcd3gw https://zuul.opendev.org/t/openstack/build/93a64aacac714a34b28f27fb12f1f668 : SUCCESS in 7m 13s\n- tooz-tox-py313-etcd3gw https://zuul.opendev.org/t/openstack/build/a6aa716b0f4c4d44a46c1722e742ea16 : SUCCESS in 5m 32s\n- tooz-tox-py310-zookeeper https://zuul.opendev.org/t/openstack/build/5ac74315e42d45eda6c1ec1e6ff120d4 : SUCCESS in 4m 08s\n- tooz-tox-py313-zookeeper https://zuul.opendev.org/t/openstack/build/02a320e6b6f24fe79d2f99b80219d777 : SUCCESS in 5m 43s\n- tooz-tox-py310-redis https://zuul.opendev.org/t/openstack/build/dbd6cf9c2a6e4c6aa8b3e8a666c8cf56 : SUCCESS in 4m 59s\n- tooz-tox-py313-redis https://zuul.opendev.org/t/openstack/build/23edb2fe3aaf44339abe9d229b622bf9 : SUCCESS in 9m 11s\n- tooz-tox-py310-sentinel https://zuul.opendev.org/t/openstack/build/4b370dea05e14f3dbe82f647d8833078 : SUCCESS in 4m 49s\n- tooz-tox-py313-sentinel https://zuul.opendev.org/t/openstack/build/c8ccd6310f2142a387173a6dd149bf52 : SUCCESS in 7m 26s\n- tooz-tox-py310-memcached https://zuul.opendev.org/t/openstack/build/3747c37f90a4412e9789c31efcfc44d7 : SUCCESS in 4m 46s\n- tooz-tox-py313-memcached https://zuul.opendev.org/t/openstack/build/6758c0b7567643cca4d3eda36d4f2e81 : SUCCESS in 5m 04s\n- tooz-tox-py310-postgresql https://zuul.opendev.org/t/openstack/build/c361d951871c4afd923e38dec07307e0 : SUCCESS in 2m 41s\n- tooz-tox-py313-postgresql https://zuul.opendev.org/t/openstack/build/8d6cffbe2125403e8c3843d4c1860dcc : SUCCESS in 4m 28s\n- tooz-tox-py310-mysql https://zuul.opendev.org/t/openstack/build/ea7d44d001f642d480869c512c5e2647 : SUCCESS in 4m 42s\n- tooz-tox-py313-mysql https://zuul.opendev.org/t/openstack/build/b8da8001782445869d8bebf6d7aa60ba : SUCCESS in 9m 20s\n- tooz-tox-py310-consul https://zuul.opendev.org/t/openstack/build/b86de74468a84c32a8fd94badc54841d : SUCCESS in 3m 07s\n- tooz-tox-py313-consul https://zuul.opendev.org/t/openstack/build/7824919877134a2782c59b9fbdfd59ec : SUCCESS in 4m 40s","accounts_in_message":[],"_revision_number":1},{"id":"33dc90e47e36d4edb82822dbfb63ff4ba6b089b6","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-03-09 18:28:01.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5cbf3e22f7c438c12994426f1dae8cb4a60377ee","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-09 18:28:20.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"737689cec7a236a1a504ae304816e91602fe6d6f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-09 19:33:23.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/208629d2f1be4166aedce7afd47fe783\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/6237b53b21e8465d99baba29917a6103 : SUCCESS in 2m 16s\n- requirements-check https://zuul.opendev.org/t/openstack/build/8de7a66b63d7431dbc7fbd7f35bcfaf4 : SUCCESS in 3m 24s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/42009fadf21346b9bf1dccc63ca89cf3 : SUCCESS in 1h 04m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d16a39a793c84e2db278c75aad79e26f : SUCCESS in 6m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2a69b2b427974cff80f2d7ea2518522c : SUCCESS in 3m 57s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3d84ee571e1f471981485740964222c6 : SUCCESS in 7m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/06dff186e9714bd4ab39586550e129fc : SUCCESS in 3m 48s\n- tooz-tox-examples https://zuul.opendev.org/t/openstack/build/1c5f14bd900645458ee47fd0bfe59045 : SUCCESS in 3m 54s\n- tooz-tox-py310-etcd3gw https://zuul.opendev.org/t/openstack/build/1443b6e013cc4439ba37bbc6da5382aa : SUCCESS in 6m 39s\n- tooz-tox-py313-etcd3gw https://zuul.opendev.org/t/openstack/build/66a0994c89fb43ae86862c0ab260937a : SUCCESS in 5m 37s\n- tooz-tox-py310-zookeeper https://zuul.opendev.org/t/openstack/build/c2aca24c5dda42c79f60e5bfae5c3cf3 : SUCCESS in 4m 19s\n- tooz-tox-py313-zookeeper https://zuul.opendev.org/t/openstack/build/f3662e0cfce94c2fa8d6e9f6eab6ce79 : SUCCESS in 4m 48s\n- tooz-tox-py310-redis https://zuul.opendev.org/t/openstack/build/04de9463c1d44d96abd8d51f10eb34d4 : SUCCESS in 6m 07s\n- tooz-tox-py313-redis https://zuul.opendev.org/t/openstack/build/974319361cd04ae9a20c6a2c4e22cc8e : SUCCESS in 6m 42s\n- tooz-tox-py310-sentinel https://zuul.opendev.org/t/openstack/build/df19edcd4b154dbeb1394c4313da270c : SUCCESS in 5m 50s\n- tooz-tox-py313-sentinel https://zuul.opendev.org/t/openstack/build/2b2ffd82180046fab701a7d7fdbea09c : SUCCESS in 6m 53s\n- tooz-tox-py310-memcached https://zuul.opendev.org/t/openstack/build/46fe19df6d914b20b080683b8bddd70a : SUCCESS in 4m 42s\n- tooz-tox-py313-memcached https://zuul.opendev.org/t/openstack/build/4cb60f79970e4d08a4fed478de8758aa : SUCCESS in 4m 46s\n- tooz-tox-py310-postgresql https://zuul.opendev.org/t/openstack/build/e2725870976141b382790da5bc99c21f : SUCCESS in 6m 07s\n- tooz-tox-py313-postgresql https://zuul.opendev.org/t/openstack/build/6acb6addaffc46518ec2bd5ab952cc83 : SUCCESS in 5m 47s\n- tooz-tox-py310-mysql https://zuul.opendev.org/t/openstack/build/c9dba0b3d96e4e198716a89f2a6280ab : SUCCESS in 4m 22s\n- tooz-tox-py313-mysql https://zuul.opendev.org/t/openstack/build/fef2c0cc55754a16bab9a83971647ba7 : SUCCESS in 6m 49s\n- tooz-tox-py310-consul https://zuul.opendev.org/t/openstack/build/d27f25d6f82a4a99bf973661dae81474 : SUCCESS in 5m 50s\n- tooz-tox-py313-consul https://zuul.opendev.org/t/openstack/build/ff53ee1eeb15443ba7cae4d0139407d2 : SUCCESS in 4m 11s","accounts_in_message":[],"_revision_number":1},{"id":"61bbb8a67a9a084e4627adbad37b1a648810be04","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-09 19:33:23.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"2e7cb014fee6f10f60ea10697dd59c8e55e503f6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-09 19:34:23.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b9bea76917d44dfbadce6de18533c289\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f9bd9a29865c40feb15ad4855c9a1e33 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"42eb79cf639783f1871e301a5b7255d58365c1bf","revisions":{"42eb79cf639783f1871e301a5b7255d58365c1bf":{"kind":"REWORK","_number":1,"created":"2026-03-09 14:20:22.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/67/979667/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/67/979667/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/67/979667/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/67/979667/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/67/979667/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/67/979667/1"}}},"commit":{"parents":[{"commit":"6a50eebef8bf4959048a64f8b76f05aa00424119","subject":"Merge \"Deprecate Consul driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/6a50eebef8bf4959048a64f8b76f05aa00424119"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-03-09 14:20:08.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-03-09 14:20:08.000000000","tz":0},"subject":"ruff: Configure hacking as external linter","message":"ruff: Configure hacking as external linter\n\nChange-Id: I4f250efae66b431cfe15453507fe4b507078074e\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/42eb79cf639783f1871e301a5b7255d58365c1bf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/42eb79cf639783f1871e301a5b7255d58365c1bf"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
