)]}'
{"id":"openstack%2Ftooz~980106","triplet_id":"openstack%2Ftooz~master~Iab667d5c9dc914095c715a645d16223c5815b968","project":"openstack/tooz","branch":"master","topic":"typing","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2026-03-14 14:41:51.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Iab667d5c9dc914095c715a645d16223c5815b968","subject":"typing: Add client, session_id properties for consul driver","status":"ABANDONED","created":"2026-03-11 15:42:02.000000000","updated":"2026-03-14 14:41:51.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"b2822e3e66a0d86686cd1567a35ff15757ebe2d0","_number":980106,"virtual_id_number":980106,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-03-12 10:38:09.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-11 16:48:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"775d7ae103c01ad92f312388f50e081c893790aa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-03-11 15:42:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7b745e7c97b57a9ce31d554c32f676295cf38ac9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-11 16:48:28.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/75a17b405882411d9f407d512ad4656f\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2fe96bbebe604397b5c2c6a1e223689f : SUCCESS in 54m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/16fc9cf5185442108d177ccbc90acce5 : SUCCESS in 6m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c4f660660f934c3eb72c8de7a4166189 : SUCCESS in 5m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/daff37c1f3d842abba5119abe0e0de5a : SUCCESS in 5m 02s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ac00374bbf6342608c17951f4bc07b50 : SUCCESS in 4m 48s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/c8bd3649fcbd4d24be937f654a423da5 : SUCCESS in 7m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/81d2fee14d014a51bdfdf4d900f73cf5 : SUCCESS in 4m 31s\n- tooz-tox-examples https://zuul.opendev.org/t/openstack/build/840d353f4d004eb18821b03f47380283 : SUCCESS in 4m 37s\n- tooz-tox-py310-etcd3gw https://zuul.opendev.org/t/openstack/build/36ca034c7ea14287843cc8855f71810d : FAILURE in 6m 31s\n- tooz-tox-py313-etcd3gw https://zuul.opendev.org/t/openstack/build/d9d5f71c26a04121908bd5269cbaee3e : FAILURE in 5m 47s\n- tooz-tox-py310-zookeeper https://zuul.opendev.org/t/openstack/build/3e3b9ae9750e4d928116bf8852c838ce : SUCCESS in 5m 29s\n- tooz-tox-py313-zookeeper https://zuul.opendev.org/t/openstack/build/fe79039d7b9a4fdaa18233b20a928f20 : SUCCESS in 4m 21s\n- tooz-tox-py310-redis https://zuul.opendev.org/t/openstack/build/0f5082c5343049c69f4dfd0f93960e3b : SUCCESS in 5m 57s\n- tooz-tox-py313-redis https://zuul.opendev.org/t/openstack/build/342768e06d774feb9c6a82e36070084b : SUCCESS in 5m 57s\n- tooz-tox-py310-sentinel https://zuul.opendev.org/t/openstack/build/f72ec12778de47b18b9427ce404c1018 : SUCCESS in 9m 26s\n- tooz-tox-py313-sentinel https://zuul.opendev.org/t/openstack/build/42e05f5fc7c1484da2f4f2b1fca2a7f1 : SUCCESS in 8m 58s\n- tooz-tox-py310-memcached https://zuul.opendev.org/t/openstack/build/fe3108db82424df182204c7d1a7967c9 : SUCCESS in 4m 21s\n- tooz-tox-py313-memcached https://zuul.opendev.org/t/openstack/build/0d8680b773b8450f890c04d09682a1a9 : SUCCESS in 3m 56s\n- tooz-tox-py310-postgresql https://zuul.opendev.org/t/openstack/build/b0e9d9bd666e49cbaf04133a94b932b1 : SUCCESS in 4m 34s\n- tooz-tox-py313-postgresql https://zuul.opendev.org/t/openstack/build/7bef1756d24b4313bbce8f455566346a : SUCCESS in 4m 46s\n- tooz-tox-py310-mysql https://zuul.opendev.org/t/openstack/build/4c88d0c7e53643b3a2032d151b224983 : SUCCESS in 4m 46s\n- tooz-tox-py313-mysql https://zuul.opendev.org/t/openstack/build/7a0e837874d64e19aa8e91cfcbe0feb8 : SUCCESS in 6m 07s\n- tooz-tox-py310-consul https://zuul.opendev.org/t/openstack/build/f9335d21b74e4db8921dd82ee3fe55dc : SUCCESS in 4m 46s\n- tooz-tox-py313-consul https://zuul.opendev.org/t/openstack/build/127fcfa67ae64a8d8e739e8aa11ebf53 : SUCCESS in 4m 52s","accounts_in_message":[],"_revision_number":1},{"id":"89bb003405433494fef8dcc81bf1232b13c3e3d1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-03-12 08:42:07.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"7b65fb445f1252c600499a7bc21dc0ef4258f4b1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-12 10:38:09.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/091b5bc8f55249adb74154d7691d9f2d\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/130a6c8b5c984c8c8eef5ed4b71796e8 : SUCCESS in 1h 35m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/44c6bbaa609041c19b45ac0b3c5e4cb2 : SUCCESS in 3m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/74ade76fd31b403594b7424d40f81013 : SUCCESS in 5m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9b3cadb1071d4deb9f9a06648d36d41b : SUCCESS in 4m 27s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2cc23d1b2683422c835eedebfdd034c6 : SUCCESS in 7m 15s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/2ab75eda7b3447e580840e914689f214 : SUCCESS in 9m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/03a04683d7b54e76b308fd26be3c7f16 : SUCCESS in 5m 16s\n- tooz-tox-examples https://zuul.opendev.org/t/openstack/build/8f59336bf0d84b47855f1758e9b1b54f : SUCCESS in 4m 44s\n- tooz-tox-py310-etcd3gw https://zuul.opendev.org/t/openstack/build/549d2b4883da4e408d4c0be21c30b84d : SUCCESS in 5m 11s\n- tooz-tox-py313-etcd3gw https://zuul.opendev.org/t/openstack/build/3f4209382947435ab35fbaf6043257e9 : SUCCESS in 5m 13s\n- tooz-tox-py310-zookeeper https://zuul.opendev.org/t/openstack/build/f600d42aea954b56a3089944b8cf67d8 : SUCCESS in 4m 39s\n- tooz-tox-py313-zookeeper https://zuul.opendev.org/t/openstack/build/a072807b5c054417827bfe425ca46dad : SUCCESS in 3m 28s\n- tooz-tox-py310-redis https://zuul.opendev.org/t/openstack/build/96c91483eccb44409160ff74e8b6dddd : SUCCESS in 4m 55s\n- tooz-tox-py313-redis https://zuul.opendev.org/t/openstack/build/963eba13b1444001866cf476c0823607 : SUCCESS in 6m 28s\n- tooz-tox-py310-sentinel https://zuul.opendev.org/t/openstack/build/6b4b45078da94ed4a30bd8f30acee679 : SUCCESS in 5m 54s\n- tooz-tox-py313-sentinel https://zuul.opendev.org/t/openstack/build/9e85c92a115c43d18631f23df6fadde0 : SUCCESS in 8m 59s\n- tooz-tox-py310-memcached https://zuul.opendev.org/t/openstack/build/21e185705a0f4cf29e6489f6e979e6c6 : SUCCESS in 3m 24s\n- tooz-tox-py313-memcached https://zuul.opendev.org/t/openstack/build/9c3f2484b39d48bd99787467cca0ec5a : SUCCESS in 5m 02s\n- tooz-tox-py310-postgresql https://zuul.opendev.org/t/openstack/build/3c1cb710fd95489999486426af0931cf : SUCCESS in 3m 45s\n- tooz-tox-py313-postgresql https://zuul.opendev.org/t/openstack/build/a2f57801d4ef40ad8830591aaf358a2e : SUCCESS in 5m 24s\n- tooz-tox-py310-mysql https://zuul.opendev.org/t/openstack/build/5ec1ad64f299466bb74cac7e2297b475 : SUCCESS in 2m 57s\n- tooz-tox-py313-mysql https://zuul.opendev.org/t/openstack/build/b8423b1067d54e78a4cfc65da0decfb3 : SUCCESS in 7m 24s\n- tooz-tox-py310-consul https://zuul.opendev.org/t/openstack/build/52cd0efb1cc44be1b4318ecfef1700f3 : SUCCESS in 5m 02s\n- tooz-tox-py313-consul https://zuul.opendev.org/t/openstack/build/8df26917daa844039ecabd30eec0a045 : SUCCESS in 3m 40s","accounts_in_message":[],"_revision_number":2},{"id":"b2822e3e66a0d86686cd1567a35ff15757ebe2d0","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-03-14 14:41:51.000000000","message":"Abandoned\n\nLet\u0027s just drop the driver instead","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"6609c3214f1020dce06916ff0039522ac13a0c22","revisions":{"2d95cffe8226ba09c827a97285bdbdcb6c84652d":{"kind":"REWORK","_number":1,"created":"2026-03-11 15:42:02.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/06/980106/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/06/980106/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/06/980106/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/06/980106/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/06/980106/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/06/980106/1"}}},"commit":{"parents":[{"commit":"eebb01aba4bb0e7489769adf5682a696b34d27c5","subject":"typing: Normalize type of lock name","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/eebb01aba4bb0e7489769adf5682a696b34d27c5"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-03-11 13:54:29.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-03-11 15:41:56.000000000","tz":0},"subject":"typing: Add client, session_id properties for consul driver","message":"typing: Add client, session_id properties for consul driver\n\nEven though this driver is deprecated, we want to be able to type the\nentire code base and need to fix this issue, where mypy thinks _client\ncan be None due to it not being initialized in __init__.\n\nChange-Id: Iab667d5c9dc914095c715a645d16223c5815b968\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/2d95cffe8226ba09c827a97285bdbdcb6c84652d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/2d95cffe8226ba09c827a97285bdbdcb6c84652d"}]},"branch":"refs/heads/master"},"6609c3214f1020dce06916ff0039522ac13a0c22":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-03-12 08:42:07.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/06/980106/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/06/980106/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/06/980106/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/06/980106/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/06/980106/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/06/980106/2"}}},"commit":{"parents":[{"commit":"0e779f164f3aba36532a59db8c2154a3212136ae","subject":"typing: Normalize type of lock name","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/0e779f164f3aba36532a59db8c2154a3212136ae"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-03-11 13:54:29.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-03-12 08:40:21.000000000","tz":0},"subject":"typing: Add client, session_id properties for consul driver","message":"typing: Add client, session_id properties for consul driver\n\nEven though this driver is deprecated, we want to be able to type the\nentire code base and need to fix this issue, where mypy thinks _client\ncan be None due to it not being initialized in __init__.\n\nChange-Id: Iab667d5c9dc914095c715a645d16223c5815b968\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/6609c3214f1020dce06916ff0039522ac13a0c22"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/6609c3214f1020dce06916ff0039522ac13a0c22"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
