)]}'
{"id":"openstack%2Ftooz~980107","triplet_id":"openstack%2Ftooz~master~I745cb15cc269a032f7cea2c84c5861de79885d39","project":"openstack/tooz","branch":"master","topic":"typing","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2026-04-10 18:17:35.000000000","reason":"\u003cGERRIT_ACCOUNT_15334\u003e replied on the change","reason_account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-04-10 22:38:41.000000000","reason":"Change was submitted"},"16137":{"account":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"last_update":"2026-04-10 22:38:41.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I745cb15cc269a032f7cea2c84c5861de79885d39","subject":"typing: Add type hints to tooz.utils","status":"MERGED","created":"2026-03-11 15:42:02.000000000","updated":"2026-04-10 22:39:39.000000000","submitted":"2026-04-10 22:38:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"980107-typing","meta_rev_id":"282afdb0f74d2c846353b6dc1c5d4c1eba74aefd","_number":980107,"virtual_id_number":980107,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-04-10 22:38:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2026-04-10 18:17:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2026-04-10 18:17:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-11 17:00:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-03-31 09:28:43.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2026-03-31 09:28:43.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"5e0d61e8242e17829375ec441961b7ed13ce680d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-03-11 15:42:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f16c85c9347eae39cb087c3bbd16cd93efe52717","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-11 17:00:34.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/505b69d8cc6f445595dbd668e2e7e13b\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/1f624b0edf48420c87a25ad415e9033c : SUCCESS in 2m 24s\n- requirements-check https://zuul.opendev.org/t/openstack/build/75d7e6748a7248a8a342da73e2b98403 : SUCCESS in 3m 04s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e2dcd09e9afc4c7f8d19d37051cc4bc1 : SUCCESS in 1h 00m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fc10a779e1bd4b0ab494ca639a29a9f8 : SUCCESS in 5m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7a7e43f72d9443019fcb96f2b3b35c74 : SUCCESS in 4m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6772ad4e53c34a469e1512347fc08c22 : SUCCESS in 5m 07s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9f090f02b0ee42f4b9c576ca9deee6e5 : SUCCESS in 5m 07s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/52379bf8367945b290e21cf913f3cf30 : SUCCESS in 8m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a028a73a63134c518fccb6d04d6a7f1c : SUCCESS in 5m 05s\n- tooz-tox-examples https://zuul.opendev.org/t/openstack/build/5a5ba5789f3844b79e9745d82146439d : SUCCESS in 4m 39s\n- tooz-tox-py310-etcd3gw https://zuul.opendev.org/t/openstack/build/3f8132a5491846f38850ea26f94a539b : FAILURE in 5m 03s\n- tooz-tox-py313-etcd3gw https://zuul.opendev.org/t/openstack/build/cb7468845403479d8e1bf4b5518d6a79 : FAILURE in 5m 23s\n- tooz-tox-py310-zookeeper https://zuul.opendev.org/t/openstack/build/2176cb9a680442c7b4e2a5784315fa83 : SUCCESS in 4m 39s\n- tooz-tox-py313-zookeeper https://zuul.opendev.org/t/openstack/build/61c17df2b02a4d1fbbc9e9ed0bea1e59 : SUCCESS in 5m 39s\n- tooz-tox-py310-redis https://zuul.opendev.org/t/openstack/build/128c23032ed0469aa4803ff156004b1e : SUCCESS in 6m 40s\n- tooz-tox-py313-redis https://zuul.opendev.org/t/openstack/build/404d57d7c104450a893be9e94ef20ab1 : SUCCESS in 9m 14s\n- tooz-tox-py310-sentinel https://zuul.opendev.org/t/openstack/build/f1dded6deb08429f920a220661584064 : SUCCESS in 6m 42s\n- tooz-tox-py313-sentinel https://zuul.opendev.org/t/openstack/build/e98074d3aa7045d79e7cc43afa6a9a6f : SUCCESS in 7m 27s\n- tooz-tox-py310-memcached https://zuul.opendev.org/t/openstack/build/759661461e8b41bab63284200b9faf87 : SUCCESS in 4m 15s\n- tooz-tox-py313-memcached https://zuul.opendev.org/t/openstack/build/b41e722494334bfd848f4600cbe6def9 : SUCCESS in 5m 14s\n- tooz-tox-py310-postgresql https://zuul.opendev.org/t/openstack/build/fbd7e044e0864c39a251a8b802463e49 : SUCCESS in 4m 25s\n- tooz-tox-py313-postgresql https://zuul.opendev.org/t/openstack/build/5e234f4a4e9d4b89987cf09f80d796fb : SUCCESS in 4m 40s\n- tooz-tox-py310-mysql https://zuul.opendev.org/t/openstack/build/a6fd9e2d6a5b40caaf3d820092503cce : SUCCESS in 3m 54s\n- tooz-tox-py313-mysql https://zuul.opendev.org/t/openstack/build/dfe4e1bd23e04e87a10fefbecb6dad8b : SUCCESS in 7m 42s\n- tooz-tox-py310-consul https://zuul.opendev.org/t/openstack/build/8aca66e9418b4c4f9d650978ece74839 : SUCCESS in 4m 47s\n- tooz-tox-py313-consul https://zuul.opendev.org/t/openstack/build/fd4de40e1e7a4466beaf33b81dc6b05f : SUCCESS in 4m 59s","accounts_in_message":[],"_revision_number":1},{"id":"a7b985b574df5c641b7f48681fe8ea21841c2c97","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-03-12 08:42:07.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"db075aa0fa308b278bdb58dbc0aae9abfe34872d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-12 10:40:58.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4388f9bc20584033b45afcca615b4cc9\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/b04db8f5706c40b5bf755e45144e9060 : SUCCESS in 5m 51s\n- requirements-check https://zuul.opendev.org/t/openstack/build/ff755b3720b34ec6bd8ed537fd3d7a00 : SUCCESS in 2m 21s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0ff0ad292aab4339b20322b27ff39988 : SUCCESS in 1h 48m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/07f3197523c74443a7fc23a74b3751b3 : SUCCESS in 4m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a0dca5cc9e764f1e95deb7b3b14cb274 : SUCCESS in 5m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/451cee2afd6f4ace9966e4b21611a59d : SUCCESS in 4m 54s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9be6c448c0e44ccc977328d3ff3334ce : SUCCESS in 4m 48s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/ea756b5f1104416c80b51cd09928e3a1 : SUCCESS in 7m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/53f86786fbab4d6681294fa8443306c5 : SUCCESS in 5m 01s\n- tooz-tox-examples https://zuul.opendev.org/t/openstack/build/d54029de60714f069b20eaa997326ca8 : SUCCESS in 5m 08s\n- tooz-tox-py310-etcd3gw https://zuul.opendev.org/t/openstack/build/38e95f0a2ac2408384ec5ed8bcb36b18 : SUCCESS in 4m 09s\n- tooz-tox-py313-etcd3gw https://zuul.opendev.org/t/openstack/build/dcf0ac7cc1cb44caab848b182b7e9483 : SUCCESS in 5m 24s\n- tooz-tox-py310-zookeeper https://zuul.opendev.org/t/openstack/build/5cc0cefe25f248b983f5dacc52239c25 : SUCCESS in 3m 23s\n- tooz-tox-py313-zookeeper https://zuul.opendev.org/t/openstack/build/6d976695f9fe4782b71e044ab630031e : SUCCESS in 4m 21s\n- tooz-tox-py310-redis https://zuul.opendev.org/t/openstack/build/d123108c2e9e41d6b3d8bbdfa5b62de2 : SUCCESS in 5m 08s\n- tooz-tox-py313-redis https://zuul.opendev.org/t/openstack/build/bc6a478f73c548c880fb41805aabcdae : SUCCESS in 8m 41s\n- tooz-tox-py310-sentinel https://zuul.opendev.org/t/openstack/build/d50b3dba77b84504986d77a9a649a25a : SUCCESS in 5m 16s\n- tooz-tox-py313-sentinel https://zuul.opendev.org/t/openstack/build/e95e7c97f4174e8a9253815a02761c9a : SUCCESS in 6m 59s\n- tooz-tox-py310-memcached https://zuul.opendev.org/t/openstack/build/4c71bc6f13514083a9c65f4e073157a2 : SUCCESS in 4m 31s\n- tooz-tox-py313-memcached https://zuul.opendev.org/t/openstack/build/17eb09eef36d41d4ba7c4a2d060cc777 : SUCCESS in 4m 53s\n- tooz-tox-py310-postgresql https://zuul.opendev.org/t/openstack/build/c5599a2b30914696a3c993120dc44dc7 : SUCCESS in 4m 58s\n- tooz-tox-py313-postgresql https://zuul.opendev.org/t/openstack/build/e9e30b7aabfc461db23bcc0d9dcd3671 : SUCCESS in 4m 49s\n- tooz-tox-py310-mysql https://zuul.opendev.org/t/openstack/build/e7835fac62f649bea71c4371f1128305 : SUCCESS in 4m 54s\n- tooz-tox-py313-mysql https://zuul.opendev.org/t/openstack/build/f7b2f0fc7f624b0da78cbaed50054d93 : SUCCESS in 7m 03s\n- tooz-tox-py310-consul https://zuul.opendev.org/t/openstack/build/0769cab8e90f4e35bc854e555aad8c70 : SUCCESS in 4m 45s\n- tooz-tox-py313-consul https://zuul.opendev.org/t/openstack/build/96a5439186794ab488b72b51c4df292a : SUCCESS in 5m 16s","accounts_in_message":[],"_revision_number":2},{"id":"6e61dd33faebfc84fd0ba5451b9092bc42376e68","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-03-14 14:40:20.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"8412937b1d6f78ebaca992d24568711fb8a8587a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-14 16:43:02.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4982042f090842aa968d97ba03bcc8f2\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/009378a8798e4edaa218707b70799741 : SUCCESS in 1m 51s\n- requirements-check https://zuul.opendev.org/t/openstack/build/fbd1ecf208f44380b9e7807ad1f211c4 : SUCCESS in 3m 33s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/48f992ec4fba45c2870180d4573f75dd : SUCCESS in 1h 57m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ee4062d1112a4b3cb6773b88ff9640d8 : SUCCESS in 5m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f03dd037a7d544f2b65feb98183bca21 : SUCCESS in 3m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/81198e2da4a9423d87e683100475c33e : SUCCESS in 4m 52s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/fdb85324ab3d4053867d464a55a49aed : SUCCESS in 5m 31s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/3e536c15edeb4194a03d89aa82ed51e6 : FAILURE in 9m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c0f2bd3d585447bbecc2980f8b30ec3 : SUCCESS in 3m 49s\n- tooz-tox-examples https://zuul.opendev.org/t/openstack/build/75c172e0d5bd412c901d7872146a54ac : SUCCESS in 3m 42s\n- tooz-tox-py310-etcd3gw https://zuul.opendev.org/t/openstack/build/a5bdee78d428491284b8416b8d484df2 : SUCCESS in 3m 31s\n- tooz-tox-py313-etcd3gw https://zuul.opendev.org/t/openstack/build/0b785c0402a0448f9c2af0525f51e661 : SUCCESS in 5m 37s\n- tooz-tox-py310-zookeeper https://zuul.opendev.org/t/openstack/build/2b5b18e1bd314504b594eae4baa8bd2e : SUCCESS in 4m 35s\n- tooz-tox-py313-zookeeper https://zuul.opendev.org/t/openstack/build/cc6cb325cb5d4299ba6fbbb1b6ea04f1 : SUCCESS in 4m 38s\n- tooz-tox-py310-redis https://zuul.opendev.org/t/openstack/build/70dd93607e2b4100b9151024fd8a2574 : SUCCESS in 4m 42s\n- tooz-tox-py313-redis https://zuul.opendev.org/t/openstack/build/de649b65702546fea01be00a82974f3c : SUCCESS in 7m 18s\n- tooz-tox-py310-sentinel https://zuul.opendev.org/t/openstack/build/46ba1e89632043139fdf412d8f0af054 : SUCCESS in 6m 43s\n- tooz-tox-py313-sentinel https://zuul.opendev.org/t/openstack/build/1d37a65188524def9852b64d1c3543f0 : SUCCESS in 7m 00s\n- tooz-tox-py310-memcached https://zuul.opendev.org/t/openstack/build/8ec629cc198c4336be284341780f2f15 : SUCCESS in 6m 06s\n- tooz-tox-py313-memcached https://zuul.opendev.org/t/openstack/build/83b73a7a149a4709aa3007b6482d3c52 : SUCCESS in 3m 18s\n- tooz-tox-py310-postgresql https://zuul.opendev.org/t/openstack/build/35cc25cbae1b4ebea6b0e9c02be1de03 : SUCCESS in 5m 51s\n- tooz-tox-py313-postgresql https://zuul.opendev.org/t/openstack/build/be64721746e8455abfaf027300824801 : SUCCESS in 3m 32s\n- tooz-tox-py310-mysql https://zuul.opendev.org/t/openstack/build/0e1456c5a6ae463896c1589aaca681dd : SUCCESS in 2m 55s\n- tooz-tox-py313-mysql https://zuul.opendev.org/t/openstack/build/edba94d60a1b4599bc422bd768c57521 : SUCCESS in 7m 50s\n- tooz-tox-py310-consul https://zuul.opendev.org/t/openstack/build/369f80d131de4946a616229fd3cda96d : SUCCESS in 3m 35s\n- tooz-tox-py313-consul https://zuul.opendev.org/t/openstack/build/7a80e0ec9cd74735a7583d409f4e4146 : SUCCESS in 5m 04s","accounts_in_message":[],"_revision_number":3},{"id":"77825fc6fd8e7320ab2d0b4662aca6ee9451ced7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-03-19 09:47:24.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"b5b2ea65ba60e2cede21935b4de6a48c4bf3e778","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-19 11:54:53.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8dd68a0d093c4b21abdda11a4e58cf83\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/fe6fc9a27c104a04945587ddbc7473d1 : SUCCESS in 3m 19s\n- requirements-check https://zuul.opendev.org/t/openstack/build/1df62f81c5c64872aa165eaff1f90026 : SUCCESS in 2m 16s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/71f5cb8e342f4bf79e07d8214f5f8e2d : SUCCESS in 2h 00m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e71dfab3ad5245c3be178b4873715d72 : SUCCESS in 6m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b033e3869bd4ea9a13835ca9a994f94 : SUCCESS in 6m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/59b40d02dc4342eaaac09c4e9e4c6d37 : SUCCESS in 5m 25s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c4c4a9f5a65348998daeed9353bee137 : SUCCESS in 5m 06s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/dc8ba5eec6e744129f3a3095ce68605e : SUCCESS in 5m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a76e3404201e4ccbad54019882b790bc : SUCCESS in 3m 36s\n- tooz-tox-examples https://zuul.opendev.org/t/openstack/build/decdce678c514ab697bb8217b454bfea : SUCCESS in 8m 54s\n- tooz-tox-py310-etcd3gw https://zuul.opendev.org/t/openstack/build/ba51bede747441859d33f32addfac8b9 : SUCCESS in 4m 45s\n- tooz-tox-py313-etcd3gw https://zuul.opendev.org/t/openstack/build/a886ed31c57245c28337ae2ff006fd12 : SUCCESS in 5m 52s\n- tooz-tox-py310-zookeeper https://zuul.opendev.org/t/openstack/build/e5b3f42799b14fc69ec01e847868631c : SUCCESS in 2m 46s\n- tooz-tox-py313-zookeeper https://zuul.opendev.org/t/openstack/build/fd098048778c4539b82ebc9884f3537f : SUCCESS in 2m 56s\n- tooz-tox-py310-redis https://zuul.opendev.org/t/openstack/build/6860c3a652914cac9564097db12961b0 : SUCCESS in 6m 40s\n- tooz-tox-py313-redis https://zuul.opendev.org/t/openstack/build/a1666f9300b14bf8b7387492ae28ab7e : SUCCESS in 7m 07s\n- tooz-tox-py310-sentinel https://zuul.opendev.org/t/openstack/build/7b3fff3c489549b7914a2898e430cb15 : SUCCESS in 8m 37s\n- tooz-tox-py313-sentinel https://zuul.opendev.org/t/openstack/build/be5ffbf383df4263a547f7f15a6f39ee : SUCCESS in 6m 26s\n- tooz-tox-py310-memcached https://zuul.opendev.org/t/openstack/build/8a52d2fb1f8242999b3a833326611861 : SUCCESS in 4m 29s\n- tooz-tox-py313-memcached https://zuul.opendev.org/t/openstack/build/df093316af5745a590680549a582a27b : SUCCESS in 5m 12s\n- tooz-tox-py310-postgresql https://zuul.opendev.org/t/openstack/build/4c064fae779244a5823c5e5a1de90e38 : SUCCESS in 4m 11s\n- tooz-tox-py313-postgresql https://zuul.opendev.org/t/openstack/build/3cdf195d617742efbb88291efff650cc : SUCCESS in 5m 11s\n- tooz-tox-py310-mysql https://zuul.opendev.org/t/openstack/build/c64e5ffb24344bf4b59ecf5f76ab8b7f : SUCCESS in 5m 34s\n- tooz-tox-py313-mysql https://zuul.opendev.org/t/openstack/build/e76025d1cd4e42f4b26dd94e1db7014f : SUCCESS in 8m 09s\n- tooz-tox-py310-consul https://zuul.opendev.org/t/openstack/build/14049e85287644098185ec7a519cd621 : SUCCESS in 5m 05s\n- tooz-tox-py313-consul https://zuul.opendev.org/t/openstack/build/077e92cffc9346fc8e86ee7448c562ef : SUCCESS in 5m 32s","accounts_in_message":[],"_revision_number":4},{"id":"b93e3fc85fc154e11ca5144b9434b41b1e9c7f9c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-03-31 09:03:29.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"ad0e9978020253fa8467c44421d621487e23aa22","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-31 10:25:40.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cff8d4bd13db4d4baaf1b0cc3e699d12\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/678479293d334b28838a0dfb118e234d : SUCCESS in 1m 41s\n- requirements-check https://zuul.opendev.org/t/openstack/build/87589dcf1b124b6e8c319f9d7c9f28d6 : SUCCESS in 3m 28s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c59482b2f7684aaa8a6de453e135c37b : SUCCESS in 1h 19m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/23a588256a0b4cc5a16795295d0465d3 : SUCCESS in 5m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c8302bd0e17c49df9535f5bae6d55eb9 : SUCCESS in 5m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e4ca9cdf884142a6adb358ac3d14b767 : SUCCESS in 4m 44s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/596ed4f24b9a455dac4c06d7986be173 : SUCCESS in 7m 02s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/94bfb1195fcc4df9849642cbe54a2278 : FAILURE in 7m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/83c77093b73c45beb39c32eb4131accc : SUCCESS in 5m 49s\n- tooz-tox-examples https://zuul.opendev.org/t/openstack/build/d1618959ddfc458da8f391b9bb42db1d : SUCCESS in 5m 17s\n- tooz-tox-py310-etcd3gw https://zuul.opendev.org/t/openstack/build/547315dc48b74b56b2d2ce447cbea5fb : SUCCESS in 4m 44s\n- tooz-tox-py313-etcd3gw https://zuul.opendev.org/t/openstack/build/2403823afbc54f9bb65b8bc4fb8ac885 : SUCCESS in 6m 17s\n- tooz-tox-py310-zookeeper https://zuul.opendev.org/t/openstack/build/a01c5328a09e402990e4b6c589ce4661 : SUCCESS in 4m 49s\n- tooz-tox-py313-zookeeper https://zuul.opendev.org/t/openstack/build/7a4c62f92da84455af17b9238715d8ac : SUCCESS in 6m 58s\n- tooz-tox-py310-redis https://zuul.opendev.org/t/openstack/build/cafaaa8b915f49acb65425c1e56c8e8d : SUCCESS in 5m 58s\n- tooz-tox-py313-redis https://zuul.opendev.org/t/openstack/build/5827be59b87f4f449b3989798c231c6a : SUCCESS in 14m 37s\n- tooz-tox-py310-sentinel https://zuul.opendev.org/t/openstack/build/5020b7ce6f2343f0b8986ae205396a8d : SUCCESS in 7m 28s\n- tooz-tox-py313-sentinel https://zuul.opendev.org/t/openstack/build/b42b31e5a4d74b78af7efe5a1136807b : SUCCESS in 5m 39s\n- tooz-tox-py310-memcached https://zuul.opendev.org/t/openstack/build/397071fbc20f4d55959adfa1b35005b5 : SUCCESS in 3m 31s\n- tooz-tox-py313-memcached https://zuul.opendev.org/t/openstack/build/cc7077c59d054a509e8c6572396b8fdc : SUCCESS in 6m 48s\n- tooz-tox-py310-postgresql https://zuul.opendev.org/t/openstack/build/8f6d6329c22a4e6a9fe449a4a79dff88 : SUCCESS in 5m 13s\n- tooz-tox-py313-postgresql https://zuul.opendev.org/t/openstack/build/32b60ae2e3a14f5ea1dbf5a70b2b8447 : SUCCESS in 5m 37s\n- tooz-tox-py310-mysql https://zuul.opendev.org/t/openstack/build/47d99e211374467fb85904ca02a741ba : SUCCESS in 5m 06s\n- tooz-tox-py313-mysql https://zuul.opendev.org/t/openstack/build/9e8d6fb1cf3e4135a1927e4c8154fd95 : SUCCESS in 5m 15s\n- tooz-tox-py310-consul https://zuul.opendev.org/t/openstack/build/3f806ed775a74b7dac86d232e1706622 : SUCCESS in 5m 00s\n- tooz-tox-py313-consul https://zuul.opendev.org/t/openstack/build/2b9eb5349d8244afa16b0bfbceb98a64 : SUCCESS in 6m 56s","accounts_in_message":[],"_revision_number":5},{"id":"db392a6959b377a73b8eb6b60d2f6725f4bd5a81","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2026-04-09 20:28:39.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"989bd7313e62f347f6e03e314e226209a787df67","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-04-10 11:32:37.000000000","message":"Patch Set 5: Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"77257dddb1effac75d91f3a88ab6e7fdecb05fd1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-10 11:33:03.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"5e472e92a54787886427e7e00157bfbf59f58cbb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-10 13:36:00.000000000","message":"Patch Set 5: Verified-2\n\n(1 comment)\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/fd72b22b205a4144b7458eecee9abf67\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/72c6c26055c645beab36b2845a506b2f : SUCCESS in 3m 50s\n- requirements-check https://zuul.opendev.org/t/openstack/build/617a8689d7904bf48e5311330aa6d24d : SUCCESS in 3m 34s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b4a341ddce53464489bece9aaab4187d : SUCCESS in 1h 57m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1dd3a207993744baa641dacfff40443d : FAILURE in 5m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9e3cf017982d42038562f113daf1eed2 : SUCCESS in 4m 38s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e6227a5748e64f93a865b558cd15076c : SUCCESS in 5m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a7a5474d59a4f06b2869a193ca8d108 : SUCCESS in 5m 05s\n- tooz-tox-examples https://zuul.opendev.org/t/openstack/build/f3000b738b174fd1a0b04d7db1a3f49f : SUCCESS in 5m 40s\n- tooz-tox-py310-etcd3gw https://zuul.opendev.org/t/openstack/build/e09e2488c6834e5791caeab9af4a0e48 : SUCCESS in 5m 28s\n- tooz-tox-py313-etcd3gw https://zuul.opendev.org/t/openstack/build/848fa7ae344847829dc47eefd4e2cdf5 : SUCCESS in 4m 51s\n- tooz-tox-py310-zookeeper https://zuul.opendev.org/t/openstack/build/8eb73c8e38014088ae79dd7ddc17b14b : SUCCESS in 4m 08s\n- tooz-tox-py313-zookeeper https://zuul.opendev.org/t/openstack/build/ae05f32976724c51951bf8bbaba0d8d1 : SUCCESS in 4m 39s\n- tooz-tox-py310-redis https://zuul.opendev.org/t/openstack/build/ee7d94aa8b0948ed9ba771c884fdf569 : SUCCESS in 6m 10s\n- tooz-tox-py313-redis https://zuul.opendev.org/t/openstack/build/eb41077f80f747f08dbc6d13a03dc737 : SUCCESS in 10m 03s\n- tooz-tox-py310-sentinel https://zuul.opendev.org/t/openstack/build/54927db0df9e4381ae6d181291eba0a4 : SUCCESS in 6m 24s\n- tooz-tox-py313-sentinel https://zuul.opendev.org/t/openstack/build/4d259c459d7844f4ab230e7637eac037 : SUCCESS in 8m 59s\n- tooz-tox-py310-memcached https://zuul.opendev.org/t/openstack/build/3ec7837130264fd7be0de9ef043f15e9 : SUCCESS in 3m 12s\n- tooz-tox-py313-memcached https://zuul.opendev.org/t/openstack/build/c771cc5bb4bd4451bccead1bb29a3386 : SUCCESS in 4m 43s\n- tooz-tox-py310-postgresql https://zuul.opendev.org/t/openstack/build/c0b19c15d1474cce85bd9cd94b272fa1 : SUCCESS in 5m 09s\n- tooz-tox-py313-postgresql https://zuul.opendev.org/t/openstack/build/6df504a7cc3c4b69ac45773a9d8400d7 : SUCCESS in 4m 27s\n- tooz-tox-py310-mysql https://zuul.opendev.org/t/openstack/build/86753ea3acd8499da12af6b1c35b4601 : SUCCESS in 2m 59s\n- tooz-tox-py313-mysql https://zuul.opendev.org/t/openstack/build/eb87b10f62ce44e184e194d5fd778485 : SUCCESS in 9m 17s\n- tooz-tox-py310-consul https://zuul.opendev.org/t/openstack/build/4ad15162312c48988a5c0de7f34f1b7d : SUCCESS in 4m 12s\n- tooz-tox-py313-consul https://zuul.opendev.org/t/openstack/build/15115da59f3743a6a690f903af4f6c24 : SUCCESS in 5m 10s","accounts_in_message":[],"_revision_number":5},{"id":"821c963c5e71ca6cc1f8005f74c51b7c999b2001","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-04-10 18:13:45.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-2\n* Workflow+1\n","accounts_in_message":[],"_revision_number":6},{"id":"bcba5123d6335d6a3b7c15053e343e0cb4419ad8","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-04-10 18:17:35.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"57ca4e416f573abb5988468f5eba6cb5ffa042b0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-10 20:34:52.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/16da7c8099ad4b8b9a585ecf3900a2f4\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/de703877643a464b97f37f36adffabae : SUCCESS in 2m 06s\n- requirements-check https://zuul.opendev.org/t/openstack/build/3e1cd905507247469188fc31da8be035 : SUCCESS in 2m 41s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fd97dcd8a07d49d4abe1de188337a9a9 : SUCCESS in 1h 36m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d0da5a9a8ee1472d80f511f2254418c7 : SUCCESS in 7m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ce373305c01a4f7bb9cd11e8d1b873ea : SUCCESS in 4m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/af1ebef2c62c4635a6d036795c9afafd : SUCCESS in 5m 06s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/80a39139236d4b5f91269a8da4df711b : SUCCESS in 4m 42s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/5f34519f22144892be772e9c5f73c3ed : SUCCESS in 10m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/203dc89403274e9f9bfc87ad5f5da77b : SUCCESS in 4m 43s\n- tooz-tox-examples https://zuul.opendev.org/t/openstack/build/4f7980225a0346c09ab649600bf7ace3 : SUCCESS in 5m 28s\n- tooz-tox-py310-etcd3gw https://zuul.opendev.org/t/openstack/build/2f00944be6194e0da86744a7ede928db : SUCCESS in 5m 08s\n- tooz-tox-py313-etcd3gw https://zuul.opendev.org/t/openstack/build/8f1b936e7daa4cb5b233c0d148b92381 : SUCCESS in 8m 13s\n- tooz-tox-py310-zookeeper https://zuul.opendev.org/t/openstack/build/5fcc3cdbf8d947bd8393767cdcc23f27 : SUCCESS in 5m 06s\n- tooz-tox-py313-zookeeper https://zuul.opendev.org/t/openstack/build/4b36821b6b794e12a7038f487ddba5f6 : SUCCESS in 5m 14s\n- tooz-tox-py310-redis https://zuul.opendev.org/t/openstack/build/186a3e76d36e41a2a195edfdcf31aae9 : SUCCESS in 5m 34s\n- tooz-tox-py313-redis https://zuul.opendev.org/t/openstack/build/eadfc7e056be4271a6ddf68b687d7bae : SUCCESS in 7m 29s\n- tooz-tox-py310-sentinel https://zuul.opendev.org/t/openstack/build/a223bec06b97449d8378e309f0e33ddf : SUCCESS in 7m 35s\n- tooz-tox-py313-sentinel https://zuul.opendev.org/t/openstack/build/fb0b56a7efe34a0b9d2823749e519e28 : SUCCESS in 6m 41s\n- tooz-tox-py310-memcached https://zuul.opendev.org/t/openstack/build/c5fead4b59674e699ecd789bb7fd8e04 : SUCCESS in 4m 23s\n- tooz-tox-py313-memcached https://zuul.opendev.org/t/openstack/build/7ae24084daf246dea4f77da847742401 : SUCCESS in 4m 16s\n- tooz-tox-py310-postgresql https://zuul.opendev.org/t/openstack/build/8742d501e3cf4ebea3310b6f40c4038a : SUCCESS in 5m 06s\n- tooz-tox-py313-postgresql https://zuul.opendev.org/t/openstack/build/0e21f466bf5845b5adc809cf5dbc01fe : SUCCESS in 5m 07s\n- tooz-tox-py310-mysql https://zuul.opendev.org/t/openstack/build/9af285c88d4f46c4bf246d4435c35069 : SUCCESS in 3m 35s\n- tooz-tox-py313-mysql https://zuul.opendev.org/t/openstack/build/f835dbaa32cd4e239b00b4e2b0d46f56 : SUCCESS in 8m 13s\n- tooz-tox-py310-consul https://zuul.opendev.org/t/openstack/build/0a2651bb60424bc69346248f1e56de72 : SUCCESS in 5m 04s\n- tooz-tox-py313-consul https://zuul.opendev.org/t/openstack/build/13056349de6e434ab33ba4f7a0e79e01 : SUCCESS in 5m 17s","accounts_in_message":[],"_revision_number":6},{"id":"ca24f29e58fdf2b54d062e166f285f9bfd0bc4c9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-10 20:35:12.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"74f6da55d3297fd577ce90d6cdc15802936dfa20","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-10 22:38:40.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0a3bd6b54e474e42a88b46177881d847\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/8a92276bf74c4c26bf767d8bca8bb820 : SUCCESS in 2m 27s\n- requirements-check https://zuul.opendev.org/t/openstack/build/8aa2ec72171c490d9fb7387fae6f8bb8 : SUCCESS in 2m 17s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5f8710f037a94db7b8ec0e7f547a7e41 : SUCCESS in 1h 58m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/96c90288d5e2413ab70c566e15aada1d : SUCCESS in 4m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/50c9db74d6f94dd68fb1dafea20cbabd : SUCCESS in 5m 45s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c27039b2e05a4e29a77814252153b682 : SUCCESS in 5m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/49c21626bf3a499fa0eba0ba2d594ec7 : SUCCESS in 6m 37s\n- tooz-tox-examples https://zuul.opendev.org/t/openstack/build/950e3c67f0b74f97ad7efc4b9c4a72a7 : SUCCESS in 2m 56s\n- tooz-tox-py310-etcd3gw https://zuul.opendev.org/t/openstack/build/264e6c98405c45379c5518695c87095a : SUCCESS in 3m 45s\n- tooz-tox-py313-etcd3gw https://zuul.opendev.org/t/openstack/build/a9b34f8c8fb340a8a2d7a18da62adcd3 : SUCCESS in 5m 14s\n- tooz-tox-py310-zookeeper https://zuul.opendev.org/t/openstack/build/20e19039e5e74bd2b15cf88c94c6f88a : SUCCESS in 3m 17s\n- tooz-tox-py313-zookeeper https://zuul.opendev.org/t/openstack/build/1d00e4d1b11f4b17a619bbfbfa374bc4 : SUCCESS in 4m 56s\n- tooz-tox-py310-redis https://zuul.opendev.org/t/openstack/build/b613860435354dd488a8c23111713d95 : SUCCESS in 6m 37s\n- tooz-tox-py313-redis https://zuul.opendev.org/t/openstack/build/afa70b50e1814932abba1cc5b9f656a0 : SUCCESS in 6m 48s\n- tooz-tox-py310-sentinel https://zuul.opendev.org/t/openstack/build/148d2ccdcd754b1bb108bd3d44fe0c9c : SUCCESS in 5m 23s\n- tooz-tox-py313-sentinel https://zuul.opendev.org/t/openstack/build/3f54e11f036542a6b6f26b68f8311f6f : SUCCESS in 6m 57s\n- tooz-tox-py310-memcached https://zuul.opendev.org/t/openstack/build/d3b846cee1ca420fbdaf035be2de4c09 : SUCCESS in 3m 03s\n- tooz-tox-py313-memcached https://zuul.opendev.org/t/openstack/build/f33ccd6c470a4084b62e5b8a99f18a0f : SUCCESS in 5m 35s\n- tooz-tox-py310-postgresql https://zuul.opendev.org/t/openstack/build/7113a0115e2b4a6e97b9b823de29618e : SUCCESS in 4m 05s\n- tooz-tox-py313-postgresql https://zuul.opendev.org/t/openstack/build/7fd2ead941db46eaaad84bda682494f9 : SUCCESS in 4m 45s\n- tooz-tox-py310-mysql https://zuul.opendev.org/t/openstack/build/03767c8855e6471c9ee3284fef802f67 : SUCCESS in 6m 50s\n- tooz-tox-py313-mysql https://zuul.opendev.org/t/openstack/build/2d30141872a64b96a50a756b78f4d072 : SUCCESS in 7m 38s\n- tooz-tox-py310-consul https://zuul.opendev.org/t/openstack/build/d2ca8b234a844348b300834e21d5a8ef : SUCCESS in 3m 03s\n- tooz-tox-py313-consul https://zuul.opendev.org/t/openstack/build/27a87ffc7ce54218bbf0c8bf12dc5c14 : SUCCESS in 5m 17s","accounts_in_message":[],"_revision_number":6},{"id":"b1571b57b783c1df65573512d9a3dd8ac2d2c1bb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-10 22:38:41.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"282afdb0f74d2c846353b6dc1c5d4c1eba74aefd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-10 22:39:39.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/43cd1f26c09347498976a03640fe7f89\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd5b521bf267457ebde642906cbc2de7 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"956ab013acde15308cf1bd037ab72006e7707d02","revisions":{"c167d1582d90b423c6765f5a13ab57a5106cd892":{"kind":"REWORK","_number":1,"created":"2026-03-11 15:42:02.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/07/980107/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/07/980107/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/07/980107/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/07/980107/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/07/980107/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/07/980107/1"}}},"commit":{"parents":[{"commit":"2d95cffe8226ba09c827a97285bdbdcb6c84652d","subject":"typing: Add client, session_id properties for consul driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/2d95cffe8226ba09c827a97285bdbdcb6c84652d"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-03-09 15:50:06.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-03-11 15:41:56.000000000","tz":0},"subject":"typing: Add type hints to tooz.utils","message":"typing: Add type hints to tooz.utils\n\nChange-Id: I745cb15cc269a032f7cea2c84c5861de79885d39\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/c167d1582d90b423c6765f5a13ab57a5106cd892"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/c167d1582d90b423c6765f5a13ab57a5106cd892"}]},"branch":"refs/heads/master"},"68c957d477c2da8a1eb206a3ec614e958d4b8443":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-03-12 08:42:07.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/07/980107/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/07/980107/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/07/980107/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/07/980107/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/07/980107/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/07/980107/2"}}},"commit":{"parents":[{"commit":"6609c3214f1020dce06916ff0039522ac13a0c22","subject":"typing: Add client, session_id properties for consul driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/6609c3214f1020dce06916ff0039522ac13a0c22"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-03-09 15:50:06.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-03-12 08:40:21.000000000","tz":0},"subject":"typing: Add type hints to tooz.utils","message":"typing: Add type hints to tooz.utils\n\nChange-Id: I745cb15cc269a032f7cea2c84c5861de79885d39\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/68c957d477c2da8a1eb206a3ec614e958d4b8443"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/68c957d477c2da8a1eb206a3ec614e958d4b8443"}]},"branch":"refs/heads/master"},"f4e9f08bd183cba1fbad5e4d03c1f6e0abfecaa9":{"kind":"REWORK","_number":3,"created":"2026-03-14 14:40:20.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/07/980107/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/07/980107/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/07/980107/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/07/980107/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/07/980107/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/07/980107/3"}}},"commit":{"parents":[{"commit":"6631d724d5eaf7f47b94b862da532a89f817d694","subject":"typing: Normalize instance and static methods","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/6631d724d5eaf7f47b94b862da532a89f817d694"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-03-09 15:50:06.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-03-14 14:40:15.000000000","tz":0},"subject":"typing: Add type hints to tooz.utils","message":"typing: Add type hints to tooz.utils\n\nChange-Id: I745cb15cc269a032f7cea2c84c5861de79885d39\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/f4e9f08bd183cba1fbad5e4d03c1f6e0abfecaa9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/f4e9f08bd183cba1fbad5e4d03c1f6e0abfecaa9"}]},"branch":"refs/heads/master"},"c2f176c4b119a1c5006ba5f31e7c4cdfb2016986":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2026-03-19 09:47:24.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/07/980107/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/07/980107/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/07/980107/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/07/980107/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/07/980107/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/07/980107/4"}}},"commit":{"parents":[{"commit":"413e9bb853d257332248829f111693f35d4d5a13","subject":"typing: Normalize instance and static methods","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/413e9bb853d257332248829f111693f35d4d5a13"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-03-09 15:50:06.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-03-19 09:46:03.000000000","tz":60},"subject":"typing: Add type hints to tooz.utils","message":"typing: Add type hints to tooz.utils\n\nChange-Id: I745cb15cc269a032f7cea2c84c5861de79885d39\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/c2f176c4b119a1c5006ba5f31e7c4cdfb2016986"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/c2f176c4b119a1c5006ba5f31e7c4cdfb2016986"}]},"branch":"refs/heads/master"},"ebd48a9a9c3d392522ae71df4ff9e3441b4c8f76":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2026-03-31 09:03:29.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/07/980107/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/07/980107/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/07/980107/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/07/980107/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/07/980107/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/07/980107/5"}}},"commit":{"parents":[{"commit":"47df98bb04b01761c70144ad3fdc05877debf153","subject":"Merge \"zookeeper: Track joined groups to fix leave-on-stop race\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/47df98bb04b01761c70144ad3fdc05877debf153"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-03-09 15:50:06.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-03-31 09:02:11.000000000","tz":120},"subject":"typing: Add type hints to tooz.utils","message":"typing: Add type hints to tooz.utils\n\nChange-Id: I745cb15cc269a032f7cea2c84c5861de79885d39\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/ebd48a9a9c3d392522ae71df4ff9e3441b4c8f76"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/ebd48a9a9c3d392522ae71df4ff9e3441b4c8f76"}]},"branch":"refs/heads/master"},"956ab013acde15308cf1bd037ab72006e7707d02":{"kind":"REWORK","_number":6,"created":"2026-04-10 18:13:45.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/07/980107/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/07/980107/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/07/980107/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/07/980107/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/07/980107/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/07/980107/6"}}},"commit":{"parents":[{"commit":"f1fbed4bed23911c5d2bff45e8d3d7a019e87502","subject":"Merge \"zookeeper: Inline handler methods\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/f1fbed4bed23911c5d2bff45e8d3d7a019e87502"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-03-09 15:50:06.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-04-10 12:00:54.000000000","tz":60},"subject":"typing: Add type hints to tooz.utils","message":"typing: Add type hints to tooz.utils\n\nChange-Id: I745cb15cc269a032f7cea2c84c5861de79885d39\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/956ab013acde15308cf1bd037ab72006e7707d02"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/956ab013acde15308cf1bd037ab72006e7707d02"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
