)]}'
{"id":"openstack%2Ftooz~980108","triplet_id":"openstack%2Ftooz~master~I8f2ebecfc02ffed0fe9de376b436565b584c5b15","project":"openstack/tooz","branch":"master","topic":"typing","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2026-04-10 18:17:40.000000000","reason":"\u003cGERRIT_ACCOUNT_15334\u003e replied on the change","reason_account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}}},"hashtags":[],"change_id":"I8f2ebecfc02ffed0fe9de376b436565b584c5b15","subject":"typing: Add type hints to tooz._retry","status":"MERGED","created":"2026-03-11 15:42:02.000000000","updated":"2026-04-10 22:40:30.000000000","submitted":"2026-04-10 22:38:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"980108-typing","meta_rev_id":"e5990801d3e031b4874842a45650a63ef0968cf8","_number":980108,"virtual_id_number":980108,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-04-10 22:38:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2026-04-10 18:13:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2026-04-10 18:17:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-11 17:38:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-04-09 20:29:08.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"78634b418758db6a23368fbc32eb628e7a1900cf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-03-11 15:42:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"91d7e07d49ba91ea51e1ca8fe8c268c93f232406","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-11 17:38:59.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e041670cb54a4627ac10b4566e43ad5e\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/32b70c8b683a4222aabd8a27265b0dab : SUCCESS in 2m 44s\n- requirements-check https://zuul.opendev.org/t/openstack/build/aac1ca15b4284d9f86914079abe57664 : SUCCESS in 3m 40s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3d9b5ece384c4c60b02c9b7163d7f11e : SUCCESS in 1h 46m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d0f8511561fa4ec0a474dba951c99c5b : SUCCESS in 5m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a52bd25202034b83b4c71e3596d842b5 : SUCCESS in 6m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c24ab6d4432c4641a7f876a0914c2d7d : SUCCESS in 3m 39s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7c940c3526da464980c887672f9e2630 : SUCCESS in 3m 42s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/e533f5682a9c45729a6cc7dc13f4f55b : SUCCESS in 7m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f1be874c6b7b422c898a39909ede168f : SUCCESS in 7m 05s\n- tooz-tox-examples https://zuul.opendev.org/t/openstack/build/ab4f3ec9239f4670b5fd2c598980a8be : SUCCESS in 4m 43s\n- tooz-tox-py310-etcd3gw https://zuul.opendev.org/t/openstack/build/c283d05744ac418ca7560fc9b03fc554 : FAILURE in 3m 44s\n- tooz-tox-py313-etcd3gw https://zuul.opendev.org/t/openstack/build/6b60c2d8b033408987251129ef8915c8 : FAILURE in 7m 49s\n- tooz-tox-py310-zookeeper https://zuul.opendev.org/t/openstack/build/7ef210e025a341f49c1c4841e68f40c7 : SUCCESS in 3m 08s\n- tooz-tox-py313-zookeeper https://zuul.opendev.org/t/openstack/build/faad5fe272554da7a31034be8e901335 : SUCCESS in 3m 10s\n- tooz-tox-py310-redis https://zuul.opendev.org/t/openstack/build/60a437fad2cd4b0581132fe972710ae4 : SUCCESS in 6m 12s\n- tooz-tox-py313-redis https://zuul.opendev.org/t/openstack/build/138ac862d8614898b10e3e3faf2bff3e : SUCCESS in 5m 32s\n- tooz-tox-py310-sentinel https://zuul.opendev.org/t/openstack/build/6b9f90039a6d428c94b62617abda734e : SUCCESS in 7m 10s\n- tooz-tox-py313-sentinel https://zuul.opendev.org/t/openstack/build/e81efe6a14d94d29ae317de7b3d34154 : SUCCESS in 6m 34s\n- tooz-tox-py310-memcached https://zuul.opendev.org/t/openstack/build/38ce48b85ff64e7d9f28b2dbf7b87d24 : SUCCESS in 3m 41s\n- tooz-tox-py313-memcached https://zuul.opendev.org/t/openstack/build/a4fe6c2ef805419ab399360599708014 : SUCCESS in 5m 05s\n- tooz-tox-py310-postgresql https://zuul.opendev.org/t/openstack/build/9da7864a9ce44988b752d00504d63289 : SUCCESS in 5m 46s\n- tooz-tox-py313-postgresql https://zuul.opendev.org/t/openstack/build/01a51162798b43c793ee764ae51cc520 : SUCCESS in 4m 38s\n- tooz-tox-py310-mysql https://zuul.opendev.org/t/openstack/build/98493c676e4d4e2495fe998f0157156d : SUCCESS in 6m 39s\n- tooz-tox-py313-mysql https://zuul.opendev.org/t/openstack/build/1525c84d092c47eea964b1d7d5f7d0bd : SUCCESS in 9m 26s\n- tooz-tox-py310-consul https://zuul.opendev.org/t/openstack/build/123996ffd5834920b73fb093cd79d3a5 : SUCCESS in 4m 21s\n- tooz-tox-py313-consul https://zuul.opendev.org/t/openstack/build/d02ce7f7cc1c4d92826b1cbfc97e57eb : SUCCESS in 3m 37s","accounts_in_message":[],"_revision_number":1},{"id":"b97cfb5b6e129ddbe2755422c9c9b57363039886","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-03-12 08:42:07.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"3402946f0aec395f0b60bb374938374a54560c22","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-12 10:44:03.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6d1a6ac8478546c58b384263fbaa5093\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/c57740d603d34129bff88875c4447bbe : SUCCESS in 2m 51s\n- requirements-check https://zuul.opendev.org/t/openstack/build/32150f6ba885419ca352496a42868a2e : SUCCESS in 3m 45s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/dedf21c57aa94a129a9d42ceb02c58bf : SUCCESS in 1h 50m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a4bf2e9926b942a3a3ec59568f7baa67 : SUCCESS in 5m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f582d4ded4774917aa70c7818837538f : SUCCESS in 6m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fbac0a126a26430ebe20b92993bd953f : SUCCESS in 4m 54s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/63dcbe7220594538bdd1800ad7b91bad : SUCCESS in 4m 51s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/da22c0016aeb4d4787f472f36f8d5aca : SUCCESS in 7m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a4cfd6a833474fbaa68435ca96b102a6 : SUCCESS in 5m 11s\n- tooz-tox-examples https://zuul.opendev.org/t/openstack/build/0d6e2895294946778bff3e66d51e2109 : SUCCESS in 5m 12s\n- tooz-tox-py310-etcd3gw https://zuul.opendev.org/t/openstack/build/f30009fecbaa4a67b47b81a7a242ad54 : SUCCESS in 4m 59s\n- tooz-tox-py313-etcd3gw https://zuul.opendev.org/t/openstack/build/076e63df93d844c19a484943242c4f83 : SUCCESS in 5m 46s\n- tooz-tox-py310-zookeeper https://zuul.opendev.org/t/openstack/build/ea22b0188a5d420c906c7ca10748d41f : SUCCESS in 4m 50s\n- tooz-tox-py313-zookeeper https://zuul.opendev.org/t/openstack/build/a3e1c7d85d6042bca65f5d64f342066c : SUCCESS in 5m 24s\n- tooz-tox-py310-redis https://zuul.opendev.org/t/openstack/build/74b114368f0c46a4ad497699a5d7f45c : SUCCESS in 5m 14s\n- tooz-tox-py313-redis https://zuul.opendev.org/t/openstack/build/62a964d442c64285a179e71f11b372d7 : SUCCESS in 7m 11s\n- tooz-tox-py310-sentinel https://zuul.opendev.org/t/openstack/build/c168483806dc4d42b948c3ba3d473bc5 : SUCCESS in 6m 50s\n- tooz-tox-py313-sentinel https://zuul.opendev.org/t/openstack/build/60c447462bfc4ca3abfbf7e18721db2a : SUCCESS in 7m 27s\n- tooz-tox-py310-memcached https://zuul.opendev.org/t/openstack/build/b84f44a6bc1c46ee92470315ac4bbcf3 : SUCCESS in 4m 30s\n- tooz-tox-py313-memcached https://zuul.opendev.org/t/openstack/build/cacb89abb3a84cfaa6217b21e858e4ee : SUCCESS in 5m 02s\n- tooz-tox-py310-postgresql https://zuul.opendev.org/t/openstack/build/13e7b0af37c34082adf69bf4f79e5bf2 : SUCCESS in 4m 13s\n- tooz-tox-py313-postgresql https://zuul.opendev.org/t/openstack/build/c2bc7cddf5da40ada08c9656c0c3fb3d : SUCCESS in 4m 36s\n- tooz-tox-py310-mysql https://zuul.opendev.org/t/openstack/build/1c7d424479ac4bff8bbd316cb99b1e4a : SUCCESS in 4m 37s\n- tooz-tox-py313-mysql https://zuul.opendev.org/t/openstack/build/90c742af81404d7a850026495912f9de : SUCCESS in 7m 33s\n- tooz-tox-py310-consul https://zuul.opendev.org/t/openstack/build/da99cddc14254cbab459ea8898df189e : SUCCESS in 4m 56s\n- tooz-tox-py313-consul https://zuul.opendev.org/t/openstack/build/faf3dee8c2684262b2ee7773f208af73 : SUCCESS in 6m 18s","accounts_in_message":[],"_revision_number":2},{"id":"d3069d168e2c4a6d86207889931c7442c7e95cea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-03-14 14:40:20.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"4bd53bd77c11886e8203aa97763553b93acd5092","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-14 16:40:39.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/49583b68da4d419291e8b50e4a657b70\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/0c751f32d0a144a6a68df57f3eb96800 : SUCCESS in 2m 53s\n- requirements-check https://zuul.opendev.org/t/openstack/build/0d3fbdfb93a1416a896971b2c5e9da93 : SUCCESS in 3m 21s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0cff661d8be84091972be9419a4f1dfa : SUCCESS in 1h 51m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5447f61bf14d460dbb975584b32dbf4e : SUCCESS in 5m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b489cc6157574d52a683a094ad589f1a : SUCCESS in 5m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/61c86046996f42028ac1adaf06cc427d : SUCCESS in 4m 45s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/29b2ff3e831a40318cfe216047466760 : SUCCESS in 5m 14s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/5e21be9ceaec43309c544f672fb0fa14 : SUCCESS in 7m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/02d6748e3a404033b749e58337dad91d : SUCCESS in 5m 09s\n- tooz-tox-examples https://zuul.opendev.org/t/openstack/build/85e6c92e631445dbb1ae123811884ede : SUCCESS in 5m 25s\n- tooz-tox-py310-etcd3gw https://zuul.opendev.org/t/openstack/build/064df6b2c75c48e0aa2429254888958f : SUCCESS in 5m 02s\n- tooz-tox-py313-etcd3gw https://zuul.opendev.org/t/openstack/build/0086e8891b764d2694cd8eaabcc25373 : SUCCESS in 5m 32s\n- tooz-tox-py310-zookeeper https://zuul.opendev.org/t/openstack/build/a10c75614b0c486f87bcc10b89016db3 : SUCCESS in 4m 34s\n- tooz-tox-py313-zookeeper https://zuul.opendev.org/t/openstack/build/5ebd4d1d410745d2a01b512bbb48bd62 : SUCCESS in 4m 53s\n- tooz-tox-py310-redis https://zuul.opendev.org/t/openstack/build/60f735417863445ab6037b5ab791372f : SUCCESS in 6m 34s\n- tooz-tox-py313-redis https://zuul.opendev.org/t/openstack/build/cce4ecc8120244f1b88c0f7f3d13612d : SUCCESS in 5m 15s\n- tooz-tox-py310-sentinel https://zuul.opendev.org/t/openstack/build/3eadd7bdc407432583a481d5cc6193a7 : SUCCESS in 6m 14s\n- tooz-tox-py313-sentinel https://zuul.opendev.org/t/openstack/build/f51567ef105b4935a1725bb5c0a55c1c : SUCCESS in 6m 48s\n- tooz-tox-py310-memcached https://zuul.opendev.org/t/openstack/build/fc53d2a75ca547f5b9d50edd1ea01f27 : SUCCESS in 4m 35s\n- tooz-tox-py313-memcached https://zuul.opendev.org/t/openstack/build/f21e2c0bd5104b36b7f2a8ab95a5b90b : SUCCESS in 4m 55s\n- tooz-tox-py310-postgresql https://zuul.opendev.org/t/openstack/build/a8289b0b1b1d49c0b24d4dea924a592a : SUCCESS in 4m 36s\n- tooz-tox-py313-postgresql https://zuul.opendev.org/t/openstack/build/20958baeb06e438d914ca7fa99a7c844 : SUCCESS in 4m 46s\n- tooz-tox-py310-mysql https://zuul.opendev.org/t/openstack/build/3fd5c12b68f146929c11d7b490809b25 : SUCCESS in 4m 30s\n- tooz-tox-py313-mysql https://zuul.opendev.org/t/openstack/build/1fc033339e4c4abe935b635f3b3091d1 : SUCCESS in 7m 57s\n- tooz-tox-py310-consul https://zuul.opendev.org/t/openstack/build/29a33d8a06394c06bd02ca88d46b8fcf : SUCCESS in 4m 52s\n- tooz-tox-py313-consul https://zuul.opendev.org/t/openstack/build/6d00131629764de7b7614790a2eb48fa : SUCCESS in 4m 16s","accounts_in_message":[],"_revision_number":3},{"id":"97be9e519f161c6556687ee04e023d1969634c35","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-03-19 09:47:24.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"9350742cc7cf2da43953267488ec6c257bda9882","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-19 10:48:41.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/102dbf94bca649958a0fcacdaaf039de\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/cdb3227577ca4e739f81ee671cbd9115 : SUCCESS in 1m 41s\n- requirements-check https://zuul.opendev.org/t/openstack/build/6a871bc642b74a11bd57d3c3077c3202 : SUCCESS in 3m 59s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b3bf4637b4fb4820a4e27dd95dcda641 : SUCCESS in 56m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4e4e382d967d4ffab9530a03d48bb331 : SUCCESS in 3m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/afb3ebcb90684480bc310990bfca9847 : SUCCESS in 3m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e85a9cce132f44bfafda863281808c5d : SUCCESS in 3m 57s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/33755ec4ab594c9aad3bd40354903b05 : SUCCESS in 6m 20s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/bdee145a212f4733b80cde259e084596 : SUCCESS in 8m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a7689580f87e454680c594e6e430ed06 : SUCCESS in 5m 07s\n- tooz-tox-examples https://zuul.opendev.org/t/openstack/build/a363837fb4a9433e97cbaa679596f948 : SUCCESS in 5m 09s\n- tooz-tox-py310-etcd3gw https://zuul.opendev.org/t/openstack/build/6f5875f9a2b84b26848f6583fff5b407 : SUCCESS in 4m 31s\n- tooz-tox-py313-etcd3gw https://zuul.opendev.org/t/openstack/build/ec0bcc6544104cc4af5e6d5e4d358004 : SUCCESS in 4m 12s\n- tooz-tox-py310-zookeeper https://zuul.opendev.org/t/openstack/build/b3a790908e6d4cd4a3917a71161dbbf8 : SUCCESS in 4m 38s\n- tooz-tox-py313-zookeeper https://zuul.opendev.org/t/openstack/build/9736473ba4f64b77b0742a3f8ff17f55 : SUCCESS in 5m 08s\n- tooz-tox-py310-redis https://zuul.opendev.org/t/openstack/build/2fe5699368fb4ed6bac7a8ca042e26de : SUCCESS in 8m 15s\n- tooz-tox-py313-redis https://zuul.opendev.org/t/openstack/build/f34823c1ac9a422180fe3e3697334f1a : SUCCESS in 7m 06s\n- tooz-tox-py310-sentinel https://zuul.opendev.org/t/openstack/build/e166c269603f46ddb81636e0ed0379e3 : SUCCESS in 7m 48s\n- tooz-tox-py313-sentinel https://zuul.opendev.org/t/openstack/build/8894f395a659411f810a6286d1ca9249 : SUCCESS in 6m 43s\n- tooz-tox-py310-memcached https://zuul.opendev.org/t/openstack/build/718363418aec4ad79f67555b66d2fc59 : SUCCESS in 8m 53s\n- tooz-tox-py313-memcached https://zuul.opendev.org/t/openstack/build/2ee7ed33bb6a4389a47fd9f4efc89438 : SUCCESS in 5m 26s\n- tooz-tox-py310-postgresql https://zuul.opendev.org/t/openstack/build/59b1da4be585497786e3f8c37023c063 : SUCCESS in 6m 37s\n- tooz-tox-py313-postgresql https://zuul.opendev.org/t/openstack/build/6be3aefc420f481485988d7370a345d9 : SUCCESS in 4m 56s\n- tooz-tox-py310-mysql https://zuul.opendev.org/t/openstack/build/e5ccba29e24946f5901f6f4845273f47 : SUCCESS in 4m 25s\n- tooz-tox-py313-mysql https://zuul.opendev.org/t/openstack/build/24a339c18dfd4753b57f22105a5237ac : SUCCESS in 7m 24s\n- tooz-tox-py310-consul https://zuul.opendev.org/t/openstack/build/508bf2ee727645bebb93a883e820b491 : SUCCESS in 6m 37s\n- tooz-tox-py313-consul https://zuul.opendev.org/t/openstack/build/f16389e79dc847c3b41c7ce63fcc2ee1 : SUCCESS in 5m 48s","accounts_in_message":[],"_revision_number":4},{"id":"cdce4f8c10dd46d132f6c9b9f95f8a07053c0158","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-03-31 09:03:29.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"7aefc99331beb1af71d47c81c1067d981cb8ae46","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-31 10:24:32.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4947e53b3e30454bad6fd365d43d6871\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/dcd41994cecd4b79814e8bc4dc44292c : SUCCESS in 2m 48s\n- requirements-check https://zuul.opendev.org/t/openstack/build/eeaf7bf813b94d47ad2b30d5e6c00347 : SUCCESS in 3m 40s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/31de8e91c73349d5b6237568cdb45cb0 : SUCCESS in 1h 17m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fbfde3f5518f4d86979a561e27d5473c : SUCCESS in 4m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/daeb48516a234412be6b54ba87e994f4 : SUCCESS in 5m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f100eb5613cc40a59cfd624dc7a25dd5 : SUCCESS in 4m 43s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9cb1249d25a54741b1b3db018931072e : SUCCESS in 6m 13s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/1010f4c98b9c4f4f8817ebcf260726f9 : SUCCESS in 7m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b21c11192fae4f74a91de9538959d5d3 : SUCCESS in 3m 14s\n- tooz-tox-examples https://zuul.opendev.org/t/openstack/build/6e714a55dbd64472bbb68a0c23e11850 : SUCCESS in 5m 07s\n- tooz-tox-py310-etcd3gw https://zuul.opendev.org/t/openstack/build/d8008d7fe5a9483bad43d9eda551903d : SUCCESS in 5m 01s\n- tooz-tox-py313-etcd3gw https://zuul.opendev.org/t/openstack/build/5037febc62f147ddbb46347c05c9113d : SUCCESS in 7m 25s\n- tooz-tox-py310-zookeeper https://zuul.opendev.org/t/openstack/build/117d5da2e39b4cb48681eb6b72b228a4 : SUCCESS in 4m 57s\n- tooz-tox-py313-zookeeper https://zuul.opendev.org/t/openstack/build/0be9d6c8a2d547729b1e58b804a1b90f : SUCCESS in 3m 27s\n- tooz-tox-py310-redis https://zuul.opendev.org/t/openstack/build/492303a1368041f7ae0605c36ff5fcba : SUCCESS in 6m 40s\n- tooz-tox-py313-redis https://zuul.opendev.org/t/openstack/build/7ba08878b61742fd9942a02a7cb1bd43 : SUCCESS in 7m 22s\n- tooz-tox-py310-sentinel https://zuul.opendev.org/t/openstack/build/509108e69c3f4080a29def7da29fef90 : SUCCESS in 6m 15s\n- tooz-tox-py313-sentinel https://zuul.opendev.org/t/openstack/build/a9880db172ad4ca2bde4710e6aa5a5bd : SUCCESS in 8m 24s\n- tooz-tox-py310-memcached https://zuul.opendev.org/t/openstack/build/53858aeaf9a04abe90ebe70558561409 : SUCCESS in 4m 10s\n- tooz-tox-py313-memcached https://zuul.opendev.org/t/openstack/build/90b0e6c26c4e4790a9d035709009be7c : SUCCESS in 5m 05s\n- tooz-tox-py310-postgresql https://zuul.opendev.org/t/openstack/build/df27c8faee5a4e6c9fbfe44ebb8d63c8 : SUCCESS in 5m 27s\n- tooz-tox-py313-postgresql https://zuul.opendev.org/t/openstack/build/d782b85deb774e0faf50be337242db82 : SUCCESS in 12m 54s\n- tooz-tox-py310-mysql https://zuul.opendev.org/t/openstack/build/3c1ed4d6bbe64e2abeafb90945a8b6ff : SUCCESS in 4m 39s\n- tooz-tox-py313-mysql https://zuul.opendev.org/t/openstack/build/e44b46c61cfe4f4f9336fe164d36173a : SUCCESS in 7m 50s\n- tooz-tox-py310-consul https://zuul.opendev.org/t/openstack/build/6f6fc143d7804d208f61f818f5ba203d : SUCCESS in 6m 48s\n- tooz-tox-py313-consul https://zuul.opendev.org/t/openstack/build/ffdf3b10ce1444d3842619dc7d60ea3d : SUCCESS in 3m 24s","accounts_in_message":[],"_revision_number":5},{"id":"604c90fc574d9f362336a0561bbc17cdabe6ab08","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2026-04-09 20:29:08.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"b848e823da00ea62068f76e4789e6a2eae133be6","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-04-10 11:32:41.000000000","message":"Patch Set 5: Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"88196350e9d8e1862b46842ca37d35259ae80fbc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-10 11:33:08.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"4178908a9a90364783c57b913399de038cbc493e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-10 13:36:04.000000000","message":"Patch Set 5: Verified-2\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/tooz/+/980107 is needed.","accounts_in_message":[],"_revision_number":5},{"id":"812bdce0852dcc902862400771e72a7cff44c8db","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-04-10 18:13:45.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-2\n* Workflow+1\n","accounts_in_message":[],"_revision_number":6},{"id":"1f81282d81848503afaed9ea65cd5d7e0a96088b","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-04-10 18:17:40.000000000","message":"Patch Set 6: Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"565b0442f35fa78109a775e034aefdeb5ad05efd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-10 20:09:40.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c328c811a62e4beeba5a15f5eb180658\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/6e9f05721d5c4838a107c4c00c20daa0 : SUCCESS in 3m 21s\n- requirements-check https://zuul.opendev.org/t/openstack/build/7050be99d152481194b07e7b9c444f42 : SUCCESS in 3m 55s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d1b780de736f47bb826dccbb3f5f59e0 : SUCCESS in 1h 47m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/38f66bb8814e4b51857de7053901958a : SUCCESS in 6m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9838457f4a2c4bc19c6eaf30e3f1e772 : SUCCESS in 3m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d7b0362b5ef043a0ba12eea0312dc2e9 : SUCCESS in 5m 07s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/62eca68b421640a08338ae3a15f77668 : SUCCESS in 4m 59s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/1a1a307cccef4c538c2e84efee7214ec : SUCCESS in 7m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/919dca4abf5a44299f21c397d2b2793f : SUCCESS in 3m 36s\n- tooz-tox-examples https://zuul.opendev.org/t/openstack/build/5967f5160e5b48169f3384042ab4fd8b : SUCCESS in 3m 56s\n- tooz-tox-py310-etcd3gw https://zuul.opendev.org/t/openstack/build/51f99a731b28494bb5cef46110908f54 : SUCCESS in 6m 23s\n- tooz-tox-py313-etcd3gw https://zuul.opendev.org/t/openstack/build/e2f03b8c31dc4266a88d7c3b0880f393 : SUCCESS in 4m 09s\n- tooz-tox-py310-zookeeper https://zuul.opendev.org/t/openstack/build/f822f8a01a874955a8985506caa6094e : SUCCESS in 4m 17s\n- tooz-tox-py313-zookeeper https://zuul.opendev.org/t/openstack/build/3d8f2c0af5394291aeca900677364a06 : SUCCESS in 3m 41s\n- tooz-tox-py310-redis https://zuul.opendev.org/t/openstack/build/c9095976f8c240c69d7a73133703897f : SUCCESS in 8m 08s\n- tooz-tox-py313-redis https://zuul.opendev.org/t/openstack/build/06b3f5e84f7846feb88abfbc121548f4 : SUCCESS in 5m 56s\n- tooz-tox-py310-sentinel https://zuul.opendev.org/t/openstack/build/5f2bf9adeae74eefb740242b98c2fc33 : SUCCESS in 6m 17s\n- tooz-tox-py313-sentinel https://zuul.opendev.org/t/openstack/build/64ed3c7ac4d54ceea59c4fbbbb916d9f : SUCCESS in 7m 09s\n- tooz-tox-py310-memcached https://zuul.opendev.org/t/openstack/build/2677ef87d404429ebb73a575ea4913dd : SUCCESS in 3m 53s\n- tooz-tox-py313-memcached https://zuul.opendev.org/t/openstack/build/158e21b767da45ecb16dbcbd68440760 : SUCCESS in 3m 59s\n- tooz-tox-py310-postgresql https://zuul.opendev.org/t/openstack/build/21d37196cf8945c68714cd231e31d880 : SUCCESS in 5m 19s\n- tooz-tox-py313-postgresql https://zuul.opendev.org/t/openstack/build/b8f3f62341ae4b52959aa5378db5dc80 : SUCCESS in 4m 22s\n- tooz-tox-py310-mysql https://zuul.opendev.org/t/openstack/build/5fbb4412ad214688aa8cb18c37ad0f8a : SUCCESS in 3m 55s\n- tooz-tox-py313-mysql https://zuul.opendev.org/t/openstack/build/88a40e7cce514f87bb66a6ab845bcfc5 : SUCCESS in 8m 05s\n- tooz-tox-py310-consul https://zuul.opendev.org/t/openstack/build/c71c79a02f964627aa8d9e037f9bd94a : SUCCESS in 3m 41s\n- tooz-tox-py313-consul https://zuul.opendev.org/t/openstack/build/864022cd310c43e1a6488390b6256d54 : SUCCESS in 7m 43s","accounts_in_message":[],"_revision_number":6},{"id":"a377e750ee3162a8b662982f86d9cd0c5325dfd4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-10 20:35:16.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"4ea3c95f2dfdcb60d01a982df262d47d2ccc8140","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-10 22:38:43.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/40a9fd9684ec44c0956f970e0caf11b7\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/7f25c79ef7084a1282304382b4ff113f : SUCCESS in 2m 45s\n- requirements-check https://zuul.opendev.org/t/openstack/build/78649f6be1484ac285ae728395435b0c : SUCCESS in 4m 15s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cc95fce440274fe2bf02ee6f04e44759 : SUCCESS in 51m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4f15de650cd244f3b51309c98132eaf8 : SUCCESS in 4m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8a073599b4974635bb45df5dada0c88d : SUCCESS in 4m 46s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/347df594918f476ebec54ab7195a67a8 : SUCCESS in 5m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1126cfc8b4204c4e9940a86e726f071e : SUCCESS in 5m 03s\n- tooz-tox-examples https://zuul.opendev.org/t/openstack/build/e25d4847cfaf43b4b4991155857b2840 : SUCCESS in 3m 13s\n- tooz-tox-py310-etcd3gw https://zuul.opendev.org/t/openstack/build/89b4df765118454f9eaa7f3221338513 : SUCCESS in 5m 09s\n- tooz-tox-py313-etcd3gw https://zuul.opendev.org/t/openstack/build/178a13d48b924e099083dbac483ebce9 : SUCCESS in 6m 10s\n- tooz-tox-py310-zookeeper https://zuul.opendev.org/t/openstack/build/d3de1b8850dd4781bde23d17d4a626f7 : SUCCESS in 3m 22s\n- tooz-tox-py313-zookeeper https://zuul.opendev.org/t/openstack/build/a55dd7fdebce460d93a910437a4b62a6 : SUCCESS in 6m 08s\n- tooz-tox-py310-redis https://zuul.opendev.org/t/openstack/build/afeffc6a4e2543669f7caddaba7f92e8 : SUCCESS in 6m 00s\n- tooz-tox-py313-redis https://zuul.opendev.org/t/openstack/build/95023a305447453fbc257b76af751901 : SUCCESS in 7m 00s\n- tooz-tox-py310-sentinel https://zuul.opendev.org/t/openstack/build/4e6ae6d2e56a4368be51f28b1d2b0284 : SUCCESS in 6m 01s\n- tooz-tox-py313-sentinel https://zuul.opendev.org/t/openstack/build/334d65410bfc4b9e9235e0a568e0c6f0 : SUCCESS in 8m 23s\n- tooz-tox-py310-memcached https://zuul.opendev.org/t/openstack/build/2f23d43940c74f6297fba4e06bbb3e05 : SUCCESS in 4m 39s\n- tooz-tox-py313-memcached https://zuul.opendev.org/t/openstack/build/755f94edca1b4823b74fab71c7a3fc7f : SUCCESS in 4m 58s\n- tooz-tox-py310-postgresql https://zuul.opendev.org/t/openstack/build/38e27ffc0cc24d4ea312cff0ab8b1b24 : SUCCESS in 4m 07s\n- tooz-tox-py313-postgresql https://zuul.opendev.org/t/openstack/build/af265a8425e446178c6d94400ac7997e : SUCCESS in 6m 56s\n- tooz-tox-py310-mysql https://zuul.opendev.org/t/openstack/build/d70067cb8ef54a50ac292b5deeb5bb84 : SUCCESS in 4m 10s\n- tooz-tox-py313-mysql https://zuul.opendev.org/t/openstack/build/9c7156ddcb9c48cc9031d3564351b808 : SUCCESS in 5m 01s\n- tooz-tox-py310-consul https://zuul.opendev.org/t/openstack/build/34b993b1bfad4e8b9f96e53adc9bc99d : SUCCESS in 4m 57s\n- tooz-tox-py313-consul https://zuul.opendev.org/t/openstack/build/4c3e630fe52746ec9febedbb1b99e8e8 : SUCCESS in 5m 25s","accounts_in_message":[],"_revision_number":6},{"id":"a80d3e817deef2ca6736e02d8e6eb13970cc635f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-10 22:38:43.000000000","message":"Change has been successfully merged\n\n5 is the latest approved patch-set.\nNo files were changed between the latest approved patch-set and the submitted one.\n","accounts_in_message":[],"_revision_number":6},{"id":"e5990801d3e031b4874842a45650a63ef0968cf8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-10 22:40:30.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/92367c6a517a4b4baee50a03b7cc6085\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c56bd4fb5474402aea895493ff38de9 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"3d93c5952173f5d67d12d59f000ef0e500570678","revisions":{"1391f27faf436982b5c6b301ff3e8dddc73c24c3":{"kind":"REWORK","_number":1,"created":"2026-03-11 15:42:02.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/08/980108/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/08/980108/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/08/980108/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/08/980108/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/08/980108/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/08/980108/1"}}},"commit":{"parents":[{"commit":"c167d1582d90b423c6765f5a13ab57a5106cd892","subject":"typing: Add type hints to tooz.utils","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/c167d1582d90b423c6765f5a13ab57a5106cd892"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-03-09 16:00:12.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-03-11 15:41:56.000000000","tz":0},"subject":"typing: Add type hints to tooz._retry","message":"typing: Add type hints to tooz._retry\n\nChange-Id: I8f2ebecfc02ffed0fe9de376b436565b584c5b15\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/1391f27faf436982b5c6b301ff3e8dddc73c24c3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/1391f27faf436982b5c6b301ff3e8dddc73c24c3"}]},"branch":"refs/heads/master"},"29aa3c0d201d2240aa713e650dc27da9c4123ee4":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-03-12 08:42:07.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/08/980108/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/08/980108/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/08/980108/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/08/980108/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/08/980108/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/08/980108/2"}}},"commit":{"parents":[{"commit":"68c957d477c2da8a1eb206a3ec614e958d4b8443","subject":"typing: Add type hints to tooz.utils","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/68c957d477c2da8a1eb206a3ec614e958d4b8443"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-03-09 16:00:12.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-03-12 08:40:21.000000000","tz":0},"subject":"typing: Add type hints to tooz._retry","message":"typing: Add type hints to tooz._retry\n\nChange-Id: I8f2ebecfc02ffed0fe9de376b436565b584c5b15\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/29aa3c0d201d2240aa713e650dc27da9c4123ee4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/29aa3c0d201d2240aa713e650dc27da9c4123ee4"}]},"branch":"refs/heads/master"},"e36edb33e96ed7812166cbd03f80cf3cfac0f638":{"kind":"REWORK","_number":3,"created":"2026-03-14 14:40:20.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/08/980108/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/08/980108/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/08/980108/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/08/980108/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/08/980108/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/08/980108/3"}}},"commit":{"parents":[{"commit":"f4e9f08bd183cba1fbad5e4d03c1f6e0abfecaa9","subject":"typing: Add type hints to tooz.utils","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/f4e9f08bd183cba1fbad5e4d03c1f6e0abfecaa9"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-03-09 16:00:12.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-03-14 14:40:15.000000000","tz":0},"subject":"typing: Add type hints to tooz._retry","message":"typing: Add type hints to tooz._retry\n\nChange-Id: I8f2ebecfc02ffed0fe9de376b436565b584c5b15\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/e36edb33e96ed7812166cbd03f80cf3cfac0f638"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/e36edb33e96ed7812166cbd03f80cf3cfac0f638"}]},"branch":"refs/heads/master"},"af5cb11eea66d269ecf2aa6906da6e114647e6c1":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2026-03-19 09:47:24.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/08/980108/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/08/980108/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/08/980108/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/08/980108/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/08/980108/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/08/980108/4"}}},"commit":{"parents":[{"commit":"c2f176c4b119a1c5006ba5f31e7c4cdfb2016986","subject":"typing: Add type hints to tooz.utils","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/c2f176c4b119a1c5006ba5f31e7c4cdfb2016986"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-03-09 16:00:12.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-03-19 09:46:03.000000000","tz":60},"subject":"typing: Add type hints to tooz._retry","message":"typing: Add type hints to tooz._retry\n\nChange-Id: I8f2ebecfc02ffed0fe9de376b436565b584c5b15\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/af5cb11eea66d269ecf2aa6906da6e114647e6c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/af5cb11eea66d269ecf2aa6906da6e114647e6c1"}]},"branch":"refs/heads/master"},"780ad195a50ff9599343aa717d4ab4d16ba6e43a":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2026-03-31 09:03:29.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/08/980108/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/08/980108/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/08/980108/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/08/980108/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/08/980108/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/08/980108/5"}}},"commit":{"parents":[{"commit":"ebd48a9a9c3d392522ae71df4ff9e3441b4c8f76","subject":"typing: Add type hints to tooz.utils","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/ebd48a9a9c3d392522ae71df4ff9e3441b4c8f76"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-03-09 16:00:12.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-03-31 09:02:11.000000000","tz":120},"subject":"typing: Add type hints to tooz._retry","message":"typing: Add type hints to tooz._retry\n\nChange-Id: I8f2ebecfc02ffed0fe9de376b436565b584c5b15\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/780ad195a50ff9599343aa717d4ab4d16ba6e43a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/780ad195a50ff9599343aa717d4ab4d16ba6e43a"}]},"branch":"refs/heads/master"},"3d93c5952173f5d67d12d59f000ef0e500570678":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2026-04-10 18:13:45.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/08/980108/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/08/980108/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/08/980108/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/08/980108/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/08/980108/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/08/980108/6"}}},"commit":{"parents":[{"commit":"956ab013acde15308cf1bd037ab72006e7707d02","subject":"typing: Add type hints to tooz.utils","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/956ab013acde15308cf1bd037ab72006e7707d02"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-03-09 16:00:12.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-04-10 12:01:47.000000000","tz":60},"subject":"typing: Add type hints to tooz._retry","message":"typing: Add type hints to tooz._retry\n\nChange-Id: I8f2ebecfc02ffed0fe9de376b436565b584c5b15\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/3d93c5952173f5d67d12d59f000ef0e500570678"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/3d93c5952173f5d67d12d59f000ef0e500570678"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
