)]}'
{"id":"openstack%2Ftooz~980109","triplet_id":"openstack%2Ftooz~master~I5bee5070d65ef44d13617e9be2e5d4b2909536a3","project":"openstack/tooz","branch":"master","topic":"typing","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2026-04-10 22:49:50.000000000","reason":"\u003cGERRIT_ACCOUNT_15334\u003e replied on the change","reason_account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}}},"hashtags":[],"change_id":"I5bee5070d65ef44d13617e9be2e5d4b2909536a3","subject":"typing: Add type hints to tooz.locking","status":"MERGED","created":"2026-03-11 15:42:02.000000000","updated":"2026-04-11 03:13:09.000000000","submitted":"2026-04-11 03:12:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"980109-typing","meta_rev_id":"4eb18cbfc8006fe5ec7d0758efa238c79ba289c4","_number":980109,"virtual_id_number":980109,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-04-11 03:12:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2026-04-10 18:13:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2026-04-10 18:17:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-11 17:47:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-04-09 20:34:17.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"f7a92cb91dca6c308761bfe3af728ca045f87813","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-03-11 15:42:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"55440999085a9baff4f2e303af438c322dca3276","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-11 17:47:47.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/55967a64d67442a88c06350e468fe437\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/e387fd69bdf248faa4d09b0c053d69ab : SUCCESS in 5m 02s\n- requirements-check https://zuul.opendev.org/t/openstack/build/ad1745825d81480f8d6d1a7872caf9b8 : SUCCESS in 5m 19s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/119b2a817b1d4f178bf32e2d5e997fbc : SUCCESS in 1h 53m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2ed93251aeed4edc95d8e8eb51906979 : SUCCESS in 7m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a0658e37c72846fca8b9ffd6149ee332 : SUCCESS in 4m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3589320dc7c245a9a12e9247350f4bef : SUCCESS in 4m 49s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2e27070e681f4deeaf399caa7921f26a : SUCCESS in 6m 46s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/5241f0c7854849d38af568592fb81122 : SUCCESS in 6m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a4a759aaffd404680c5e6513d41d4a6 : SUCCESS in 5m 02s\n- tooz-tox-examples https://zuul.opendev.org/t/openstack/build/2742f839772d4d1c961f33b8e070d0d2 : SUCCESS in 7m 02s\n- tooz-tox-py310-etcd3gw https://zuul.opendev.org/t/openstack/build/2d04d0c0d42d473d8e857b899e387889 : FAILURE in 4m 49s\n- tooz-tox-py313-etcd3gw https://zuul.opendev.org/t/openstack/build/397605b66c624931a49b67ca0494e186 : FAILURE in 5m 19s\n- tooz-tox-py310-zookeeper https://zuul.opendev.org/t/openstack/build/a41e58c7fc3948b0a40dbc65ddd9783b : SUCCESS in 7m 29s\n- tooz-tox-py313-zookeeper https://zuul.opendev.org/t/openstack/build/9d9dd4612c6b4f4b86cb658ca5735ec9 : SUCCESS in 5m 07s\n- tooz-tox-py310-redis https://zuul.opendev.org/t/openstack/build/5d205308c21943a0b423f000b723c6bf : SUCCESS in 6m 04s\n- tooz-tox-py313-redis https://zuul.opendev.org/t/openstack/build/24d099034f694cb8b94712ada95a34c3 : SUCCESS in 6m 44s\n- tooz-tox-py310-sentinel https://zuul.opendev.org/t/openstack/build/72c8e98774ca4dd69aa92fdf6f324dd1 : SUCCESS in 6m 12s\n- tooz-tox-py313-sentinel https://zuul.opendev.org/t/openstack/build/5a5a1e923be4413fa2527d3ffaf9f72d : SUCCESS in 5m 07s\n- tooz-tox-py310-memcached https://zuul.opendev.org/t/openstack/build/38e58555e7824ca28c2e239cc8768eaa : SUCCESS in 3m 15s\n- tooz-tox-py313-memcached https://zuul.opendev.org/t/openstack/build/261f9b7a143147209da3ea55f192ce08 : SUCCESS in 4m 54s\n- tooz-tox-py310-postgresql https://zuul.opendev.org/t/openstack/build/177f1944fee84869a18d1e015657384c : SUCCESS in 4m 22s\n- tooz-tox-py313-postgresql https://zuul.opendev.org/t/openstack/build/b7f275b7306648da8ed51e2ccee56536 : SUCCESS in 6m 48s\n- tooz-tox-py310-mysql https://zuul.opendev.org/t/openstack/build/fdc545416b13482a877bf9ad1ef9dd93 : SUCCESS in 4m 03s\n- tooz-tox-py313-mysql https://zuul.opendev.org/t/openstack/build/0db4f5775ea541cbb455d571f5a2f40a : SUCCESS in 6m 36s\n- tooz-tox-py310-consul https://zuul.opendev.org/t/openstack/build/92276da666a74801b8b56ae10594b6ef : SUCCESS in 6m 22s\n- tooz-tox-py313-consul https://zuul.opendev.org/t/openstack/build/86d4fb92ce8544eb95db58b56775ed47 : SUCCESS in 4m 57s","accounts_in_message":[],"_revision_number":1},{"id":"ab0e769392dcea14fb01a079fa321cfda5d798c7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-03-12 08:42:07.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"a8704cfa836978dd588a7a363a713e5fa6eb2dd7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-12 10:27:25.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/51ea93195b2f4e5b98f94759382c7a28\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/53e4dea32a7f43639fe2f0dbbe95181c : SUCCESS in 3m 20s\n- requirements-check https://zuul.opendev.org/t/openstack/build/c8bb154dd7c74b9b831594e63389b9ff : SUCCESS in 3m 40s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/934e62ef6f374eb4968fd882c3c1632c : SUCCESS in 1h 31m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cc63879806d24880b545d0a008983a98 : SUCCESS in 4m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/97c1150d411f448f86a800f4e9d34ba8 : SUCCESS in 5m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/82bb237a6e654099a59b61558938335f : SUCCESS in 5m 43s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e0e0e54dcd94402d912ff91f6dafe80e : SUCCESS in 5m 14s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/255f08504cee419f8dd5646e9d577fd3 : SUCCESS in 7m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c7c85f52165e46ecb9bef53012fd5df5 : SUCCESS in 5m 28s\n- tooz-tox-examples https://zuul.opendev.org/t/openstack/build/8e6a2e95b7e749f885190a2f48e20fc1 : SUCCESS in 5m 11s\n- tooz-tox-py310-etcd3gw https://zuul.opendev.org/t/openstack/build/8017697a1582414ba5e28e300787b6fe : SUCCESS in 6m 56s\n- tooz-tox-py313-etcd3gw https://zuul.opendev.org/t/openstack/build/a46853a1fdb44d2aa53766314fe4afe6 : SUCCESS in 5m 52s\n- tooz-tox-py310-zookeeper https://zuul.opendev.org/t/openstack/build/7ccdef404e424ec2805b59c7a4e55997 : SUCCESS in 4m 38s\n- tooz-tox-py313-zookeeper https://zuul.opendev.org/t/openstack/build/cf566c068a9e49ad8be5dba8cb24e046 : SUCCESS in 3m 36s\n- tooz-tox-py310-redis https://zuul.opendev.org/t/openstack/build/848bac274d5d436587abc868ed15a49f : SUCCESS in 6m 29s\n- tooz-tox-py313-redis https://zuul.opendev.org/t/openstack/build/24f85958855f484ba1094ab3675c2ca1 : SUCCESS in 6m 34s\n- tooz-tox-py310-sentinel https://zuul.opendev.org/t/openstack/build/b39d034e21bc4b8da1efef2b1a2dcffb : SUCCESS in 5m 40s\n- tooz-tox-py313-sentinel https://zuul.opendev.org/t/openstack/build/74d48996cda64e309d56363a16faeec3 : SUCCESS in 8m 26s\n- tooz-tox-py310-memcached https://zuul.opendev.org/t/openstack/build/31db687c4d5c4bc7a3fec7e377a0b9f1 : SUCCESS in 3m 46s\n- tooz-tox-py313-memcached https://zuul.opendev.org/t/openstack/build/d7d5a759c84e42ef9ce4eea6af88f396 : SUCCESS in 6m 04s\n- tooz-tox-py310-postgresql https://zuul.opendev.org/t/openstack/build/0f567d0e23ee4f48862ede66eb5c9d48 : SUCCESS in 5m 35s\n- tooz-tox-py313-postgresql https://zuul.opendev.org/t/openstack/build/399aae45c8d74466bbf680cb1ad250c1 : SUCCESS in 4m 27s\n- tooz-tox-py310-mysql https://zuul.opendev.org/t/openstack/build/a852a20485cd4010943f58c300e41c70 : SUCCESS in 5m 10s\n- tooz-tox-py313-mysql https://zuul.opendev.org/t/openstack/build/2e440fd409e3499b868f2ecf0ff4fe76 : SUCCESS in 7m 54s\n- tooz-tox-py310-consul https://zuul.opendev.org/t/openstack/build/b31e3cfa943d41228d50e1aaca94ff2b : SUCCESS in 4m 53s\n- tooz-tox-py313-consul https://zuul.opendev.org/t/openstack/build/c416cb86733e4d9aaf41a5389bcadb87 : SUCCESS in 6m 18s","accounts_in_message":[],"_revision_number":2},{"id":"b9d0da4d2856c59ec6736cb732591b8f1780f5cb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-03-14 14:40:20.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"7b85ac330cc3bcad918a8d7e096ab86b8c979fd2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-14 16:14:28.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/02138465a8de4cf3ae3ec8856c4bf0e9\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/59da613a63ae4dadaa016106e0dea37f : SUCCESS in 2m 57s\n- requirements-check https://zuul.opendev.org/t/openstack/build/23376e30e333458990f9455b8914d185 : SUCCESS in 4m 34s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/da3cda5a68fd49798d773dc15ff59094 : SUCCESS in 1h 17m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a6bba794195749409dbee897c57835a8 : SUCCESS in 6m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/499c3c9b0f9945979a7a0dda540db7ac : SUCCESS in 7m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a0933652b0774f1ab0359be0612f1168 : SUCCESS in 4m 54s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7af1f8b9055c425eb6662decdc31f7eb : SUCCESS in 4m 58s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/509a45eeb59d4b68a8a9db0493cb8db7 : SUCCESS in 6m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a2dfeb184ae46a199aa70ec16559583 : SUCCESS in 3m 39s\n- tooz-tox-examples https://zuul.opendev.org/t/openstack/build/64fab778583a4f75bf05f4b72ef4c868 : SUCCESS in 4m 54s\n- tooz-tox-py310-etcd3gw https://zuul.opendev.org/t/openstack/build/8252450e9de74a5bbd299a10d0e340a2 : SUCCESS in 5m 10s\n- tooz-tox-py313-etcd3gw https://zuul.opendev.org/t/openstack/build/cf1f9dd7d4054e8d8776c48ff9b97f5c : SUCCESS in 5m 02s\n- tooz-tox-py310-zookeeper https://zuul.opendev.org/t/openstack/build/5f523ea1d79041adb84f18d0815442b0 : FAILURE in 4m 28s\n- tooz-tox-py313-zookeeper https://zuul.opendev.org/t/openstack/build/0f143e83fbc44821875e6c292bfaf682 : SUCCESS in 5m 04s\n- tooz-tox-py310-redis https://zuul.opendev.org/t/openstack/build/cc1ad1f4e62b472c8d5bc24ea471dd3f : SUCCESS in 5m 13s\n- tooz-tox-py313-redis https://zuul.opendev.org/t/openstack/build/c0e67cb6e1f842c69871ceda62a17fee : SUCCESS in 7m 01s\n- tooz-tox-py310-sentinel https://zuul.opendev.org/t/openstack/build/c7f0b79906f3468991cd2f05bd55d703 : SUCCESS in 6m 54s\n- tooz-tox-py313-sentinel https://zuul.opendev.org/t/openstack/build/81c95fe23cd8459cb9213e0d8c9f0198 : SUCCESS in 4m 46s\n- tooz-tox-py310-memcached https://zuul.opendev.org/t/openstack/build/9fee6756cd8f4f9f93aff7a79fcbd7d7 : SUCCESS in 4m 07s\n- tooz-tox-py313-memcached https://zuul.opendev.org/t/openstack/build/d29f06cd2a194fdc87933f5a1c6e1642 : SUCCESS in 7m 16s\n- tooz-tox-py310-postgresql https://zuul.opendev.org/t/openstack/build/a9c0555667ae4783af2b1538624846d9 : SUCCESS in 4m 08s\n- tooz-tox-py313-postgresql https://zuul.opendev.org/t/openstack/build/4c58e250689c4e1bb85ab68186d69276 : SUCCESS in 7m 44s\n- tooz-tox-py310-mysql https://zuul.opendev.org/t/openstack/build/12f699d7fa2a4386a8bdbb1bc5d3eafe : SUCCESS in 4m 45s\n- tooz-tox-py313-mysql https://zuul.opendev.org/t/openstack/build/6845d24af8ac4854b3e299043e3d09ff : SUCCESS in 9m 33s\n- tooz-tox-py310-consul https://zuul.opendev.org/t/openstack/build/cfc3bc7918e2403db1c259bc1e161f37 : SUCCESS in 4m 56s\n- tooz-tox-py313-consul https://zuul.opendev.org/t/openstack/build/0e8c14b4533e46f0bea34a2207636e6e : SUCCESS in 4m 13s","accounts_in_message":[],"_revision_number":3},{"id":"3352268acf4dfc632f012f7f393665e9a5251268","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-03-18 12:04:41.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"40637338393606e36825c726668a0ec740c80557","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-18 12:58:06.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c5602657e1ab49c29dde3517f04a00ea\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/435a4bc2721c436cbf92e910bb448b47 : SUCCESS in 1m 52s\n- requirements-check https://zuul.opendev.org/t/openstack/build/b377a350d9c047ab8673c7398e3440fc : SUCCESS in 2m 32s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ed0645c1a6394371a1a55271a4b146e3 : SUCCESS in 52m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/22a562ec77ec4ef49bde178b0eabe14a : SUCCESS in 4m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4363bb2894ca4f2183721d450c1f5c1d : SUCCESS in 3m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8a2c523243904fcb89467e06ebcfc894 : SUCCESS in 3m 18s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d10c77a4248d44fe8ca31855d12ba152 : SUCCESS in 5m 10s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/78a526639f0546189746d791fe0bfc4d : SUCCESS in 4m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/40d6a190160547f1b8ebad68fcb11ced : SUCCESS in 3m 19s\n- tooz-tox-examples https://zuul.opendev.org/t/openstack/build/750eb0013bb643b98b7e266339f44a93 : SUCCESS in 5m 04s\n- tooz-tox-py310-etcd3gw https://zuul.opendev.org/t/openstack/build/d80fc6a028b944da877e38fbfcc310fe : SUCCESS in 4m 44s\n- tooz-tox-py313-etcd3gw https://zuul.opendev.org/t/openstack/build/4be232ca14974417b06ba5d4018bc752 : SUCCESS in 5m 04s\n- tooz-tox-py310-zookeeper https://zuul.opendev.org/t/openstack/build/4c60bda019524f9798711600d5ffce12 : SUCCESS in 4m 19s\n- tooz-tox-py313-zookeeper https://zuul.opendev.org/t/openstack/build/84f03d5a02fa452a8219a328c4ccf049 : SUCCESS in 3m 43s\n- tooz-tox-py310-redis https://zuul.opendev.org/t/openstack/build/517642eb68554f5f80fe6d2c41a8a284 : SUCCESS in 5m 55s\n- tooz-tox-py313-redis https://zuul.opendev.org/t/openstack/build/77d6d84ddafa47de87109e970da68db0 : SUCCESS in 6m 27s\n- tooz-tox-py310-sentinel https://zuul.opendev.org/t/openstack/build/61c6963a0ad740f3a185227515cd0964 : SUCCESS in 6m 15s\n- tooz-tox-py313-sentinel https://zuul.opendev.org/t/openstack/build/f72a35d807a0430da8b7de505c446586 : SUCCESS in 6m 49s\n- tooz-tox-py310-memcached https://zuul.opendev.org/t/openstack/build/11f778c969be4b6097fb764cc88b6504 : SUCCESS in 4m 30s\n- tooz-tox-py313-memcached https://zuul.opendev.org/t/openstack/build/88fefce903d44944839b81cce16fc565 : SUCCESS in 3m 22s\n- tooz-tox-py310-postgresql https://zuul.opendev.org/t/openstack/build/df1e8b14295f4a2cbb9154ba644427c2 : SUCCESS in 4m 25s\n- tooz-tox-py313-postgresql https://zuul.opendev.org/t/openstack/build/c300318d9ece4f339292d37f1ff10624 : SUCCESS in 7m 36s\n- tooz-tox-py310-mysql https://zuul.opendev.org/t/openstack/build/e1cb2ee279264ab3ba1d68fd077aba9e : SUCCESS in 4m 21s\n- tooz-tox-py313-mysql https://zuul.opendev.org/t/openstack/build/949b6276f5a14453b4ea3ba26044f97a : SUCCESS in 7m 21s\n- tooz-tox-py310-consul https://zuul.opendev.org/t/openstack/build/f9f46a168c4c40c18115333e95dde061 : SUCCESS in 3m 18s\n- tooz-tox-py313-consul https://zuul.opendev.org/t/openstack/build/5263da13b3424e66b83827bfe4c7e0bc : SUCCESS in 4m 02s","accounts_in_message":[],"_revision_number":3},{"id":"e0cb3c667162b3d71936312ee7785487d64f0ed0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-03-19 09:47:24.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"616e5d29049e3150bc9e717b8edd46f60e1c0a1b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-19 11:37:42.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7a53de95f2764ecc8870d42dc27d6b99\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/f1700c567a1545b08fb071a47c33bfa6 : SUCCESS in 3m 18s\n- requirements-check https://zuul.opendev.org/t/openstack/build/1fd2638975954de4879934354ed4a886 : SUCCESS in 3m 36s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/55214eb4a7d3425fb608ee322e0c4c87 : SUCCESS in 1h 44m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/839f6453d15e4575a1437dddd1ebffb1 : SUCCESS in 6m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/299c8415fa7549cb9b3798ba83116611 : SUCCESS in 3m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/395d723c44054171bf5adb22d22ade36 : SUCCESS in 5m 22s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/196d10731ec0499d8f90f7acf0d6e483 : SUCCESS in 5m 31s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/51e6cfac3e364576b8bf8cab21771126 : SUCCESS in 9m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef10d0e15b474d59986091c461493c5e : SUCCESS in 5m 30s\n- tooz-tox-examples https://zuul.opendev.org/t/openstack/build/514c45f010894a5ab04474543944262f : SUCCESS in 5m 36s\n- tooz-tox-py310-etcd3gw https://zuul.opendev.org/t/openstack/build/5864e10960db48cb845a8eb5185e292e : SUCCESS in 5m 32s\n- tooz-tox-py313-etcd3gw https://zuul.opendev.org/t/openstack/build/c15d40e3001c43c991f8178db7ea5d67 : SUCCESS in 6m 03s\n- tooz-tox-py310-zookeeper https://zuul.opendev.org/t/openstack/build/b0e5af358c9c48229b82acf2ea3179be : SUCCESS in 5m 07s\n- tooz-tox-py313-zookeeper https://zuul.opendev.org/t/openstack/build/0b17a77344884d1c8fd6574cd85a8208 : SUCCESS in 5m 01s\n- tooz-tox-py310-redis https://zuul.opendev.org/t/openstack/build/350cb5f4137147a0bed4498dca34f6ac : SUCCESS in 6m 53s\n- tooz-tox-py313-redis https://zuul.opendev.org/t/openstack/build/00562e0561b249e9bbc1741b520da8de : SUCCESS in 6m 56s\n- tooz-tox-py310-sentinel https://zuul.opendev.org/t/openstack/build/49ee5787aed84bb09eca5f39c39fe72b : SUCCESS in 6m 14s\n- tooz-tox-py313-sentinel https://zuul.opendev.org/t/openstack/build/a0954e4b7fbf483db558114052c07b21 : SUCCESS in 6m 36s\n- tooz-tox-py310-memcached https://zuul.opendev.org/t/openstack/build/b78ba1bda03b456db7ed3a6a3fb4184e : SUCCESS in 4m 22s\n- tooz-tox-py313-memcached https://zuul.opendev.org/t/openstack/build/c476084967ce48babb7a36f6e0a8892f : SUCCESS in 5m 26s\n- tooz-tox-py310-postgresql https://zuul.opendev.org/t/openstack/build/e18d7e28f49145baab66839301a476e3 : SUCCESS in 5m 10s\n- tooz-tox-py313-postgresql https://zuul.opendev.org/t/openstack/build/4bcef90541554994ab6f4834eb6ee7fe : SUCCESS in 6m 28s\n- tooz-tox-py310-mysql https://zuul.opendev.org/t/openstack/build/5187cc138521420485c99bc1071bf655 : SUCCESS in 4m 37s\n- tooz-tox-py313-mysql https://zuul.opendev.org/t/openstack/build/2ad1b6c287bc42b28761577965e0a14a : SUCCESS in 8m 13s\n- tooz-tox-py310-consul https://zuul.opendev.org/t/openstack/build/ed23d4c11b234a1db73ed597d71dbf40 : SUCCESS in 3m 44s\n- tooz-tox-py313-consul https://zuul.opendev.org/t/openstack/build/f464c8134a4a4bffb5ab8cafe3ad9eee : SUCCESS in 5m 58s","accounts_in_message":[],"_revision_number":4},{"id":"a5127f1c032e13f3e3f54d748cf258d0ca72d040","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-03-31 09:03:29.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"6228fc92c532dfdea3c9fe365824ce5d06360604","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-31 11:17:45.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4001ba0af1ac4b58986017b09a0e8ad4\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/a971e49c5411451085f1b00bb3fa03d1 : SUCCESS in 2m 38s\n- requirements-check https://zuul.opendev.org/t/openstack/build/c8d7438a31af4d48b9b567a7cdac68d1 : SUCCESS in 3m 09s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9d78ca9a99834035aca738ad06a90d88 : SUCCESS in 2h 08m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4b465e763aa148609e46c38b0c25224a : SUCCESS in 6m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b29f52a355d842eba36306bf99545d67 : SUCCESS in 5m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5853374b7f6a41229d951e43fdd0ccdd : SUCCESS in 4m 36s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ec130f45a49640e5895c6c32ee1ac06c : SUCCESS in 3m 41s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/858ee3ec2af94b8f94eb1f3098db539a : SUCCESS in 4m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/58535ff619c549e4b3eaa030630487fc : SUCCESS in 5m 07s\n- tooz-tox-examples https://zuul.opendev.org/t/openstack/build/4ecfc331d2e5441ea7f6c8d5fc5f9354 : SUCCESS in 7m 00s\n- tooz-tox-py310-etcd3gw https://zuul.opendev.org/t/openstack/build/f932d7568cb64618b2cde63ff47dbde5 : SUCCESS in 5m 56s\n- tooz-tox-py313-etcd3gw https://zuul.opendev.org/t/openstack/build/510e2c5d6d5244feb95bfd671df1dfdd : SUCCESS in 5m 28s\n- tooz-tox-py310-zookeeper https://zuul.opendev.org/t/openstack/build/9cc5e203750a49b5b8539172a9c19eab : SUCCESS in 3m 36s\n- tooz-tox-py313-zookeeper https://zuul.opendev.org/t/openstack/build/b414fb90e3784433b1a05795ee8830ce : SUCCESS in 6m 48s\n- tooz-tox-py310-redis https://zuul.opendev.org/t/openstack/build/ebf1413a8a60419197ccb6996a3318a0 : SUCCESS in 5m 37s\n- tooz-tox-py313-redis https://zuul.opendev.org/t/openstack/build/f15cca242f584a8e9dca9ea130c09957 : SUCCESS in 5m 49s\n- tooz-tox-py310-sentinel https://zuul.opendev.org/t/openstack/build/075cb5455f85434a8f1633094c3d1952 : SUCCESS in 5m 30s\n- tooz-tox-py313-sentinel https://zuul.opendev.org/t/openstack/build/92a72d4539784c94846556c1d3f04c06 : SUCCESS in 7m 04s\n- tooz-tox-py310-memcached https://zuul.opendev.org/t/openstack/build/d76223a8b640452a8a04b206edf1134b : SUCCESS in 6m 40s\n- tooz-tox-py313-memcached https://zuul.opendev.org/t/openstack/build/1bca02dc906f45babe0d0709c6d1da82 : SUCCESS in 5m 55s\n- tooz-tox-py310-postgresql https://zuul.opendev.org/t/openstack/build/8c872392f57d4d20bc613f8dfd2cfce2 : SUCCESS in 3m 26s\n- tooz-tox-py313-postgresql https://zuul.opendev.org/t/openstack/build/66ea1b2a77934c1cb5bf13ba7b6a9621 : SUCCESS in 6m 11s\n- tooz-tox-py310-mysql https://zuul.opendev.org/t/openstack/build/9b1716df06774e06886cec73c91a952c : SUCCESS in 3m 03s\n- tooz-tox-py313-mysql https://zuul.opendev.org/t/openstack/build/d9a4c294d2a044bd923756eace5f02cf : SUCCESS in 8m 11s\n- tooz-tox-py310-consul https://zuul.opendev.org/t/openstack/build/f18bd95ab02e4fa2b0b1dfaa3430ba87 : SUCCESS in 4m 55s\n- tooz-tox-py313-consul https://zuul.opendev.org/t/openstack/build/1f16326885004ab6abd87e43514cd63f : SUCCESS in 6m 21s","accounts_in_message":[],"_revision_number":5},{"id":"6c43acdeb74efdcd054a8e77609ed3e8ffabd286","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2026-04-09 20:34:17.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"d4278b29213de6df151aa05a02d5b1a9450e1070","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-04-10 11:32:45.000000000","message":"Patch Set 5: Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"60aba7e1ccaa7126867f54e93bbd215fb9224d7e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-10 11:33:13.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"a21b9a2c4ff7bba3eb9e2b9d84f169ddf98c26a8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-10 13:36:04.000000000","message":"Patch Set 5: Verified-2\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/tooz/+/980108 is needed.","accounts_in_message":[],"_revision_number":5},{"id":"827b1adbb09ddc71c15a86fc3ee7971f251013e3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-04-10 18:13:45.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-2\n* Workflow+1\n","accounts_in_message":[],"_revision_number":6},{"id":"57658cba9beb7de835461ea9738ed542aa2ac53b","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-04-10 18:17:44.000000000","message":"Patch Set 6: Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"29b64cbe86cf13e9532640503c42d2f900a2a6c7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-10 20:31:35.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/72732c3fce9642ef92cb32bbc9733295\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/173c89197b284674ae452aadef9cd746 : SUCCESS in 2m 55s\n- requirements-check https://zuul.opendev.org/t/openstack/build/2da4dda6e1904fd9a7563adab1525445 : SUCCESS in 4m 23s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1a27bfcf5f664e14a6ff0e798d149447 : SUCCESS in 2h 07m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/139a2bc2ea524a2db1b0dbb610738c3c : SUCCESS in 6m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e7bb4ce25fb744bab33385dfd877c7f5 : SUCCESS in 5m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/21bd8fb787ce48348bd32d354631d221 : SUCCESS in 5m 40s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ae4721b75c754dad9a757cf03ee41877 : SUCCESS in 7m 41s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/4cd748108d114723921ae242af36f104 : SUCCESS in 9m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ac8ec69d393413eb32a2cbdde04aa2b : SUCCESS in 4m 02s\n- tooz-tox-examples https://zuul.opendev.org/t/openstack/build/e199f64273824d359df5a9648515018b : SUCCESS in 3m 43s\n- tooz-tox-py310-etcd3gw https://zuul.opendev.org/t/openstack/build/360273677660413582e25458f317a854 : SUCCESS in 5m 52s\n- tooz-tox-py313-etcd3gw https://zuul.opendev.org/t/openstack/build/b0d277d78473470a9bb95e8a53363e0f : SUCCESS in 7m 11s\n- tooz-tox-py310-zookeeper https://zuul.opendev.org/t/openstack/build/d9b1c4aa38da40098fe62fcf18e69089 : SUCCESS in 5m 00s\n- tooz-tox-py313-zookeeper https://zuul.opendev.org/t/openstack/build/e05f337069444c5d86aae2b02b2b78dd : SUCCESS in 6m 03s\n- tooz-tox-py310-redis https://zuul.opendev.org/t/openstack/build/39e2065d6c8a456da6337a952befaeb5 : SUCCESS in 6m 40s\n- tooz-tox-py313-redis https://zuul.opendev.org/t/openstack/build/39ae75ca6f7f4ab3ae6353c57fc69e70 : SUCCESS in 9m 52s\n- tooz-tox-py310-sentinel https://zuul.opendev.org/t/openstack/build/d8da6ca08eb34cc78f45a9f2f9aa0a76 : SUCCESS in 6m 51s\n- tooz-tox-py313-sentinel https://zuul.opendev.org/t/openstack/build/a801b15dc21a4150b127057fdf78c0b5 : SUCCESS in 9m 39s\n- tooz-tox-py310-memcached https://zuul.opendev.org/t/openstack/build/867470ee6e1b44538f2952b8e71e52f4 : SUCCESS in 6m 37s\n- tooz-tox-py313-memcached https://zuul.opendev.org/t/openstack/build/7df12d23cdae44ccaa569dc93b305fc1 : SUCCESS in 7m 35s\n- tooz-tox-py310-postgresql https://zuul.opendev.org/t/openstack/build/52895e1bc5104ddfb79d7fe67dcb3f00 : SUCCESS in 4m 17s\n- tooz-tox-py313-postgresql https://zuul.opendev.org/t/openstack/build/14ac0418865b4666aa68f85612acec48 : SUCCESS in 5m 47s\n- tooz-tox-py310-mysql https://zuul.opendev.org/t/openstack/build/b720a90d58f54a829a7eedc3fa15e270 : SUCCESS in 5m 03s\n- tooz-tox-py313-mysql https://zuul.opendev.org/t/openstack/build/413986d73ec34760b304c5a72615bb9c : SUCCESS in 7m 52s\n- tooz-tox-py310-consul https://zuul.opendev.org/t/openstack/build/8ee692204b614f309564dfd36347a54f : SUCCESS in 5m 36s\n- tooz-tox-py313-consul https://zuul.opendev.org/t/openstack/build/22cfee1b8b2d45b4b905576031536856 : SUCCESS in 6m 28s","accounts_in_message":[],"_revision_number":6},{"id":"7b9b9508f8b32ac75d9e0c5f7bb9b0d910780605","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-10 20:35:19.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"b324d44dd5aacbf2d513ab283bc1acaa1b4f4c95","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-10 22:38:45.000000000","message":"Patch Set 6: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1b26c1b7d7784d758a58c0b0d7484034\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/2777a14d70344ee58f83d590bc047089 : SUCCESS in 2m 50s\n- requirements-check https://zuul.opendev.org/t/openstack/build/ee4d104e89bb41eca63e90c45d8fb0d5 : SUCCESS in 3m 59s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1fbf4768ab7a4d1d88bb4b8579d41bb9 : FAILURE in 1h 17m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/51b2ff4643414e969d50863997115e41 : SUCCESS in 5m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4a639296bb954ee09d9fd52486db4a1e : SUCCESS in 5m 03s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/66dc555e84f54fe4b329f953f0961611 : SUCCESS in 6m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb0a54856b9b450c8a1faee2dd7638ee : SUCCESS in 4m 54s\n- tooz-tox-examples https://zuul.opendev.org/t/openstack/build/661cc762c6db4c4996d48b45095454d2 : SUCCESS in 6m 23s\n- tooz-tox-py310-etcd3gw https://zuul.opendev.org/t/openstack/build/86be9ec90fa54f32bc750ad1141814f3 : SUCCESS in 8m 48s\n- tooz-tox-py313-etcd3gw https://zuul.opendev.org/t/openstack/build/770a4b7cca4c4547a80ca24395dca756 : SUCCESS in 5m 52s\n- tooz-tox-py310-zookeeper https://zuul.opendev.org/t/openstack/build/b8eb1ece4f964eb9be1ae6f6d70530cb : SUCCESS in 4m 39s\n- tooz-tox-py313-zookeeper https://zuul.opendev.org/t/openstack/build/00452628238f47fd8fa13d73809e8750 : SUCCESS in 5m 05s\n- tooz-tox-py310-redis https://zuul.opendev.org/t/openstack/build/e336f89008794def87673daf2e3b53e4 : SUCCESS in 6m 21s\n- tooz-tox-py313-redis https://zuul.opendev.org/t/openstack/build/610c6ca498e347f1b0717af95fa6158c : SUCCESS in 10m 17s\n- tooz-tox-py310-sentinel https://zuul.opendev.org/t/openstack/build/118699207f4141f1851b514070297ca3 : SUCCESS in 7m 28s\n- tooz-tox-py313-sentinel https://zuul.opendev.org/t/openstack/build/6c482189dc7c496196af487056c26189 : SUCCESS in 6m 40s\n- tooz-tox-py310-memcached https://zuul.opendev.org/t/openstack/build/ed5e426d45da4ee5b6a76097de570a01 : SUCCESS in 4m 29s\n- tooz-tox-py313-memcached https://zuul.opendev.org/t/openstack/build/f66705f9c3e44b12ad953f1767b7b700 : SUCCESS in 9m 48s\n- tooz-tox-py310-postgresql https://zuul.opendev.org/t/openstack/build/2902e14479344b11abb73c4f0bfffb82 : SUCCESS in 4m 42s\n- tooz-tox-py313-postgresql https://zuul.opendev.org/t/openstack/build/e1eca462d943468184cb8370289fed01 : SUCCESS in 4m 45s\n- tooz-tox-py310-mysql https://zuul.opendev.org/t/openstack/build/1bde226f0f5340609196451e82b2fbf9 : SUCCESS in 4m 42s\n- tooz-tox-py313-mysql https://zuul.opendev.org/t/openstack/build/3b98017a0e8547a8816b597185aa653a : SUCCESS in 7m 57s\n- tooz-tox-py310-consul https://zuul.opendev.org/t/openstack/build/373aead9680945189b4c222e48df618e : SUCCESS in 5m 33s\n- tooz-tox-py313-consul https://zuul.opendev.org/t/openstack/build/9fd5ad1d98ca41bf9a20cebbd23e9724 : SUCCESS in 5m 41s","accounts_in_message":[],"_revision_number":6},{"id":"e4f7456507c3c0710ed846faed77c529fb66bbc4","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-04-10 22:49:50.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"87731fea20d3f9aeb8e1b8c4d00c50475d007427","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-11 01:08:47.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a6dc37ddef8840b6a0c39a3ba59fbc78\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/8024d81082584968895f7fd742199736 : SUCCESS in 2m 39s\n- requirements-check https://zuul.opendev.org/t/openstack/build/7041faaee74d4f63bca9b3709feb5809 : SUCCESS in 3m 44s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d79fc5a90ea54a84a441f74e6ba245b5 : SUCCESS in 2h 14m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2a5b146e69ce4024bff5b670c7f582a1 : SUCCESS in 4m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4dd13c884567440989bc1aa098d433ed : SUCCESS in 5m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c5a405fec1ba4478b6a77770e1f2ff4b : SUCCESS in 4m 43s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4ec37ed8674342909f4c5a1c3caf35b9 : SUCCESS in 5m 32s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/ae6007dc2ef24579add65f0654f32453 : FAILURE in 5m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa3dfc5a61a94531b25c7d58c62b5f4e : SUCCESS in 4m 57s\n- tooz-tox-examples https://zuul.opendev.org/t/openstack/build/56fb504c20404a30965a1698e5e90cab : SUCCESS in 4m 52s\n- tooz-tox-py310-etcd3gw https://zuul.opendev.org/t/openstack/build/fd407d1e2ce4473bb9424710cae3fc38 : SUCCESS in 5m 03s\n- tooz-tox-py313-etcd3gw https://zuul.opendev.org/t/openstack/build/e9a1cb7d8b8a4dffa54729b6d3cdd564 : SUCCESS in 5m 25s\n- tooz-tox-py310-zookeeper https://zuul.opendev.org/t/openstack/build/2ee9b9641506493db287ee3eae00755e : SUCCESS in 4m 19s\n- tooz-tox-py313-zookeeper https://zuul.opendev.org/t/openstack/build/e71d29aba7734d679051a7198163ed21 : SUCCESS in 5m 09s\n- tooz-tox-py310-redis https://zuul.opendev.org/t/openstack/build/65109842547a4960a616f0d5c7d6b3a4 : SUCCESS in 7m 27s\n- tooz-tox-py313-redis https://zuul.opendev.org/t/openstack/build/42711c18fb8c4df2aac57aa36136244f : SUCCESS in 8m 34s\n- tooz-tox-py310-sentinel https://zuul.opendev.org/t/openstack/build/c18f978db96549ccb1278890b19e7976 : SUCCESS in 5m 55s\n- tooz-tox-py313-sentinel https://zuul.opendev.org/t/openstack/build/ea87986a18284831a8aba4c7ef85dc78 : SUCCESS in 7m 30s\n- tooz-tox-py310-memcached https://zuul.opendev.org/t/openstack/build/db22ca1a0e4740eeb4db88645c8a6c90 : SUCCESS in 5m 33s\n- tooz-tox-py313-memcached https://zuul.opendev.org/t/openstack/build/0e7814f920274a79a8dadbe0b4c36075 : SUCCESS in 5m 19s\n- tooz-tox-py310-postgresql https://zuul.opendev.org/t/openstack/build/34f92052ca1c46cab6fce43b656da3ae : SUCCESS in 4m 10s\n- tooz-tox-py313-postgresql https://zuul.opendev.org/t/openstack/build/27489d6f73ee4ff0add9f852b8617614 : SUCCESS in 3m 42s\n- tooz-tox-py310-mysql https://zuul.opendev.org/t/openstack/build/970d0536c3764e06ad36420aa1f188dd : SUCCESS in 4m 35s\n- tooz-tox-py313-mysql https://zuul.opendev.org/t/openstack/build/cde3dd12dbd5415cb674672a7cfbf9a1 : SUCCESS in 8m 17s\n- tooz-tox-py310-consul https://zuul.opendev.org/t/openstack/build/94ed3967e6474ba2886eeafad24f821d : SUCCESS in 4m 49s\n- tooz-tox-py313-consul https://zuul.opendev.org/t/openstack/build/453e4704fb9245bc98b289642e5ac5f9 : SUCCESS in 7m 37s","accounts_in_message":[],"_revision_number":6},{"id":"ae959f41f2e7b8d38b110acfd14e22e81d7d488d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-11 01:09:07.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"947d166c07da081250c40c0f21875863b5338754","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-11 03:12:09.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e5e4ba27f0d04344949e98757503bb20\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/7dea4e2fc062421895fae08e6de37723 : SUCCESS in 2m 14s\n- requirements-check https://zuul.opendev.org/t/openstack/build/cdfb22eb08374d658eaed8e6b2c6f727 : SUCCESS in 4m 20s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c01b13ed6ad04fdd9aa2c90fc9033388 : SUCCESS in 1h 58m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/92c147e7339f4b46b501f75c3736111a : SUCCESS in 5m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5790070c59104db5a68da24a6a5b9826 : SUCCESS in 4m 06s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b09d816acba14852a77ebc0d1c566880 : SUCCESS in 5m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a405494008994bfda054f1822e48ad3e : SUCCESS in 3m 02s\n- tooz-tox-examples https://zuul.opendev.org/t/openstack/build/79657da6841642b98b2c602de147ef25 : SUCCESS in 4m 11s\n- tooz-tox-py310-etcd3gw https://zuul.opendev.org/t/openstack/build/2b9078b2e3b24a279a33debde319b527 : SUCCESS in 5m 17s\n- tooz-tox-py313-etcd3gw https://zuul.opendev.org/t/openstack/build/e1dd8f96add64325adaa5e64bae4ccf2 : SUCCESS in 7m 03s\n- tooz-tox-py310-zookeeper https://zuul.opendev.org/t/openstack/build/aa06950208aa481a9c3e6d6f15846de3 : SUCCESS in 4m 11s\n- tooz-tox-py313-zookeeper https://zuul.opendev.org/t/openstack/build/19b7cccb7a0542bd86d480578b6b9ad3 : SUCCESS in 4m 17s\n- tooz-tox-py310-redis https://zuul.opendev.org/t/openstack/build/e235bc58f2ec4a67b2b12ccabf85c2c4 : SUCCESS in 4m 46s\n- tooz-tox-py313-redis https://zuul.opendev.org/t/openstack/build/d9236cff3dd14395aa31f975bd8397d3 : SUCCESS in 6m 54s\n- tooz-tox-py310-sentinel https://zuul.opendev.org/t/openstack/build/5bbf1ee2891f4dcb80f0337506f78504 : SUCCESS in 5m 20s\n- tooz-tox-py313-sentinel https://zuul.opendev.org/t/openstack/build/f4dd8ef3129a41ae88d569338ed0d6d0 : SUCCESS in 4m 59s\n- tooz-tox-py310-memcached https://zuul.opendev.org/t/openstack/build/c0d459053b56406a9ac9f9dda85d8dc4 : SUCCESS in 4m 29s\n- tooz-tox-py313-memcached https://zuul.opendev.org/t/openstack/build/3dcd5479ee9e42d681e325a7db9584c2 : SUCCESS in 4m 58s\n- tooz-tox-py310-postgresql https://zuul.opendev.org/t/openstack/build/3960c5a7253c41aea032b8cb6e4d0de3 : SUCCESS in 3m 55s\n- tooz-tox-py313-postgresql https://zuul.opendev.org/t/openstack/build/0b31ad41ae404c04a7d6584aa30bf39c : SUCCESS in 5m 31s\n- tooz-tox-py310-mysql https://zuul.opendev.org/t/openstack/build/05eb4de59f074de5ab6c7967e128e8fa : SUCCESS in 3m 59s\n- tooz-tox-py313-mysql https://zuul.opendev.org/t/openstack/build/895b0b50c80d4c96af52ed48f90fec66 : SUCCESS in 7m 23s\n- tooz-tox-py310-consul https://zuul.opendev.org/t/openstack/build/905dd3744c494c7bb5f18d586f7182d6 : SUCCESS in 4m 20s\n- tooz-tox-py313-consul https://zuul.opendev.org/t/openstack/build/3d49305af873444a976becc8796e376f : SUCCESS in 7m 37s","accounts_in_message":[],"_revision_number":6},{"id":"d879366c8fb016b788255591b7225651d31c9122","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-11 03:12:10.000000000","message":"Change has been successfully merged\n\n5 is the latest approved patch-set.\nNo files were changed between the latest approved patch-set and the submitted one.\n","accounts_in_message":[],"_revision_number":6},{"id":"4eb18cbfc8006fe5ec7d0758efa238c79ba289c4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-11 03:13:09.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/627f748141a14c438c69365fedb02051\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a270a80a55f4c6db39f42d01287b50c : SUCCESS in 42s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"32c396e8db5ab3675a33c7c175d040ed7dd7a17f","revisions":{"fe579d1b3b53f18c1a110c4be81a1b8ba1dc1433":{"kind":"REWORK","_number":1,"created":"2026-03-11 15:42:02.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/09/980109/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/09/980109/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/09/980109/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/09/980109/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/09/980109/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/09/980109/1"}}},"commit":{"parents":[{"commit":"1391f27faf436982b5c6b301ff3e8dddc73c24c3","subject":"typing: Add type hints to tooz._retry","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/1391f27faf436982b5c6b301ff3e8dddc73c24c3"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-03-09 16:11:16.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-03-11 15:41:56.000000000","tz":0},"subject":"typing: Add type hints to tooz.locking","message":"typing: Add type hints to tooz.locking\n\nChange-Id: I5bee5070d65ef44d13617e9be2e5d4b2909536a3\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/fe579d1b3b53f18c1a110c4be81a1b8ba1dc1433"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/fe579d1b3b53f18c1a110c4be81a1b8ba1dc1433"}]},"branch":"refs/heads/master"},"1be0ae2062aa9ebc30f1ab3385bbb531f25c4fa5":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-03-12 08:42:07.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/09/980109/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/09/980109/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/09/980109/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/09/980109/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/09/980109/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/09/980109/2"}}},"commit":{"parents":[{"commit":"29aa3c0d201d2240aa713e650dc27da9c4123ee4","subject":"typing: Add type hints to tooz._retry","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/29aa3c0d201d2240aa713e650dc27da9c4123ee4"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-03-09 16:11:16.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-03-12 08:40:21.000000000","tz":0},"subject":"typing: Add type hints to tooz.locking","message":"typing: Add type hints to tooz.locking\n\nChange-Id: I5bee5070d65ef44d13617e9be2e5d4b2909536a3\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/1be0ae2062aa9ebc30f1ab3385bbb531f25c4fa5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/1be0ae2062aa9ebc30f1ab3385bbb531f25c4fa5"}]},"branch":"refs/heads/master"},"7b927563054a29d213122081d4b8f40df7f436e8":{"kind":"REWORK","_number":3,"created":"2026-03-14 14:40:20.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/09/980109/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/09/980109/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/09/980109/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/09/980109/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/09/980109/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/09/980109/3"}}},"commit":{"parents":[{"commit":"e36edb33e96ed7812166cbd03f80cf3cfac0f638","subject":"typing: Add type hints to tooz._retry","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/e36edb33e96ed7812166cbd03f80cf3cfac0f638"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-03-09 16:11:16.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-03-14 14:40:15.000000000","tz":0},"subject":"typing: Add type hints to tooz.locking","message":"typing: Add type hints to tooz.locking\n\nChange-Id: I5bee5070d65ef44d13617e9be2e5d4b2909536a3\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/7b927563054a29d213122081d4b8f40df7f436e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/7b927563054a29d213122081d4b8f40df7f436e8"}]},"branch":"refs/heads/master"},"b9a3cbe26ec6cb3db37e10e98363e4b8c3dfb0c9":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2026-03-19 09:47:24.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/09/980109/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/09/980109/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/09/980109/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/09/980109/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/09/980109/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/09/980109/4"}}},"commit":{"parents":[{"commit":"af5cb11eea66d269ecf2aa6906da6e114647e6c1","subject":"typing: Add type hints to tooz._retry","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/af5cb11eea66d269ecf2aa6906da6e114647e6c1"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-03-09 16:11:16.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-03-19 09:46:03.000000000","tz":60},"subject":"typing: Add type hints to tooz.locking","message":"typing: Add type hints to tooz.locking\n\nChange-Id: I5bee5070d65ef44d13617e9be2e5d4b2909536a3\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/b9a3cbe26ec6cb3db37e10e98363e4b8c3dfb0c9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/b9a3cbe26ec6cb3db37e10e98363e4b8c3dfb0c9"}]},"branch":"refs/heads/master"},"0f7b7e57dec9c27b6f9f9246e13757e53c6746a4":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2026-03-31 09:03:29.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/09/980109/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/09/980109/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/09/980109/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/09/980109/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/09/980109/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/09/980109/5"}}},"commit":{"parents":[{"commit":"780ad195a50ff9599343aa717d4ab4d16ba6e43a","subject":"typing: Add type hints to tooz._retry","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/780ad195a50ff9599343aa717d4ab4d16ba6e43a"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-03-09 16:11:16.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-03-31 09:02:11.000000000","tz":120},"subject":"typing: Add type hints to tooz.locking","message":"typing: Add type hints to tooz.locking\n\nChange-Id: I5bee5070d65ef44d13617e9be2e5d4b2909536a3\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/0f7b7e57dec9c27b6f9f9246e13757e53c6746a4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/0f7b7e57dec9c27b6f9f9246e13757e53c6746a4"}]},"branch":"refs/heads/master"},"32c396e8db5ab3675a33c7c175d040ed7dd7a17f":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2026-04-10 18:13:45.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/09/980109/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/09/980109/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/09/980109/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/09/980109/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/09/980109/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/09/980109/6"}}},"commit":{"parents":[{"commit":"3d93c5952173f5d67d12d59f000ef0e500570678","subject":"typing: Add type hints to tooz._retry","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/3d93c5952173f5d67d12d59f000ef0e500570678"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-03-09 16:11:16.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-04-10 12:02:21.000000000","tz":60},"subject":"typing: Add type hints to tooz.locking","message":"typing: Add type hints to tooz.locking\n\nChange-Id: I5bee5070d65ef44d13617e9be2e5d4b2909536a3\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/32c396e8db5ab3675a33c7c175d040ed7dd7a17f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/32c396e8db5ab3675a33c7c175d040ed7dd7a17f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
