)]}'
{"id":"openstack%2Ftooz~980110","triplet_id":"openstack%2Ftooz~master~Id0e1ead0db5b58fc1e19bedb67893dc001304d39","project":"openstack/tooz","branch":"master","topic":"typing","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2026-04-10 22:49:55.000000000","reason":"\u003cGERRIT_ACCOUNT_15334\u003e replied on the change","reason_account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}}},"hashtags":[],"change_id":"Id0e1ead0db5b58fc1e19bedb67893dc001304d39","subject":"typing: Add type hints to tooz.hashring","status":"MERGED","created":"2026-03-11 15:42:02.000000000","updated":"2026-04-11 03:14:03.000000000","submitted":"2026-04-11 03:12:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"980110-typing","meta_rev_id":"db27d8682e4d465c60bb406aa38cd911732c6138","_number":980110,"virtual_id_number":980110,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-04-11 03:12:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2026-04-10 18:13:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2026-04-10 18:17:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-11 16:48:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-04-09 20:36:05.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"90b733fa2497d6eb0dc3c538da345b0d1ace52ca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-03-11 15:42:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ff59a487f66eddc188a5f5725e4b1f1e9a8bfb0b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-11 16:48:37.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/73201744da904cb2a22355fcef17eb02\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/641f19378b8d4c9096080cf99199a776 : SUCCESS in 2m 40s\n- requirements-check https://zuul.opendev.org/t/openstack/build/051426e9c60f4eaabd27d36137c2a61f : SUCCESS in 6m 17s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a45dd05127bb4f0eb1ce49cb223df692 : SUCCESS in 52m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e45c834f450d4f34a6755d6ef396b02b : SUCCESS in 5m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/42fa8e17cdfa492eaec4325c4b23d204 : SUCCESS in 4m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/30588104267047a18c53958f7d48c2ed : SUCCESS in 4m 45s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2e8407a53c5140388ae9978978c91bef : SUCCESS in 7m 21s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/56bc47fc3e194ba192e37f28026660b7 : SUCCESS in 10m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cede8f5bea1440ff8ce8ba476132c1ca : SUCCESS in 5m 10s\n- tooz-tox-examples https://zuul.opendev.org/t/openstack/build/759f985d492844d4a5405e4a5aed12e0 : SUCCESS in 6m 16s\n- tooz-tox-py310-etcd3gw https://zuul.opendev.org/t/openstack/build/680fa5f8f6104f00aaef55c54f3d9c67 : FAILURE in 3m 33s\n- tooz-tox-py313-etcd3gw https://zuul.opendev.org/t/openstack/build/540a1a9723194644822b60f9452a6742 : FAILURE in 5m 29s\n- tooz-tox-py310-zookeeper https://zuul.opendev.org/t/openstack/build/6a48ecef2c40425583ceb79e09f7e480 : SUCCESS in 5m 08s\n- tooz-tox-py313-zookeeper https://zuul.opendev.org/t/openstack/build/6bdf2889ef0b43df99942d91d92b3427 : SUCCESS in 4m 43s\n- tooz-tox-py310-redis https://zuul.opendev.org/t/openstack/build/910b61e5c1cb45c99683216806a461dd : SUCCESS in 7m 16s\n- tooz-tox-py313-redis https://zuul.opendev.org/t/openstack/build/b8eebd2703b54a219bb646c2419e0c7d : SUCCESS in 6m 38s\n- tooz-tox-py310-sentinel https://zuul.opendev.org/t/openstack/build/cdcc2df6265d436baddd7e94c811ace6 : SUCCESS in 9m 19s\n- tooz-tox-py313-sentinel https://zuul.opendev.org/t/openstack/build/bc710d5d1898429d9f53b64fe1d2a36c : SUCCESS in 8m 42s\n- tooz-tox-py310-memcached https://zuul.opendev.org/t/openstack/build/e18a863a966341e2aad4ec56370a8be8 : SUCCESS in 4m 25s\n- tooz-tox-py313-memcached https://zuul.opendev.org/t/openstack/build/2af87507f77b4495b41b7e9f6fc87c2c : SUCCESS in 4m 37s\n- tooz-tox-py310-postgresql https://zuul.opendev.org/t/openstack/build/1fe9335ac60a4b449527fc4863cdf467 : SUCCESS in 4m 51s\n- tooz-tox-py313-postgresql https://zuul.opendev.org/t/openstack/build/23a525db80e847e19a00e25995b288d7 : SUCCESS in 3m 33s\n- tooz-tox-py310-mysql https://zuul.opendev.org/t/openstack/build/afb8c966047441e5ad6b7631954b98a5 : SUCCESS in 6m 06s\n- tooz-tox-py313-mysql https://zuul.opendev.org/t/openstack/build/bcb1d0164d724871b5807bb663936333 : SUCCESS in 7m 11s\n- tooz-tox-py310-consul https://zuul.opendev.org/t/openstack/build/79e562924e684b12ada85b701353930d : SUCCESS in 3m 24s\n- tooz-tox-py313-consul https://zuul.opendev.org/t/openstack/build/90ed1cc60a4d463eae1a8e20f12da4ac : SUCCESS in 5m 42s","accounts_in_message":[],"_revision_number":1},{"id":"f380636c4e0b0c7d7bbe96208e442c1fba3e501a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-03-12 08:42:07.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"4c1355b0598f44d982225acdaa5ae9608c973143","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-12 10:42:40.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cd8697b6bfdc4813874ae81aded1beeb\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/88c7e597a2a542829855eda9da364775 : SUCCESS in 2m 51s\n- requirements-check https://zuul.opendev.org/t/openstack/build/87dac5325e9a47129fa08103564fdda7 : SUCCESS in 4m 42s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ff1438a531444aadbb66bf66b94dd6e9 : SUCCESS in 1h 38m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/665c3838c32b44278e4eebaa66f89f00 : SUCCESS in 6m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0ad8fa3fca7046ad9098d8bb6edabfe4 : SUCCESS in 6m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e57aa88d67ca4562ac1ab224f3a5ea27 : SUCCESS in 4m 15s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1fe545f89948466685a4eb470a5960cb : SUCCESS in 4m 19s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/10e29107a8ea47d7b8b39d92fae72ca7 : FAILURE in 5m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cefdc3a326594892995af615aa069001 : SUCCESS in 5m 55s\n- tooz-tox-examples https://zuul.opendev.org/t/openstack/build/f67227191b9e44ffba8b84b05b1cd6d5 : SUCCESS in 5m 20s\n- tooz-tox-py310-etcd3gw https://zuul.opendev.org/t/openstack/build/c7b4f20775b3497da1f99669f7319669 : SUCCESS in 5m 12s\n- tooz-tox-py313-etcd3gw https://zuul.opendev.org/t/openstack/build/55ccc4fdd29b4c59b34b042c5d34902c : SUCCESS in 6m 24s\n- tooz-tox-py310-zookeeper https://zuul.opendev.org/t/openstack/build/418fd64e0a6a41b7bc06231c413e951b : SUCCESS in 4m 40s\n- tooz-tox-py313-zookeeper https://zuul.opendev.org/t/openstack/build/0019ce02b8cc49b78227a429412b6810 : SUCCESS in 5m 19s\n- tooz-tox-py310-redis https://zuul.opendev.org/t/openstack/build/5ab354599df749bb80789f5bb7bf3c94 : SUCCESS in 6m 42s\n- tooz-tox-py313-redis https://zuul.opendev.org/t/openstack/build/dc7f03901c92466a91254f8fa610d01e : SUCCESS in 6m 29s\n- tooz-tox-py310-sentinel https://zuul.opendev.org/t/openstack/build/df33cea5ec904069acdf54aaa296ee6a : SUCCESS in 6m 16s\n- tooz-tox-py313-sentinel https://zuul.opendev.org/t/openstack/build/068f6f2a7d2748e59bc7040e27544bad : SUCCESS in 7m 26s\n- tooz-tox-py310-memcached https://zuul.opendev.org/t/openstack/build/aa64d6edc81243bdb13e4c9ca973430b : SUCCESS in 5m 07s\n- tooz-tox-py313-memcached https://zuul.opendev.org/t/openstack/build/df6ecc7aac0a4a2b827e3e125ad95c5c : SUCCESS in 4m 25s\n- tooz-tox-py310-postgresql https://zuul.opendev.org/t/openstack/build/f6479cb0d16c4d37a484526b5bfb8e39 : SUCCESS in 4m 56s\n- tooz-tox-py313-postgresql https://zuul.opendev.org/t/openstack/build/3897dab4c3e741429b451257388799fd : SUCCESS in 5m 13s\n- tooz-tox-py310-mysql https://zuul.opendev.org/t/openstack/build/fab7c39af0ed4a26af681e90467a5060 : SUCCESS in 4m 57s\n- tooz-tox-py313-mysql https://zuul.opendev.org/t/openstack/build/0180e3eadb1e46a7ae680e615bb26438 : SUCCESS in 8m 04s\n- tooz-tox-py310-consul https://zuul.opendev.org/t/openstack/build/0fa2c431ab0e4acaa062d7433461c71f : SUCCESS in 4m 26s\n- tooz-tox-py313-consul https://zuul.opendev.org/t/openstack/build/4f7e51d2e3e3485b9fe2c4bf8bf61ff3 : SUCCESS in 5m 24s","accounts_in_message":[],"_revision_number":2},{"id":"846d2cf66c38874c67e9bcbfb9f98bdc8b5c8cbe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-03-14 14:40:20.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"76792462115e827c6ca2c53182450361b7267135","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-14 15:58:19.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/15717432a20449ed930b8497aa40fdbc\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/619ce644f7eb4b019bab33183975d7aa : SUCCESS in 2m 24s\n- requirements-check https://zuul.opendev.org/t/openstack/build/62769bb78ecf4601b5744ca3e88dce48 : SUCCESS in 3m 45s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f0eacfdad98249ca9ca681a8546c8b3b : SUCCESS in 1h 00m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f5cd8f41ed9f4d77897d2ffc1268a131 : SUCCESS in 5m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ec3901211d3f47088eda3f8ed7583d88 : SUCCESS in 5m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/35c76623c86e4846a2582263d55a22b9 : SUCCESS in 4m 50s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d85e86ce34f84f1885b3cc819f297f01 : SUCCESS in 5m 21s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/395908722dba44df951f702746e9fbe6 : SUCCESS in 9m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d6f916c663dd42949b47b994a02e5ab1 : SUCCESS in 3m 25s\n- tooz-tox-examples https://zuul.opendev.org/t/openstack/build/9c663275f2c8492cba71fdeea279d4d1 : SUCCESS in 6m 23s\n- tooz-tox-py310-etcd3gw https://zuul.opendev.org/t/openstack/build/b98cdda605a44a1393fe608a76c64128 : SUCCESS in 4m 40s\n- tooz-tox-py313-etcd3gw https://zuul.opendev.org/t/openstack/build/f8ad9c386a4544968a4a534d76031f38 : SUCCESS in 4m 44s\n- tooz-tox-py310-zookeeper https://zuul.opendev.org/t/openstack/build/7ea5f80d203d483d9dccf875f0fe5a10 : SUCCESS in 4m 10s\n- tooz-tox-py313-zookeeper https://zuul.opendev.org/t/openstack/build/9d6916cec5d74d2ab1091fe5c79ad858 : SUCCESS in 7m 12s\n- tooz-tox-py310-redis https://zuul.opendev.org/t/openstack/build/e80c75f28fce4bd8b001fd56cd04d41d : SUCCESS in 6m 27s\n- tooz-tox-py313-redis https://zuul.opendev.org/t/openstack/build/2462a7b5942343ba988b4a64b3713b6c : SUCCESS in 7m 43s\n- tooz-tox-py310-sentinel https://zuul.opendev.org/t/openstack/build/dd0a808dd2144dcf982459f0924c844e : SUCCESS in 6m 33s\n- tooz-tox-py313-sentinel https://zuul.opendev.org/t/openstack/build/8d995504c9c64aa7bc37c332b0e4863c : SUCCESS in 6m 29s\n- tooz-tox-py310-memcached https://zuul.opendev.org/t/openstack/build/ce2fec757bde4f049ff69b3e38a1df26 : SUCCESS in 4m 26s\n- tooz-tox-py313-memcached https://zuul.opendev.org/t/openstack/build/584977dd1fbd41ff83f2f15a02bd0ab5 : SUCCESS in 5m 37s\n- tooz-tox-py310-postgresql https://zuul.opendev.org/t/openstack/build/3f8a93e372c64184954bab09fb22d6c4 : SUCCESS in 3m 57s\n- tooz-tox-py313-postgresql https://zuul.opendev.org/t/openstack/build/b0223bbc29934502a8f2fe9079302e10 : SUCCESS in 5m 00s\n- tooz-tox-py310-mysql https://zuul.opendev.org/t/openstack/build/0a70704a5c144f85949657132193e07e : SUCCESS in 4m 43s\n- tooz-tox-py313-mysql https://zuul.opendev.org/t/openstack/build/65e9efd4d83746d1b4598eae4a65dd82 : SUCCESS in 9m 22s\n- tooz-tox-py310-consul https://zuul.opendev.org/t/openstack/build/1b7c78db482f47d1bdc5804f287d1362 : SUCCESS in 4m 38s\n- tooz-tox-py313-consul https://zuul.opendev.org/t/openstack/build/3ad1e47e848540bba6e1aa37db124964 : SUCCESS in 5m 35s","accounts_in_message":[],"_revision_number":3},{"id":"13cfe4eb270d81c3bda1af4e79a7fcfd5cd60ecf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-03-19 09:47:24.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"a83ad94c57a1a8e593ac69a1153c6f7cbee47509","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-19 11:37:57.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f700ec47aae949d7875585cad9ad7743\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/ed02ea1f829449c1b10d4679e2ca4295 : SUCCESS in 2m 40s\n- requirements-check https://zuul.opendev.org/t/openstack/build/808be3761a1242e58c748f97e2dca95b : SUCCESS in 6m 42s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/463706338ebd4f6cad9c5f927d41e856 : SUCCESS in 1h 40m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0cc9dfeed8d14395b86b3a3919bad3f2 : SUCCESS in 6m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d1a986ba0ae84d199420012cc5539998 : SUCCESS in 5m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/594942313c1a40b39979212188db1f75 : SUCCESS in 6m 13s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5f1e815312a04c2aaeb0721ea5cb313b : SUCCESS in 7m 04s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/2653e4b0a7b047d1bf1eae8c7e23a388 : SUCCESS in 8m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/06d8a06f67944bf6ac21056ef40ffb5e : SUCCESS in 6m 44s\n- tooz-tox-examples https://zuul.opendev.org/t/openstack/build/067ea543908048e7b688211842277164 : SUCCESS in 6m 11s\n- tooz-tox-py310-etcd3gw https://zuul.opendev.org/t/openstack/build/43dd0d72c48a4615b62daa0258477309 : SUCCESS in 6m 38s\n- tooz-tox-py313-etcd3gw https://zuul.opendev.org/t/openstack/build/74bd116fbada47c4b91b9116d0988e8d : SUCCESS in 7m 23s\n- tooz-tox-py310-zookeeper https://zuul.opendev.org/t/openstack/build/491888705f7242179227916d8ba9d7a6 : SUCCESS in 5m 35s\n- tooz-tox-py313-zookeeper https://zuul.opendev.org/t/openstack/build/f7df2eca06294b45a14005246c9ce7e7 : SUCCESS in 4m 59s\n- tooz-tox-py310-redis https://zuul.opendev.org/t/openstack/build/cf6643c7cf4b46dca187df4de18b28fe : SUCCESS in 7m 27s\n- tooz-tox-py313-redis https://zuul.opendev.org/t/openstack/build/7cfaf6623dae4fd8b469b4986d6dff17 : SUCCESS in 8m 36s\n- tooz-tox-py310-sentinel https://zuul.opendev.org/t/openstack/build/0ed01f39df644ababff5b4c5969f9af4 : SUCCESS in 6m 48s\n- tooz-tox-py313-sentinel https://zuul.opendev.org/t/openstack/build/7c54a124a051423db6dcfd4b1aa8103e : SUCCESS in 4m 53s\n- tooz-tox-py310-memcached https://zuul.opendev.org/t/openstack/build/11c6b1eb2ffc472b843c959f06e1d4ee : SUCCESS in 2m 56s\n- tooz-tox-py313-memcached https://zuul.opendev.org/t/openstack/build/640115f79a3248328d266336e7eeb800 : SUCCESS in 5m 39s\n- tooz-tox-py310-postgresql https://zuul.opendev.org/t/openstack/build/649b9e99ef5f4a3f9a0bb26f94276a0b : SUCCESS in 5m 14s\n- tooz-tox-py313-postgresql https://zuul.opendev.org/t/openstack/build/33bfa870db154cd0930384c792b3b22d : SUCCESS in 6m 23s\n- tooz-tox-py310-mysql https://zuul.opendev.org/t/openstack/build/d74f119b6f394bf8af10c40b37b2a8c8 : SUCCESS in 4m 57s\n- tooz-tox-py313-mysql https://zuul.opendev.org/t/openstack/build/395539b8582c4d13bbe30f7bc8de3ad4 : SUCCESS in 6m 19s\n- tooz-tox-py310-consul https://zuul.opendev.org/t/openstack/build/97402541161b40ca94b0a1fb2cd9d946 : SUCCESS in 5m 25s\n- tooz-tox-py313-consul https://zuul.opendev.org/t/openstack/build/7555e3f90c294136a71df8f92ef2f8e9 : SUCCESS in 5m 31s","accounts_in_message":[],"_revision_number":4},{"id":"6a23346e8fa9a59b7f9a9284b4f9e0ddd5ba64f8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-03-31 09:03:29.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"f3609988e5556b5bc17d294fd3b5e7aa7382357d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-31 10:12:39.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fc5369dc62c14cd68b92c933ae65f5a4\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/b5b6d097021445788a37d03665238dc1 : SUCCESS in 3m 14s\n- requirements-check https://zuul.opendev.org/t/openstack/build/bdf59a82978c494599e1b04aa73c5331 : SUCCESS in 3m 43s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6fad85ac42b04049b42b38bfb395e44d : SUCCESS in 1h 06m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e116d2564cc541bc94dcc84e054e2deb : SUCCESS in 6m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3ba7b8c2f1b04d449957e79486af7253 : SUCCESS in 6m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c6fb7db9406546d88614e23ffd85d496 : SUCCESS in 6m 33s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7d037780360f416cab8b342c6a911247 : SUCCESS in 5m 27s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/1327ed01a08d44b7acbbf891d4b2afba : SUCCESS in 8m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b0a27b3af99c40d8a1043058530b529a : SUCCESS in 3m 22s\n- tooz-tox-examples https://zuul.opendev.org/t/openstack/build/31c8eab17caa4e27a1365f6fdaa22fdb : SUCCESS in 3m 14s\n- tooz-tox-py310-etcd3gw https://zuul.opendev.org/t/openstack/build/fc8b814348d54b58900086a8c87e41c8 : SUCCESS in 5m 26s\n- tooz-tox-py313-etcd3gw https://zuul.opendev.org/t/openstack/build/d5aee8baad76493ca347fea9a74713c6 : SUCCESS in 6m 32s\n- tooz-tox-py310-zookeeper https://zuul.opendev.org/t/openstack/build/0a55f311f3ab439991361e105ffe1dd7 : SUCCESS in 3m 01s\n- tooz-tox-py313-zookeeper https://zuul.opendev.org/t/openstack/build/b3a04deec49944d3aa69ad9982a93db7 : SUCCESS in 3m 43s\n- tooz-tox-py310-redis https://zuul.opendev.org/t/openstack/build/c67687144dea43f79f8f2bf059a53a21 : SUCCESS in 6m 50s\n- tooz-tox-py313-redis https://zuul.opendev.org/t/openstack/build/d182ce33595348358028d1b0725f6379 : SUCCESS in 11m 16s\n- tooz-tox-py310-sentinel https://zuul.opendev.org/t/openstack/build/110592161ca044f2bd1edc11048f56ae : SUCCESS in 6m 35s\n- tooz-tox-py313-sentinel https://zuul.opendev.org/t/openstack/build/fdbd8c9ba9fe4c05b8dbce828bcc3e66 : SUCCESS in 7m 40s\n- tooz-tox-py310-memcached https://zuul.opendev.org/t/openstack/build/194e028020c74dbbb5053e5f48759ca3 : SUCCESS in 6m 01s\n- tooz-tox-py313-memcached https://zuul.opendev.org/t/openstack/build/1e7332b71c5c4197aa77c9940670dc39 : SUCCESS in 7m 16s\n- tooz-tox-py310-postgresql https://zuul.opendev.org/t/openstack/build/20cd686dd29e4ca4848976b4e381b344 : SUCCESS in 5m 01s\n- tooz-tox-py313-postgresql https://zuul.opendev.org/t/openstack/build/97e6b47027a444de9643e87b51d20ef7 : SUCCESS in 6m 24s\n- tooz-tox-py310-mysql https://zuul.opendev.org/t/openstack/build/7855eedf08744e1292f2a14593604de4 : SUCCESS in 4m 41s\n- tooz-tox-py313-mysql https://zuul.opendev.org/t/openstack/build/712d0d8039a646f0903cc9e3d2a7bf33 : SUCCESS in 7m 45s\n- tooz-tox-py310-consul https://zuul.opendev.org/t/openstack/build/283adeaf627a4c47a98f24431f1c5116 : SUCCESS in 3m 33s\n- tooz-tox-py313-consul https://zuul.opendev.org/t/openstack/build/766f3506e5da43b8b9c7cc659948425c : SUCCESS in 6m 03s","accounts_in_message":[],"_revision_number":5},{"id":"119cf2a2698a0537b228a474b83b3a6864200020","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2026-04-09 20:36:05.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"c02194cd6765ed650da40c4241722d577ef1729f","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-04-10 11:32:48.000000000","message":"Patch Set 5: Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"ea856c5d8c3e2e3d7860ab6c0551679befe75a0b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-10 11:33:19.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"bb0c19af9c362f1cedcd978bfbe12be9a84f62fe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-10 13:36:05.000000000","message":"Patch Set 5: Verified-2\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/tooz/+/980109 is needed.","accounts_in_message":[],"_revision_number":5},{"id":"d90ac874f04c4ce5b4f66ccd5a5e515162a1a349","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-04-10 18:13:45.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-2\n* Workflow+1\n","accounts_in_message":[],"_revision_number":6},{"id":"81bc7374f3fa1816bc2d6dae49323d6ba4084a49","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-04-10 18:17:48.000000000","message":"Patch Set 6: Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"baccdf0e182105f2efaa7a708c4f7bf908c83f4d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-10 20:22:41.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e89807c552fc419ca92c920f71f50e86\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/b1bc2cb31d884f2ea02a15f155d30702 : SUCCESS in 2m 43s\n- requirements-check https://zuul.opendev.org/t/openstack/build/d9efa32e074c479fa60474cc4986b7b5 : SUCCESS in 3m 55s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b13cfa3269964c5a9e5234aa986a80cb : SUCCESS in 1h 58m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bb7f8015d6e548b6ae113d5a022f4d19 : SUCCESS in 4m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/971348040ffa4697870076a21c27f18d : SUCCESS in 5m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3c8d8b028a3e4ea3a74186ac2dacee2a : SUCCESS in 4m 30s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/cd5eb0f0741f40c0b90330c1c9dcd067 : SUCCESS in 8m 31s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/527be4f2b60a476690a3d81738198ca3 : SUCCESS in 9m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bff6a6efea5d4a8497d763d628232209 : SUCCESS in 5m 51s\n- tooz-tox-examples https://zuul.opendev.org/t/openstack/build/190cb5b77dcb4de8b61c991ee2188955 : SUCCESS in 5m 52s\n- tooz-tox-py310-etcd3gw https://zuul.opendev.org/t/openstack/build/cfb38a0298b74cba8667a9cb5a84c238 : SUCCESS in 5m 48s\n- tooz-tox-py313-etcd3gw https://zuul.opendev.org/t/openstack/build/75a5e16993b8436f921749750ae36451 : SUCCESS in 6m 07s\n- tooz-tox-py310-zookeeper https://zuul.opendev.org/t/openstack/build/b4862c8224d04ee3a1d0d35e1a87c506 : SUCCESS in 5m 14s\n- tooz-tox-py313-zookeeper https://zuul.opendev.org/t/openstack/build/07dfc137ac4944f9ab105c4e8dfbdbbb : SUCCESS in 8m 43s\n- tooz-tox-py310-redis https://zuul.opendev.org/t/openstack/build/6e976ae79e6b418eb4047dc4f262b3de : SUCCESS in 6m 30s\n- tooz-tox-py313-redis https://zuul.opendev.org/t/openstack/build/66068f85fe484f4490babe3a4dc02ff0 : SUCCESS in 7m 20s\n- tooz-tox-py310-sentinel https://zuul.opendev.org/t/openstack/build/98c5954000164c19b8c00952831563de : SUCCESS in 6m 20s\n- tooz-tox-py313-sentinel https://zuul.opendev.org/t/openstack/build/db00e634bf954fab8cc3185820f8a3c2 : SUCCESS in 7m 22s\n- tooz-tox-py310-memcached https://zuul.opendev.org/t/openstack/build/62fe5ea81f4e4a39b74fb4ebb08dd254 : SUCCESS in 5m 55s\n- tooz-tox-py313-memcached https://zuul.opendev.org/t/openstack/build/d9a114d0960a4eb3baac8582102207af : SUCCESS in 5m 44s\n- tooz-tox-py310-postgresql https://zuul.opendev.org/t/openstack/build/c095d1eded354f2d8b21a4733c4b7b15 : SUCCESS in 4m 50s\n- tooz-tox-py313-postgresql https://zuul.opendev.org/t/openstack/build/129ba66756a84b59ac039a070e535cf3 : SUCCESS in 5m 00s\n- tooz-tox-py310-mysql https://zuul.opendev.org/t/openstack/build/6e4d38bc8ab74381848ba7e6df7e0a86 : SUCCESS in 3m 33s\n- tooz-tox-py313-mysql https://zuul.opendev.org/t/openstack/build/50f88835cf7147a9b732aebc65e8ae5a : SUCCESS in 8m 07s\n- tooz-tox-py310-consul https://zuul.opendev.org/t/openstack/build/3055b603397f4dfd85e9f4ffe46b85d1 : SUCCESS in 4m 54s\n- tooz-tox-py313-consul https://zuul.opendev.org/t/openstack/build/a6f071adef5b4f48926d816c5c3f834b : SUCCESS in 6m 12s","accounts_in_message":[],"_revision_number":6},{"id":"1d2c52bf5a523416f3c04373b5cfd5020c2a1baa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-10 20:35:24.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"2a0b124b39d8934da0b7bfbf6221818a0ac8359a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-10 22:38:47.000000000","message":"Patch Set 6: Verified-2\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/tooz/+/980109 is needed.","accounts_in_message":[],"_revision_number":6},{"id":"92c4f7f35792dfefe8f156d49b6db1d02ba1b982","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-04-10 22:49:55.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"291c737712fd4c04b6a709ae5a957ff1a6b603ac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-11 00:54:37.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/44cb1a4867234ef5ac43d9ab5048a5ab\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/747e148785854a00a1562f6540de9db5 : SUCCESS in 1m 40s\n- requirements-check https://zuul.opendev.org/t/openstack/build/9175a4f07df1400da67e1809fcfbf607 : SUCCESS in 3m 40s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/44768ab568264037826781eade68ffa7 : SUCCESS in 1h 55m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b4f5c6d5917149a3b716a1372bd50db3 : SUCCESS in 7m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e57ae9311deb4333b414d9cfd2ddcd26 : SUCCESS in 8m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0fc1a87e8a684d2d8eb482a9901d1ba5 : SUCCESS in 5m 59s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/564c765c2e58403c81c1794eb1c0a1e5 : SUCCESS in 5m 13s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/79c4e6923d0f4242b64776545b95545f : FAILURE in 8m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/455f527405814476892a86853f1274b2 : SUCCESS in 5m 35s\n- tooz-tox-examples https://zuul.opendev.org/t/openstack/build/2f0f97b9f1854daaa2e820c92d1a640c : SUCCESS in 4m 49s\n- tooz-tox-py310-etcd3gw https://zuul.opendev.org/t/openstack/build/459cb2f271044205835cdff0406e0c09 : SUCCESS in 3m 49s\n- tooz-tox-py313-etcd3gw https://zuul.opendev.org/t/openstack/build/ae5239d01c8e4f9e97a63388ac3f78c9 : SUCCESS in 5m 32s\n- tooz-tox-py310-zookeeper https://zuul.opendev.org/t/openstack/build/768652a869d4463392916a593e2226dc : SUCCESS in 4m 23s\n- tooz-tox-py313-zookeeper https://zuul.opendev.org/t/openstack/build/55cec2b0f1c246ba923ed3b2ba3d205e : SUCCESS in 5m 14s\n- tooz-tox-py310-redis https://zuul.opendev.org/t/openstack/build/69c24880d17042629bccdcce13dbc6f3 : SUCCESS in 6m 28s\n- tooz-tox-py313-redis https://zuul.opendev.org/t/openstack/build/2c29e820277b48dba68cf6a52a63a6b8 : SUCCESS in 8m 00s\n- tooz-tox-py310-sentinel https://zuul.opendev.org/t/openstack/build/3038bafde5c94d9695fc60360a994f76 : SUCCESS in 6m 37s\n- tooz-tox-py313-sentinel https://zuul.opendev.org/t/openstack/build/797e1e795ba445ac88a67406e2e0e6b9 : SUCCESS in 5m 11s\n- tooz-tox-py310-memcached https://zuul.opendev.org/t/openstack/build/3a09c2d612da46d69620fd841b0bbd32 : SUCCESS in 4m 35s\n- tooz-tox-py313-memcached https://zuul.opendev.org/t/openstack/build/fdeb0c2c2876489eb04cf8fed7dde3de : SUCCESS in 5m 11s\n- tooz-tox-py310-postgresql https://zuul.opendev.org/t/openstack/build/939c5282cfe74273a4ee20544d37d232 : SUCCESS in 4m 14s\n- tooz-tox-py313-postgresql https://zuul.opendev.org/t/openstack/build/681d40f8313d43f59a8523e6e14f92fd : SUCCESS in 5m 02s\n- tooz-tox-py310-mysql https://zuul.opendev.org/t/openstack/build/6ff61debaf8749b1b66e5fc5c3fbd2c1 : SUCCESS in 4m 12s\n- tooz-tox-py313-mysql https://zuul.opendev.org/t/openstack/build/9271769eb5284b1a910c3a2c393d2c58 : SUCCESS in 7m 24s\n- tooz-tox-py310-consul https://zuul.opendev.org/t/openstack/build/35228035d2e547d59259c3effb5630d2 : SUCCESS in 5m 19s\n- tooz-tox-py313-consul https://zuul.opendev.org/t/openstack/build/def20ca7892c4901b6c24f90747cc7cc : SUCCESS in 5m 39s","accounts_in_message":[],"_revision_number":6},{"id":"2cbf8e1bd66d8a6908639ec74528357090a7a0e9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-11 01:09:11.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"355c2b55f2c41980558485d0ee4bb7133e3c740d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-11 03:12:13.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d4057308638f48b68bf73db721f696ee\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/2bdb99fffd804014b551ce0f45e20dff : SUCCESS in 2m 07s\n- requirements-check https://zuul.opendev.org/t/openstack/build/f2e10661a5db46e985f2823bce89d609 : SUCCESS in 3m 19s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bc72aeb8b0064ab3b2aff85402dea8e0 : SUCCESS in 54m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dcf65375e18445dbb850fa96d180e175 : SUCCESS in 5m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/66bc2332a4ea477ba360c585cd6c4da8 : SUCCESS in 3m 53s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8f623b18c953470d988b03d130b367f5 : SUCCESS in 4m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2402c69937b34d1ca77914086f7c48b4 : SUCCESS in 2m 45s\n- tooz-tox-examples https://zuul.opendev.org/t/openstack/build/03bf100a5a994c36a11244b441797df8 : SUCCESS in 3m 16s\n- tooz-tox-py310-etcd3gw https://zuul.opendev.org/t/openstack/build/be848137305c4f1c912794aef5c0081e : SUCCESS in 4m 13s\n- tooz-tox-py313-etcd3gw https://zuul.opendev.org/t/openstack/build/f0c2d2bb45d34032a19989041eaca9f9 : SUCCESS in 5m 26s\n- tooz-tox-py310-zookeeper https://zuul.opendev.org/t/openstack/build/7e8496767bda40ada98ddc035e10bca6 : SUCCESS in 4m 17s\n- tooz-tox-py313-zookeeper https://zuul.opendev.org/t/openstack/build/2169664c27d647f49a1f034ccd9d30d2 : SUCCESS in 4m 38s\n- tooz-tox-py310-redis https://zuul.opendev.org/t/openstack/build/014cbc2750a4486b8f29b7e6e5f7cf95 : SUCCESS in 6m 02s\n- tooz-tox-py313-redis https://zuul.opendev.org/t/openstack/build/35bf913a98cc404faf58ac1bb43c45d4 : SUCCESS in 5m 29s\n- tooz-tox-py310-sentinel https://zuul.opendev.org/t/openstack/build/1247d1ca29eb4204990babbacef02cb4 : SUCCESS in 6m 07s\n- tooz-tox-py313-sentinel https://zuul.opendev.org/t/openstack/build/05d158c98df84436baca23fe73eb90a5 : SUCCESS in 9m 54s\n- tooz-tox-py310-memcached https://zuul.opendev.org/t/openstack/build/aeac81942cab405c8839cd0547abc7e1 : SUCCESS in 3m 01s\n- tooz-tox-py313-memcached https://zuul.opendev.org/t/openstack/build/ff8be5eff1214ca6996fa1d441d46700 : SUCCESS in 4m 41s\n- tooz-tox-py310-postgresql https://zuul.opendev.org/t/openstack/build/e86bdcbef8a243c1815006ee601d1828 : SUCCESS in 4m 02s\n- tooz-tox-py313-postgresql https://zuul.opendev.org/t/openstack/build/4476a5e42f65484891902e0cdeaa2a3d : SUCCESS in 4m 30s\n- tooz-tox-py310-mysql https://zuul.opendev.org/t/openstack/build/901118f274cb473db27f5ef8296e9ab4 : SUCCESS in 4m 11s\n- tooz-tox-py313-mysql https://zuul.opendev.org/t/openstack/build/586434c6d9944b98b9a4d830f0409f16 : SUCCESS in 7m 07s\n- tooz-tox-py310-consul https://zuul.opendev.org/t/openstack/build/b789a6d925a34bc2bb1a5af992a43d28 : SUCCESS in 4m 21s\n- tooz-tox-py313-consul https://zuul.opendev.org/t/openstack/build/68c6c9ec0c804e1981501dd846558204 : SUCCESS in 4m 56s","accounts_in_message":[],"_revision_number":6},{"id":"4fd1ffdcd8fc6689087ee143df86c020f6fd26b4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-11 03:12:13.000000000","message":"Change has been successfully merged\n\n5 is the latest approved patch-set.\nNo files were changed between the latest approved patch-set and the submitted one.\n","accounts_in_message":[],"_revision_number":6},{"id":"db27d8682e4d465c60bb406aa38cd911732c6138","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-11 03:14:03.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cb2eb91179a34295b06bfd7db36e1d54\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a4fda547cea46c594ced94a28b35ed6 : SUCCESS in 45s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"abca4e997f3f0ea0af63db85f7c9e242a4104d37","revisions":{"85449f17a1459d6711fe2ba7c154e0796f33558b":{"kind":"REWORK","_number":1,"created":"2026-03-11 15:42:02.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/10/980110/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/10/980110/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/10/980110/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/10/980110/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/10/980110/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/10/980110/1"}}},"commit":{"parents":[{"commit":"fe579d1b3b53f18c1a110c4be81a1b8ba1dc1433","subject":"typing: Add type hints to tooz.locking","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/fe579d1b3b53f18c1a110c4be81a1b8ba1dc1433"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-03-09 16:21:22.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-03-11 15:41:56.000000000","tz":0},"subject":"typing: Add type hints to tooz.hashring","message":"typing: Add type hints to tooz.hashring\n\nChange-Id: Id0e1ead0db5b58fc1e19bedb67893dc001304d39\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/85449f17a1459d6711fe2ba7c154e0796f33558b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/85449f17a1459d6711fe2ba7c154e0796f33558b"}]},"branch":"refs/heads/master"},"47d66fb94c8249569faf4d0680b7c27c8bdb850d":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-03-12 08:42:07.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/10/980110/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/10/980110/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/10/980110/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/10/980110/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/10/980110/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/10/980110/2"}}},"commit":{"parents":[{"commit":"1be0ae2062aa9ebc30f1ab3385bbb531f25c4fa5","subject":"typing: Add type hints to tooz.locking","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/1be0ae2062aa9ebc30f1ab3385bbb531f25c4fa5"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-03-09 16:21:22.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-03-12 08:40:21.000000000","tz":0},"subject":"typing: Add type hints to tooz.hashring","message":"typing: Add type hints to tooz.hashring\n\nChange-Id: Id0e1ead0db5b58fc1e19bedb67893dc001304d39\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/47d66fb94c8249569faf4d0680b7c27c8bdb850d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/47d66fb94c8249569faf4d0680b7c27c8bdb850d"}]},"branch":"refs/heads/master"},"a43a445e8a167027218527f286c2b28fd988c83f":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2026-03-14 14:40:20.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/10/980110/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/10/980110/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/10/980110/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/10/980110/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/10/980110/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/10/980110/3"}}},"commit":{"parents":[{"commit":"7b927563054a29d213122081d4b8f40df7f436e8","subject":"typing: Add type hints to tooz.locking","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/7b927563054a29d213122081d4b8f40df7f436e8"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-03-09 16:21:22.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-03-14 14:40:15.000000000","tz":0},"subject":"typing: Add type hints to tooz.hashring","message":"typing: Add type hints to tooz.hashring\n\nChange-Id: Id0e1ead0db5b58fc1e19bedb67893dc001304d39\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/a43a445e8a167027218527f286c2b28fd988c83f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/a43a445e8a167027218527f286c2b28fd988c83f"}]},"branch":"refs/heads/master"},"61882d0ce9e3d908383a1fee9a8aecf2a0493aaa":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2026-03-19 09:47:24.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/10/980110/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/10/980110/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/10/980110/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/10/980110/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/10/980110/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/10/980110/4"}}},"commit":{"parents":[{"commit":"b9a3cbe26ec6cb3db37e10e98363e4b8c3dfb0c9","subject":"typing: Add type hints to tooz.locking","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/b9a3cbe26ec6cb3db37e10e98363e4b8c3dfb0c9"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-03-09 16:21:22.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-03-19 09:46:03.000000000","tz":60},"subject":"typing: Add type hints to tooz.hashring","message":"typing: Add type hints to tooz.hashring\n\nChange-Id: Id0e1ead0db5b58fc1e19bedb67893dc001304d39\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/61882d0ce9e3d908383a1fee9a8aecf2a0493aaa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/61882d0ce9e3d908383a1fee9a8aecf2a0493aaa"}]},"branch":"refs/heads/master"},"f281db268b7474e5924db0ac291eb70c61295c3e":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2026-03-31 09:03:29.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/10/980110/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/10/980110/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/10/980110/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/10/980110/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/10/980110/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/10/980110/5"}}},"commit":{"parents":[{"commit":"0f7b7e57dec9c27b6f9f9246e13757e53c6746a4","subject":"typing: Add type hints to tooz.locking","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/0f7b7e57dec9c27b6f9f9246e13757e53c6746a4"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-03-09 16:21:22.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-03-31 09:02:11.000000000","tz":120},"subject":"typing: Add type hints to tooz.hashring","message":"typing: Add type hints to tooz.hashring\n\nChange-Id: Id0e1ead0db5b58fc1e19bedb67893dc001304d39\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/f281db268b7474e5924db0ac291eb70c61295c3e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/f281db268b7474e5924db0ac291eb70c61295c3e"}]},"branch":"refs/heads/master"},"abca4e997f3f0ea0af63db85f7c9e242a4104d37":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2026-04-10 18:13:45.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/10/980110/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/10/980110/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/10/980110/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/10/980110/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/10/980110/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/10/980110/6"}}},"commit":{"parents":[{"commit":"32c396e8db5ab3675a33c7c175d040ed7dd7a17f","subject":"typing: Add type hints to tooz.locking","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/32c396e8db5ab3675a33c7c175d040ed7dd7a17f"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-03-09 16:21:22.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-04-10 12:02:57.000000000","tz":60},"subject":"typing: Add type hints to tooz.hashring","message":"typing: Add type hints to tooz.hashring\n\nChange-Id: Id0e1ead0db5b58fc1e19bedb67893dc001304d39\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/abca4e997f3f0ea0af63db85f7c9e242a4104d37"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/abca4e997f3f0ea0af63db85f7c9e242a4104d37"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
