)]}'
{"id":"openstack%2Ftooz~980111","triplet_id":"openstack%2Ftooz~master~Id822991a16c57342182abd0a846cc2f2a9931c06","project":"openstack/tooz","branch":"master","topic":"typing","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2026-04-23 14:42:05.000000000","reason":"\u003cGERRIT_ACCOUNT_15334\u003e replied on the change","reason_account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},"16137":{"account":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"last_update":"2026-04-21 09:26:59.000000000","reason":"\u003cGERRIT_ACCOUNT_16137\u003e replied on the change","reason_account":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}},"hashtags":[],"change_id":"Id822991a16c57342182abd0a846cc2f2a9931c06","subject":"typing: Add type hints to tooz.coordination, tooz.partitioner","status":"MERGED","created":"2026-03-11 15:42:02.000000000","updated":"2026-04-23 16:51:32.000000000","submitted":"2026-04-23 16:51:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"980111-typing","meta_rev_id":"470e6d5e1980ca1b964bb1b3bf6487df3c707581","_number":980111,"virtual_id_number":980111,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-04-23 16:51:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2026-04-21 09:26:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2026-04-23 14:42:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-11 17:43:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-04-13 08:00:51.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"72edaeac7a942180b22c800d6f147f48850a5e80","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-03-11 15:42:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"65ae5107a03658d19049c5a9efcd60cce7c9f8b9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-11 17:43:07.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1e27581a6f6f4ee9b504afcff51a054a\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/12374d5770d841b8a3b105f053e18953 : SUCCESS in 3m 53s\n- requirements-check https://zuul.opendev.org/t/openstack/build/85830fe9c0474dbc9aef6896b004b7cc : SUCCESS in 3m 27s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/347fbd43a80d4340a90e9453e2c79442 : SUCCESS in 1h 44m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fd09bd51e29249029d60c8faa40aa0e1 : SUCCESS in 6m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/672504d5253a4f92b5ad63d5ef9b8a9e : SUCCESS in 3m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9c75c64309f6479daccebdf04261fa72 : SUCCESS in 5m 13s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3a385eea893e4227ad26a83cf0c8c33a : SUCCESS in 6m 55s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/4217be53789441dda84874a323fd3113 : SUCCESS in 7m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/541fd9a5535a4a0492a439b410c6394f : SUCCESS in 6m 37s\n- tooz-tox-examples https://zuul.opendev.org/t/openstack/build/652a41615fa44bdd881d33edee718bb2 : SUCCESS in 5m 31s\n- tooz-tox-py310-etcd3gw https://zuul.opendev.org/t/openstack/build/0161bca07ef54ca4a9726c383aa75780 : FAILURE in 5m 30s\n- tooz-tox-py313-etcd3gw https://zuul.opendev.org/t/openstack/build/9189f31aa9154cc6816db3defa13bbb0 : FAILURE in 5m 54s\n- tooz-tox-py310-zookeeper https://zuul.opendev.org/t/openstack/build/2fab08a8e9934e28bdfe439618da20f4 : SUCCESS in 4m 41s\n- tooz-tox-py313-zookeeper https://zuul.opendev.org/t/openstack/build/4a7d8d5b000c4abc88ccab3d1a8b7988 : SUCCESS in 5m 25s\n- tooz-tox-py310-redis https://zuul.opendev.org/t/openstack/build/fef28d5fa785477788687a8b94beaf06 : SUCCESS in 9m 23s\n- tooz-tox-py313-redis https://zuul.opendev.org/t/openstack/build/40bd9b8434524ab4903a65d2188fe7bb : SUCCESS in 6m 45s\n- tooz-tox-py310-sentinel https://zuul.opendev.org/t/openstack/build/bcb923953ab14759bd1517b5d28bfe9d : SUCCESS in 7m 14s\n- tooz-tox-py313-sentinel https://zuul.opendev.org/t/openstack/build/03d96ff9163142acadb2687aa0404ed6 : SUCCESS in 6m 52s\n- tooz-tox-py310-memcached https://zuul.opendev.org/t/openstack/build/071a420e3fc24e0f82927d2b37c7e8fd : SUCCESS in 4m 28s\n- tooz-tox-py313-memcached https://zuul.opendev.org/t/openstack/build/3b2cb8449136486ab75ce392789e0e27 : SUCCESS in 3m 39s\n- tooz-tox-py310-postgresql https://zuul.opendev.org/t/openstack/build/72d31d05400a46be9282512027129725 : SUCCESS in 4m 37s\n- tooz-tox-py313-postgresql https://zuul.opendev.org/t/openstack/build/cedff829f87d4fc181e88f03f53d0f13 : SUCCESS in 5m 02s\n- tooz-tox-py310-mysql https://zuul.opendev.org/t/openstack/build/4d7d5d48adc24d8680420aac35d0449e : SUCCESS in 4m 51s\n- tooz-tox-py313-mysql https://zuul.opendev.org/t/openstack/build/f9fa0f71764045d5bff8a19cd43d7fd0 : SUCCESS in 7m 43s\n- tooz-tox-py310-consul https://zuul.opendev.org/t/openstack/build/72a919d55e7a4d44909b9e131c7ae456 : SUCCESS in 5m 07s\n- tooz-tox-py313-consul https://zuul.opendev.org/t/openstack/build/7690bf156c614b68a17aba739099dea9 : SUCCESS in 5m 03s","accounts_in_message":[],"_revision_number":1},{"id":"cf95b1a6b7e9448b5c35cfc35126716461ea8b2d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-03-12 08:42:07.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"62579489664b83c877d5a06ba0ed19327343a85f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-12 11:12:55.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/220cac2be979428d9a472ad8db2be150\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/1dc2f3819fc44c8fbe8f8769240165fe : SUCCESS in 3m 23s\n- requirements-check https://zuul.opendev.org/t/openstack/build/f97253096484479db712265fd100547e : SUCCESS in 3m 49s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/009e190d91a4451fb283ba9a27c570e2 : SUCCESS in 1h 56m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2d89dde316c8494dbd3a36af42c37d71 : SUCCESS in 5m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/026bdc88fac5429cbb1c9459072d3569 : SUCCESS in 6m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/12a18e9876a04786a494da40773f101a : SUCCESS in 5m 14s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2a497ea15a6d41ac930abe30de9e2fac : SUCCESS in 7m 26s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/e3df99fa52cc41f892a383f89705cdd8 : SUCCESS in 7m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/152ac32acda74e9b889c998606289b6b : SUCCESS in 5m 17s\n- tooz-tox-examples https://zuul.opendev.org/t/openstack/build/bfffd40fab984f9e986233ac09e844af : SUCCESS in 3m 55s\n- tooz-tox-py310-etcd3gw https://zuul.opendev.org/t/openstack/build/bfa7e647535a424eaba6329732264211 : SUCCESS in 5m 26s\n- tooz-tox-py313-etcd3gw https://zuul.opendev.org/t/openstack/build/b5c1136416f2481ab2e04f29c4796b47 : SUCCESS in 7m 50s\n- tooz-tox-py310-zookeeper https://zuul.opendev.org/t/openstack/build/fa8e6206fd774791b59efefa161c9ffd : SUCCESS in 5m 10s\n- tooz-tox-py313-zookeeper https://zuul.opendev.org/t/openstack/build/2d9d1927e48e4bb69bc07314a1cdaabb : FAILURE in 5m 09s\n- tooz-tox-py310-redis https://zuul.opendev.org/t/openstack/build/8a396fd36b254640aa8459c90ae35c89 : SUCCESS in 6m 57s\n- tooz-tox-py313-redis https://zuul.opendev.org/t/openstack/build/b2b306ced3474c71a2c07216543a431e : SUCCESS in 9m 02s\n- tooz-tox-py310-sentinel https://zuul.opendev.org/t/openstack/build/00e0e6f4b1794fc6a0445f0d6e1f4a68 : SUCCESS in 5m 06s\n- tooz-tox-py313-sentinel https://zuul.opendev.org/t/openstack/build/51afed55854d4006a8efd65697d124a8 : SUCCESS in 7m 31s\n- tooz-tox-py310-memcached https://zuul.opendev.org/t/openstack/build/c62a02da16e142969960024dadfb5127 : SUCCESS in 4m 56s\n- tooz-tox-py313-memcached https://zuul.opendev.org/t/openstack/build/089d7f01aa214c999e172ceac65769f5 : SUCCESS in 5m 58s\n- tooz-tox-py310-postgresql https://zuul.opendev.org/t/openstack/build/b4e69844bcac4c458c33220f4b04c7be : SUCCESS in 7m 55s\n- tooz-tox-py313-postgresql https://zuul.opendev.org/t/openstack/build/c257b2a095ed40a582997e315e98b8d9 : SUCCESS in 5m 16s\n- tooz-tox-py310-mysql https://zuul.opendev.org/t/openstack/build/e17315b7dfbf40918078806e2248bc9f : SUCCESS in 5m 08s\n- tooz-tox-py313-mysql https://zuul.opendev.org/t/openstack/build/e9d59d7ef8e24773afc8b87bcce2e6f4 : SUCCESS in 8m 02s\n- tooz-tox-py310-consul https://zuul.opendev.org/t/openstack/build/d5643f01a281400593c6e88db48f68cf : SUCCESS in 5m 03s\n- tooz-tox-py313-consul https://zuul.opendev.org/t/openstack/build/fc1ce05847bb472288d5595a76045b05 : SUCCESS in 5m 48s","accounts_in_message":[],"_revision_number":2},{"id":"152c8e649d22b5dfd97a5710ffe12af5d80d7d30","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-03-14 14:40:20.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"5ffc64387c87ab9529c12819e84411fabde4f529","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-14 16:27:55.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2c5f6782386b4ed68f8eff2872cc9812\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/50270f6991324083b756ce49b6ae39d5 : SUCCESS in 2m 39s\n- requirements-check https://zuul.opendev.org/t/openstack/build/25b182ef7e8d4659981c391aa85bf30a : SUCCESS in 2m 46s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/358fe3757ca44294b3048e6c18542769 : SUCCESS in 1h 35m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7c7bbd9c25ca41e69a353e62bb40eb5d : SUCCESS in 3m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2a52e76c88664a6cb49c8e1e86076e9d : SUCCESS in 5m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8d4d393cf19345f09d76272c57810162 : SUCCESS in 4m 36s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4e54d0f1c69648e3a5547bcc876ad450 : SUCCESS in 5m 13s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/e38296dd013c4628be162e0f61a04787 : SUCCESS in 7m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f30078b7d7749a2bf55ea1a65091a2d : SUCCESS in 5m 04s\n- tooz-tox-examples https://zuul.opendev.org/t/openstack/build/c541c7a0c9404c4188e0ff664188224e : SUCCESS in 5m 04s\n- tooz-tox-py310-etcd3gw https://zuul.opendev.org/t/openstack/build/35b77e886f2e4b23b19441d8580282cb : SUCCESS in 5m 05s\n- tooz-tox-py313-etcd3gw https://zuul.opendev.org/t/openstack/build/e0a867be48d34ac488c127957f5107be : SUCCESS in 5m 20s\n- tooz-tox-py310-zookeeper https://zuul.opendev.org/t/openstack/build/1482e736d3a642dba429a9680f594e71 : SUCCESS in 4m 42s\n- tooz-tox-py313-zookeeper https://zuul.opendev.org/t/openstack/build/8c9ed8a3a5884fddb80190cc636be1bb : SUCCESS in 4m 09s\n- tooz-tox-py310-redis https://zuul.opendev.org/t/openstack/build/ad74b23fdb6041b9b993e5354c16066c : SUCCESS in 6m 11s\n- tooz-tox-py313-redis https://zuul.opendev.org/t/openstack/build/5edd4b8a6bb640fdbc6a9bb5a0310c41 : SUCCESS in 5m 14s\n- tooz-tox-py310-sentinel https://zuul.opendev.org/t/openstack/build/cea7eae928a241ca9d619608196f26a1 : SUCCESS in 5m 03s\n- tooz-tox-py313-sentinel https://zuul.opendev.org/t/openstack/build/08e5177d09624640bae946dc0f495a14 : SUCCESS in 6m 21s\n- tooz-tox-py310-memcached https://zuul.opendev.org/t/openstack/build/ba00f62447e74012a4587e0a70a5f995 : SUCCESS in 4m 28s\n- tooz-tox-py313-memcached https://zuul.opendev.org/t/openstack/build/c18bf42b940f4d1ab359b089d169dc50 : SUCCESS in 5m 37s\n- tooz-tox-py310-postgresql https://zuul.opendev.org/t/openstack/build/6b34cf978acb468a88f07b5b41019a2d : SUCCESS in 2m 38s\n- tooz-tox-py313-postgresql https://zuul.opendev.org/t/openstack/build/0df48c1c9c8a48bf938ab8b810c63693 : SUCCESS in 3m 14s\n- tooz-tox-py310-mysql https://zuul.opendev.org/t/openstack/build/d6ca29f9476a42e2a9028e7a35a155f8 : SUCCESS in 4m 29s\n- tooz-tox-py313-mysql https://zuul.opendev.org/t/openstack/build/3755bc3876714920bb6124436f081680 : SUCCESS in 5m 06s\n- tooz-tox-py310-consul https://zuul.opendev.org/t/openstack/build/46f9f88500b84317a498bef5b84e1f6c : SUCCESS in 5m 04s\n- tooz-tox-py313-consul https://zuul.opendev.org/t/openstack/build/934047daeefc46e7af30e3d8b091a3e9 : SUCCESS in 5m 05s","accounts_in_message":[],"_revision_number":3},{"id":"0c8208cab6a9f9d9a0aa82279425f81b68f6b020","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-03-19 09:47:24.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"19863db294f55ffad39293e542af738b8ed8a050","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-19 10:52:35.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2254072ab69b48e9b9796be9efeddab9\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/9b0a94628c374230a8cce64ffb700a45 : SUCCESS in 3m 01s\n- requirements-check https://zuul.opendev.org/t/openstack/build/7c988b066b884e6a97296b3f3754ff34 : SUCCESS in 4m 00s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8d591ae3e1fc4a19a8d42f948bc4ae46 : SUCCESS in 1h 00m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/027a5f12905841eeac25de08d8e98f91 : SUCCESS in 9m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/66db6461a281487f9f8879bfbbdaf468 : SUCCESS in 5m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/067269334c064bb8bcaba95b9dfe4eb7 : SUCCESS in 5m 08s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3625ecc5ee6c4654b83933904ae2f622 : SUCCESS in 5m 10s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/301e7b28d1834697ae4e10c315e3ce54 : SUCCESS in 7m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c479bdbfcbc4fe3b7149fef103af1b0 : SUCCESS in 3m 52s\n- tooz-tox-examples https://zuul.opendev.org/t/openstack/build/224412b8f8c14dd3bb80f3181872e23d : SUCCESS in 5m 41s\n- tooz-tox-py310-etcd3gw https://zuul.opendev.org/t/openstack/build/15c029b0d3774e4caba2f6881425ebcd : SUCCESS in 4m 45s\n- tooz-tox-py313-etcd3gw https://zuul.opendev.org/t/openstack/build/db5aca3181704b40ac74d18a6db50d95 : SUCCESS in 6m 22s\n- tooz-tox-py310-zookeeper https://zuul.opendev.org/t/openstack/build/3b840150c379414cb1932a8676406e7a : SUCCESS in 4m 46s\n- tooz-tox-py313-zookeeper https://zuul.opendev.org/t/openstack/build/6eeb3842575448a7a6e09ee63bbb54f9 : SUCCESS in 3m 20s\n- tooz-tox-py310-redis https://zuul.opendev.org/t/openstack/build/af43ac26dc274b68aa7d51fb08823079 : SUCCESS in 6m 49s\n- tooz-tox-py313-redis https://zuul.opendev.org/t/openstack/build/eacd4de3c6e94c2db5634aaa98d933e1 : SUCCESS in 6m 04s\n- tooz-tox-py310-sentinel https://zuul.opendev.org/t/openstack/build/0aa9234bfa594781bbf886b629783830 : SUCCESS in 6m 31s\n- tooz-tox-py313-sentinel https://zuul.opendev.org/t/openstack/build/95413b23a6af4a78877a006b9948a116 : SUCCESS in 7m 48s\n- tooz-tox-py310-memcached https://zuul.opendev.org/t/openstack/build/8a0a4557a81248e6bf90ef871e078bda : SUCCESS in 6m 05s\n- tooz-tox-py313-memcached https://zuul.opendev.org/t/openstack/build/267b452404a442848c82a9d0901b661a : SUCCESS in 4m 06s\n- tooz-tox-py310-postgresql https://zuul.opendev.org/t/openstack/build/3d831ca8fcac4498b85eec38c91f385b : SUCCESS in 2m 58s\n- tooz-tox-py313-postgresql https://zuul.opendev.org/t/openstack/build/4347f042613c4a86ba40347e030e710b : SUCCESS in 4m 51s\n- tooz-tox-py310-mysql https://zuul.opendev.org/t/openstack/build/af85bff186734013b5f94cd35f9e0d7d : SUCCESS in 6m 13s\n- tooz-tox-py313-mysql https://zuul.opendev.org/t/openstack/build/b050ce3e61764b9480de280b6f72a7e2 : SUCCESS in 10m 54s\n- tooz-tox-py310-consul https://zuul.opendev.org/t/openstack/build/170359cf207b4750b5dce35d7e54e834 : SUCCESS in 5m 44s\n- tooz-tox-py313-consul https://zuul.opendev.org/t/openstack/build/587b9f8bed384da69edf762bbe179982 : SUCCESS in 5m 08s","accounts_in_message":[],"_revision_number":4},{"id":"c066d77d8c014f963bac421886754d4aa051af6f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-03-31 09:03:29.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"e7570e99d7467d1e589d747f378053d7487f2851","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-31 10:53:04.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2a7140f0dba44af39ed48c45f170e812\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/90f1c1c3d743498cb3b14f20da88a3b7 : SUCCESS in 5m 23s\n- requirements-check https://zuul.opendev.org/t/openstack/build/f74034fed28e480196f7588fc87e286e : SUCCESS in 3m 50s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5cef27eead6b45228d07764656163014 : FAILURE in 1h 39m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/27f86e7cd7f14a0481d0a9f62d696d2b : SUCCESS in 5m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d51a6b3a7cd4e59b873fe7098aa5ea0 : SUCCESS in 7m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/242d4a92cf2c404d8a7e73ef00c6a68c : SUCCESS in 5m 05s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/109259dee1e640a2af4b17ce233adf9f : SUCCESS in 7m 46s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/ca95fff1598f440bbb22db5e5c3a850d : SUCCESS in 7m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/91a8ba17e02a4805b3640d1ee22b6c1d : SUCCESS in 4m 26s\n- tooz-tox-examples https://zuul.opendev.org/t/openstack/build/884cb39b6e9d4a7b892e0f7b3ff6848c : SUCCESS in 5m 12s\n- tooz-tox-py310-etcd3gw https://zuul.opendev.org/t/openstack/build/7b18924f731d4abb9285e509b2d164d5 : SUCCESS in 4m 42s\n- tooz-tox-py313-etcd3gw https://zuul.opendev.org/t/openstack/build/ee65ecbb017d440c9ed2e9376c4daa55 : SUCCESS in 5m 27s\n- tooz-tox-py310-zookeeper https://zuul.opendev.org/t/openstack/build/b272f27f50b2420f99a5326eba3cc861 : SUCCESS in 4m 39s\n- tooz-tox-py313-zookeeper https://zuul.opendev.org/t/openstack/build/b5ed086f459540808347212b00ce2157 : SUCCESS in 5m 53s\n- tooz-tox-py310-redis https://zuul.opendev.org/t/openstack/build/b1b9cc1e68de41298efebb362c16004c : SUCCESS in 5m 48s\n- tooz-tox-py313-redis https://zuul.opendev.org/t/openstack/build/8839cfe712314658aa293b019e45f437 : SUCCESS in 6m 59s\n- tooz-tox-py310-sentinel https://zuul.opendev.org/t/openstack/build/4d0c99e7c7734dff8b72fb28e73a497b : SUCCESS in 6m 50s\n- tooz-tox-py313-sentinel https://zuul.opendev.org/t/openstack/build/23a671652c3f4ede992a8ba4ca5a89f0 : SUCCESS in 8m 25s\n- tooz-tox-py310-memcached https://zuul.opendev.org/t/openstack/build/80b848223129428bbc4d28c752b03093 : SUCCESS in 4m 37s\n- tooz-tox-py313-memcached https://zuul.opendev.org/t/openstack/build/306f05b64b07402099db2be643915fc9 : SUCCESS in 3m 24s\n- tooz-tox-py310-postgresql https://zuul.opendev.org/t/openstack/build/178ea44d95f64f71ade1714e471e54ee : SUCCESS in 4m 38s\n- tooz-tox-py313-postgresql https://zuul.opendev.org/t/openstack/build/7ae9c465a52f43aca985c927eb81c0c2 : SUCCESS in 6m 24s\n- tooz-tox-py310-mysql https://zuul.opendev.org/t/openstack/build/72da49dd87624508b07f6a4d30cc8390 : SUCCESS in 5m 04s\n- tooz-tox-py313-mysql https://zuul.opendev.org/t/openstack/build/713e7217dc5546e6a1494120ca2d49b1 : SUCCESS in 9m 55s\n- tooz-tox-py310-consul https://zuul.opendev.org/t/openstack/build/f9be9f84fc144909b981c8cff8817775 : SUCCESS in 3m 34s\n- tooz-tox-py313-consul https://zuul.opendev.org/t/openstack/build/6ebcde9ab8a344249bad5465d7048337 : SUCCESS in 10m 32s","accounts_in_message":[],"_revision_number":5},{"id":"1d0937348a5646b9bc6c490d34e4e3b84bd6b50b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-04-10 18:13:45.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"9d209770086cd237d85828d2e2281c5c52f103d0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-10 20:08:23.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3b48b52677e84e4d8117c4dbb030b46f\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/57ae31603c4d4fe9a57917e00df3fcae : SUCCESS in 2m 25s\n- requirements-check https://zuul.opendev.org/t/openstack/build/bd2c9aae560945cabdd4d67820c9befb : SUCCESS in 5m 02s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5aa10985187c454281f67893e31fb4e4 : SUCCESS in 1h 42m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d8991b2b42074f04882a3013ddebc29c : SUCCESS in 6m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/73137ba12ee34cd0a3ff43ca06b33acd : SUCCESS in 5m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c12e57b3a0b44f6badf378532965523b : SUCCESS in 5m 55s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c4d73c4ad53545e7ac3ce55fcc728ec9 : SUCCESS in 7m 48s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/77ff83f9e11d4bec84e3613fd93a3d5f : SUCCESS in 8m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/36a0755c635940429d26961f7c06f933 : SUCCESS in 5m 40s\n- tooz-tox-examples https://zuul.opendev.org/t/openstack/build/44a6d5b2ab4a458697852bc61c97b805 : SUCCESS in 5m 20s\n- tooz-tox-py310-etcd3gw https://zuul.opendev.org/t/openstack/build/8da740983a824f14af84b479d526f509 : SUCCESS in 5m 16s\n- tooz-tox-py313-etcd3gw https://zuul.opendev.org/t/openstack/build/93a5da6e64614715917d58a18ad91177 : SUCCESS in 5m 51s\n- tooz-tox-py310-zookeeper https://zuul.opendev.org/t/openstack/build/0b3c632331134781a66fce40abafe23f : SUCCESS in 4m 36s\n- tooz-tox-py313-zookeeper https://zuul.opendev.org/t/openstack/build/3ca638d232374aff88ad13f17b75e58d : SUCCESS in 5m 20s\n- tooz-tox-py310-redis https://zuul.opendev.org/t/openstack/build/0ac727dfd45d4169a5a8a73306b1a655 : SUCCESS in 6m 36s\n- tooz-tox-py313-redis https://zuul.opendev.org/t/openstack/build/9fd1f3430f5b451e9c299c2b2ae29256 : SUCCESS in 7m 34s\n- tooz-tox-py310-sentinel https://zuul.opendev.org/t/openstack/build/91d082bef1314c4399cc2e9d523b9054 : SUCCESS in 6m 50s\n- tooz-tox-py313-sentinel https://zuul.opendev.org/t/openstack/build/7c8313d3ccd04515a292243933a87467 : SUCCESS in 6m 08s\n- tooz-tox-py310-memcached https://zuul.opendev.org/t/openstack/build/5b12c918c8e744dbaea49a0b9dea524b : SUCCESS in 6m 20s\n- tooz-tox-py313-memcached https://zuul.opendev.org/t/openstack/build/9f69f355e0744a4cbbde19ba3a239901 : SUCCESS in 5m 51s\n- tooz-tox-py310-postgresql https://zuul.opendev.org/t/openstack/build/3560dee629d549d38d56e16b231e6dad : SUCCESS in 5m 42s\n- tooz-tox-py313-postgresql https://zuul.opendev.org/t/openstack/build/ac94ee4e279f4976b7b551d6ba41afb5 : SUCCESS in 4m 40s\n- tooz-tox-py310-mysql https://zuul.opendev.org/t/openstack/build/ce1bc444cbc2401cacf88688fbac035e : SUCCESS in 4m 45s\n- tooz-tox-py313-mysql https://zuul.opendev.org/t/openstack/build/2ddf4f80a2914d758b850fd3ed9f9b3e : SUCCESS in 5m 12s\n- tooz-tox-py310-consul https://zuul.opendev.org/t/openstack/build/9ef0a6191dda4530b948494629c6323f : SUCCESS in 4m 38s\n- tooz-tox-py313-consul https://zuul.opendev.org/t/openstack/build/c68b9ee6174e475081783c3474b272c8 : SUCCESS in 5m 57s","accounts_in_message":[],"_revision_number":6},{"id":"4812d83c991a5276816d2d17fb016d7e15cc0e68","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2026-04-13 08:00:51.000000000","message":"Patch Set 6: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"ae2caea7f22b3785b792b2796c1469c20da05594","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-04-14 09:16:57.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"f095cd20f87a85d1e189e42c13b2628d28c2ff9c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-04-14 09:45:17.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":7},{"id":"0dc97c52029931c14c271de80cbc1fa70b813384","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-04-14 10:23:17.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"6a3fc552802ed5fa8495064d5272566c468b759f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-14 11:38:06.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/38e007f0f88043438344f4ccc0ae5519\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/d009dfa0a22d4db2a0fc58199b4352e8 : SUCCESS in 2m 47s\n- requirements-check https://zuul.opendev.org/t/openstack/build/cf737f49de5a4bec82650aa05717a3dc : SUCCESS in 4m 15s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fd89834ad24d47aaa553d53354b12fac : SUCCESS in 1h 48m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9d7dfa4bea3f4778a5b5df463f13dced : SUCCESS in 7m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e7b85895d08a4410a0e5164cb2771f79 : SUCCESS in 5m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ddc6d83d5caf41aeacd078bd00e399b9 : SUCCESS in 6m 32s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ee3f2f5aaf124b54993d06044a2c1045 : SUCCESS in 4m 16s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/a2f21cc439534692941e7d775bc44c31 : SUCCESS in 8m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8912fc030b14223a28be5637f092ef7 : SUCCESS in 5m 25s\n- tooz-tox-examples https://zuul.opendev.org/t/openstack/build/8feb1fd122f74330b07b0b480f89fd1a : SUCCESS in 6m 45s\n- tooz-tox-py310-etcd3gw https://zuul.opendev.org/t/openstack/build/00203b1ef2fa4db1a3a2f8ce64af469c : SUCCESS in 4m 19s\n- tooz-tox-py313-etcd3gw https://zuul.opendev.org/t/openstack/build/6298c26f7b4943dfa1962ffdfb54ed2d : SUCCESS in 6m 21s\n- tooz-tox-py310-zookeeper https://zuul.opendev.org/t/openstack/build/79d4c5876b394fc687cb40d013912969 : SUCCESS in 4m 29s\n- tooz-tox-py313-zookeeper https://zuul.opendev.org/t/openstack/build/b456baa87e264ac08a8c6f6410d0b771 : SUCCESS in 7m 43s\n- tooz-tox-py310-redis https://zuul.opendev.org/t/openstack/build/3def07a8482e4646acf306b4e948a9a6 : SUCCESS in 6m 44s\n- tooz-tox-py313-redis https://zuul.opendev.org/t/openstack/build/d91411edd3014c8fbe850c50450924cd : SUCCESS in 7m 27s\n- tooz-tox-py310-sentinel https://zuul.opendev.org/t/openstack/build/8a776cce4ecc484391701dbe1e398c61 : SUCCESS in 8m 02s\n- tooz-tox-py313-sentinel https://zuul.opendev.org/t/openstack/build/dc809bdb831c4237a58bde2cbd08fa77 : SUCCESS in 6m 59s\n- tooz-tox-py310-memcached https://zuul.opendev.org/t/openstack/build/c9a24a1593a34504b9d0d4437ad31adb : SUCCESS in 6m 22s\n- tooz-tox-py313-memcached https://zuul.opendev.org/t/openstack/build/96d7613ce97c4e02beb9d7a38755803c : SUCCESS in 4m 46s\n- tooz-tox-py310-postgresql https://zuul.opendev.org/t/openstack/build/5cd55522357842dcbf3cbd73b19c4a3a : SUCCESS in 5m 09s\n- tooz-tox-py313-postgresql https://zuul.opendev.org/t/openstack/build/718489209cae4316879cd36e13235c75 : SUCCESS in 4m 30s\n- tooz-tox-py310-mysql https://zuul.opendev.org/t/openstack/build/29c91cc83a4845f58e6d85c7be357b5d : SUCCESS in 5m 15s\n- tooz-tox-py313-mysql https://zuul.opendev.org/t/openstack/build/4f7c47f94e2c4dafbec0c327ac3ee142 : SUCCESS in 7m 49s\n- tooz-tox-py310-consul https://zuul.opendev.org/t/openstack/build/11b04d0298b14bd98088b51de5fdf546 : SUCCESS in 5m 45s\n- tooz-tox-py313-consul https://zuul.opendev.org/t/openstack/build/3cb6b92401b948d8839ee2467ede58bd : SUCCESS in 7m 13s","accounts_in_message":[],"_revision_number":7},{"id":"4535c1709d5b1d10847c4821a8a47e64a4fa4632","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2026-04-21 09:26:59.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"49a488f1a69b10d3792324ab86373faf1a2dcafd","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-04-23 14:42:05.000000000","message":"Patch Set 7: Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"00136bfb0f299508216acce86fb4300e4d385fdd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-23 14:42:42.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"4b80a936ea627c1beadb27f7991ee078610ff4ae","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-23 16:51:32.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c0d83fcc56934cd6a05c69c7e372f280\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/b33f46e235aa4370bb0c99f01900d322 : SUCCESS in 2m 26s\n- requirements-check https://zuul.opendev.org/t/openstack/build/6ac75ab42d35433f9ec6141dc157683a : SUCCESS in 2m 39s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6caf77bd6529484080c69c0126335b9f : SUCCESS in 1h 44m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4e9f12ed1f754365b6e438f868a0cb55 : SUCCESS in 5m 31s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ee29228def1c4c17adb8b5278d594ae0 : SUCCESS in 4m 02s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/320e5f89ad2e45b79b68e670a016f28f : SUCCESS in 6m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1cc5cf4253a9484e85141dae0d646478 : SUCCESS in 4m 59s\n- tooz-tox-examples https://zuul.opendev.org/t/openstack/build/efeea1d6b8934e748d3e2a3166b5a400 : SUCCESS in 5m 11s\n- tooz-tox-py310-etcd3gw https://zuul.opendev.org/t/openstack/build/6affb0cdaf1b42319af21f8ab43ecbfd : SUCCESS in 3m 38s\n- tooz-tox-py313-etcd3gw https://zuul.opendev.org/t/openstack/build/bc00d134e1584d75a5cb080eb3d7b061 : SUCCESS in 6m 55s\n- tooz-tox-py310-zookeeper https://zuul.opendev.org/t/openstack/build/96adb3f00e5a42e187f56b1de57f64ea : SUCCESS in 4m 42s\n- tooz-tox-py313-zookeeper https://zuul.opendev.org/t/openstack/build/8da7306c522044d7ae4a52e2475c4628 : SUCCESS in 7m 43s\n- tooz-tox-py310-redis https://zuul.opendev.org/t/openstack/build/16c42bfb55e84bcb89bd16b3197a5d7f : SUCCESS in 5m 33s\n- tooz-tox-py313-redis https://zuul.opendev.org/t/openstack/build/5068609a1b1742629d15266c7193d74d : SUCCESS in 6m 07s\n- tooz-tox-py310-sentinel https://zuul.opendev.org/t/openstack/build/3a17d96deba0470696ffc7fb7e96be51 : SUCCESS in 6m 03s\n- tooz-tox-py313-sentinel https://zuul.opendev.org/t/openstack/build/99b0c589a01c412fb275015592fe4526 : SUCCESS in 6m 28s\n- tooz-tox-py310-memcached https://zuul.opendev.org/t/openstack/build/df00ccbe31cf4351b8a2b58ded348f0a : SUCCESS in 4m 16s\n- tooz-tox-py313-memcached https://zuul.opendev.org/t/openstack/build/af5d4e091bbf4c88b2386b5edf9e854a : SUCCESS in 6m 37s\n- tooz-tox-py310-postgresql https://zuul.opendev.org/t/openstack/build/32e69c45ba4d4b3794e71effe6ffe766 : SUCCESS in 3m 58s\n- tooz-tox-py313-postgresql https://zuul.opendev.org/t/openstack/build/ab9fe13c74f945c6a3e0cc40677716b0 : SUCCESS in 4m 54s\n- tooz-tox-py310-mysql https://zuul.opendev.org/t/openstack/build/e2aeb529cf9549eab2970f5be2a0aebd : SUCCESS in 4m 37s\n- tooz-tox-py313-mysql https://zuul.opendev.org/t/openstack/build/fa6a4effcb2a47efaf892428f54fbe88 : SUCCESS in 6m 59s\n- tooz-tox-py310-consul https://zuul.opendev.org/t/openstack/build/0c193df4b64d45b7bc941cdc53b7cebe : SUCCESS in 4m 22s\n- tooz-tox-py313-consul https://zuul.opendev.org/t/openstack/build/342848bea8ee46fe811df3a524847d2a : SUCCESS in 4m 41s","accounts_in_message":[],"_revision_number":7},{"id":"470e6d5e1980ca1b964bb1b3bf6487df3c707581","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-23 16:51:32.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"bf925b9b301208a64b5e28c2e382da58949468f8","revisions":{"ad22b7ecad07ec71b42174f989ae3df2512a1a5e":{"kind":"REWORK","_number":1,"created":"2026-03-11 15:42:02.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/11/980111/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/11/980111/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/11/980111/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/11/980111/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/11/980111/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/11/980111/1"}}},"commit":{"parents":[{"commit":"85449f17a1459d6711fe2ba7c154e0796f33558b","subject":"typing: Add type hints to tooz.hashring","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/85449f17a1459d6711fe2ba7c154e0796f33558b"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-03-09 16:29:20.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-03-11 15:41:56.000000000","tz":0},"subject":"typing: Add type hints to tooz.coordination, tooz.partitioner","message":"typing: Add type hints to tooz.coordination, tooz.partitioner\n\nThese are handled together since they are very closely coupled.\n\nChange-Id: Id822991a16c57342182abd0a846cc2f2a9931c06\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/ad22b7ecad07ec71b42174f989ae3df2512a1a5e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/ad22b7ecad07ec71b42174f989ae3df2512a1a5e"}]},"branch":"refs/heads/master"},"2e1ff3bd0105ffbd089620c5f40e3c80061fac53":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-03-12 08:42:07.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/11/980111/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/11/980111/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/11/980111/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/11/980111/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/11/980111/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/11/980111/2"}}},"commit":{"parents":[{"commit":"47d66fb94c8249569faf4d0680b7c27c8bdb850d","subject":"typing: Add type hints to tooz.hashring","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/47d66fb94c8249569faf4d0680b7c27c8bdb850d"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-03-09 16:29:20.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-03-12 08:40:21.000000000","tz":0},"subject":"typing: Add type hints to tooz.coordination, tooz.partitioner","message":"typing: Add type hints to tooz.coordination, tooz.partitioner\n\nThese are handled together since they are very closely coupled.\n\nChange-Id: Id822991a16c57342182abd0a846cc2f2a9931c06\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/2e1ff3bd0105ffbd089620c5f40e3c80061fac53"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/2e1ff3bd0105ffbd089620c5f40e3c80061fac53"}]},"branch":"refs/heads/master"},"157beb3dd639ed799bc23c998fee266c2231c790":{"kind":"REWORK","_number":3,"created":"2026-03-14 14:40:20.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/11/980111/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/11/980111/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/11/980111/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/11/980111/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/11/980111/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/11/980111/3"}}},"commit":{"parents":[{"commit":"a43a445e8a167027218527f286c2b28fd988c83f","subject":"typing: Add type hints to tooz.hashring","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/a43a445e8a167027218527f286c2b28fd988c83f"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-03-09 16:29:20.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-03-14 14:40:15.000000000","tz":0},"subject":"typing: Add type hints to tooz.coordination, tooz.partitioner","message":"typing: Add type hints to tooz.coordination, tooz.partitioner\n\nThese are handled together since they are very closely coupled.\n\nChange-Id: Id822991a16c57342182abd0a846cc2f2a9931c06\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/157beb3dd639ed799bc23c998fee266c2231c790"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/157beb3dd639ed799bc23c998fee266c2231c790"}]},"branch":"refs/heads/master"},"2355ac4823aedfe0be104c914608f37b04c31155":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2026-03-19 09:47:24.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/11/980111/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/11/980111/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/11/980111/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/11/980111/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/11/980111/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/11/980111/4"}}},"commit":{"parents":[{"commit":"61882d0ce9e3d908383a1fee9a8aecf2a0493aaa","subject":"typing: Add type hints to tooz.hashring","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/61882d0ce9e3d908383a1fee9a8aecf2a0493aaa"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-03-09 16:29:20.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-03-19 09:46:03.000000000","tz":60},"subject":"typing: Add type hints to tooz.coordination, tooz.partitioner","message":"typing: Add type hints to tooz.coordination, tooz.partitioner\n\nThese are handled together since they are very closely coupled.\n\nChange-Id: Id822991a16c57342182abd0a846cc2f2a9931c06\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/2355ac4823aedfe0be104c914608f37b04c31155"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/2355ac4823aedfe0be104c914608f37b04c31155"}]},"branch":"refs/heads/master"},"037736c6dd2806da8257917003f94b63fc8df906":{"kind":"REWORK","_number":5,"created":"2026-03-31 09:03:29.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/11/980111/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/11/980111/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/11/980111/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/11/980111/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/11/980111/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/11/980111/5"}}},"commit":{"parents":[{"commit":"f281db268b7474e5924db0ac291eb70c61295c3e","subject":"typing: Add type hints to tooz.hashring","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/f281db268b7474e5924db0ac291eb70c61295c3e"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-03-09 16:29:20.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-03-31 09:02:21.000000000","tz":120},"subject":"typing: Add type hints to tooz.coordination, tooz.partitioner","message":"typing: Add type hints to tooz.coordination, tooz.partitioner\n\nThese are handled together since they are very closely coupled.\n\nChange-Id: Id822991a16c57342182abd0a846cc2f2a9931c06\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/037736c6dd2806da8257917003f94b63fc8df906"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/037736c6dd2806da8257917003f94b63fc8df906"}]},"branch":"refs/heads/master"},"c09665ee7d0a38a446a036f81b56477e360c76aa":{"kind":"REWORK","_number":6,"created":"2026-04-10 18:13:45.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/11/980111/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/11/980111/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/11/980111/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/11/980111/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/11/980111/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/11/980111/6"}}},"commit":{"parents":[{"commit":"abca4e997f3f0ea0af63db85f7c9e242a4104d37","subject":"typing: Add type hints to tooz.hashring","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/abca4e997f3f0ea0af63db85f7c9e242a4104d37"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-03-09 16:29:20.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-04-10 12:03:32.000000000","tz":60},"subject":"typing: Add type hints to tooz.coordination, tooz.partitioner","message":"typing: Add type hints to tooz.coordination, tooz.partitioner\n\nThese are handled together since they are very closely coupled.\n\nChange-Id: Id822991a16c57342182abd0a846cc2f2a9931c06\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/c09665ee7d0a38a446a036f81b56477e360c76aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/c09665ee7d0a38a446a036f81b56477e360c76aa"}]},"branch":"refs/heads/master"},"bf925b9b301208a64b5e28c2e382da58949468f8":{"kind":"REWORK","_number":7,"created":"2026-04-14 09:45:17.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/11/980111/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/11/980111/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/11/980111/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/11/980111/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/11/980111/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/11/980111/7"}}},"commit":{"parents":[{"commit":"e71185216f432d66f5f25844df329715bf4a08cd","subject":"coordination: Change join_group capabilities default","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/e71185216f432d66f5f25844df329715bf4a08cd"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-03-09 16:29:20.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-04-14 09:45:09.000000000","tz":60},"subject":"typing: Add type hints to tooz.coordination, tooz.partitioner","message":"typing: Add type hints to tooz.coordination, tooz.partitioner\n\nThese are handled together since they are very closely coupled.\n\nChange-Id: Id822991a16c57342182abd0a846cc2f2a9931c06\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/bf925b9b301208a64b5e28c2e382da58949468f8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/bf925b9b301208a64b5e28c2e382da58949468f8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
