)]}'
{"id":"openstack%2Ftooz~980112","triplet_id":"openstack%2Ftooz~master~Ib325103ba0ab4b502e3fee6d0a60470e6b9cd510","project":"openstack/tooz","branch":"master","topic":"typing","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2026-03-14 21:02:19.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ib325103ba0ab4b502e3fee6d0a60470e6b9cd510","subject":"typing: Add type hints to tooz.drivers","status":"ABANDONED","created":"2026-03-11 15:42:02.000000000","updated":"2026-03-14 21:02:19.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"77ce082564c2efca2eaa6ef377a654856c792180","_number":980112,"virtual_id_number":980112,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-03-12 10:55:59.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-11 17:37:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a24d3a79237b54b09b0d02e33048740a672dc5d8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-03-11 15:42:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"870d9b057df790dc3fd1768881d884d2a10bb41c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-11 17:37:13.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c6910849c3434c0ca863c6a4dfce1b3f\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/c0d8fa6e5ee94d6b9c5b5fbaa6a54881 : SUCCESS in 3m 34s\n- requirements-check https://zuul.opendev.org/t/openstack/build/f6cd03f1c21749acb8442f4865c69ec6 : SUCCESS in 3m 24s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a6c8f25d12644afba55fbeec1ff9fba5 : SUCCESS in 1h 43m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c5ed15672c4348e5ad4a87dc89a293ec : SUCCESS in 6m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ed3e23397c7746369d80e39e1d8a21c7 : SUCCESS in 7m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/742d39c0aa644b6eb6f258af5e9dab20 : SUCCESS in 5m 17s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/076a1a653ad14ed4a1a4666d6674835a : SUCCESS in 5m 10s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/5d8e84cc19c3458fb49e753a9cb21575 : FAILURE in 7m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/11b2767b7430472891b46e9ed9cf881f : SUCCESS in 4m 57s\n- tooz-tox-examples https://zuul.opendev.org/t/openstack/build/497d2d225d234bbead2df17f2863c32a : SUCCESS in 3m 11s\n- tooz-tox-py310-etcd3gw https://zuul.opendev.org/t/openstack/build/4cc12a35dfc343219fa45bba9373a13c : FAILURE in 5m 02s\n- tooz-tox-py313-etcd3gw https://zuul.opendev.org/t/openstack/build/b9db7c4cfb4743308d828e6b19d0f8d3 : FAILURE in 4m 07s\n- tooz-tox-py310-zookeeper https://zuul.opendev.org/t/openstack/build/53ccab6fbf5849859dd4e679cdb8a276 : FAILURE in 4m 51s\n- tooz-tox-py313-zookeeper https://zuul.opendev.org/t/openstack/build/811c14e35ffd491d9454458c8071bd18 : FAILURE in 5m 38s\n- tooz-tox-py310-redis https://zuul.opendev.org/t/openstack/build/5456167066bd43e4bf875901b965835f : FAILURE in 9m 32s\n- tooz-tox-py313-redis https://zuul.opendev.org/t/openstack/build/9ec220550c21475495ac5afd4450cde0 : FAILURE in 7m 19s\n- tooz-tox-py310-sentinel https://zuul.opendev.org/t/openstack/build/ded95ac6fd1c4c69977edd06d8ba3dfe : FAILURE in 6m 39s\n- tooz-tox-py313-sentinel https://zuul.opendev.org/t/openstack/build/0beafc4dadc344dda09444d1bd4cb889 : FAILURE in 5m 51s\n- tooz-tox-py310-memcached https://zuul.opendev.org/t/openstack/build/a6fb654ff6e6463098afd46d23970b11 : SUCCESS in 4m 31s\n- tooz-tox-py313-memcached https://zuul.opendev.org/t/openstack/build/25fcaa42dff649149746b52a54cf9a8c : SUCCESS in 5m 20s\n- tooz-tox-py310-postgresql https://zuul.opendev.org/t/openstack/build/61e9faf6297044589d15b3d553d15337 : SUCCESS in 4m 03s\n- tooz-tox-py313-postgresql https://zuul.opendev.org/t/openstack/build/41a36a63d92d409fbb2f8cdaa34cff5e : SUCCESS in 6m 41s\n- tooz-tox-py310-mysql https://zuul.opendev.org/t/openstack/build/895925e4d20e4a7aa5e14614012fb3ed : SUCCESS in 4m 54s\n- tooz-tox-py313-mysql https://zuul.opendev.org/t/openstack/build/f7e049a704a64de18bc26eae0e95d5a1 : SUCCESS in 7m 16s\n- tooz-tox-py310-consul https://zuul.opendev.org/t/openstack/build/fe0955c889a44c3b9a8658f49df07796 : FAILURE in 5m 15s\n- tooz-tox-py313-consul https://zuul.opendev.org/t/openstack/build/c88e3e75cbb74edeb6345b053f347e36 : FAILURE in 5m 07s","accounts_in_message":[],"_revision_number":1},{"id":"b7d068e6f814810f14aead74fc43004592136366","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-03-12 08:42:07.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"419eb54fc000d600dfc65524c66b0af513117eae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-12 10:55:59.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d64f202fd01d44faadfc27ee43283c8d\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/23e25f882de249489856e2bdba94ef61 : SUCCESS in 3m 14s\n- requirements-check https://zuul.opendev.org/t/openstack/build/de41cd82eb07443a8531b959b14f79d7 : SUCCESS in 3m 47s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d225388e57994be2aedeaff3fc0a3270 : SUCCESS in 1h 49m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5a1c49e14e5348219ec46528bbcbbf7d : SUCCESS in 5m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0a7a83cd711a4295a7db59557f131030 : SUCCESS in 6m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8a30b7bfe74241189b9062600722e7ef : SUCCESS in 4m 47s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a51ee916c4924650a23a96eed80981e8 : SUCCESS in 5m 36s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/262939976bd543a5abe27087796abb2c : SUCCESS in 9m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/32cafbe4901d463c972a1147a88d75ce : SUCCESS in 4m 00s\n- tooz-tox-examples https://zuul.opendev.org/t/openstack/build/a18e25716bef4352a09817fd872684a6 : SUCCESS in 5m 01s\n- tooz-tox-py310-etcd3gw https://zuul.opendev.org/t/openstack/build/90bc96f347dc46c9af645b1fff030095 : FAILURE in 5m 26s\n- tooz-tox-py313-etcd3gw https://zuul.opendev.org/t/openstack/build/e1ea47aa37cd42498cca0c89c908de40 : FAILURE in 6m 26s\n- tooz-tox-py310-zookeeper https://zuul.opendev.org/t/openstack/build/a3f245f2244d4eba9c4680bfdae6c4d5 : FAILURE in 4m 22s\n- tooz-tox-py313-zookeeper https://zuul.opendev.org/t/openstack/build/954e53dfeeb0475a9e351f6843c4432b : FAILURE in 4m 51s\n- tooz-tox-py310-redis https://zuul.opendev.org/t/openstack/build/2111f83226474af08ed5b4836db0c6d7 : FAILURE in 7m 50s\n- tooz-tox-py313-redis https://zuul.opendev.org/t/openstack/build/d5261666629f42a0abab5c6721a3df68 : FAILURE in 5m 48s\n- tooz-tox-py310-sentinel https://zuul.opendev.org/t/openstack/build/5ce478be244e4ad3b4fe7346277e5976 : FAILURE in 6m 48s\n- tooz-tox-py313-sentinel https://zuul.opendev.org/t/openstack/build/5ee6ce020f61425792d39f1f2537b020 : FAILURE in 6m 45s\n- tooz-tox-py310-memcached https://zuul.opendev.org/t/openstack/build/d2652dece50c48daa61c51caf221fa32 : SUCCESS in 4m 59s\n- tooz-tox-py313-memcached https://zuul.opendev.org/t/openstack/build/165ce1a4b775473da0ffcd74e7441627 : SUCCESS in 5m 38s\n- tooz-tox-py310-postgresql https://zuul.opendev.org/t/openstack/build/fc3d83245f414e4e9633dc230cdb0e12 : SUCCESS in 5m 56s\n- tooz-tox-py313-postgresql https://zuul.opendev.org/t/openstack/build/52e970b1df9a493fab64a50df68a6efe : SUCCESS in 3m 32s\n- tooz-tox-py310-mysql https://zuul.opendev.org/t/openstack/build/1253193c8d3249c6b20d4a5aebee9fea : SUCCESS in 5m 13s\n- tooz-tox-py313-mysql https://zuul.opendev.org/t/openstack/build/5845de3e93494af7be8c18708a5dda4d : SUCCESS in 7m 18s\n- tooz-tox-py310-consul https://zuul.opendev.org/t/openstack/build/57c54a23a2584945bfd18e9195c27973 : FAILURE in 4m 52s\n- tooz-tox-py313-consul https://zuul.opendev.org/t/openstack/build/e3c358f0f5e44391af804a9e35408d8c : FAILURE in 5m 41s","accounts_in_message":[],"_revision_number":2},{"id":"77ce082564c2efca2eaa6ef377a654856c792180","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-03-14 21:02:19.000000000","message":"Abandoned\n\nThis has been split up","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"53d13ae86769f32e5617015a5b90f23b30bd9fd5","revisions":{"e1b5fb0bcacec207491ca873900b6ceb14fdb909":{"kind":"REWORK","_number":1,"created":"2026-03-11 15:42:02.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/12/980112/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/12/980112/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/12/980112/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/12/980112/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/12/980112/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/12/980112/1"}}},"commit":{"parents":[{"commit":"ad22b7ecad07ec71b42174f989ae3df2512a1a5e","subject":"typing: Add type hints to tooz.coordination, tooz.partitioner","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/ad22b7ecad07ec71b42174f989ae3df2512a1a5e"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-03-11 11:26:48.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-03-11 15:41:56.000000000","tz":0},"subject":"typing: Add type hints to tooz.drivers","message":"typing: Add type hints to tooz.drivers\n\nWe can do these all in one fell swoop since we have already typed\nthe rest of the code base.\n\nChange-Id: Ib325103ba0ab4b502e3fee6d0a60470e6b9cd510\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/e1b5fb0bcacec207491ca873900b6ceb14fdb909"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/e1b5fb0bcacec207491ca873900b6ceb14fdb909"}]},"branch":"refs/heads/master"},"53d13ae86769f32e5617015a5b90f23b30bd9fd5":{"kind":"REWORK","_number":2,"created":"2026-03-12 08:42:07.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/12/980112/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/12/980112/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/12/980112/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/12/980112/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/12/980112/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/12/980112/2"}}},"commit":{"parents":[{"commit":"2e1ff3bd0105ffbd089620c5f40e3c80061fac53","subject":"typing: Add type hints to tooz.coordination, tooz.partitioner","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/2e1ff3bd0105ffbd089620c5f40e3c80061fac53"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-03-11 11:26:48.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-03-12 08:40:43.000000000","tz":0},"subject":"typing: Add type hints to tooz.drivers","message":"typing: Add type hints to tooz.drivers\n\nWe can do these all in one fell swoop since we have already typed\nthe rest of the code base.\n\nChange-Id: Ib325103ba0ab4b502e3fee6d0a60470e6b9cd510\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/53d13ae86769f32e5617015a5b90f23b30bd9fd5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/53d13ae86769f32e5617015a5b90f23b30bd9fd5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
