)]}'
{"id":"openstack%2Ftooz~980556","triplet_id":"openstack%2Ftooz~master~Ic71b40d59c31e695448f46df00f2bc03479f075d","project":"openstack/tooz","branch":"master","topic":"trivial","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2026-05-05 16:05:04.000000000","reason":"Change was submitted"},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-05-05 16:05:04.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic71b40d59c31e695448f46df00f2bc03479f075d","subject":"trivial: Add explanation to test scripts","status":"MERGED","created":"2026-03-14 14:00:20.000000000","updated":"2026-05-05 16:07:13.000000000","submitted":"2026-05-05 16:05:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"980556-trivial","meta_rev_id":"60650c822f54f714a5e16823115bec0056d3f1e1","_number":980556,"virtual_id_number":980556,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-05-05 16:05:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2026-05-05 12:34:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2026-05-05 12:34:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2026-05-05 13:01:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2026-05-05 12:37:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-14 15:53:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-03-19 09:49:30.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2026-04-09 20:39:22.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2026-05-05 13:01:03.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"}],"messages":[{"id":"a238f25c3f63f3c4bbc80a206687a56ca08d39a9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-03-14 14:00:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e141a75c1207d4bc9bbe9908c2739aa1ad8591d0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-14 15:53:02.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ecc272b80cb74ffe9cf23b9b6d4d299c\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a6d355a0f6d24041a98208922ec3ce70 : SUCCESS in 1h 46m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b46774a0aec441b6b5c3e57fb9b20894 : SUCCESS in 6m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/575b95001bab4626b454568afc587f82 : SUCCESS in 6m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/de5c8abb720041f8bb08e2f27fca5336 : SUCCESS in 4m 09s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1b87f18d11ac4b5eb0f11ba01ab66ceb : SUCCESS in 10m 12s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/6e688ed6823a47cfbfba19746c5624e4 : SUCCESS in 7m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de195ca1705741bb8d353e0c9c22992d : SUCCESS in 6m 03s\n- tooz-tox-examples https://zuul.opendev.org/t/openstack/build/9ae32511f44e4a58af32c3ba462daed4 : SUCCESS in 4m 49s\n- tooz-tox-py310-etcd3gw https://zuul.opendev.org/t/openstack/build/72c0a6d9bc444d2eaf462b287a94dd7c : SUCCESS in 4m 37s\n- tooz-tox-py313-etcd3gw https://zuul.opendev.org/t/openstack/build/803cdfa40279425b86c1554fd95e909e : SUCCESS in 6m 30s\n- tooz-tox-py310-zookeeper https://zuul.opendev.org/t/openstack/build/2a18837d90804f32ba9c173cf130274a : SUCCESS in 4m 06s\n- tooz-tox-py313-zookeeper https://zuul.opendev.org/t/openstack/build/b254e8fa63674e8c8f32598a33646148 : SUCCESS in 7m 14s\n- tooz-tox-py310-redis https://zuul.opendev.org/t/openstack/build/4eb29cf482744d359710052690f4a4e8 : SUCCESS in 5m 24s\n- tooz-tox-py313-redis https://zuul.opendev.org/t/openstack/build/55ce55e9147c4b57b130b5ce994f1b8b : SUCCESS in 6m 36s\n- tooz-tox-py310-sentinel https://zuul.opendev.org/t/openstack/build/a2380089ad8947d089d8e6332927a113 : SUCCESS in 6m 13s\n- tooz-tox-py313-sentinel https://zuul.opendev.org/t/openstack/build/4fc5c6bb50ab4a409485bc7ad32f9608 : SUCCESS in 7m 40s\n- tooz-tox-py310-memcached https://zuul.opendev.org/t/openstack/build/ce40e9e2e0f4445fa52d22cc0963663d : SUCCESS in 2m 52s\n- tooz-tox-py313-memcached https://zuul.opendev.org/t/openstack/build/fcb5659616f442539f02f28b4188adeb : SUCCESS in 4m 51s\n- tooz-tox-py310-postgresql https://zuul.opendev.org/t/openstack/build/522a4022d93f4e668753a00e63a467a1 : SUCCESS in 4m 00s\n- tooz-tox-py313-postgresql https://zuul.opendev.org/t/openstack/build/352768af08954526b04be19b5fd5ff59 : SUCCESS in 2m 44s\n- tooz-tox-py310-mysql https://zuul.opendev.org/t/openstack/build/54a7571e2da2474d950fd944e2ae7cec : SUCCESS in 4m 00s\n- tooz-tox-py313-mysql https://zuul.opendev.org/t/openstack/build/f29f8534ee574cefb7ea4b9bc95b1e81 : SUCCESS in 8m 20s\n- tooz-tox-py310-consul https://zuul.opendev.org/t/openstack/build/a466e36a62d74b109c6916460ac303c6 : FAILURE in 2m 42s\n- tooz-tox-py313-consul https://zuul.opendev.org/t/openstack/build/a7cbcb76e02d49d785d98d607a4397b9 : FAILURE in 4m 21s","accounts_in_message":[],"_revision_number":1},{"id":"f3ac201c8bdc313fff33fc1096625638a8246e02","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-03-19 09:49:06.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"00fa57ea7222121768fff1dc9942e28cd4cb2689","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-19 10:55:13.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a6cf985dce004e6ea692e495d688f98d\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0801f1c480f7441784fc9bb99c01bc73 : SUCCESS in 1h 01m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a319df7080fc44aabde7924d58261095 : SUCCESS in 5m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2c432a7444e946c1ae39d170fb818ea4 : SUCCESS in 6m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1f1db44d182442d58409e0e3a3416d4d : SUCCESS in 5m 11s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ba472c6eb14c4699a585aa358894b323 : SUCCESS in 5m 58s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/287b0902ddc34470a2cd2e1d0f73ef7e : SUCCESS in 7m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c19d529b7944c6abb67a1513619972e : SUCCESS in 5m 18s\n- tooz-tox-examples https://zuul.opendev.org/t/openstack/build/1e8117985fb94424bb0d766df2d928cb : SUCCESS in 4m 52s\n- tooz-tox-py310-etcd3gw https://zuul.opendev.org/t/openstack/build/9c80859d8f544f8a9d1ef186a448deed : SUCCESS in 6m 25s\n- tooz-tox-py313-etcd3gw https://zuul.opendev.org/t/openstack/build/c15605e2c4ae417c99cca62b4eb694cb : SUCCESS in 3m 33s\n- tooz-tox-py310-zookeeper https://zuul.opendev.org/t/openstack/build/637f9d13818a42a88535916950377151 : SUCCESS in 4m 27s\n- tooz-tox-py313-zookeeper https://zuul.opendev.org/t/openstack/build/e6558d2d844949dbbe6bfeee486cae0a : SUCCESS in 5m 20s\n- tooz-tox-py310-redis https://zuul.opendev.org/t/openstack/build/63d3fa3dd6ec4d7eb534a47357d00aaa : SUCCESS in 6m 25s\n- tooz-tox-py313-redis https://zuul.opendev.org/t/openstack/build/711298fabe5b4c009e2b576849a395cc : SUCCESS in 6m 01s\n- tooz-tox-py310-sentinel https://zuul.opendev.org/t/openstack/build/d1666457ee8a4d729da874300d5db26d : SUCCESS in 6m 51s\n- tooz-tox-py313-sentinel https://zuul.opendev.org/t/openstack/build/46cf0386a7e0464bbc4c0d986beed826 : SUCCESS in 7m 42s\n- tooz-tox-py310-memcached https://zuul.opendev.org/t/openstack/build/004f48b7c0bd4f0cb38ce5d54273b50a : SUCCESS in 3m 12s\n- tooz-tox-py313-memcached https://zuul.opendev.org/t/openstack/build/1898f61492954bcd8d1d12ac89fd6bbe : SUCCESS in 6m 32s\n- tooz-tox-py310-postgresql https://zuul.opendev.org/t/openstack/build/7cf6cf0404ee48969787bd5c5c85dc12 : SUCCESS in 4m 54s\n- tooz-tox-py313-postgresql https://zuul.opendev.org/t/openstack/build/30626f5fcff24474baa89c82b5fc0df9 : SUCCESS in 6m 17s\n- tooz-tox-py310-mysql https://zuul.opendev.org/t/openstack/build/a46496f3de1d4ef2af475c9b32f1eda2 : SUCCESS in 5m 06s\n- tooz-tox-py313-mysql https://zuul.opendev.org/t/openstack/build/f658be081518495c9e716569e9ce7e53 : SUCCESS in 8m 11s","accounts_in_message":[],"_revision_number":2},{"id":"06f13478580a3c4a093fe1a51c94107e4e54cd25","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-03-20 11:02:39.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"d3a3aa784437714a369289258ecf2ba0a54d0ad7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-20 12:59:55.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fa9bf2e1586a4bfcaa2399e2c57ddc81\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/928e721010c94771808f9879969de5c0 : SUCCESS in 1h 50m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/aba82b55917f4f5a946cfdd2077a8ff7 : SUCCESS in 3m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9d4557ea0a3d45ae9777223ce7d11a11 : SUCCESS in 5m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0a02a3388421427ab1e361e31d0067d2 : SUCCESS in 3m 23s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/59862bedab23401cb0dc24766a17f465 : SUCCESS in 5m 08s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/bfe7bf22eb8942388a60a700f419f5e2 : SUCCESS in 7m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec7c94a1197e447a9e01ecc86b443fc9 : SUCCESS in 5m 50s\n- tooz-tox-examples https://zuul.opendev.org/t/openstack/build/04c91f35a51c46bfafae1c42611b2123 : SUCCESS in 4m 44s\n- tooz-tox-py310-etcd3gw https://zuul.opendev.org/t/openstack/build/28c3ba39416146ed841c0137bc33b498 : SUCCESS in 4m 46s\n- tooz-tox-py313-etcd3gw https://zuul.opendev.org/t/openstack/build/3cf961938c2649f3b50b5c36a9eb8603 : SUCCESS in 5m 11s\n- tooz-tox-py310-zookeeper https://zuul.opendev.org/t/openstack/build/2c428506f2a44ec0b84fbfbfdcefca1d : SUCCESS in 4m 08s\n- tooz-tox-py313-zookeeper https://zuul.opendev.org/t/openstack/build/f1d2cf3bd2094737ad34b0b2d9d6f3bb : SUCCESS in 4m 24s\n- tooz-tox-py310-redis https://zuul.opendev.org/t/openstack/build/dcd603272697472e92fb228843ee345c : SUCCESS in 6m 20s\n- tooz-tox-py313-redis https://zuul.opendev.org/t/openstack/build/1b9941fbd9ae47f2b92b30d63143703f : SUCCESS in 7m 35s\n- tooz-tox-py310-sentinel https://zuul.opendev.org/t/openstack/build/42983a494e7b42429bd4c16a9819c702 : SUCCESS in 6m 09s\n- tooz-tox-py313-sentinel https://zuul.opendev.org/t/openstack/build/6bfa03b3de1c4fd99f028d8d3e18f4c4 : SUCCESS in 6m 48s\n- tooz-tox-py310-memcached https://zuul.opendev.org/t/openstack/build/880a7b2e1ec64abe9d90b8385e2bb491 : SUCCESS in 4m 18s\n- tooz-tox-py313-memcached https://zuul.opendev.org/t/openstack/build/b856a3a3a2af4195b664610302140b92 : SUCCESS in 5m 22s\n- tooz-tox-py310-postgresql https://zuul.opendev.org/t/openstack/build/a5c9ed777ce54f759102c336d959730b : SUCCESS in 3m 01s\n- tooz-tox-py313-postgresql https://zuul.opendev.org/t/openstack/build/8ab4db4942a84a81a0ae765f17d254fa : SUCCESS in 2m 45s\n- tooz-tox-py310-mysql https://zuul.opendev.org/t/openstack/build/d04457e891f44b499cd30b9637b6c7fe : SUCCESS in 4m 24s\n- tooz-tox-py313-mysql https://zuul.opendev.org/t/openstack/build/5fa1d6ea28e2476ca1f9c1c86c95089e : SUCCESS in 7m 36s","accounts_in_message":[],"_revision_number":3},{"id":"64bb80bbbabba610ca39b3fa7ac81bf580e453df","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2026-04-09 20:39:22.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"a11d2c1b64622ce6ca39704cadc90ae6835cef48","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"real_author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-04-26 16:02:27.000000000","message":"Patch Set 4: Patch Set 3 was rebased on behalf of \u003cGERRIT_ACCOUNT_15334\u003e\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}],"_revision_number":4},{"id":"e9fb51270d416609304c265bc0b023bac65ef293","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-04-26 16:02:41.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"2631af0376d2b479c4df1003c0bf0b6cc461c6d0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-26 17:01:44.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/884c8480ee0f491d9b162307ec5440bd\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8be74049bba84902a7ec85b92dbc2efc : SUCCESS in 57m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2303af120be74f689f02d42f74ef0c25 : SUCCESS in 5m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba7cecee86334f7ca89169539c420d9e : SUCCESS in 3m 31s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fb510ab2d33648869bf3a4cdb98ad6ff : SUCCESS in 4m 15s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2b047051220a4d39a27ae25166427c96 : SUCCESS in 4m 34s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/1a79ed4859f94347992463c80d49963a : FAILURE in 9m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/daf6558e6f214550ab3a7a30d63eb86c : SUCCESS in 6m 11s\n- tooz-tox-examples https://zuul.opendev.org/t/openstack/build/a40a72b7078540d0ab42529d786d0335 : SUCCESS in 4m 42s\n- tooz-tox-py310-etcd3gw https://zuul.opendev.org/t/openstack/build/66d8818c38fc4147a176a68a2218fc08 : SUCCESS in 3m 49s\n- tooz-tox-py313-etcd3gw https://zuul.opendev.org/t/openstack/build/7d8747db2403471795b7580dadef0f71 : SUCCESS in 3m 46s\n- tooz-tox-py310-zookeeper https://zuul.opendev.org/t/openstack/build/c23326906b9047168f67ebdb764792bb : SUCCESS in 3m 07s\n- tooz-tox-py313-zookeeper https://zuul.opendev.org/t/openstack/build/e6653a1e09344f3a8fc7e4344b646693 : SUCCESS in 3m 32s\n- tooz-tox-py310-redis https://zuul.opendev.org/t/openstack/build/2cf564edf31d41a79e08c2c3f8849349 : SUCCESS in 5m 36s\n- tooz-tox-py313-redis https://zuul.opendev.org/t/openstack/build/96cd7e58360f43c79784871f6bf5254f : SUCCESS in 5m 52s\n- tooz-tox-py310-sentinel https://zuul.opendev.org/t/openstack/build/c294dfb7f317489bb6560be6efac3c14 : SUCCESS in 5m 14s\n- tooz-tox-py313-sentinel https://zuul.opendev.org/t/openstack/build/930f396dd0ae4341b915d8c08cd58afa : SUCCESS in 6m 00s\n- tooz-tox-py310-memcached https://zuul.opendev.org/t/openstack/build/21b6609e1fb84152838a337023efe388 : SUCCESS in 4m 07s\n- tooz-tox-py313-memcached https://zuul.opendev.org/t/openstack/build/4b12d3df599d4d38884e7ccf78758a76 : SUCCESS in 6m 18s\n- tooz-tox-py310-postgresql https://zuul.opendev.org/t/openstack/build/99be7dfb116b4042ab0679170aac0abb : SUCCESS in 3m 51s\n- tooz-tox-py313-postgresql https://zuul.opendev.org/t/openstack/build/8a04877f299f4de09b30736b81a4cc20 : SUCCESS in 4m 25s\n- tooz-tox-py310-mysql https://zuul.opendev.org/t/openstack/build/010988ce6c6440189dfb3ac88320a184 : SUCCESS in 3m 46s\n- tooz-tox-py313-mysql https://zuul.opendev.org/t/openstack/build/a06d6ad9c4a940ddb5b26503f62bf547 : SUCCESS in 7m 01s","accounts_in_message":[],"_revision_number":4},{"id":"15eb3b70e8fadb4aa437b037bc5cf9ba18920319","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-05-05 12:34:36.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n* Workflow+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":5},{"id":"107485b1bdb8d52f3a37aa740637147af1ffd459","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-05-05 12:37:02.000000000","message":"Patch Set 5: Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"25e9f942e40327b164c2f6bea68fb3c9707264ce","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2026-05-05 13:01:03.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"9f97f3a93d150ac9e0cb12f8fce4e98f6d36d046","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-05 14:22:58.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e8aa6841c6d840ebaa685888fc2f422b\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4dd19158ec6a4f33bd5479ca96bf5272 : SUCCESS in 1h 38m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/04774f1e008b4bc0985747e248d6d62c : SUCCESS in 7m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e9538efbb0a249d4af9e7470320e8c30 : SUCCESS in 3m 41s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d479b0fbf24848858656ec6335066eec : SUCCESS in 5m 16s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ad0d9fc3e35147088e3f9c72d7f0e4bd : SUCCESS in 5m 55s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/83584e0daff64224aa3cdcb15c5b72e9 : FAILURE in 9m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e094cfeef4b45d1a291be1d5449fccb : SUCCESS in 4m 39s\n- tooz-tox-examples https://zuul.opendev.org/t/openstack/build/97417533b0ca4451b820bdc332a0af63 : SUCCESS in 3m 20s\n- tooz-tox-py310-etcd3gw https://zuul.opendev.org/t/openstack/build/aed446ee4b9e41a09cf614d4bf132233 : SUCCESS in 3m 57s\n- tooz-tox-py313-etcd3gw https://zuul.opendev.org/t/openstack/build/67b61d4ed6fb4143bd982995a8e5e8fe : SUCCESS in 7m 58s\n- tooz-tox-py310-zookeeper https://zuul.opendev.org/t/openstack/build/9ac74380792d42a099e9c054fe606052 : SUCCESS in 4m 07s\n- tooz-tox-py313-zookeeper https://zuul.opendev.org/t/openstack/build/fa31c6e23f5845169e7561ed21fe4887 : SUCCESS in 5m 03s\n- tooz-tox-py310-redis https://zuul.opendev.org/t/openstack/build/0e2a2dc8313b498d9dc7dbdf8109bd16 : SUCCESS in 6m 37s\n- tooz-tox-py313-redis https://zuul.opendev.org/t/openstack/build/826a433b06ce44eca875aeab35263ec7 : SUCCESS in 6m 37s\n- tooz-tox-py310-sentinel https://zuul.opendev.org/t/openstack/build/162728f99fd945be9a91187afdfee201 : SUCCESS in 6m 03s\n- tooz-tox-py313-sentinel https://zuul.opendev.org/t/openstack/build/fa3de39475204d8cace585c30013066b : SUCCESS in 6m 25s\n- tooz-tox-py310-memcached https://zuul.opendev.org/t/openstack/build/61770865b6074c3296178e001a2b3799 : SUCCESS in 3m 26s\n- tooz-tox-py313-memcached https://zuul.opendev.org/t/openstack/build/fd02f2b3057e49e2a725c912c5c44637 : SUCCESS in 4m 59s\n- tooz-tox-py310-postgresql https://zuul.opendev.org/t/openstack/build/16e54bd628464e58830f9ea4489ef9e0 : SUCCESS in 4m 48s\n- tooz-tox-py313-postgresql https://zuul.opendev.org/t/openstack/build/c4ce14460e5b4adea28225c97fe56ea0 : SUCCESS in 6m 46s\n- tooz-tox-py310-mysql https://zuul.opendev.org/t/openstack/build/2df9a85053ed40cb89596c96547b04ae : SUCCESS in 4m 01s\n- tooz-tox-py313-mysql https://zuul.opendev.org/t/openstack/build/d6df8f5d357c47b88388b0398bfd68f0 : SUCCESS in 7m 12s\n- tooz-tox-py310-consul https://zuul.opendev.org/t/openstack/build/f09473f2e75648d7a707271157682b59 : SUCCESS in 4m 45s\n- tooz-tox-py313-consul https://zuul.opendev.org/t/openstack/build/399cc738e0a94dbfb98fa09eb3878efa : SUCCESS in 5m 32s","accounts_in_message":[],"_revision_number":5},{"id":"06fd7ad8d1e42311624f58d8fe60c2c3129f2142","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-05 14:23:26.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"468dceff1b44c12887d22c16b31a7e787c8d1cb9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-05 16:05:04.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c83e9941a01f4e9583e0ffaa36a5dd98\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/17ac1419704b44099c71ef67a8395ea3 : SUCCESS in 1h 37m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bda2aa4c32b744c9a75cbe1356370555 : SUCCESS in 6m 02s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8e6d5be654c14044b2d95a62e5afaf5a : SUCCESS in 6m 49s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/502aa5e54c704b088fdfe8c09b6bd70f : SUCCESS in 5m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7cd68f9a2a6247028c9bf178881d9c82 : SUCCESS in 4m 45s\n- tooz-tox-examples https://zuul.opendev.org/t/openstack/build/d301bd4920724a0c80acc005c42d86f7 : SUCCESS in 5m 00s\n- tooz-tox-py310-etcd3gw https://zuul.opendev.org/t/openstack/build/e47f0b86a63349bf83d76884ab314915 : SUCCESS in 4m 55s\n- tooz-tox-py313-etcd3gw https://zuul.opendev.org/t/openstack/build/acd59525fc4340838c4b17ff16b525ba : SUCCESS in 6m 14s\n- tooz-tox-py310-zookeeper https://zuul.opendev.org/t/openstack/build/603adf19898247718bdf343232bf5de0 : SUCCESS in 4m 00s\n- tooz-tox-py313-zookeeper https://zuul.opendev.org/t/openstack/build/7a778b3ae6044b55b39b904bb252bef8 : SUCCESS in 4m 02s\n- tooz-tox-py310-redis https://zuul.opendev.org/t/openstack/build/6c970003fa33422391a90ad42b3d00f8 : SUCCESS in 8m 10s\n- tooz-tox-py313-redis https://zuul.opendev.org/t/openstack/build/9e9aadfd43394df18353adf900087fc1 : SUCCESS in 7m 35s\n- tooz-tox-py310-sentinel https://zuul.opendev.org/t/openstack/build/eeafcf7999fc4254aad0ca07e57bf170 : SUCCESS in 6m 15s\n- tooz-tox-py313-sentinel https://zuul.opendev.org/t/openstack/build/6a700d7b5aad4938a110c14b115b87d3 : SUCCESS in 6m 38s\n- tooz-tox-py310-memcached https://zuul.opendev.org/t/openstack/build/82529c8b8a6240e3a363de85c2e17357 : SUCCESS in 6m 55s\n- tooz-tox-py313-memcached https://zuul.opendev.org/t/openstack/build/f8e7124416c545468939501794875260 : SUCCESS in 5m 49s\n- tooz-tox-py310-postgresql https://zuul.opendev.org/t/openstack/build/509258dcf7844bfe83d0d0c3364e6ab8 : SUCCESS in 3m 34s\n- tooz-tox-py313-postgresql https://zuul.opendev.org/t/openstack/build/6e24dec475cb493aabaf5852bd814520 : SUCCESS in 5m 26s\n- tooz-tox-py310-mysql https://zuul.opendev.org/t/openstack/build/6e4cb827620343069121d6a1775e8eb3 : SUCCESS in 4m 29s\n- tooz-tox-py313-mysql https://zuul.opendev.org/t/openstack/build/75a04af6f5c446b1bedb6ef114d1c850 : SUCCESS in 7m 56s\n- tooz-tox-py310-consul https://zuul.opendev.org/t/openstack/build/606ce7b621f048f880489b4728fc1549 : SUCCESS in 4m 16s\n- tooz-tox-py313-consul https://zuul.opendev.org/t/openstack/build/9dac9cfc2e374402819bcacffe422b48 : SUCCESS in 4m 02s","accounts_in_message":[],"_revision_number":5},{"id":"14c3ebda686af293cbad15c156b14b64db4f3d5c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-05 16:05:04.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"60650c822f54f714a5e16823115bec0056d3f1e1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-05 16:07:13.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/90989e31fd224feba1cc4afbf93542ae\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21eea1a107514c0f9cc0498bfc1afd56 : SUCCESS in 1m 06s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"d4820f6eeb44b096c7faf1998a9252ae8cdfdba5","revisions":{"ec1ae4051e002a9f29c5344f99593dfeec7d4aee":{"kind":"REWORK","_number":1,"created":"2026-03-14 14:00:20.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/56/980556/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/56/980556/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/56/980556/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/56/980556/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/56/980556/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/56/980556/1"}}},"commit":{"parents":[{"commit":"8302cdbe97360c6b92fb6ec978874aab1d519942","subject":"Remove consul driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/8302cdbe97360c6b92fb6ec978874aab1d519942"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-03-13 22:50:19.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-03-14 14:00:07.000000000","tz":0},"subject":"trivial: Add explanation to test scripts","message":"trivial: Add explanation to test scripts\n\nChange-Id: Ic71b40d59c31e695448f46df00f2bc03479f075d\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/ec1ae4051e002a9f29c5344f99593dfeec7d4aee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/ec1ae4051e002a9f29c5344f99593dfeec7d4aee"}]},"branch":"refs/heads/master"},"e4be0f98850de6b1eb9d9f7f1300387e8aee1822":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-03-19 09:49:06.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/56/980556/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/56/980556/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/56/980556/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/56/980556/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/56/980556/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/56/980556/2"}}},"commit":{"parents":[{"commit":"496bd33614af9161142f7f513b40ef8744e17661","subject":"Remove consul driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/496bd33614af9161142f7f513b40ef8744e17661"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-03-13 22:50:19.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-03-19 09:48:53.000000000","tz":60},"subject":"trivial: Add explanation to test scripts","message":"trivial: Add explanation to test scripts\n\nChange-Id: Ic71b40d59c31e695448f46df00f2bc03479f075d\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/e4be0f98850de6b1eb9d9f7f1300387e8aee1822"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/e4be0f98850de6b1eb9d9f7f1300387e8aee1822"}]},"branch":"refs/heads/master"},"e70b0e683ec82d4e14a3d8a21a500601b71f7960":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2026-03-20 11:02:39.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/56/980556/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/56/980556/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/56/980556/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/56/980556/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/56/980556/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/56/980556/3"}}},"commit":{"parents":[{"commit":"6838928f77f9d1fcf6323f1790ae720e60d567bd","subject":"Remove consul driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/6838928f77f9d1fcf6323f1790ae720e60d567bd"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-03-13 22:50:19.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-03-20 11:02:32.000000000","tz":60},"subject":"trivial: Add explanation to test scripts","message":"trivial: Add explanation to test scripts\n\nChange-Id: Ic71b40d59c31e695448f46df00f2bc03479f075d\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/e70b0e683ec82d4e14a3d8a21a500601b71f7960"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/e70b0e683ec82d4e14a3d8a21a500601b71f7960"}]},"branch":"refs/heads/master"},"ac875daf79191cf9cf097dcd596f974974d088d4":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2026-04-26 16:02:27.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"real_uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/56/980556/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/56/980556/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/56/980556/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/56/980556/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/56/980556/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/56/980556/4"}}},"commit":{"parents":[{"commit":"74fc3dbe7c58b770f6eb1f615a7f4c75fce3f46f","subject":"Remove consul driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/74fc3dbe7c58b770f6eb1f615a7f4c75fce3f46f"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-03-13 22:50:19.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-04-26 16:02:27.000000000","tz":0},"subject":"trivial: Add explanation to test scripts","message":"trivial: Add explanation to test scripts\n\nChange-Id: Ic71b40d59c31e695448f46df00f2bc03479f075d\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/ac875daf79191cf9cf097dcd596f974974d088d4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/ac875daf79191cf9cf097dcd596f974974d088d4"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"ours":"e70b0e683ec82d4e14a3d8a21a500601b71f7960","theirs":"74fc3dbe7c58b770f6eb1f615a7f4c75fce3f46f","contains_conflicts":false}},"d4820f6eeb44b096c7faf1998a9252ae8cdfdba5":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2026-05-05 12:34:36.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/56/980556/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/56/980556/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/56/980556/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/56/980556/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/56/980556/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/56/980556/5"}}},"commit":{"parents":[{"commit":"f022fae5f86c1e41648f574e5fccd91323330df0","subject":"Merge \"typing: Enable strict checking\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/f022fae5f86c1e41648f574e5fccd91323330df0"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-03-13 22:50:19.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-05-05 12:28:28.000000000","tz":60},"subject":"trivial: Add explanation to test scripts","message":"trivial: Add explanation to test scripts\n\nChange-Id: Ic71b40d59c31e695448f46df00f2bc03479f075d\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/d4820f6eeb44b096c7faf1998a9252ae8cdfdba5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/d4820f6eeb44b096c7faf1998a9252ae8cdfdba5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
