)]}'
{"id":"openstack%2Ftooz~980558","triplet_id":"openstack%2Ftooz~master~Ic13ff11143ff78cce4def53a23d3538ac02aed45","project":"openstack/tooz","branch":"master","topic":"typing","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2026-03-18 17:42:19.000000000","reason":"\u003cGERRIT_ACCOUNT_15334\u003e replied on the change","reason_account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}}},"hashtags":[],"change_id":"Ic13ff11143ff78cce4def53a23d3538ac02aed45","subject":"tests: Add test for str IDs","status":"MERGED","created":"2026-03-14 14:26:12.000000000","updated":"2026-03-18 20:33:00.000000000","submitted":"2026-03-18 20:32:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"980558-typing","meta_rev_id":"d90029ac48e9ed16b07336622442a00147f2584f","_number":980558,"virtual_id_number":980558,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-03-18 20:32:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2026-03-18 12:07:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2026-03-18 12:07:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-14 16:40:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9f2912df9a60325a42a14e0250e6b03e495b7975","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-03-14 14:26:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"319d026045fca5aa754e02922d920d5d33705cb7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-03-14 14:36:54.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"8a50a3446412310c7ef0db8f5df564ec4498c00c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-03-14 14:37:03.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"5edf2d26f7cfa183ccec09cafac3966af0c4f039","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-14 16:40:08.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d005b063fe7149d68e593485f127e121\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/aba3372e7f534142bcc8eadb24222a86 : SUCCESS in 1h 57m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b5dafc2e79a04c0889f97959edcca97b : SUCCESS in 5m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/313f1c16471a4e3b96f0c072189e91d8 : SUCCESS in 5m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/71c2739c2de54706bf1326e6450d9840 : SUCCESS in 3m 59s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/66fc678f88ed4dae96d13c28c19996e2 : SUCCESS in 8m 06s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/e9c3d126c1f7492abb90b4ba0a0987e8 : SUCCESS in 4m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ace80081296c460994d8d9759bd1ebb2 : SUCCESS in 5m 24s\n- tooz-tox-examples https://zuul.opendev.org/t/openstack/build/db8db967ebcc48aabc69d26788608aaa : SUCCESS in 3m 01s\n- tooz-tox-py310-etcd3gw https://zuul.opendev.org/t/openstack/build/3ab8e819826a476abbdf0867a7845ec2 : SUCCESS in 7m 58s\n- tooz-tox-py313-etcd3gw https://zuul.opendev.org/t/openstack/build/69704e17171e470db565c602e2bc13e3 : SUCCESS in 5m 58s\n- tooz-tox-py310-zookeeper https://zuul.opendev.org/t/openstack/build/b44ea7d4e25b485194fb5475389dab3b : FAILURE in 6m 02s\n- tooz-tox-py313-zookeeper https://zuul.opendev.org/t/openstack/build/7eafaba29f344c0b9dbbebb74c428637 : SUCCESS in 3m 07s\n- tooz-tox-py310-redis https://zuul.opendev.org/t/openstack/build/7e8dbca7c8d1466690890de2e9220e51 : SUCCESS in 5m 48s\n- tooz-tox-py313-redis https://zuul.opendev.org/t/openstack/build/624098d03f3b49d9be369e7c5a92bc8b : SUCCESS in 6m 56s\n- tooz-tox-py310-sentinel https://zuul.opendev.org/t/openstack/build/1014cdf3d99a4de280ecedcdb2f93cc7 : SUCCESS in 5m 48s\n- tooz-tox-py313-sentinel https://zuul.opendev.org/t/openstack/build/e265831b2b134b6994f7a3b08bc4768d : SUCCESS in 7m 20s\n- tooz-tox-py310-memcached https://zuul.opendev.org/t/openstack/build/fdc74a93aeb84a34be88f2d8baff72fc : SUCCESS in 4m 35s\n- tooz-tox-py313-memcached https://zuul.opendev.org/t/openstack/build/79af63e9f16943c698ec74a0d08726d1 : SUCCESS in 5m 28s\n- tooz-tox-py310-postgresql https://zuul.opendev.org/t/openstack/build/8b14f516ed4140918e83a26837648257 : SUCCESS in 2m 55s\n- tooz-tox-py313-postgresql https://zuul.opendev.org/t/openstack/build/a4d62a0c70364625993526cf526d22a9 : SUCCESS in 4m 11s\n- tooz-tox-py310-mysql https://zuul.opendev.org/t/openstack/build/e0925ae38fcd47a1999dd8a4b48bb822 : SUCCESS in 5m 20s\n- tooz-tox-py313-mysql https://zuul.opendev.org/t/openstack/build/1d6b75942ab8460fa82e18fb734179f0 : SUCCESS in 6m 17s\n- tooz-tox-py310-consul https://zuul.opendev.org/t/openstack/build/5b59a3533b2f43f89b2792615d2e29b1 : SUCCESS in 5m 00s\n- tooz-tox-py313-consul https://zuul.opendev.org/t/openstack/build/26dff19dda854e96b537e17021151fda : SUCCESS in 5m 09s","accounts_in_message":[],"_revision_number":3},{"id":"8c87c3de00ea69b229d33156cd0dba4e6a980d2b","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-03-18 12:07:33.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"3f5a0aca58e0f80a0c5c6c49d50ee3b853f0111e","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-03-18 12:07:38.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"9e240b90fdd8afc24f1ac8347bd4ea1a61441201","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-18 13:58:37.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fe40720e203d4ef5b774c1af25602c6c\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0d9c0336a6944ce79c5d27e825c3df5d : SUCCESS in 1h 45m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8039b4490a8c41df81ba75436d2a1fd9 : SUCCESS in 5m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a420050da97a4f4abf4ce5eee620e124 : SUCCESS in 6m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/de3d68fe98aa4a309373eac403abfa9c : SUCCESS in 4m 38s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/21e89fec9de041f498a938043b7be5f0 : SUCCESS in 6m 50s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/7133a9aec5294eba81c81380bf46a8fb : SUCCESS in 5m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f7b9f23932a045d6a3ce14dde3f8ce1c : SUCCESS in 4m 44s\n- tooz-tox-examples https://zuul.opendev.org/t/openstack/build/50f49ae9bb704f0bb03252f2fdb1164d : SUCCESS in 4m 31s\n- tooz-tox-py310-etcd3gw https://zuul.opendev.org/t/openstack/build/64c84e3221ce4142876b6630d04d7461 : SUCCESS in 4m 46s\n- tooz-tox-py313-etcd3gw https://zuul.opendev.org/t/openstack/build/1d5c6dc04fc843aca7f6f06b3825a942 : SUCCESS in 6m 53s\n- tooz-tox-py310-zookeeper https://zuul.opendev.org/t/openstack/build/3bcb174a81e84059a9ba9b705a245a92 : SUCCESS in 3m 01s\n- tooz-tox-py313-zookeeper https://zuul.opendev.org/t/openstack/build/62eb2b335fdc4b04b5c6b2d8bccf6925 : SUCCESS in 5m 54s\n- tooz-tox-py310-redis https://zuul.opendev.org/t/openstack/build/165ef21891d248878edbf3cb8d53b331 : SUCCESS in 6m 38s\n- tooz-tox-py313-redis https://zuul.opendev.org/t/openstack/build/35176e5bc7394a2297e5d7925a89f0ab : SUCCESS in 6m 06s\n- tooz-tox-py310-sentinel https://zuul.opendev.org/t/openstack/build/a4e239af9f7e441681eeab2e049fc6bf : SUCCESS in 5m 57s\n- tooz-tox-py313-sentinel https://zuul.opendev.org/t/openstack/build/378640c6c3e94ca6a5684a6491b5e22a : SUCCESS in 6m 21s\n- tooz-tox-py310-memcached https://zuul.opendev.org/t/openstack/build/431550bff5fa45d1a71cf0dd52ad0948 : SUCCESS in 4m 11s\n- tooz-tox-py313-memcached https://zuul.opendev.org/t/openstack/build/14e4c16c02e6447289ead2d50c0ef78a : SUCCESS in 4m 42s\n- tooz-tox-py310-postgresql https://zuul.opendev.org/t/openstack/build/dbaabf5cf7504f588b1190f5a309a2f0 : SUCCESS in 3m 17s\n- tooz-tox-py313-postgresql https://zuul.opendev.org/t/openstack/build/0111a1f8d4e147508e1fa529df2a4885 : SUCCESS in 4m 21s\n- tooz-tox-py310-mysql https://zuul.opendev.org/t/openstack/build/b4575f69613e4249ae0b22a35916f0de : SUCCESS in 4m 18s\n- tooz-tox-py313-mysql https://zuul.opendev.org/t/openstack/build/92f804cb5ddd44648137d5dc75f9d698 : SUCCESS in 9m 43s\n- tooz-tox-py310-consul https://zuul.opendev.org/t/openstack/build/c8d12c3c425c44569fd2965fafa35e66 : SUCCESS in 3m 20s\n- tooz-tox-py313-consul https://zuul.opendev.org/t/openstack/build/152c841602304c7da18070afba0c5431 : SUCCESS in 3m 27s","accounts_in_message":[],"_revision_number":3},{"id":"c4104407e899558faa22c9889fe32a1ffa0993b8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-18 13:59:04.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"5e430ab47b2a3f43c171cccce1ba7912a25b2d3f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-18 15:16:32.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/712ed780036d4b3087969dc6e87d1968\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2df0b0bb20764bd5ade5163ec1233b8b : SUCCESS in 1h 00m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fd2eba3b02484194ab6ff7a0aa3050ac : SUCCESS in 6m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/04e6ce957ad2431392a53ab1e9ca9e95 : SUCCESS in 4m 37s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4a5423c5d68a43c99037867e174eab3d : SUCCESS in 5m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1fea9c236e1b47f497533ce2c485b8da : SUCCESS in 5m 05s\n- tooz-tox-examples https://zuul.opendev.org/t/openstack/build/77b845d6e2db4392bd0a83cb9593a116 : SUCCESS in 4m 35s\n- tooz-tox-py310-etcd3gw https://zuul.opendev.org/t/openstack/build/b3c0b857ef7840c6ad95e855e17bdc7b : SUCCESS in 4m 39s\n- tooz-tox-py313-etcd3gw https://zuul.opendev.org/t/openstack/build/0c5fc4c9393c4c058164c2df09bd7439 : SUCCESS in 5m 44s\n- tooz-tox-py310-zookeeper https://zuul.opendev.org/t/openstack/build/220b5765a4f44cebb779e75eea11e1bd : SUCCESS in 4m 44s\n- tooz-tox-py313-zookeeper https://zuul.opendev.org/t/openstack/build/a6ed0ce65e9e4f3694eea6de017d407e : FAILURE in 3m 05s\n- tooz-tox-py310-redis https://zuul.opendev.org/t/openstack/build/b0218c267853498f958cb5ebe739a16e : SUCCESS in 6m 36s\n- tooz-tox-py313-redis https://zuul.opendev.org/t/openstack/build/4649b4c97cb2487688551e5c329ae953 : SUCCESS in 9m 03s\n- tooz-tox-py310-sentinel https://zuul.opendev.org/t/openstack/build/9bc0d56133814b56a8c332ed992f37fb : SUCCESS in 5m 43s\n- tooz-tox-py313-sentinel https://zuul.opendev.org/t/openstack/build/e7cd595ca42143e98925ae97b7453845 : SUCCESS in 5m 39s\n- tooz-tox-py310-memcached https://zuul.opendev.org/t/openstack/build/0071414ea6584e7faf4fe604bf756205 : SUCCESS in 5m 28s\n- tooz-tox-py313-memcached https://zuul.opendev.org/t/openstack/build/28bfe490132f47bdbb95768932970fb0 : SUCCESS in 6m 28s\n- tooz-tox-py310-postgresql https://zuul.opendev.org/t/openstack/build/6644c48c22154d489daf047e359ff03c : SUCCESS in 4m 20s\n- tooz-tox-py313-postgresql https://zuul.opendev.org/t/openstack/build/bee434d8ca3b4258bed24e29e93df5a9 : SUCCESS in 5m 00s\n- tooz-tox-py310-mysql https://zuul.opendev.org/t/openstack/build/cc48e7c7f79c46a1bd8d6f50f087004b : SUCCESS in 3m 37s\n- tooz-tox-py313-mysql https://zuul.opendev.org/t/openstack/build/6b6caf3161b640a8bf61cdc4a04a6875 : SUCCESS in 7m 04s\n- tooz-tox-py310-consul https://zuul.opendev.org/t/openstack/build/6c3102a8f78d4dd19c155f67eab78f1c : SUCCESS in 3m 52s\n- tooz-tox-py313-consul https://zuul.opendev.org/t/openstack/build/20458bdd6ed14d149b65788937a004e1 : SUCCESS in 4m 01s","accounts_in_message":[],"_revision_number":3},{"id":"927838064db7d8fb2b23ccdd7281ff3a368ea751","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-03-18 17:42:19.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"1bfc94ce099358cfb19212a21eca65582faf0cfd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-18 19:38:48.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1d3c1663fbf249d08a897477460aef19\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0a224453b2af4cc38bce759136af4b2a : SUCCESS in 1h 50m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1b5f9722e1e241c192f065974aaf468f : SUCCESS in 5m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8bcb9631c6184ddeabb01311cd065e69 : SUCCESS in 5m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1d66a13b6def4a4b8d5a0bb0587e1509 : SUCCESS in 4m 18s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/82ec15a4bf244cefb7a7dfad812da903 : SUCCESS in 6m 58s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/20d83bc33f5c4514b21fbcbdeb465ed5 : SUCCESS in 4m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc6628f3d22347df9cb12434c032bd88 : SUCCESS in 5m 06s\n- tooz-tox-examples https://zuul.opendev.org/t/openstack/build/3a9d97f8fe414779b60eb8ba3e9b60bc : SUCCESS in 3m 18s\n- tooz-tox-py310-etcd3gw https://zuul.opendev.org/t/openstack/build/6426ffb10dab47368e6119573e9f56de : SUCCESS in 4m 33s\n- tooz-tox-py313-etcd3gw https://zuul.opendev.org/t/openstack/build/14a6bb1a18c2438ca39e607dc736a1e9 : SUCCESS in 4m 00s\n- tooz-tox-py310-zookeeper https://zuul.opendev.org/t/openstack/build/5bb6d84233aa45008ed1657df39f8579 : SUCCESS in 4m 23s\n- tooz-tox-py313-zookeeper https://zuul.opendev.org/t/openstack/build/575445b1be7b4501ade4582e826e3d0c : SUCCESS in 3m 05s\n- tooz-tox-py310-redis https://zuul.opendev.org/t/openstack/build/899400224c5b454c858c05ca1ca05a69 : SUCCESS in 5m 50s\n- tooz-tox-py313-redis https://zuul.opendev.org/t/openstack/build/981aba2108d1430080d4508f73e0a39b : SUCCESS in 6m 36s\n- tooz-tox-py310-sentinel https://zuul.opendev.org/t/openstack/build/bb3bca79963747b79e96efa8fe34f6c8 : SUCCESS in 6m 04s\n- tooz-tox-py313-sentinel https://zuul.opendev.org/t/openstack/build/b8eea026b98f4bc5b57cf9ae49112f2d : SUCCESS in 8m 16s\n- tooz-tox-py310-memcached https://zuul.opendev.org/t/openstack/build/cee1fa2351074a5eb9b322092ba3aa54 : SUCCESS in 2m 57s\n- tooz-tox-py313-memcached https://zuul.opendev.org/t/openstack/build/93b0abeb0e304ce28acb9825f9c91e41 : SUCCESS in 3m 42s\n- tooz-tox-py310-postgresql https://zuul.opendev.org/t/openstack/build/e2347dd39e694ef89df019717a9c409a : SUCCESS in 4m 28s\n- tooz-tox-py313-postgresql https://zuul.opendev.org/t/openstack/build/dc59e9b8b342479595b7d3c356027f06 : SUCCESS in 2m 49s\n- tooz-tox-py310-mysql https://zuul.opendev.org/t/openstack/build/f52fcc43a74b42a49f85c303a3b3d705 : SUCCESS in 4m 41s\n- tooz-tox-py313-mysql https://zuul.opendev.org/t/openstack/build/8ce4ca196fc04385b40c00365555b93b : SUCCESS in 7m 36s\n- tooz-tox-py310-consul https://zuul.opendev.org/t/openstack/build/00b0c87e5e054555910100faea346fc0 : SUCCESS in 3m 16s\n- tooz-tox-py313-consul https://zuul.opendev.org/t/openstack/build/a3ff86e0253c43cba29725bfb24222be : SUCCESS in 4m 03s","accounts_in_message":[],"_revision_number":3},{"id":"5333628b76db6398a72e26a66868a5934eb0f958","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-18 19:39:06.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"f7d6edcaa5e43aad78a7982b46919f718c82770a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-18 20:32:14.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/42611c26ba814f9199685875aaac3984\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/14f5341bcebf47ffab46976aaf13db03 : SUCCESS in 52m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3271e822b8b34edca59f922ba246b2d6 : SUCCESS in 5m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/828144406961475daa031f1a01b3ad18 : SUCCESS in 3m 51s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/21243d92cb8847e9b67e2e34405db9d3 : SUCCESS in 5m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ed83aca978b47b1aab9ef82cf9d038f : SUCCESS in 4m 38s\n- tooz-tox-examples https://zuul.opendev.org/t/openstack/build/7608953acaec4ab39c810510eabbc145 : SUCCESS in 4m 54s\n- tooz-tox-py310-etcd3gw https://zuul.opendev.org/t/openstack/build/1a49f201cb02462789b12b7461605dc6 : SUCCESS in 6m 08s\n- tooz-tox-py313-etcd3gw https://zuul.opendev.org/t/openstack/build/9c6180037643429893ad43edd5324f9b : SUCCESS in 4m 50s\n- tooz-tox-py310-zookeeper https://zuul.opendev.org/t/openstack/build/88d235c2d4b14a1080b0e0853be49d0f : SUCCESS in 4m 12s\n- tooz-tox-py313-zookeeper https://zuul.opendev.org/t/openstack/build/ab4f0c7dc65d40849f5b592559dc06a7 : SUCCESS in 6m 38s\n- tooz-tox-py310-redis https://zuul.opendev.org/t/openstack/build/3689cc2e48ba4eeaaabd4afa7e21a25c : SUCCESS in 4m 52s\n- tooz-tox-py313-redis https://zuul.opendev.org/t/openstack/build/54ebad1d7a534489966e87ca91241dde : SUCCESS in 6m 39s\n- tooz-tox-py310-sentinel https://zuul.opendev.org/t/openstack/build/fe182c815cbd457894495cd63ae15a01 : SUCCESS in 6m 30s\n- tooz-tox-py313-sentinel https://zuul.opendev.org/t/openstack/build/697a0def04ca4e58a069821cc1588be1 : SUCCESS in 5m 38s\n- tooz-tox-py310-memcached https://zuul.opendev.org/t/openstack/build/a57c2caa69784c1ca764427a8c733ace : SUCCESS in 3m 48s\n- tooz-tox-py313-memcached https://zuul.opendev.org/t/openstack/build/e7d6a95bae4d494f8cac4cab0f60cf8e : SUCCESS in 6m 28s\n- tooz-tox-py310-postgresql https://zuul.opendev.org/t/openstack/build/5dcc4dca416e4de69bc5359787348759 : SUCCESS in 3m 52s\n- tooz-tox-py313-postgresql https://zuul.opendev.org/t/openstack/build/4fb08bdd83ec434790b14960cf28cf27 : SUCCESS in 5m 02s\n- tooz-tox-py310-mysql https://zuul.opendev.org/t/openstack/build/01c1f1ba634c4eee877e82d2b62bd769 : SUCCESS in 5m 37s\n- tooz-tox-py313-mysql https://zuul.opendev.org/t/openstack/build/d30a58ec1c9d471bbd993c7f5c39583a : SUCCESS in 6m 30s\n- tooz-tox-py310-consul https://zuul.opendev.org/t/openstack/build/cd6b12f2c94748389807b58669d8473d : SUCCESS in 4m 28s\n- tooz-tox-py313-consul https://zuul.opendev.org/t/openstack/build/e6d98b843364453290bcea8a056c1a4e : SUCCESS in 4m 37s","accounts_in_message":[],"_revision_number":3},{"id":"8f36d6d46dc8d1861e9a5cb5cb338b75324d70a8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-18 20:32:15.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"d90029ac48e9ed16b07336622442a00147f2584f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-18 20:33:00.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1a7eb5c6af6c4b5881a8364967ec62e1\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b939a9956e8d46fe82ab14641575e18c : SUCCESS in 35s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"29be6b7ea0d98185e3cebe276579e2658ed2daf5","revisions":{"33bc28622f5232ad2719b08298a0d7fb55c1133e":{"kind":"REWORK","_number":1,"created":"2026-03-14 14:26:12.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/58/980558/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/58/980558/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/58/980558/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/58/980558/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/58/980558/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/58/980558/1"}}},"commit":{"parents":[{"commit":"f8b1767103f097793ee1d196446177379884351e","subject":"tests: Prove behavior will str member_id/group_id","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/f8b1767103f097793ee1d196446177379884351e"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-03-14 14:12:06.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-03-14 14:22:58.000000000","tz":0},"subject":"tests: Add test for str IDs","message":"tests: Add test for str IDs\n\nHelp us determine runtime behavior for different backends when users\npass str IDs rather than byte IDs we expect.\n\nChange-Id: Ic13ff11143ff78cce4def53a23d3538ac02aed45\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/33bc28622f5232ad2719b08298a0d7fb55c1133e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/33bc28622f5232ad2719b08298a0d7fb55c1133e"}]},"branch":"refs/heads/master"},"1028babbf69445a537edd5f39c7ac78736c047e9":{"kind":"NO_CHANGE","_number":2,"created":"2026-03-14 14:36:54.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/58/980558/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/58/980558/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/58/980558/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/58/980558/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/58/980558/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/58/980558/2"}}},"commit":{"parents":[{"commit":"64f85470581b900b229a58ce4d41f8a432d9bfdf","subject":"tests: Validate behavior of file driver with str IDs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/64f85470581b900b229a58ce4d41f8a432d9bfdf"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-03-14 14:12:06.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-03-14 14:36:54.000000000","tz":0},"subject":"tests: Add test for str IDs","message":"tests: Add test for str IDs\n\nHelp us determine runtime behavior for different backends when users\npass str IDs rather than byte IDs we expect.\n\nChange-Id: Ic13ff11143ff78cce4def53a23d3538ac02aed45\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/1028babbf69445a537edd5f39c7ac78736c047e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/1028babbf69445a537edd5f39c7ac78736c047e9"}]},"branch":"refs/heads/master","description":"Rebase"},"29be6b7ea0d98185e3cebe276579e2658ed2daf5":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2026-03-14 14:37:03.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/58/980558/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/58/980558/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/58/980558/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/58/980558/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/58/980558/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/58/980558/3"}}},"commit":{"parents":[{"commit":"64f85470581b900b229a58ce4d41f8a432d9bfdf","subject":"tests: Validate behavior of file driver with str IDs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/64f85470581b900b229a58ce4d41f8a432d9bfdf"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-03-14 14:12:06.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-03-14 14:37:03.000000000","tz":0},"subject":"tests: Add test for str IDs","message":"tests: Add test for str IDs\n\nHelp us determine runtime behavior for different backends when users\npass str IDs rather than the byte IDs we expect.\n\nChange-Id: Ic13ff11143ff78cce4def53a23d3538ac02aed45\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/29be6b7ea0d98185e3cebe276579e2658ed2daf5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/29be6b7ea0d98185e3cebe276579e2658ed2daf5"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
