)]}'
{"id":"openstack%2Ftooz~980559","triplet_id":"openstack%2Ftooz~master~I435ee9571d782b28c1a11d7ff47d113bad657c1a","project":"openstack/tooz","branch":"master","topic":"typing","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2026-03-18 12:02:23.000000000","reason":"\u003cGERRIT_ACCOUNT_15334\u003e replied on the change","reason_account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}}},"hashtags":[],"change_id":"I435ee9571d782b28c1a11d7ff47d113bad657c1a","subject":"typing: Add missing dependency","status":"MERGED","created":"2026-03-14 14:40:20.000000000","updated":"2026-03-18 16:08:03.000000000","submitted":"2026-03-18 16:06:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"980559-typing","meta_rev_id":"b2451dc4d796a6152f81a9e5049f5ec381611ce6","_number":980559,"virtual_id_number":980559,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-03-18 16:06:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2026-03-18 12:02:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2026-03-18 12:02:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-14 16:10:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c7aecdfe1d84dea6c7645fe2c342dfd47efd8dd1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-03-14 14:40:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"35aa6b8e86c0a251a57aaf3c23ba70248be7ece8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-14 16:10:47.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1973fd1373894862be27ac132230caa3\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0ccb18b48061436ab376656c557b2c61 : SUCCESS in 1h 26m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4ad7d25b335d4d84822f26d00b2fa12b : SUCCESS in 5m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9bc8ff29ea794f59970a181d3164d3d0 : SUCCESS in 3m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cdc5123246374212a0e68a2239427605 : SUCCESS in 4m 12s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/08afb0ee571848748ce5fb56e7bc2c0c : SUCCESS in 5m 26s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/5202e78f4987483cb496a1b3134bfbae : SUCCESS in 5m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/167aedd4c38e4f26b9857322e08940a1 : SUCCESS in 3m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c86841529fb14cfabe6d41f17d88e5f2 : SUCCESS in 2m 39s\n- tooz-tox-examples https://zuul.opendev.org/t/openstack/build/029b4512e3874f2d9ecc9c674e805e81 : SUCCESS in 5m 51s\n- tooz-tox-py310-etcd3gw https://zuul.opendev.org/t/openstack/build/ec37ad278758485ab35d2d12cead8cf3 : SUCCESS in 4m 51s\n- tooz-tox-py313-etcd3gw https://zuul.opendev.org/t/openstack/build/22679330289e4a23beb07e62223a0246 : SUCCESS in 5m 14s\n- tooz-tox-py310-zookeeper https://zuul.opendev.org/t/openstack/build/e17031dd5a8f4ae2a6e27c92c5376e69 : SUCCESS in 3m 44s\n- tooz-tox-py313-zookeeper https://zuul.opendev.org/t/openstack/build/86889c3d694844dcb548da1088a67bad : FAILURE in 2m 57s\n- tooz-tox-py310-redis https://zuul.opendev.org/t/openstack/build/14b4f9deb4a04ae8b6c118360cb74f1d : SUCCESS in 6m 26s\n- tooz-tox-py313-redis https://zuul.opendev.org/t/openstack/build/c10bbb6895cf4b238123b7824b4f73b2 : SUCCESS in 8m 26s\n- tooz-tox-py310-sentinel https://zuul.opendev.org/t/openstack/build/9368e2210fd84036a1a742466bd20963 : SUCCESS in 4m 46s\n- tooz-tox-py313-sentinel https://zuul.opendev.org/t/openstack/build/cb3e64feddc54d129801bef17ae79549 : SUCCESS in 6m 32s\n- tooz-tox-py310-memcached https://zuul.opendev.org/t/openstack/build/7b2ef8849ac54b27bc250e5dc47d4f8b : SUCCESS in 3m 43s\n- tooz-tox-py313-memcached https://zuul.opendev.org/t/openstack/build/439c2725918a423480b7a7e142b2b706 : SUCCESS in 5m 05s\n- tooz-tox-py310-postgresql https://zuul.opendev.org/t/openstack/build/c5bf791fa2474b2695bf3c1542881d01 : SUCCESS in 4m 30s\n- tooz-tox-py313-postgresql https://zuul.opendev.org/t/openstack/build/a10833b9f5164c9c94be6ecc9b537c09 : SUCCESS in 4m 50s\n- tooz-tox-py310-mysql https://zuul.opendev.org/t/openstack/build/000afa90e1c140ecaaaf6511cf9d46b6 : SUCCESS in 4m 05s\n- tooz-tox-py313-mysql https://zuul.opendev.org/t/openstack/build/f67ba88dac504b8d8d467e51e90e21c6 : SUCCESS in 7m 21s\n- tooz-tox-py310-consul https://zuul.opendev.org/t/openstack/build/434226032fa64a4896c80d7f197626da : SUCCESS in 4m 31s\n- tooz-tox-py313-consul https://zuul.opendev.org/t/openstack/build/119aab4bae66462f9cace98dfce94c94 : SUCCESS in 5m 08s","accounts_in_message":[],"_revision_number":1},{"id":"ad6ccaed486710ba7181c061892ca9649be34aba","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-03-18 12:02:23.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d9dae08f44f0e3565587f53c164d183f748bedca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-18 14:00:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/09334418a4b54b7b840875418121acec\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/17f0e2e6b3c948a89911970aa9bcdad4 : SUCCESS in 1h 52m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a24628228ef74891b6d8cccfcd307cc6 : SUCCESS in 7m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ad8df1596564dce8dee7aab78a27673 : SUCCESS in 5m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b67b61e2562c4418b6e42e49ecc15b7f : SUCCESS in 4m 45s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8db8698e2d67461aa7a87874f430637f : SUCCESS in 6m 33s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/9a6a9ef624554b6294203a96d538509f : SUCCESS in 10m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5936b9586c2244938879e846661e457f : SUCCESS in 2m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b9858cbbcdc34199b98999d0133519b3 : SUCCESS in 2m 51s\n- tooz-tox-examples https://zuul.opendev.org/t/openstack/build/87ffaec684de47c2a2ba9111581d219e : SUCCESS in 4m 59s\n- tooz-tox-py310-etcd3gw https://zuul.opendev.org/t/openstack/build/f8444a708a6c415eaa9e57e24fb6105b : SUCCESS in 4m 59s\n- tooz-tox-py313-etcd3gw https://zuul.opendev.org/t/openstack/build/ae9ce96d4eef494b96e75fe5d4290dc3 : SUCCESS in 6m 20s\n- tooz-tox-py310-zookeeper https://zuul.opendev.org/t/openstack/build/9d0728c3664142aeb4773b543c552ad2 : SUCCESS in 4m 10s\n- tooz-tox-py313-zookeeper https://zuul.opendev.org/t/openstack/build/641d55045b9b43bda605ca6f47c2d7c6 : SUCCESS in 5m 13s\n- tooz-tox-py310-redis https://zuul.opendev.org/t/openstack/build/3f426144203045e7b9c9f2044b84fb55 : SUCCESS in 7m 11s\n- tooz-tox-py313-redis https://zuul.opendev.org/t/openstack/build/6dd8727615d94f05ad03857f6ca8e305 : SUCCESS in 5m 14s\n- tooz-tox-py310-sentinel https://zuul.opendev.org/t/openstack/build/ac6d90fd04fb469c9f9c98b831a1a11f : SUCCESS in 6m 13s\n- tooz-tox-py313-sentinel https://zuul.opendev.org/t/openstack/build/f2bbbb5bf86444b180cddddb51b8198c : SUCCESS in 6m 47s\n- tooz-tox-py310-memcached https://zuul.opendev.org/t/openstack/build/d022bfa7caa743b3ba0fea297faca0bf : SUCCESS in 3m 16s\n- tooz-tox-py313-memcached https://zuul.opendev.org/t/openstack/build/b4ce28a31cc746dbb96cc22f16e4418a : SUCCESS in 7m 08s\n- tooz-tox-py310-postgresql https://zuul.opendev.org/t/openstack/build/60bbcc61433244f9a8b3c0d458353fca : SUCCESS in 4m 39s\n- tooz-tox-py313-postgresql https://zuul.opendev.org/t/openstack/build/632ece5a1dcc49c5874bfc13c6f5fcc7 : SUCCESS in 5m 09s\n- tooz-tox-py310-mysql https://zuul.opendev.org/t/openstack/build/90eecbf8df3045c4b8db47b8c90f0d0c : SUCCESS in 6m 05s\n- tooz-tox-py313-mysql https://zuul.opendev.org/t/openstack/build/26765e4f93e542b5b3e360555731d68d : SUCCESS in 7m 43s\n- tooz-tox-py310-consul https://zuul.opendev.org/t/openstack/build/1c19f5b0d5fc473499f62b62e20759c7 : SUCCESS in 4m 37s\n- tooz-tox-py313-consul https://zuul.opendev.org/t/openstack/build/27e35a9d41de4b1e89fcebc011ca3119 : SUCCESS in 5m 23s","accounts_in_message":[],"_revision_number":1},{"id":"815ec12d80e5a13d3f3b399d0f925823091183cc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-18 14:01:03.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f89a29b85762ed4cfab18db906a58600508e324a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-18 16:06:41.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8107dbfa7dc8424ea1f62f54bf92e6f1\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b57cb38e2eb247968819b521dc658db0 : SUCCESS in 1h 51m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/534b268145c2446aad713df8b68826f1 : SUCCESS in 6m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/57b5dcd0b7ea4fa9aefb7ae3b6d58463 : SUCCESS in 4m 19s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/044283b43b7c41a4b4e379a4124e4761 : SUCCESS in 3m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1bcf893264a144d0bbcf02ece65d61e9 : SUCCESS in 5m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/76bf62cb4ec444a986c2da30bf4dd7f1 : SUCCESS in 3m 56s\n- tooz-tox-examples https://zuul.opendev.org/t/openstack/build/271c187ebada46bdb8d767891807884f : SUCCESS in 5m 25s\n- tooz-tox-py310-etcd3gw https://zuul.opendev.org/t/openstack/build/813b6953af4a45698d46a8768c5c0cae : SUCCESS in 6m 17s\n- tooz-tox-py313-etcd3gw https://zuul.opendev.org/t/openstack/build/066a3e4a421f4f91a6b63bfe55eb0dc9 : SUCCESS in 7m 05s\n- tooz-tox-py310-zookeeper https://zuul.opendev.org/t/openstack/build/3893a23a5da74fa4b08ccd933f8b4c6d : SUCCESS in 4m 16s\n- tooz-tox-py313-zookeeper https://zuul.opendev.org/t/openstack/build/e930bc960d5a47c3a00db43b0bd77850 : SUCCESS in 5m 33s\n- tooz-tox-py310-redis https://zuul.opendev.org/t/openstack/build/4d61492956ef47d49e25ca6541b62cf0 : SUCCESS in 8m 54s\n- tooz-tox-py313-redis https://zuul.opendev.org/t/openstack/build/f169ca6d6400450ba97b60bf8841a301 : SUCCESS in 6m 33s\n- tooz-tox-py310-sentinel https://zuul.opendev.org/t/openstack/build/5d1d7ce07b3e4d03b7655a90c0da96b1 : SUCCESS in 6m 04s\n- tooz-tox-py313-sentinel https://zuul.opendev.org/t/openstack/build/24a2b6ab1f7042afaf1297004e30c434 : SUCCESS in 7m 02s\n- tooz-tox-py310-memcached https://zuul.opendev.org/t/openstack/build/85df09ba8daf44f2a8c48d3bc8ef8e65 : SUCCESS in 3m 47s\n- tooz-tox-py313-memcached https://zuul.opendev.org/t/openstack/build/ecf542a1e59b412b82871605cf5ae173 : SUCCESS in 3m 59s\n- tooz-tox-py310-postgresql https://zuul.opendev.org/t/openstack/build/b4e131051b4e4cc9b370cf9043627eb9 : SUCCESS in 5m 06s\n- tooz-tox-py313-postgresql https://zuul.opendev.org/t/openstack/build/468a0a62c44f4d598694d21f42c8a4fb : SUCCESS in 4m 53s\n- tooz-tox-py310-mysql https://zuul.opendev.org/t/openstack/build/7a66a365357f4fa9a42ddbbb5e9489dc : SUCCESS in 4m 23s\n- tooz-tox-py313-mysql https://zuul.opendev.org/t/openstack/build/9906c68837994d4fabc485272794c1ba : SUCCESS in 4m 58s\n- tooz-tox-py310-consul https://zuul.opendev.org/t/openstack/build/86a7d758ca6b4c16811f8924c02ffef1 : SUCCESS in 5m 55s\n- tooz-tox-py313-consul https://zuul.opendev.org/t/openstack/build/93854d30cab44ca29aa37044afe57739 : SUCCESS in 6m 03s","accounts_in_message":[],"_revision_number":1},{"id":"f120819b1d13f6475c8a52a8b2c3571ea1e9d57f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-18 16:06:41.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"b2451dc4d796a6152f81a9e5049f5ec381611ce6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-18 16:08:03.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a2ab47a0da2b45ee8c210c453c5a9b73\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/14671585ae0e4ccdbddc8602a87d75f8 : SUCCESS in 1m 02s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/49c1f1cd56534fcfbf8600b3aaf8bea5 : SUCCESS in 40s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d60124f8a08f0084d6b56737b16d3cd07fa4338c","revisions":{"d60124f8a08f0084d6b56737b16d3cd07fa4338c":{"kind":"REWORK","_number":1,"created":"2026-03-14 14:40:20.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/59/980559/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/59/980559/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/59/980559/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/59/980559/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/59/980559/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/59/980559/1"}}},"commit":{"parents":[{"commit":"2c20044bc05e4a543ead196631e0e27c46289c9b","subject":"Merge \"typing: Add argument to tooz.NotImplementedError\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/2c20044bc05e4a543ead196631e0e27c46289c9b"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-03-09 15:25:44.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-03-14 14:40:05.000000000","tz":0},"subject":"typing: Add missing dependency","message":"typing: Add missing dependency\n\nChange-Id: I435ee9571d782b28c1a11d7ff47d113bad657c1a\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/d60124f8a08f0084d6b56737b16d3cd07fa4338c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/d60124f8a08f0084d6b56737b16d3cd07fa4338c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
