)]}'
{"id":"openstack%2Ftooz~980561","triplet_id":"openstack%2Ftooz~master~I38343fddd9d5060534b008ef3822ad0b6ddf55b7","project":"openstack/tooz","branch":"master","topic":"typing","hashtags":[],"change_id":"I38343fddd9d5060534b008ef3822ad0b6ddf55b7","subject":"gitignore: Ignore pifpaf artifacts","status":"MERGED","created":"2026-03-14 14:40:20.000000000","updated":"2026-03-18 16:08:57.000000000","submitted":"2026-03-18 16:06:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"980561-typing","meta_rev_id":"0a8eb65f7a8e178550b7c06d12989dda27e7a27b","_number":980561,"virtual_id_number":980561,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-03-18 16:06:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2026-03-18 12:03:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2026-03-18 12:03:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-14 16:29:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"18dcba53221812175411de7a39d6c72050764907","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-03-14 14:40:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"477f89bc9c09ef583f536dbec4d7e2e4ea13a90c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-14 16:29:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/106b0c784ad84771b6f17a94573d5687\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ba549c2689254c94a9a76473dd5eb8dd : SUCCESS in 1h 38m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d0712ccd353647afb38b412854ff54d2 : SUCCESS in 5m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b3bd4b7d0194265b9e5e070bc1761f5 : SUCCESS in 3m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0ddf05e193ef4d31b222fd0969b41753 : SUCCESS in 6m 02s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/56ac8022b30b473e907f2f29435efbab : SUCCESS in 6m 49s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/b04e24e257c04ed18c1b2edf5e44157c : SUCCESS in 7m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d05fe60132c146dc8f1a06d04e8557bd : SUCCESS in 4m 48s\n- tooz-tox-examples https://zuul.opendev.org/t/openstack/build/b857d057a4754be098bd883d005b2e00 : SUCCESS in 3m 47s\n- tooz-tox-py310-etcd3gw https://zuul.opendev.org/t/openstack/build/3809488930744fd09c4dc33c5c6d9c0e : SUCCESS in 4m 32s\n- tooz-tox-py313-etcd3gw https://zuul.opendev.org/t/openstack/build/3a229f8799684aa8804a9cd751ad9b51 : SUCCESS in 4m 23s\n- tooz-tox-py310-zookeeper https://zuul.opendev.org/t/openstack/build/d87d00cda9cc44a4b20a47abbb502101 : SUCCESS in 5m 02s\n- tooz-tox-py313-zookeeper https://zuul.opendev.org/t/openstack/build/74327b61f6f34681b5c8a5a0f5d0c972 : SUCCESS in 4m 56s\n- tooz-tox-py310-redis https://zuul.opendev.org/t/openstack/build/6f41ba7b3b2047ddb43f8d228c38c07e : SUCCESS in 5m 48s\n- tooz-tox-py313-redis https://zuul.opendev.org/t/openstack/build/8e32b8cfbf90457db9bc2a3ad7bf47e5 : SUCCESS in 7m 01s\n- tooz-tox-py310-sentinel https://zuul.opendev.org/t/openstack/build/9f1124f17e674f1cb794285f04eb171d : SUCCESS in 7m 36s\n- tooz-tox-py313-sentinel https://zuul.opendev.org/t/openstack/build/966511779c3e4f8ca288d644b0f2caf7 : SUCCESS in 6m 07s\n- tooz-tox-py310-memcached https://zuul.opendev.org/t/openstack/build/a6c64cac1a834a0abfd934fd0de60a55 : SUCCESS in 4m 01s\n- tooz-tox-py313-memcached https://zuul.opendev.org/t/openstack/build/463edc5659df4164a8b2d97af645f715 : SUCCESS in 3m 22s\n- tooz-tox-py310-postgresql https://zuul.opendev.org/t/openstack/build/135284bde6ad4fe6a36ebbf99536c987 : SUCCESS in 3m 24s\n- tooz-tox-py313-postgresql https://zuul.opendev.org/t/openstack/build/ad76956548d842798c362f20074815ec : SUCCESS in 3m 16s\n- tooz-tox-py310-mysql https://zuul.opendev.org/t/openstack/build/3c3ca8156798461ba20e4cef1a08161d : SUCCESS in 4m 05s\n- tooz-tox-py313-mysql https://zuul.opendev.org/t/openstack/build/66b1116b61a345189772ebd0e24c426d : SUCCESS in 7m 24s\n- tooz-tox-py310-consul https://zuul.opendev.org/t/openstack/build/a543951740464d729957e9a0d8d1df49 : SUCCESS in 3m 44s\n- tooz-tox-py313-consul https://zuul.opendev.org/t/openstack/build/738846b5c1de4400a84781c5a58af453 : SUCCESS in 4m 01s","accounts_in_message":[],"_revision_number":1},{"id":"ecd269a8be1cb9ff719bbaba47d649960d928358","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-03-18 12:03:12.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"586be032dcd62bcf27e59efd12503edf2614237b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-18 14:01:16.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4f8337d5f806d45552737cd71affa696dd292c80","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-18 16:06:45.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/55e03719fefa43bc87533d15ca840ca9\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c7737886f2d9462dae972398f82343be : SUCCESS in 1h 48m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1158582655c54683b732a45eaad5c808 : SUCCESS in 5m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c6d75b088daa496d9c0c9ad30e8fcdeb : SUCCESS in 5m 44s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/31d46a4ad058435293c3d2886b7330d4 : SUCCESS in 5m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9001ba4ae0b24815b02e0204063907b2 : SUCCESS in 5m 32s\n- tooz-tox-examples https://zuul.opendev.org/t/openstack/build/a2eb3c2356eb4478978193db7d22166a : SUCCESS in 3m 54s\n- tooz-tox-py310-etcd3gw https://zuul.opendev.org/t/openstack/build/4cf9ac9242404f66a4fea520af5f68ec : SUCCESS in 4m 15s\n- tooz-tox-py313-etcd3gw https://zuul.opendev.org/t/openstack/build/dd01ef4a9ee349f69cda2b362d43a5bf : SUCCESS in 7m 19s\n- tooz-tox-py310-zookeeper https://zuul.opendev.org/t/openstack/build/e3ba9b24b88e44a797d4b503df59ec24 : SUCCESS in 6m 14s\n- tooz-tox-py313-zookeeper https://zuul.opendev.org/t/openstack/build/c2448a0eb4fa46b8acc1404b226b2236 : SUCCESS in 5m 26s\n- tooz-tox-py310-redis https://zuul.opendev.org/t/openstack/build/a6b8136ac2c141b9b235856a838e8437 : SUCCESS in 6m 12s\n- tooz-tox-py313-redis https://zuul.opendev.org/t/openstack/build/4c19f4e0dc384654b8e70e56fa261ec6 : SUCCESS in 6m 27s\n- tooz-tox-py310-sentinel https://zuul.opendev.org/t/openstack/build/39f5f01c3425439c8b72d56f9bc4fb55 : SUCCESS in 7m 07s\n- tooz-tox-py313-sentinel https://zuul.opendev.org/t/openstack/build/2476a4acfbba417482c112b911df88d2 : SUCCESS in 5m 41s\n- tooz-tox-py310-memcached https://zuul.opendev.org/t/openstack/build/fc4c9e6d16724ca08140a2675d3db2fb : SUCCESS in 4m 20s\n- tooz-tox-py313-memcached https://zuul.opendev.org/t/openstack/build/1403a7956ee641e7a59a9b357a414ab8 : SUCCESS in 5m 00s\n- tooz-tox-py310-postgresql https://zuul.opendev.org/t/openstack/build/9a08e9c5f41a4257819e44f984c91f54 : SUCCESS in 4m 27s\n- tooz-tox-py313-postgresql https://zuul.opendev.org/t/openstack/build/02591a04b8a547fb9438f88c8250e4dc : SUCCESS in 4m 05s\n- tooz-tox-py310-mysql https://zuul.opendev.org/t/openstack/build/080b6528f12543e88da8b7690cd32f22 : SUCCESS in 4m 53s\n- tooz-tox-py313-mysql https://zuul.opendev.org/t/openstack/build/8171b5a7058448c182f011ff010394ca : SUCCESS in 7m 54s\n- tooz-tox-py310-consul https://zuul.opendev.org/t/openstack/build/0629777695e344d58ea414b8c3130474 : SUCCESS in 4m 59s\n- tooz-tox-py313-consul https://zuul.opendev.org/t/openstack/build/ed394cb593014449940c75fbadbdab85 : SUCCESS in 5m 30s","accounts_in_message":[],"_revision_number":1},{"id":"b4fceeb14f49ddcb444965366f98bd2d5d992e4e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-18 16:06:45.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"0a8eb65f7a8e178550b7c06d12989dda27e7a27b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-18 16:08:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/db4790acb9a74389a43e5f86db4cc41f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a7c1e2559e434aca891aa5aceacde5f5 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7df2cb5248a47d61886b2921fefcdd639e165784","revisions":{"7df2cb5248a47d61886b2921fefcdd639e165784":{"kind":"REWORK","_number":1,"created":"2026-03-14 14:40:20.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/61/980561/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/61/980561/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/61/980561/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/61/980561/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/61/980561/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/61/980561/1"}}},"commit":{"parents":[{"commit":"859343d7c6aca0c9f86dcc7dc1a1b2364b3a73e6","subject":"Decode provided IDs for exceptions, event repr","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/859343d7c6aca0c9f86dcc7dc1a1b2364b3a73e6"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-03-12 21:54:47.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-03-14 14:40:14.000000000","tz":0},"subject":"gitignore: Ignore pifpaf artifacts","message":"gitignore: Ignore pifpaf artifacts\n\nThese are generated by our unit tests.\n\nChange-Id: I38343fddd9d5060534b008ef3822ad0b6ddf55b7\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/7df2cb5248a47d61886b2921fefcdd639e165784"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/7df2cb5248a47d61886b2921fefcdd639e165784"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
