)]}'
{"id":"openstack%2Ftooz~980563","triplet_id":"openstack%2Ftooz~master~I621dc87412d44fd87bb9056a6304fbe71f52ca69","project":"openstack/tooz","branch":"master","topic":"typing","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2026-04-23 14:42:08.000000000","reason":"\u003cGERRIT_ACCOUNT_15334\u003e replied on the change","reason_account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}}},"hashtags":[],"change_id":"I621dc87412d44fd87bb9056a6304fbe71f52ca69","subject":"typing: Add type hints to tooz.driver.etcd3gw","status":"MERGED","created":"2026-03-14 14:40:20.000000000","updated":"2026-04-23 16:51:35.000000000","submitted":"2026-04-23 16:51:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"980563-typing","meta_rev_id":"d733c00d1542ad514abc00ae63d3270aa57b1e3b","_number":980563,"virtual_id_number":980563,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-04-23 16:51:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2026-04-14 09:45:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2026-04-23 14:42:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-14 15:46:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-04-13 08:11:26.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"2420bb31f996ea4b863072ac65a14c5949888e8e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-03-14 14:40:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"94b3bb7cfbb039dd8c6cfcc6d28a3bfcec32ce33","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-14 15:46:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0713c551e36747b695bf4e6cedeec6bb\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/f8eebe9a6f5d4e24ae384d211a89bf32 : SUCCESS in 2m 44s\n- requirements-check https://zuul.opendev.org/t/openstack/build/6cf8882087bf4ac3b38926db2a1dfc49 : SUCCESS in 2m 48s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c74f8afda3964b16b4b3e7ac8f1f0d54 : SUCCESS in 57m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/adef2fbc0698493597219e56019c9798 : SUCCESS in 5m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b5e114bbb40c422b86766e35b2395bc3 : SUCCESS in 5m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c4dd3be2544a4ea182fe401bd088f304 : SUCCESS in 4m 20s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/73c8a115ee2148d98c69521e36280c74 : SUCCESS in 4m 54s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/dd1ab3f1eed64290b365b8918f3d5e17 : SUCCESS in 7m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff553d50e9fc468b9afce9aded76e3c7 : SUCCESS in 5m 12s\n- tooz-tox-examples https://zuul.opendev.org/t/openstack/build/b2201d65bee141cd953d29157bf8011a : SUCCESS in 3m 45s\n- tooz-tox-py310-etcd3gw https://zuul.opendev.org/t/openstack/build/42d17efa3f154be5aeff341416627d3d : SUCCESS in 4m 13s\n- tooz-tox-py313-etcd3gw https://zuul.opendev.org/t/openstack/build/0c1eddd422cb48549f33d75d6f5f2b24 : SUCCESS in 4m 20s\n- tooz-tox-py310-zookeeper https://zuul.opendev.org/t/openstack/build/89ef568beb30403393c88bf891351455 : SUCCESS in 4m 26s\n- tooz-tox-py313-zookeeper https://zuul.opendev.org/t/openstack/build/8030427cfc0c4645bea4119985f6ee11 : SUCCESS in 4m 59s\n- tooz-tox-py310-redis https://zuul.opendev.org/t/openstack/build/dbea499c5344433f8d945675347f06bd : SUCCESS in 5m 05s\n- tooz-tox-py313-redis https://zuul.opendev.org/t/openstack/build/fd8cdf97814f401cb9ce6592f160dbcb : SUCCESS in 5m 15s\n- tooz-tox-py310-sentinel https://zuul.opendev.org/t/openstack/build/93d26d06e749465bbc66faa0103eb963 : SUCCESS in 5m 15s\n- tooz-tox-py313-sentinel https://zuul.opendev.org/t/openstack/build/b3514065ff9b470ba21f95407e62644b : SUCCESS in 6m 12s\n- tooz-tox-py310-memcached https://zuul.opendev.org/t/openstack/build/075312b583b04ddbba0e24450e7355a0 : SUCCESS in 4m 46s\n- tooz-tox-py313-memcached https://zuul.opendev.org/t/openstack/build/8ae07e222f064481a9e2458289152323 : SUCCESS in 4m 15s\n- tooz-tox-py310-postgresql https://zuul.opendev.org/t/openstack/build/4c183f7f6d344eca879a536b04334a4a : SUCCESS in 3m 42s\n- tooz-tox-py313-postgresql https://zuul.opendev.org/t/openstack/build/8641acdafc60472b8418edaae9a2d802 : SUCCESS in 5m 06s\n- tooz-tox-py310-mysql https://zuul.opendev.org/t/openstack/build/1b969b72625d474a98d8e19f3a955b54 : SUCCESS in 4m 14s\n- tooz-tox-py313-mysql https://zuul.opendev.org/t/openstack/build/d73dfb7e10df41eea7d9ad94e5391e60 : SUCCESS in 4m 16s\n- tooz-tox-py310-consul https://zuul.opendev.org/t/openstack/build/1f9cc78b8fdf4286a8db21d6e6f9197b : SUCCESS in 3m 33s\n- tooz-tox-py313-consul https://zuul.opendev.org/t/openstack/build/8f2d80b53a4b42ada45f79f99b01957f : SUCCESS in 4m 06s","accounts_in_message":[],"_revision_number":1},{"id":"7fa102bc91c57160e449582d3ff4917d84a9d07f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-03-19 09:47:24.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"8c9423e2ab9e72352a502ade1aef9eae4339f931","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-19 11:18:47.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a63c3a2d7ed94587b8b5a10188f3a570\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/73469285e19f4da2ab542d0e63efac5b : SUCCESS in 4m 00s\n- requirements-check https://zuul.opendev.org/t/openstack/build/c2dca6221c3b4293974521cb8a5bfcc5 : SUCCESS in 3m 47s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b7549cebfe284243a908b5c45a11f8eb : SUCCESS in 1h 23m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c9635540dabe4437b4297b7588ece9f8 : SUCCESS in 6m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/df2c2ac7e4c0488c921f6235de5605f3 : SUCCESS in 3m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/519fc1ea57b84903bb5dac0b9c3795cc : SUCCESS in 5m 08s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/cf06d51445e24900aed774844778f7c3 : SUCCESS in 6m 12s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/13ce5cddd3ff4634a4ad194a3b97a614 : SUCCESS in 5m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1de9f5702679436fa03b00dde60e08be : SUCCESS in 3m 36s\n- tooz-tox-examples https://zuul.opendev.org/t/openstack/build/fdbae77f5eef4db589aeb3fbc0d62d4a : SUCCESS in 5m 07s\n- tooz-tox-py310-etcd3gw https://zuul.opendev.org/t/openstack/build/722b5446e0484a2e9d0fcd1ff34d3944 : SUCCESS in 4m 36s\n- tooz-tox-py313-etcd3gw https://zuul.opendev.org/t/openstack/build/cb97c5e5debb453eb2748240c48e905a : SUCCESS in 5m 42s\n- tooz-tox-py310-zookeeper https://zuul.opendev.org/t/openstack/build/75472bcafc5a4cb6b7845bd1ee97c8c2 : SUCCESS in 5m 25s\n- tooz-tox-py313-zookeeper https://zuul.opendev.org/t/openstack/build/7388d86272224f088f8175cc465151ce : SUCCESS in 5m 42s\n- tooz-tox-py310-redis https://zuul.opendev.org/t/openstack/build/afe50da241794fe88a822d5e1fe5c768 : SUCCESS in 6m 48s\n- tooz-tox-py313-redis https://zuul.opendev.org/t/openstack/build/90a6be3baea847ce8946d2d4fa32897c : SUCCESS in 6m 10s\n- tooz-tox-py310-sentinel https://zuul.opendev.org/t/openstack/build/269c36fafa674683841491564dde69cb : SUCCESS in 7m 43s\n- tooz-tox-py313-sentinel https://zuul.opendev.org/t/openstack/build/838a852c2f9b4f4b88277480bb445588 : SUCCESS in 7m 21s\n- tooz-tox-py310-memcached https://zuul.opendev.org/t/openstack/build/57442b8bef8f42d8a0579dec8a489abe : SUCCESS in 4m 18s\n- tooz-tox-py313-memcached https://zuul.opendev.org/t/openstack/build/b5634e01e09543acb4950e5df8e8df95 : SUCCESS in 3m 56s\n- tooz-tox-py310-postgresql https://zuul.opendev.org/t/openstack/build/e877270db42247b1a128d9a2552ed035 : SUCCESS in 3m 25s\n- tooz-tox-py313-postgresql https://zuul.opendev.org/t/openstack/build/161c0f527c4a450fb517760bf1d8878f : SUCCESS in 6m 45s\n- tooz-tox-py310-mysql https://zuul.opendev.org/t/openstack/build/f41702e62b2946a991c0384cab745f86 : SUCCESS in 4m 26s\n- tooz-tox-py313-mysql https://zuul.opendev.org/t/openstack/build/1480196259544d51aeb5924b132529bf : SUCCESS in 7m 10s\n- tooz-tox-py310-consul https://zuul.opendev.org/t/openstack/build/fa0ba95160c048d9b0fe7d877169e711 : SUCCESS in 4m 13s\n- tooz-tox-py313-consul https://zuul.opendev.org/t/openstack/build/f5b6b2b6c1144950b12d92914555b3a8 : SUCCESS in 5m 09s","accounts_in_message":[],"_revision_number":2},{"id":"f10ae88d4456b96caf881d9186e9d85565e0bbaf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-03-31 09:03:29.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"0dbd4212159272cf9b106cd4d839b3f1ed74ec72","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-31 10:55:37.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1f3d68db2c9a4a3b8bafe6cf0e22b1e2\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/358bfc70f60345e884ece6ee3c6b02dd : SUCCESS in 3m 09s\n- requirements-check https://zuul.opendev.org/t/openstack/build/d50541e48d064bde8bc26c1bde79d415 : SUCCESS in 3m 58s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a680f39a2b7240d090f879d328945b39 : SUCCESS in 1h 35m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ee1069f19ab041828a1f58fda0049e4c : SUCCESS in 6m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/21eb4bf86ce1486c8373975d9be041bf : SUCCESS in 3m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c7fa31e0924c4265a1a907f45b4a5fb5 : SUCCESS in 3m 33s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d3a9e44cef034c1a821953976fd037f7 : SUCCESS in 3m 59s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/697f7156a5d241b4a6d3b8e23dba2516 : FAILURE in 6m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/afa2c13c85524f20b44706de5a1d6473 : SUCCESS in 5m 27s\n- tooz-tox-examples https://zuul.opendev.org/t/openstack/build/06535f5966c341ef8b903f14507debd9 : SUCCESS in 4m 35s\n- tooz-tox-py310-etcd3gw https://zuul.opendev.org/t/openstack/build/db2db1c00d104daebaff24793abb0e0c : SUCCESS in 4m 37s\n- tooz-tox-py313-etcd3gw https://zuul.opendev.org/t/openstack/build/edbafe6138174d0285848b4be4ee27a5 : SUCCESS in 6m 27s\n- tooz-tox-py310-zookeeper https://zuul.opendev.org/t/openstack/build/f774a602233148448749229c834d422f : SUCCESS in 4m 55s\n- tooz-tox-py313-zookeeper https://zuul.opendev.org/t/openstack/build/d8d66b6ed312443fb8a202ea455e135d : SUCCESS in 5m 12s\n- tooz-tox-py310-redis https://zuul.opendev.org/t/openstack/build/ff0d76837cc1498baccba4d82452c066 : SUCCESS in 6m 52s\n- tooz-tox-py313-redis https://zuul.opendev.org/t/openstack/build/e12fd45983e44515ab4d2f587f80b30a : SUCCESS in 5m 13s\n- tooz-tox-py310-sentinel https://zuul.opendev.org/t/openstack/build/d64bbef7bc9a4784969d46e9abef0814 : SUCCESS in 6m 51s\n- tooz-tox-py313-sentinel https://zuul.opendev.org/t/openstack/build/5044f27865bb4a5eb35a23568d688a86 : SUCCESS in 7m 42s\n- tooz-tox-py310-memcached https://zuul.opendev.org/t/openstack/build/4cc99f1536a54fa6b9b2e7b19bc2f98f : SUCCESS in 4m 32s\n- tooz-tox-py313-memcached https://zuul.opendev.org/t/openstack/build/7a2a306c826d436c81cb2d19b4d627ce : SUCCESS in 3m 34s\n- tooz-tox-py310-postgresql https://zuul.opendev.org/t/openstack/build/b3fd4fbdc1d24660bf7c213548e7900e : SUCCESS in 4m 51s\n- tooz-tox-py313-postgresql https://zuul.opendev.org/t/openstack/build/9de57815337b4764920699b2a0d4d2f2 : SUCCESS in 6m 58s\n- tooz-tox-py310-mysql https://zuul.opendev.org/t/openstack/build/541ed073f93b4900b23c4bc1c1d0c0c3 : SUCCESS in 4m 41s\n- tooz-tox-py313-mysql https://zuul.opendev.org/t/openstack/build/050f22dfbd1c4d7fa3f38c79b9082337 : SUCCESS in 7m 58s\n- tooz-tox-py310-consul https://zuul.opendev.org/t/openstack/build/58c39a72de4b43c6a91e2aa64d4cc37f : SUCCESS in 3m 39s\n- tooz-tox-py313-consul https://zuul.opendev.org/t/openstack/build/c5bb00fd913a46c0b07cc0b9a0b6f222 : SUCCESS in 5m 21s","accounts_in_message":[],"_revision_number":3},{"id":"2b408ea515333c610c1290acefd4546c6b9f3cae","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-04-10 18:13:45.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"f03a89da597f6bb6e4e42555857f976f1b564f6a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-10 19:20:36.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ea3973f638c445e483a782911e22c9ba\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/ceede6156a124902b780a0405eeb5cff : SUCCESS in 2m 51s\n- requirements-check https://zuul.opendev.org/t/openstack/build/3b6ba28658814ac19a3a77310b676d21 : SUCCESS in 3m 30s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d8fe7726c14b45da8706bc5335f849c1 : SUCCESS in 1h 00m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a11eebbafb8045b5a255f98315bd9f6e : SUCCESS in 6m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/36eb0922472b4dcb948d93e119adf219 : SUCCESS in 4m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/90d22ba310b44db7b268bf3f1f29f16d : SUCCESS in 5m 05s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e585ccd464534fe2bcbd5be0216ad71a : SUCCESS in 5m 38s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/e5dc18f5927848febce3ccdf32e0142e : SUCCESS in 7m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/09e127be518649829b06bb35bccecb7d : SUCCESS in 5m 53s\n- tooz-tox-examples https://zuul.opendev.org/t/openstack/build/f1c38d1fea044a3ab3bbcc17b330679d : SUCCESS in 5m 07s\n- tooz-tox-py310-etcd3gw https://zuul.opendev.org/t/openstack/build/2085d4294bc645d8af957eefbd0f6c75 : FAILURE in 4m 53s\n- tooz-tox-py313-etcd3gw https://zuul.opendev.org/t/openstack/build/1ffb053379d345a79500ac25071a30a6 : FAILURE in 6m 36s\n- tooz-tox-py310-zookeeper https://zuul.opendev.org/t/openstack/build/ff8632c1451542e7b62753be8515222b : SUCCESS in 5m 59s\n- tooz-tox-py313-zookeeper https://zuul.opendev.org/t/openstack/build/c293bb37931f46c78594628cddd3e741 : SUCCESS in 3m 37s\n- tooz-tox-py310-redis https://zuul.opendev.org/t/openstack/build/232c749e05294b8482d274ad47990dd2 : SUCCESS in 7m 56s\n- tooz-tox-py313-redis https://zuul.opendev.org/t/openstack/build/2307b72d9351431a865c661257faee5f : SUCCESS in 6m 58s\n- tooz-tox-py310-sentinel https://zuul.opendev.org/t/openstack/build/769192a28dc84cb9b796915cc112fbae : SUCCESS in 7m 11s\n- tooz-tox-py313-sentinel https://zuul.opendev.org/t/openstack/build/68b3305205ee46888a7e2e3e6496ac1a : SUCCESS in 7m 25s\n- tooz-tox-py310-memcached https://zuul.opendev.org/t/openstack/build/b5f367bd2b324d2abf04773c22913fd8 : SUCCESS in 5m 01s\n- tooz-tox-py313-memcached https://zuul.opendev.org/t/openstack/build/f641cb2c0fb64e2f840f6ae0db262ae8 : SUCCESS in 5m 54s\n- tooz-tox-py310-postgresql https://zuul.opendev.org/t/openstack/build/482b3c660a0f4321b3a4da74c8aa0d81 : SUCCESS in 3m 36s\n- tooz-tox-py313-postgresql https://zuul.opendev.org/t/openstack/build/cd83cbe188e64a8fb747d1cac097621c : SUCCESS in 3m 21s\n- tooz-tox-py310-mysql https://zuul.opendev.org/t/openstack/build/e302eea386a9458db0dce6768bda754f : SUCCESS in 4m 57s\n- tooz-tox-py313-mysql https://zuul.opendev.org/t/openstack/build/712c91472a4f4123b0b0cea8f7774fb2 : SUCCESS in 8m 42s\n- tooz-tox-py310-consul https://zuul.opendev.org/t/openstack/build/0005679be1c34eada770bb47aa00d308 : SUCCESS in 4m 44s\n- tooz-tox-py313-consul https://zuul.opendev.org/t/openstack/build/eb0de06ed3884bf9867057f27ccd89ab : SUCCESS in 4m 56s","accounts_in_message":[],"_revision_number":4},{"id":"214f1b55896c4d50f606f56139e0bf7003136e10","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-04-10 22:50:18.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"10ce8c4ced0f587ee41cdb16a60ee19b0dc70591","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-04-10 22:56:02.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"39c3fa199d00d1478072497ee1230b99e3561363","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-11 00:14:20.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6c351d83331840b5b98a358a0331d7e0\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/b099634af083499aa220e0afd40938ba : SUCCESS in 3m 06s\n- requirements-check https://zuul.opendev.org/t/openstack/build/b3f80335866c4f349d5d3c61bb9cb163 : SUCCESS in 3m 27s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b236dcbd893e4c6d8e45ec76ddaca9d0 : SUCCESS in 1h 17m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/985c95a0645548bf9584f77d4729d84d : SUCCESS in 4m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3bc4be6eef5f46bf8f5c133fbfbc6af6 : SUCCESS in 4m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3a90cb87a9044b8e9dd2de6256055547 : SUCCESS in 4m 41s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6e4899cfb75d450ebef0eb93affc21a7 : SUCCESS in 4m 02s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/0834075f6292484986e19975cf3f7dc7 : FAILURE in 6m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/45d326df5eaa4df6b9f0a5c731275889 : SUCCESS in 6m 01s\n- tooz-tox-examples https://zuul.opendev.org/t/openstack/build/b0146853552a45428989816a9142706a : SUCCESS in 3m 22s\n- tooz-tox-py310-etcd3gw https://zuul.opendev.org/t/openstack/build/96ac22b2a1c1470fafb7af5e8a81eda5 : SUCCESS in 4m 45s\n- tooz-tox-py313-etcd3gw https://zuul.opendev.org/t/openstack/build/ec5825c7389e4cf5b681b7305a768329 : SUCCESS in 5m 16s\n- tooz-tox-py310-zookeeper https://zuul.opendev.org/t/openstack/build/601843d9c0464eab81d76ce963758238 : SUCCESS in 3m 56s\n- tooz-tox-py313-zookeeper https://zuul.opendev.org/t/openstack/build/7e8e53c7a93a47dbb583d6332c96b4a4 : SUCCESS in 4m 41s\n- tooz-tox-py310-redis https://zuul.opendev.org/t/openstack/build/a751a702aaa142e8a821c886a794cf13 : SUCCESS in 6m 24s\n- tooz-tox-py313-redis https://zuul.opendev.org/t/openstack/build/960cc388694f46ebb94c34f559355e6c : SUCCESS in 7m 42s\n- tooz-tox-py310-sentinel https://zuul.opendev.org/t/openstack/build/2956e8ebe41446debef88f8af8f5637a : SUCCESS in 6m 24s\n- tooz-tox-py313-sentinel https://zuul.opendev.org/t/openstack/build/e7c6b4df896e4f01a30a1a2e1b6f5258 : SUCCESS in 6m 20s\n- tooz-tox-py310-memcached https://zuul.opendev.org/t/openstack/build/efa85ebff4904648a59e6f2224eca707 : SUCCESS in 4m 03s\n- tooz-tox-py313-memcached https://zuul.opendev.org/t/openstack/build/9c228d9781714a528c1cf4215f30950f : SUCCESS in 3m 25s\n- tooz-tox-py310-postgresql https://zuul.opendev.org/t/openstack/build/3c6da1a309374954ad6af05465995eb9 : SUCCESS in 4m 04s\n- tooz-tox-py313-postgresql https://zuul.opendev.org/t/openstack/build/a780f2bda2e1475ba55494c8bf5e12c9 : SUCCESS in 4m 39s\n- tooz-tox-py310-mysql https://zuul.opendev.org/t/openstack/build/2b8661cb0b024db68c06c12d677eccbc : SUCCESS in 4m 38s\n- tooz-tox-py313-mysql https://zuul.opendev.org/t/openstack/build/ba0ff774727446e99fc8c36a193637ab : SUCCESS in 7m 11s\n- tooz-tox-py310-consul https://zuul.opendev.org/t/openstack/build/9176e42b2bf24b9db1b66c6f5c77f6a3 : SUCCESS in 4m 29s\n- tooz-tox-py313-consul https://zuul.opendev.org/t/openstack/build/96bdce9e910e412ebdf3c58478b66e2e : SUCCESS in 4m 43s","accounts_in_message":[],"_revision_number":5},{"id":"8ba0411f72224d77956855e8b9eb4f627f1857df","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-04-11 11:25:43.000000000","message":"Uploaded patch set 6: New patch set was added with same tree, parent tree, and commit message as Patch Set 5.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"d28b1f4bdfe954d70fb4ad1749338b5ead8f6f41","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-11 13:24:54.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8e5c8e5be6c4460095a086028a1fae40\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/1ad70cb9b03a4ec8905676de9b84a45b : SUCCESS in 2m 47s\n- requirements-check https://zuul.opendev.org/t/openstack/build/ebb484c8dd9e4feabdcc0e0bc6cccb24 : SUCCESS in 3m 34s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4a0fdb3938ae47e7bd86f08231658599 : SUCCESS in 1h 53m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/478af56341a842f19e61f4a7fffa9e76 : SUCCESS in 7m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dd1b928ed6de4a3d9a45683245c80e24 : SUCCESS in 5m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/768fc0a45efc413399afc3a3fb2fd53f : SUCCESS in 4m 38s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9ab517753ed64c7eb5604d74fc14f960 : SUCCESS in 5m 01s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/499ec719fd784903a11cef4dec83e1dc : SUCCESS in 7m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57a9deeb774a493f966659f97ce2f3ec : SUCCESS in 4m 45s\n- tooz-tox-examples https://zuul.opendev.org/t/openstack/build/16e5326a62b44e1d8dbf59a06637a229 : SUCCESS in 3m 14s\n- tooz-tox-py310-etcd3gw https://zuul.opendev.org/t/openstack/build/1b4216b0f95e4ec88016c4627dbc3dff : SUCCESS in 5m 03s\n- tooz-tox-py313-etcd3gw https://zuul.opendev.org/t/openstack/build/7b12918a03d141bcbab7e923d6cc85e1 : SUCCESS in 5m 27s\n- tooz-tox-py310-zookeeper https://zuul.opendev.org/t/openstack/build/8438bc4a4c5b4722acfa4bc38bbe89ac : SUCCESS in 4m 20s\n- tooz-tox-py313-zookeeper https://zuul.opendev.org/t/openstack/build/fcaec560070042f8914cba290e1076a2 : SUCCESS in 5m 13s\n- tooz-tox-py310-redis https://zuul.opendev.org/t/openstack/build/6a168dea7dbe487e917fa5914626cc18 : SUCCESS in 6m 08s\n- tooz-tox-py313-redis https://zuul.opendev.org/t/openstack/build/9afaaa01b8294503aa2fb92377f06577 : SUCCESS in 5m 43s\n- tooz-tox-py310-sentinel https://zuul.opendev.org/t/openstack/build/76292f2fe23c46109efe024bacf8099d : SUCCESS in 6m 13s\n- tooz-tox-py313-sentinel https://zuul.opendev.org/t/openstack/build/86491613994c401e9e8cc2112254fa61 : SUCCESS in 6m 46s\n- tooz-tox-py310-memcached https://zuul.opendev.org/t/openstack/build/b79db3bc2d8e4ca5b4aa07c69dd8376b : SUCCESS in 5m 00s\n- tooz-tox-py313-memcached https://zuul.opendev.org/t/openstack/build/ac18bf53bc404fffa7815e0e4b2898ee : SUCCESS in 7m 29s\n- tooz-tox-py310-postgresql https://zuul.opendev.org/t/openstack/build/ee3fe24c54fb4d7c95e05c5a2d3ef5a2 : SUCCESS in 4m 35s\n- tooz-tox-py313-postgresql https://zuul.opendev.org/t/openstack/build/9147c6a279e34d9ab9741a8469c865ed : SUCCESS in 5m 09s\n- tooz-tox-py310-mysql https://zuul.opendev.org/t/openstack/build/910a2395c0cd47aa99499dc2e1922ec9 : SUCCESS in 3m 04s\n- tooz-tox-py313-mysql https://zuul.opendev.org/t/openstack/build/d5559439aa68433090fa61780e1e43f1 : SUCCESS in 7m 29s\n- tooz-tox-py310-consul https://zuul.opendev.org/t/openstack/build/ca7f0c30075542f0b5663d53ae6a712b : SUCCESS in 4m 25s\n- tooz-tox-py313-consul https://zuul.opendev.org/t/openstack/build/27cca9fc487146599ba0e290bf5d6bd8 : SUCCESS in 5m 19s","accounts_in_message":[],"_revision_number":6},{"id":"87b4098e95c31bb852694ad9c051b70e423d6e84","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2026-04-13 08:11:26.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"3e472abc0a64a40b1d130cfd14dd39b9b6f73310","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-04-14 09:45:17.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":7},{"id":"1bf0d3c7a0654c7846bc915f3019bdda26f04679","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-14 10:49:16.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/93befc4411214e288dac4fe46ac35e20\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/e359ceed1dce48a7883772cc3150455f : SUCCESS in 2m 08s\n- requirements-check https://zuul.opendev.org/t/openstack/build/cae5226d532c450982a0f9c6c0af5309 : SUCCESS in 2m 43s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1c265111caae4d1abf9b1b621a831e8c : SUCCESS in 1h 00m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8a11797b74614bddb3fe3f30e465fdae : SUCCESS in 4m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9275e7c711f04bf3b52b11eb2833035e : SUCCESS in 4m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/736b7347aca841f7b7c34e230c68f5d2 : SUCCESS in 4m 47s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8e569e5c2e6f4fd8a31641cba8a0a809 : SUCCESS in 6m 03s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/0bc4c581d5dc451682f6106c12188c48 : FAILURE in 6m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ff12d216bd74efab6017068dcf03965 : SUCCESS in 4m 04s\n- tooz-tox-examples https://zuul.opendev.org/t/openstack/build/d99ca8edcdd9419abfc14b212b32bf56 : SUCCESS in 6m 07s\n- tooz-tox-py310-etcd3gw https://zuul.opendev.org/t/openstack/build/e53bde509f2d4e79aa5f24b06eaa4671 : SUCCESS in 4m 03s\n- tooz-tox-py313-etcd3gw https://zuul.opendev.org/t/openstack/build/0f19ba03ae424a1e8fd9170ad472143c : SUCCESS in 7m 34s\n- tooz-tox-py310-zookeeper https://zuul.opendev.org/t/openstack/build/a69172a896c14f55b5b44591ffe2643a : SUCCESS in 3m 45s\n- tooz-tox-py313-zookeeper https://zuul.opendev.org/t/openstack/build/7c471f54f03c4ffcb243ac0e10d1b570 : SUCCESS in 5m 22s\n- tooz-tox-py310-redis https://zuul.opendev.org/t/openstack/build/22d006621816479eadb9e68dbe42cb0e : SUCCESS in 5m 48s\n- tooz-tox-py313-redis https://zuul.opendev.org/t/openstack/build/65f4633cbc374871b695694854236f60 : SUCCESS in 5m 14s\n- tooz-tox-py310-sentinel https://zuul.opendev.org/t/openstack/build/bd5f2ea0ab58482b96c8d0d4a5db3be5 : SUCCESS in 6m 25s\n- tooz-tox-py313-sentinel https://zuul.opendev.org/t/openstack/build/a4288ae337aa4e1d962e7565e9f04d70 : SUCCESS in 7m 12s\n- tooz-tox-py310-memcached https://zuul.opendev.org/t/openstack/build/f372f9754d9a4dbb85105dd2c0689da5 : SUCCESS in 4m 19s\n- tooz-tox-py313-memcached https://zuul.opendev.org/t/openstack/build/e036fb6319a548ac89c7924b4d4fae6d : SUCCESS in 5m 53s\n- tooz-tox-py310-postgresql https://zuul.opendev.org/t/openstack/build/ea8f6f93d9554e92ab66a8869c54b421 : SUCCESS in 5m 31s\n- tooz-tox-py313-postgresql https://zuul.opendev.org/t/openstack/build/3041d8e777884dc6bae369bcf538a066 : SUCCESS in 5m 38s\n- tooz-tox-py310-mysql https://zuul.opendev.org/t/openstack/build/25e4f301d7cf436b8c7a4fd2fc5475d2 : SUCCESS in 4m 32s\n- tooz-tox-py313-mysql https://zuul.opendev.org/t/openstack/build/40ba051109c147f8952d07b14979e243 : SUCCESS in 8m 18s\n- tooz-tox-py310-consul https://zuul.opendev.org/t/openstack/build/89d642e64473424e85781418d4e03f32 : SUCCESS in 4m 00s\n- tooz-tox-py313-consul https://zuul.opendev.org/t/openstack/build/db9183d8521144cd867fc9d2a8f78130 : SUCCESS in 5m 40s","accounts_in_message":[],"_revision_number":7},{"id":"0084834e48cfae647168fb6cab8e5d41f6f5c584","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-04-23 14:42:08.000000000","message":"Patch Set 7: Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"b2ac3d0cebaa86bf38e8981c742e49441104eb2c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-23 14:42:44.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"46e89e98cff164ed003299e40e4735afc20f5bc7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-23 16:51:34.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c6d6677cbdfe46b9a233d0ac2c525087\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/6bf3d82b308e48ae928363a01bd20e77 : SUCCESS in 3m 10s\n- requirements-check https://zuul.opendev.org/t/openstack/build/96a36a9b8efd4998b8e54cfe4e102b16 : SUCCESS in 3m 41s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0a8ec7ba308246f198d9b9cbd2576df2 : SUCCESS in 1h 53m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b28d8c156a34452b4d5e820cb95cb37 : SUCCESS in 6m 05s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/275d82b273f94b95b4ece392cc30dd5c : SUCCESS in 5m 11s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8fa89497473444beaf7538725f91a97a : SUCCESS in 5m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1124a8a26385407891934f3ece917c6f : SUCCESS in 4m 16s\n- tooz-tox-examples https://zuul.opendev.org/t/openstack/build/03fc2fd409de4e54924ff14d3bdda812 : SUCCESS in 5m 30s\n- tooz-tox-py310-etcd3gw https://zuul.opendev.org/t/openstack/build/fbc4d3be9f3d4558b9c0ce5bcbaeadd0 : SUCCESS in 4m 36s\n- tooz-tox-py313-etcd3gw https://zuul.opendev.org/t/openstack/build/a4010e90ee3b4ce59109f8146b1975bc : SUCCESS in 5m 31s\n- tooz-tox-py310-zookeeper https://zuul.opendev.org/t/openstack/build/5b8343e6656949c38706b723636d37e0 : SUCCESS in 4m 30s\n- tooz-tox-py313-zookeeper https://zuul.opendev.org/t/openstack/build/7ba5c046a8ec48cc98eb96e5227f1f73 : SUCCESS in 4m 33s\n- tooz-tox-py310-redis https://zuul.opendev.org/t/openstack/build/0cbfa2aa888d426ba4fa8f10f2ec7225 : SUCCESS in 5m 48s\n- tooz-tox-py313-redis https://zuul.opendev.org/t/openstack/build/fc303e53880e4d3dae8e5af7411cfae6 : SUCCESS in 7m 41s\n- tooz-tox-py310-sentinel https://zuul.opendev.org/t/openstack/build/ba514254f25a45ce96194847b090ab54 : SUCCESS in 6m 24s\n- tooz-tox-py313-sentinel https://zuul.opendev.org/t/openstack/build/c2d8ea31747844ec9213749b2efda36f : SUCCESS in 6m 40s\n- tooz-tox-py310-memcached https://zuul.opendev.org/t/openstack/build/9a3bccc949e447f48ab083d1aac8bfc5 : SUCCESS in 3m 36s\n- tooz-tox-py313-memcached https://zuul.opendev.org/t/openstack/build/a094da5102a9417a82a29cc9fcf05746 : SUCCESS in 4m 56s\n- tooz-tox-py310-postgresql https://zuul.opendev.org/t/openstack/build/d77126b027ee4f358ab5fe6106d9bbe7 : SUCCESS in 4m 24s\n- tooz-tox-py313-postgresql https://zuul.opendev.org/t/openstack/build/6eaa59f724ac4442976672b2cff550ea : SUCCESS in 5m 09s\n- tooz-tox-py310-mysql https://zuul.opendev.org/t/openstack/build/4a94666a3e374582a0dda63c31bb0282 : SUCCESS in 4m 58s\n- tooz-tox-py313-mysql https://zuul.opendev.org/t/openstack/build/85068b16f728468b88079de2271ff0d7 : SUCCESS in 7m 28s\n- tooz-tox-py310-consul https://zuul.opendev.org/t/openstack/build/1999423eab48467da4dc36e58a196e26 : SUCCESS in 4m 25s\n- tooz-tox-py313-consul https://zuul.opendev.org/t/openstack/build/34bb79f908d4417eb6b0b153ebea0a45 : SUCCESS in 6m 46s","accounts_in_message":[],"_revision_number":7},{"id":"d733c00d1542ad514abc00ae63d3270aa57b1e3b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-23 16:51:35.000000000","message":"Change has been successfully merged\n\n6 is the latest approved patch-set.\nNo files were changed between the latest approved patch-set and the submitted one.\n","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"334ee42e938f6838b53f506718777582f35e882f","revisions":{"749285cfe233848b31e1d9130e78f9eff80f7faa":{"kind":"REWORK","_number":1,"created":"2026-03-14 14:40:20.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/63/980563/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/63/980563/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/63/980563/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/63/980563/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/63/980563/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/63/980563/1"}}},"commit":{"parents":[{"commit":"157beb3dd639ed799bc23c998fee266c2231c790","subject":"typing: Add type hints to tooz.coordination, tooz.partitioner","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/157beb3dd639ed799bc23c998fee266c2231c790"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-03-13 12:51:51.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-03-14 14:40:15.000000000","tz":0},"subject":"typing: Add type hints to tooz.driver.etcd3gw","message":"typing: Add type hints to tooz.driver.etcd3gw\n\nThis was recently typed so it\u0027s a good first candidate.\n\nChange-Id: I621dc87412d44fd87bb9056a6304fbe71f52ca69\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/749285cfe233848b31e1d9130e78f9eff80f7faa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/749285cfe233848b31e1d9130e78f9eff80f7faa"}]},"branch":"refs/heads/master"},"8293e079db1f80659f82a975fd7c9ef8dd3e3e52":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-03-19 09:47:24.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/63/980563/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/63/980563/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/63/980563/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/63/980563/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/63/980563/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/63/980563/2"}}},"commit":{"parents":[{"commit":"2355ac4823aedfe0be104c914608f37b04c31155","subject":"typing: Add type hints to tooz.coordination, tooz.partitioner","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/2355ac4823aedfe0be104c914608f37b04c31155"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-03-13 12:51:51.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-03-19 09:46:03.000000000","tz":60},"subject":"typing: Add type hints to tooz.driver.etcd3gw","message":"typing: Add type hints to tooz.driver.etcd3gw\n\nThis was recently typed so it\u0027s a good first candidate.\n\nChange-Id: I621dc87412d44fd87bb9056a6304fbe71f52ca69\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/8293e079db1f80659f82a975fd7c9ef8dd3e3e52"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/8293e079db1f80659f82a975fd7c9ef8dd3e3e52"}]},"branch":"refs/heads/master"},"3afbf85c422553bed38d05cd3468fa1e0f7ffcad":{"kind":"REWORK","_number":3,"created":"2026-03-31 09:03:29.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/63/980563/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/63/980563/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/63/980563/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/63/980563/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/63/980563/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/63/980563/3"}}},"commit":{"parents":[{"commit":"037736c6dd2806da8257917003f94b63fc8df906","subject":"typing: Add type hints to tooz.coordination, tooz.partitioner","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/037736c6dd2806da8257917003f94b63fc8df906"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-03-13 12:51:51.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-03-31 09:02:23.000000000","tz":120},"subject":"typing: Add type hints to tooz.driver.etcd3gw","message":"typing: Add type hints to tooz.driver.etcd3gw\n\nThis was recently typed so it\u0027s a good first candidate.\n\nChange-Id: I621dc87412d44fd87bb9056a6304fbe71f52ca69\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/3afbf85c422553bed38d05cd3468fa1e0f7ffcad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/3afbf85c422553bed38d05cd3468fa1e0f7ffcad"}]},"branch":"refs/heads/master"},"9b321f179ec1064605b4d56de809a13345df1221":{"kind":"REWORK","_number":4,"created":"2026-04-10 18:13:45.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/63/980563/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/63/980563/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/63/980563/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/63/980563/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/63/980563/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/63/980563/4"}}},"commit":{"parents":[{"commit":"c09665ee7d0a38a446a036f81b56477e360c76aa","subject":"typing: Add type hints to tooz.coordination, tooz.partitioner","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/c09665ee7d0a38a446a036f81b56477e360c76aa"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-03-13 12:51:51.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-04-10 14:50:14.000000000","tz":60},"subject":"typing: Add type hints to tooz.driver.etcd3gw","message":"typing: Add type hints to tooz.driver.etcd3gw\n\nThis was recently typed so it\u0027s a good first candidate.\n\nChange-Id: I621dc87412d44fd87bb9056a6304fbe71f52ca69\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/9b321f179ec1064605b4d56de809a13345df1221"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/9b321f179ec1064605b4d56de809a13345df1221"}]},"branch":"refs/heads/master"},"60537c7e0a549dcb8d0694f4b749325232ecc1df":{"kind":"REWORK","_number":5,"created":"2026-04-10 22:56:02.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/63/980563/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/63/980563/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/63/980563/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/63/980563/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/63/980563/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/63/980563/5"}}},"commit":{"parents":[{"commit":"c09665ee7d0a38a446a036f81b56477e360c76aa","subject":"typing: Add type hints to tooz.coordination, tooz.partitioner","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/c09665ee7d0a38a446a036f81b56477e360c76aa"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-03-13 12:51:51.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-04-10 22:55:48.000000000","tz":60},"subject":"typing: Add type hints to tooz.driver.etcd3gw","message":"typing: Add type hints to tooz.driver.etcd3gw\n\nThis was recently typed so it\u0027s a good first candidate.\n\nChange-Id: I621dc87412d44fd87bb9056a6304fbe71f52ca69\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/60537c7e0a549dcb8d0694f4b749325232ecc1df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/60537c7e0a549dcb8d0694f4b749325232ecc1df"}]},"branch":"refs/heads/master"},"bc13f1177083ea16a156fe2697f58a899b0ffc3a":{"kind":"NO_CHANGE","_number":6,"created":"2026-04-11 11:25:43.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/63/980563/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/63/980563/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/63/980563/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/63/980563/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/63/980563/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/63/980563/6"}}},"commit":{"parents":[{"commit":"c09665ee7d0a38a446a036f81b56477e360c76aa","subject":"typing: Add type hints to tooz.coordination, tooz.partitioner","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/c09665ee7d0a38a446a036f81b56477e360c76aa"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-03-13 12:51:51.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-04-11 11:25:33.000000000","tz":60},"subject":"typing: Add type hints to tooz.driver.etcd3gw","message":"typing: Add type hints to tooz.driver.etcd3gw\n\nThis was recently typed so it\u0027s a good first candidate.\n\nChange-Id: I621dc87412d44fd87bb9056a6304fbe71f52ca69\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/bc13f1177083ea16a156fe2697f58a899b0ffc3a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/bc13f1177083ea16a156fe2697f58a899b0ffc3a"}]},"branch":"refs/heads/master"},"334ee42e938f6838b53f506718777582f35e882f":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2026-04-14 09:45:17.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/63/980563/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/63/980563/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/63/980563/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/63/980563/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/63/980563/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/63/980563/7"}}},"commit":{"parents":[{"commit":"bf925b9b301208a64b5e28c2e382da58949468f8","subject":"typing: Add type hints to tooz.coordination, tooz.partitioner","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/bf925b9b301208a64b5e28c2e382da58949468f8"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-03-13 12:51:51.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-04-14 09:45:09.000000000","tz":60},"subject":"typing: Add type hints to tooz.driver.etcd3gw","message":"typing: Add type hints to tooz.driver.etcd3gw\n\nThis was recently typed so it\u0027s a good first candidate.\n\nChange-Id: I621dc87412d44fd87bb9056a6304fbe71f52ca69\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/334ee42e938f6838b53f506718777582f35e882f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/334ee42e938f6838b53f506718777582f35e882f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
