)]}'
{"id":"openstack%2Ftooz~984101","triplet_id":"openstack%2Ftooz~master~Id9218ab2042d4273bb2f3ef1c2561a4851e6927d","project":"openstack/tooz","branch":"master","topic":"typing","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2026-05-01 17:31:08.000000000","reason":"\u003cGERRIT_ACCOUNT_15334\u003e replied on the change","reason_account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-05-01 19:19:40.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id9218ab2042d4273bb2f3ef1c2561a4851e6927d","subject":"zookeeper: Normalize path generation","status":"MERGED","created":"2026-04-10 18:13:45.000000000","updated":"2026-05-01 19:21:31.000000000","submitted":"2026-05-01 19:19:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"984101-typing","meta_rev_id":"6fb54a18fda297c17e477bcb85617b2ce867f53f","_number":984101,"virtual_id_number":984101,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-05-01 19:19:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2026-04-27 12:00:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2026-04-27 12:00:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2026-05-01 17:31:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-10 20:04:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-04-24 13:34:05.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2026-04-26 14:02:34.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"}],"messages":[{"id":"2b685ed27d5b4c2a1d066f4464da9dc70e32ab53","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-04-10 18:13:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ee060ad1742aa6970cffa720175f79ace078b653","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-10 20:04:26.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f4ee6695068e49af80347c0fd309f98c\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9af006d68cb34c758aa828f91edcc78b : SUCCESS in 1h 09m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0a43b5fbe5e34758b561248eff86420f : SUCCESS in 6m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/11574eefa57a44ac888ae3157ab3523c : SUCCESS in 6m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/35b8d305d6164925890c625c80b33e78 : SUCCESS in 5m 56s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/898fab24768942a7b62f43f760ec4bb0 : SUCCESS in 6m 43s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/8fce80cebf3e4a958b6e03d27c691d5e : SUCCESS in 5m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d865f19a98943f888a0647f533a9728 : SUCCESS in 4m 05s\n- tooz-tox-examples https://zuul.opendev.org/t/openstack/build/da9e26bb58804671b25780b97b154040 : SUCCESS in 5m 34s\n- tooz-tox-py310-etcd3gw https://zuul.opendev.org/t/openstack/build/8a5fd2358b57413cbb3923b4a4d65587 : FAILURE in 6m 11s\n- tooz-tox-py313-etcd3gw https://zuul.opendev.org/t/openstack/build/cf3ccd15f59047c2b773fbb02d20e3f0 : FAILURE in 5m 57s\n- tooz-tox-py310-zookeeper https://zuul.opendev.org/t/openstack/build/2cfb799c74f943e0af57f2055869862a : SUCCESS in 5m 05s\n- tooz-tox-py313-zookeeper https://zuul.opendev.org/t/openstack/build/1828986bd7024d12b7c366a0e536d7b3 : SUCCESS in 3m 47s\n- tooz-tox-py310-redis https://zuul.opendev.org/t/openstack/build/50ca54b15d8b402ba04866ee62467605 : SUCCESS in 7m 35s\n- tooz-tox-py313-redis https://zuul.opendev.org/t/openstack/build/68bf8abbdac4404494a22ccc15942ed0 : SUCCESS in 7m 11s\n- tooz-tox-py310-sentinel https://zuul.opendev.org/t/openstack/build/b170eeab4ffc412bbe119edb0d527419 : SUCCESS in 6m 57s\n- tooz-tox-py313-sentinel https://zuul.opendev.org/t/openstack/build/78877b974ca2417aa30e38badc63002d : SUCCESS in 8m 20s\n- tooz-tox-py310-memcached https://zuul.opendev.org/t/openstack/build/06d816a4544842fc977430fbf14ecdfb : SUCCESS in 5m 44s\n- tooz-tox-py313-memcached https://zuul.opendev.org/t/openstack/build/20434284244d4e9db214fb1f7b383351 : SUCCESS in 6m 03s\n- tooz-tox-py310-postgresql https://zuul.opendev.org/t/openstack/build/0a3fe930642044bcaa2cc2d27b9f9af5 : SUCCESS in 5m 29s\n- tooz-tox-py313-postgresql https://zuul.opendev.org/t/openstack/build/13718da467c545119a0f117884846f8a : SUCCESS in 4m 44s\n- tooz-tox-py310-mysql https://zuul.opendev.org/t/openstack/build/6d73d4a2b90d472ead8aece46efe94d4 : SUCCESS in 5m 55s\n- tooz-tox-py313-mysql https://zuul.opendev.org/t/openstack/build/1dbdc54a8d3342438b5f72e41315b691 : SUCCESS in 9m 24s\n- tooz-tox-py310-consul https://zuul.opendev.org/t/openstack/build/b6585452782e4051aacc6863fd8d8553 : SUCCESS in 5m 02s\n- tooz-tox-py313-consul https://zuul.opendev.org/t/openstack/build/ef5182df10984135834e67241d5cd3d7 : SUCCESS in 6m 14s","accounts_in_message":[],"_revision_number":1},{"id":"fd7eda6fcb1fcc30dd021bda9047bae4118f97b2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-04-11 11:25:43.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"374de5b60bd58a629e3a260e7299f1bee7503ed8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-11 13:31:29.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/47b47648a6ab4fab9a667faf6bcb7623\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/baded386dc754fa780a46af0636d3946 : SUCCESS in 1h 57m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/747e2ca2b3d34181b011037cbb5981ca : SUCCESS in 6m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/045db6d026164989889ac21150650d96 : SUCCESS in 5m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/af088d889fa844c681e26641bbcdf4fd : SUCCESS in 4m 40s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4dbf5c1fcb9c4327989d2f4725013cb8 : SUCCESS in 5m 42s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/c2b70393750743c3903e198ccbb872c5 : FAILURE in 6m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e41286e4c36242e5a269dfdea04ca54a : SUCCESS in 6m 05s\n- tooz-tox-examples https://zuul.opendev.org/t/openstack/build/2eb4431377ae4c3381c3504241c02920 : SUCCESS in 3m 08s\n- tooz-tox-py310-etcd3gw https://zuul.opendev.org/t/openstack/build/936fa4ebfdc544e8a5cc90338be1ae9d : SUCCESS in 5m 21s\n- tooz-tox-py313-etcd3gw https://zuul.opendev.org/t/openstack/build/7143e9dab07c4aad89c1e4dc554ba019 : SUCCESS in 5m 40s\n- tooz-tox-py310-zookeeper https://zuul.opendev.org/t/openstack/build/460ebb4b1182451a9b569c9c5ad1bb9d : SUCCESS in 4m 24s\n- tooz-tox-py313-zookeeper https://zuul.opendev.org/t/openstack/build/16684e61e3a248ea8129787f7f380db8 : SUCCESS in 5m 04s\n- tooz-tox-py310-redis https://zuul.opendev.org/t/openstack/build/3e210d361ef94602b33e80dbfa13575c : SUCCESS in 6m 31s\n- tooz-tox-py313-redis https://zuul.opendev.org/t/openstack/build/ad6beb504e1e47a994662d169fb99eb6 : SUCCESS in 8m 51s\n- tooz-tox-py310-sentinel https://zuul.opendev.org/t/openstack/build/96e767b904e84ad1b14583f4d35b15d5 : SUCCESS in 5m 37s\n- tooz-tox-py313-sentinel https://zuul.opendev.org/t/openstack/build/c3406350ee9a4b6c8c92fe9d13196250 : SUCCESS in 6m 57s\n- tooz-tox-py310-memcached https://zuul.opendev.org/t/openstack/build/fe19ecf51d1645fba63d3525a0df159f : SUCCESS in 4m 37s\n- tooz-tox-py313-memcached https://zuul.opendev.org/t/openstack/build/43c5aa2efb5f45fe855e0ff75d4e819a : SUCCESS in 5m 36s\n- tooz-tox-py310-postgresql https://zuul.opendev.org/t/openstack/build/1a63c8eb7e794d798e0ac4e17a402d3a : SUCCESS in 3m 59s\n- tooz-tox-py313-postgresql https://zuul.opendev.org/t/openstack/build/4d12469df33d4414b242db489e0d26f4 : SUCCESS in 3m 12s\n- tooz-tox-py310-mysql https://zuul.opendev.org/t/openstack/build/bbbdb3eb310b491082778c522698761c : SUCCESS in 4m 34s\n- tooz-tox-py313-mysql https://zuul.opendev.org/t/openstack/build/270effb8611d45d88a269f0d95b4d7c8 : SUCCESS in 7m 45s\n- tooz-tox-py310-consul https://zuul.opendev.org/t/openstack/build/4b0fd683e54d4f33b7b7dc6fccaad26e : SUCCESS in 3m 34s\n- tooz-tox-py313-consul https://zuul.opendev.org/t/openstack/build/eb0f538c6fdc4060b64673cc7d36d6f6 : SUCCESS in 5m 32s","accounts_in_message":[],"_revision_number":2},{"id":"1007d9dea3f2aa2716372ade96333fe43f20a9ce","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-04-14 09:45:17.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"ee2bdf12159547f13866e389012887ddb51134b6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-14 11:44:26.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fca86d652f674ab792c86f71e0d4a273\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/81389cc5c02e485c91b5d7ea049993b1 : SUCCESS in 1h 48m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9d0a2a58bc1c4ed5856da9bb251b3ee2 : SUCCESS in 5m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ecd2440323bb4a5a9fcb08a044603921 : SUCCESS in 6m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5d44e5d97389461dbd64ef077e5f9628 : SUCCESS in 5m 57s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/46745d9f1f7d463e8ce0ae819746d605 : SUCCESS in 5m 42s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/8e426c4ec3974402a55535b7489ed5f9 : SUCCESS in 7m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/690ffb1f1eef4ed2bf532fcf76459aa8 : SUCCESS in 5m 14s\n- tooz-tox-examples https://zuul.opendev.org/t/openstack/build/81764f432cd7413f9326d932f404f54b : SUCCESS in 6m 08s\n- tooz-tox-py310-etcd3gw https://zuul.opendev.org/t/openstack/build/2856ef506b754cd5a321ccb4c195e717 : SUCCESS in 5m 27s\n- tooz-tox-py313-etcd3gw https://zuul.opendev.org/t/openstack/build/a19e0deb86784fc9b9bd891999fc7cfb : SUCCESS in 5m 04s\n- tooz-tox-py310-zookeeper https://zuul.opendev.org/t/openstack/build/53db706f90b44c25ae59ec4b6c4105f3 : SUCCESS in 4m 19s\n- tooz-tox-py313-zookeeper https://zuul.opendev.org/t/openstack/build/7324928e29324a12a5eec9e2a65bb808 : SUCCESS in 4m 53s\n- tooz-tox-py310-redis https://zuul.opendev.org/t/openstack/build/1fce9c9cd26b4978a2e840fe35fcc982 : SUCCESS in 7m 07s\n- tooz-tox-py313-redis https://zuul.opendev.org/t/openstack/build/c62b3a52bfa944f0a149207c6f2789d3 : SUCCESS in 7m 11s\n- tooz-tox-py310-sentinel https://zuul.opendev.org/t/openstack/build/9ff5af8b5353488a86b23cb17c67e66c : SUCCESS in 7m 59s\n- tooz-tox-py313-sentinel https://zuul.opendev.org/t/openstack/build/20ca8263f6ce4a3ca0b6200c69839547 : SUCCESS in 5m 58s\n- tooz-tox-py310-memcached https://zuul.opendev.org/t/openstack/build/8756b75d91944514b576103dd72c0590 : SUCCESS in 5m 29s\n- tooz-tox-py313-memcached https://zuul.opendev.org/t/openstack/build/e5948d1b15b4451185ae076d80382290 : SUCCESS in 5m 35s\n- tooz-tox-py310-postgresql https://zuul.opendev.org/t/openstack/build/7910a4b80fa44b739114155e6792e6d9 : SUCCESS in 4m 33s\n- tooz-tox-py313-postgresql https://zuul.opendev.org/t/openstack/build/355a5f58030d4a98af5e783cfc4e0d31 : SUCCESS in 4m 59s\n- tooz-tox-py310-mysql https://zuul.opendev.org/t/openstack/build/ad818914f4704ab7b37b5dcf41b3a449 : SUCCESS in 5m 09s\n- tooz-tox-py313-mysql https://zuul.opendev.org/t/openstack/build/368a0ab7221740aba990d8621d459be9 : SUCCESS in 7m 03s\n- tooz-tox-py310-consul https://zuul.opendev.org/t/openstack/build/cc8aa614c4e74f19adccd18bf339010f : SUCCESS in 5m 16s\n- tooz-tox-py313-consul https://zuul.opendev.org/t/openstack/build/b8a0ef7bf050416bb190983ec1420170 : SUCCESS in 4m 19s","accounts_in_message":[],"_revision_number":3},{"id":"f311e45f6c587dd2d929dd8ca9c133e3f36f9f92","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2026-04-24 13:34:05.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"d60225450509519b1bf837092d7876f17eb9cb3b","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-04-26 14:02:34.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"8830b40546f42e156407f0fc446d4ba5bd00eede","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-04-27 12:00:15.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n* Workflow+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"f5ec02cf9ae118d52b22c146754bfce0d99de75b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-27 13:29:31.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d7f3e1f0b33a4233ae48a5a9537df3f3\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5bbb3c381a0c41839217e11b4c9a5d40 : SUCCESS in 1h 25m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/793dfb4fcde1412d96b5b22fb0409b01 : SUCCESS in 6m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/402df64fd4ec4a2d856bb31bbff82052 : SUCCESS in 6m 13s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a77c4531d4d045b2b630dbe5ce924204 : SUCCESS in 5m 41s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/dc0d642c0f6a4f5a8f2d52ce9fc15073 : SUCCESS in 5m 27s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/8e2e29976d5d441cad873fd379df7c39 : SUCCESS in 7m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/331d15a2bb844009a168de4de25cbffc : SUCCESS in 5m 34s\n- tooz-tox-examples https://zuul.opendev.org/t/openstack/build/32e8e2e6d9454ab9a9c8388db87ac80b : SUCCESS in 5m 08s\n- tooz-tox-py310-etcd3gw https://zuul.opendev.org/t/openstack/build/3b889d82a181468ca9c5ab3009d34909 : SUCCESS in 4m 36s\n- tooz-tox-py313-etcd3gw https://zuul.opendev.org/t/openstack/build/11589ec08ca14866945e198fa0e28488 : SUCCESS in 4m 33s\n- tooz-tox-py310-zookeeper https://zuul.opendev.org/t/openstack/build/9267247f1ae84e27b8a875a31b2ed019 : SUCCESS in 4m 50s\n- tooz-tox-py313-zookeeper https://zuul.opendev.org/t/openstack/build/30cddb5f3f774fa08a5d8f6735554efc : SUCCESS in 5m 45s\n- tooz-tox-py310-redis https://zuul.opendev.org/t/openstack/build/c81f1e381426415cbd15c462428fd84b : SUCCESS in 6m 23s\n- tooz-tox-py313-redis https://zuul.opendev.org/t/openstack/build/9432c0c848dd474fa7790f0285cb64ec : SUCCESS in 7m 12s\n- tooz-tox-py310-sentinel https://zuul.opendev.org/t/openstack/build/e91f9e6070c84706a335f9506c07eb65 : SUCCESS in 6m 15s\n- tooz-tox-py313-sentinel https://zuul.opendev.org/t/openstack/build/b4ca9fe2e63f4cfcac25ff818a591f78 : SUCCESS in 7m 09s\n- tooz-tox-py310-memcached https://zuul.opendev.org/t/openstack/build/8108e00a660a43a89c61f1e9ef84af09 : SUCCESS in 4m 47s\n- tooz-tox-py313-memcached https://zuul.opendev.org/t/openstack/build/e88f5097a06a4780b770ed553000d507 : SUCCESS in 5m 16s\n- tooz-tox-py310-postgresql https://zuul.opendev.org/t/openstack/build/2555366b76b14febbbe50742a9d32110 : SUCCESS in 4m 33s\n- tooz-tox-py313-postgresql https://zuul.opendev.org/t/openstack/build/357d560548ef44038e2058b9b3dda233 : SUCCESS in 5m 25s\n- tooz-tox-py310-mysql https://zuul.opendev.org/t/openstack/build/b3068771a4a046baaf1562dc3b7f609b : SUCCESS in 4m 33s\n- tooz-tox-py313-mysql https://zuul.opendev.org/t/openstack/build/de0f0242738f43c2a2ab8efcc289deb2 : SUCCESS in 7m 56s\n- tooz-tox-py310-consul https://zuul.opendev.org/t/openstack/build/db177ca217cf4eefb3569a3e05e27128 : SUCCESS in 4m 34s\n- tooz-tox-py313-consul https://zuul.opendev.org/t/openstack/build/b6d76f2410994245bc74e6858bfaa357 : SUCCESS in 5m 29s","accounts_in_message":[],"_revision_number":4},{"id":"0a955d8bd5478f0761fc22cbfc28196533fae799","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-05-01 17:31:08.000000000","message":"Patch Set 4: Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"08d854545fabdd61b7a4a3e95bbbbf08a7b9fc9c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-01 17:31:40.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"ec8c333f74fa933bbaa0d4d770a7d9e3e2263f88","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-01 19:19:40.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/98cd5f300d6e4c2ab6630e218af3b6d7\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e74b551d390b4a89b78502ebbac9fff8 : SUCCESS in 1h 05m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7f15389d00c14c5cac53c4f5827a1ff3 : SUCCESS in 6m 10s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d323e8763a2e413981c43c29390f25d3 : SUCCESS in 4m 27s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f066a9af1b0344de96d06e55b510b0ca : SUCCESS in 4m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/234fe24f60e541f598e882ba7e295e7f : SUCCESS in 6m 37s\n- tooz-tox-examples https://zuul.opendev.org/t/openstack/build/126c157802f440948b200502fea4d0ef : SUCCESS in 4m 52s\n- tooz-tox-py310-etcd3gw https://zuul.opendev.org/t/openstack/build/7de272ade64342b4b094b46567a49d4f : SUCCESS in 4m 55s\n- tooz-tox-py313-etcd3gw https://zuul.opendev.org/t/openstack/build/892a0c8dd04345a0b061ddc82d84630b : SUCCESS in 5m 26s\n- tooz-tox-py310-zookeeper https://zuul.opendev.org/t/openstack/build/e812e7d706e644469755baf3a12c24da : SUCCESS in 4m 21s\n- tooz-tox-py313-zookeeper https://zuul.opendev.org/t/openstack/build/0c696ef8e41c4964a2f640476f5d194c : SUCCESS in 4m 12s\n- tooz-tox-py310-redis https://zuul.opendev.org/t/openstack/build/21bd5a594fc748179a8a780403e0d8b5 : SUCCESS in 5m 12s\n- tooz-tox-py313-redis https://zuul.opendev.org/t/openstack/build/0296ff0777524577a2cd2bc653d8c056 : SUCCESS in 6m 42s\n- tooz-tox-py310-sentinel https://zuul.opendev.org/t/openstack/build/42e384780e2c4cb081e8647ba0238f06 : SUCCESS in 6m 09s\n- tooz-tox-py313-sentinel https://zuul.opendev.org/t/openstack/build/809d603eb83c404699b3a9693f9bebec : SUCCESS in 7m 21s\n- tooz-tox-py310-memcached https://zuul.opendev.org/t/openstack/build/d048e43e77274185b77fa452c432676f : SUCCESS in 4m 08s\n- tooz-tox-py313-memcached https://zuul.opendev.org/t/openstack/build/c1cfece12e284e809542481fb84bb857 : SUCCESS in 8m 58s\n- tooz-tox-py310-postgresql https://zuul.opendev.org/t/openstack/build/700e389f62904b27be71778f79ef24f5 : SUCCESS in 4m 05s\n- tooz-tox-py313-postgresql https://zuul.opendev.org/t/openstack/build/812b5794fd0348a6815593dd3c99a83a : SUCCESS in 4m 56s\n- tooz-tox-py310-mysql https://zuul.opendev.org/t/openstack/build/d203b63b2eb5420bae966a7212f42017 : SUCCESS in 4m 37s\n- tooz-tox-py313-mysql https://zuul.opendev.org/t/openstack/build/82dc8ede1d884a80b1203aef5ed96697 : SUCCESS in 7m 07s\n- tooz-tox-py310-consul https://zuul.opendev.org/t/openstack/build/3fdb66eff6314a8091d39f8478428b83 : SUCCESS in 4m 15s\n- tooz-tox-py313-consul https://zuul.opendev.org/t/openstack/build/b3467f68dd454cd087a15fe5c6c270c9 : SUCCESS in 5m 16s","accounts_in_message":[],"_revision_number":4},{"id":"cd799e519e615a6f63b41010f7a5df163d7459d6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-01 19:19:40.000000000","message":"Change has been successfully merged\n\n3 is the latest approved patch-set.\nNo files were changed between the latest approved patch-set and the submitted one.\n","accounts_in_message":[],"_revision_number":4},{"id":"6fb54a18fda297c17e477bcb85617b2ce867f53f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-01 19:21:31.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/55125c6cbbcb486397f7ef8d50f97a17\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ba774cd85e44c4499472ac337cff48d : SUCCESS in 42s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"70d1370794d27086ca74cf431566dbf702e80cd5","revisions":{"19d0999d76771f36169b00bf9836295076ac3906":{"kind":"REWORK","_number":1,"created":"2026-04-10 18:13:45.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/01/984101/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/01/984101/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/01/984101/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/01/984101/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/01/984101/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/01/984101/1"}}},"commit":{"parents":[{"commit":"8e9acac6f400207e3ab9d6fb2e76542059b17f56","subject":"zookeeper: Remove unnecessary staticmethod wrappers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/8e9acac6f400207e3ab9d6fb2e76542059b17f56"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-04-10 11:52:04.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-04-10 14:54:29.000000000","tz":60},"subject":"zookeeper: Normalize path generation","message":"zookeeper: Normalize path generation\n\nChange-Id: Id9218ab2042d4273bb2f3ef1c2561a4851e6927d\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/19d0999d76771f36169b00bf9836295076ac3906"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/19d0999d76771f36169b00bf9836295076ac3906"}]},"branch":"refs/heads/master"},"b69343ff69add0d63a419281f2623161b8234382":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-04-11 11:25:43.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/01/984101/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/01/984101/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/01/984101/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/01/984101/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/01/984101/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/01/984101/2"}}},"commit":{"parents":[{"commit":"c3e93cc1a203c24ec0c3b1521f8304065d3e6d93","subject":"zookeeper: Remove unnecessary staticmethod wrappers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/c3e93cc1a203c24ec0c3b1521f8304065d3e6d93"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-04-10 11:52:04.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-04-11 11:25:35.000000000","tz":60},"subject":"zookeeper: Normalize path generation","message":"zookeeper: Normalize path generation\n\nChange-Id: Id9218ab2042d4273bb2f3ef1c2561a4851e6927d\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/b69343ff69add0d63a419281f2623161b8234382"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/b69343ff69add0d63a419281f2623161b8234382"}]},"branch":"refs/heads/master"},"57b5c0ef449420ed9e9df4b441d410c0ccc1e936":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2026-04-14 09:45:17.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/01/984101/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/01/984101/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/01/984101/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/01/984101/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/01/984101/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/01/984101/3"}}},"commit":{"parents":[{"commit":"7c5760b778922270a3e31e8d18458dea8f739ed4","subject":"zookeeper: Remove unnecessary staticmethod wrappers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/7c5760b778922270a3e31e8d18458dea8f739ed4"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-04-10 11:52:04.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-04-14 09:45:09.000000000","tz":60},"subject":"zookeeper: Normalize path generation","message":"zookeeper: Normalize path generation\n\nChange-Id: Id9218ab2042d4273bb2f3ef1c2561a4851e6927d\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/57b5c0ef449420ed9e9df4b441d410c0ccc1e936"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/57b5c0ef449420ed9e9df4b441d410c0ccc1e936"}]},"branch":"refs/heads/master"},"70d1370794d27086ca74cf431566dbf702e80cd5":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2026-04-27 12:00:15.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/01/984101/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/01/984101/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/01/984101/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/01/984101/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/01/984101/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/01/984101/4"}}},"commit":{"parents":[{"commit":"bd4d7684658ccba46ab624cf9cceeed707835ac7","subject":"zookeeper: Remove unnecessary staticmethod wrappers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/bd4d7684658ccba46ab624cf9cceeed707835ac7"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-04-10 11:52:04.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-04-27 11:27:55.000000000","tz":60},"subject":"zookeeper: Normalize path generation","message":"zookeeper: Normalize path generation\n\nChange-Id: Id9218ab2042d4273bb2f3ef1c2561a4851e6927d\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/70d1370794d27086ca74cf431566dbf702e80cd5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/70d1370794d27086ca74cf431566dbf702e80cd5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
