)]}'
{"id":"openstack%2Ftooz~984292","triplet_id":"openstack%2Ftooz~master~I71ecd8d9e0b8895bbbbdf5ed190b7d548d3d05b2","project":"openstack/tooz","branch":"master","topic":"bug/2148145","attention_set":{"36741":{"account":{"_account_id":36741,"name":"Eunkyung Kim","display_name":"eunkyung","email":"ek121.kim@samsung.com","username":"eunkyung"},"last_update":"2026-10-01 10:32:47.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2026-10-02 12:53:51.000000000","reason":"\u003cGERRIT_ACCOUNT_15334\u003e replied on the change","reason_account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}}},"hashtags":[],"change_id":"I71ecd8d9e0b8895bbbbdf5ed190b7d548d3d05b2","subject":"zookeeper: Add retry parameters\u0027 parsing logic","status":"NEW","created":"2026-04-13 03:40:53.000000000","updated":"2026-10-02 12:53:51.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":10,"unresolved_comment_count":5,"has_review_started":true,"meta_rev_id":"069e898212d16a217f7b0710c5c1a135ff6368da","_number":984292,"virtual_id_number":984292,"owner":{"_account_id":36741,"name":"Eunkyung Kim","display_name":"eunkyung","email":"ek121.kim@samsung.com","username":"eunkyung"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-10-02 02:40:40.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":-1,"date":"2026-10-02 12:53:51.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-13 04:49:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-10 13:45:30.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"CC"},{"updated":"2026-10-02 12:53:51.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"}],"messages":[{"id":"fec283a4d725f02fab87bb94185f192deac54f9d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36741,"name":"Eunkyung Kim","display_name":"eunkyung","email":"ek121.kim@samsung.com","username":"eunkyung"},"date":"2026-04-13 03:40:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ec462d803516960dafad7d6544f4fc9196e01f36","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36741,"name":"Eunkyung Kim","display_name":"eunkyung","email":"ek121.kim@samsung.com","username":"eunkyung"},"date":"2026-04-13 03:45:23.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"dbfc3386f23f3f4d667464a72f4f3fce77bf9c74","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-13 04:49:13.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/45c087874295462bbecd1228c3b05e8c\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9f1235eeece64a4da2a522f278427517 : SUCCESS in 1h 02m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fe24f2b2a9004ae6a9f1b546ef6fab39 : SUCCESS in 5m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/090c302dffb34e8494c744604783ca6c : SUCCESS in 3m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/986a30930a0a45599ec8beacdf40c903 : SUCCESS in 3m 40s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f1bd698489104ba2af9f61c0ddff5abf : SUCCESS in 3m 51s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/44132723c76e46a5833a47081d880f3a : SUCCESS in 7m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4eb1d8e657eb4bba9a588fe2379fbe8e : SUCCESS in 4m 07s\n- tooz-tox-examples https://zuul.opendev.org/t/openstack/build/b8004c53a02f44e8a64bb09bd2fb832d : SUCCESS in 3m 15s\n- tooz-tox-py310-etcd3gw https://zuul.opendev.org/t/openstack/build/ac9d30f71e7e4406bf9d971b4ec3c22a : SUCCESS in 3m 35s\n- tooz-tox-py313-etcd3gw https://zuul.opendev.org/t/openstack/build/9e241d4784644471b9e869a84d953626 : SUCCESS in 6m 44s\n- tooz-tox-py310-zookeeper https://zuul.opendev.org/t/openstack/build/895d27553ef0457b8c5d292aa88f621f : SUCCESS in 4m 15s\n- tooz-tox-py313-zookeeper https://zuul.opendev.org/t/openstack/build/f1cc38f29e124fcb95fd0eadcec175ce : SUCCESS in 4m 59s\n- tooz-tox-py310-redis https://zuul.opendev.org/t/openstack/build/33f2f344160e4866a7600f975d0e0572 : SUCCESS in 6m 02s\n- tooz-tox-py313-redis https://zuul.opendev.org/t/openstack/build/a32ac22b7c574ed9ae5d514fda9f257e : SUCCESS in 8m 30s\n- tooz-tox-py310-sentinel https://zuul.opendev.org/t/openstack/build/016d2c381b654b669d5c4998deccafd1 : SUCCESS in 4m 49s\n- tooz-tox-py313-sentinel https://zuul.opendev.org/t/openstack/build/d824de229adc401e85df45163781f7c7 : SUCCESS in 6m 17s\n- tooz-tox-py310-memcached https://zuul.opendev.org/t/openstack/build/2f3adc5fa75a47acb392902dc2b4e77b : SUCCESS in 3m 27s\n- tooz-tox-py313-memcached https://zuul.opendev.org/t/openstack/build/da9f3dbd395143f787959cba1d01c8fd : SUCCESS in 4m 49s\n- tooz-tox-py310-postgresql https://zuul.opendev.org/t/openstack/build/9cc6d478ae2d4eb99701ff933266266a : SUCCESS in 3m 56s\n- tooz-tox-py313-postgresql https://zuul.opendev.org/t/openstack/build/a104f604cd43466f90b58c95b812cd57 : SUCCESS in 4m 56s\n- tooz-tox-py310-mysql https://zuul.opendev.org/t/openstack/build/9919da4f131f421fb404b13dfec2d2f4 : SUCCESS in 4m 13s\n- tooz-tox-py313-mysql https://zuul.opendev.org/t/openstack/build/c03ce40be12844b0b9601cc5856b5565 : SUCCESS in 6m 37s\n- tooz-tox-py310-consul https://zuul.opendev.org/t/openstack/build/f4aaf032ee6c455ba55d9b2a222d9628 : SUCCESS in 4m 36s\n- tooz-tox-py313-consul https://zuul.opendev.org/t/openstack/build/21207807517245fa81b41ddfef334984 : SUCCESS in 3m 12s","accounts_in_message":[],"_revision_number":2},{"id":"640c33ad502ab802b3b4280918be804e491d9b60","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-06-10 13:44:43.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"b44b6ccc96c5820c27fdae5155379a7f138adf95","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-06-10 13:45:30.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1c354bd6e269c77b569f2fcc71298db0b767cbbb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-10 14:55:02.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/72de73edd1194ac19572b482be15b68a\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c5b89bc7ccd14243bb0c0c691bc33c0b : SUCCESS in 1h 08m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d80b5ea158f948a98636d7a169d8e9ba : SUCCESS in 6m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a064c88291c94dc7bcdff7db301686e0 : FAILURE in 4m 30s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b403fcd7f01841769d693c1beff315ea : SUCCESS in 4m 28s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e6d2502bfe29441689985772130ff4fd : SUCCESS in 5m 19s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/a964f34cb1ec45d28acebec2c0c71fad : SUCCESS in 7m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e97e0779d53411dbcfb2d0eb477816f : SUCCESS in 5m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9925b9d16abd4555bee99eac46c86712 : SUCCESS in 4m 04s\n- tooz-tox-examples https://zuul.opendev.org/t/openstack/build/c7c6f871f8e246d3a7ae9cd93875b4a1 : SUCCESS in 5m 04s\n- tooz-tox-py310-etcd3gw https://zuul.opendev.org/t/openstack/build/635890229d014e0192aa39a1f6bba9ff : FAILURE in 4m 23s\n- tooz-tox-py313-etcd3gw https://zuul.opendev.org/t/openstack/build/ee30e688ea594514b5f30e3e16a1e554 : SUCCESS in 6m 01s\n- tooz-tox-py310-zookeeper https://zuul.opendev.org/t/openstack/build/62346fd5bbbe4353b97077de2f7a59c8 : FAILURE in 4m 17s\n- tooz-tox-py313-zookeeper https://zuul.opendev.org/t/openstack/build/b455ac979be440d7b26adf5f6f854e66 : SUCCESS in 5m 00s\n- tooz-tox-py310-redis https://zuul.opendev.org/t/openstack/build/3f892cab639c48f7baa94c72d978d916 : FAILURE in 4m 14s\n- tooz-tox-py313-redis https://zuul.opendev.org/t/openstack/build/deb06218fd354e1b9a213ca75a320179 : SUCCESS in 6m 09s\n- tooz-tox-py310-sentinel https://zuul.opendev.org/t/openstack/build/99bf82f440d24fc0a7e28e79ce929754 : FAILURE in 4m 29s\n- tooz-tox-py313-sentinel https://zuul.opendev.org/t/openstack/build/49ecaad067354d6d89c1122af9302a75 : SUCCESS in 5m 48s\n- tooz-tox-py310-memcached https://zuul.opendev.org/t/openstack/build/ed483663d5e246c29636159bbbc6dd67 : FAILURE in 4m 44s\n- tooz-tox-py313-memcached https://zuul.opendev.org/t/openstack/build/a982fbebb41f463aa5e7063a1a7b882e : SUCCESS in 5m 53s\n- tooz-tox-py310-postgresql https://zuul.opendev.org/t/openstack/build/647a98b15b954fc7a59a50ae26e4fbfe : FAILURE in 4m 10s\n- tooz-tox-py313-postgresql https://zuul.opendev.org/t/openstack/build/6642ddda39164df59b41b69110474618 : SUCCESS in 4m 48s\n- tooz-tox-py310-mysql https://zuul.opendev.org/t/openstack/build/bc2035d161764cae9de1ffe99d050039 : FAILURE in 4m 29s\n- tooz-tox-py313-mysql https://zuul.opendev.org/t/openstack/build/15dc5dc034654457a3803c7704b1a7a8 : SUCCESS in 8m 23s\n- tooz-tox-py310-consul https://zuul.opendev.org/t/openstack/build/9834f8eea2c34757a783c49798a144ff : FAILURE in 4m 16s\n- tooz-tox-py313-consul https://zuul.opendev.org/t/openstack/build/bc4cee2000bd4173a0c6c45f3be4717f : SUCCESS in 5m 16s","accounts_in_message":[],"_revision_number":3},{"id":"e0265bc4e0c9a2d761c811078cf3700059895c85","author":{"_account_id":36741,"name":"Eunkyung Kim","display_name":"eunkyung","email":"ek121.kim@samsung.com","username":"eunkyung"},"date":"2026-06-29 05:58:16.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"9e169156d77c0e875ca9b9a033b4baf072cfd702","author":{"_account_id":36741,"name":"Eunkyung Kim","display_name":"eunkyung","email":"ek121.kim@samsung.com","username":"eunkyung"},"date":"2026-09-21 11:07:26.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"cc630d694465c3ea6bb4b91d7fa13b65363bdb3a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 12:44:06.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/186fc153f5884b2588c841c1d5bf8148\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/98deb30aa8c64c27ad0b6b5427fbe1e4 : SUCCESS in 1h 29m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c7bc42b254104bb5bad92d9716ac5574 : SUCCESS in 6m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b84bb4578fba4bc794704a978ee177b2 : SUCCESS in 8m 39s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f7f3de49889b4f4b9798f42927856699 : SUCCESS in 5m 28s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/68a22f93a29445209db0bfb2ad1581c3 : SUCCESS in 3m 53s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/3f185074cc5244c7b4a9bd43521797f3 : SUCCESS in 8m 25s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/6ff265c427814b00a98c271519083d65 : SUCCESS in 9m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d5db3fc5c12f49cab4ba06177a3a7c3b : SUCCESS in 6m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b38ae62b49134f7bb50832555fdbf5a7 : SUCCESS in 4m 52s\n- tooz-tox-examples https://zuul.opendev.org/t/openstack/build/347e261c55b84ff1994424d4170a3a60 : SUCCESS in 5m 00s\n- tooz-tox-py311-etcd3gw https://zuul.opendev.org/t/openstack/build/ee09224ee4e04a90a9c0bbf4a568fbee : SUCCESS in 5m 50s\n- tooz-tox-py314-etcd3gw https://zuul.opendev.org/t/openstack/build/b6d6005fe93045519109d4db881c7cec : SUCCESS in 11m 10s\n- tooz-tox-py311-zookeeper https://zuul.opendev.org/t/openstack/build/755be44c8b624c74a2f3141408d93798 : SUCCESS in 6m 13s\n- tooz-tox-py314-zookeeper https://zuul.opendev.org/t/openstack/build/5f424c73cb5b4bcfa1436a0ba9a68413 : SUCCESS in 7m 16s\n- tooz-tox-py311-redis https://zuul.opendev.org/t/openstack/build/55d1b269461146afb009f3a3ddb6d64f : SUCCESS in 6m 22s\n- tooz-tox-py314-redis https://zuul.opendev.org/t/openstack/build/3d096c3593684245809967112f42eb92 : SUCCESS in 6m 20s\n- tooz-tox-py311-sentinel https://zuul.opendev.org/t/openstack/build/48352910a9714fb39a798176f5a0ccf5 : SUCCESS in 7m 52s\n- tooz-tox-py314-sentinel https://zuul.opendev.org/t/openstack/build/5a12b31f7c9f4646be449bfaeb52c309 : SUCCESS in 10m 54s\n- tooz-tox-py311-memcached https://zuul.opendev.org/t/openstack/build/2852b347625649c88a6514c02bb05ec5 : SUCCESS in 3m 48s\n- tooz-tox-py314-memcached https://zuul.opendev.org/t/openstack/build/17af67d741534cb6b5d52a8f841ee0f3 : SUCCESS in 8m 07s\n- tooz-tox-py311-postgresql https://zuul.opendev.org/t/openstack/build/05494d32decf4d5ba8da7725ae14c916 : SUCCESS in 5m 05s\n- tooz-tox-py314-postgresql https://zuul.opendev.org/t/openstack/build/76d6601e59eb4658b7fd21dca8c95b4d : SUCCESS in 4m 37s\n- tooz-tox-py311-mysql https://zuul.opendev.org/t/openstack/build/77c7948bdf03476db93483e86e928355 : SUCCESS in 8m 33s\n- tooz-tox-py314-mysql https://zuul.opendev.org/t/openstack/build/bb33e9c400e847dfaadb8ce95746a652 : SUCCESS in 8m 20s","accounts_in_message":[],"_revision_number":3},{"id":"509132414cd94b070bcc086eac9d505721bc6a90","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36741,"name":"Eunkyung Kim","display_name":"eunkyung","email":"ek121.kim@samsung.com","username":"eunkyung"},"date":"2026-10-01 08:57:15.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"05e10d4243ce37005db9bc3108185904d82e64c6","author":{"_account_id":36741,"name":"Eunkyung Kim","display_name":"eunkyung","email":"ek121.kim@samsung.com","username":"eunkyung"},"date":"2026-10-01 09:01:34.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"a30f9e10afbfe9af7f027d2c3fd13fdd0edfde86","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-01 10:32:47.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a46dda50504d40d891a8e337e3fd69e7\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1023075df0fe485aa80a947ada62cf0b : SUCCESS in 1h 29m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/af288fa0664a4573838b1f3c08fcf175 : SUCCESS in 5m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d4ea4b274dd84043a51f8f4dc58cbd4a : FAILURE in 2m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3aafa51cc3fd4aa899e4981a9d8df152 : SUCCESS in 4m 22s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/2f1a5c1c2466475dbebbdf9056c06326 : SUCCESS in 7m 17s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/5038212dcecb4f5a8dfe8f6861ebf2e7 : SUCCESS in 7m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c69220dbc4e246099d754691577dc282 : SUCCESS in 4m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/22d838fd1bb54a77afa3b1b33681e357 : SUCCESS in 3m 32s\n- tooz-tox-examples https://zuul.opendev.org/t/openstack/build/3bf3764fcd2e4b149c659877d4befbb8 : SUCCESS in 5m 14s\n- tooz-tox-py311-etcd3gw https://zuul.opendev.org/t/openstack/build/57b7640bf99e4e93a3ed4a89edd5fe09 : SUCCESS in 5m 27s\n- tooz-tox-py314-etcd3gw https://zuul.opendev.org/t/openstack/build/2845e5c4f38247abbb5b51934548b533 : SUCCESS in 9m 08s\n- tooz-tox-py311-zookeeper https://zuul.opendev.org/t/openstack/build/98388aebfd2d40b39d3bcfab53955095 : SUCCESS in 4m 38s\n- tooz-tox-py314-zookeeper https://zuul.opendev.org/t/openstack/build/a9704846866c4ed2974fc49b70d4ffaa : SUCCESS in 4m 51s\n- tooz-tox-py311-redis https://zuul.opendev.org/t/openstack/build/719818403cab49b3b26b03fcc1299cbd : SUCCESS in 5m 44s\n- tooz-tox-py314-redis https://zuul.opendev.org/t/openstack/build/08befe564d3e4cddba93a3e41178a3c3 : SUCCESS in 8m 40s\n- tooz-tox-py311-sentinel https://zuul.opendev.org/t/openstack/build/6cbce18c947243618563eb40cb79796a : SUCCESS in 5m 16s\n- tooz-tox-py314-sentinel https://zuul.opendev.org/t/openstack/build/e887c61163494e2a815eaa95a01a4498 : SUCCESS in 8m 28s\n- tooz-tox-py311-memcached https://zuul.opendev.org/t/openstack/build/463ee988775247ca9b5cb44048cf8eb5 : SUCCESS in 3m 39s\n- tooz-tox-py314-memcached https://zuul.opendev.org/t/openstack/build/44a8cad2869b418b9f3c95b49d84321c : SUCCESS in 5m 17s\n- tooz-tox-py311-postgresql https://zuul.opendev.org/t/openstack/build/56e4cfbeb8f64a77b189325c3d22117b : SUCCESS in 4m 59s\n- tooz-tox-py314-postgresql https://zuul.opendev.org/t/openstack/build/8230ea6c8a9e45d588ff6187c67b9dce : SUCCESS in 8m 58s\n- tooz-tox-py311-mysql https://zuul.opendev.org/t/openstack/build/c7be58074ed543c28c3082170aec35f0 : SUCCESS in 4m 35s\n- tooz-tox-py314-mysql https://zuul.opendev.org/t/openstack/build/7260e877a90345ff8ab6b5b3488e5a65 : SUCCESS in 8m 07s","accounts_in_message":[],"_revision_number":4},{"id":"fd7d0f990f2024846bba81d090125ccd7a4d6bdf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36741,"name":"Eunkyung Kim","display_name":"eunkyung","email":"ek121.kim@samsung.com","username":"eunkyung"},"date":"2026-10-02 01:50:19.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":5},{"id":"ae9911fe29006601450ac9d69042a94a8e4c1014","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-02 02:40:40.000000000","message":"Patch Set 5: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b65474ebdb1042e891b269a4d27256ad\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4754d3b90be6416d97f50c8e1e5aadbc : SUCCESS in 49m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9609303af8164d4fbeb8b3e0d8e37842 : SUCCESS in 3m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/26fd7a7e181842a2b2e550ceaf4520c4 : FAILURE in 3m 56s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c1f608d4f99e41dab155e8a349cd06a4 : SUCCESS in 5m 50s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/08105716e30440ccb729c0b54d948f82 : SUCCESS in 4m 55s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/148c088fb5d64d75bbe6eed5bb0563b2 : SUCCESS in 7m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e64f65eff326410293d39d6d18fba5dd : SUCCESS in 3m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/24aa23dda151496f8c894285c6c4a77d : SUCCESS in 4m 14s\n- tooz-tox-examples https://zuul.opendev.org/t/openstack/build/50806de496f841b984889e6134629602 : SUCCESS in 2m 50s\n- tooz-tox-py311-etcd3gw https://zuul.opendev.org/t/openstack/build/ca8dd3a7dce24a36b58991183e24180e : SUCCESS in 4m 12s\n- tooz-tox-py314-etcd3gw https://zuul.opendev.org/t/openstack/build/ef74e4fb08444be090c657ba86bb888e : SUCCESS in 9m 45s\n- tooz-tox-py311-zookeeper https://zuul.opendev.org/t/openstack/build/d0860fb19e174821a04c6d31e91a8a5c : SUCCESS in 6m 56s\n- tooz-tox-py314-zookeeper https://zuul.opendev.org/t/openstack/build/3c0c59dab9a64b99acf9d5ead81cca6f : SUCCESS in 7m 30s\n- tooz-tox-py311-redis https://zuul.opendev.org/t/openstack/build/aa395fd1e4b143cd810a12f616b5bff7 : SUCCESS in 5m 36s\n- tooz-tox-py314-redis https://zuul.opendev.org/t/openstack/build/6a3da72015ff40d0b103a62062e51786 : SUCCESS in 5m 46s\n- tooz-tox-py311-sentinel https://zuul.opendev.org/t/openstack/build/9c4318a5bc354344aa09bc11a1ba5d74 : SUCCESS in 7m 36s\n- tooz-tox-py314-sentinel https://zuul.opendev.org/t/openstack/build/a1c8c1a357e1452987e099c3cf4021ee : SUCCESS in 7m 58s\n- tooz-tox-py311-memcached https://zuul.opendev.org/t/openstack/build/b330cbe6b89d453c80a6015f96c2f51b : SUCCESS in 4m 38s\n- tooz-tox-py314-memcached https://zuul.opendev.org/t/openstack/build/4f0d7c59893e444da307eddf738d2e4e : SUCCESS in 6m 33s\n- tooz-tox-py311-postgresql https://zuul.opendev.org/t/openstack/build/c341221bfbb64e41855c5ae6e5a19a08 : SUCCESS in 6m 05s\n- tooz-tox-py314-postgresql https://zuul.opendev.org/t/openstack/build/827e91a0d7e04893a67dc0584f121909 : SUCCESS in 6m 45s\n- tooz-tox-py311-mysql https://zuul.opendev.org/t/openstack/build/e144e6b290ee4b0cbb3025935f60ad54 : SUCCESS in 6m 55s\n- tooz-tox-py314-mysql https://zuul.opendev.org/t/openstack/build/b8d68d7f4cbf4b9fb3c8493633385ca5 : SUCCESS in 7m 32s","accounts_in_message":[],"_revision_number":5},{"id":"069e898212d16a217f7b0710c5c1a135ff6368da","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-10-02 12:53:51.000000000","message":"Patch Set 5: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"13be5959294fc0e084f636d2cdb9d8349676392b","revisions":{"16c95c441abfcac5defe7528d4daa01b3495405b":{"kind":"REWORK","_number":1,"created":"2026-04-13 03:40:53.000000000","uploader":{"_account_id":36741,"name":"Eunkyung Kim","display_name":"eunkyung","email":"ek121.kim@samsung.com","username":"eunkyung"},"ref":"refs/changes/92/984292/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/92/984292/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/92/984292/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/92/984292/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/92/984292/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/92/984292/1"}}},"commit":{"parents":[{"commit":"abca4e997f3f0ea0af63db85f7c9e242a4104d37","subject":"typing: Add type hints to tooz.hashring","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/abca4e997f3f0ea0af63db85f7c9e242a4104d37"}]}],"author":{"name":"Eunkyung99","email":"ek121.kim@samsung.com","date":"2026-04-13 03:38:46.000000000","tz":540},"committer":{"name":"Eunkyung99","email":"ek121.kim@samsung.com","date":"2026-04-13 03:40:21.000000000","tz":540},"subject":"zookeeper: Add retry parameters\u0027 parsing logic","message":"zookeeper: Add retry parameters\u0027 parsing logic\n\nTo provide \"connection_retry\" and \"command_retry\" parameters to KazooClient,\nthe options parsed from url config should be passed as a KazooRetry object\nor a dictionary type.\nTherefore, the logic has been added to convert the retry options to\nthe corresponding type and pass it.\n\nCloses-bug: #2148145\nChange-Id: I71ecd8d9e0b8895bbbbdf5ed190b7d548d3d05b2\nSigned-off-by: eunkyung \u003cek121.kim@samsung.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/16c95c441abfcac5defe7528d4daa01b3495405b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/16c95c441abfcac5defe7528d4daa01b3495405b"}]},"branch":"refs/heads/master"},"5ca704470ca83718d9bd617a3181dce6e967390c":{"kind":"REWORK","_number":2,"created":"2026-04-13 03:45:23.000000000","uploader":{"_account_id":36741,"name":"Eunkyung Kim","display_name":"eunkyung","email":"ek121.kim@samsung.com","username":"eunkyung"},"ref":"refs/changes/92/984292/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/92/984292/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/92/984292/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/92/984292/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/92/984292/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/92/984292/2"}}},"commit":{"parents":[{"commit":"abca4e997f3f0ea0af63db85f7c9e242a4104d37","subject":"typing: Add type hints to tooz.hashring","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/abca4e997f3f0ea0af63db85f7c9e242a4104d37"}]}],"author":{"name":"Eunkyung99","email":"ek121.kim@samsung.com","date":"2026-04-13 03:38:46.000000000","tz":540},"committer":{"name":"Eunkyung99","email":"ek121.kim@samsung.com","date":"2026-04-13 03:44:31.000000000","tz":540},"subject":"zookeeper: Add retry parameters\u0027 parsing logic","message":"zookeeper: Add retry parameters\u0027 parsing logic\n\nTo provide \"connection_retry\" and \"command_retry\" parameters to KazooClient,\nthe options parsed from url config should be passed as a KazooRetry object\nor a dictionary type.\nTherefore, the logic has been added to convert the retry options to\nthe corresponding type and pass it.\n\nCloses-bug: #2148145\nChange-Id: I71ecd8d9e0b8895bbbbdf5ed190b7d548d3d05b2\nSigned-off-by: eunkyung \u003cek121.kim@samsung.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/5ca704470ca83718d9bd617a3181dce6e967390c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/5ca704470ca83718d9bd617a3181dce6e967390c"}]},"branch":"refs/heads/master"},"9f0186a8c0017f59a4861e97452e174b685e0749":{"kind":"REWORK","_number":3,"created":"2026-06-10 13:44:43.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/92/984292/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/92/984292/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/92/984292/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/92/984292/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/92/984292/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/92/984292/3"}}},"commit":{"parents":[{"commit":"12e22ef600b547b1636a38f8920910ac1dbc91f9","subject":"Fix mypy errors","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/12e22ef600b547b1636a38f8920910ac1dbc91f9"}]}],"author":{"name":"Eunkyung99","email":"ek121.kim@samsung.com","date":"2026-04-13 03:38:46.000000000","tz":540},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-06-10 13:44:32.000000000","tz":60},"subject":"zookeeper: Add retry parameters\u0027 parsing logic","message":"zookeeper: Add retry parameters\u0027 parsing logic\n\nTo provide \"connection_retry\" and \"command_retry\" parameters to\nKazooClient, the options parsed from url config should be passed as a\nKazooRetry object or a dictionary type. Therefore, the logic has been\nadded to convert the retry options to the corresponding type and pass\nit.\n\nChange-Id: I71ecd8d9e0b8895bbbbdf5ed190b7d548d3d05b2\nSigned-off-by: eunkyung \u003cek121.kim@samsung.com\u003e\nCo-authored-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nCloses-bug: #2148145\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/9f0186a8c0017f59a4861e97452e174b685e0749"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/9f0186a8c0017f59a4861e97452e174b685e0749"}]},"branch":"refs/heads/master"},"6796d5b6c4ba0131f39f018de5702415aff767cb":{"kind":"REWORK","_number":4,"created":"2026-10-01 08:57:15.000000000","uploader":{"_account_id":36741,"name":"Eunkyung Kim","display_name":"eunkyung","email":"ek121.kim@samsung.com","username":"eunkyung"},"ref":"refs/changes/92/984292/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/92/984292/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/92/984292/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/92/984292/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/92/984292/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/92/984292/4"}}},"commit":{"parents":[{"commit":"12e22ef600b547b1636a38f8920910ac1dbc91f9","subject":"Fix mypy errors","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/12e22ef600b547b1636a38f8920910ac1dbc91f9"}]}],"author":{"name":"Eunkyung99","email":"ek121.kim@samsung.com","date":"2026-04-13 03:38:46.000000000","tz":540},"committer":{"name":"Eunkyung99","email":"ek121.kim@samsung.com","date":"2026-10-01 08:56:25.000000000","tz":540},"subject":"zookeeper: Add retry parameters\u0027 parsing logic","message":"zookeeper: Add retry parameters\u0027 parsing logic\n\nTo provide \"connection_retry\" and \"command_retry\" parameters to\nKazooClient, the options parsed from url config should be passed as a\nKazooRetry object or a dictionary type. Therefore, the logic has been\nadded to convert the retry options to the corresponding type and pass\nit.\n\nChange-Id: I71ecd8d9e0b8895bbbbdf5ed190b7d548d3d05b2\nSigned-off-by: eunkyung \u003cek121.kim@samsung.com\u003e\nCo-authored-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nCloses-bug: #2148145\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/6796d5b6c4ba0131f39f018de5702415aff767cb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/6796d5b6c4ba0131f39f018de5702415aff767cb"}]},"branch":"refs/heads/master"},"13be5959294fc0e084f636d2cdb9d8349676392b":{"kind":"REWORK","_number":5,"created":"2026-10-02 01:50:19.000000000","uploader":{"_account_id":36741,"name":"Eunkyung Kim","display_name":"eunkyung","email":"ek121.kim@samsung.com","username":"eunkyung"},"ref":"refs/changes/92/984292/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/92/984292/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/92/984292/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/92/984292/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/92/984292/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/92/984292/5"}}},"commit":{"parents":[{"commit":"12e22ef600b547b1636a38f8920910ac1dbc91f9","subject":"Fix mypy errors","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/12e22ef600b547b1636a38f8920910ac1dbc91f9"}]}],"author":{"name":"Eunkyung99","email":"ek121.kim@samsung.com","date":"2026-04-13 03:38:46.000000000","tz":540},"committer":{"name":"Eunkyung99","email":"ek121.kim@samsung.com","date":"2026-10-02 01:49:31.000000000","tz":540},"subject":"zookeeper: Add retry parameters\u0027 parsing logic","message":"zookeeper: Add retry parameters\u0027 parsing logic\n\nTo provide \"connection_retry\" and \"command_retry\" parameters to\nKazooClient, the options parsed from url config should be passed as a\nKazooRetry object or a dictionary type. Therefore, the logic has been\nadded to convert the retry options to the corresponding type and pass\nit.\n\nChange-Id: I71ecd8d9e0b8895bbbbdf5ed190b7d548d3d05b2\nSigned-off-by: eunkyung \u003cek121.kim@samsung.com\u003e\nCo-authored-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nCloses-bug: #2148145\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/13be5959294fc0e084f636d2cdb9d8349676392b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/13be5959294fc0e084f636d2cdb9d8349676392b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
