)]}'
{"id":"openstack%2Ftooz~984524","triplet_id":"openstack%2Ftooz~master~I7ac9a4353acd94dccc71722eaa086d146e4830ad","project":"openstack/tooz","branch":"master","topic":"typing","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2026-04-23 14:42:00.000000000","reason":"\u003cGERRIT_ACCOUNT_15334\u003e replied on the change","reason_account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}}},"hashtags":[],"change_id":"I7ac9a4353acd94dccc71722eaa086d146e4830ad","subject":"coordination: Change join_group capabilities default","status":"MERGED","created":"2026-04-14 09:45:17.000000000","updated":"2026-04-23 16:52:11.000000000","submitted":"2026-04-23 16:51:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"984524-typing","meta_rev_id":"94af032a1c426f4bd138618c6c9b834fc0ea33ff","_number":984524,"virtual_id_number":984524,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-04-23 16:51:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2026-04-21 09:22:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2026-04-23 14:42:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-14 10:57:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-04-21 09:22:00.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"08a40399ed5a7a90b7ecdb219eb07621764f6047","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-04-14 09:45:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e53ef56127e34c1af6d32aecfa4f095325b1e33d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-14 10:57:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b46ccee9c5aa4caa9938fd13a7ab1bf1\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2c2d717768d04eabbbe89ac963d26cf4 : SUCCESS in 1h 04m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9ef6af783e964087a1d864234903d2b3 : SUCCESS in 6m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9a85e10287a044dbb41a5b80b267121b : SUCCESS in 3m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c3af96bc15d042d0bfe4813ba4b2ca8b : SUCCESS in 4m 12s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/fba09318fa68442d80545b9cfad0573d : SUCCESS in 5m 44s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/4837bce989e94ffc9059d0adc2e92a14 : SUCCESS in 7m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2136db6196744cdab7d4720e80f2a36b : SUCCESS in 3m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/073a55e580044f42b513a601719b54b5 : SUCCESS in 3m 33s\n- tooz-tox-examples https://zuul.opendev.org/t/openstack/build/15c526bf7e244c7e90610413122137e1 : SUCCESS in 5m 44s\n- tooz-tox-py310-etcd3gw https://zuul.opendev.org/t/openstack/build/773a33418338440582b05c98e13da115 : SUCCESS in 4m 38s\n- tooz-tox-py313-etcd3gw https://zuul.opendev.org/t/openstack/build/67f690ad2b9943a7ae03e027a5d179ef : SUCCESS in 5m 52s\n- tooz-tox-py310-zookeeper https://zuul.opendev.org/t/openstack/build/4af77b0a44dd4b29be0bad1941ca4f64 : SUCCESS in 4m 49s\n- tooz-tox-py313-zookeeper https://zuul.opendev.org/t/openstack/build/ec82ad6557544a4c8edd98e13e322b75 : SUCCESS in 5m 15s\n- tooz-tox-py310-redis https://zuul.opendev.org/t/openstack/build/926703df08674d41b8f40ff080784dfc : SUCCESS in 6m 22s\n- tooz-tox-py313-redis https://zuul.opendev.org/t/openstack/build/e1a1ab44e8a7418bae3fbd3972a034cd : SUCCESS in 5m 00s\n- tooz-tox-py310-sentinel https://zuul.opendev.org/t/openstack/build/72c068bddb724f0e8b9d1656a584ef09 : SUCCESS in 6m 43s\n- tooz-tox-py313-sentinel https://zuul.opendev.org/t/openstack/build/9b74cca21f7a492098ea665bf4447db3 : SUCCESS in 5m 10s\n- tooz-tox-py310-memcached https://zuul.opendev.org/t/openstack/build/40b210712cbe43b4aa7f62c58c7b4dfd : SUCCESS in 3m 59s\n- tooz-tox-py313-memcached https://zuul.opendev.org/t/openstack/build/bf137b8f68a641fc865e699ba5f293e7 : SUCCESS in 6m 03s\n- tooz-tox-py310-postgresql https://zuul.opendev.org/t/openstack/build/a0c14092bd7e418b9564bec0711d24ab : SUCCESS in 4m 30s\n- tooz-tox-py313-postgresql https://zuul.opendev.org/t/openstack/build/738e6e2ddead40c493429cbb55c133db : SUCCESS in 5m 35s\n- tooz-tox-py310-mysql https://zuul.opendev.org/t/openstack/build/e0231610a4154c2a825c523aaf15b2bd : SUCCESS in 5m 22s\n- tooz-tox-py313-mysql https://zuul.opendev.org/t/openstack/build/f897b27b102b40d89a325a4aa917effa : SUCCESS in 5m 17s\n- tooz-tox-py310-consul https://zuul.opendev.org/t/openstack/build/6ac8a5faa1d746d38e7207963295f71c : SUCCESS in 4m 36s\n- tooz-tox-py313-consul https://zuul.opendev.org/t/openstack/build/10facc1f45624e84a8a860b57dec644b : SUCCESS in 5m 52s","accounts_in_message":[],"_revision_number":1},{"id":"6049d16826025186a4d641b5a06365c053eb3e5c","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2026-04-21 09:22:00.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"2582daed8da7b51131b0df3aa20da95817936a9a","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-04-23 14:42:00.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f30328aeccdf31462d096d37f4bcc1d6499b667d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-23 14:42:34.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2f11715774866c5b4f5ff653205eea3d20c636e0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-23 16:51:19.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/63b59de4621342ffa9eaed6d4504430b\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c1cf6ce50dbf4dd996e232a5d6a20df2 : SUCCESS in 2h 02m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5a6084a255de4240a82a4e64d681f6bf : SUCCESS in 5m 22s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/11459d895f31485db00912aa7d6c3405 : SUCCESS in 5m 11s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/748ec5b805b94cb18af13ad9e8521d08 : SUCCESS in 7m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f2107afd8c44768b8652959c03806b8 : SUCCESS in 3m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/511d4188c5be4d9bad4c235e2c2e4d2c : SUCCESS in 4m 54s\n- tooz-tox-examples https://zuul.opendev.org/t/openstack/build/97b5b3bca9b647679030deb83fe80a90 : SUCCESS in 3m 24s\n- tooz-tox-py310-etcd3gw https://zuul.opendev.org/t/openstack/build/aa7e4efb34d64b36afa3e18553e423e3 : SUCCESS in 3m 48s\n- tooz-tox-py313-etcd3gw https://zuul.opendev.org/t/openstack/build/4e95e1103c7e4a5780c292cae58c38e0 : SUCCESS in 6m 44s\n- tooz-tox-py310-zookeeper https://zuul.opendev.org/t/openstack/build/28304f27756d4cd595679d5729c24774 : SUCCESS in 4m 19s\n- tooz-tox-py313-zookeeper https://zuul.opendev.org/t/openstack/build/1558b23dd4aa489eb268ae772d54cf39 : SUCCESS in 12m 45s\n- tooz-tox-py310-redis https://zuul.opendev.org/t/openstack/build/40095b41ca594180a7a4a6cbf0e8d24f : SUCCESS in 5m 40s\n- tooz-tox-py313-redis https://zuul.opendev.org/t/openstack/build/b766651d938b4db8af919b03e061d47c : SUCCESS in 8m 10s\n- tooz-tox-py310-sentinel https://zuul.opendev.org/t/openstack/build/af4923bb5cf6403fb6b0c2e19f52e4b0 : SUCCESS in 5m 15s\n- tooz-tox-py313-sentinel https://zuul.opendev.org/t/openstack/build/08c2145e8f7e4a1cb6a9364e69f423da : SUCCESS in 8m 30s\n- tooz-tox-py310-memcached https://zuul.opendev.org/t/openstack/build/1d14b8c06cfe462c958d03b038c617b1 : SUCCESS in 4m 01s\n- tooz-tox-py313-memcached https://zuul.opendev.org/t/openstack/build/1b7b0e295de34f2b88df2d7e4db4438a : SUCCESS in 3m 41s\n- tooz-tox-py310-postgresql https://zuul.opendev.org/t/openstack/build/61a1b55db9d3455ba75446ed0ee7fd74 : SUCCESS in 4m 22s\n- tooz-tox-py313-postgresql https://zuul.opendev.org/t/openstack/build/5c729a45c2294433b3fcd8384fb00359 : SUCCESS in 5m 14s\n- tooz-tox-py310-mysql https://zuul.opendev.org/t/openstack/build/11c5427c5f8f4bbfaf8b8371a1adb8d1 : SUCCESS in 3m 09s\n- tooz-tox-py313-mysql https://zuul.opendev.org/t/openstack/build/d9603b8bf3b742f98d00be1c91b992c6 : SUCCESS in 6m 16s\n- tooz-tox-py310-consul https://zuul.opendev.org/t/openstack/build/62afbf3a060c4857a9222669d2e376e6 : SUCCESS in 4m 32s\n- tooz-tox-py313-consul https://zuul.opendev.org/t/openstack/build/231883c8a9f842108d39e6efc7b737b8 : SUCCESS in 5m 55s","accounts_in_message":[],"_revision_number":1},{"id":"ee9c409e9cee66461f403aa5a4d703001ef7a292","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-23 16:51:19.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"94af032a1c426f4bd138618c6c9b834fc0ea33ff","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-23 16:52:11.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/53be36b1c558462bba12f2aed8db95db\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1923c710d41348de9aa279baa2a070bb : SUCCESS in 40s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c3f86cccc28b49529721aa50e531a8dc : SUCCESS in 39s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e71185216f432d66f5f25844df329715bf4a08cd","revisions":{"e71185216f432d66f5f25844df329715bf4a08cd":{"kind":"REWORK","_number":1,"created":"2026-04-14 09:45:17.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/24/984524/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/24/984524/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/24/984524/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/24/984524/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/24/984524/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/24/984524/1"}}},"commit":{"parents":[{"commit":"abca4e997f3f0ea0af63db85f7c9e242a4104d37","subject":"typing: Add type hints to tooz.hashring","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/abca4e997f3f0ea0af63db85f7c9e242a4104d37"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-04-14 09:36:53.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-04-14 09:44:48.000000000","tz":60},"subject":"coordination: Change join_group capabilities default","message":"coordination: Change join_group capabilities default\n\nThis defaulted to an empty binary string, but the only in-tree user we\nhave (join_partitioned_group, which calls join_group_create, which calls\njoin_group) called it with a dictionary, which seems to be a fairly sane\nuse of the API. None makes more sense in this instance. This should have\nno breaking impact since None is serializable and both are false\u0027y\nvalues. A (potentially unnecessary) release note is included to this\neffect.\n\nDoing this allows us to present a more consistent API to users and\noffers a path to eventually insist on dictionary values for\ncapabilities.\n\nChange-Id: I7ac9a4353acd94dccc71722eaa086d146e4830ad\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/e71185216f432d66f5f25844df329715bf4a08cd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/e71185216f432d66f5f25844df329715bf4a08cd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
