)]}'
{"id":"openstack%2Ftooz~986261","triplet_id":"openstack%2Ftooz~master~I310a8a93491bc82a1a11cd314c12d493e5de82a0","project":"openstack/tooz","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-04-27 18:36:58.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I310a8a93491bc82a1a11cd314c12d493e5de82a0","subject":"tox: Use new constraints option","status":"MERGED","created":"2026-04-25 15:52:57.000000000","updated":"2026-04-27 18:38:01.000000000","submitted":"2026-04-27 18:36:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"986261","meta_rev_id":"7c92457a0c5f7b60d86aef55daa6904402793c0a","_number":986261,"virtual_id_number":986261,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-04-27 18:36:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2026-04-27 15:15:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2026-04-27 15:15:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-25 17:40:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-04-27 15:15:17.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"}],"messages":[{"id":"3018dc9a2b1c16ae5749cf222cfedbb60d14b68b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-04-25 15:52:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"563cb7949dcede47253dbd919267b979b1f289ee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-25 17:40:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/115a815aa95b4a72b64dff24beee57a3\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c1a0fcef7d3e47d0b4fcac11d902d6d3 : SUCCESS in 1h 42m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6f2ba78eecbe4e518b864d3f3bbdf44e : SUCCESS in 5m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/45d3cc757d9f483d8672e32acfccb8f3 : SUCCESS in 5m 08s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e36f55935ab04a3bb93c313ba25a9eb3 : SUCCESS in 3m 27s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e2fd5c7d0c7b4a858c41dd2fef5399b5 : SUCCESS in 4m 57s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/d3d7f93fa0cc460088a1e235fa89735c : SUCCESS in 4m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8041d4595ce449bb909a32af20680b6a : SUCCESS in 3m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c4255d2a27fb4971b6edaab23a0dfa54 : SUCCESS in 3m 35s\n- tooz-tox-examples https://zuul.opendev.org/t/openstack/build/9ce84859825c42769532fafe84dc1cef : SUCCESS in 4m 33s\n- tooz-tox-py310-etcd3gw https://zuul.opendev.org/t/openstack/build/00fe4d230373419b914f71a65577b432 : SUCCESS in 4m 43s\n- tooz-tox-py313-etcd3gw https://zuul.opendev.org/t/openstack/build/df382b4d836a4e42b7e0867c68ebaaba : SUCCESS in 4m 40s\n- tooz-tox-py310-zookeeper https://zuul.opendev.org/t/openstack/build/954492a6cee6466fba84004e8c6ab0e2 : SUCCESS in 4m 52s\n- tooz-tox-py313-zookeeper https://zuul.opendev.org/t/openstack/build/aa80e84e533a4ae796675f021115e629 : SUCCESS in 4m 35s\n- tooz-tox-py310-redis https://zuul.opendev.org/t/openstack/build/fd891ec8714d4a1fb7bbb862106932ce : SUCCESS in 4m 59s\n- tooz-tox-py313-redis https://zuul.opendev.org/t/openstack/build/b010b5df38054513baf97e67f668d76b : SUCCESS in 6m 17s\n- tooz-tox-py310-sentinel https://zuul.opendev.org/t/openstack/build/0626b7a95f6a44ef9148780dba3bf77d : SUCCESS in 5m 17s\n- tooz-tox-py313-sentinel https://zuul.opendev.org/t/openstack/build/d7c927c27e8c4ce4b9e6491e787428b6 : SUCCESS in 5m 43s\n- tooz-tox-py310-memcached https://zuul.opendev.org/t/openstack/build/8c14c991fae14e6398864c22cf15f0cc : SUCCESS in 4m 16s\n- tooz-tox-py313-memcached https://zuul.opendev.org/t/openstack/build/e095762e626e4c15945e2155a646c8ad : SUCCESS in 5m 26s\n- tooz-tox-py310-postgresql https://zuul.opendev.org/t/openstack/build/fbd629d3f9ed4d8290370a7afeeb8004 : SUCCESS in 2m 57s\n- tooz-tox-py313-postgresql https://zuul.opendev.org/t/openstack/build/95bd8d38d3e44f12b91e1328f795e30a : SUCCESS in 5m 10s\n- tooz-tox-py310-mysql https://zuul.opendev.org/t/openstack/build/4873f908d8974a568d697315c5659a73 : SUCCESS in 4m 44s\n- tooz-tox-py313-mysql https://zuul.opendev.org/t/openstack/build/37edf42981d04960b5c3f8fab5b4ac71 : SUCCESS in 7m 24s\n- tooz-tox-py310-consul https://zuul.opendev.org/t/openstack/build/7f5036d795ea41ec97b117756636ad38 : SUCCESS in 4m 24s\n- tooz-tox-py313-consul https://zuul.opendev.org/t/openstack/build/4d4788783c3a4a5ead2dd2a498c105b2 : SUCCESS in 4m 52s","accounts_in_message":[],"_revision_number":1},{"id":"507b29e646f35053bb7e0a39dbbc90c8c70197d0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-04-27 14:25:33.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"2b0068bff721e5970d6e2b4af1453b993eda9357","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-04-27 15:15:17.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0818a7eb92ae8dcadf9bbf5323594aa2441e3872","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-27 16:29:03.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a73b73ed4df448898d72b6b217ce026a\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4b5406452a4e48cc847766449f9689e2 : SUCCESS in 1h 59m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/426904b20d2341af85a868e6aa6ef658 : SUCCESS in 5m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fe0082f6425647569fbcd47c93fb6623 : SUCCESS in 5m 22s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8af41dcfe26342fc8b27bb38fe820433 : SUCCESS in 7m 37s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2b867babd5b94271b14253162abd6f83 : SUCCESS in 5m 27s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/041063ef11d245c38f17880ae955e58c : FAILURE in 9m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/32128f5fb3444c19a5fb80c2d7fd271e : SUCCESS in 5m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3ee4c9c7f0af4a87b76a9b6600251501 : SUCCESS in 3m 43s\n- tooz-tox-examples https://zuul.opendev.org/t/openstack/build/f87ca169e8884afcb35b2452b3c1a20d : SUCCESS in 4m 48s\n- tooz-tox-py310-etcd3gw https://zuul.opendev.org/t/openstack/build/0e522c8db6a54dc5a91c9d38adc8bce8 : SUCCESS in 4m 40s\n- tooz-tox-py313-etcd3gw https://zuul.opendev.org/t/openstack/build/dc13896e7d9e406fa0a767fbe7f60e0c : SUCCESS in 7m 17s\n- tooz-tox-py310-zookeeper https://zuul.opendev.org/t/openstack/build/4ff00f58fc114de2addb25105e2b24e8 : SUCCESS in 6m 03s\n- tooz-tox-py313-zookeeper https://zuul.opendev.org/t/openstack/build/283e8c26d8bc4aeb974a980b7f8d1b97 : SUCCESS in 4m 56s\n- tooz-tox-py310-redis https://zuul.opendev.org/t/openstack/build/5095d9e105d04eaeb6f809c7e27a1b9c : SUCCESS in 5m 40s\n- tooz-tox-py313-redis https://zuul.opendev.org/t/openstack/build/c6b8f72570c943968ec0556ccd333e7e : SUCCESS in 6m 39s\n- tooz-tox-py310-sentinel https://zuul.opendev.org/t/openstack/build/e0cd742e458b44e2882cab79656e1729 : SUCCESS in 6m 33s\n- tooz-tox-py313-sentinel https://zuul.opendev.org/t/openstack/build/701eb66ae0cf40a0be1bc7bb5fe352a8 : SUCCESS in 10m 32s\n- tooz-tox-py310-memcached https://zuul.opendev.org/t/openstack/build/7912f7944e5f4c87a82c8721ecacac86 : SUCCESS in 4m 42s\n- tooz-tox-py313-memcached https://zuul.opendev.org/t/openstack/build/0e46a2fda8ae4d8286c4a60f5dacf2b5 : SUCCESS in 5m 16s\n- tooz-tox-py310-postgresql https://zuul.opendev.org/t/openstack/build/6602f29b3fc94e67b5c8adb2a2c6bfff : SUCCESS in 5m 59s\n- tooz-tox-py313-postgresql https://zuul.opendev.org/t/openstack/build/a3dec69749db46c1839757255575cff9 : SUCCESS in 5m 06s\n- tooz-tox-py310-mysql https://zuul.opendev.org/t/openstack/build/82fff9f71836400ba60038c39752794e : SUCCESS in 6m 07s\n- tooz-tox-py313-mysql https://zuul.opendev.org/t/openstack/build/c5681530a5174dc089f99bb30cd6dd2d : SUCCESS in 6m 54s\n- tooz-tox-py310-consul https://zuul.opendev.org/t/openstack/build/0e66cb3249a44acebb5d041d83939428 : SUCCESS in 4m 31s\n- tooz-tox-py313-consul https://zuul.opendev.org/t/openstack/build/07b78f32486e412a97e7940634a59da7 : SUCCESS in 5m 34s","accounts_in_message":[],"_revision_number":2},{"id":"027f5f4ed42aa0060f68c9c315ba9b1fe88d1de6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-27 16:29:28.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"289c09fd842afdadd53c59d468c8f38c1ec0d129","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-27 18:36:58.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8c6667ecd0044eeeb13e9165cc750c04\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6fff803715cd460eb2f949d6ea409ad6 : SUCCESS in 2h 02m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6dbabddf81114f8aa8aae40a18091dca : SUCCESS in 3m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3b24bd9e55ad40ee821515f0579ef5a7 : SUCCESS in 5m 00s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/92093ad15f97478a86a4a3a8c102dbcf : SUCCESS in 5m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5705f62dec184e3b97c374100afccb83 : SUCCESS in 5m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7fb4ccc0ceb440e5989e0f77bf12d93a : SUCCESS in 4m 16s\n- tooz-tox-examples https://zuul.opendev.org/t/openstack/build/c2ecb2e58f444bb49898b41647d9011d : SUCCESS in 5m 30s\n- tooz-tox-py310-etcd3gw https://zuul.opendev.org/t/openstack/build/85484140e5ff4469a1f560f371d5eb32 : SUCCESS in 5m 06s\n- tooz-tox-py313-etcd3gw https://zuul.opendev.org/t/openstack/build/f4180314dc5445d2a2ff0f1e1846a271 : SUCCESS in 8m 12s\n- tooz-tox-py310-zookeeper https://zuul.opendev.org/t/openstack/build/44caa5c7d17f41d490e80bca4b834f91 : SUCCESS in 5m 06s\n- tooz-tox-py313-zookeeper https://zuul.opendev.org/t/openstack/build/8e973040d0b145a392a451b5bcc058b5 : SUCCESS in 3m 47s\n- tooz-tox-py310-redis https://zuul.opendev.org/t/openstack/build/f9ea077a15644c079f9c32eed74d9967 : SUCCESS in 5m 27s\n- tooz-tox-py313-redis https://zuul.opendev.org/t/openstack/build/4443fa3739db4f43959e99482849667b : SUCCESS in 5m 49s\n- tooz-tox-py310-sentinel https://zuul.opendev.org/t/openstack/build/8410a16a46944fe58ca3b3411ed0fbef : SUCCESS in 6m 00s\n- tooz-tox-py313-sentinel https://zuul.opendev.org/t/openstack/build/8ed78d3c889846ecaec051e785ed0346 : SUCCESS in 5m 11s\n- tooz-tox-py310-memcached https://zuul.opendev.org/t/openstack/build/18684e129ba34f348079d351fe8ed783 : SUCCESS in 4m 40s\n- tooz-tox-py313-memcached https://zuul.opendev.org/t/openstack/build/65478765f7144ff1876a9986453c5e00 : SUCCESS in 5m 46s\n- tooz-tox-py310-postgresql https://zuul.opendev.org/t/openstack/build/41b0b658ef7b498591290d370cb9ebea : SUCCESS in 5m 41s\n- tooz-tox-py313-postgresql https://zuul.opendev.org/t/openstack/build/e99d0a6c0aa24832aa43512c12e9f53c : SUCCESS in 3m 03s\n- tooz-tox-py310-mysql https://zuul.opendev.org/t/openstack/build/d4cd4ca366d24aca92671754e65a0277 : SUCCESS in 4m 12s\n- tooz-tox-py313-mysql https://zuul.opendev.org/t/openstack/build/2b5d0cb4a3d24614b35f3d11b7d7734b : SUCCESS in 4m 49s\n- tooz-tox-py310-consul https://zuul.opendev.org/t/openstack/build/ad5b3f02eb0f46f688d2f20558b60db6 : SUCCESS in 4m 23s\n- tooz-tox-py313-consul https://zuul.opendev.org/t/openstack/build/9d45f10feb0b4d299c85720ca730c761 : SUCCESS in 5m 01s","accounts_in_message":[],"_revision_number":2},{"id":"5702312ff1c6d0eadfd51bd347d380bd33b39b98","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-27 18:36:58.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"7c92457a0c5f7b60d86aef55daa6904402793c0a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-27 18:38:01.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3fab0e6ae11e4eb9baeee15cf5d323a3\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1aaf8f4d2cea4f819a2b7f75ed705b56 : SUCCESS in 49s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/94cecbb9231d447c8581054cbf55d48e : SUCCESS in 46s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"951ad95e5ff070a17c227aad5007e1b416ab41b8","revisions":{"44a2c6aa5c565b014e0233d69ebf93a7c81f5d10":{"kind":"REWORK","_number":1,"created":"2026-04-25 15:52:57.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/61/986261/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/61/986261/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/61/986261/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/61/986261/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/61/986261/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/61/986261/1"}}},"commit":{"parents":[{"commit":"abca4e997f3f0ea0af63db85f7c9e242a4104d37","subject":"typing: Add type hints to tooz.hashring","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/abca4e997f3f0ea0af63db85f7c9e242a4104d37"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-04-25 15:52:40.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-04-25 15:52:40.000000000","tz":540},"subject":"tox: Use new constraints option","message":"tox: Use new constraints option\n\nIt is a preferred way to define the constraints file.\n\nAlso fix inconsistent indentations.\n\nChange-Id: I310a8a93491bc82a1a11cd314c12d493e5de82a0\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/44a2c6aa5c565b014e0233d69ebf93a7c81f5d10"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/44a2c6aa5c565b014e0233d69ebf93a7c81f5d10"}]},"branch":"refs/heads/master"},"951ad95e5ff070a17c227aad5007e1b416ab41b8":{"kind":"REWORK","_number":2,"created":"2026-04-27 14:25:33.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/61/986261/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/61/986261/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/61/986261/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/61/986261/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/61/986261/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/61/986261/2"}}},"commit":{"parents":[{"commit":"ad84459cebf7cff3c505d7fb553f718b7acabba7","subject":"Merge \"Remove etcd driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/ad84459cebf7cff3c505d7fb553f718b7acabba7"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-04-25 15:52:40.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-04-27 14:25:19.000000000","tz":540},"subject":"tox: Use new constraints option","message":"tox: Use new constraints option\n\nIt is a preferred way to define the constraints file.\n\nAlso fix inconsistent indentations.\n\nChange-Id: I310a8a93491bc82a1a11cd314c12d493e5de82a0\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/951ad95e5ff070a17c227aad5007e1b416ab41b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/951ad95e5ff070a17c227aad5007e1b416ab41b8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
