)]}'
{"id":"openstack%2Ftooz~992606","triplet_id":"openstack%2Ftooz~master~Iddba39cd5d29a21a3d2d0430cb8b9d7157bf2748","project":"openstack/tooz","branch":"master","topic":"drop-python-310","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2026-06-30 04:40:08.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iddba39cd5d29a21a3d2d0430cb8b9d7157bf2748","subject":"Remove unnecessary debtcollector dependency","status":"MERGED","created":"2026-06-10 10:09:26.000000000","updated":"2026-06-30 04:41:08.000000000","submitted":"2026-06-30 04:40:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"992606-drop-python-310","meta_rev_id":"6565ad50e1fc809ead77d77b7db00880c5b58f03","_number":992606,"virtual_id_number":992606,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-06-30 04:40:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2026-06-30 02:46:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":1,"date":"2026-06-30 02:46:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-10 11:19:21.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2026-06-10 11:56:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"11e583d5cb83bd783f895c39127ccdbe90d3a7c0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-06-10 10:09:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"400e4e51620dadb4136e542d12074ee8073dfedd","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-06-10 11:19:21.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"10f7291e66ca53e7f8bf44ec622c831cda6250cb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-10 11:56:49.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7d884247590c40daab41d108f337ab9b\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/827de76979884b2788d8f1fa2174ccdc : SUCCESS in 3m 12s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/084be1ae064941a08a0df92132b91b88 : SUCCESS in 1h 44m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/385afd3ef70440c286bec741e1886415 : SUCCESS in 6m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0a92cdbb64c94755ba3cd8973f463e01 : FAILURE in 7m 46s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ee7f1ea9e2f347d58b9745205f147ee3 : SUCCESS in 7m 30s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a375b4057e944a55ab6949aa58c299b9 : SUCCESS in 5m 56s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/65d4c6f7058c4e6a98b97f4856df607e : FAILURE in 6m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/383ce5e3eac947338a4eafea2e5dde93 : SUCCESS in 4m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bec7c65297ae43c18e658bacfa2a309d : SUCCESS in 5m 30s\n- tooz-tox-examples https://zuul.opendev.org/t/openstack/build/614c607ca2824d849af8be28a05978d2 : SUCCESS in 5m 02s\n- tooz-tox-py311-etcd3gw https://zuul.opendev.org/t/openstack/build/ad4699d4ab3a4016b0c447f719815e05 : SUCCESS in 4m 52s\n- tooz-tox-py314-etcd3gw https://zuul.opendev.org/t/openstack/build/0ea41352394a412bb99ab13fcc856cd3 : SUCCESS in 8m 05s\n- tooz-tox-py311-zookeeper https://zuul.opendev.org/t/openstack/build/b036be8826be4deca4e3e5271f5b6a45 : SUCCESS in 5m 28s\n- tooz-tox-py314-zookeeper https://zuul.opendev.org/t/openstack/build/54dfddad414b472d9f24edbd199bf5db : SUCCESS in 7m 37s\n- tooz-tox-py311-redis https://zuul.opendev.org/t/openstack/build/f06a9707cfc74022888d51ec17f39e85 : SUCCESS in 6m 00s\n- tooz-tox-py314-redis https://zuul.opendev.org/t/openstack/build/3283b8153f014ba0a7e7b755ae6fc7fb : SUCCESS in 6m 55s\n- tooz-tox-py311-sentinel https://zuul.opendev.org/t/openstack/build/fb4fd3c8fbd7418593b14f47ec682046 : SUCCESS in 5m 21s\n- tooz-tox-py314-sentinel https://zuul.opendev.org/t/openstack/build/1a7a7e85dbbf495db4b7d2a35fcf592d : SUCCESS in 9m 11s\n- tooz-tox-py311-memcached https://zuul.opendev.org/t/openstack/build/958a4962fd4a4ae6b602633493224b11 : SUCCESS in 4m 09s\n- tooz-tox-py314-memcached https://zuul.opendev.org/t/openstack/build/5f0c7d86e069423ebf89612d6fbf0c7e : SUCCESS in 7m 51s\n- tooz-tox-py311-postgresql https://zuul.opendev.org/t/openstack/build/ab361b99507141059c547f11e4c3fb5f : SUCCESS in 4m 51s\n- tooz-tox-py314-postgresql https://zuul.opendev.org/t/openstack/build/16edcef14b19438faac1251df42da004 : SUCCESS in 9m 32s\n- tooz-tox-py311-mysql https://zuul.opendev.org/t/openstack/build/c9d5608edebc4b5fa6e977bb70148405 : FAILURE in 4m 41s\n- tooz-tox-py314-mysql https://zuul.opendev.org/t/openstack/build/2046b9cbba954e4f9a45180e92ae36a4 : SUCCESS in 8m 13s\n- tooz-tox-py311-consul https://zuul.opendev.org/t/openstack/build/4e701f2a61794bfb97599b7212a9f3a2 : SUCCESS in 4m 50s\n- tooz-tox-py314-consul https://zuul.opendev.org/t/openstack/build/f91dd01fb5004e9bb3ca2c5f61bd9e74 : SUCCESS in 6m 41s","accounts_in_message":[],"_revision_number":1},{"id":"70b68501f472aabd755f19ef78d9de3421426a31","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-06-10 12:40:49.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"2350173751fd376184969731ab28b44ca3f1eaf9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-10 14:30:53.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7c7ee9f8f0a44e798266748bd58b891e\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/cb6474f8da3342dbb57e249c755392f4 : SUCCESS in 6m 46s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/83fbd5f6530e404d9b512ff39a5551a4 : SUCCESS in 1h 39m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e5e43bc8c86c42ae84e5b06584ff0e05 : SUCCESS in 6m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/189252a17f694b8496aeeefcecf0c948 : SUCCESS in 7m 23s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4a7ab27ede00485489e9be208386feac : SUCCESS in 4m 13s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/98645f0ca2e44a1c90edc21eddbb7d55 : SUCCESS in 5m 42s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/b3172b1c3dc84226b0775ea1ffdf9ad3 : SUCCESS in 7m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d8559593b384dd799cb20ef63834d30 : SUCCESS in 5m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/321b180845594b0e85436f865c0f9739 : SUCCESS in 5m 40s\n- tooz-tox-examples https://zuul.opendev.org/t/openstack/build/519cb46944b24c95993b897e95fe2e1f : SUCCESS in 6m 09s\n- tooz-tox-py311-etcd3gw https://zuul.opendev.org/t/openstack/build/6a2f250419e442bcbd12c91e866628ae : SUCCESS in 8m 34s\n- tooz-tox-py314-etcd3gw https://zuul.opendev.org/t/openstack/build/2bf26c2052d444089779deed8a99cac5 : SUCCESS in 6m 34s\n- tooz-tox-py311-zookeeper https://zuul.opendev.org/t/openstack/build/57c16fdd040c429dbdacf012e3f148ed : SUCCESS in 6m 55s\n- tooz-tox-py314-zookeeper https://zuul.opendev.org/t/openstack/build/d5ff58e9f1df41049058b8dc907ff206 : FAILURE in 8m 33s\n- tooz-tox-py311-redis https://zuul.opendev.org/t/openstack/build/318517e7f1e1487eb6191b9c9ce32143 : SUCCESS in 6m 48s\n- tooz-tox-py314-redis https://zuul.opendev.org/t/openstack/build/3ddfbf05242f49f98608e0c545ea10f4 : SUCCESS in 6m 25s\n- tooz-tox-py311-sentinel https://zuul.opendev.org/t/openstack/build/3af3fa576506478fa9aa6c3719884c61 : SUCCESS in 6m 49s\n- tooz-tox-py314-sentinel https://zuul.opendev.org/t/openstack/build/f5b6420bbd0f47ab88886f59c15d5cee : SUCCESS in 8m 41s\n- tooz-tox-py311-memcached https://zuul.opendev.org/t/openstack/build/0ec4e56a6d1f4e21937ab502dbbfe9e9 : SUCCESS in 3m 54s\n- tooz-tox-py314-memcached https://zuul.opendev.org/t/openstack/build/984c2f85da4f4e6fa4906234e4d92913 : SUCCESS in 8m 22s\n- tooz-tox-py311-postgresql https://zuul.opendev.org/t/openstack/build/41c3a794ebf54a5dadb008add59f8f49 : SUCCESS in 4m 23s\n- tooz-tox-py314-postgresql https://zuul.opendev.org/t/openstack/build/41027ae60e3b43369e50724d9cdebc03 : FAILURE in 4m 42s\n- tooz-tox-py311-mysql https://zuul.opendev.org/t/openstack/build/539cee2629ed4108867dc5ae9884ef46 : FAILURE in 5m 20s\n- tooz-tox-py314-mysql https://zuul.opendev.org/t/openstack/build/ba41744c25fc455cad85ae6f081bd87a : SUCCESS in 8m 59s\n- tooz-tox-py311-consul https://zuul.opendev.org/t/openstack/build/4f3544278aa74218ac1c36a253579c39 : SUCCESS in 4m 54s\n- tooz-tox-py314-consul https://zuul.opendev.org/t/openstack/build/0a6fd1c9c83e4c47b0d2a26eb5321446 : SUCCESS in 8m 54s","accounts_in_message":[],"_revision_number":2},{"id":"9123efb2e31ca0664710dcbb3c5bebf24e4ca4b3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"real_author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-06-18 10:44:05.000000000","message":"Patch Set 3: Patch Set 2 was rebased on behalf of \u003cGERRIT_ACCOUNT_15334\u003e\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}],"_revision_number":3},{"id":"8dec87f26355a65811ca9fcfff9548a6ce288d58","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-18 13:06:10.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6c62773c6a7c4a89ba7dece46dabd6db\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/a27874446b734ec2877794faef6c2b20 : SUCCESS in 2m 10s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a74086c7f13d4357966e282a9eef96ed : SUCCESS in 2h 11m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fe641fbc854748f8a405d12c3a4c308b : SUCCESS in 6m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5dea0b86666746279b0d10587149cd7b : SUCCESS in 5m 55s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/887e0c64b9bb48f390800c1889849f70 : SUCCESS in 4m 18s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/115651a56bf840428bacf312b9945050 : SUCCESS in 5m 42s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/d457cec063334d5bbef44da62a81e791 : SUCCESS in 8m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e7fcb22fff4744ada7ddb33d8146a656 : SUCCESS in 5m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3300d398c4f74104b491f86ddb417f64 : SUCCESS in 5m 07s\n- tooz-tox-examples https://zuul.opendev.org/t/openstack/build/24a068e147a74167a3c5680c4a152bfe : SUCCESS in 6m 01s\n- tooz-tox-py311-etcd3gw https://zuul.opendev.org/t/openstack/build/a57765a7617f4f799dd801d2bc2ec0ea : SUCCESS in 5m 42s\n- tooz-tox-py314-etcd3gw https://zuul.opendev.org/t/openstack/build/81c88ef6a40a424182b62bf7c4cff1b4 : FAILURE in 5m 45s\n- tooz-tox-py311-zookeeper https://zuul.opendev.org/t/openstack/build/cf01dec73ab84c90bbcbdb04b2558ead : SUCCESS in 5m 11s\n- tooz-tox-py314-zookeeper https://zuul.opendev.org/t/openstack/build/67d2a586a9644d639f8cf59d4017a416 : FAILURE in 14m 35s\n- tooz-tox-py311-redis https://zuul.opendev.org/t/openstack/build/9fedba64476b4c639eea4914dc7c8142 : SUCCESS in 6m 42s\n- tooz-tox-py314-redis https://zuul.opendev.org/t/openstack/build/ea8d6b04c59d420a9892b7e1d7551ea0 : SUCCESS in 12m 03s\n- tooz-tox-py311-sentinel https://zuul.opendev.org/t/openstack/build/16544d9735654b999cfcbec4a03e905a : SUCCESS in 6m 25s\n- tooz-tox-py314-sentinel https://zuul.opendev.org/t/openstack/build/c99191e38ea94dd1b1caa75bb01ee6e5 : FAILURE in 5m 31s\n- tooz-tox-py311-memcached https://zuul.opendev.org/t/openstack/build/4e94c5782ef3480fa61868a0392ef308 : SUCCESS in 6m 38s\n- tooz-tox-py314-memcached https://zuul.opendev.org/t/openstack/build/ba40cfdddebe4f7ca8efde2e6eabe85d : SUCCESS in 5m 12s\n- tooz-tox-py311-postgresql https://zuul.opendev.org/t/openstack/build/ce7041746b3a4583a6b3be0395888237 : SUCCESS in 4m 36s\n- tooz-tox-py314-postgresql https://zuul.opendev.org/t/openstack/build/b12d8f5d0fbd404f809fa0d8d7431b5d : SUCCESS in 7m 31s\n- tooz-tox-py311-mysql https://zuul.opendev.org/t/openstack/build/41a78652dda34ef9acc52a61d4629c56 : SUCCESS in 7m 52s\n- tooz-tox-py314-mysql https://zuul.opendev.org/t/openstack/build/d7dc59178c4b4e3e8a5218d117af7e24 : SUCCESS in 8m 03s\n- tooz-tox-py311-consul https://zuul.opendev.org/t/openstack/build/e947313756a4452183ab15122241b601 : SUCCESS in 6m 02s\n- tooz-tox-py314-consul https://zuul.opendev.org/t/openstack/build/42230479b4bd4e34a8371a34e4148acf : SUCCESS in 8m 28s","accounts_in_message":[],"_revision_number":3},{"id":"a83204e55462ed7fa88ae3d06aae59d416e51908","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-06-29 14:12:51.000000000","message":"Patch Set 4: Patch Set 3 was rebased\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"c53d921c33b2deae647ec28c1fdd429423b7a62b","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-06-29 14:12:57.000000000","message":"Patch Set 4: Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"4941c30fc6292f61b123b86584db858b2d8eee9a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-29 15:45:58.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b90e5f1e6a294f94a1fb6d9080bd5c5c\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/75f76f9b6f514ba6be049915032ccaed : SUCCESS in 3m 20s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/45dcacfa88fa4c7088a790b42f8d41dd : SUCCESS in 1h 29m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/89d541b1496747a2acd249c42665a7fa : SUCCESS in 5m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a5f3a3563aa24692981f6878ee5159c8 : SUCCESS in 3m 42s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c20ab49801f643b88bbbf2c5bac05496 : SUCCESS in 5m 58s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/72d7b9ad9c4a4f32945ab9013b178635 : SUCCESS in 5m 15s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/f4541064474b413cae514e89391589d4 : SUCCESS in 7m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/993b486cbc444f87a1c2b2fb20cea123 : SUCCESS in 3m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a4b4cb38867c47fa9cd54289780164c8 : SUCCESS in 5m 12s\n- tooz-tox-examples https://zuul.opendev.org/t/openstack/build/1e8b0208f37e479992c22594fa3b4c85 : SUCCESS in 4m 52s\n- tooz-tox-py311-etcd3gw https://zuul.opendev.org/t/openstack/build/8e18ef884f46489b98894afbe6f2aa6d : SUCCESS in 5m 46s\n- tooz-tox-py314-etcd3gw https://zuul.opendev.org/t/openstack/build/2c1517a6ef834b628d3870fe02b09b06 : SUCCESS in 9m 52s\n- tooz-tox-py311-zookeeper https://zuul.opendev.org/t/openstack/build/22a738c1224443a586fffbceeffafb65 : SUCCESS in 4m 59s\n- tooz-tox-py314-zookeeper https://zuul.opendev.org/t/openstack/build/84851c10b4594d638c7dc1635d6b709a : SUCCESS in 8m 28s\n- tooz-tox-py311-redis https://zuul.opendev.org/t/openstack/build/94d374783b1a4a73b739fac1dd438b0b : SUCCESS in 5m 54s\n- tooz-tox-py314-redis https://zuul.opendev.org/t/openstack/build/afcbd15c41694b30bfedd6680320325b : SUCCESS in 8m 05s\n- tooz-tox-py311-sentinel https://zuul.opendev.org/t/openstack/build/24c4225373974d16add64616cfb4707d : SUCCESS in 5m 27s\n- tooz-tox-py314-sentinel https://zuul.opendev.org/t/openstack/build/ded91f243cd949b58f6e46ddb2e085ba : FAILURE in 7m 28s\n- tooz-tox-py311-memcached https://zuul.opendev.org/t/openstack/build/e4cd8cc34fa14126850da8bc6a7f2e6e : SUCCESS in 5m 15s\n- tooz-tox-py314-memcached https://zuul.opendev.org/t/openstack/build/52daace97db246448bfe139762f801e5 : FAILURE in 9m 21s\n- tooz-tox-py311-postgresql https://zuul.opendev.org/t/openstack/build/ec2f155985a94a20985116a83f62d36f : SUCCESS in 4m 23s\n- tooz-tox-py314-postgresql https://zuul.opendev.org/t/openstack/build/866a6bdf717f416ba6d3f04101644681 : FAILURE in 7m 41s\n- tooz-tox-py311-mysql https://zuul.opendev.org/t/openstack/build/16042061c2ff49efb5ce78e8149b8ac3 : SUCCESS in 7m 33s\n- tooz-tox-py314-mysql https://zuul.opendev.org/t/openstack/build/527c4fb5527346da8e2f00746137fd7b : SUCCESS in 9m 44s\n- tooz-tox-py311-consul https://zuul.opendev.org/t/openstack/build/db4516b9efc445cba000e515d9678e5f : SUCCESS in 6m 28s\n- tooz-tox-py314-consul https://zuul.opendev.org/t/openstack/build/5126dc48fd894af28bd5d9ffb5bdfba2 : SUCCESS in 8m 03s","accounts_in_message":[],"_revision_number":4},{"id":"4ccb8f4be817bd16788addae71f34ee83a2e96a3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-06-29 20:28:32.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n* Workflow+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":5},{"id":"5e4c09632f5073b5892d06e8272be853c1b9d923","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-29 22:33:49.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bfb79ed5325f44d79323deb29e615183\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/4972c062262f4215a1978e459f483b81 : SUCCESS in 2m 23s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/27c28e62084b4c599e4968c007cb0536 : SUCCESS in 1h 48m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a01af9b6521f44bbbef8041c6e717b87 : SUCCESS in 4m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c5ddc03edb0b43fd805d28d2a79b9659 : SUCCESS in 7m 47s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4f85a079a6dc41b59ff1710f9e655fb2 : SUCCESS in 5m 43s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8c54588fd0d9442890ecd95cfcdaf093 : SUCCESS in 5m 42s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/c7e2d2024ea64167ab8a7e0c8a2c2efe : SUCCESS in 8m 22s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/d58ed34bc7c24d6db6daecd37c9e2605 : SUCCESS in 9m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f1058df8de541a6ae80cdb9af303170 : SUCCESS in 6m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3fe13512b3d94ae9b563fb3a62335ca6 : SUCCESS in 5m 42s\n- tooz-tox-examples https://zuul.opendev.org/t/openstack/build/6fc58e44cdc646879dbbf71a5ade4d54 : SUCCESS in 6m 01s\n- tooz-tox-py311-etcd3gw https://zuul.opendev.org/t/openstack/build/6cc4e1ae45fc4d99abb917e28a66322a : SUCCESS in 6m 09s\n- tooz-tox-py314-etcd3gw https://zuul.opendev.org/t/openstack/build/ed4ca8df26974d3681d222640d44968c : SUCCESS in 8m 38s\n- tooz-tox-py311-zookeeper https://zuul.opendev.org/t/openstack/build/d06cb36be8864a5bb5e875802c4ecea1 : SUCCESS in 4m 15s\n- tooz-tox-py314-zookeeper https://zuul.opendev.org/t/openstack/build/cf2da1372418452d9fb45d16ca08308c : SUCCESS in 7m 55s\n- tooz-tox-py311-redis https://zuul.opendev.org/t/openstack/build/00ac2e6f3cb943c0a7950f7f84583ff7 : SUCCESS in 7m 02s\n- tooz-tox-py314-redis https://zuul.opendev.org/t/openstack/build/eef1d85f5c014725ba0a4bf23cf1d7c0 : SUCCESS in 9m 22s\n- tooz-tox-py311-sentinel https://zuul.opendev.org/t/openstack/build/51e2f5a34a954159a498335178983dfc : SUCCESS in 6m 56s\n- tooz-tox-py314-sentinel https://zuul.opendev.org/t/openstack/build/e0db1bdd92654b258600192febc8ed74 : SUCCESS in 8m 50s\n- tooz-tox-py311-memcached https://zuul.opendev.org/t/openstack/build/e3b16a4fe3ac4205839efb0d7cee5f15 : SUCCESS in 7m 10s\n- tooz-tox-py314-memcached https://zuul.opendev.org/t/openstack/build/ea35695db85b410eb74ff39e7b6db629 : SUCCESS in 9m 50s\n- tooz-tox-py311-postgresql https://zuul.opendev.org/t/openstack/build/733f4bb399dc423890860353aca0e364 : SUCCESS in 3m 54s\n- tooz-tox-py314-postgresql https://zuul.opendev.org/t/openstack/build/21843e62c2104cb186e89deb19d41752 : SUCCESS in 8m 05s\n- tooz-tox-py311-mysql https://zuul.opendev.org/t/openstack/build/9bf2af2038194f9182c341e903922407 : SUCCESS in 7m 24s\n- tooz-tox-py314-mysql https://zuul.opendev.org/t/openstack/build/c9ab1cc33a554bf3af60b166c6aea052 : SUCCESS in 8m 21s\n- tooz-tox-py311-consul https://zuul.opendev.org/t/openstack/build/b322b7d28110455e95a1094fecc44d8b : SUCCESS in 4m 32s\n- tooz-tox-py314-consul https://zuul.opendev.org/t/openstack/build/58c92f3774ef4cd7827e2f82783fb6b3 : SUCCESS in 6m 07s","accounts_in_message":[],"_revision_number":5},{"id":"499835e53ce69f274f400642d673a1db926b8d5b","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-06-30 02:46:55.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"f03d0354df57b2b619d2c1678f59d80ee4c2d0d4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-30 02:48:26.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"5b3306df12cf279a1bce0eb06fa448cbf21e5788","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-30 04:40:08.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/622981fb8b3e4c149aeb3923e04e272b\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/6de363f654844e64bd70fd0587cd64ac : SUCCESS in 3m 50s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/93e412e63e684f8d835be66530542e1a : SUCCESS in 1h 46m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4e098c528ac64d9a9686baa42f903f5b : SUCCESS in 7m 40s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f78ec1502bea4529868621cf42b8afbc : SUCCESS in 5m 39s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/12167bc77bf946a19995525db66451a5 : SUCCESS in 4m 35s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/ede6da5d5abe49f691a81b04d9d6a977 : SUCCESS in 7m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/02554c35e80044dba899a7a57d4fd69c : SUCCESS in 6m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/168bfd135ea447f0a2b459c7bbf08be0 : SUCCESS in 5m 53s\n- tooz-tox-examples https://zuul.opendev.org/t/openstack/build/d52bec67cce740949f6faf19d155aea4 : SUCCESS in 5m 44s\n- tooz-tox-py311-etcd3gw https://zuul.opendev.org/t/openstack/build/dd03c3ea897f46dfb7953c00b5cfd2df : SUCCESS in 6m 25s\n- tooz-tox-py314-etcd3gw https://zuul.opendev.org/t/openstack/build/29881a795eda49ab867c7267e2a438c7 : SUCCESS in 10m 06s\n- tooz-tox-py311-zookeeper https://zuul.opendev.org/t/openstack/build/11aec19037d447c6bf0f2484351b42bb : SUCCESS in 7m 10s\n- tooz-tox-py314-zookeeper https://zuul.opendev.org/t/openstack/build/8c19e25ec3b44ec582cfd5ffc147ba31 : SUCCESS in 6m 13s\n- tooz-tox-py311-redis https://zuul.opendev.org/t/openstack/build/6c0ac7aec8c84ba58ea2b75adb001a37 : SUCCESS in 6m 35s\n- tooz-tox-py314-redis https://zuul.opendev.org/t/openstack/build/c8e2a39debef4265b20819c45981bf79 : SUCCESS in 9m 15s\n- tooz-tox-py311-sentinel https://zuul.opendev.org/t/openstack/build/8b3dfd1a71b846acba95f9994abc9a52 : SUCCESS in 9m 24s\n- tooz-tox-py314-sentinel https://zuul.opendev.org/t/openstack/build/bb82f4f6c86740d89aee96d4f2e5b72c : SUCCESS in 9m 48s\n- tooz-tox-py311-memcached https://zuul.opendev.org/t/openstack/build/9f580419766e4a898f1881e604c5e7de : SUCCESS in 5m 48s\n- tooz-tox-py314-memcached https://zuul.opendev.org/t/openstack/build/a1e0fdfd26fc45f2a6a4a5fddc29b9dc : SUCCESS in 8m 32s\n- tooz-tox-py311-postgresql https://zuul.opendev.org/t/openstack/build/3a2bf234f9db49eba0bd0ac2b84de1da : SUCCESS in 5m 05s\n- tooz-tox-py314-postgresql https://zuul.opendev.org/t/openstack/build/05d8f7b77c034b97b1ee82609ba7af00 : SUCCESS in 4m 48s\n- tooz-tox-py311-mysql https://zuul.opendev.org/t/openstack/build/ac77e96bcaf0423cbc00238ef476fc1f : SUCCESS in 6m 11s\n- tooz-tox-py314-mysql https://zuul.opendev.org/t/openstack/build/a0e90732ffbd43f39272b328ad6414b2 : SUCCESS in 7m 58s\n- tooz-tox-py311-consul https://zuul.opendev.org/t/openstack/build/f523d9a2a01c4578b04fddea7a53c0c3 : SUCCESS in 5m 44s\n- tooz-tox-py314-consul https://zuul.opendev.org/t/openstack/build/0026b7db8fac4544a3f97ee1a6135c42 : SUCCESS in 7m 29s","accounts_in_message":[],"_revision_number":5},{"id":"37644379a8ec33c775dd15902f01416cf4f8762c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-30 04:40:08.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"6565ad50e1fc809ead77d77b7db00880c5b58f03","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-30 04:41:08.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ae6c7e54baa64e70a34d2d52bdeac188\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c64269647e754acfb2319049efdcea9b : SUCCESS in 47s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"13fba2f0ce881259c0e9367e2f365e6b75ec049c","revisions":{"1c807cc8f3c80079622912461c3a52f590c0bd13":{"kind":"REWORK","_number":1,"created":"2026-06-10 10:09:26.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/06/992606/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/06/992606/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/06/992606/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/06/992606/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/06/992606/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/06/992606/1"}}},"commit":{"parents":[{"commit":"0269194f4f81a591b1543b6f82184b407902ba26","subject":"Drop Python 3.10","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/0269194f4f81a591b1543b6f82184b407902ba26"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-06-10 10:09:06.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-06-10 10:09:06.000000000","tz":60},"subject":"Remove unnecessary debtcollector dependency","message":"Remove unnecessary debtcollector dependency\n\nChange-Id: Iddba39cd5d29a21a3d2d0430cb8b9d7157bf2748\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/1c807cc8f3c80079622912461c3a52f590c0bd13"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/1c807cc8f3c80079622912461c3a52f590c0bd13"}]},"branch":"refs/heads/master"},"81cef3d9502a612132edd33ddbcad409e2054166":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-06-10 12:40:49.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/06/992606/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/06/992606/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/06/992606/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/06/992606/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/06/992606/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/06/992606/2"}}},"commit":{"parents":[{"commit":"4988293a1c1b1001b722087b4218a8341495f951","subject":"Drop Python 3.10","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/4988293a1c1b1001b722087b4218a8341495f951"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-06-10 10:09:06.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-06-10 12:40:49.000000000","tz":0},"subject":"Remove unnecessary debtcollector dependency","message":"Remove unnecessary debtcollector dependency\n\nChange-Id: Iddba39cd5d29a21a3d2d0430cb8b9d7157bf2748\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/81cef3d9502a612132edd33ddbcad409e2054166"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/81cef3d9502a612132edd33ddbcad409e2054166"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"0269194f4f81a591b1543b6f82184b407902ba26","ours":"1c807cc8f3c80079622912461c3a52f590c0bd13","theirs":"4988293a1c1b1001b722087b4218a8341495f951","merge_strategy":"recursive","contains_conflicts":false}},"112203cd79015a4c1e956c69a85b7051bdd714bf":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2026-06-18 10:44:05.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"real_uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/06/992606/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/06/992606/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/06/992606/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/06/992606/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/06/992606/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/06/992606/3"}}},"commit":{"parents":[{"commit":"371a1397e5b07409ecd67f7fb5f1785789db4ef5","subject":"Drop Python 3.10","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/371a1397e5b07409ecd67f7fb5f1785789db4ef5"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-06-10 10:09:06.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-06-18 10:44:05.000000000","tz":0},"subject":"Remove unnecessary debtcollector dependency","message":"Remove unnecessary debtcollector dependency\n\nChange-Id: Iddba39cd5d29a21a3d2d0430cb8b9d7157bf2748\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/112203cd79015a4c1e956c69a85b7051bdd714bf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/112203cd79015a4c1e956c69a85b7051bdd714bf"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"4988293a1c1b1001b722087b4218a8341495f951","ours":"81cef3d9502a612132edd33ddbcad409e2054166","theirs":"371a1397e5b07409ecd67f7fb5f1785789db4ef5","merge_strategy":"recursive","contains_conflicts":false}},"0becfab13d818cdfce2fbebafd2f6dd3051fbfc9":{"kind":"NO_CHANGE","_number":4,"created":"2026-06-29 14:12:51.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/06/992606/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/06/992606/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/06/992606/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/06/992606/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/06/992606/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/06/992606/4"}}},"commit":{"parents":[{"commit":"62460f36e5cf8ce235976428e5537f12db6fcc72","subject":"Drop Python 3.10","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/62460f36e5cf8ce235976428e5537f12db6fcc72"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-06-10 10:09:06.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-06-29 14:12:51.000000000","tz":0},"subject":"Remove unnecessary debtcollector dependency","message":"Remove unnecessary debtcollector dependency\n\nChange-Id: Iddba39cd5d29a21a3d2d0430cb8b9d7157bf2748\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/0becfab13d818cdfce2fbebafd2f6dd3051fbfc9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/0becfab13d818cdfce2fbebafd2f6dd3051fbfc9"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"371a1397e5b07409ecd67f7fb5f1785789db4ef5","ours":"112203cd79015a4c1e956c69a85b7051bdd714bf","theirs":"62460f36e5cf8ce235976428e5537f12db6fcc72","merge_strategy":"recursive","contains_conflicts":false}},"13fba2f0ce881259c0e9367e2f365e6b75ec049c":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2026-06-29 20:28:32.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/06/992606/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/06/992606/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/06/992606/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/06/992606/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/06/992606/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/06/992606/5"}}},"commit":{"parents":[{"commit":"1133aa21be5aaab32f11d7452c412e5149a50d34","subject":"zuul: Use openstack-python3-next-jobs template","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/1133aa21be5aaab32f11d7452c412e5149a50d34"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-06-10 10:09:06.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-06-29 20:28:27.000000000","tz":60},"subject":"Remove unnecessary debtcollector dependency","message":"Remove unnecessary debtcollector dependency\n\nChange-Id: Iddba39cd5d29a21a3d2d0430cb8b9d7157bf2748\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/13fba2f0ce881259c0e9367e2f365e6b75ec049c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/13fba2f0ce881259c0e9367e2f365e6b75ec049c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
