)]}'
{"id":"openstack%2Ftooz~994482","triplet_id":"openstack%2Ftooz~master~Ia72e03393397f3bfc9cb56c2c6641df675292fc1","project":"openstack/tooz","branch":"master","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2026-06-29 19:39:42.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ia72e03393397f3bfc9cb56c2c6641df675292fc1","subject":"Fix mypy errors","status":"ABANDONED","created":"2026-06-23 16:39:21.000000000","updated":"2026-06-29 19:39:42.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"a9a06a97acff81edef120b1e06618086305b134d","_number":994482,"virtual_id_number":994482,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-06-23 19:06:06.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-23 19:06:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8616e2b625a24360685bf176ffbf1585316a3630","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-06-23 16:39:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fece9ddd7e931b9ba139e8d91124197e18be4120","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-23 19:06:06.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6795505117e947e491815dc36d30d04c\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/af2a62beafa24c978cdf383bb5397886 : SUCCESS in 2h 17m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/84ddfdfb0d104dd29956c1dd5c4469cc : SUCCESS in 6m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/283e47d34aae4ebcacb17981689fa942 : SUCCESS in 5m 45s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1e10e9c6b09c40c48640339c1dca5bcf : SUCCESS in 3m 32s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5e360faec01e4749bfca4f47bee6353c : SUCCESS in 4m 18s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/9b9d72c43f9a4fa89c449e15421b7a3d : SUCCESS in 7m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa3202c40fd94347b19886c4a2dfaa53 : SUCCESS in 6m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e977a4592f8d416b8bc1d058bba59fdf : SUCCESS in 5m 27s\n- tooz-tox-examples https://zuul.opendev.org/t/openstack/build/44980a80bc154b7da9cc8976bec83005 : SUCCESS in 5m 11s\n- tooz-tox-py310-etcd3gw https://zuul.opendev.org/t/openstack/build/a1c782afaeb8454f8561c240301d0904 : FAILURE in 3m 43s\n- tooz-tox-py313-etcd3gw https://zuul.opendev.org/t/openstack/build/4cbeca7443f94b568c8d13b29accc951 : SUCCESS in 6m 17s\n- tooz-tox-py310-zookeeper https://zuul.opendev.org/t/openstack/build/f286a17ab6444f62a1b786e3df17604e : FAILURE in 3m 12s\n- tooz-tox-py313-zookeeper https://zuul.opendev.org/t/openstack/build/9ed2a4a4aeb945399b305d8ff7d9a278 : SUCCESS in 5m 06s\n- tooz-tox-py310-redis https://zuul.opendev.org/t/openstack/build/5917d356b1b44c11a3daa0e48325e88a : FAILURE in 2m 56s\n- tooz-tox-py313-redis https://zuul.opendev.org/t/openstack/build/0aa474e83391402aa28490cb32a70d00 : SUCCESS in 8m 42s\n- tooz-tox-py310-sentinel https://zuul.opendev.org/t/openstack/build/c93666b8abbf4e088fa860e5bdb155a1 : FAILURE in 4m 43s\n- tooz-tox-py313-sentinel https://zuul.opendev.org/t/openstack/build/93d97d5252de48be8dc68e5a480616f3 : SUCCESS in 6m 27s\n- tooz-tox-py310-memcached https://zuul.opendev.org/t/openstack/build/5de867c1fb744d05867ebcfd3bcec0d2 : FAILURE in 3m 24s\n- tooz-tox-py313-memcached https://zuul.opendev.org/t/openstack/build/24fa7426055f4c37a4cb42fb547565c3 : SUCCESS in 4m 54s\n- tooz-tox-py310-postgresql https://zuul.opendev.org/t/openstack/build/104365f03e244fa7a3f4926c877ff4aa : FAILURE in 3m 28s\n- tooz-tox-py313-postgresql https://zuul.opendev.org/t/openstack/build/c58f39f369144d4a95c95c127e729d29 : SUCCESS in 4m 49s\n- tooz-tox-py310-mysql https://zuul.opendev.org/t/openstack/build/e539aada57174e84aa514903586555de : FAILURE in 5m 03s\n- tooz-tox-py313-mysql https://zuul.opendev.org/t/openstack/build/cf9e72192bcb4449a7bd37f57b08a35c : SUCCESS in 8m 34s\n- tooz-tox-py310-consul https://zuul.opendev.org/t/openstack/build/04edec308a9546f4ae8522c92570e0ce : FAILURE in 4m 26s\n- tooz-tox-py313-consul https://zuul.opendev.org/t/openstack/build/4aabe0f1ee8a44a3a766f7bafc214b46 : SUCCESS in 4m 28s","accounts_in_message":[],"_revision_number":1},{"id":"bdc330634e3cf42d622cfe91b3a8714c8abfee03","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-06-29 14:11:08.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"73ae6763b9f0a078744ba617a2a2e2a8e71df2ff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-29 15:36:50.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ee6fea98e816454aa922b6d17e4d7b0e\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d393d1afb2f24f02bc0ec7ee7e8ace55 : SUCCESS in 1h 23m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/690d78e8c7ef4dc991eb87d4c3521ac0 : SUCCESS in 3m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/83c29331337e4428b228a5ab2b724097 : SUCCESS in 3m 12s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/01b87447903f4efaaf0a5d2535f7529a : SUCCESS in 3m 53s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ab7586d0011c4392b2da98283031d618 : SUCCESS in 3m 30s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/319d944cef6f445e803d88eae39e92aa : FAILURE in 9m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d6b48b01c986445fb4839a0e0d6ba6a4 : SUCCESS in 6m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ebccdaf750a644aa9eeca17532421399 : SUCCESS in 4m 51s\n- tooz-tox-examples https://zuul.opendev.org/t/openstack/build/4534fcb90b364390bcdc2fec7fd9ea94 : SUCCESS in 4m 49s\n- tooz-tox-py310-etcd3gw https://zuul.opendev.org/t/openstack/build/c11693cac6ab42718613e4e9697c67e7 : FAILURE in 2m 17s\n- tooz-tox-py313-etcd3gw https://zuul.opendev.org/t/openstack/build/7796d94ce37841b588dc140f9e628997 : SUCCESS in 5m 43s\n- tooz-tox-py310-zookeeper https://zuul.opendev.org/t/openstack/build/b9dad389285d439989aca61e0d20bd0e : FAILURE in 2m 33s\n- tooz-tox-py313-zookeeper https://zuul.opendev.org/t/openstack/build/5b74427f0f6645f39166731d1df9444b : SUCCESS in 4m 43s\n- tooz-tox-py310-redis https://zuul.opendev.org/t/openstack/build/2618934260f8402bb03ae43336a9b532 : FAILURE in 2m 36s\n- tooz-tox-py313-redis https://zuul.opendev.org/t/openstack/build/eb8748a2ecab424d957ad6b0d3e59080 : SUCCESS in 5m 05s\n- tooz-tox-py310-sentinel https://zuul.opendev.org/t/openstack/build/8ea87e44d9e6475e9c6214bbff56daf6 : FAILURE in 3m 16s\n- tooz-tox-py313-sentinel https://zuul.opendev.org/t/openstack/build/71fb68e5675047bda811d3736e020aff : SUCCESS in 6m 47s\n- tooz-tox-py310-memcached https://zuul.opendev.org/t/openstack/build/7f1de4b01e6646f3bb95c606d86c7d03 : FAILURE in 3m 17s\n- tooz-tox-py313-memcached https://zuul.opendev.org/t/openstack/build/e7ce30db25de4682bdb1bbad16730132 : SUCCESS in 7m 07s\n- tooz-tox-py310-postgresql https://zuul.opendev.org/t/openstack/build/3fe720c52087494ebf17e7b6154acda4 : FAILURE in 2m 31s\n- tooz-tox-py313-postgresql https://zuul.opendev.org/t/openstack/build/06b458a56e6949e6bed115ed8d015a11 : SUCCESS in 3m 30s\n- tooz-tox-py310-mysql https://zuul.opendev.org/t/openstack/build/b1bfa623a7fd4f028b0f7ef95c4d2d45 : FAILURE in 3m 34s\n- tooz-tox-py313-mysql https://zuul.opendev.org/t/openstack/build/f90b2d50a4244f1eaff7a705f649bf01 : SUCCESS in 7m 40s\n- tooz-tox-py310-consul https://zuul.opendev.org/t/openstack/build/9f1b707f944940bbb5fe64f34b42bb78 : FAILURE in 2m 30s\n- tooz-tox-py313-consul https://zuul.opendev.org/t/openstack/build/a0b45ceed2a94f4383b2fdb3691ce78d : SUCCESS in 4m 53s","accounts_in_message":[],"_revision_number":1},{"id":"a9a06a97acff81edef120b1e06618086305b134d","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-06-29 19:39:42.000000000","message":"Abandoned\n\nsquashed into https://review.opendev.org/c/openstack/tooz/+/992605/","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"eddf649c8a9f8208cbe2d65005a671a5eae4c091","revisions":{"eddf649c8a9f8208cbe2d65005a671a5eae4c091":{"kind":"REWORK","_number":1,"created":"2026-06-23 16:39:21.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/82/994482/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/82/994482/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/82/994482/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/82/994482/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/82/994482/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/82/994482/1"}}},"commit":{"parents":[{"commit":"12e22ef600b547b1636a38f8920910ac1dbc91f9","subject":"Fix mypy errors","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/12e22ef600b547b1636a38f8920910ac1dbc91f9"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-06-23 16:32:10.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-06-23 16:32:10.000000000","tz":60},"subject":"Fix mypy errors","message":"Fix mypy errors\n\nChange-Id: Ia72e03393397f3bfc9cb56c2c6641df675292fc1\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/eddf649c8a9f8208cbe2d65005a671a5eae4c091"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/eddf649c8a9f8208cbe2d65005a671a5eae4c091"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
