)]}'
{"id":"openstack%2Ftooz~994665","triplet_id":"openstack%2Ftooz~master~Ib7c5c08931e14bc3205d32928c6f5c50f5951d63","project":"openstack/tooz","branch":"master","topic":"python-next-jobs","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2026-06-30 04:26:47.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib7c5c08931e14bc3205d32928c6f5c50f5951d63","subject":"zuul: Use openstack-python3-next-jobs template","status":"MERGED","created":"2026-06-24 10:19:06.000000000","updated":"2026-06-30 04:29:42.000000000","submitted":"2026-06-30 04:26:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"994665-python-next-jobs","meta_rev_id":"ce9a23ba6012b84c3afa377e1c41fdcfe63d4a81","_number":994665,"virtual_id_number":994665,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-06-30 04:26:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2026-06-30 02:46:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2026-06-30 02:46:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-24 12:27:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-30 02:46:45.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"}],"messages":[{"id":"ea8bfc4ba10699f8718d416755246cc61354495a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-06-24 10:19:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8f48a58a4241eb27328b6fc9d45e1706b43296cf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-24 12:27:28.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/144641575271463c85b5fa51eaf7720e\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6b2ca01f1b47412c97398cb130434738 : SUCCESS in 1h 49m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8f3d09d6fb5e499582c018032dcf4ecf : SUCCESS in 7m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3e07ee5c0bb941018e239459b0d57fe4 : FAILURE in 7m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/26ad8df4492740beac409a67f2633ad6 : SUCCESS in 5m 53s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/75cd708ef1e040a2ac1b1978733abe37 : SUCCESS in 6m 48s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/a25eb236ba58413bb29bbb42e4df196f : SUCCESS in 9m 08s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/12ea5898867046a2b28bd870c5e9a280 : FAILURE in 9m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/08a0a0dd47d54628bfa28c8c67ae4cee : SUCCESS in 5m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d0307e4eb831481f97ba569e7ba02a4e : SUCCESS in 5m 15s\n- tooz-tox-examples https://zuul.opendev.org/t/openstack/build/e059d707bc8848dda1a64476479323b5 : SUCCESS in 5m 00s\n- tooz-tox-py310-etcd3gw https://zuul.opendev.org/t/openstack/build/ab9c00bf9863458899142d4aca1c8c90 : FAILURE in 4m 35s\n- tooz-tox-py313-etcd3gw https://zuul.opendev.org/t/openstack/build/5cb3b39aedf24257901e24c9b21688f5 : SUCCESS in 7m 16s\n- tooz-tox-py310-zookeeper https://zuul.opendev.org/t/openstack/build/dc76909cb72a4552829300fcff0ffe82 : FAILURE in 3m 31s\n- tooz-tox-py313-zookeeper https://zuul.opendev.org/t/openstack/build/33e425d80707466cb987a850c4d7729d : SUCCESS in 6m 46s\n- tooz-tox-py310-redis https://zuul.opendev.org/t/openstack/build/6d42efcd1be247d79a73b4f93e361c77 : FAILURE in 5m 06s\n- tooz-tox-py313-redis https://zuul.opendev.org/t/openstack/build/1a7d65d9af7546c39a5221a273a6bc3e : SUCCESS in 6m 48s\n- tooz-tox-py310-sentinel https://zuul.opendev.org/t/openstack/build/b49158985b404a1081cd605d4cca5aed : FAILURE in 3m 58s\n- tooz-tox-py313-sentinel https://zuul.opendev.org/t/openstack/build/5111cf348e7646478a86e8899623196f : SUCCESS in 4m 26s\n- tooz-tox-py310-memcached https://zuul.opendev.org/t/openstack/build/71b55ca7a5dd46a894739c9ddc7dc1bf : FAILURE in 4m 44s\n- tooz-tox-py313-memcached https://zuul.opendev.org/t/openstack/build/c5025363517c4ce9a12682bb5ec217c2 : SUCCESS in 6m 10s\n- tooz-tox-py310-postgresql https://zuul.opendev.org/t/openstack/build/df2ff0f3e584435eb32017d0aced6fe7 : FAILURE in 4m 25s\n- tooz-tox-py313-postgresql https://zuul.opendev.org/t/openstack/build/56d71d85522e4ce88b0f3c9c965e5282 : SUCCESS in 5m 54s\n- tooz-tox-py310-mysql https://zuul.opendev.org/t/openstack/build/4ff993f0051e47e78eeb96a3cbf61986 : FAILURE in 4m 51s\n- tooz-tox-py313-mysql https://zuul.opendev.org/t/openstack/build/b41b06aeb05b4043828c0dcdbe675f95 : SUCCESS in 7m 31s\n- tooz-tox-py310-consul https://zuul.opendev.org/t/openstack/build/0c578ac2082e4a7bac9a3697352cfd01 : FAILURE in 5m 07s\n- tooz-tox-py313-consul https://zuul.opendev.org/t/openstack/build/2decbca2e05543ff85db95a0ba7fb9c8 : SUCCESS in 5m 53s","accounts_in_message":[],"_revision_number":1},{"id":"b135714721bc94c45f61d6e6e354812a902f4d3b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-06-29 14:15:05.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"2c2ba454a0c8e7cd450b4866a42d5ba0cb44d638","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-29 15:13:43.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8063205ce68040e686b1b6a7659fd279\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0d5134934dd040b3bcfe84465fa20c0c : SUCCESS in 55m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/14e6a0e3c9cb477db0574bed54142eb9 : SUCCESS in 3m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d1fb0b2ba3d94a0db1743222fb6c0205 : SUCCESS in 3m 11s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/36cbc1944eda4730b8178993d2fd2e96 : SUCCESS in 5m 04s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/90a34419c5c04e7986fe221ae843a406 : SUCCESS in 3m 27s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/8cd9d30204e24b59a8800bb42be8727e : SUCCESS in 8m 32s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/7012ea720ad846e8b85da4c21aae75a0 : SUCCESS in 5m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bcd80a6792a84763bd40efb2fb232c04 : SUCCESS in 4m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/41b7f98af4a34f5fb27d333c63bdc74b : SUCCESS in 5m 50s\n- tooz-tox-examples https://zuul.opendev.org/t/openstack/build/1e8d51d83832473d900c6f5520865367 : SUCCESS in 5m 35s\n- tooz-tox-py311-etcd3gw https://zuul.opendev.org/t/openstack/build/6272c6f640064d98ac0b338c014e9520 : SUCCESS in 5m 46s\n- tooz-tox-py314-etcd3gw https://zuul.opendev.org/t/openstack/build/eb6006baf8914306914476cd20f33174 : SUCCESS in 8m 05s\n- tooz-tox-py311-zookeeper https://zuul.opendev.org/t/openstack/build/9bf2e35a952945c7bb276d62b5a86a6d : SUCCESS in 4m 16s\n- tooz-tox-py314-zookeeper https://zuul.opendev.org/t/openstack/build/11b0a550b8554be8b56c62511616924f : SUCCESS in 7m 32s\n- tooz-tox-py311-redis https://zuul.opendev.org/t/openstack/build/e5393e2775644728848152660068fb1e : SUCCESS in 5m 51s\n- tooz-tox-py314-redis https://zuul.opendev.org/t/openstack/build/489c90a889b94eb18e8f1b93f34ab9af : FAILURE in 10m 14s\n- tooz-tox-py311-sentinel https://zuul.opendev.org/t/openstack/build/f363758466b6481fbe4ac255d9be00d1 : SUCCESS in 5m 50s\n- tooz-tox-py314-sentinel https://zuul.opendev.org/t/openstack/build/52235bd0615f434b84cb8ffbce3300ee : SUCCESS in 9m 25s\n- tooz-tox-py311-memcached https://zuul.opendev.org/t/openstack/build/06ffcd10b3854c7986ed84e797dcdfaf : SUCCESS in 5m 11s\n- tooz-tox-py314-memcached https://zuul.opendev.org/t/openstack/build/13210f2c96f04559a6fdf18acfeb030f : SUCCESS in 7m 51s\n- tooz-tox-py311-postgresql https://zuul.opendev.org/t/openstack/build/fd3512617bb2400082a0cfbb64afe200 : SUCCESS in 4m 37s\n- tooz-tox-py314-postgresql https://zuul.opendev.org/t/openstack/build/4f7f71aea3294b378b10400fd5f267c9 : SUCCESS in 9m 23s\n- tooz-tox-py311-mysql https://zuul.opendev.org/t/openstack/build/9c952015bbc34e2ea0f22c387ae00d81 : SUCCESS in 4m 46s\n- tooz-tox-py314-mysql https://zuul.opendev.org/t/openstack/build/a81ced38a7394d449146d6520bed586a : SUCCESS in 8m 24s\n- tooz-tox-py311-consul https://zuul.opendev.org/t/openstack/build/566c45f558934ef2a67bd93b03d9d9a0 : SUCCESS in 5m 15s\n- tooz-tox-py314-consul https://zuul.opendev.org/t/openstack/build/411df45f3d584d4a9c39fbdcbccfccf7 : SUCCESS in 5m 30s","accounts_in_message":[],"_revision_number":2},{"id":"c662356a5e631c833f410225d5457256dcf91a2c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-06-29 20:28:32.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"20d9b401d52787a1496d3f0a30b3c979bff42468","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-29 22:31:54.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c4619a25bb294328818613b4ac404f4f\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4669b53d21cd4e4a8dabb0348a869478 : SUCCESS in 1h 41m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/12d5ebc351ab4c4c859747d7d18bcb7b : SUCCESS in 6m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f17302ae2574435d90e69468fcba20cd : SUCCESS in 5m 54s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/91e2de4588774441adadaee109177742 : SUCCESS in 5m 44s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/63ad6482a8b2484287bb6940fdcf00bd : SUCCESS in 7m 08s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/a3202ead94cd4be2b92dcdb3d8137df0 : SUCCESS in 6m 27s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/9a14c01037774aa4b8d10a1879a5ad65 : SUCCESS in 8m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5aa99242259f45f5894367de70b418f5 : SUCCESS in 3m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3f12b8d6080b470e81613cc6ae53e556 : SUCCESS in 3m 25s\n- tooz-tox-examples https://zuul.opendev.org/t/openstack/build/2dfe544ed5e349c1a5c9696faca3edce : SUCCESS in 5m 55s\n- tooz-tox-py311-etcd3gw https://zuul.opendev.org/t/openstack/build/e3428c17676a41e294796f3a54befda8 : SUCCESS in 6m 22s\n- tooz-tox-py314-etcd3gw https://zuul.opendev.org/t/openstack/build/00bd787518214638abdb6002f63603de : SUCCESS in 10m 23s\n- tooz-tox-py311-zookeeper https://zuul.opendev.org/t/openstack/build/585802aa3b4542129ca3afaf6e67e400 : SUCCESS in 5m 43s\n- tooz-tox-py314-zookeeper https://zuul.opendev.org/t/openstack/build/16639a52cf984cd29a3dac00fdaecfa6 : SUCCESS in 5m 00s\n- tooz-tox-py311-redis https://zuul.opendev.org/t/openstack/build/1d8d6b02c9b04fe195694699a4e778e6 : SUCCESS in 4m 54s\n- tooz-tox-py314-redis https://zuul.opendev.org/t/openstack/build/27831a7f63ea49c4bb4f6314bb8e251e : SUCCESS in 9m 06s\n- tooz-tox-py311-sentinel https://zuul.opendev.org/t/openstack/build/c56ff2b0970441feb1675919e8d5a8c1 : SUCCESS in 9m 12s\n- tooz-tox-py314-sentinel https://zuul.opendev.org/t/openstack/build/5b627423c5704e979788386ffa05a52b : SUCCESS in 7m 03s\n- tooz-tox-py311-memcached https://zuul.opendev.org/t/openstack/build/f5f24cd75e3a40159eb0e82fe5023abf : SUCCESS in 6m 41s\n- tooz-tox-py314-memcached https://zuul.opendev.org/t/openstack/build/0f9ecbcc2b1448289d4cf807e55e42f5 : SUCCESS in 5m 55s\n- tooz-tox-py311-postgresql https://zuul.opendev.org/t/openstack/build/baa8c56a09d0468db063f8e9853bb448 : SUCCESS in 3m 13s\n- tooz-tox-py314-postgresql https://zuul.opendev.org/t/openstack/build/4eeb415d1c3d441098db43accdcfc9b1 : SUCCESS in 8m 21s\n- tooz-tox-py311-mysql https://zuul.opendev.org/t/openstack/build/15885e7c2ddf4624a3bf5d802981f084 : SUCCESS in 7m 28s\n- tooz-tox-py314-mysql https://zuul.opendev.org/t/openstack/build/d4f1ee927980408782126d1d012c948e : SUCCESS in 8m 54s\n- tooz-tox-py311-consul https://zuul.opendev.org/t/openstack/build/3ab3914ffb6a42b982467bc2c01a7d8f : SUCCESS in 5m 42s\n- tooz-tox-py314-consul https://zuul.opendev.org/t/openstack/build/32722b8438044d349c20d3d83c9e6a2e : SUCCESS in 10m 26s","accounts_in_message":[],"_revision_number":3},{"id":"782f3f718880eff95b5fad2e2992cb36f67d5214","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-06-30 02:46:45.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"b12f07b77ff7ffbd9ef11ad2c5b99083023476c2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-30 02:48:25.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"7131309690107a0d699f8381e2319cd2e8026e7e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-30 04:26:47.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d2167f75e57340fe9f0427e85fcc7ea8\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/94cc655ed11b49208d4a25ba0d50ba30 : SUCCESS in 58m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f0fb0d1016054dd7ae2dfb3cd9ac106d : SUCCESS in 6m 56s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6712e569226c4c5cba05419597f08e89 : SUCCESS in 5m 26s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ec1c4b75c378431f9654272eda12c09b : SUCCESS in 5m 38s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/01ff6a74771c4f5f9496104adbec04e5 : SUCCESS in 8m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd05e049fba14b35bfaaf1a144db09eb : SUCCESS in 3m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8a6b729cc4874ed38f226620648bfbba : SUCCESS in 5m 18s\n- tooz-tox-examples https://zuul.opendev.org/t/openstack/build/fd5968231fe345c3b1b323ea49898629 : SUCCESS in 3m 50s\n- tooz-tox-py311-etcd3gw https://zuul.opendev.org/t/openstack/build/b2585740189b4c7db3b20bafe45e4505 : SUCCESS in 7m 16s\n- tooz-tox-py314-etcd3gw https://zuul.opendev.org/t/openstack/build/0c6baf78547b4464a5a1a87f9a86f6c9 : SUCCESS in 6m 25s\n- tooz-tox-py311-zookeeper https://zuul.opendev.org/t/openstack/build/375da1c0b9b7487db4f878e0499abd8f : SUCCESS in 7m 00s\n- tooz-tox-py314-zookeeper https://zuul.opendev.org/t/openstack/build/b84cdb580d0d4c5589e9c96e7941b234 : SUCCESS in 7m 49s\n- tooz-tox-py311-redis https://zuul.opendev.org/t/openstack/build/bb4dc0d36dd846a786e87650c2131021 : SUCCESS in 6m 31s\n- tooz-tox-py314-redis https://zuul.opendev.org/t/openstack/build/756ec533a800476b8aa3f715d6b4b484 : SUCCESS in 9m 18s\n- tooz-tox-py311-sentinel https://zuul.opendev.org/t/openstack/build/e2dc67bdc5004bed895ea8495176efe8 : SUCCESS in 5m 51s\n- tooz-tox-py314-sentinel https://zuul.opendev.org/t/openstack/build/6648b1c560f34356b08317c9d5a13669 : SUCCESS in 9m 17s\n- tooz-tox-py311-memcached https://zuul.opendev.org/t/openstack/build/45b69db4f45a40dc9ce0967e1dc8179d : SUCCESS in 5m 35s\n- tooz-tox-py314-memcached https://zuul.opendev.org/t/openstack/build/4d2d650170fe40ff9b14a2b8fd0dc75d : SUCCESS in 5m 32s\n- tooz-tox-py311-postgresql https://zuul.opendev.org/t/openstack/build/664fc14a37e54bb8aab4e51bddefcc2e : SUCCESS in 5m 19s\n- tooz-tox-py314-postgresql https://zuul.opendev.org/t/openstack/build/82f1afaa14e44ab99fc44be69195d0c7 : SUCCESS in 7m 31s\n- tooz-tox-py311-mysql https://zuul.opendev.org/t/openstack/build/cc77a78c62d44cbf9ae61e97d04a2514 : SUCCESS in 8m 02s\n- tooz-tox-py314-mysql https://zuul.opendev.org/t/openstack/build/2dbfd667fbac4e9bb1abd175fad78f89 : SUCCESS in 5m 00s\n- tooz-tox-py311-consul https://zuul.opendev.org/t/openstack/build/eb4f773168624b73a054dbe2f5035415 : SUCCESS in 5m 58s\n- tooz-tox-py314-consul https://zuul.opendev.org/t/openstack/build/24330d16af2145c4957cd3ea92617db4 : SUCCESS in 7m 53s","accounts_in_message":[],"_revision_number":3},{"id":"9a62d3dfdcb7b29bff3c2c63bf0b3fab01a9170d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-30 04:26:47.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"ce9a23ba6012b84c3afa377e1c41fdcfe63d4a81","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-30 04:29:42.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7afa0131e2f94123abcc353b19ad91fc\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5aee7c5a3236407ab73f289d3bbbbb00 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"1133aa21be5aaab32f11d7452c412e5149a50d34","revisions":{"86fbd02b82dc0258cbdd67622e0d82c30249fa7b":{"kind":"REWORK","_number":1,"created":"2026-06-24 10:19:06.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/65/994665/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/65/994665/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/65/994665/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/65/994665/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/65/994665/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/65/994665/1"}}},"commit":{"parents":[{"commit":"ff09a44e5b6a7c21d031af94444d1badbf64e156","subject":"pre-commit: Bump versions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/ff09a44e5b6a7c21d031af94444d1badbf64e156"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-06-24 10:16:19.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-06-24 10:16:19.000000000","tz":60},"subject":"zuul: Use openstack-python3-next-jobs template","message":"zuul: Use openstack-python3-next-jobs template\n\nTest the project against the latest and greatest Python versions.\n\nChange-Id: Ib7c5c08931e14bc3205d32928c6f5c50f5951d63\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/86fbd02b82dc0258cbdd67622e0d82c30249fa7b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/86fbd02b82dc0258cbdd67622e0d82c30249fa7b"}]},"branch":"refs/heads/master"},"96138c470df6668052c170f38b7754f3e35170ed":{"kind":"REWORK","_number":2,"created":"2026-06-29 14:15:05.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/65/994665/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/65/994665/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/65/994665/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/65/994665/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/65/994665/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/65/994665/2"}}},"commit":{"parents":[{"commit":"1044a7da73c446ddd4b4af31d977e32a373b7e7b","subject":"pre-commit: Bump versions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/1044a7da73c446ddd4b4af31d977e32a373b7e7b"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-06-24 10:16:19.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-06-29 14:14:51.000000000","tz":60},"subject":"zuul: Use openstack-python3-next-jobs template","message":"zuul: Use openstack-python3-next-jobs template\n\nTest the project against the latest and greatest Python versions.\n\nChange-Id: Ib7c5c08931e14bc3205d32928c6f5c50f5951d63\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/96138c470df6668052c170f38b7754f3e35170ed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/96138c470df6668052c170f38b7754f3e35170ed"}]},"branch":"refs/heads/master"},"1133aa21be5aaab32f11d7452c412e5149a50d34":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2026-06-29 20:28:32.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/65/994665/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/65/994665/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/65/994665/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/65/994665/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/65/994665/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/65/994665/3"}}},"commit":{"parents":[{"commit":"b81c6d7695ec0e8c458054533b7cc67ff7d1ce97","subject":"pre-commit: Bump versions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/b81c6d7695ec0e8c458054533b7cc67ff7d1ce97"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-06-24 10:16:19.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-06-29 20:28:27.000000000","tz":60},"subject":"zuul: Use openstack-python3-next-jobs template","message":"zuul: Use openstack-python3-next-jobs template\n\nTest the project against the latest and greatest Python versions.\n\nChange-Id: Ib7c5c08931e14bc3205d32928c6f5c50f5951d63\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/1133aa21be5aaab32f11d7452c412e5149a50d34"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/1133aa21be5aaab32f11d7452c412e5149a50d34"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
