)]}'
{"id":"openstack%2Ftooz~995417","triplet_id":"openstack%2Ftooz~master~I7507ac2907453562b64c135a3a46adcc007df4a2","project":"openstack/tooz","branch":"master","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2026-06-30 04:26:43.000000000","reason":"Change was submitted"},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-06-30 02:46:27.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"I7507ac2907453562b64c135a3a46adcc007df4a2","subject":"Add Python 3.14 support","status":"MERGED","created":"2026-06-29 20:28:32.000000000","updated":"2026-06-30 04:26:43.000000000","submitted":"2026-06-30 04:26:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"995417","meta_rev_id":"3082b8eabd838ba2d24d5c30bece0bd9148c52f5","_number":995417,"virtual_id_number":995417,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-06-30 04:26:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2026-06-29 22:25:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2026-06-30 02:46:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":1,"date":"2026-06-30 02:46:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-29 22:25:20.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2026-06-29 22:26:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0993878c071405de41432c0685dc838a6403a2cb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-06-29 20:28:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"27c33ece9c39df45f5b10c54182a348cc31d3709","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-06-29 22:25:20.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"dcce48520d5774231d8863acfe08576cb5d80d25","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-29 22:26:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6c4e0a10e1b549298354970bf0e859ea\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/ffe65fe6788f4dca8cffd25af7732924 : SUCCESS in 3m 32s\n- requirements-check https://zuul.opendev.org/t/openstack/build/acf9322c80be4b6f8e838eb4f46f4571 : SUCCESS in 2m 29s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e83ab21ab1d2402f9225a50441e38ac3 : SUCCESS in 1h 38m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/04f2889cb930417bb0e3f710bd7a7b28 : SUCCESS in 5m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4fee283fd9064d5ab900f71f7e7baa69 : SUCCESS in 6m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/eb1e920afb644689903b237649ae4884 : SUCCESS in 6m 12s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9ca3e0a98bb047abb08fed0b87c873bb : SUCCESS in 5m 46s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/cdfe8034d3f44254b71560ee5dad409b : SUCCESS in 8m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c214edadee934a219570ed88f6b27d53 : SUCCESS in 6m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2938209345854da496536e03ac79bfa9 : SUCCESS in 4m 11s\n- tooz-tox-examples https://zuul.opendev.org/t/openstack/build/948c873b6c07463c8b4b5b1688fb1571 : SUCCESS in 6m 35s\n- tooz-tox-py311-etcd3gw https://zuul.opendev.org/t/openstack/build/cfe5e347cd98406282bbfd881139eed6 : SUCCESS in 5m 32s\n- tooz-tox-py314-etcd3gw https://zuul.opendev.org/t/openstack/build/6c5b1092a7804c389f1408e9e0d53536 : SUCCESS in 8m 08s\n- tooz-tox-py311-zookeeper https://zuul.opendev.org/t/openstack/build/e29ca0d62cbc4058ad8f1809d7544f32 : SUCCESS in 6m 04s\n- tooz-tox-py314-zookeeper https://zuul.opendev.org/t/openstack/build/c6ce6fd0e3224154bdcb677403b1b0e6 : SUCCESS in 7m 47s\n- tooz-tox-py311-redis https://zuul.opendev.org/t/openstack/build/f0bcdcbd2e7641cca06267464f84ae92 : SUCCESS in 6m 55s\n- tooz-tox-py314-redis https://zuul.opendev.org/t/openstack/build/9e04bffaa1174fb495f369d4211065bd : SUCCESS in 10m 00s\n- tooz-tox-py311-sentinel https://zuul.opendev.org/t/openstack/build/f427be4a955d4a4585fedd3195112f33 : SUCCESS in 6m 09s\n- tooz-tox-py314-sentinel https://zuul.opendev.org/t/openstack/build/0844fa9a19854c78a8102682e817b29f : SUCCESS in 9m 02s\n- tooz-tox-py311-memcached https://zuul.opendev.org/t/openstack/build/2866b1d8da0147b1925b30d7d13eed67 : SUCCESS in 5m 17s\n- tooz-tox-py314-memcached https://zuul.opendev.org/t/openstack/build/10288549f7c549e1a869ebce35d7b040 : SUCCESS in 5m 25s\n- tooz-tox-py311-postgresql https://zuul.opendev.org/t/openstack/build/efc47a65fed74d4f894d2657fb4d62f3 : SUCCESS in 5m 12s\n- tooz-tox-py314-postgresql https://zuul.opendev.org/t/openstack/build/1b8b893f9ee14f41bc493eaad6c5169b : SUCCESS in 7m 36s\n- tooz-tox-py311-mysql https://zuul.opendev.org/t/openstack/build/aaf4b3af82a94cef91febfcffff1172d : SUCCESS in 7m 59s\n- tooz-tox-py314-mysql https://zuul.opendev.org/t/openstack/build/621c0d77b87b4fe18430b4a66d0979cd : SUCCESS in 5m 08s\n- tooz-tox-py311-consul https://zuul.opendev.org/t/openstack/build/e208e088d36b473589f61dc73322ed54 : SUCCESS in 4m 15s\n- tooz-tox-py314-consul https://zuul.opendev.org/t/openstack/build/5ac59d38db824b84b0d62f5e46ee71a0 : SUCCESS in 8m 48s","accounts_in_message":[],"_revision_number":1},{"id":"39485a0778be986c8ac8caddac7962018668fc61","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-06-30 02:46:27.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3efefaebfe9d056470b2141b690094d3ad20983a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-30 02:47:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6b8f50d7c1c32888cd73c5c7791509776e6f7bd9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-30 04:26:43.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/31abe0b743b24bbe870762197f6a691f\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/0f465f4286644e5686dcd6bb413e790f : SUCCESS in 2m 27s\n- requirements-check https://zuul.opendev.org/t/openstack/build/6e1b589b788742e388540d039e3e8bd5 : SUCCESS in 2m 51s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/93e6b900542e4e05bfc1ecc5384116da : SUCCESS in 57m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/755e441ff9cc436187d4642b110bde6b : SUCCESS in 6m 21s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/be9b842a41694a418391944864aaf5c9 : SUCCESS in 6m 07s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8eb2ec2731e64212895af6516248f834 : SUCCESS in 5m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/10d58b57007f4398887810d7adb75686 : SUCCESS in 5m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a301e3dcd04d4500b643bdc3456cd698 : SUCCESS in 5m 07s\n- tooz-tox-examples https://zuul.opendev.org/t/openstack/build/e94cfad945a84c3f9b89a84dddc59999 : SUCCESS in 5m 23s\n- tooz-tox-py311-etcd3gw https://zuul.opendev.org/t/openstack/build/06ff57c66ce54510b9111c85ac3a0f3d : SUCCESS in 6m 11s\n- tooz-tox-py314-etcd3gw https://zuul.opendev.org/t/openstack/build/5d12d6acdde9438eb50afba5fdc25fa6 : SUCCESS in 11m 12s\n- tooz-tox-py311-zookeeper https://zuul.opendev.org/t/openstack/build/57286ce622a9412490b7dc20d3fbb66d : SUCCESS in 4m 55s\n- tooz-tox-py314-zookeeper https://zuul.opendev.org/t/openstack/build/c786fbd130984236b1f9657e255e9329 : SUCCESS in 6m 44s\n- tooz-tox-py311-redis https://zuul.opendev.org/t/openstack/build/d80e29052d934e7981d14a231249b26d : SUCCESS in 6m 16s\n- tooz-tox-py314-redis https://zuul.opendev.org/t/openstack/build/74a896ac545843eb83eab31836a6655a : SUCCESS in 9m 16s\n- tooz-tox-py311-sentinel https://zuul.opendev.org/t/openstack/build/162fce108668491d8fd7e8298bbf4fc5 : SUCCESS in 6m 43s\n- tooz-tox-py314-sentinel https://zuul.opendev.org/t/openstack/build/ce091dd323e74c3c83117e2f88e0a14c : SUCCESS in 11m 31s\n- tooz-tox-py311-memcached https://zuul.opendev.org/t/openstack/build/5b0f7f4042ae425ab621864973d3af0e : SUCCESS in 5m 06s\n- tooz-tox-py314-memcached https://zuul.opendev.org/t/openstack/build/1eb8813f0c4642aa8572430046c17d6b : SUCCESS in 7m 44s\n- tooz-tox-py311-postgresql https://zuul.opendev.org/t/openstack/build/c08800cccdc24bfca0746686b84bd645 : SUCCESS in 5m 09s\n- tooz-tox-py314-postgresql https://zuul.opendev.org/t/openstack/build/81d34d8737804078b31a48850a1bede4 : SUCCESS in 5m 19s\n- tooz-tox-py311-mysql https://zuul.opendev.org/t/openstack/build/085a96cfbf1e460fb3256689f9abef35 : SUCCESS in 10m 40s\n- tooz-tox-py314-mysql https://zuul.opendev.org/t/openstack/build/055105c151fc4dffb8fffbe5342ad85c : SUCCESS in 7m 04s\n- tooz-tox-py311-consul https://zuul.opendev.org/t/openstack/build/8a7d66ab1d6b47229cfa1c43d4961260 : SUCCESS in 6m 02s\n- tooz-tox-py314-consul https://zuul.opendev.org/t/openstack/build/1fb1fcd4b9a84aa4ba45b91b065fd8a5 : SUCCESS in 7m 31s","accounts_in_message":[],"_revision_number":1},{"id":"3082b8eabd838ba2d24d5c30bece0bd9148c52f5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-30 04:26:43.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c3ba08149382fbae3695e5fe5f187c39530d600b","revisions":{"c3ba08149382fbae3695e5fe5f187c39530d600b":{"kind":"REWORK","_number":1,"created":"2026-06-29 20:28:32.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/17/995417/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/17/995417/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/17/995417/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/17/995417/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/17/995417/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/17/995417/1"}}},"commit":{"parents":[{"commit":"32fc851ab8a4959d60ab55574fa65168d3add785","subject":"tests: Fix ProcessPoolExecutor race on Python 3.14+","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/32fc851ab8a4959d60ab55574fa65168d3add785"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-06-29 20:23:29.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-06-29 20:28:23.000000000","tz":60},"subject":"Add Python 3.14 support","message":"Add Python 3.14 support\n\nChange-Id: I7507ac2907453562b64c135a3a46adcc007df4a2\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/c3ba08149382fbae3695e5fe5f187c39530d600b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/c3ba08149382fbae3695e5fe5f187c39530d600b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
