)]}'
{"id":"openstack%2Ftooz~996933","triplet_id":"openstack%2Ftooz~master~I6f04f5d09cffc825d336d5a11bf27370bbf69004","project":"openstack/tooz","branch":"master","topic":"tox-use-extras","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-08-23 23:01:07.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I6f04f5d09cffc825d336d5a11bf27370bbf69004","subject":"Remove duplicated optional dependencies from test requirements","status":"MERGED","created":"2026-07-12 10:48:21.000000000","updated":"2026-08-23 23:01:57.000000000","submitted":"2026-08-23 23:01:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"996933-tox-use-extras","meta_rev_id":"db8d90d2fc92cc86ce40f8338e48240914bb3f34","_number":996933,"virtual_id_number":996933,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-08-23 23:01:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2026-08-23 22:11:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2026-08-23 22:11:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-07-12 11:53:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-08-23 22:11:40.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"}],"messages":[{"id":"9a6d43d2db88b35e640e74cbd4d79a121e63f92e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-07-12 10:48:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"80aefe0127a057d63ea9942f8ecfd980749933e5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-12 11:53:36.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b8693d845aa346339ac89d1e95a2520d\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/044b41dacbf04112aaffcb6fc63a7801 : SUCCESS in 2m 53s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d4abe720288b4b5e97a22df4aa4b54d7 : SUCCESS in 1h 04m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a00cb8484ec84a0790b9405a33e6e51f : SUCCESS in 5m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6d00a9ee97224cec9d0cf2e98907f8f3 : SUCCESS in 4m 44s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3755421e77d14451be0eaaba319d9685 : SUCCESS in 5m 36s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6dd611925c434f6a812dc16c328de085 : SUCCESS in 3m 18s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/d6761b99ec6f459897e64732e9521adf : SUCCESS in 8m 35s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/a3f84fe6a67b4a0280ed11bd1739dfc0 : SUCCESS in 5m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f11df3587f3f44119aaf8575bfe06802 : FAILURE in 3m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7289af1189994dbeb588a6bfd4b9dcb4 : SUCCESS in 4m 40s\n- tooz-tox-examples https://zuul.opendev.org/t/openstack/build/398ca50561f8473e9e8b5b668699b285 : SUCCESS in 4m 19s\n- tooz-tox-py311-etcd3gw https://zuul.opendev.org/t/openstack/build/a510fffd57b04a2786f9a5434fcb1b1d : SUCCESS in 8m 42s\n- tooz-tox-py314-etcd3gw https://zuul.opendev.org/t/openstack/build/93d213c461d64f3c948e072b15e8e1ef : SUCCESS in 8m 24s\n- tooz-tox-py311-zookeeper https://zuul.opendev.org/t/openstack/build/8acf760114944cc197fa4988842e1ec3 : SUCCESS in 4m 15s\n- tooz-tox-py314-zookeeper https://zuul.opendev.org/t/openstack/build/4d1660799a97409ebec3c56ba175661e : SUCCESS in 7m 14s\n- tooz-tox-py311-redis https://zuul.opendev.org/t/openstack/build/5ad484a82ab0460586134eb2ffec0634 : SUCCESS in 6m 57s\n- tooz-tox-py314-redis https://zuul.opendev.org/t/openstack/build/10ca2bdc3976440ca757661976df3488 : SUCCESS in 9m 54s\n- tooz-tox-py311-sentinel https://zuul.opendev.org/t/openstack/build/49be192fbd12476ea6c01249972d7a20 : SUCCESS in 7m 30s\n- tooz-tox-py314-sentinel https://zuul.opendev.org/t/openstack/build/2298cbf9d1614536a5a49bf0f6980414 : SUCCESS in 12m 32s\n- tooz-tox-py311-memcached https://zuul.opendev.org/t/openstack/build/6baafceec709478f862f88a15f55c3a3 : SUCCESS in 10m 18s\n- tooz-tox-py314-memcached https://zuul.opendev.org/t/openstack/build/23c291e37a244cfdaba9f0d58f578d8b : SUCCESS in 7m 12s\n- tooz-tox-py311-postgresql https://zuul.opendev.org/t/openstack/build/6f46a3eebfd34aaea562f98d6942ca04 : SUCCESS in 4m 34s\n- tooz-tox-py314-postgresql https://zuul.opendev.org/t/openstack/build/d1add132c5f84c9d9fb16d8af77ec001 : SUCCESS in 10m 05s\n- tooz-tox-py311-mysql https://zuul.opendev.org/t/openstack/build/a723639ad32c43988f2c1392360354a7 : SUCCESS in 6m 12s\n- tooz-tox-py314-mysql https://zuul.opendev.org/t/openstack/build/fd1fd7eea0674cb799786bbd8044a2e9 : SUCCESS in 4m 29s","accounts_in_message":[],"_revision_number":1},{"id":"1ef281a74ac96d01dd24c7e07db4fe071f387ad0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-07-12 13:32:33.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"cb613a913ee326529b3738dbbac82324f189c7dd","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-07-12 13:32:50.000000000","message":"Topic set to tox-user-extras","accounts_in_message":[],"_revision_number":2},{"id":"978ba876c6288af05a1bd9c3883d9f6c57770c9c","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-07-12 13:37:27.000000000","message":"Topic tox-user-extras removed","accounts_in_message":[],"_revision_number":2},{"id":"00200cfdd4630f0cacce3468fe3c22c8167f161f","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-07-12 13:37:34.000000000","message":"Topic set to tox-use-extras","accounts_in_message":[],"_revision_number":2},{"id":"52de19dd0262b3bcba76b245b5d133ce7a30f8e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-12 14:18:51.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5f5eca8d2c5a48d18aa700960d02eecc\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/31a3dd2ccca841429b9980627572bcbb : SUCCESS in 3m 06s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/39b64ecf56d442859055dded8334f50c : SUCCESS in 45m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a6cc3b200b564891901ba6300d865dd4 : SUCCESS in 4m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/07675fbff7f548eb866431411e4447fe : SUCCESS in 3m 59s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ac4ed5bd6aea4a4ba460a646fdb6af9f : SUCCESS in 4m 48s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5b380edfc1c846908e4d4c603893ac0f : SUCCESS in 5m 00s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/828cd782e13f45c8845dee95c7e51d91 : SUCCESS in 7m 09s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/cd7f942b4bbb4ad689f0e52d7922b9a5 : SUCCESS in 5m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f5ea7c109bc14dcb95baae32f42883f4 : SUCCESS in 4m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/506ea92adf864a139b2b1aea8c202d74 : SUCCESS in 4m 19s\n- tooz-tox-examples https://zuul.opendev.org/t/openstack/build/6b8a4aef71e54dd1a6d113139dc23c3e : SUCCESS in 3m 52s\n- tooz-tox-py311-etcd3gw https://zuul.opendev.org/t/openstack/build/d6bfc707debc454bbc84da05475a8b69 : SUCCESS in 4m 53s\n- tooz-tox-py314-etcd3gw https://zuul.opendev.org/t/openstack/build/68f455f0f4944138ad574cfce61001bf : SUCCESS in 8m 03s\n- tooz-tox-py311-zookeeper https://zuul.opendev.org/t/openstack/build/bbd7b22e80a04f3898abce87d9338834 : SUCCESS in 4m 35s\n- tooz-tox-py314-zookeeper https://zuul.opendev.org/t/openstack/build/0f2503f40f8e434a85b3b5129ceb5923 : SUCCESS in 7m 02s\n- tooz-tox-py311-redis https://zuul.opendev.org/t/openstack/build/c5ec0df4220943a9abec26e17f8fef70 : SUCCESS in 5m 36s\n- tooz-tox-py314-redis https://zuul.opendev.org/t/openstack/build/0d6382cc82674ef3984daa483d7668c9 : SUCCESS in 8m 44s\n- tooz-tox-py311-sentinel https://zuul.opendev.org/t/openstack/build/a0ea2632bbd24ab8b2e743d30b213ede : SUCCESS in 4m 45s\n- tooz-tox-py314-sentinel https://zuul.opendev.org/t/openstack/build/2f046319809147328804f0aebe4c6f9d : SUCCESS in 9m 04s\n- tooz-tox-py311-memcached https://zuul.opendev.org/t/openstack/build/69be91d10f7b4463a8feb7ba2b2fbf39 : SUCCESS in 4m 38s\n- tooz-tox-py314-memcached https://zuul.opendev.org/t/openstack/build/978e1834938a4185b3182b1acfdd9e16 : SUCCESS in 7m 41s\n- tooz-tox-py311-postgresql https://zuul.opendev.org/t/openstack/build/cfb569f934c04ed1aaca1523b49f52ce : SUCCESS in 3m 42s\n- tooz-tox-py314-postgresql https://zuul.opendev.org/t/openstack/build/9e2c08f507d346ff87ff3bc8f1dd50d1 : SUCCESS in 8m 57s\n- tooz-tox-py311-mysql https://zuul.opendev.org/t/openstack/build/91ffec08770844498693973ef9ee4d97 : SUCCESS in 7m 14s\n- tooz-tox-py314-mysql https://zuul.opendev.org/t/openstack/build/8b51f88fc08c492390c1ee421eb93137 : SUCCESS in 7m 41s","accounts_in_message":[],"_revision_number":2},{"id":"7821461da64d6a00beb1388dcb8c8ec3086c8442","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-08-23 22:11:40.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"fdcfe7c0fccd8d2acda10aab38a7931edb60084e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-23 22:12:12.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"313d9a24b41ea56bd9956f8fd72189f131a09c02","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-23 23:01:07.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ac6a46f2dec34842bdcce6c438d84ea3\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/48863b6eaeca4c6fbe67ec022cefd126 : SUCCESS in 4m 32s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1868e2934c6248c0bb9a56f02805e5c3 : SUCCESS in 48m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/23c2b994b29b4aceb01b5c731c05250b : SUCCESS in 3m 46s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c3275e1a64744423afd64278fc9f34a6 : SUCCESS in 5m 48s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4b6e9788a3ae41c7b76d3131f39f302c : SUCCESS in 4m 51s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/abef77781798480b81709b244725cf0c : SUCCESS in 5m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac7dfa4f647f4c2c829adde7be68fd00 : SUCCESS in 5m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3e60dbabf58c4185ba4cf205873aaabd : SUCCESS in 5m 51s\n- tooz-tox-examples https://zuul.opendev.org/t/openstack/build/9169cd5ad7e6401789e19b1a3b8c1dbf : SUCCESS in 4m 46s\n- tooz-tox-py311-etcd3gw https://zuul.opendev.org/t/openstack/build/6e880a93f3e24c4584ff18ee97a173f3 : SUCCESS in 5m 16s\n- tooz-tox-py314-etcd3gw https://zuul.opendev.org/t/openstack/build/3958d7f1e9764a138ec632d29738ae25 : SUCCESS in 8m 20s\n- tooz-tox-py311-zookeeper https://zuul.opendev.org/t/openstack/build/22bc6791d66b4c81b87b056fe35ba86e : SUCCESS in 5m 16s\n- tooz-tox-py314-zookeeper https://zuul.opendev.org/t/openstack/build/93d04cd077574e61b52b4fb242fe9c8f : SUCCESS in 7m 22s\n- tooz-tox-py311-redis https://zuul.opendev.org/t/openstack/build/7a4fceecbf1d498085dbefb023277d72 : SUCCESS in 11m 03s\n- tooz-tox-py314-redis https://zuul.opendev.org/t/openstack/build/eae2cd8ca889442bb2819c0e611100f3 : SUCCESS in 8m 30s\n- tooz-tox-py311-sentinel https://zuul.opendev.org/t/openstack/build/f46feef8f8524db597a759ae8524d805 : SUCCESS in 5m 41s\n- tooz-tox-py314-sentinel https://zuul.opendev.org/t/openstack/build/9a5d29c27dfa4d05bd3c776f23aba75e : SUCCESS in 8m 02s\n- tooz-tox-py311-memcached https://zuul.opendev.org/t/openstack/build/e3a4b94883d649949991fbce89542592 : SUCCESS in 5m 02s\n- tooz-tox-py314-memcached https://zuul.opendev.org/t/openstack/build/cd3be859738b49aa9a96531567b4fbaf : SUCCESS in 7m 11s\n- tooz-tox-py311-postgresql https://zuul.opendev.org/t/openstack/build/1a07d7ffb2fa48dcb5bc6e4628fc33b2 : SUCCESS in 4m 45s\n- tooz-tox-py314-postgresql https://zuul.opendev.org/t/openstack/build/abbc27191b9e4cd993556dd16ba7ba0f : SUCCESS in 7m 22s\n- tooz-tox-py311-mysql https://zuul.opendev.org/t/openstack/build/8a677553641f40cdbc106e83b597751b : SUCCESS in 7m 39s\n- tooz-tox-py314-mysql https://zuul.opendev.org/t/openstack/build/7549b3ce4ba9481e889e253abe2bfb0a : SUCCESS in 8m 01s","accounts_in_message":[],"_revision_number":2},{"id":"c4703e65c123a5750c431803f41f29cedb8e2e6a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-23 23:01:07.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"db8d90d2fc92cc86ce40f8338e48240914bb3f34","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-23 23:01:57.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4d62deb7e3fa4dbd9c7f7a83ea0ca932\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3471ffe7d864443eacc1302f65abbb78 : SUCCESS in 38s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/06e125d9a6414464854166754e7d11ee : SUCCESS in 37s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"746ff4493786faaed06ede2e49b813003d1be2ac","revisions":{"f72739713570c2529e0a32fbfebac672b74fe11e":{"kind":"REWORK","_number":1,"created":"2026-07-12 10:48:21.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/33/996933/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/33/996933/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/33/996933/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/33/996933/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/33/996933/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/33/996933/1"}}},"commit":{"parents":[{"commit":"21fdc625dc938495ba4cff0204e87ec6fff0920e","subject":"Remove consul driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/21fdc625dc938495ba4cff0204e87ec6fff0920e"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-07-12 10:47:47.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-07-12 10:47:47.000000000","tz":540},"subject":"Remove duplicated optional dependencies from test requirements","message":"Remove duplicated optional dependencies from test requirements\n\n... and use extras to define these, instead.\n\nChange-Id: I6f04f5d09cffc825d336d5a11bf27370bbf69004\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/f72739713570c2529e0a32fbfebac672b74fe11e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/f72739713570c2529e0a32fbfebac672b74fe11e"}]},"branch":"refs/heads/master"},"746ff4493786faaed06ede2e49b813003d1be2ac":{"kind":"REWORK","_number":2,"created":"2026-07-12 13:32:33.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/33/996933/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/33/996933/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/33/996933/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/33/996933/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/33/996933/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/33/996933/2"}}},"commit":{"parents":[{"commit":"21fdc625dc938495ba4cff0204e87ec6fff0920e","subject":"Remove consul driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/21fdc625dc938495ba4cff0204e87ec6fff0920e"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-07-12 10:47:47.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-07-12 13:32:24.000000000","tz":540},"subject":"Remove duplicated optional dependencies from test requirements","message":"Remove duplicated optional dependencies from test requirements\n\n... and use extras to define these, instead.\n\nChange-Id: I6f04f5d09cffc825d336d5a11bf27370bbf69004\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/746ff4493786faaed06ede2e49b813003d1be2ac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/746ff4493786faaed06ede2e49b813003d1be2ac"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
