)]}'
{"id":"openstack%2Ftooz~997877","triplet_id":"openstack%2Ftooz~master~I63e44b31891e35c0dfb6a4319b62f5774d6b0cbd","project":"openstack/tooz","branch":"master","hashtags":[],"change_id":"I63e44b31891e35c0dfb6a4319b62f5774d6b0cbd","subject":"zookeeper: Fix wrong member used by get_member_capabilities","status":"MERGED","created":"2026-07-19 15:08:40.000000000","updated":"2026-07-19 17:45:25.000000000","submitted":"2026-07-19 17:44:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"997877","meta_rev_id":"aa01af7681a55edac5e858e02d7982c771994e71","_number":997877,"virtual_id_number":997877,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-07-19 17:44:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2026-07-19 16:54:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2026-07-19 16:54:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-07-19 16:53:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7e3b4fab999b9f98f616223964b00accb73335dc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-07-19 15:08:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"87572012de7f54ee0ecc033eeb163056a2f7a24a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-19 16:53:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e3a5d4de5a094e0a8edd4a945963f7b8\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9bdfc8d33251475c8d46b6d230bae3ae : SUCCESS in 1h 39m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1927d1b9573842a5a0d12cc93c6cce41 : SUCCESS in 6m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d8899a9afb54979bfa3ba82934e561b : SUCCESS in 5m 59s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6d1bb23daecc47cd90ba65f6ea2d7b9b : SUCCESS in 5m 44s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/83a9fe7ecf3b4fce9c68dce696d567c8 : SUCCESS in 3m 53s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/7517e7fe114e4f258a7f1d803cd1192d : SUCCESS in 5m 14s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/0ddaa793ce664edb943ca866af9f5234 : SUCCESS in 6m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9996b02d24c44a5484b7c2c9f5871dbc : SUCCESS in 2m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dafa437f2173457bac0b52c650723552 : SUCCESS in 3m 19s\n- tooz-tox-examples https://zuul.opendev.org/t/openstack/build/d81f7755ccc74594b5f37009c4264a19 : SUCCESS in 5m 07s\n- tooz-tox-py311-etcd3gw https://zuul.opendev.org/t/openstack/build/2effe0421de74fd7b491fc442f67b5ea : SUCCESS in 5m 11s\n- tooz-tox-py314-etcd3gw https://zuul.opendev.org/t/openstack/build/f9b5afee1ab24252990dd5141403c35e : SUCCESS in 8m 07s\n- tooz-tox-py311-zookeeper https://zuul.opendev.org/t/openstack/build/a8f7147b01e0401f95d69d9ecb14b296 : SUCCESS in 4m 26s\n- tooz-tox-py314-zookeeper https://zuul.opendev.org/t/openstack/build/346ce6482939411fbe8c65153bc01636 : SUCCESS in 9m 35s\n- tooz-tox-py311-redis https://zuul.opendev.org/t/openstack/build/e4092dacbfb947fead995c3830dd9b50 : SUCCESS in 5m 39s\n- tooz-tox-py314-redis https://zuul.opendev.org/t/openstack/build/e45195336671419eb8a87417c7b59752 : SUCCESS in 7m 29s\n- tooz-tox-py311-sentinel https://zuul.opendev.org/t/openstack/build/4777203307a547daa9c4defaaf56a2af : SUCCESS in 4m 11s\n- tooz-tox-py314-sentinel https://zuul.opendev.org/t/openstack/build/7d498d50610c4c9ab976cc3bc56fc8f7 : SUCCESS in 10m 11s\n- tooz-tox-py311-memcached https://zuul.opendev.org/t/openstack/build/346086e37ff7410684f9e4edf00b106b : SUCCESS in 4m 48s\n- tooz-tox-py314-memcached https://zuul.opendev.org/t/openstack/build/b20efc317b2644489ccb9554cbf6b5e4 : SUCCESS in 7m 23s\n- tooz-tox-py311-postgresql https://zuul.opendev.org/t/openstack/build/69e00c5217214d98ab77b0014e768172 : SUCCESS in 2m 47s\n- tooz-tox-py314-postgresql https://zuul.opendev.org/t/openstack/build/07c1dfd7bb814acebfd7599c79298572 : SUCCESS in 7m 12s\n- tooz-tox-py311-mysql https://zuul.opendev.org/t/openstack/build/0a503784de99458c9b5a38e8037533ff : SUCCESS in 7m 49s\n- tooz-tox-py314-mysql https://zuul.opendev.org/t/openstack/build/f635601a4e7f4fb6aca0b63789a0266a : SUCCESS in 8m 13s","accounts_in_message":[],"_revision_number":1},{"id":"644d7700dfc2e5e961a1dfcca25cdd2486e9077a","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-07-19 16:54:31.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d24c07e2c437d2443c903d2a9beced446ba7c065","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-19 16:54:53.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"cfae1ae0c764a89269696e50bb8c39378a26c58a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-19 17:44:24.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f0eb13739fff43d3877050cca1c6e903\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3d88e7335760464b92ab79e1f72b67b0 : SUCCESS in 48m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/77db5cb2b87d4d4fa44cf7cc14a7faf0 : SUCCESS in 6m 45s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/aecb01c907984df68904120cbc5169b4 : SUCCESS in 4m 42s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/dc7ceddcc7494b388e81222a08ef4150 : SUCCESS in 5m 17s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/8d9bbb7edd51499bb93c061de18926d2 : SUCCESS in 6m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/beda0c84189947c8a58fc134dc31d835 : SUCCESS in 4m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/34d397c370e24ec3b6d2e2f6a60b0ee4 : SUCCESS in 6m 54s\n- tooz-tox-examples https://zuul.opendev.org/t/openstack/build/f411cd6ae8594e3facf37669374863d4 : SUCCESS in 5m 09s\n- tooz-tox-py311-etcd3gw https://zuul.opendev.org/t/openstack/build/dc8709f62c1d4cb6a412998ab40cb098 : SUCCESS in 4m 51s\n- tooz-tox-py314-etcd3gw https://zuul.opendev.org/t/openstack/build/02ed4ccf12be44c0ad099b00da175698 : SUCCESS in 6m 59s\n- tooz-tox-py311-zookeeper https://zuul.opendev.org/t/openstack/build/e6fa944c21d242dc8d31cc9157f0f786 : SUCCESS in 4m 22s\n- tooz-tox-py314-zookeeper https://zuul.opendev.org/t/openstack/build/aeb86065545d4822acf0e3d3493fc76a : SUCCESS in 6m 21s\n- tooz-tox-py311-redis https://zuul.opendev.org/t/openstack/build/0ac8ba3e83ba40109b2e296d85b805ae : SUCCESS in 4m 50s\n- tooz-tox-py314-redis https://zuul.opendev.org/t/openstack/build/7a8e27a5ad3b4acbb4a1fc577a51ad46 : SUCCESS in 7m 17s\n- tooz-tox-py311-sentinel https://zuul.opendev.org/t/openstack/build/87607dc7c8ae48aea3adfca5fb55d891 : SUCCESS in 5m 44s\n- tooz-tox-py314-sentinel https://zuul.opendev.org/t/openstack/build/bf0499f860be447cae96350a3c6dd3eb : SUCCESS in 7m 14s\n- tooz-tox-py311-memcached https://zuul.opendev.org/t/openstack/build/11139f141a644ca6b75b66a2bc737905 : SUCCESS in 4m 32s\n- tooz-tox-py314-memcached https://zuul.opendev.org/t/openstack/build/d6c5791957ff4380a24daf584f9362b0 : SUCCESS in 5m 10s\n- tooz-tox-py311-postgresql https://zuul.opendev.org/t/openstack/build/794285ca0868413fb70d90607e969213 : SUCCESS in 4m 27s\n- tooz-tox-py314-postgresql https://zuul.opendev.org/t/openstack/build/b8340352cb2947c0aa2ebe846b162996 : SUCCESS in 9m 35s\n- tooz-tox-py311-mysql https://zuul.opendev.org/t/openstack/build/652831e6276943088636c760356fb3ed : SUCCESS in 6m 02s\n- tooz-tox-py314-mysql https://zuul.opendev.org/t/openstack/build/3c811dc345d4418880a7a4c80a7e60e5 : SUCCESS in 10m 04s","accounts_in_message":[],"_revision_number":1},{"id":"3e46cd5a6ce6f4ca2312dc8541e0f483a11b14b8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-19 17:44:24.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"aa01af7681a55edac5e858e02d7982c771994e71","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-19 17:45:25.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/59d94786e1b04d9597cf135d41a902da\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/71b5bc5e5b9d40929d602b3d323cc3bc : SUCCESS in 50s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"173183cc848672149e206b88b89a47daa6b7d43d","revisions":{"173183cc848672149e206b88b89a47daa6b7d43d":{"kind":"REWORK","_number":1,"created":"2026-07-19 15:08:40.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/77/997877/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/77/997877/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/77/997877/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/77/997877/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/77/997877/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/77/997877/1"}}},"commit":{"parents":[{"commit":"21fdc625dc938495ba4cff0204e87ec6fff0920e","subject":"Remove consul driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/21fdc625dc938495ba4cff0204e87ec6fff0920e"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-07-19 15:07:17.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-07-19 15:08:08.000000000","tz":540},"subject":"zookeeper: Fix wrong member used by get_member_capabilities","message":"zookeeper: Fix wrong member used by get_member_capabilities\n\nThe method should select the member according to the specified id, not\nto the own member id.\n\nCo-Authored-By: Adam Harwell \u003cflux.adam@gmail.com\u003e\nChange-Id: I63e44b31891e35c0dfb6a4319b62f5774d6b0cbd\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/173183cc848672149e206b88b89a47daa6b7d43d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/173183cc848672149e206b88b89a47daa6b7d43d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
