)]}'
{"id":"openstack%2Ftooz~997919","triplet_id":"openstack%2Ftooz~master~I58181a3e4530ae8ac26772bb6ab89a0ef297cd91","project":"openstack/tooz","branch":"master","topic":"etcd-distribution","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-09-21 15:16:21.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I58181a3e4530ae8ac26772bb6ab89a0ef297cd91","subject":"Use etcd from distribution","status":"MERGED","created":"2026-07-20 02:16:54.000000000","updated":"2026-09-21 15:19:17.000000000","submitted":"2026-09-21 15:16:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"997919-etcd-distribution","meta_rev_id":"9f1ced486dc50112c76116e031203b059ea02a13","_number":997919,"virtual_id_number":997919,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-21 15:16:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2026-09-21 13:37:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2026-08-26 11:54:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2026-09-21 13:37:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-07-20 05:21:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-08-26 11:54:04.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2026-09-21 13:37:34.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"}],"messages":[{"id":"82b094db14d43d3d9fac9a9c054ee4f790b9f999","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-07-20 02:16:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4dd6c62aac8126a57dde070f18bfa1e297e49dbc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-07-20 02:21:33.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"6cc61492eda4e558a2d94e5b2634acb5b9623055","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-07-20 04:40:17.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"c97429134f392eda97c4b29b70c8a0a464856ee8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-20 05:21:38.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/26fe27bef6ec40f4a6e5e833506525d5\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a845d9bd1d564aab8ef3c99e7f4f2974 : SUCCESS in 39m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fa3586f884884c87a7f8cbafc7cd2b62 : SUCCESS in 3m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d3417de419704909af1162ddfe24230f : SUCCESS in 3m 09s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/73b21a885b8d4314ba386ab7af7d8102 : SUCCESS in 3m 59s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a8f5ace998f24fd5bd83c9e5e4e546a1 : SUCCESS in 6m 11s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/13c40739197f498584ccf81eb4f6a272 : SUCCESS in 4m 31s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/0bd17cefb83f4bcdb8ae95c3f850cde1 : SUCCESS in 10m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d7d7a991a9f7438f97e6a981a4e6833f : SUCCESS in 5m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e4aadb82403b43ed864f8b0a35e7b5cb : SUCCESS in 2m 28s\n- tooz-tox-examples https://zuul.opendev.org/t/openstack/build/19a6f7b4ec7f4bd7a68940b97490e906 : FAILURE in 4m 05s\n- tooz-tox-py311-etcd3gw https://zuul.opendev.org/t/openstack/build/70db5c1fcca2403da9d179111ee766fc : ERROR Unable to find playbook /var/lib/zuul/builds/70db5c1fcca2403da9d179111ee766fc/untrusted/project_2/opendev.org/openstack/tooz/playbooks/pre.yaml in 12s\n- tooz-tox-py314-etcd3gw https://zuul.opendev.org/t/openstack/build/10e743a474024fca960661817bf0bae2 : ERROR Unable to find playbook /var/lib/zuul/builds/10e743a474024fca960661817bf0bae2/untrusted/project_2/opendev.org/openstack/tooz/playbooks/pre.yaml in 12s\n- tooz-tox-py311-zookeeper https://zuul.opendev.org/t/openstack/build/ffb8dd97655a4600add36b89d0d63969 : ERROR Unable to find playbook /var/lib/zuul/builds/ffb8dd97655a4600add36b89d0d63969/untrusted/project_2/opendev.org/openstack/tooz/playbooks/pre.yaml in 11s\n- tooz-tox-py314-zookeeper https://zuul.opendev.org/t/openstack/build/ff3e0a259c7a4503b481f7cb5f49bc83 : ERROR Unable to find playbook /var/lib/zuul/builds/ff3e0a259c7a4503b481f7cb5f49bc83/untrusted/project_2/opendev.org/openstack/tooz/playbooks/pre.yaml in 13s\n- tooz-tox-py311-redis https://zuul.opendev.org/t/openstack/build/11124bb8dcc841d4b75628355ec2370c : ERROR Unable to find playbook /var/lib/zuul/builds/11124bb8dcc841d4b75628355ec2370c/untrusted/project_2/opendev.org/openstack/tooz/playbooks/pre.yaml in 13s\n- tooz-tox-py314-redis https://zuul.opendev.org/t/openstack/build/129f86fddfd041c58438647ca99205cf : ERROR Unable to find playbook /var/lib/zuul/builds/129f86fddfd041c58438647ca99205cf/untrusted/project_2/opendev.org/openstack/tooz/playbooks/pre.yaml in 10s\n- tooz-tox-py311-sentinel https://zuul.opendev.org/t/openstack/build/5f7a2e5fdfd6498e8d7796efd1350139 : ERROR Unable to find playbook /var/lib/zuul/builds/5f7a2e5fdfd6498e8d7796efd1350139/untrusted/project_2/opendev.org/openstack/tooz/playbooks/pre.yaml in 10s\n- tooz-tox-py314-sentinel https://zuul.opendev.org/t/openstack/build/095b1cef343f414fa30998429fa4b3be : ERROR Unable to find playbook /var/lib/zuul/builds/095b1cef343f414fa30998429fa4b3be/untrusted/project_2/opendev.org/openstack/tooz/playbooks/pre.yaml in 10s\n- tooz-tox-py311-memcached https://zuul.opendev.org/t/openstack/build/6eb453813cb04506a22029f795949cc1 : ERROR Unable to find playbook /var/lib/zuul/builds/6eb453813cb04506a22029f795949cc1/untrusted/project_2/opendev.org/openstack/tooz/playbooks/pre.yaml in 10s\n- tooz-tox-py314-memcached https://zuul.opendev.org/t/openstack/build/d3e6753b894e447da3f9025c6d0e6a74 : ERROR Unable to find playbook /var/lib/zuul/builds/d3e6753b894e447da3f9025c6d0e6a74/untrusted/project_2/opendev.org/openstack/tooz/playbooks/pre.yaml in 10s\n- tooz-tox-py311-postgresql https://zuul.opendev.org/t/openstack/build/0f14a787b4a042e2a560998de589115d : ERROR Unable to find playbook /var/lib/zuul/builds/0f14a787b4a042e2a560998de589115d/untrusted/project_2/opendev.org/openstack/tooz/playbooks/pre.yaml in 10s\n- tooz-tox-py314-postgresql https://zuul.opendev.org/t/openstack/build/0fcfe95447004a429c7726f8e06daf54 : ERROR Unable to find playbook /var/lib/zuul/builds/0fcfe95447004a429c7726f8e06daf54/untrusted/project_2/opendev.org/openstack/tooz/playbooks/pre.yaml in 13s\n- tooz-tox-py311-mysql https://zuul.opendev.org/t/openstack/build/47e4f34384ef4277a8f5167b2bb199de : ERROR Unable to find playbook /var/lib/zuul/builds/47e4f34384ef4277a8f5167b2bb199de/untrusted/project_2/opendev.org/openstack/tooz/playbooks/pre.yaml in 10s\n- tooz-tox-py314-mysql https://zuul.opendev.org/t/openstack/build/8bfc3d53e6d24b4c952ff9025ed7dec1 : ERROR Unable to find playbook /var/lib/zuul/builds/8bfc3d53e6d24b4c952ff9025ed7dec1/untrusted/project_2/opendev.org/openstack/tooz/playbooks/pre.yaml in 12s","accounts_in_message":[],"_revision_number":3},{"id":"1dfd665cc6d0715d6cb647ef47f87c7bafb7adf6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-07-20 07:25:46.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"360d7ae30bd3e777a7105afa4125bc37eab54e6c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-20 09:00:29.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/97992b1815944743a7d8a513b5cc3f71\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7f5be813cd6c40cdad9892fe3eff31ef : SUCCESS in 1h 28m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b3d375540d054e559edcafe3e130f438 : SUCCESS in 5m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/916bbca9c3be4354b6385f7b2df8b5c1 : SUCCESS in 5m 00s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9a72aee16f0c4beeab357702e7a14eb3 : SUCCESS in 4m 56s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/fbf026fb15054231bc727a74fd04d9d0 : SUCCESS in 5m 10s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/9409f2a3fb0a4ae69f362c61e472c9d9 : SUCCESS in 6m 44s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/519510715617494a89abfef396c4799c : SUCCESS in 8m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f8e1c44c75a4fd388bc5552fa7e6920 : SUCCESS in 4m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/be76bda837ab4417a122e8d1c133c995 : SUCCESS in 5m 31s\n- tooz-tox-examples https://zuul.opendev.org/t/openstack/build/30659a4b8a5745268e019a2a1ddf2bbd : FAILURE in 5m 05s\n- tooz-tox-py311-etcd3gw https://zuul.opendev.org/t/openstack/build/27c887a14d4343db96fa1a6171319ca6 : RETRY_LIMIT in 2m 56s\n- tooz-tox-py314-etcd3gw https://zuul.opendev.org/t/openstack/build/506801d2a19d4fa7a81650217b724c6f : RETRY_LIMIT in 4m 50s\n- tooz-tox-py311-zookeeper https://zuul.opendev.org/t/openstack/build/2d932683eea64739a3e69e408eb5a7aa : RETRY_LIMIT in 2m 34s\n- tooz-tox-py314-zookeeper https://zuul.opendev.org/t/openstack/build/364da400b30347d6a6a615f6e7cb2bec : RETRY_LIMIT in 4m 55s\n- tooz-tox-py311-redis https://zuul.opendev.org/t/openstack/build/da1b5741e41c4a5c8c17172c51c17eab : RETRY_LIMIT in 2m 27s\n- tooz-tox-py314-redis https://zuul.opendev.org/t/openstack/build/2f91d258e6ce43ecbcaa245f97758638 : RETRY_LIMIT in 5m 16s\n- tooz-tox-py311-sentinel https://zuul.opendev.org/t/openstack/build/0e2f80b1110a4f849ec81bccd01f7b69 : RETRY_LIMIT in 2m 53s\n- tooz-tox-py314-sentinel https://zuul.opendev.org/t/openstack/build/829f474467b14229be4319262467bebe : RETRY_LIMIT in 4m 44s\n- tooz-tox-py311-memcached https://zuul.opendev.org/t/openstack/build/0e7fb74333e24d73854f0706344efcae : RETRY_LIMIT in 2m 40s\n- tooz-tox-py314-memcached https://zuul.opendev.org/t/openstack/build/27ee7c26d1ba489991d6dfe222d1c045 : RETRY_LIMIT in 4m 56s\n- tooz-tox-py311-postgresql https://zuul.opendev.org/t/openstack/build/d3d6642d1e8f4197b5bf26b6baf91a0b : RETRY_LIMIT in 2m 23s\n- tooz-tox-py314-postgresql https://zuul.opendev.org/t/openstack/build/8fe0b7a4e7b84fa6b801b8abbd4d2711 : RETRY_LIMIT in 4m 50s\n- tooz-tox-py311-mysql https://zuul.opendev.org/t/openstack/build/8252aeeef4be4911bc5181af8363f514 : RETRY_LIMIT in 5m 34s\n- tooz-tox-py314-mysql https://zuul.opendev.org/t/openstack/build/17749c98abae4bd3ae99978cc55e1f57 : RETRY_LIMIT in 5m 06s","accounts_in_message":[],"_revision_number":4},{"id":"1d1907f28c2567f4cae1ebe7385ad962f6d41c1b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-07-20 11:17:32.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":5},{"id":"f8be8e03f0dbe6450fee7e061c7e8a8431c380d1","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-07-20 11:38:05.000000000","message":"Topic set to etcd-distribution","accounts_in_message":[],"_revision_number":5},{"id":"9186fb56f8f8ce43de2f2061abfeae45ebfe8302","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-20 13:24:49.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/65737fc3e4894c14b62702d2e45b8c24\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c81bfa3e6455434ab15969d71c512cb1 : SUCCESS in 1h 56m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2725c5f4bde8491989335b79e80365f5 : SUCCESS in 4m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/543f98f07ab54acab105052ac0a9f325 : SUCCESS in 7m 04s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e305eb90177a4874953264707274e268 : SUCCESS in 3m 47s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/37caa6f5a05a4ee6b88213ec78586a71 : SUCCESS in 4m 41s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/8f3fcf4c69b145f6b47ad668e942bb7a : SUCCESS in 5m 08s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/25122cd2feab427aa5563a090bf39594 : SUCCESS in 8m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7697addeb473452fa335c31444ae4367 : SUCCESS in 4m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f6700037d4f24b71acc61e2174ad0d4c : SUCCESS in 5m 59s\n- tooz-tox-examples https://zuul.opendev.org/t/openstack/build/d424af0bb3244dbda1ac64a0691c9b41 : FAILURE in 5m 15s\n- tooz-tox-py311-etcd3gw https://zuul.opendev.org/t/openstack/build/e34128eaf4cc4201a7f09f017a5a6520 : SUCCESS in 4m 10s\n- tooz-tox-py314-etcd3gw https://zuul.opendev.org/t/openstack/build/b60c214c89b149c8a24b98e13fcdfc09 : SUCCESS in 7m 45s\n- tooz-tox-py311-zookeeper https://zuul.opendev.org/t/openstack/build/0db57a5174144621a23e1df7f2f229d1 : SUCCESS in 7m 10s\n- tooz-tox-py314-zookeeper https://zuul.opendev.org/t/openstack/build/1c6b925678d5495aac0f02becec42dd2 : SUCCESS in 7m 46s\n- tooz-tox-py311-redis https://zuul.opendev.org/t/openstack/build/537c7397dd65462a9d790d2f635ea5ad : SUCCESS in 3m 51s\n- tooz-tox-py314-redis https://zuul.opendev.org/t/openstack/build/180089b062654b12be6c49cd18fd5aab : SUCCESS in 10m 00s\n- tooz-tox-py311-sentinel https://zuul.opendev.org/t/openstack/build/329a09737980417484f1a78c50a7b2a3 : SUCCESS in 5m 03s\n- tooz-tox-py314-sentinel https://zuul.opendev.org/t/openstack/build/81cf0a30e4dd42b0b402a79affef78aa : SUCCESS in 8m 12s\n- tooz-tox-py311-memcached https://zuul.opendev.org/t/openstack/build/c68fe67e1e5e472ab65214408970b134 : RETRY_LIMIT in 2m 57s\n- tooz-tox-py314-memcached https://zuul.opendev.org/t/openstack/build/1ebcad7554c64ac2ad4f3e9aeec3c5e6 : RETRY_LIMIT in 6m 37s\n- tooz-tox-py311-postgresql https://zuul.opendev.org/t/openstack/build/d4bd4579e11e433eaa4e0e27200a01bb : RETRY_LIMIT in 3m 57s\n- tooz-tox-py314-postgresql https://zuul.opendev.org/t/openstack/build/479e8cabc8144e49a1f6436b719dff24 : RETRY_LIMIT in 6m 22s\n- tooz-tox-py311-mysql https://zuul.opendev.org/t/openstack/build/b369e4ff2d1843608135252c84ce4aba : RETRY_LIMIT in 5m 51s\n- tooz-tox-py314-mysql https://zuul.opendev.org/t/openstack/build/0dac1fb39e46410d8d61902996e12f51 : RETRY_LIMIT in 5m 43s","accounts_in_message":[],"_revision_number":5},{"id":"051fd34c21a07ddf12e2d8511ef33437aa77ebf0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-07-20 13:31:58.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":6},{"id":"e7178251aad07d04517c635f5faaa4581511f1a1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-20 15:01:26.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d225a5420d5644c09937b67a7a41c434\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/623a698c97e440ce8768ed6a80793336 : SUCCESS in 1h 26m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bcc69eb9d5ed43fe9890ce84f04a98f7 : SUCCESS in 6m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/48afd1569c3e40a9b1c69b055793339c : SUCCESS in 6m 39s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/bc0b063506e84fd6979e7affb737f184 : SUCCESS in 4m 20s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/81694929ee88490e9dfca3c5ab654bef : SUCCESS in 4m 44s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/05a94d7863324554b6fabce006def121 : SUCCESS in 7m 58s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/e7a79d02db5d472cbd578ee7cd6401b5 : SUCCESS in 8m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d4d2e5d0bc7745f3a013d797314e9781 : SUCCESS in 5m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/24d45b15a0184fd3b152c5d64df00a0e : SUCCESS in 5m 23s\n- tooz-tox-examples https://zuul.opendev.org/t/openstack/build/03f7f76b090744dcae6766a17787d3ed : FAILURE in 4m 51s\n- tooz-tox-py311-etcd3gw https://zuul.opendev.org/t/openstack/build/c24f3ffe712e41b8ac0b84e8dfcc8c05 : SUCCESS in 5m 07s\n- tooz-tox-py314-etcd3gw https://zuul.opendev.org/t/openstack/build/76e8e5cef4e84256a32e0b91b2a1a82d : SUCCESS in 7m 23s\n- tooz-tox-py311-zookeeper https://zuul.opendev.org/t/openstack/build/f0cad2c874df4ae49f9f1d1528be6444 : SUCCESS in 4m 56s\n- tooz-tox-py314-zookeeper https://zuul.opendev.org/t/openstack/build/6b8f9627010740518a630882ffa2e3a1 : SUCCESS in 8m 44s\n- tooz-tox-py311-redis https://zuul.opendev.org/t/openstack/build/de334ba5a3a54ce9ac2ba173f35a1981 : SUCCESS in 7m 20s\n- tooz-tox-py314-redis https://zuul.opendev.org/t/openstack/build/9ea3a652cdb9451a81012f26f2a17e6e : SUCCESS in 10m 22s\n- tooz-tox-py311-sentinel https://zuul.opendev.org/t/openstack/build/3c2681cac7994eaf993d8e8d837eb778 : SUCCESS in 6m 59s\n- tooz-tox-py314-sentinel https://zuul.opendev.org/t/openstack/build/bd046cf3a29b441d86a713ec288a9ad7 : SUCCESS in 8m 45s\n- tooz-tox-py311-memcached https://zuul.opendev.org/t/openstack/build/60d49f10a7e04f3f9075e6e9eba806f8 : SUCCESS in 4m 52s\n- tooz-tox-py314-memcached https://zuul.opendev.org/t/openstack/build/bb81077ef5dc4380a31e549a6bf751be : SUCCESS in 7m 36s\n- tooz-tox-py311-postgresql https://zuul.opendev.org/t/openstack/build/9f6edb4b6e9e4938b3239bebf6a53577 : SUCCESS in 4m 48s\n- tooz-tox-py314-postgresql https://zuul.opendev.org/t/openstack/build/601240cf58104ca197cae4fd7ffa3664 : SUCCESS in 6m 55s\n- tooz-tox-py311-mysql https://zuul.opendev.org/t/openstack/build/e7e636341dfa4a6f9bd9f077dc62d51c : SUCCESS in 7m 28s\n- tooz-tox-py314-mysql https://zuul.opendev.org/t/openstack/build/9c976d3d85e842b8b63aa8fb5a9ac38d : SUCCESS in 7m 47s","accounts_in_message":[],"_revision_number":6},{"id":"13f1545bf297f05418e2b9ff9aac1aee447fd605","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-07-20 17:13:03.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":7},{"id":"a6d046f7cd88758d7fea582e316975053b5ea9a7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-20 18:47:43.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3f262e28e8a04184ae1aef4fab856452\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2a33d2ef73b54bafa79bb9f0dfc4e985 : SUCCESS in 1h 29m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dd79234502d54d03b2f70ddb396a6af7 : SUCCESS in 4m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d148010d8f284296a953f0c469d29d5d : SUCCESS in 3m 41s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/78f2cc65ef7b4e218d0fd4b8c5461948 : SUCCESS in 4m 57s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ad37cb4d2a3a478198b4968981c9e3c4 : SUCCESS in 4m 52s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/8c8953d91a324d4caa5b3081241df083 : SUCCESS in 6m 29s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/309bafd7b75d42258bdf563e77c45466 : SUCCESS in 8m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f39ce1dc3e4431aa0faeb4e19a22073 : SUCCESS in 4m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bc6f1ce828894533b4f3d42bdfb2e8be : SUCCESS in 6m 09s\n- tooz-tox-examples https://zuul.opendev.org/t/openstack/build/d6783496789140aabd467e5bbd6cd8fe : SUCCESS in 4m 25s\n- tooz-tox-py311-etcd3gw https://zuul.opendev.org/t/openstack/build/30ae550e84144bbc867aafb44441e12b : SUCCESS in 4m 20s\n- tooz-tox-py314-etcd3gw https://zuul.opendev.org/t/openstack/build/b6fb25ed675842a9bd573f5ff9557a10 : SUCCESS in 8m 00s\n- tooz-tox-py311-zookeeper https://zuul.opendev.org/t/openstack/build/db30e2a89fbb4649854f82f8bf50fb18 : SUCCESS in 5m 09s\n- tooz-tox-py314-zookeeper https://zuul.opendev.org/t/openstack/build/0e47a07f88ad4f43947f7ae9c598d711 : SUCCESS in 7m 40s\n- tooz-tox-py311-redis https://zuul.opendev.org/t/openstack/build/ec12f3b8f59640cda55f8a9891cdec2e : SUCCESS in 7m 14s\n- tooz-tox-py314-redis https://zuul.opendev.org/t/openstack/build/dd2f11ae27d84e44afbbc6e11d044e36 : SUCCESS in 8m 02s\n- tooz-tox-py311-sentinel https://zuul.opendev.org/t/openstack/build/b3326fd97a2b4e0d913eeb5860dfef22 : SUCCESS in 5m 01s\n- tooz-tox-py314-sentinel https://zuul.opendev.org/t/openstack/build/909a522430914c5aaf08b85e20343b8b : SUCCESS in 8m 12s\n- tooz-tox-py311-memcached https://zuul.opendev.org/t/openstack/build/8922ae5cbada4386815ed43d13a22865 : SUCCESS in 4m 55s\n- tooz-tox-py314-memcached https://zuul.opendev.org/t/openstack/build/9bcbb937e261441fa41435f4ead4294f : SUCCESS in 7m 37s\n- tooz-tox-py311-postgresql https://zuul.opendev.org/t/openstack/build/3f9521600d9e42e7aae2b2827da1b4ea : SUCCESS in 4m 22s\n- tooz-tox-py314-postgresql https://zuul.opendev.org/t/openstack/build/a8beaa4e39da4d6598f10571d4be33d1 : SUCCESS in 5m 12s\n- tooz-tox-py311-mysql https://zuul.opendev.org/t/openstack/build/bbe8b79b75b640fdabc6d5f421c25249 : RETRY_LIMIT in 5m 09s\n- tooz-tox-py314-mysql https://zuul.opendev.org/t/openstack/build/cd2c8c0c8ba542438e006340eba6184b : RETRY_LIMIT in 3m 54s","accounts_in_message":[],"_revision_number":7},{"id":"df1572cd78a893dedb7fb5bbbdd06e80dad39d79","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-08-24 16:48:33.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":8},{"id":"082150c904fbd81a1b1fa5afe8585278982a41ee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-24 18:00:40.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/13683415703948258099217e868452d1\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9380e72aaeda422e97b5f674ae933393 : SUCCESS in 1h 09m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7ea43d0dd336488db3b4390d6b8214b8 : SUCCESS in 4m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1f25ec9a59f44476840280d0b230d1cf : SUCCESS in 3m 00s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8bb153d4beec49a886ad402e1a659313 : FAILURE in 5m 02s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/dfe70fb9e27a443bb17f3f4d8e51ad54 : SUCCESS in 5m 21s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/3f6ffd2d8700478db2a6a173cb83060a : SUCCESS in 6m 41s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/ccdc71d9e62d40ec87888596b7e48e40 : SUCCESS in 9m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/71c781940d15401f81e0c417087999d7 : SUCCESS in 3m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cfceecab52344e3c8e4ccdd114885dc4 : SUCCESS in 3m 41s\n- tooz-tox-examples https://zuul.opendev.org/t/openstack/build/0165e62c52164f5ea78a4bbce938ee20 : SUCCESS in 4m 39s\n- tooz-tox-py311-etcd3gw https://zuul.opendev.org/t/openstack/build/dd1c6f292db44f609432f6f702d37e48 : RETRY_LIMIT in 3m 04s\n- tooz-tox-py314-etcd3gw https://zuul.opendev.org/t/openstack/build/c4efb604eafa4d499f31f115adbaee81 : RETRY_LIMIT in 5m 36s\n- tooz-tox-py311-zookeeper https://zuul.opendev.org/t/openstack/build/3f76e06166474267bb7cc620982a850c : RETRY_LIMIT in 3m 26s\n- tooz-tox-py314-zookeeper https://zuul.opendev.org/t/openstack/build/3fabefcf0adc4d6abf2e277d9921ecdd : RETRY_LIMIT in 5m 56s\n- tooz-tox-py311-redis https://zuul.opendev.org/t/openstack/build/cce997106b64476b830f1e5002aa4708 : RETRY_LIMIT in 2m 39s\n- tooz-tox-py314-redis https://zuul.opendev.org/t/openstack/build/112a44e987a84e31b807e95dcfd2fb43 : RETRY_LIMIT in 6m 43s\n- tooz-tox-py311-sentinel https://zuul.opendev.org/t/openstack/build/5561c661ec174a0b9e6655525c87c2da : RETRY_LIMIT in 2m 50s\n- tooz-tox-py314-sentinel https://zuul.opendev.org/t/openstack/build/919ea9a3f9534422843cacadbba1ab6e : RETRY_LIMIT in 5m 24s\n- tooz-tox-py311-memcached https://zuul.opendev.org/t/openstack/build/98c0836175ad4d83b48b85636a6d3c5c : RETRY_LIMIT in 2m 56s\n- tooz-tox-py314-memcached https://zuul.opendev.org/t/openstack/build/b5d15b8727b541fcba424f1c8f6d78c5 : RETRY_LIMIT in 5m 21s\n- tooz-tox-py311-postgresql https://zuul.opendev.org/t/openstack/build/783e64a10fa64e7cb31a4e815c4026b1 : RETRY_LIMIT in 2m 22s\n- tooz-tox-py314-postgresql https://zuul.opendev.org/t/openstack/build/18a1c432d62845a597e6587bcbdfff70 : RETRY_LIMIT in 5m 11s\n- tooz-tox-py311-mysql https://zuul.opendev.org/t/openstack/build/dcffadca532d4be8bdafd7ac2ce99f0a : RETRY_LIMIT in 3m 35s\n- tooz-tox-py314-mysql https://zuul.opendev.org/t/openstack/build/b8a444e3eaff40238404da62ec7f0ec7 : RETRY_LIMIT in 6m 00s","accounts_in_message":[],"_revision_number":8},{"id":"16b77abd15d8370b952b28430daf0b39531e6811","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-08-24 18:10:52.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":9},{"id":"fa1ef359c47e61986cb85e4649338842ca056043","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-24 19:55:52.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c8160538bdd8427aa696e7f5b507389a\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a117b05ba4a8458d89aef7ccca26cdc1 : SUCCESS in 1h 37m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/194ad096f2fd4183a86a4aaf43445229 : SUCCESS in 6m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b9b02200bc44936a536342af3ec9a10 : SUCCESS in 5m 21s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b62c443bfa8a4cc6b4c2cf55d81003c2 : SUCCESS in 3m 08s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5be2cbc665f54c219452c5d7c6477dd7 : SUCCESS in 4m 36s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/af0570537c464e8cb245952b9884247f : SUCCESS in 5m 10s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/dad644d36d684026b29cd26f6eb9b820 : SUCCESS in 5m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/26ce8f35932d4b92be45303a40e0f088 : SUCCESS in 4m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/312c2be9deaa484986669cc3ec685d7c : SUCCESS in 2m 18s\n- tooz-tox-examples https://zuul.opendev.org/t/openstack/build/ebd7616943c14c3381712565f3c3ce87 : SUCCESS in 2m 49s\n- tooz-tox-py311-etcd3gw https://zuul.opendev.org/t/openstack/build/795932fba0284d69ad3df342a4524774 : SUCCESS in 5m 40s\n- tooz-tox-py314-etcd3gw https://zuul.opendev.org/t/openstack/build/5723851215e04268826537758c0d6680 : SUCCESS in 5m 16s\n- tooz-tox-py311-zookeeper https://zuul.opendev.org/t/openstack/build/f2811c358e2c4b819d6d3d501be4bc0e : SUCCESS in 5m 56s\n- tooz-tox-py314-zookeeper https://zuul.opendev.org/t/openstack/build/b33a30763aa54305b78b9f6c88907d0b : SUCCESS in 7m 22s\n- tooz-tox-py311-redis https://zuul.opendev.org/t/openstack/build/83d615ca76574da2bb7278beb0e72178 : SUCCESS in 7m 11s\n- tooz-tox-py314-redis https://zuul.opendev.org/t/openstack/build/441db45d7a5942b9bbb5f95b7aff1e1c : SUCCESS in 5m 35s\n- tooz-tox-py311-sentinel https://zuul.opendev.org/t/openstack/build/f3baf48e32cc4e7e8dca476671890367 : SUCCESS in 5m 40s\n- tooz-tox-py314-sentinel https://zuul.opendev.org/t/openstack/build/2c81c10ed5344ff19d46f94f3998b9a2 : SUCCESS in 5m 27s\n- tooz-tox-py311-memcached https://zuul.opendev.org/t/openstack/build/7c84ccd5284d4d3cad0750eccb25b1fc : SUCCESS in 4m 35s\n- tooz-tox-py314-memcached https://zuul.opendev.org/t/openstack/build/6fa15e8b94aa4a8083fcded9e006017a : SUCCESS in 9m 11s\n- tooz-tox-py311-postgresql https://zuul.opendev.org/t/openstack/build/2b0066e90a004d9fbe317cf6e2962246 : SUCCESS in 2m 30s\n- tooz-tox-py314-postgresql https://zuul.opendev.org/t/openstack/build/2db2bfdd5d744810ab27f18d7691fe5f : SUCCESS in 7m 29s\n- tooz-tox-py311-mysql https://zuul.opendev.org/t/openstack/build/da71e3d2630f4f68baf502424de4110e : RETRY_LIMIT in 5m 19s\n- tooz-tox-py314-mysql https://zuul.opendev.org/t/openstack/build/b3f385d62173437c85419663f402f9fe : RETRY_LIMIT in 3m 31s","accounts_in_message":[],"_revision_number":9},{"id":"6392fb9b38d30567f101bce98f49677da25ab3af","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-08-25 16:14:27.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":10},{"id":"18272651c73971c59a4e81f5c9b062a23c47f675","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-25 18:24:13.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f253c85d96eb4e528cd3083a4df1220c\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/edf427eb0da3498c8b7b7462a1b64cf6 : SUCCESS in 1h 33m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8c07a480ced84d35ab42686e48bc2c85 : SUCCESS in 5m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8d2f3525ad674af49db1d5b31c822d8e : SUCCESS in 5m 13s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7bcf82eec8704443b80c117458b52293 : SUCCESS in 5m 56s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ac4d94dc3b154697ba81999e78ad065d : SUCCESS in 4m 32s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/30c7644f8ef34fd1810bf2888eea5844 : SUCCESS in 8m 18s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/f77aae3e9e6c4263972abbc32c12aa2d : SUCCESS in 11m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8beaefcf283a453cb7bcbb28b633ef46 : SUCCESS in 7m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f6d00efae09f4c89acae2038ee6230ba : SUCCESS in 4m 07s\n- tooz-tox-examples https://zuul.opendev.org/t/openstack/build/8b03775a86c24d5e962cadbf1788e968 : SUCCESS in 3m 48s\n- tooz-tox-py311-etcd3gw https://zuul.opendev.org/t/openstack/build/e89464311a5847f29d032e143d461571 : SUCCESS in 6m 17s\n- tooz-tox-py314-etcd3gw https://zuul.opendev.org/t/openstack/build/20b16ccb6bb74c8981a904715fff4d2e : SUCCESS in 7m 28s\n- tooz-tox-py311-zookeeper https://zuul.opendev.org/t/openstack/build/bdee8e619d374982880b1bfb5ccf9870 : SUCCESS in 4m 44s\n- tooz-tox-py314-zookeeper https://zuul.opendev.org/t/openstack/build/4c3f623a3062442a8f6f8c9d4123126a : SUCCESS in 10m 52s\n- tooz-tox-py311-redis https://zuul.opendev.org/t/openstack/build/4d98642a970d40c08df67b58ede3c5dc : SUCCESS in 5m 40s\n- tooz-tox-py314-redis https://zuul.opendev.org/t/openstack/build/6f3cf5d3d65f4dd9b3b44b42aac2cb91 : SUCCESS in 9m 01s\n- tooz-tox-py311-sentinel https://zuul.opendev.org/t/openstack/build/540bcde4dc134aa2b653f10547fba484 : SUCCESS in 10m 01s\n- tooz-tox-py314-sentinel https://zuul.opendev.org/t/openstack/build/5addeb21852945eb8d782ae4614a8f40 : SUCCESS in 11m 30s\n- tooz-tox-py311-memcached https://zuul.opendev.org/t/openstack/build/6a34d8f683174742a75cbb2c59ffb83c : SUCCESS in 4m 31s\n- tooz-tox-py314-memcached https://zuul.opendev.org/t/openstack/build/4ac8fc42d29340d6a02be76df1b47f18 : SUCCESS in 6m 07s\n- tooz-tox-py311-postgresql https://zuul.opendev.org/t/openstack/build/6e9937164b164bda82848d335759ae37 : SUCCESS in 4m 33s\n- tooz-tox-py314-postgresql https://zuul.opendev.org/t/openstack/build/ff6df8fffd8a4495a57c4e8d7959249a : SUCCESS in 6m 34s\n- tooz-tox-py311-mysql https://zuul.opendev.org/t/openstack/build/e6c7015a17774226b26c7135267a0f60 : SUCCESS in 8m 49s\n- tooz-tox-py314-mysql https://zuul.opendev.org/t/openstack/build/5b92d134266e4a6699bd4ac32b465e9d : SUCCESS in 6m 29s","accounts_in_message":[],"_revision_number":10},{"id":"0dbce7bad9aed0e82e51e177ae6d2e4226654352","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2026-08-26 11:54:04.000000000","message":"Patch Set 10: Code-Review+2","accounts_in_message":[],"_revision_number":10},{"id":"c71d014f4d08f866936ce705ebf72a2feebb225e","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-21 13:37:34.000000000","message":"Patch Set 10: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":10},{"id":"a595feeb248b01a81475d3a6e967712e4e99c3d8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 13:38:13.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"0f2f82638da823f132b63d158ddf0b8eb4dc1ac5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 15:16:21.000000000","message":"Patch Set 10: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0b23e7da4e4d42f58a6d341783cfbd3a\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/165c5942bb0f4b5b8e6da446d0571e0d : SUCCESS in 1h 36m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/800ce18f5e3242a79ad3a56e9a2d7e1e : SUCCESS in 6m 27s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/cfd1abc44bb04d9483053be5d1340b5b : SUCCESS in 4m 51s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5e7dff50d9c84918a31f786501a8fad2 : SUCCESS in 6m 43s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/5b01a349e02a4a43adbe4bb5b6a13701 : SUCCESS in 9m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/59459451daa14a6e992fd77e36dd328f : SUCCESS in 4m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/89bf96f3032146fb8d3df4cd7171c8cd : SUCCESS in 3m 11s\n- tooz-tox-examples https://zuul.opendev.org/t/openstack/build/1328ae55f2274c9797af4b9d04b55106 : SUCCESS in 5m 15s\n- tooz-tox-py311-etcd3gw https://zuul.opendev.org/t/openstack/build/496165ee51d240988aa5aa18a525babe : SUCCESS in 4m 49s\n- tooz-tox-py314-etcd3gw https://zuul.opendev.org/t/openstack/build/14b8990514e444138f3d09c31cfbdeaa : SUCCESS in 7m 58s\n- tooz-tox-py311-zookeeper https://zuul.opendev.org/t/openstack/build/05893f98387c4148b348a4902dc5ddd1 : SUCCESS in 7m 34s\n- tooz-tox-py314-zookeeper https://zuul.opendev.org/t/openstack/build/d39f3615600b448a9555df120db0d6a1 : SUCCESS in 5m 47s\n- tooz-tox-py311-redis https://zuul.opendev.org/t/openstack/build/0ff4a89577b042d38e597e0d694c0b81 : SUCCESS in 5m 51s\n- tooz-tox-py314-redis https://zuul.opendev.org/t/openstack/build/eaee537676fd4c9f861c799e406f7079 : SUCCESS in 10m 16s\n- tooz-tox-py311-sentinel https://zuul.opendev.org/t/openstack/build/6aa6ece744d7450d997cdf064202161c : SUCCESS in 4m 49s\n- tooz-tox-py314-sentinel https://zuul.opendev.org/t/openstack/build/72ee32e7d82f43dea5d135279d4aea07 : SUCCESS in 6m 51s\n- tooz-tox-py311-memcached https://zuul.opendev.org/t/openstack/build/12c03943c6fd4af9b2351953f081092d : SUCCESS in 4m 48s\n- tooz-tox-py314-memcached https://zuul.opendev.org/t/openstack/build/3babaeff4fe540959cd42cc3965153c3 : SUCCESS in 7m 20s\n- tooz-tox-py311-postgresql https://zuul.opendev.org/t/openstack/build/e03e9baf3aa94c43aa89fe37d1122ed0 : SUCCESS in 4m 28s\n- tooz-tox-py314-postgresql https://zuul.opendev.org/t/openstack/build/d265fd973cf44d1cbb39ad87876c74b4 : SUCCESS in 6m 35s\n- tooz-tox-py311-mysql https://zuul.opendev.org/t/openstack/build/ed46394a99a841b688ee18c12a4ee3a0 : SUCCESS in 6m 57s\n- tooz-tox-py314-mysql https://zuul.opendev.org/t/openstack/build/7fd95e8e0a5b44199c6e36aae7f35a79 : SUCCESS in 9m 32s","accounts_in_message":[],"_revision_number":10},{"id":"2e2dad083e0bbce4cbe5582a7d09c7a223a044c5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 15:16:21.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":10},{"id":"9f1ced486dc50112c76116e031203b059ea02a13","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 15:19:17.000000000","message":"Patch Set 10:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4cac9f1595794822b8b388ab77008273\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/07daf713a62346a8aecda258b649429e : SUCCESS in 1m 01s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/711affa871ec44d7ab3b925d715542fc : SUCCESS in 47s","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"2a8d47077904ee20464c1f6e349f1751dcd57374","revisions":{"1c06b9b9e3a4a159c913b73152cd2e6ea042608d":{"kind":"REWORK","_number":1,"created":"2026-07-20 02:16:54.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/19/997919/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/19/997919/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/19/997919/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/19/997919/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/19/997919/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/19/997919/1"}}},"commit":{"parents":[{"commit":"173183cc848672149e206b88b89a47daa6b7d43d","subject":"zookeeper: Fix wrong member used by get_member_capabilities","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/173183cc848672149e206b88b89a47daa6b7d43d"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-07-20 02:13:14.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-07-20 02:16:06.000000000","tz":540},"subject":"Use etcd from distribution","message":"Use etcd from distribution\n\netcd has been available in Ubuntu and Debian for some time.\n\nUse the distribution package to install it, following the other\nbackends like redis.\n\nChange-Id: I58181a3e4530ae8ac26772bb6ab89a0ef297cd91\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/1c06b9b9e3a4a159c913b73152cd2e6ea042608d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/1c06b9b9e3a4a159c913b73152cd2e6ea042608d"}]},"branch":"refs/heads/master"},"d774dbffcaf7afb620665f50bde266efb307cd36":{"kind":"REWORK","_number":2,"created":"2026-07-20 02:21:33.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/19/997919/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/19/997919/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/19/997919/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/19/997919/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/19/997919/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/19/997919/2"}}},"commit":{"parents":[{"commit":"173183cc848672149e206b88b89a47daa6b7d43d","subject":"zookeeper: Fix wrong member used by get_member_capabilities","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/173183cc848672149e206b88b89a47daa6b7d43d"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-07-20 02:13:14.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-07-20 02:21:22.000000000","tz":540},"subject":"Use etcd from distribution","message":"Use etcd from distribution\n\netcd has been available in Ubuntu and Debian for some time.\n\nUse the distribution package to install it, following the other\nbackends like redis.\n\nChange-Id: I58181a3e4530ae8ac26772bb6ab89a0ef297cd91\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/d774dbffcaf7afb620665f50bde266efb307cd36"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/d774dbffcaf7afb620665f50bde266efb307cd36"}]},"branch":"refs/heads/master"},"1812e8de90b128212481c3b214b841a80c2c00fa":{"kind":"REWORK","_number":3,"created":"2026-07-20 04:40:17.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/19/997919/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/19/997919/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/19/997919/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/19/997919/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/19/997919/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/19/997919/3"}}},"commit":{"parents":[{"commit":"173183cc848672149e206b88b89a47daa6b7d43d","subject":"zookeeper: Fix wrong member used by get_member_capabilities","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/173183cc848672149e206b88b89a47daa6b7d43d"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-07-20 02:13:14.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-07-20 04:40:02.000000000","tz":540},"subject":"Use etcd from distribution","message":"Use etcd from distribution\n\netcd has been available in Ubuntu and Debian for some time.\n\nUse the distribution package to install it, following the other\nbackends like redis.\n\nChange-Id: I58181a3e4530ae8ac26772bb6ab89a0ef297cd91\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/1812e8de90b128212481c3b214b841a80c2c00fa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/1812e8de90b128212481c3b214b841a80c2c00fa"}]},"branch":"refs/heads/master"},"8c2b7a93845b3d9dee6079d086af94268186c7a5":{"kind":"REWORK","_number":4,"created":"2026-07-20 07:25:46.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/19/997919/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/19/997919/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/19/997919/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/19/997919/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/19/997919/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/19/997919/4"}}},"commit":{"parents":[{"commit":"173183cc848672149e206b88b89a47daa6b7d43d","subject":"zookeeper: Fix wrong member used by get_member_capabilities","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/173183cc848672149e206b88b89a47daa6b7d43d"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-07-20 02:13:14.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-07-20 07:25:10.000000000","tz":540},"subject":"Use etcd from distribution","message":"Use etcd from distribution\n\netcd has been available in Ubuntu and Debian for some time.\n\nUse the distribution package to install it, following the other\nbackends like redis.\n\nChange-Id: I58181a3e4530ae8ac26772bb6ab89a0ef297cd91\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/8c2b7a93845b3d9dee6079d086af94268186c7a5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/8c2b7a93845b3d9dee6079d086af94268186c7a5"}]},"branch":"refs/heads/master"},"f367f5afe68632b470f7127eb0d1064f79509ee2":{"kind":"REWORK","_number":5,"created":"2026-07-20 11:17:32.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/19/997919/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/19/997919/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/19/997919/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/19/997919/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/19/997919/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/19/997919/5"}}},"commit":{"parents":[{"commit":"173183cc848672149e206b88b89a47daa6b7d43d","subject":"zookeeper: Fix wrong member used by get_member_capabilities","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/173183cc848672149e206b88b89a47daa6b7d43d"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-07-20 02:13:14.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-07-20 11:17:21.000000000","tz":540},"subject":"Use etcd from distribution","message":"Use etcd from distribution\n\netcd has been available in Ubuntu and Debian for some time.\n\nUse the distribution package to install it, following the other\nbackends like redis.\n\nChange-Id: I58181a3e4530ae8ac26772bb6ab89a0ef297cd91\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/f367f5afe68632b470f7127eb0d1064f79509ee2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/f367f5afe68632b470f7127eb0d1064f79509ee2"}]},"branch":"refs/heads/master"},"a40882a4deed298084785475d6b5946d3a9d5fd4":{"kind":"REWORK","_number":6,"created":"2026-07-20 13:31:58.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/19/997919/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/19/997919/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/19/997919/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/19/997919/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/19/997919/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/19/997919/6"}}},"commit":{"parents":[{"commit":"173183cc848672149e206b88b89a47daa6b7d43d","subject":"zookeeper: Fix wrong member used by get_member_capabilities","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/173183cc848672149e206b88b89a47daa6b7d43d"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-07-20 02:13:14.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-07-20 13:31:43.000000000","tz":540},"subject":"Use etcd from distribution","message":"Use etcd from distribution\n\netcd has been available in Ubuntu and Debian for some time.\n\nUse the distribution package to install it, following the other\nbackends like redis.\n\nChange-Id: I58181a3e4530ae8ac26772bb6ab89a0ef297cd91\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/a40882a4deed298084785475d6b5946d3a9d5fd4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/a40882a4deed298084785475d6b5946d3a9d5fd4"}]},"branch":"refs/heads/master"},"53074fe5316d13c61e147901aa5747ca57d1b4d2":{"kind":"REWORK","_number":7,"created":"2026-07-20 17:13:03.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/19/997919/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/19/997919/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/19/997919/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/19/997919/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/19/997919/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/19/997919/7"}}},"commit":{"parents":[{"commit":"173183cc848672149e206b88b89a47daa6b7d43d","subject":"zookeeper: Fix wrong member used by get_member_capabilities","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/173183cc848672149e206b88b89a47daa6b7d43d"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-07-20 02:13:14.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-07-20 17:12:53.000000000","tz":540},"subject":"Use etcd from distribution","message":"Use etcd from distribution\n\netcd has been available in Ubuntu and Debian for some time.\n\nUse the distribution package to install it, following the other\nbackends like redis.\n\nChange-Id: I58181a3e4530ae8ac26772bb6ab89a0ef297cd91\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/53074fe5316d13c61e147901aa5747ca57d1b4d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/53074fe5316d13c61e147901aa5747ca57d1b4d2"}]},"branch":"refs/heads/master"},"5cfbee19afc5c00c69e7388a0215c352f7408803":{"kind":"REWORK","_number":8,"created":"2026-08-24 16:48:33.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/19/997919/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/19/997919/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/19/997919/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/19/997919/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/19/997919/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/19/997919/8"}}},"commit":{"parents":[{"commit":"173183cc848672149e206b88b89a47daa6b7d43d","subject":"zookeeper: Fix wrong member used by get_member_capabilities","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/173183cc848672149e206b88b89a47daa6b7d43d"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-07-20 02:13:14.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-08-24 16:48:19.000000000","tz":540},"subject":"Use etcd from distribution","message":"Use etcd from distribution\n\netcd has been available in Ubuntu and Debian for some time.\n\nUse the distribution package to install it, following the other\nbackends like redis.\n\nChange-Id: I58181a3e4530ae8ac26772bb6ab89a0ef297cd91\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/5cfbee19afc5c00c69e7388a0215c352f7408803"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/5cfbee19afc5c00c69e7388a0215c352f7408803"}]},"branch":"refs/heads/master"},"ccc24c0b1ebaa5d17bd60b7c020e0a8c11a8e44b":{"kind":"REWORK","_number":9,"created":"2026-08-24 18:10:52.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/19/997919/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/19/997919/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/19/997919/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/19/997919/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/19/997919/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/19/997919/9"}}},"commit":{"parents":[{"commit":"173183cc848672149e206b88b89a47daa6b7d43d","subject":"zookeeper: Fix wrong member used by get_member_capabilities","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/173183cc848672149e206b88b89a47daa6b7d43d"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-07-20 02:13:14.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-08-24 18:10:34.000000000","tz":540},"subject":"Use etcd from distribution","message":"Use etcd from distribution\n\netcd has been available in Ubuntu and Debian for some time.\n\nUse the distribution package to install it, following the other\nbackends like redis.\n\nChange-Id: I58181a3e4530ae8ac26772bb6ab89a0ef297cd91\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/ccc24c0b1ebaa5d17bd60b7c020e0a8c11a8e44b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/ccc24c0b1ebaa5d17bd60b7c020e0a8c11a8e44b"}]},"branch":"refs/heads/master"},"2a8d47077904ee20464c1f6e349f1751dcd57374":{"kind":"REWORK","_number":10,"created":"2026-08-25 16:14:27.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/19/997919/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/19/997919/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/19/997919/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/19/997919/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/19/997919/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/19/997919/10"}}},"commit":{"parents":[{"commit":"173183cc848672149e206b88b89a47daa6b7d43d","subject":"zookeeper: Fix wrong member used by get_member_capabilities","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/173183cc848672149e206b88b89a47daa6b7d43d"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-07-20 02:13:14.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-08-25 16:14:16.000000000","tz":540},"subject":"Use etcd from distribution","message":"Use etcd from distribution\n\netcd has been available in Ubuntu and Debian for some time.\n\nUse the distribution package to install it, following the other\nbackends like redis.\n\nChange-Id: I58181a3e4530ae8ac26772bb6ab89a0ef297cd91\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/2a8d47077904ee20464c1f6e349f1751dcd57374"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/2a8d47077904ee20464c1f6e349f1751dcd57374"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
