)]}'
{"id":"openstack%2Ftooz~999317","triplet_id":"openstack%2Ftooz~master~I4f5cbb6b35d7352bd4372d8ceda9c82ea05ecfc2","project":"openstack/tooz","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-07-31 02:29:07.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"I4f5cbb6b35d7352bd4372d8ceda9c82ea05ecfc2","subject":"Refactor retry wrapper","status":"MERGED","created":"2026-07-30 17:10:23.000000000","updated":"2026-07-31 04:08:20.000000000","submitted":"2026-07-31 04:07:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"999317","meta_rev_id":"a33d956be2cf8a8972c82bffebe9f14e3258b072","_number":999317,"virtual_id_number":999317,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-07-31 04:07:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2026-07-31 02:29:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2026-07-31 02:29:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-07-30 19:06:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3729ac770fd2de76442ceb97097d22bc923d1251","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-07-30 17:10:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f20026ae0c5ba781dc1fc54f28e536bea8b9b9b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-30 19:06:26.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/808d659d768d4b288bf15033aea5896d\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4554fe20baf5406a997ec94aa5d4124d : SUCCESS in 1h 50m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1b69f787f0cb471cbdc965d64f628e71 : SUCCESS in 5m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1269882e3fba4ded9a210d51d162f814 : FAILURE in 5m 01s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/199164f8c1884b389f44617e4b30f398 : SUCCESS in 5m 00s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2f1a7b6bfce64e81b3edc1be82dc70cc : SUCCESS in 5m 37s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/9ee34701b8304eaaa91a00f5717d9297 : SUCCESS in 7m 26s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/ac629de56a4741fba36f895a79be7395 : SUCCESS in 8m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/593b9e6c035544b0b633a6b6bf527898 : SUCCESS in 5m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/48196e3917664bf29b3483f516f3104f : SUCCESS in 6m 09s\n- tooz-tox-examples https://zuul.opendev.org/t/openstack/build/1054ddf1831749b59a8f1759397bc24c : SUCCESS in 7m 19s\n- tooz-tox-py311-etcd3gw https://zuul.opendev.org/t/openstack/build/1764f0aaca75423ba72bcc46870693a6 : SUCCESS in 5m 45s\n- tooz-tox-py314-etcd3gw https://zuul.opendev.org/t/openstack/build/5f4068ae23674c029472c101b559f02c : SUCCESS in 8m 33s\n- tooz-tox-py311-zookeeper https://zuul.opendev.org/t/openstack/build/403256a69c1c45caa142d2f4d897c0e7 : SUCCESS in 3m 11s\n- tooz-tox-py314-zookeeper https://zuul.opendev.org/t/openstack/build/ff7ab176afe44f91ab14c97a0b93f6d3 : SUCCESS in 6m 17s\n- tooz-tox-py311-redis https://zuul.opendev.org/t/openstack/build/fadd9fd48b974161b1f81f1daa149079 : SUCCESS in 4m 40s\n- tooz-tox-py314-redis https://zuul.opendev.org/t/openstack/build/c989ff00936847b6adb3ddfb8d01232a : SUCCESS in 8m 26s\n- tooz-tox-py311-sentinel https://zuul.opendev.org/t/openstack/build/2b82299f600d402695c2aaba40e7c1b6 : SUCCESS in 4m 26s\n- tooz-tox-py314-sentinel https://zuul.opendev.org/t/openstack/build/9c58ef54bf224c699b6b10bd034a8fff : SUCCESS in 11m 17s\n- tooz-tox-py311-memcached https://zuul.opendev.org/t/openstack/build/eb9abfcf7da64e939c30144256e466e4 : SUCCESS in 4m 44s\n- tooz-tox-py314-memcached https://zuul.opendev.org/t/openstack/build/07eb2f697b134b23ba487e0e0abd5880 : SUCCESS in 4m 53s\n- tooz-tox-py311-postgresql https://zuul.opendev.org/t/openstack/build/bdd1171740eb4c36a0450e06f5f71031 : SUCCESS in 3m 09s\n- tooz-tox-py314-postgresql https://zuul.opendev.org/t/openstack/build/14484117d5e84b2fb9b42c41a67a358d : SUCCESS in 4m 13s\n- tooz-tox-py311-mysql https://zuul.opendev.org/t/openstack/build/10c4d410eda74b0dba8db62df7e371e9 : SUCCESS in 6m 38s\n- tooz-tox-py314-mysql https://zuul.opendev.org/t/openstack/build/cd7c454db5694bdf8ec8ce9f6833dee4 : SUCCESS in 6m 01s","accounts_in_message":[],"_revision_number":1},{"id":"82471fac4f7410eeb30212a83f78e2f3b504a791","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-07-30 23:42:31.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"75c9f12f37427909b96946625d769dd13769d708","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-31 01:21:02.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f6035212385a499ca02d6137719df376\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5a67b700b06d4b6abf9be22129579b41 : SUCCESS in 1h 33m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/28b5219037b347b4967594990452a837 : SUCCESS in 9m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/92a75d520bbb4713b42716adad8b782d : SUCCESS in 5m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b73b1768c6b44a8b9730bfbd3bc8595a : SUCCESS in 8m 14s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d916867a29444aea9f0dff09abb5b4ad : SUCCESS in 5m 12s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/f8a12f3bbaf844658872d1a4a70913eb : SUCCESS in 10m 49s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/0f7ac7ceabc74396ba27f0be69e379e5 : SUCCESS in 8m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f84d9f27a6074b31973f8e1ecdbe639e : SUCCESS in 4m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d2b8191936dd48e89adccc574cabc105 : SUCCESS in 3m 10s\n- tooz-tox-examples https://zuul.opendev.org/t/openstack/build/47ab347d129749ab951b5bd0ab05fff2 : SUCCESS in 4m 36s\n- tooz-tox-py311-etcd3gw https://zuul.opendev.org/t/openstack/build/8ca303b70cbf4d6fb933c9c68138f305 : SUCCESS in 4m 37s\n- tooz-tox-py314-etcd3gw https://zuul.opendev.org/t/openstack/build/b37f707ccdfd48eba574c25cbf70ba46 : SUCCESS in 7m 50s\n- tooz-tox-py311-zookeeper https://zuul.opendev.org/t/openstack/build/3b0c23a508484becb1d137924ae99466 : SUCCESS in 3m 14s\n- tooz-tox-py314-zookeeper https://zuul.opendev.org/t/openstack/build/250c9371525f4a169c791b67d758389d : SUCCESS in 7m 26s\n- tooz-tox-py311-redis https://zuul.opendev.org/t/openstack/build/e47b4b6c916c4b938525915d990dc644 : SUCCESS in 6m 04s\n- tooz-tox-py314-redis https://zuul.opendev.org/t/openstack/build/a0cdc2d12797458f8f68557351e3acab : SUCCESS in 7m 57s\n- tooz-tox-py311-sentinel https://zuul.opendev.org/t/openstack/build/d3bc96392315432ea06475a6875bd03c : SUCCESS in 6m 11s\n- tooz-tox-py314-sentinel https://zuul.opendev.org/t/openstack/build/0267df5f994e4763947edbba271737fa : SUCCESS in 6m 18s\n- tooz-tox-py311-memcached https://zuul.opendev.org/t/openstack/build/76ec158c130b4760abafbb4ca000d468 : SUCCESS in 4m 58s\n- tooz-tox-py314-memcached https://zuul.opendev.org/t/openstack/build/2215e07fecd6480aba6e9b650fea15a0 : SUCCESS in 4m 46s\n- tooz-tox-py311-postgresql https://zuul.opendev.org/t/openstack/build/5e3492192ad24fb497c8e0ed04eac2cf : SUCCESS in 4m 23s\n- tooz-tox-py314-postgresql https://zuul.opendev.org/t/openstack/build/435adeaea32142b8b32b0bf9669f4530 : SUCCESS in 7m 50s\n- tooz-tox-py311-mysql https://zuul.opendev.org/t/openstack/build/6116987a31884bcbaaf7166f49431cd4 : SUCCESS in 6m 23s\n- tooz-tox-py314-mysql https://zuul.opendev.org/t/openstack/build/6c553dc1f20c41e7ba14bce45cd67db8 : SUCCESS in 9m 00s","accounts_in_message":[],"_revision_number":2},{"id":"32edd30612c5d8acaf65aa75e5bfdebfc16a74f3","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-07-31 02:29:07.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"6196ca3f60545ccca27e7ce92b4b9f431456768e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-31 02:29:34.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"12097204fa533923e020ccd2251fc752718211a4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-31 04:07:21.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8c9ba88a79b74390be34dbc5b99d51c9\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e489dda1b9e84326b069dd90b8061038 : SUCCESS in 1h 19m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6a7b1d4e8b104db18848d4731ebf7e7f : SUCCESS in 6m 38s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2ab37faed29b4960b72992b8f81fdbbd : SUCCESS in 6m 18s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/68a15036b71c4bc48217c9b421557021 : SUCCESS in 5m 44s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/afd11d9a3fc74615943cdb8d2e7877b5 : SUCCESS in 8m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/013a1e36f690479090600a0f42a9b4bf : SUCCESS in 5m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/16c891674ae44434a3c3986bcdd8db17 : SUCCESS in 6m 45s\n- tooz-tox-examples https://zuul.opendev.org/t/openstack/build/f72747c2c7bd4fca8f2544adfbed6521 : SUCCESS in 6m 37s\n- tooz-tox-py311-etcd3gw https://zuul.opendev.org/t/openstack/build/3503a9a6909d4c789fb82fef65ba0f4b : SUCCESS in 6m 01s\n- tooz-tox-py314-etcd3gw https://zuul.opendev.org/t/openstack/build/7f5db7ff08e34224aa163985a48e3069 : SUCCESS in 8m 54s\n- tooz-tox-py311-zookeeper https://zuul.opendev.org/t/openstack/build/b008907de9704b3ea27c90cb8d67d810 : SUCCESS in 5m 43s\n- tooz-tox-py314-zookeeper https://zuul.opendev.org/t/openstack/build/db3c93bb0ab741daa52b81691e04c48e : SUCCESS in 8m 00s\n- tooz-tox-py311-redis https://zuul.opendev.org/t/openstack/build/afd1443f7b91419aa85448205413a821 : SUCCESS in 6m 15s\n- tooz-tox-py314-redis https://zuul.opendev.org/t/openstack/build/cc8df911125d472aab447ac77dcb1d1f : SUCCESS in 8m 23s\n- tooz-tox-py311-sentinel https://zuul.opendev.org/t/openstack/build/97b8e6cf7a574832a00e2f96b1f8047d : SUCCESS in 7m 07s\n- tooz-tox-py314-sentinel https://zuul.opendev.org/t/openstack/build/f84de388301a4e42b0ec83bd76e069fa : SUCCESS in 7m 17s\n- tooz-tox-py311-memcached https://zuul.opendev.org/t/openstack/build/3ce26350fee6461fbce820b86d5019e3 : SUCCESS in 4m 12s\n- tooz-tox-py314-memcached https://zuul.opendev.org/t/openstack/build/66fa610cff774a0dbaacae8975ffd902 : SUCCESS in 7m 52s\n- tooz-tox-py311-postgresql https://zuul.opendev.org/t/openstack/build/bf6ba0c2511140bf89455c69049a8f5b : SUCCESS in 4m 59s\n- tooz-tox-py314-postgresql https://zuul.opendev.org/t/openstack/build/872ae9987d52415185764374173b9dba : SUCCESS in 8m 48s\n- tooz-tox-py311-mysql https://zuul.opendev.org/t/openstack/build/bd136457d1a84d8d97e1c71b77c99d70 : SUCCESS in 9m 18s\n- tooz-tox-py314-mysql https://zuul.opendev.org/t/openstack/build/ac7e7ad04f7946178dee9592b195b6ae : SUCCESS in 8m 21s","accounts_in_message":[],"_revision_number":2},{"id":"e6ecfce72d9604c02a9c47df7fb8e1cfb2c70892","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-31 04:07:22.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"a33d956be2cf8a8972c82bffebe9f14e3258b072","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-31 04:08:20.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/53fac9f88bd94edf9adacae9358a6b74\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a7fcd5c0a6f34f308bf0439407040f1b : SUCCESS in 46s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"21ab4f3c732046057494edabf6849265eaff36a4","revisions":{"f4d9d955ff4d4bb5e0d6dcb98ed876a052dbc083":{"kind":"REWORK","_number":1,"created":"2026-07-30 17:10:23.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/17/999317/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/17/999317/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/17/999317/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/17/999317/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/17/999317/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/17/999317/1"}}},"commit":{"parents":[{"commit":"100d17f1a87195f28662c4c602b2d1158742d091","subject":"retry: Don\u0027t treat bool as a numeric stop_max_delay","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/100d17f1a87195f28662c4c602b2d1158742d091"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-07-30 17:09:50.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-07-30 17:09:50.000000000","tz":540},"subject":"Refactor retry wrapper","message":"Refactor retry wrapper\n\n... to make the logic to ignore stop argument more visible.\n\nChange-Id: I4f5cbb6b35d7352bd4372d8ceda9c82ea05ecfc2\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/f4d9d955ff4d4bb5e0d6dcb98ed876a052dbc083"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/f4d9d955ff4d4bb5e0d6dcb98ed876a052dbc083"}]},"branch":"refs/heads/master"},"21ab4f3c732046057494edabf6849265eaff36a4":{"kind":"REWORK","_number":2,"created":"2026-07-30 23:42:31.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/17/999317/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/17/999317/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/17/999317/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/17/999317/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/17/999317/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/17/999317/2"}}},"commit":{"parents":[{"commit":"100d17f1a87195f28662c4c602b2d1158742d091","subject":"retry: Don\u0027t treat bool as a numeric stop_max_delay","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/100d17f1a87195f28662c4c602b2d1158742d091"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-07-30 17:09:50.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-07-30 23:42:17.000000000","tz":540},"subject":"Refactor retry wrapper","message":"Refactor retry wrapper\n\n... to make the logic to ignore stop argument more visible.\n\nChange-Id: I4f5cbb6b35d7352bd4372d8ceda9c82ea05ecfc2\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/21ab4f3c732046057494edabf6849265eaff36a4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/21ab4f3c732046057494edabf6849265eaff36a4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
