)]}'
{"id":"openstack%2Ftosca-parser~1004302","triplet_id":"openstack%2Ftosca-parser~master~I60d23c9577334b55318bd30be902b647984bb994","project":"openstack/tosca-parser","branch":"master","hashtags":[],"change_id":"I60d23c9577334b55318bd30be902b647984bb994","subject":"Add minimum pyproject.toml","status":"NEW","created":"2026-09-06 03:55:45.000000000","updated":"2026-09-06 04:05:26.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"a8ef167be5c353bcf1d22a2f655b95a0283e9f07","_number":1004302,"virtual_id_number":1004302,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-09-06 04:05:26.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-06 04:05:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"833ae885b932cc3b47d649ee81fe1f3c4271151c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-06 03:55:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a8ef167be5c353bcf1d22a2f655b95a0283e9f07","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-06 04:05:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/39fa37eb643e4c6aa851d234dcc5fda5\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/dd04a86783924fec95a0100da0b53eaa : SUCCESS in 2m 35s\n- requirements-check https://zuul.opendev.org/t/openstack/build/40f37f11a282414f8402675752ed4fd7 : SUCCESS in 1m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2df0c593d7e64025ada3b9c75fd4beb3 : SUCCESS in 2m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/08267c3e082a4f04a4e5f80093da85a8 : SUCCESS in 3m 01s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1228ba1236c14b0d8738a5494f05ffa7 : SUCCESS in 1m 54s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d61348a3bf3b4e958997ffbeb36a603e : SUCCESS in 2m 59s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/8b627b5dd07246ca9e17dd0aaf9dd57d : SUCCESS in 4m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/156794ebd9d54f3eae24f1d82fb6484e : SUCCESS in 3m 08s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3ac6fea108ae905c588bcba058e9dfc4dd8ab4ac","revisions":{"3ac6fea108ae905c588bcba058e9dfc4dd8ab4ac":{"kind":"REWORK","_number":1,"created":"2026-09-06 03:55:45.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/02/1004302/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tosca-parser","ref":"refs/changes/02/1004302/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tosca-parser refs/changes/02/1004302/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tosca-parser refs/changes/02/1004302/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tosca-parser refs/changes/02/1004302/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tosca-parser refs/changes/02/1004302/1"}}},"commit":{"parents":[{"commit":"11b686a724db8194a28d10d4709c314bce17a53f","subject":"Merge \"Validate occurrences in relationships\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tosca-parser/commit/11b686a724db8194a28d10d4709c314bce17a53f"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-06 03:50:58.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-06 03:53:35.000000000","tz":540},"subject":"Add minimum pyproject.toml","message":"Add minimum pyproject.toml\n\nIt\u0027s now required by the requirements-check job.\n\nChange-Id: I60d23c9577334b55318bd30be902b647984bb994\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tosca-parser/commit/3ac6fea108ae905c588bcba058e9dfc4dd8ab4ac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tosca-parser/commit/3ac6fea108ae905c588bcba058e9dfc4dd8ab4ac"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
