)]}'
{"id":"openstack%2Ftosca-parser~1004305","triplet_id":"openstack%2Ftosca-parser~master~I1ca66c21c0bde89f8be00ccb77d2948b49e67e12","project":"openstack/tosca-parser","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-09-06 14:32:57.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_9816\u003e by clicking the attention icon","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"I1ca66c21c0bde89f8be00ccb77d2948b49e67e12","subject":"Refactor tox.ini","status":"NEW","created":"2026-09-06 03:55:45.000000000","updated":"2026-09-06 14:32:57.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"2e03bf6fb41fbb6a3b8a6b5c600bc3974aceb037","_number":1004305,"virtual_id_number":1004305,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-09-06 10:44:05.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-06 04:07:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e6ab0f602393f7494ac61fdc27051a494638e6a3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-06 03:55:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cadf29580f218f94907d35d89cda8ee03b9cc02c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-06 04:07:07.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/edf48297c2a043ffb9961e9fef0058d7\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/257b754a055d45e49afeaf4ac2473c53 : FAILURE in 2m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ec35f544eda648dfbbd567438654ad2f : FAILURE in 2m 13s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/50e037a6f7c9480196b5c882d46bd8e8 : FAILURE in 2m 38s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/17965a0cd49e427784cfaf2ec86a1ee2 : FAILURE in 1m 52s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/11e41ba742e14b2d92f33ec6bf549ef1 : FAILURE in 3m 00s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/8d8a81aa38fd4741b60b98ee6890029e : FAILURE in 4m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4bae408578364bc19a60a88d6c22c4d3 : FAILURE in 1m 54s","accounts_in_message":[],"_revision_number":1},{"id":"8b7a70b06e0f421454eae6dd59336a0c1d2c88ff","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-06 10:28:52.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"7f43180cf6d76871d2ed764652ca2076d617eed9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-06 10:44:05.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c923f54bad6b4a429db14201bc99797d\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9471a5460dc341d2a27f43ffeb571f21 : SUCCESS in 3m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aae39f6df32449ae81be9f2c6a8ce8a9 : SUCCESS in 2m 10s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ae77c56ac31041fe8c0e780250737282 : SUCCESS in 3m 35s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2e2ddda9e6e2422cbb55e72970ba4e77 : SUCCESS in 3m 12s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/bacc77674f08417ab54fa14156fcc8a6 : SUCCESS in 8m 12s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/343c34dd0fb2426f918082ac238185ce : SUCCESS in 3m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e35e51e9c87a4767a1290b386dce953e : SUCCESS in 3m 09s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"f9a424f6231447cf0c0ed019d52dbe624dfe7042","revisions":{"4e9f4fac4e252298d9952375fb050fa0fdcb7771":{"kind":"REWORK","_number":1,"created":"2026-09-06 03:55:45.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/05/1004305/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tosca-parser","ref":"refs/changes/05/1004305/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tosca-parser refs/changes/05/1004305/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tosca-parser refs/changes/05/1004305/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tosca-parser refs/changes/05/1004305/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tosca-parser refs/changes/05/1004305/1"}}},"commit":{"parents":[{"commit":"a268f976693894ae3a637adf28200f8921d0b950","subject":"tox: Drop redundant passenv","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tosca-parser/commit/a268f976693894ae3a637adf28200f8921d0b950"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-06 03:55:06.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-06 03:55:06.000000000","tz":540},"subject":"Refactor tox.ini","message":"Refactor tox.ini\n\nRemove a few redundant options.\n\nUpdate min version to use the constraints option, which is now\npreferred.\n\nChange-Id: I1ca66c21c0bde89f8be00ccb77d2948b49e67e12\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tosca-parser/commit/4e9f4fac4e252298d9952375fb050fa0fdcb7771"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tosca-parser/commit/4e9f4fac4e252298d9952375fb050fa0fdcb7771"}]},"branch":"refs/heads/master"},"f9a424f6231447cf0c0ed019d52dbe624dfe7042":{"kind":"REWORK","_number":2,"created":"2026-09-06 10:28:52.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/05/1004305/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tosca-parser","ref":"refs/changes/05/1004305/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tosca-parser refs/changes/05/1004305/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tosca-parser refs/changes/05/1004305/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tosca-parser refs/changes/05/1004305/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tosca-parser refs/changes/05/1004305/2"}}},"commit":{"parents":[{"commit":"a268f976693894ae3a637adf28200f8921d0b950","subject":"tox: Drop redundant passenv","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tosca-parser/commit/a268f976693894ae3a637adf28200f8921d0b950"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-06 03:55:06.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-06 10:28:52.000000000","tz":0},"subject":"Refactor tox.ini","message":"Refactor tox.ini\n\nRemove a few redundant options.\n\nUpdate min version to use the constraints option, which is now\npreferred.\n\nChange-Id: I1ca66c21c0bde89f8be00ccb77d2948b49e67e12\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tosca-parser/commit/f9a424f6231447cf0c0ed019d52dbe624dfe7042"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tosca-parser/commit/f9a424f6231447cf0c0ed019d52dbe624dfe7042"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
