)]}'
{"id":"openstack%2Ftraining-guides~654552","triplet_id":"openstack%2Ftraining-guides~master~I2d43c43a2825f592502116683162edaa387e6d26","project":"openstack/training-guides","branch":"master","topic":"opendev-gerrit","hashtags":[],"change_id":"I2d43c43a2825f592502116683162edaa387e6d26","subject":"Update openstackdocstheme options","status":"MERGED","created":"2019-04-22 15:58:07.000000000","updated":"2019-04-24 00:02:01.000000000","submitted":"2019-04-24 00:02:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"654552-1556064121310-354c625b","meta_rev_id":"b16ffd20467f04fc00399dbac282b92353668d3e","_number":654552,"virtual_id_number":654552,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"value":0,"_account_id":17765,"name":"Frank Kloeker","email":"f.kloeker@telekom.de","username":"eumel8"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2019-04-24 00:02:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17765,"name":"Frank Kloeker","email":"f.kloeker@telekom.de","username":"eumel8"},"all":[{"value":0,"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"value":2,"date":"2019-04-23 23:52:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17765,"name":"Frank Kloeker","email":"f.kloeker@telekom.de","username":"eumel8"},{"value":2,"date":"2019-04-23 14:37:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17765,"name":"Frank Kloeker","email":"f.kloeker@telekom.de","username":"eumel8"},"all":[{"value":0,"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"value":1,"date":"2019-04-23 23:52:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17765,"name":"Frank Kloeker","email":"f.kloeker@telekom.de","username":"eumel8"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"_account_id":17765,"name":"Frank Kloeker","email":"f.kloeker@telekom.de","username":"eumel8"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-22 18:37:44.000000000","updated_by":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"reviewer":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"state":"REVIEWER"},{"updated":"2019-04-23 14:37:40.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2019-04-23 23:52:58.000000000","updated_by":{"_account_id":17765,"name":"Frank Kloeker","email":"f.kloeker@telekom.de","username":"eumel8"},"reviewer":{"_account_id":17765,"name":"Frank Kloeker","email":"f.kloeker@telekom.de","username":"eumel8"},"state":"REVIEWER"},{"updated":"2019-04-24 00:02:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8bbb79512bffbe21afc4aa2943b4d56cfd0383fe","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-04-22 15:58:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"950650ad3f132ffda19b8df4e524249e13f73700","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 16:26:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-tox-manuals-checkbuild http://logs.openstack.org/52/654552/1/check/build-tox-manuals-checkbuild/8807f17/publish-docs/ : SUCCESS in 4m 08s","accounts_in_message":[],"_revision_number":1},{"id":"1330011b2e9e9b025b637d12bbd5d8f2e33a5731","author":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"date":"2019-04-22 18:37:44.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"3aa929ef9f06381b270e5b45f3780a7c08d07833","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-04-23 14:29:02.000000000","message":"Patch Set 1:\n\n(8 comments)\n\nOne open question. The rest are nits that we might want to fix up later","accounts_in_message":[],"_revision_number":1},{"id":"d0a6ba0ae18d59bb950b8208f270a93307fbd433","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-04-23 14:36:33.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"dd9f034cc5bc3d86ffa2e4249ca81ea30eddf9bf","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-04-23 14:37:40.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"7a042f5451c19c526667639a5479ff8d531cdaf1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 14:47:38.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-tox-manuals-checkbuild http://logs.openstack.org/52/654552/2/check/build-tox-manuals-checkbuild/b214bf7/publish-docs/ : SUCCESS in 5m 46s","accounts_in_message":[],"_revision_number":2},{"id":"a39dfffa4303aa8d506ccf6cb7da460d6c932e91","author":{"_account_id":17765,"name":"Frank Kloeker","email":"f.kloeker@telekom.de","username":"eumel8"},"date":"2019-04-23 23:52:58.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"b8b14df6f220555de7a0bc29a915269851608a0a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 23:53:11.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"187fd5aecbee37fca16d6f3b7d0306ebaf409e85","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-24 00:02:01.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-tox-manuals-checkbuild http://logs.openstack.org/52/654552/2/gate/build-tox-manuals-checkbuild/1b982e8/publish-docs/ : SUCCESS in 4m 05s","accounts_in_message":[],"_revision_number":2},{"id":"ca0d6044adde85d2dc1429fe7ef84c70c166c85c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-24 00:02:01.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"85e5de8101761a1faef4659c185c6086a2eea7ac","revisions":{"b1b3914fdb98350b7637afc2ff0f8a94b242aaf0":{"kind":"REWORK","_number":1,"created":"2019-04-22 15:58:07.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/52/654552/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/training-guides","ref":"refs/changes/52/654552/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/training-guides refs/changes/52/654552/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/training-guides refs/changes/52/654552/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/training-guides refs/changes/52/654552/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/training-guides refs/changes/52/654552/1"}}},"commit":{"parents":[{"commit":"ee4df7f88aa4331a99d1cb046ae6468751f5c77b","subject":"Update for OpenDev changes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/training-guides/commit/ee4df7f88aa4331a99d1cb046ae6468751f5c77b"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-04-22 15:57:16.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-04-22 15:57:16.000000000","tz":120},"subject":"Update openstackdocstheme options","message":"Update openstackdocstheme options\n\nRequire at least version 1.20 and cleanup the conf.py files and use\nnewer option settings.\n\nChange-Id: I2d43c43a2825f592502116683162edaa387e6d26\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/training-guides/commit/b1b3914fdb98350b7637afc2ff0f8a94b242aaf0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/training-guides/commit/b1b3914fdb98350b7637afc2ff0f8a94b242aaf0"}]},"branch":"refs/heads/master"},"85e5de8101761a1faef4659c185c6086a2eea7ac":{"kind":"REWORK","_number":2,"created":"2019-04-23 14:36:33.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/52/654552/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/training-guides","ref":"refs/changes/52/654552/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/training-guides refs/changes/52/654552/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/training-guides refs/changes/52/654552/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/training-guides refs/changes/52/654552/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/training-guides refs/changes/52/654552/2"}}},"commit":{"parents":[{"commit":"ee4df7f88aa4331a99d1cb046ae6468751f5c77b","subject":"Update for OpenDev changes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/training-guides/commit/ee4df7f88aa4331a99d1cb046ae6468751f5c77b"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-04-22 15:57:16.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-04-23 14:36:01.000000000","tz":120},"subject":"Update openstackdocstheme options","message":"Update openstackdocstheme options\n\nRequire at least version 1.20 and cleanup the conf.py files and use\nnewer option settings.\n\nAs part of cleanup remove many unused options and comments.\n\nChange-Id: I2d43c43a2825f592502116683162edaa387e6d26\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/training-guides/commit/85e5de8101761a1faef4659c185c6086a2eea7ac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/training-guides/commit/85e5de8101761a1faef4659c185c6086a2eea7ac"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
