)]}'
{"id":"openstack%2Ftraining-guides~688604","triplet_id":"openstack%2Ftraining-guides~master~I8eba155304c05e61466b8885bde2bc2777ca6dff","project":"openstack/training-guides","branch":"master","topic":"update_exercises","hashtags":[],"change_id":"I8eba155304c05e61466b8885bde2bc2777ca6dff","subject":"Update Task Tracking Exercises","status":"MERGED","created":"2019-10-15 02:43:34.000000000","updated":"2019-10-15 18:51:09.000000000","submitted":"2019-10-15 18:47:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"688604-1571165276842-33a7a8e1","meta_rev_id":"415802d998b06eadc04e201ba3526520a4547d65","_number":688604,"virtual_id_number":688604,"owner":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-15 18:47:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":2,"date":"2019-10-15 18:36:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":2,"date":"2019-10-15 16:46:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":1,"date":"2019-10-15 18:36:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-15 16:46:32.000000000","updated_by":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"reviewer":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"state":"REVIEWER"},{"updated":"2019-10-15 18:36:36.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2019-10-15 18:47:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a377199327e6231178463a9d460d122004afa671","author":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"date":"2019-10-15 02:43:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c778fd67008e0b7af0cf978b6e3ab3b31f3b7489","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-15 02:49:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-tox-manuals-publishdocs https://zuul.opendev.org/t/openstack/build/a30c04943a2a4261be93b48a4b2b1a01 : SUCCESS in 4m 39s","accounts_in_message":[],"_revision_number":1},{"id":"aceb46b5a5e5bf8717547257ac7262f4bf5660fa","author":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"date":"2019-10-15 09:35:03.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)\n\nThanks Kendall!\n\nLooks good overall, I only added one nit picking question to consider.\n\nThanks,\nIldikó","accounts_in_message":[],"_revision_number":1},{"id":"6698b56caff080e58d5c830ee66f16d7243dd58e","author":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"date":"2019-10-15 15:38:09.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e79fb1e1579697103954442603dbbe1f6dd20cc0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-15 15:48:51.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-tox-manuals-publishdocs https://zuul.opendev.org/t/openstack/build/75baeefb0ee348b6adf81c277ea1d83d : SUCCESS in 4m 44s","accounts_in_message":[],"_revision_number":2},{"id":"5d8f08d767e6c68f9ee481590d0e85dd53e4b752","author":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"date":"2019-10-15 16:46:32.000000000","message":"Patch Set 2: Code-Review+2\n\nLooks good to me, thanks for the quick update.\n\nCheers,\nIldikó","accounts_in_message":[],"_revision_number":2},{"id":"dde69bd347c257f7ac72cd7c9aeff0282ef003af","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-10-15 18:36:36.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"9d5ae1d0788a08badfae057f5c8e667c0ef57de4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-15 18:36:47.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"e3ff4a9dc285aadf8164cde5696b7567b507337d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-15 18:47:56.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-tox-manuals-publishdocs https://zuul.opendev.org/t/openstack/build/db39941edadd48ba8616fd268c294bd9 : SUCCESS in 5m 05s","accounts_in_message":[],"_revision_number":2},{"id":"2e6e97ee50a6ec98f61bd96d79fd8db402d0837c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-15 18:47:56.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"f1228351d9572bfaef3b6e039bab6102579de76e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-15 18:51:09.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-manuals https://zuul.opendev.org/t/openstack/build/a7bfcb7b30c14aea864fca307904308e : SUCCESS in 2m 47s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"49d4861081e720328dab44a1f80da95c19b0858d","revisions":{"10713ab9240594cc8b9f8e42ffbd7d900f9a5849":{"kind":"REWORK","_number":1,"created":"2019-10-15 02:43:34.000000000","uploader":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"ref":"refs/changes/04/688604/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/training-guides","ref":"refs/changes/04/688604/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/training-guides refs/changes/04/688604/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/training-guides refs/changes/04/688604/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/training-guides refs/changes/04/688604/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/training-guides refs/changes/04/688604/1"}}},"commit":{"parents":[{"commit":"27e5c2c1e9dc79bf1f20fef422dcc9159d274050","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/training-guides/commit/27e5c2c1e9dc79bf1f20fef422dcc9159d274050"}]}],"author":{"name":"Kendall Nelson","email":"kennelson11@gmail.com","date":"2019-10-15 02:40:58.000000000","tz":-420},"committer":{"name":"Kendall Nelson","email":"kennelson11@gmail.com","date":"2019-10-15 02:40:58.000000000","tz":-420},"subject":"Update Task Tracking Exercises","message":"Update Task Tracking Exercises\n\nWe still occasionally hit the race condition where when a story is\ncreated with multiple tasks there is a deadlock. To avoid hitting this\nin the upstream training, let us have the students make stories with\none task.\n\nThis patch also adds clarifying phrasing to the other exercies in the\ntask tracking section.\n\nChange-Id: I8eba155304c05e61466b8885bde2bc2777ca6dff\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/training-guides/commit/10713ab9240594cc8b9f8e42ffbd7d900f9a5849"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/training-guides/commit/10713ab9240594cc8b9f8e42ffbd7d900f9a5849"}]},"branch":"refs/heads/master"},"49d4861081e720328dab44a1f80da95c19b0858d":{"kind":"REWORK","_number":2,"created":"2019-10-15 15:38:09.000000000","uploader":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"ref":"refs/changes/04/688604/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/training-guides","ref":"refs/changes/04/688604/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/training-guides refs/changes/04/688604/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/training-guides refs/changes/04/688604/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/training-guides refs/changes/04/688604/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/training-guides refs/changes/04/688604/2"}}},"commit":{"parents":[{"commit":"27e5c2c1e9dc79bf1f20fef422dcc9159d274050","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/training-guides/commit/27e5c2c1e9dc79bf1f20fef422dcc9159d274050"}]}],"author":{"name":"Kendall Nelson","email":"kennelson11@gmail.com","date":"2019-10-15 02:40:58.000000000","tz":-420},"committer":{"name":"Kendall Nelson","email":"kennelson11@gmail.com","date":"2019-10-15 15:38:00.000000000","tz":-420},"subject":"Update Task Tracking Exercises","message":"Update Task Tracking Exercises\n\nWe still occasionally hit the race condition where when a story is\ncreated with multiple tasks there is a deadlock. To avoid hitting this\nin the upstream training, let us have the students make stories with\none task.\n\nThis patch also adds clarifying phrasing to the other exercies in the\ntask tracking section.\n\nChange-Id: I8eba155304c05e61466b8885bde2bc2777ca6dff\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/training-guides/commit/49d4861081e720328dab44a1f80da95c19b0858d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/training-guides/commit/49d4861081e720328dab44a1f80da95c19b0858d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
