)]}'
{"id":"openstack%2Ftraining-guides~720574","triplet_id":"openstack%2Ftraining-guides~master~Ifb4c5bdff7314388ace38f98c8e1ca3433ab9d64","project":"openstack/training-guides","branch":"master","topic":"cap-sphinx","hashtags":[],"change_id":"Ifb4c5bdff7314388ace38f98c8e1ca3433ab9d64","subject":"Cap Sphinx version to fix builds","status":"MERGED","created":"2020-04-16 17:48:53.000000000","updated":"2020-04-17 09:13:01.000000000","submitted":"2020-04-17 09:08:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"720574-1587114485227-eff0fd40","meta_rev_id":"06055361e55ccb9046d949062f84410ea6eb638f","_number":720574,"virtual_id_number":720574,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-17 09:08:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":2,"date":"2020-04-17 08:59:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":2,"date":"2020-04-16 18:52:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},{"value":1,"date":"2020-04-17 02:46:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":1,"date":"2020-04-17 08:59:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-16 18:52:22.000000000","updated_by":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"reviewer":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"state":"REVIEWER"},{"updated":"2020-04-17 02:46:26.000000000","updated_by":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"reviewer":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"state":"REVIEWER"},{"updated":"2020-04-17 09:08:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4c45269217564f44cc77cd740f88a6cc42340760","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-04-16 17:48:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3cf205bd6f1ce256cf7337840c456141d4325ac9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-16 17:58:26.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-tox-manuals-publishdocs https://zuul.opendev.org/t/openstack/build/eb9c119e1bad41e8a99b2dc0b8345760 : FAILURE in 3m 41s","accounts_in_message":[],"_revision_number":1},{"id":"831b888953f27f499272287457fd6b0bb2563872","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-04-16 17:59:10.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"946830e95976c656a8281f121852d2bf6e282293","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-16 18:10:00.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-tox-manuals-publishdocs https://zuul.opendev.org/t/openstack/build/c769149a79704d418490754f364df506 : SUCCESS in 9m 23s","accounts_in_message":[],"_revision_number":2},{"id":"b25b85d05fad43e8757a30688153ab3c764ab18a","author":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"date":"2020-04-16 18:52:22.000000000","message":"Patch Set 2: Code-Review+2\n\nSadness. Thanks Andreas for catching and fixing the issue!","accounts_in_message":[],"_revision_number":2},{"id":"242ffda3e655b6336d21a7792775347b4e1ce194","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2020-04-17 02:46:26.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"8b691c855ecd5f7a800122b239f3fc143a2c8bba","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-04-17 08:59:19.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nself-approving to fix build","accounts_in_message":[],"_revision_number":2},{"id":"14a1053ac06d0fdb9d3782fbc93a0aeeab22b22f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-17 08:59:30.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b3f9967cc4338a03d7bf2713f4c2c9339568f9b5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-17 09:08:04.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-tox-manuals-publishdocs https://zuul.opendev.org/t/openstack/build/9895ef45bc2e48329db391a528996fa9 : SUCCESS in 6m 21s","accounts_in_message":[],"_revision_number":2},{"id":"67d7d2d8f63a388c588cae71507b6f1ef43a8fc1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-17 09:08:05.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"129e99411a7ef0bc18d87e2a2b3281612c81b06e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-17 09:13:01.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-manuals https://zuul.opendev.org/t/openstack/build/e28580a50523417884704ad090f6abe3 : SUCCESS in 4m 31s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c570c3749d13e729d5a104de75985ed1bb16358e","revisions":{"65557790997354a3173da41b8aab236bb84b8d37":{"kind":"REWORK","_number":1,"created":"2020-04-16 17:48:53.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/74/720574/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/training-guides","ref":"refs/changes/74/720574/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/training-guides refs/changes/74/720574/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/training-guides refs/changes/74/720574/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/training-guides refs/changes/74/720574/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/training-guides refs/changes/74/720574/1"}}},"commit":{"parents":[{"commit":"f9075743d7a84020144c18a67f74d1cbf9350b45","subject":"Merge \"Small cleanups\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/training-guides/commit/f9075743d7a84020144c18a67f74d1cbf9350b45"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-04-16 17:48:01.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-04-16 17:48:01.000000000","tz":120},"subject":"Cap Sphinx version to fix builds","message":"Cap Sphinx version to fix builds\n\nThe guide fails building with hieroglyph 1.0 and Sphinx 3.0.\n\nUntil there\u0027s a new release of hieroglyph, cap Sphinx version.\n\nUpstream issue:\nhttps://github.com/nyergler/hieroglyph/issues/166\n\nChange-Id: Ifb4c5bdff7314388ace38f98c8e1ca3433ab9d64\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/training-guides/commit/65557790997354a3173da41b8aab236bb84b8d37"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/training-guides/commit/65557790997354a3173da41b8aab236bb84b8d37"}]},"branch":"refs/heads/master"},"c570c3749d13e729d5a104de75985ed1bb16358e":{"kind":"REWORK","_number":2,"created":"2020-04-16 17:59:10.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/74/720574/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/training-guides","ref":"refs/changes/74/720574/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/training-guides refs/changes/74/720574/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/training-guides refs/changes/74/720574/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/training-guides refs/changes/74/720574/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/training-guides refs/changes/74/720574/2"}}},"commit":{"parents":[{"commit":"f9075743d7a84020144c18a67f74d1cbf9350b45","subject":"Merge \"Small cleanups\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/training-guides/commit/f9075743d7a84020144c18a67f74d1cbf9350b45"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-04-16 17:48:01.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-04-16 17:59:03.000000000","tz":120},"subject":"Cap Sphinx version to fix builds","message":"Cap Sphinx version to fix builds\n\nThe guide fails building with hieroglyph 1.0 and Sphinx 3.0.\n\nUntil there\u0027s a new release of hieroglyph, cap Sphinx version.\n\nUpstream issue:\nhttps://github.com/nyergler/hieroglyph/issues/166\n\nChange-Id: Ifb4c5bdff7314388ace38f98c8e1ca3433ab9d64\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/training-guides/commit/c570c3749d13e729d5a104de75985ed1bb16358e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/training-guides/commit/c570c3749d13e729d5a104de75985ed1bb16358e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
