)]}'
{"id":"openstack%2Ftraining-guides~758474","triplet_id":"openstack%2Ftraining-guides~master~I78b2acdebea41482df77ab8b42ad5d9a0a67fae1","project":"openstack/training-guides","branch":"master","topic":"update-task-tracking-exercises","hashtags":[],"change_id":"I78b2acdebea41482df77ab8b42ad5d9a0a67fae1","subject":"Modify Task Tracking Exercises","status":"MERGED","created":"2020-10-15 18:04:36.000000000","updated":"2020-10-15 18:56:28.000000000","submitted":"2020-10-15 18:54:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"758474-1602788040414-880e6410","meta_rev_id":"1ad3bb70d26119f799beb4f466a771b5bc2da006","_number":758474,"virtual_id_number":758474,"owner":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-15 18:54:00.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":2,"date":"2020-10-15 18:37:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":1,"date":"2020-10-15 18:37:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-15 18:21:03.000000000","updated_by":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"reviewer":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"state":"REVIEWER"},{"updated":"2020-10-15 18:37:05.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2020-10-15 18:54:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"46627435b5585b6237cfb17c43f029142403badc","author":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"date":"2020-10-15 18:04:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ed719d60605df239055865d19fd4a05fe07c70ba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-15 18:14:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-tox-manuals-publishdocs https://zuul.opendev.org/t/openstack/build/07f14a887d7a43e586880fa99d854c73 : SUCCESS in 5m 27s","accounts_in_message":[],"_revision_number":1},{"id":"1bc4e217e0874c50767fd42f5a95e289f95893ab","author":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"date":"2020-10-15 18:21:03.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)\n\nThanks!\n\nAdded a nitpick in line in case you need to upload another version. Or we can have a student fixing it. :)","accounts_in_message":[],"_revision_number":1},{"id":"7113305914d20f410ddd6d8c02b6ccf69ba3b564","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-10-15 18:35:11.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"d2c3ce9bce15dfb9e661d902cbc2b0719574c2e2","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-10-15 18:35:18.000000000","message":"Patch Set 2:\n\nfixed space","accounts_in_message":[],"_revision_number":2},{"id":"de8a6b4c42fb3bc6de0fe6652cd3a59e0517beda","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-10-15 18:36:19.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0b40738355ad9689f4e477e953c1135291f270a3","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-10-15 18:36:35.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"efd51720493a301bb30a67578df1fd2a5f0f15db","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-10-15 18:37:05.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\nFixed spaces, approving based on Ildiko\u0027s prior +2.","accounts_in_message":[],"_revision_number":3},{"id":"c61f22eb3097c72714f1404174b8f943f2931bb2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-15 18:43:48.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-tox-manuals-publishdocs https://zuul.opendev.org/t/openstack/build/0bc17ad203e34a9bacb9c7a91230bd2c : SUCCESS in 5m 51s","accounts_in_message":[],"_revision_number":3},{"id":"4b610fb96446090db9650366b7ec85c47e7f1a46","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-15 18:43:59.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"4d9577af773f834dcb8b4ba6e9ea58423630a2c2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-15 18:54:00.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"9da2f962f0cee97ff95887b1b66abd0becece6af","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-15 18:54:00.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-tox-manuals-publishdocs https://zuul.opendev.org/t/openstack/build/de5a0a2afe45407a9b0b0f496712a1de : SUCCESS in 4m 55s","accounts_in_message":[],"_revision_number":3},{"id":"37ab5a663e489dd12b8a57d2c7b0216ae01334a8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-15 18:56:28.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-manuals https://zuul.opendev.org/t/openstack/build/56308cc3de4b4539a03085c5f9454ccd : SUCCESS in 2m 06s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"f5a3d5f8ff1d10c264032e0cfc5ba922d2b80690","revisions":{"6798483799f2b42168004067c62ea9df408110bf":{"kind":"REWORK","_number":1,"created":"2020-10-15 18:04:36.000000000","uploader":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"ref":"refs/changes/74/758474/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/training-guides","ref":"refs/changes/74/758474/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/training-guides refs/changes/74/758474/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/training-guides refs/changes/74/758474/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/training-guides refs/changes/74/758474/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/training-guides refs/changes/74/758474/1"}}},"commit":{"parents":[{"commit":"7869be1455b423501b4a63855393710199e5ce4f","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/training-guides/commit/7869be1455b423501b4a63855393710199e5ce4f"}]}],"author":{"name":"Kendall Nelson","email":"kennelson11@gmail.com","date":"2020-10-15 18:03:27.000000000","tz":-420},"committer":{"name":"Kendall Nelson","email":"kennelson11@gmail.com","date":"2020-10-15 18:04:30.000000000","tz":-420},"subject":"Modify Task Tracking Exercises","message":"Modify Task Tracking Exercises\n\nReorder exercises, mark two of the StoryBoard ones\nas optional.\n\nChange-Id: I78b2acdebea41482df77ab8b42ad5d9a0a67fae1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/training-guides/commit/6798483799f2b42168004067c62ea9df408110bf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/training-guides/commit/6798483799f2b42168004067c62ea9df408110bf"}]},"branch":"refs/heads/master"},"b05c74283bd86119ecc4ba4be90f26f0863c0de4":{"kind":"REWORK","_number":2,"created":"2020-10-15 18:35:11.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/74/758474/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/training-guides","ref":"refs/changes/74/758474/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/training-guides refs/changes/74/758474/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/training-guides refs/changes/74/758474/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/training-guides refs/changes/74/758474/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/training-guides refs/changes/74/758474/2"}}},"commit":{"parents":[{"commit":"7869be1455b423501b4a63855393710199e5ce4f","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/training-guides/commit/7869be1455b423501b4a63855393710199e5ce4f"}]}],"author":{"name":"Kendall Nelson","email":"kennelson11@gmail.com","date":"2020-10-15 18:03:27.000000000","tz":-420},"committer":{"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","date":"2020-10-15 18:35:04.000000000","tz":0},"subject":"Modify Task Tracking Exercises","message":"Modify Task Tracking Exercises\n\nReorder exercises, mark two of the StoryBoard ones\nas optional.\n\nChange-Id: I78b2acdebea41482df77ab8b42ad5d9a0a67fae1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/training-guides/commit/b05c74283bd86119ecc4ba4be90f26f0863c0de4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/training-guides/commit/b05c74283bd86119ecc4ba4be90f26f0863c0de4"}]},"branch":"refs/heads/master"},"f5a3d5f8ff1d10c264032e0cfc5ba922d2b80690":{"kind":"REWORK","_number":3,"created":"2020-10-15 18:36:35.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/74/758474/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/training-guides","ref":"refs/changes/74/758474/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/training-guides refs/changes/74/758474/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/training-guides refs/changes/74/758474/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/training-guides refs/changes/74/758474/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/training-guides refs/changes/74/758474/3"}}},"commit":{"parents":[{"commit":"7869be1455b423501b4a63855393710199e5ce4f","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/training-guides/commit/7869be1455b423501b4a63855393710199e5ce4f"}]}],"author":{"name":"Kendall Nelson","email":"kennelson11@gmail.com","date":"2020-10-15 18:03:27.000000000","tz":-420},"committer":{"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","date":"2020-10-15 18:36:31.000000000","tz":0},"subject":"Modify Task Tracking Exercises","message":"Modify Task Tracking Exercises\n\nReorder exercises, mark two of the StoryBoard ones\nas optional.\n\nChange-Id: I78b2acdebea41482df77ab8b42ad5d9a0a67fae1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/training-guides/commit/f5a3d5f8ff1d10c264032e0cfc5ba922d2b80690"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/training-guides/commit/f5a3d5f8ff1d10c264032e0cfc5ba922d2b80690"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
