)]}'
{"id":"openstack%2Ftraining-guides~895599","triplet_id":"openstack%2Ftraining-guides~master~I9ed81d5edefe904fcf7e6f9aed2a00d756299d04","project":"openstack/training-guides","branch":"master","topic":"bug/2035226","attention_set":{},"removed_from_attention_set":{"841":{"account":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"last_update":"2023-09-22 16:26:47.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I9ed81d5edefe904fcf7e6f9aed2a00d756299d04","subject":"generatepot-rst.sh: Drop UUID filtering","status":"MERGED","created":"2023-09-18 08:11:06.000000000","updated":"2023-09-22 16:28:50.000000000","submitted":"2023-09-22 16:26:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"895599-bug/2035226","meta_rev_id":"dc4b4da11ff74a9779974d049fcb0a584ed22e45","_number":895599,"virtual_id_number":895599,"owner":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-22 16:26:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"all":[{"value":2,"date":"2023-09-22 16:15:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"all":[{"value":1,"date":"2023-09-22 16:15:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-18 08:16:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-22 16:15:11.000000000","updated_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"reviewer":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"state":"REVIEWER"}],"messages":[{"id":"ccae07b9b367d3d980712f610f92b525204ae9e5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2023-09-18 08:11:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f03ec38b092f20add0e9a5ac85f3c953d473b3be","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-18 08:16:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ec957bc2af1b4d69903c832390214888\n\n- build-tox-manuals-publishdocs https://zuul.opendev.org/t/openstack/build/2ef7db38c553424a914a1b209be0b9a0 : SUCCESS in 4m 20s","accounts_in_message":[],"_revision_number":1},{"id":"69d9823326475d41e4d36164ee1370be3d30e151","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2023-09-22 16:15:11.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1515150a29fd749158212d354495b82cdc10988f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-22 16:15:29.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"256118f2a88ba3f5b82d47e2ab0b569b2dce3401","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-22 16:26:47.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aa4bfbcedafa4d1f818d23246e144605\n\n- build-tox-manuals-publishdocs https://zuul.opendev.org/t/openstack/build/0f60b8869974477db18b71e943258abf : SUCCESS in 6m 18s","accounts_in_message":[],"_revision_number":1},{"id":"48062817432a121c9b154ca4a9e18ce9168f18c5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-22 16:26:47.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"dc4b4da11ff74a9779974d049fcb0a584ed22e45","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-22 16:28:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/036c35d9a1dd48818d64fd146fc135d8\n\n- promote-openstack-manuals https://zuul.opendev.org/t/openstack/build/ba404a1eafd8431faf376227f4b1983d : SUCCESS in 1m 44s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"31b6059f55675ee9d5f1d04b57227ae48e479d51","revisions":{"31b6059f55675ee9d5f1d04b57227ae48e479d51":{"kind":"REWORK","_number":1,"created":"2023-09-18 08:11:06.000000000","uploader":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"ref":"refs/changes/99/895599/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/training-guides","ref":"refs/changes/99/895599/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/training-guides refs/changes/99/895599/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/training-guides refs/changes/99/895599/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/training-guides refs/changes/99/895599/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/training-guides refs/changes/99/895599/1"}}},"commit":{"parents":[{"commit":"1521a8784dd78cb8680671b016e5e82f7c4d92ae","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/training-guides/commit/1521a8784dd78cb8680671b016e5e82f7c4d92ae"}]}],"author":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2023-09-18 08:09:56.000000000","tz":540},"committer":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2023-09-18 08:10:47.000000000","tz":540},"subject":"generatepot-rst.sh: Drop UUID filtering","message":"generatepot-rst.sh: Drop UUID filtering\n\ntools/generatepot-rst.sh has an awk filter to drop UUID from POT files,\nbut it is unnecessary now. It was introduced to strip UUID information\nwhich was added by default in Sphinx \u003c1.3 [1] in openstack-manuals.\nSphinx \u003e\u003d1.3 does not output UUID information by default [2],\n so is is no longer needed.\n\nIn addition, (g)awk 5.0 or later (adopted in Ubuntu 20.04 or later)\ncomplains the current awk regexp. I think it is the time to drop it\nrather than fixing the regexp.\n\n[1] https://opendev.org/openstack/openstack-manuals/commit/993647f316fbe33a513baffb5cc4ad156355da8a\n[2] https://www.sphinx-doc.org/en/master/usage/configuration.html#confval-gettext_uuid\n\nRelated-Bug: #2035226\nChange-Id: I9ed81d5edefe904fcf7e6f9aed2a00d756299d04\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/training-guides/commit/31b6059f55675ee9d5f1d04b57227ae48e479d51"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/training-guides/commit/31b6059f55675ee9d5f1d04b57227ae48e479d51"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
