)]}'
{"id":"openstack%2Ftraining-labs~701092","triplet_id":"openstack%2Ftraining-labs~master~I7defb743e5791ca897cdc6736b5031ed015d9a02","project":"openstack/training-labs","branch":"master","topic":"train","hashtags":[],"change_id":"I7defb743e5791ca897cdc6736b5031ed015d9a02","subject":"setup_glance: replace openstack client with glance client","status":"MERGED","created":"2020-01-04 14:53:41.000000000","updated":"2020-01-22 15:55:14.000000000","submitted":"2020-01-22 15:55:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"701092-1579708514244-46a14a57","meta_rev_id":"ad662b3d1b2b8b88c7fc3d6c2a7edd0616ea5b62","_number":701092,"virtual_id_number":701092,"owner":{"_account_id":11109,"name":"Roger Luethi","email":"rl@patchworkscience.org","username":"rluethi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":20770,"name":"Julen Larrucea","email":"julenl@gmail.com","username":"julen"},{"value":0,"_account_id":17765,"name":"Frank Kloeker","email":"f.kloeker@telekom.de","username":"eumel8"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-22 15:55:14.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":20770,"name":"Julen Larrucea","email":"julenl@gmail.com","username":"julen"},"all":[{"value":2,"date":"2020-01-16 16:55:50.000000000","_account_id":20770,"name":"Julen Larrucea","email":"julenl@gmail.com","username":"julen"},{"value":2,"date":"2020-01-22 15:47:18.000000000","_account_id":17765,"name":"Frank Kloeker","email":"f.kloeker@telekom.de","username":"eumel8"},{"value":2,"date":"2020-01-15 15:14:16.000000000","_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17765,"name":"Frank Kloeker","email":"f.kloeker@telekom.de","username":"eumel8"},"all":[{"value":0,"_account_id":20770,"name":"Julen Larrucea","email":"julenl@gmail.com","username":"julen"},{"value":1,"date":"2020-01-22 15:47:18.000000000","_account_id":17765,"name":"Frank Kloeker","email":"f.kloeker@telekom.de","username":"eumel8"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":17765,"name":"Frank Kloeker","email":"f.kloeker@telekom.de","username":"eumel8"},{"_account_id":20770,"name":"Julen Larrucea","email":"julenl@gmail.com","username":"julen"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-15 15:14:16.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2020-01-16 16:55:50.000000000","updated_by":{"_account_id":20770,"name":"Julen Larrucea","email":"julenl@gmail.com","username":"julen"},"reviewer":{"_account_id":20770,"name":"Julen Larrucea","email":"julenl@gmail.com","username":"julen"},"state":"REVIEWER"},{"updated":"2020-01-22 15:47:18.000000000","updated_by":{"_account_id":17765,"name":"Frank Kloeker","email":"f.kloeker@telekom.de","username":"eumel8"},"reviewer":{"_account_id":17765,"name":"Frank Kloeker","email":"f.kloeker@telekom.de","username":"eumel8"},"state":"REVIEWER"},{"updated":"2020-01-22 15:55:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e5e402d1337da0922015a87b2dd2872f2d4a15a0","author":{"_account_id":11109,"name":"Roger Luethi","email":"rl@patchworkscience.org","username":"rluethi"},"date":"2020-01-04 14:53:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2c3ab973f0c7026d190fdde2f791e41c59f612c9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-04 14:58:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- training-labs-scripts https://zuul.opendev.org/t/openstack/build/b8f9966d4d7041e8a28ced8c726e4c60 : SUCCESS in 3m 25s","accounts_in_message":[],"_revision_number":1},{"id":"9ea9ed7ce0a21fba3172a93fa94588838d38c776","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-01-15 15:14:16.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"398eea7b22ce73bdc3a903d433b3de9e731aa0ba","author":{"_account_id":20770,"name":"Julen Larrucea","email":"julenl@gmail.com","username":"julen"},"date":"2020-01-16 16:55:50.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"53e4ce8aff7f811bb42d7e4c216080e5f086693e","author":{"_account_id":17765,"name":"Frank Kloeker","email":"f.kloeker@telekom.de","username":"eumel8"},"date":"2020-01-22 15:47:18.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nlgtm, thx","accounts_in_message":[],"_revision_number":1},{"id":"2ccb3dc8f90dfacda777709f5243fc58f2bfd780","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-22 15:47:30.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"553e618858a69ba1472c1008156873680bae815a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-22 15:55:14.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"ce26b624aa13ab43314a66acd971f5eb021ebcf4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-22 15:55:14.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- training-labs-scripts https://zuul.opendev.org/t/openstack/build/7208260a2b3b443a8742756b9d077c5d : SUCCESS in 2m 36s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"78bdb0f17785a0a052d1bdf4f14f285ed8bfb28f","revisions":{"78bdb0f17785a0a052d1bdf4f14f285ed8bfb28f":{"kind":"REWORK","_number":1,"created":"2020-01-04 14:53:41.000000000","uploader":{"_account_id":11109,"name":"Roger Luethi","email":"rl@patchworkscience.org","username":"rluethi"},"ref":"refs/changes/92/701092/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/training-labs","ref":"refs/changes/92/701092/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/training-labs refs/changes/92/701092/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/training-labs refs/changes/92/701092/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/training-labs refs/changes/92/701092/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/training-labs refs/changes/92/701092/1"}}},"commit":{"parents":[{"commit":"148ef079e4100483670185e9754ae57dce11a5c4","subject":"Introduce placement-api as a separate service","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/training-labs/commit/148ef079e4100483670185e9754ae57dce11a5c4"}]}],"author":{"name":"Roger Luethi","email":"rl@patchworkscience.org","date":"2019-12-07 12:29:34.000000000","tz":60},"committer":{"name":"Roger Luethi","email":"rl@patchworkscience.org","date":"2020-01-04 14:51:44.000000000","tz":60},"subject":"setup_glance: replace openstack client with glance client","message":"setup_glance: replace openstack client with glance client\n\nSee openstack/glance repo:\n\n82697983c0cbeead47dfb6df9789ad88a8995af7\n  \"Use `glance` as example in docs instead of osc\"\n\nChange-Id: I7defb743e5791ca897cdc6736b5031ed015d9a02\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/training-labs/commit/78bdb0f17785a0a052d1bdf4f14f285ed8bfb28f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/training-labs/commit/78bdb0f17785a0a052d1bdf4f14f285ed8bfb28f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
