)]}'
{"id":"openstack%2Ftripleo-ci~766621","triplet_id":"openstack%2Ftripleo-ci~master~Ic31ebeca3dcb20b092ae2bfd890070c2c1b7bd66","project":"openstack/tripleo-ci","branch":"master","hashtags":[],"change_id":"Ic31ebeca3dcb20b092ae2bfd890070c2c1b7bd66","subject":"remove c7-train c7000-multinode-oooq-container-upgrades","status":"ABANDONED","created":"2020-12-10 21:46:55.000000000","updated":"2021-09-14 15:37:36.000000000","total_comment_count":2,"unresolved_comment_count":2,"has_review_started":true,"meta_rev_id":"45dad378325c52bc72f08b07aa779634012d91f4","_number":766621,"virtual_id_number":766621,"owner":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"all":[{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-12-14 03:39:35.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2020-12-14 02:26:49.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":30750,"name":"amolkahat","display_name":"Amol Kahat","email":"amolkahat@gmail.com","username":"amolkahat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"all":[{"value":-1,"date":"2020-12-14 09:17:59.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":2,"date":"2020-12-14 00:16:39.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":1,"date":"2020-12-14 05:00:19.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":30750,"name":"amolkahat","display_name":"Amol Kahat","email":"amolkahat@gmail.com","username":"amolkahat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":30750,"name":"amolkahat","display_name":"Amol Kahat","email":"amolkahat@gmail.com","username":"amolkahat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":30750,"name":"amolkahat","display_name":"Amol Kahat","email":"amolkahat@gmail.com","username":"amolkahat"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-10 23:31:55.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-11 04:10:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-14 05:00:19.000000000","updated_by":{"_account_id":30750,"name":"amolkahat","display_name":"Amol Kahat","email":"amolkahat@gmail.com","username":"amolkahat"},"reviewer":{"_account_id":30750,"name":"amolkahat","display_name":"Amol Kahat","email":"amolkahat@gmail.com","username":"amolkahat"},"state":"REVIEWER"},{"updated":"2020-12-14 09:17:59.000000000","updated_by":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"reviewer":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"state":"REVIEWER"}],"messages":[{"id":"06f8dbdd4c90bdee2000855ec2a62cea58fbbaa7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2020-12-10 21:46:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1c80370764b780f309824d479367b5ab8533e81a","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-12-10 23:31:55.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-ussuri-branch https://review.rdoproject.org/zuul/build/ca7f86d842e945aba803ed5487bfde8e : FAILURE in 1h 41m 53s (non-voting)\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/7dc54aed92434df8b61fef578a2c15f1 : FAILURE in 1h 13m 34s","accounts_in_message":[],"_revision_number":1},{"id":"4a0e58b7a1f63e208cdc36272b1f58b9a3fbe9d8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-11 04:10:41.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/34ce5f5c7d654ff6ba0078f93a091028 : SUCCESS in 3h 09m 38s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/2b2d97ac05f049eab17f8199d5da03c6 : SUCCESS in 3h 08m 41s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/7ac9df16d9e3436e8494249cbe158ca3 : SUCCESS in 5h 05m 21s\n- tripleo-ci-centos-7-content-provider-train https://zuul.opendev.org/t/openstack/build/4fa09da2cd394f7c94ff7872d236b60f : SUCCESS in 3h 23m 23s\n- tripleo-ci-centos-7-content-provider-queens https://zuul.opendev.org/t/openstack/build/5d2192e009d446c1b20f8c10e5e52fb8 : SUCCESS in 2h 54m 27s\n- tripleo-ci-centos-8-containers-multinode-victoria https://zuul.opendev.org/t/openstack/build/0d5a8c864d084f1fb3d3c154bb3decf9 : SUCCESS in 1h 59m 08s\n- tripleo-ci-centos-8-containers-multinode-ussuri https://zuul.opendev.org/t/openstack/build/ed128cd4b2324ae4bb7681b5c17f31b9 : SUCCESS in 1h 59m 55s\n- tripleo-ci-centos-8-containers-multinode-train https://zuul.opendev.org/t/openstack/build/f59ebb23ca8d4b3d93c7f4567df19d21 : TIMED_OUT in 3h 34m 55s\n- tripleo-ci-centos-7-containers-multinode-train https://zuul.opendev.org/t/openstack/build/f3b16d65d99f4502abcbd5f7a1e1418c : SUCCESS in 2h 07m 27s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-queens https://zuul.opendev.org/t/openstack/build/f7f43546e853430699f3d30cf1393fd6 : SUCCESS in 1h 41m 17s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/ca57f4e1ed0346d0949cd9ea15bd7972 : SUCCESS in 6h 21m 35s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/8ba958545f0c4c15810657aba98be30e : SUCCESS in 3h 25m 01s\n- tripleo-ci-centos-8-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/6767e58111bf401dbd514333fa607289 : SUCCESS in 1h 35m 48s (non-voting)\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/275c43d990d542e1b189d10eaab033d1 : SUCCESS in 1h 35m 45s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/6858773ff860424f91ed63d3da82405f : SUCCESS in 1h 38m 26s (non-voting)\n- tripleo-ci-centos-8-containers-undercloud-minion https://zuul.opendev.org/t/openstack/build/1541523106cb4bc6b564d856a822b013 : SUCCESS in 1h 09m 47s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/13afe74ccf19443e88bbd4cdc1a394b4 : SUCCESS in 1h 10m 24s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/a9d62fe407d747c987ca350a101c8698 : SUCCESS in 5m 47s\n- tripleo-tox-molecule https://zuul.opendev.org/t/openstack/build/4078b4fc316b41b290e0c51c742c671a : SUCCESS in 9m 40s","accounts_in_message":[],"_revision_number":1},{"id":"1104ac8ba0e676a7932cd4274ab00ac48905fa04","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2020-12-14 00:15:52.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"ad80c5474f6d0f4463aa2578f937a0052fd57402","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2020-12-14 00:16:39.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"962a75a3a1d0aae99a74c62519cbd8fafa1e5c10","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-12-14 02:26:49.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-ussuri-branch https://review.rdoproject.org/zuul/build/85819e2d14d24d28a1449b679c741294 : FAILURE in 1h 13m 22s (non-voting)\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/bb3940d0e2fd4421bb1a8c3a9bd0841d : FAILURE in 2h 09m 05s","accounts_in_message":[],"_revision_number":2},{"id":"27e58b5938bbadc5f0be7a210860771c4fa38951","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-14 03:39:35.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/fff84bbb89c74f4f88b08b4565798a5e : SUCCESS in 3h 03m 05s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/d7bbd19924c34804b69f8fec0e61d081 : SUCCESS in 3h 08m 07s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/34f6014944cd4a448d3c6d5ceb02ce98 : SUCCESS in 2h 52m 55s\n- tripleo-ci-centos-7-content-provider-train https://zuul.opendev.org/t/openstack/build/4b1424e9873749d9bb6c802a0a45c2dc : SUCCESS in 3h 21m 26s\n- tripleo-ci-centos-7-content-provider-queens https://zuul.opendev.org/t/openstack/build/a7d82f5227cb41b7b9ed1641feebff19 : SUCCESS in 2h 55m 15s\n- tripleo-ci-centos-8-containers-multinode-victoria https://zuul.opendev.org/t/openstack/build/4b07f3ab186447e98b36a54f5defe441 : SUCCESS in 1h 58m 48s\n- tripleo-ci-centos-8-containers-multinode-ussuri https://zuul.opendev.org/t/openstack/build/a481c363d44347c695bba124808fc0bb : SUCCESS in 1h 59m 11s\n- tripleo-ci-centos-8-containers-multinode-train https://zuul.opendev.org/t/openstack/build/417782340e7c418782ad2a339ef8c1b1 : SUCCESS in 1h 57m 25s\n- tripleo-ci-centos-7-containers-multinode-train https://zuul.opendev.org/t/openstack/build/93bb889d288f4322806447b787fb0f75 : SUCCESS in 2h 01m 41s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-queens https://zuul.opendev.org/t/openstack/build/3ccfba3cc3e54719826251b298cca58c : SUCCESS in 1h 46m 21s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/6d4a32baadb04d9bab3e49de31bf1a4a : SUCCESS in 3h 08m 47s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/c05eb4c2f65b43008d2f4fc2d4edeee7 : SUCCESS in 1h 58m 50s\n- tripleo-ci-centos-8-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/e26740e11ddc4862ab7a09653161fb09 : SUCCESS in 1h 52m 27s (non-voting)\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/fd85e739c9564390b5eb6a97a36ac9d6 : SUCCESS in 1h 52m 09s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/7c1972afe7fa4696a5691e383c32818c : SUCCESS in 1h 59m 46s (non-voting)\n- tripleo-ci-centos-8-containers-undercloud-minion https://zuul.opendev.org/t/openstack/build/fccba71499f34d9b8989cca674d9f7d2 : SUCCESS in 1h 13m 47s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/38b4e373f2384405a9369070a916a9ab : SUCCESS in 1h 26m 48s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/6639d7a1cb4941cdb85a9b81a19b965d : SUCCESS in 4m 13s\n- tripleo-tox-molecule https://zuul.opendev.org/t/openstack/build/b56e3646ec2644debb73a9c015e413a9 : SUCCESS in 8m 38s","accounts_in_message":[],"_revision_number":2},{"id":"e81001ca149121e19b70b98f638a14b349a768dc","author":{"_account_id":30750,"name":"amolkahat","display_name":"Amol Kahat","email":"amolkahat@gmail.com","username":"amolkahat"},"date":"2020-12-14 05:00:19.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"f284339292b471ae857241db5f434831ebd160bf","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2020-12-14 09:17:59.000000000","message":"Patch Set 2: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"45dad378325c52bc72f08b07aa779634012d91f4","tag":"autogenerated:gerrit:abandon","author":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"date":"2021-09-14 15:37:36.000000000","message":"Abandoned\n\nAbandoning this patch per the TripleO Patch Abandonment guidelines\n(https://specs.openstack.org/openstack/tripleo-specs/specs/policy/patch-abandonment.html).\nIf you wish to have this restored and cannot do so yourself, please reach out\nvia #tripleo on OFTC or the OpenStack Dev mailing list.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c54bfd221f061f31fece5283b413c233c766043d","revisions":{"3614c54ad956f06f65f669c624861055e061c5fb":{"kind":"REWORK","_number":1,"created":"2020-12-10 21:46:55.000000000","uploader":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"ref":"refs/changes/21/766621/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-ci","ref":"refs/changes/21/766621/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-ci refs/changes/21/766621/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-ci refs/changes/21/766621/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-ci refs/changes/21/766621/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-ci refs/changes/21/766621/1"}}},"commit":{"parents":[{"commit":"c66ef1c5b61d7d34ebd55a65c8670996e1dd71a0","subject":"Merge \"Assure lower-constraints.txt do not trigger any deployments\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-ci/commit/c66ef1c5b61d7d34ebd55a65c8670996e1dd71a0"}]}],"author":{"name":"Wes Hayutin","email":"weshayutin@gmail.com","date":"2020-12-10 21:46:02.000000000","tz":-420},"committer":{"name":"Wes Hayutin","email":"weshayutin@gmail.com","date":"2020-12-10 21:46:07.000000000","tz":-420},"subject":"remove c7-train c7000-multinode-oooq-container-upgrades","message":"remove c7-train c7000-multinode-oooq-container-upgrades\n\nIt\u0027s my understanding that upgrade jobs were\nbanned from c7-train\n\nChange-Id: Ic31ebeca3dcb20b092ae2bfd890070c2c1b7bd66\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-ci/commit/3614c54ad956f06f65f669c624861055e061c5fb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-ci/commit/3614c54ad956f06f65f669c624861055e061c5fb"}]},"branch":"refs/heads/master"},"c54bfd221f061f31fece5283b413c233c766043d":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-12-14 00:15:52.000000000","uploader":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"ref":"refs/changes/21/766621/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-ci","ref":"refs/changes/21/766621/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-ci refs/changes/21/766621/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-ci refs/changes/21/766621/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-ci refs/changes/21/766621/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-ci refs/changes/21/766621/2"}}},"commit":{"parents":[{"commit":"c33b75aab5565326612c9ae5028c1ca7a1efca0f","subject":"exclude ubi-8-containers from train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-ci/commit/c33b75aab5565326612c9ae5028c1ca7a1efca0f"}]}],"author":{"name":"Wes Hayutin","email":"weshayutin@gmail.com","date":"2020-12-10 21:46:02.000000000","tz":-420},"committer":{"name":"wes hayutin","email":"whayutin@redhat.com","date":"2020-12-14 00:15:52.000000000","tz":0},"subject":"remove c7-train c7000-multinode-oooq-container-upgrades","message":"remove c7-train c7000-multinode-oooq-container-upgrades\n\nIt\u0027s my understanding that upgrade jobs were\nbanned from c7-train\n\nChange-Id: Ic31ebeca3dcb20b092ae2bfd890070c2c1b7bd66\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-ci/commit/c54bfd221f061f31fece5283b413c233c766043d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-ci/commit/c54bfd221f061f31fece5283b413c233c766043d"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
