)]}'
{"id":"openstack%2Ftripleo-ci~773413","triplet_id":"openstack%2Ftripleo-ci~master~I84db635c86f8d372bd8d14b38546c281f6ee40cb","project":"openstack/tripleo-ci","branch":"master","topic":"bz_1914061","hashtags":[],"change_id":"I84db635c86f8d372bd8d14b38546c281f6ee40cb","subject":"Check container build status to avoid push and pull","status":"ABANDONED","created":"2021-02-01 15:25:06.000000000","updated":"2021-09-14 15:38:08.000000000","total_comment_count":9,"unresolved_comment_count":8,"has_review_started":true,"meta_rev_id":"2acd7e6c4fe992b5aaa5a8e95327b967eddf45f8","_number":773413,"virtual_id_number":773413,"owner":{"_account_id":30750,"name":"amolkahat","display_name":"Amol Kahat","email":"amolkahat@gmail.com","username":"amolkahat"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-02-18 11:32:42.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-02-18 12:04:08.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"all":[{"value":1,"date":"2021-02-18 09:52:14.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"value":-1,"date":"2021-02-18 11:25:27.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-01 15:26:37.000000000","updated_by":{"_account_id":30750,"name":"amolkahat","display_name":"Amol Kahat","email":"amolkahat@gmail.com","username":"amolkahat"},"reviewer":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"state":"REVIEWER"},{"updated":"2021-02-01 15:26:37.000000000","updated_by":{"_account_id":30750,"name":"amolkahat","display_name":"Amol Kahat","email":"amolkahat@gmail.com","username":"amolkahat"},"reviewer":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"state":"REVIEWER"},{"updated":"2021-02-01 15:26:37.000000000","updated_by":{"_account_id":30750,"name":"amolkahat","display_name":"Amol Kahat","email":"amolkahat@gmail.com","username":"amolkahat"},"reviewer":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"state":"REVIEWER"},{"updated":"2021-02-01 15:26:37.000000000","updated_by":{"_account_id":30750,"name":"amolkahat","display_name":"Amol Kahat","email":"amolkahat@gmail.com","username":"amolkahat"},"reviewer":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"state":"REVIEWER"},{"updated":"2021-02-01 16:17:41.000000000","updated_by":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"reviewer":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"state":"REVIEWER"},{"updated":"2021-02-01 18:53:29.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-02 01:01:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-08 09:48:34.000000000","updated_by":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"reviewer":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"state":"REVIEWER"},{"updated":"2021-02-11 07:54:39.000000000","updated_by":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2021-02-18 08:10:18.000000000","updated_by":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"reviewer":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"state":"REVIEWER"}],"messages":[{"id":"b63e97f15aa184cce7651fdf11a80964af311c89","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30750,"name":"amolkahat","display_name":"Amol Kahat","email":"amolkahat@gmail.com","username":"amolkahat"},"date":"2021-02-01 15:25:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a9e486acbb64422b9fc970fd87cb5d1d2572689a","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2021-02-01 16:17:41.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"f3507c6f76f1d9c701d109ef0f1af46eff870d31","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-02-01 18:53:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-ussuri-branch https://review.rdoproject.org/zuul/build/d532924ad2b34fea8344fc63e0679101 : SUCCESS in 2h 43m 35s (non-voting)\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/8c552974c33b46aa94d97ffeb4e51c5b : SUCCESS in 2h 53m 22s","accounts_in_message":[],"_revision_number":1},{"id":"0a44b280d813b928ace8a595cce0a7b37efe21df","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2021-02-01 19:41:27.000000000","message":"Patch Set 1: Code-Review-1\n\nBuild report should ignore errors, it\u0027s just a report and runs in post playbook. We need to find why run playbook doesn\u0027t fail when it should.","accounts_in_message":[],"_revision_number":1},{"id":"f16c8b00acdf9ec79517dbdfd8d023f7dba78b19","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-02 01:01:26.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tripleo-build-containers-centos-7-buildah-train https://zuul.opendev.org/t/openstack/build/9854be275a4e4cd49a0790ec860c3324 : POST_FAILURE in 53m 50s\n- tripleo-build-containers-centos-7-buildah-stein https://zuul.opendev.org/t/openstack/build/952112cbbb6d4119876e2241561d866c : POST_FAILURE in 18m 51s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/dca2c95fe97b4a89a39c03fca771d9b8 : SUCCESS in 1h 07m 31s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/0bc3ea26a20d4ff99a1d966fe4b7f912 : SUCCESS in 1h 06m 45s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/9cb0346f057b490db5e96e1bc21a6aa1 : SUCCESS in 51m 47s\n- tripleo-ci-centos-7-content-provider-train https://zuul.opendev.org/t/openstack/build/fdbb106db21b4740893d8e8d680ab6b4 : POST_FAILURE in 1h 15m 06s\n- tripleo-ci-centos-7-content-provider-queens https://zuul.opendev.org/t/openstack/build/5fd26fb5d6244c098965aa8fa0b29287 : POST_FAILURE in 16m 38s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/b649c761086d459f9568f7ebfbd49059 : SUCCESS in 1h 04m 07s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/2dfd499c7e134e6c88c4666c5bbf85ce : SUCCESS in 5m 04s\n- tripleo-tox-molecule https://zuul.opendev.org/t/openstack/build/7edacab3980940a292f735ea90c5d43a : SUCCESS in 9m 27s","accounts_in_message":[],"_revision_number":1},{"id":"60e7149239668280ffaa2426706083bb5c19aefb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30750,"name":"amolkahat","display_name":"Amol Kahat","email":"amolkahat@gmail.com","username":"amolkahat"},"date":"2021-02-04 10:06:56.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a148560dcd6030229c1cd465db6e5aae765bdd58","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-04 11:29:54.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tripleo-build-containers-centos-7-buildah-train https://zuul.opendev.org/t/openstack/build/9ab89323c95b4f22a346e3d0241fdb65 : FAILURE in 1h 02m 01s\n- tripleo-build-containers-centos-7-buildah-stein https://zuul.opendev.org/t/openstack/build/c4500e1bf4f746ff85a01d00b9a53151 : FAILURE in 1h 02m 52s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/c523e35e76f74ed29d6f5e819ffedc7c : SUCCESS in 1h 06m 19s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/f34473fe9e914441a3865698f9aeacdf : SUCCESS in 1h 08m 55s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/99f3ccd649ed4b04803a5c4a49efbe2d : SUCCESS in 54m 15s\n- tripleo-ci-centos-7-content-provider-train https://zuul.opendev.org/t/openstack/build/14cb45976376448cba577e64bb37dbf8 : FAILURE in 1h 15m 53s\n- tripleo-ci-centos-7-content-provider-queens https://zuul.opendev.org/t/openstack/build/33176e0e1d2e4f7ea074f5388acb015c : FAILURE in 53m 14s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/d2437eaf95a04685a218f827347344ab : SUCCESS in 1h 06m 14s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/7b34d226ec26412d95ae93a40eb7c63a : SUCCESS in 4m 12s\n- tripleo-tox-molecule https://zuul.opendev.org/t/openstack/build/7d3f5c7b333e4c50bdf038c7d82dd9ee : SUCCESS in 7m 56s","accounts_in_message":[],"_revision_number":2},{"id":"4162b00137999befb79a5aaf8f748925335fd382","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-02-04 13:29:17.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-victoria-branch https://review.rdoproject.org/zuul/build/fed502f97f7c42959afa8d48b23a061a : SUCCESS in 2h 55m 38s (non-voting)\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/393acf9f811d4eabaa262a78cf7e2d86 : SUCCESS in 3h 12m 35s","accounts_in_message":[],"_revision_number":2},{"id":"d27b6e7490e4413659b15508059a80db683c9834","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30750,"name":"amolkahat","display_name":"Amol Kahat","email":"amolkahat@gmail.com","username":"amolkahat"},"date":"2021-02-05 08:09:46.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"a395efdf73c6f296ea986e98b8e4fbc2f8b6feb3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-05 09:31:04.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tripleo-build-containers-centos-7-buildah-train https://zuul.opendev.org/t/openstack/build/5e682eab8c974874a26547292aae47f2 : FAILURE in 1h 02m 33s\n- tripleo-build-containers-centos-7-buildah-stein https://zuul.opendev.org/t/openstack/build/91eacbd81ca443c98973e6afd0676c86 : FAILURE in 1h 01m 39s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/b0bfcf06ba4b4fd888141a99645f359b : SUCCESS in 1h 01m 23s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/d7018137363e45bd99d62ff217b269d0 : SUCCESS in 58m 06s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/54ec9ee2af3a42ae802b480888c3a195 : SUCCESS in 52m 43s\n- tripleo-ci-centos-7-content-provider-train https://zuul.opendev.org/t/openstack/build/e242157f5b2f4ed2a043114a90eb9713 : FAILURE in 1h 16m 56s\n- tripleo-ci-centos-7-content-provider-queens https://zuul.opendev.org/t/openstack/build/95b1b4e5d38b4194bd15fb5efca1dfee : FAILURE in 40m 34s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/3c1b3ac3c246442f91a3c442bc56788b : SUCCESS in 1h 14m 27s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/c266bfe91c714338a712d4d2657c574c : SUCCESS in 4m 45s\n- tripleo-tox-molecule https://zuul.opendev.org/t/openstack/build/f894a9c4245f467f9a34df6e76f0f194 : SUCCESS in 8m 05s","accounts_in_message":[],"_revision_number":3},{"id":"565f9ad1a9a81adc30683bb09bed6c6e0bc83793","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30750,"name":"amolkahat","display_name":"Amol Kahat","email":"amolkahat@gmail.com","username":"amolkahat"},"date":"2021-02-05 09:48:43.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"1c0d7f06ac8cbdef1bdd6348a008ae3028beb870","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-05 11:12:01.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-7-buildah-train https://zuul.opendev.org/t/openstack/build/d72755015829421d85b6a6da81dfb1ee : SUCCESS in 1h 02m 55s\n- tripleo-build-containers-centos-7-buildah-stein https://zuul.opendev.org/t/openstack/build/00b1d743cad340a2bdeb29123a3d6eda : SUCCESS in 1h 01m 06s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/31c275c28a874ea0af3a07eac841603d : SUCCESS in 59m 10s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/43ccc258c74a43c5a728b65ea34c455c : SUCCESS in 1h 05m 15s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/28564af58f0841eaac2acecd85fe1aa5 : SUCCESS in 56m 24s\n- tripleo-ci-centos-7-content-provider-train https://zuul.opendev.org/t/openstack/build/605c1f5c0386456d8ad42baf56aa5601 : SUCCESS in 1h 19m 46s\n- tripleo-ci-centos-7-content-provider-queens https://zuul.opendev.org/t/openstack/build/0ad315bbd3fe45d49263fda712962e1a : SUCCESS in 1h 14m 27s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/b4524960b1d54bd1a51884e480f58241 : SUCCESS in 1h 05m 49s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/19b706f594954127b7b28a36786195ee : SUCCESS in 4m 34s\n- tripleo-tox-molecule https://zuul.opendev.org/t/openstack/build/f828703e4d044966bd8ee1b92d71c2b8 : SUCCESS in 8m 14s","accounts_in_message":[],"_revision_number":4},{"id":"b3863134a94f50bfaeb8a2fb72525c3191db4126","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-02-05 13:35:49.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-victoria-branch https://review.rdoproject.org/zuul/build/b11b6e67013443b6858b09ce9afeb2c2 : SUCCESS in 2h 29m 39s (non-voting)\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/829ac177c9fb4d9382fb263fb6a94f5b : SUCCESS in 3h 00m 15s","accounts_in_message":[],"_revision_number":4},{"id":"4ba4113550c08ec604000aa512e34a55e87cb971","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2021-02-08 09:35:17.000000000","message":"Patch Set 4:\n\nCan we send a dmn patch in tripleo-common by changing the container name in https://opendev.org/openstack/tripleo-common/src/branch/stable/queens/container-images/overcloud_containers.yaml#L7 and add depends-on here whether it works or not?","accounts_in_message":[],"_revision_number":4},{"id":"dd239724d2ebf71e28e80ef0b296cb509ab83d90","author":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"date":"2021-02-08 09:48:34.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"6ff8ad54918be3d54e4a44a558c8f26995c3c216","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30750,"name":"amolkahat","display_name":"Amol Kahat","email":"amolkahat@gmail.com","username":"amolkahat"},"date":"2021-02-08 09:49:37.000000000","message":"Uploaded patch set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"18c12eb47a67e9a3220abe04969cfbacbbb111e6","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2021-02-08 10:26:24.000000000","message":"Patch Set 5: Code-Review-1\n\nI don\u0027t think searching text in files is a good solution. We need to discover why previous tasks that actually build images don\u0027t fail when they should. This requires more investigation.","accounts_in_message":[],"_revision_number":5},{"id":"9f7b621cff6c3a52650592e5999bb2c07d807e65","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-08 11:54:57.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-7-buildah-train https://zuul.opendev.org/t/openstack/build/4f7b3d6820484498909aa94e00e6be6d : SUCCESS in 1h 40m 41s\n- tripleo-build-containers-centos-7-buildah-stein https://zuul.opendev.org/t/openstack/build/0c639642736f4f27ad661a3f2813abd4 : SUCCESS in 55m 27s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/014bd9ab5a8e410798573b5799d8be03 : SUCCESS in 47m 29s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/419ee872a515492fa956f5acda1f8acc : SUCCESS in 1h 01m 53s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/1687168f620740a395d7d2a35db632bd : SUCCESS in 49m 33s\n- tripleo-ci-centos-7-content-provider-train https://zuul.opendev.org/t/openstack/build/cdd393e2c7904defabb40af51cef969f : SUCCESS in 1h 18m 03s\n- tripleo-ci-centos-7-content-provider-queens https://zuul.opendev.org/t/openstack/build/b72de81a97274929be10f935247975e3 : SUCCESS in 2h 01m 35s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/c40e36abe3984c8483c8fd8d32f4f618 : SUCCESS in 1h 13m 02s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/3d62475b0a7541ce88bc5e091fc1f52d : SUCCESS in 4m 12s\n- tripleo-tox-molecule https://zuul.opendev.org/t/openstack/build/15b54e66cb7a4e5f99745d311999cf83 : SUCCESS in 7m 16s","accounts_in_message":[],"_revision_number":5},{"id":"af31198986b9e3369f118265ee1a4a3ada42c13d","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-02-08 12:28:31.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-victoria-branch https://review.rdoproject.org/zuul/build/09b5877650bc48588d08987ed50b7bb5 : SUCCESS in 2h 33m 36s (non-voting)\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/58ee863a457c4761ab2db3bcb74fa0d9 : SUCCESS in 2h 36m 41s","accounts_in_message":[],"_revision_number":5},{"id":"847f8ed6e0c4b0bf14ea8e07b3ab89d0f7ab768f","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2021-02-11 07:55:17.000000000","message":"Patch Set 5:\n\n(1 comment)\n\nright, this should be th","accounts_in_message":[],"_revision_number":5},{"id":"b824dc4ec12c68d8ce20637933f1e94a6d4aaeab","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30750,"name":"amolkahat","display_name":"Amol Kahat","email":"amolkahat@gmail.com","username":"amolkahat"},"date":"2021-02-17 09:39:15.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"7fc762586cc921005369f924b66ed3300f34657b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-17 11:44:26.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tripleo-build-containers-centos-7-buildah-train https://zuul.opendev.org/t/openstack/build/9e83b42514124902a91e3dc7ae075fc8 : SUCCESS in 1h 08m 45s\n- tripleo-build-containers-centos-7-buildah-stein https://zuul.opendev.org/t/openstack/build/a96d4e903b094d689151618dfb40967d : SUCCESS in 1h 01m 31s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/39a885d1e9734fa1b0c9bee3ed321349 : SUCCESS in 1h 06m 16s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/c99eb99014de45af889cb0d8d51318e2 : SUCCESS in 1h 09m 49s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/a2581e3d17824ea9861f494abb843ce2 : SUCCESS in 1h 04m 36s\n- tripleo-ci-centos-7-content-provider-train https://zuul.opendev.org/t/openstack/build/efd58e702eef4bc7a55b32da7eb1a743 : SUCCESS in 1h 21m 30s\n- tripleo-ci-centos-7-content-provider-queens https://zuul.opendev.org/t/openstack/build/76e7718e3c5a4450b72c0ca7f8ab49a7 : SUCCESS in 1h 42m 00s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/973a00d624ea4653b701f7e2057a9834 : SUCCESS in 1h 08m 16s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/a4697e08e1304323a0bbe027fdd3b7e5 : SUCCESS in 5m 54s\n- tripleo-tox-molecule https://zuul.opendev.org/t/openstack/build/bec96e826646490dbddadb3545e1e47c : FAILURE in 7m 19s","accounts_in_message":[],"_revision_number":6},{"id":"3a2edd91b07bf8ae225d8a2f0914d6952e2bcfa7","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-02-17 11:58:45.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-victoria-branch https://review.rdoproject.org/zuul/build/fb41849b219f4f15a128cd703bf97d09 : SUCCESS in 2h 16m 13s (non-voting)\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/cabe199cf31b4b70bf4bff9a5f9b48a4 : SUCCESS in 2h 17m 16s","accounts_in_message":[],"_revision_number":6},{"id":"ae598d3acb792d9651d0cc8a951c4dd66c44c476","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2021-02-18 04:57:44.000000000","message":"Patch Set 6:\n\nhttps://7a5854cdc73e90042b49-88701ef6bf8de11ee3b199ef1b631f6f.ssl.cf5.rackcdn.com/773413/6/check/tripleo-tox-molecule/bec96e8/tox/reports.html\n\n msg: \u0027Error building molecule_local/centos - code: None, message: toomanyrequests: You have reached your pull rate limit. You may increase the limit by authenticating and upgrading: https://www.docker.com/increase-rate-limit, logs: [\u0027\u0027Step 1/2 : FROM docker.io/centos:8\u0027\u0027, \u0027\u0027\\n\u0027\u0027]\u0027","accounts_in_message":[],"_revision_number":6},{"id":"81fb271346abb2f8ff9983f09baebb27dbb6c47d","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2021-02-18 04:57:56.000000000","message":"Patch Set 6:\n\nrecheck docker ratelimit","accounts_in_message":[],"_revision_number":6},{"id":"81743b3b2d58142cd07ce8638798f1a7d226eb23","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-18 06:17:41.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-7-buildah-train https://zuul.opendev.org/t/openstack/build/b01a2380b6434e2f9d42ee86d29f1de8 : SUCCESS in 33m 19s\n- tripleo-build-containers-centos-7-buildah-stein https://zuul.opendev.org/t/openstack/build/b6c88e6437184b108ac6850ded876048 : SUCCESS in 59m 24s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/d0b74b9dcc9b4e569ffb2f3a4e06f684 : SUCCESS in 56m 25s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/4e6655efdd18437a98ab1d569742aa3b : SUCCESS in 1h 11m 19s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/a62bc739489e4d02bfdf09915cff6150 : SUCCESS in 45m 00s\n- tripleo-ci-centos-7-content-provider-train https://zuul.opendev.org/t/openstack/build/741ef2179eca418d9cc815eb47482941 : SUCCESS in 1h 03m 22s\n- tripleo-ci-centos-7-content-provider-queens https://zuul.opendev.org/t/openstack/build/2204991aaacc490ba7888fda3b1db0ee : SUCCESS in 1h 08m 42s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/a79b7dd3a4874035836ad96c9b7001e7 : SUCCESS in 53m 58s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/5f3260c350654b878eb7f8d53036b0d7 : SUCCESS in 4m 23s\n- tripleo-tox-molecule https://zuul.opendev.org/t/openstack/build/7f665f8d4f8949e5b1e561c3170eab78 : SUCCESS in 8m 02s","accounts_in_message":[],"_revision_number":6},{"id":"c174947c722eb6ffed0833dfc0a6491a75babce1","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-02-18 07:12:02.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-victoria-branch https://review.rdoproject.org/zuul/build/b986914f8b984ca3b6fdc71394fc0535 : SUCCESS in 2h 12m 51s (non-voting)\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/3c98422d116445b39828d1cb05823bab : FAILURE in 1h 08m 23s","accounts_in_message":[],"_revision_number":6},{"id":"509fad91f8eca5de679090dbe57ba9ae3830653c","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2021-02-18 08:10:18.000000000","message":"Patch Set 6: Code-Review+2\n\n(1 comment)\n\nlgtm, inline bikeshedding :) about the variable name","accounts_in_message":[],"_revision_number":6},{"id":"8e59f0969b215c1c60387fdbb7d2bd4a9dcf9329","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2021-02-18 08:12:22.000000000","message":"Patch Set 6: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"e50655d5f8ba51117544aee0df6ccaf3cfe3d302","author":{"_account_id":30750,"name":"amolkahat","display_name":"Amol Kahat","email":"amolkahat@gmail.com","username":"amolkahat"},"date":"2021-02-18 09:27:39.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"9c04897c41cebc63bf74dd9db7375fb628751ecd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30750,"name":"amolkahat","display_name":"Amol Kahat","email":"amolkahat@gmail.com","username":"amolkahat"},"date":"2021-02-18 09:30:38.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"4f268165acc28de0137727a4a38532459253d5a9","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2021-02-18 09:52:14.000000000","message":"Patch Set 7: Code-Review+1\n\nthank you ping when you are ready to merge and remove the depends-on","accounts_in_message":[],"_revision_number":7},{"id":"68632f8fe4b5fc5917e6ddf41fee807c77c39b5f","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2021-02-18 09:55:32.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"41088c784c52ed8a2eb1d877b584a65f22795996","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2021-02-18 11:25:27.000000000","message":"Patch Set 7: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":7},{"id":"a2631386dd1f56ae1306b9d9bbd31a76e79e90f6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-18 11:32:42.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tripleo-build-containers-centos-7-buildah-train https://zuul.opendev.org/t/openstack/build/1e81099864514b7ea1dfbf853a7571bb : SUCCESS in 41m 43s\n- tripleo-build-containers-centos-7-buildah-stein https://zuul.opendev.org/t/openstack/build/aa92260526fd4d81a40e4bff4292cdf3 : SUCCESS in 1h 23m 19s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/1307c307c20243779be2337816337c93 : SUCCESS in 52m 05s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/2e1bf168e9474188b8abb448cc71beb1 : SUCCESS in 58m 48s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/708306484acf4f8b8c0f52dbea95a581 : SUCCESS in 1h 03m 56s\n- tripleo-ci-centos-7-content-provider-train https://zuul.opendev.org/t/openstack/build/fc81b76c7d794db3b2a751c8e1a0813a : FAILURE in 24m 47s\n- tripleo-ci-centos-7-content-provider-queens https://zuul.opendev.org/t/openstack/build/416d8ed87a4540bd8d29fbf8d8263977 : SUCCESS in 1h 26m 52s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/66cd8ae4d78c4536b716d42d80342d43 : SUCCESS in 57m 42s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/632b0307604942628531bec3e58cd628 : SUCCESS in 6m 07s\n- tripleo-tox-molecule https://zuul.opendev.org/t/openstack/build/61bb200adcae46d3a0aaffded2dd3f9a : SUCCESS in 9m 04s","accounts_in_message":[],"_revision_number":7},{"id":"ee5ccaa8b7c8136a455a0ea86cab67f846ad2144","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-02-18 12:04:08.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-victoria-branch https://review.rdoproject.org/zuul/build/613447149b5841078f51500cc6ac2598 : SUCCESS in 2h 18m 12s (non-voting)\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/9820e8c2246c4112ae413c0785705009 : SUCCESS in 2h 31m 33s","accounts_in_message":[],"_revision_number":7},{"id":"2acd7e6c4fe992b5aaa5a8e95327b967eddf45f8","tag":"autogenerated:gerrit:abandon","author":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"date":"2021-09-14 15:38:08.000000000","message":"Abandoned\n\nAbandoning this patch per the TripleO Patch Abandonment guidelines\n(https://specs.openstack.org/openstack/tripleo-specs/specs/policy/patch-abandonment.html).\nIf you wish to have this restored and cannot do so yourself, please reach out\nvia #tripleo on OFTC or the OpenStack Dev mailing list.","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"790a7e52ce58e80f9beb1770ee66093caca1046c","revisions":{"9552cd2a953497a9612317368ce142dca0e86f81":{"kind":"REWORK","_number":1,"created":"2021-02-01 15:25:06.000000000","uploader":{"_account_id":30750,"name":"amolkahat","display_name":"Amol Kahat","email":"amolkahat@gmail.com","username":"amolkahat"},"ref":"refs/changes/13/773413/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-ci","ref":"refs/changes/13/773413/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-ci refs/changes/13/773413/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-ci refs/changes/13/773413/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-ci refs/changes/13/773413/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-ci refs/changes/13/773413/1"}}},"commit":{"parents":[{"commit":"e4ad95dd2a4c03d2b6f0f40f242000600553ada2","subject":"Merge \"remove tripleo-ci-centos-7-standalone-upgrade-train\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-ci/commit/e4ad95dd2a4c03d2b6f0f40f242000600553ada2"}]}],"author":{"name":"Amol Kahat","email":"amolkahat@gmail.com","date":"2021-02-01 15:18:06.000000000","tz":330},"committer":{"name":"Amol Kahat","email":"amolkahat@gmail.com","date":"2021-02-01 15:24:31.000000000","tz":330},"subject":"Removed ignore_errors from build-reports.yaml","message":"Removed ignore_errors from build-reports.yaml\n\nSome tasks are failing and zuul will not get\nnotified because of ignore_errors.\n\nCloses-Bug: #1914061\n\nChange-Id: I84db635c86f8d372bd8d14b38546c281f6ee40cb\nSigned-off-by: Amol Kahat \u003camolkahat@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-ci/commit/9552cd2a953497a9612317368ce142dca0e86f81"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-ci/commit/9552cd2a953497a9612317368ce142dca0e86f81"}]},"branch":"refs/heads/master"},"0dc54feb585960d7dcb3f8cb8e43e17606056513":{"kind":"REWORK","_number":2,"created":"2021-02-04 10:06:56.000000000","uploader":{"_account_id":30750,"name":"amolkahat","display_name":"Amol Kahat","email":"amolkahat@gmail.com","username":"amolkahat"},"ref":"refs/changes/13/773413/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-ci","ref":"refs/changes/13/773413/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-ci refs/changes/13/773413/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-ci refs/changes/13/773413/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-ci refs/changes/13/773413/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-ci refs/changes/13/773413/2"}}},"commit":{"parents":[{"commit":"e4ad95dd2a4c03d2b6f0f40f242000600553ada2","subject":"Merge \"remove tripleo-ci-centos-7-standalone-upgrade-train\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-ci/commit/e4ad95dd2a4c03d2b6f0f40f242000600553ada2"}]}],"author":{"name":"Amol Kahat","email":"amolkahat@gmail.com","date":"2021-02-01 15:18:06.000000000","tz":330},"committer":{"name":"Amol Kahat","email":"amolkahat@gmail.com","date":"2021-02-04 10:06:30.000000000","tz":330},"subject":"Check container build status to avoid push and pull","message":"Check container build status to avoid push and pull\n\nContainer build status is not getting validated which\nleads to failure of container push and pull.\n\nCloses-Bug: #1914061\n\nChange-Id: I84db635c86f8d372bd8d14b38546c281f6ee40cb\nSigned-off-by: Amol Kahat \u003camolkahat@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-ci/commit/0dc54feb585960d7dcb3f8cb8e43e17606056513"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-ci/commit/0dc54feb585960d7dcb3f8cb8e43e17606056513"}]},"branch":"refs/heads/master"},"4561595fcd230297d3d3b182a90bf536052427f4":{"kind":"REWORK","_number":3,"created":"2021-02-05 08:09:46.000000000","uploader":{"_account_id":30750,"name":"amolkahat","display_name":"Amol Kahat","email":"amolkahat@gmail.com","username":"amolkahat"},"ref":"refs/changes/13/773413/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-ci","ref":"refs/changes/13/773413/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-ci refs/changes/13/773413/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-ci refs/changes/13/773413/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-ci refs/changes/13/773413/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-ci refs/changes/13/773413/3"}}},"commit":{"parents":[{"commit":"e9329e38808ca73daca43183a13b4ce0515f2f7e","subject":"Remove tripleoclient file pattern- Standalone scen","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-ci/commit/e9329e38808ca73daca43183a13b4ce0515f2f7e"}]}],"author":{"name":"Amol Kahat","email":"amolkahat@gmail.com","date":"2021-02-01 15:18:06.000000000","tz":330},"committer":{"name":"Amol Kahat","email":"amolkahat@gmail.com","date":"2021-02-05 08:09:36.000000000","tz":330},"subject":"Check container build status to avoid push and pull","message":"Check container build status to avoid push and pull\n\nContainer build status is not getting validated which\nleads to failure of container push and pull.\n\nCloses-Bug: #1914061\n\nChange-Id: I84db635c86f8d372bd8d14b38546c281f6ee40cb\nSigned-off-by: Amol Kahat \u003camolkahat@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-ci/commit/4561595fcd230297d3d3b182a90bf536052427f4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-ci/commit/4561595fcd230297d3d3b182a90bf536052427f4"}]},"branch":"refs/heads/master"},"8d2f4c0c3832012fe4ab43053a7448a5af9b01cf":{"kind":"REWORK","_number":4,"created":"2021-02-05 09:48:43.000000000","uploader":{"_account_id":30750,"name":"amolkahat","display_name":"Amol Kahat","email":"amolkahat@gmail.com","username":"amolkahat"},"ref":"refs/changes/13/773413/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-ci","ref":"refs/changes/13/773413/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-ci refs/changes/13/773413/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-ci refs/changes/13/773413/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-ci refs/changes/13/773413/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-ci refs/changes/13/773413/4"}}},"commit":{"parents":[{"commit":"e9329e38808ca73daca43183a13b4ce0515f2f7e","subject":"Remove tripleoclient file pattern- Standalone scen","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-ci/commit/e9329e38808ca73daca43183a13b4ce0515f2f7e"}]}],"author":{"name":"Amol Kahat","email":"amolkahat@gmail.com","date":"2021-02-01 15:18:06.000000000","tz":330},"committer":{"name":"Amol Kahat","email":"amolkahat@gmail.com","date":"2021-02-05 09:48:28.000000000","tz":330},"subject":"Check container build status to avoid push and pull","message":"Check container build status to avoid push and pull\n\nContainer build status is not getting validated which\nleads to failure of container push and pull.\n\nCloses-Bug: #1914061\n\nChange-Id: I84db635c86f8d372bd8d14b38546c281f6ee40cb\nSigned-off-by: Amol Kahat \u003camolkahat@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-ci/commit/8d2f4c0c3832012fe4ab43053a7448a5af9b01cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-ci/commit/8d2f4c0c3832012fe4ab43053a7448a5af9b01cf"}]},"branch":"refs/heads/master"},"57e03a868bcd8ec44099eb3999372721f2bdc876":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2021-02-08 09:49:37.000000000","uploader":{"_account_id":30750,"name":"amolkahat","display_name":"Amol Kahat","email":"amolkahat@gmail.com","username":"amolkahat"},"ref":"refs/changes/13/773413/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-ci","ref":"refs/changes/13/773413/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-ci refs/changes/13/773413/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-ci refs/changes/13/773413/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-ci refs/changes/13/773413/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-ci refs/changes/13/773413/5"}}},"commit":{"parents":[{"commit":"e9329e38808ca73daca43183a13b4ce0515f2f7e","subject":"Remove tripleoclient file pattern- Standalone scen","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-ci/commit/e9329e38808ca73daca43183a13b4ce0515f2f7e"}]}],"author":{"name":"Amol Kahat","email":"amolkahat@gmail.com","date":"2021-02-01 15:18:06.000000000","tz":330},"committer":{"name":"Amol Kahat","email":"amolkahat@gmail.com","date":"2021-02-08 09:48:47.000000000","tz":330},"subject":"Check container build status to avoid push and pull","message":"Check container build status to avoid push and pull\n\nContainer build status is not getting validated which\nleads to failure of container push and pull.\n\nDepends-On: https://review.opendev.org/c/openstack/tripleo-common/+/774424\n\nCloses-Bug: #1914061\n\nChange-Id: I84db635c86f8d372bd8d14b38546c281f6ee40cb\nSigned-off-by: Amol Kahat \u003camolkahat@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-ci/commit/57e03a868bcd8ec44099eb3999372721f2bdc876"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-ci/commit/57e03a868bcd8ec44099eb3999372721f2bdc876"}]},"branch":"refs/heads/master"},"67b457bb3383c15ee7d30d2ef632cc2ad84a15a5":{"kind":"REWORK","_number":6,"created":"2021-02-17 09:39:15.000000000","uploader":{"_account_id":30750,"name":"amolkahat","display_name":"Amol Kahat","email":"amolkahat@gmail.com","username":"amolkahat"},"ref":"refs/changes/13/773413/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-ci","ref":"refs/changes/13/773413/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-ci refs/changes/13/773413/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-ci refs/changes/13/773413/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-ci refs/changes/13/773413/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-ci refs/changes/13/773413/6"}}},"commit":{"parents":[{"commit":"695e57fbbb5ca1fdf09e88f114af78d341306ff5","subject":"Merge \"Fix conditions for Executing build-report\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-ci/commit/695e57fbbb5ca1fdf09e88f114af78d341306ff5"}]}],"author":{"name":"Amol Kahat","email":"amolkahat@gmail.com","date":"2021-02-01 15:18:06.000000000","tz":330},"committer":{"name":"Amol Kahat","email":"amolkahat@gmail.com","date":"2021-02-17 09:38:50.000000000","tz":330},"subject":"Check container build status to avoid push and pull","message":"Check container build status to avoid push and pull\n\nContainer build status is not getting validated which\nleads to failure of container push and pull.\n\nDepends-On: https://review.opendev.org/c/openstack/tripleo-common/+/774424\n\nCloses-Bug: #1914061\n\nChange-Id: I84db635c86f8d372bd8d14b38546c281f6ee40cb\nSigned-off-by: Amol Kahat \u003camolkahat@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-ci/commit/67b457bb3383c15ee7d30d2ef632cc2ad84a15a5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-ci/commit/67b457bb3383c15ee7d30d2ef632cc2ad84a15a5"}]},"branch":"refs/heads/master"},"790a7e52ce58e80f9beb1770ee66093caca1046c":{"kind":"REWORK","_number":7,"created":"2021-02-18 09:30:38.000000000","uploader":{"_account_id":30750,"name":"amolkahat","display_name":"Amol Kahat","email":"amolkahat@gmail.com","username":"amolkahat"},"ref":"refs/changes/13/773413/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-ci","ref":"refs/changes/13/773413/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-ci refs/changes/13/773413/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-ci refs/changes/13/773413/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-ci refs/changes/13/773413/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-ci refs/changes/13/773413/7"}}},"commit":{"parents":[{"commit":"695e57fbbb5ca1fdf09e88f114af78d341306ff5","subject":"Merge \"Fix conditions for Executing build-report\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-ci/commit/695e57fbbb5ca1fdf09e88f114af78d341306ff5"}]}],"author":{"name":"Amol Kahat","email":"amolkahat@gmail.com","date":"2021-02-01 15:18:06.000000000","tz":330},"committer":{"name":"Amol Kahat","email":"amolkahat@gmail.com","date":"2021-02-18 09:30:21.000000000","tz":330},"subject":"Check container build status to avoid push and pull","message":"Check container build status to avoid push and pull\n\nContainer build status is not getting validated which\nleads to failure of container push and pull.\n\nDepends-On: https://review.opendev.org/c/openstack/tripleo-common/+/774424\n\nCloses-Bug: #1914061\n\nChange-Id: I84db635c86f8d372bd8d14b38546c281f6ee40cb\nSigned-off-by: Amol Kahat \u003camolkahat@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-ci/commit/790a7e52ce58e80f9beb1770ee66093caca1046c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-ci/commit/790a7e52ce58e80f9beb1770ee66093caca1046c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
